)]}'
{"id":"openstack%2Fneutron~905913","triplet_id":"openstack%2Fneutron~stable%2Fyoga~Iaf15e560e1b1ec31618b2ebc6206a938463c1094","project":"openstack/neutron","branch":"stable/yoga","topic":"bug/2040172-stable/zed-stable/yoga","attention_set":{},"removed_from_attention_set":{"24245":{"account":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"last_update":"2024-01-18 17:54:20.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iaf15e560e1b1ec31618b2ebc6206a938463c1094","subject":"[OVN] DB sync host/physnet - filter on agent_type","status":"MERGED","created":"2024-01-18 14:20:11.000000000","updated":"2024-01-18 17:55:40.000000000","submitted":"2024-01-18 17:54:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"905913-bug/2040172-stable/zed-stable/yoga","cherry_pick_of_change":903685,"cherry_pick_of_patch_set":1,"meta_rev_id":"dfa8e0a7d4bab78f1a95d2202a307e635f4be4de","_number":905913,"virtual_id_number":905913,"owner":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-01-18 17:54:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2024-01-18 16:20:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2024-01-18 15:42:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2024-01-18 16:20:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-01-18 15:05:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-01-18 15:42:59.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2024-01-18 15:50:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-01-18 16:20:52.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"}],"messages":[{"id":"7d95edd62dfdb0ae10cd5e17677717ec488357ed","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"date":"2024-01-18 14:20:11.000000000","message":"Patch Set 1: Cherry Picked from branch stable/zed.","accounts_in_message":[],"_revision_number":1},{"id":"6f23d6cdb3e5939dfe4e283b48e4b10d16c81cd0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-18 15:05:54.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4067720728584bb18ba56bfdc48021b8\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a2f4035f0fb44d698f5b400009d6dd2b : SUCCESS in 43m 49s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9721727e1a734339981a3c5acbf6e484 : SUCCESS in 43m 18s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"965214aba39e8fb6d877333d5303c223bdbe13de","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-01-18 15:42:59.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"99c8e3361cd3e2294c1ee1e8fe6be326e220f84c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-18 15:50:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1528fa95c2914355bff2ef60c43aec8a\n\n- neutron-tempest-plugin-api-yoga https://zuul.opendev.org/t/openstack/build/52be7e04327f469f9d1ca80f381e11a6 : SUCCESS in 55m 47s\n- neutron-tempest-plugin-scenario-ovn-yoga https://zuul.opendev.org/t/openstack/build/6b39351ef5bf4f109fc0f44f960f36cb : SUCCESS in 1h 03m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f9b6fe72ae9b40cca85d8209a4fed55c : SUCCESS in 34m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6c734137fd1a4b7abf4664fff4ba5a72 : SUCCESS in 16m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0182adef3c644aafb877077e90a5544c : SUCCESS in 27m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4a7dcd3df9d044379b3e8b56d4876d5d : SUCCESS in 24m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/629f02b8a2264aa881fc5d0861459c91 : SUCCESS in 9m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3bc52bea5d824f01b3c6f35f45111d88 : SUCCESS in 14m 22s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/6e8457c052954cf3b4ec92fbb48d656a : SUCCESS in 59m 23s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d429a675fc884d7ba0dd9feb87900aa6 : SUCCESS in 1h 03m 38s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/9c1678872dbc466a8f83ef94e71be404 : SUCCESS in 1h 20m 01s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"53834b986c630d29f90507788fe9a44ee0779666","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-01-18 16:20:52.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a749fb2e73a171c0b49348bce08c37b1081a6929","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-18 16:21:38.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"04a07f224964428cbf293f264de9049035b9c35d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-18 17:54:20.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e3c5a44bcf674117bd48b1f871d1ec9a\n\n- neutron-tempest-plugin-api-yoga https://zuul.opendev.org/t/openstack/build/315ae356001c46feafe4c07cf6640159 : SUCCESS in 1h 00m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3c870552e957441aaa43655900deb1e6 : SUCCESS in 14m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/268159f83b12415d864d1796a5d7f662 : SUCCESS in 19m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/08b68d5115fb4794b27fc9797424ef50 : SUCCESS in 25m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6dfb4c6c507a461787c1420e50291bdc : SUCCESS in 10m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6af83366ce744b1782222afc36856874 : SUCCESS in 17m 28s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/35bc5072c5834db48abf220889546902 : SUCCESS in 57m 22s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/38c3c400f49242e2a96afe00b5e264e9 : SUCCESS in 51m 45s","accounts_in_message":[],"_revision_number":1},{"id":"e8e0d5252cc10c363d9b7e854df3b85c446c51f0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-18 17:54:20.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"dfa8e0a7d4bab78f1a95d2202a307e635f4be4de","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-18 17:55:40.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3dbdd4fa7b2647888404fd5d9c092cbb\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/692d38693d4a4d0386b660afe664d7e5 : SUCCESS in 56s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f7bfe65440f54422a2546e1e8a86b68d : SUCCESS in 54s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"37abd72fed99525c302182d276d1c175f5d01a70","revisions":{"37abd72fed99525c302182d276d1c175f5d01a70":{"kind":"REWORK","_number":1,"created":"2024-01-18 14:20:11.000000000","uploader":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"ref":"refs/changes/13/905913/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/13/905913/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/905913/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/905913/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/905913/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/13/905913/1"}}},"commit":{"parents":[{"commit":"7d8ea71b3af14bfc6ef47229e9d2f5020f659f10","subject":"Merge \"Register Chassis_Private table in BaseOvnSbIdl\" into stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7d8ea71b3af14bfc6ef47229e9d2f5020f659f10"}]}],"author":{"name":"Harald Jensås","email":"hjensas@redhat.com","date":"2023-10-23 15:29:00.000000000","tz":120},"committer":{"name":"Harald Jensås","email":"hjensas@redhat.com","date":"2024-01-18 14:20:11.000000000","tz":0},"subject":"[OVN] DB sync host/physnet - filter on agent_type","message":"[OVN] DB sync host/physnet - filter on agent_type\n\nWhen syncing hostname and physical networks, filter neutron\nhosts on agent_type. Only segmenthostmappings for hosts\nwith agent \u0027OVN Controller agent\u0027 should be cleaned up.\n\nSince change: I935186b6ee95f0cae8dc05869d9742c8fb3353c3 there\nis de-duplication of segmenthostmapping updates from agents.\nIf the OVN DB sync clears/deletes mappings for hosts owned by\nother agents/plugins the mappings are never re-created.\n\nCloses-Bug: #2040172\n\nChange-Id: Iaf15e560e1b1ec31618b2ebc6206a938463c1094\nSigned-off-by: Harald Jensås \u003chjensas@redhat.com\u003e\n(cherry picked from commit 71d69cf6277ba553354512209a2bff61c013f8ea)\n(cherry picked from commit 4eb234a4476761ca2581dba45655caceb21fc1ee)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/37abd72fed99525c302182d276d1c175f5d01a70"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/37abd72fed99525c302182d276d1c175f5d01a70"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
