)]}'
{"id":"openstack%2Fneutron~906006","triplet_id":"openstack%2Fneutron~master~Id9d6d841133241bbc87a589117468c4e699c310a","project":"openstack/neutron","branch":"master","topic":"fix-keyword-arg-before-vararg","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2024-04-22 15:49:02.000000000","reason":"Change was submitted"},"9656":{"account":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"last_update":"2024-02-21 15:59:06.000000000","reason":"\u003cGERRIT_ACCOUNT_9656\u003e replied on the change","reason_account":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"}},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2024-02-20 13:57:31.000000000","reason":"\u003cGERRIT_ACCOUNT_8313\u003e replied on the change","reason_account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2024-04-22 15:49:02.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id9d6d841133241bbc87a589117468c4e699c310a","subject":"Fix keyword-arg-before-vararg warnings","status":"MERGED","created":"2024-01-17 21:09:01.000000000","updated":"2024-04-22 15:51:15.000000000","submitted":"2024-04-22 15:49:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"906006-fix-keyword-arg-before-vararg","meta_rev_id":"79e374ff3fbcce303b55ac6d75ba73f97c6e1e88","_number":906006,"virtual_id_number":906006,"owner":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2024-02-19 21:53:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-04-22 15:49:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2024-04-22 13:35:02.000000000","_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"value":2,"date":"2024-02-20 13:57:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-02-20 02:25:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":1,"date":"2024-04-22 13:35:02.000000000","_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-01-17 21:53:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-01-17 22:09:33.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-01-17 22:54:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-01-18 20:48:04.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-02-06 17:24:34.000000000","updated_by":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"reviewer":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"state":"REVIEWER"},{"updated":"2024-02-13 01:00:41.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2024-02-13 01:00:41.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2024-02-20 02:25:04.000000000","updated_by":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"reviewer":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"state":"REVIEWER"}],"messages":[{"id":"ffe52236c7923fc6856843a9c0464a0d496a0964","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-01-17 21:09:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ac04bf3746b12daf84ba618ea7bb26386d5b949a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-17 21:53:46.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8ae53e28ed2e40509f040ec130d061b4\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6830dfb0c8f942319f3c8f1e52342d30 : FAILURE in 42m 05s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/8e768274ae204b6aa6451ab377844281 : FAILURE in 17m 27s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/76cefc929aed4dc39d914b6df359e66f : FAILURE in 15m 04s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1a35bcb01f45de623764996f49d1bd058247b6d5","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2024-01-17 22:09:33.000000000","message":"Patch Set 1:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/January-2024/Arista_Tempest_Testrefs/changes/06/906006/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"b73bde265a9fbf4230684b001e8c77a9cb66dd89","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-17 22:54:44.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a02689cad6ba485fb97acfcd41c1f0f0\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/fba12706740e4a1582b34f2f3948369e : FAILURE in 14m 28s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/58ec5930a7cc499b8a4c66bb7ce393ba : FAILURE in 16m 49s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/7c272a5ab82849d4b5a8127e09111689 : FAILURE in 14m 30s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/d39522da5557480ea638cbfc2d32e9df : FAILURE in 19m 44s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/bbcabfa8bed64bc99e1659c55aead363 : FAILURE in 16m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5b7858cfbd864037bef5c62b11a70ed2 : FAILURE in 24m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/27a9bff3206949cf85df7bc267f99a80 : SUCCESS in 9m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ab0ded2762344ce39ea9036c0b9d8d28 : FAILURE in 21m 16s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b16def152c2d472caa74ad4ea2ba7ba0 : FAILURE in 22m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a9bab48e77a4c52a06b6de83a9aac39 : SUCCESS in 10m 40s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/3715655294974d8fa0861842e22aae78 : FAILURE in 22m 27s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/509480b5528f470f8f492b40bfb4c6f1 : FAILURE in 56m 46s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/4df24cf43ecf4eb7a73c7c8b0dfaef05 : FAILURE in 1h 01m 34s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/705e4f2b987646e4a5f987d332853309 : FAILURE in 48m 58s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/1c7a410dabd443ff9825321731d36e01 : FAILURE in 1h 38m 48s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/149c597d04704e5c83d700c8e853741a : FAILURE in 53m 06s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/89ce44a9672c4f6395d38cb051ac75e5 : FAILURE in 55m 23s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/3d22263b4e194b1189862f2130ed3b32 : FAILURE in 30m 45s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/c03de1e5eea3459d8ce84e42d1cb0ba4 : FAILURE in 31m 59s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/bf86b4f31fbc4ddfb718fcb605723f1f : FAILURE in 29m 00s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/2459f353162a4d59a2c06fc9f0ae8d40 : FAILURE in 37m 12s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/329ee5cfbd8a4463851a83f824e4e3dc : FAILURE in 29m 39s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/0b7a61da8a4049998244b92d31a70c14 : FAILURE in 29m 33s","accounts_in_message":[],"_revision_number":1},{"id":"404a3fe736a581e29b6d16a9d8ccd12c5fa1c082","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-01-18 14:51:45.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"1fcfcfe305d92bc505afbf157446573383d1029b","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-01-18 20:16:11.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"758f22e1c2c0c9173795653b4c65e81fdd722df3","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2024-01-18 20:48:04.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/January-2024/Arista_Tempest_Testrefs/changes/06/906006/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"5e3f682d38c8976462cffe02af8ef65a31801873","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-01-18 20:49:22.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":2},{"id":"7bdf63b449571a707f409be1879f670cd1b0c2b5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-18 20:51:46.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8f9d936726654160a7963e7cd7e4a9f7\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9c5684fc4c9f48f8b92da73c71e4980e : SUCCESS in 17m 48s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/96ac5a17e2164b80a6706dc9c963f807 : SUCCESS in 17m 47s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/1bc369ddabb7404ab84c6bbe95196e61 : SUCCESS in 34m 42s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ea39f268c3169fb8b6d8ba6f8cb6d6c0c3801cc7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-18 22:45:16.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b0bc7dbd3b5b4100bdd4869d53772332\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/886af7fba51f4106a46ba823fa5693d7 : SUCCESS in 1h 20m 45s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/7fa05936c8124e24aead6e392ccd2ebb : SUCCESS in 1h 15m 46s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/c6490f81cff0429fa0d748b5b11a3f19 : SUCCESS in 1h 15m 45s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/cff02133cd8e44bb9c841e4644e6fbf8 : SUCCESS in 1h 29m 04s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/053ac862b0bc47fda93dae2ce9cdc65c : SUCCESS in 32m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e40fcd22b0734197944569b0b97f7640 : SUCCESS in 41m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d347e3b88d2e4261b13d2ee576cdc8a2 : SUCCESS in 11m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4e3ff978ce3d4086a03b2b1e33168b8b : SUCCESS in 28m 29s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/fcd15a9bbf2743e6b1ad4126cc42fa93 : SUCCESS in 22m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e9af02807e343b798cecc1ef8039200 : SUCCESS in 11m 11s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/8453115d0ff749cdbf0b1fcab7dfd316 : SUCCESS in 27m 33s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/6d098e788f4340b9a6243923b581ef1e : SUCCESS in 59m 23s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/7d79bf255a544acb8fc4afa81da2815b : SUCCESS in 1h 09m 03s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/4a94c92ed7d944a2818262b75b121a8b : SUCCESS in 59m 40s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/bf2c6f53c26b492cabfde2db4c57694c : SUCCESS in 1h 45m 04s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/4385efa82a024c509c666fd7e0556b6a : SUCCESS in 1h 08m 02s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/442b0d159bc24c59a40dec0efc722edf : SUCCESS in 1h 01m 52s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/e82e7b2092954a69b5f383c72260e467 : SUCCESS in 1h 29m 27s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/2f1f45ba9b6143a89de9f061975dcecb : SUCCESS in 1h 23m 51s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/15e5b53cb10b415f9a66805e6047df15 : SUCCESS in 1h 10m 06s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/408997ded5a9427191104dfe01355a22 : SUCCESS in 2h 07m 46s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/f136893a40a94c51890df60a3186394a : SUCCESS in 2h 05m 57s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/8708f92f0a0940aeb543de3b516cce0d : SUCCESS in 1h 10m 17s","accounts_in_message":[],"_revision_number":2},{"id":"412d4ef23e85bd4d4530cdf5f8c0e6705d96c7a6","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2024-02-06 17:24:34.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"f5931bc42443eb0265b8378e11f4d1fca28b9944","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-02-16 10:10:35.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"18ffd8f4b6bbf9037bf6e47af6348f031bf8670b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-02-19 21:21:44.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"64e532f786d463ee3db7e334bfa65c7bfffb628b","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-02-19 21:21:57.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"b3cdbb6a8a8cf781c68284b6ea83a38340eecf57","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2024-02-19 21:53:07.000000000","message":"Patch Set 3: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/February-2024/Arista_Tempest_Testrefs/changes/06/906006/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"3e6cc8e6527aea707457f2bfe747eb8d207dc7c5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-19 22:04:53.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f12f9346cb4e47c6bdd11820e58d42a0\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1fc7ba8e890141b9a82c4965600017f6 : SUCCESS in 30m 33s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/65cc52c3d4924944889ae8920f70986c : SUCCESS in 18m 44s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/e9a3efb43db741ba8a00899a18ccff52 : SUCCESS in 41m 09s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"aa874e5d870d2cc7fff80cefc227664a6bdc047a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-19 23:31:41.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/559537bb9b77403299e8a08ae00ccfe3\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/e9c4089ec9664f248183468bf59b2bf0 : SUCCESS in 1h 22m 27s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/967505609a53439eb32750527d720e63 : SUCCESS in 2h 01m 43s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/0eeecdd860a945b5b89997475f4ef3a7 : SUCCESS in 2h 07m 45s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/9256796c1c1447e39db7567f0b3de0fa : SUCCESS in 1h 29m 51s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/be86d11efdce4126b7611f93a7d0c3cf : SUCCESS in 31m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f1ddf33fbbd4461a826804e240b3e461 : SUCCESS in 40m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/33383ad7357c4058bd122a89ed3d7b43 : SUCCESS in 9m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e51a3902e4af4ecf93a84d4ede5644a1 : SUCCESS in 19m 15s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/431ec498f0c148e49450228b4409ae9c : SUCCESS in 23m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa1f4c0fef064a28a4e97b0f78106add : SUCCESS in 10m 20s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/d79a153b62654555a514935198f13c50 : SUCCESS in 21m 10s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/f84d4ac06ba341a79fb80d8170a3482f : SUCCESS in 59m 36s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/ae5cd088da5d41879549bd1028f503f7 : SUCCESS in 1h 05m 46s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/4e616e9389f64eb3ac2436ce2c4ee635 : SUCCESS in 52m 48s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/2b4a447ba8114465b0df93b3cc96aeb3 : SUCCESS in 1h 16m 39s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/786c394a0639407fa61f5548ef48a2fd : SUCCESS in 1h 04m 25s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/55db99bfd42b4a58a5528542d985ad43 : SUCCESS in 49m 13s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/f7c50bdd3bc649e082ded3da4f655334 : SUCCESS in 1h 20m 29s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/0faa6fe913de45f7a9b00272f2aa2136 : SUCCESS in 1h 21m 21s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/c4592885548c471886ea0c76a418882b : SUCCESS in 1h 01m 50s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/c7b1ff96e8e340cfb821e9ec2e56d59f : SUCCESS in 1h 13m 59s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/0fe9d46d48584ad4b40351dbfd1bfb99 : SUCCESS in 1h 01m 19s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/3acd7dd0f8c249948d5e137c645857bf : SUCCESS in 1h 04m 10s","accounts_in_message":[],"_revision_number":3},{"id":"9f796835b670a388eb867c1351fe480cc13dfec9","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2024-02-20 02:25:04.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"cd9c69a91bbfa848e341b3f002036bd4d6c44f2a","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-02-20 13:57:31.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"6b0112351c35d5dcd90b900fdac1be4c8509ae33","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2024-02-21 15:59:06.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"621db5d4bee040e307b30b5faa2c8f46734dbd4b","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2024-04-22 13:35:02.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"1aefa6977b6803838035d3226aca4926fa20d88f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-22 13:35:37.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"76e691450ba2be4dadeb2879ab93ecdd6a4bca1a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-22 15:49:02.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/718abc47a3164548b8f39af8c12bec32\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/48439c944b8748399f9182faaf28f8c1 : SUCCESS in 1h 15m 39s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/da3a1e872aa648fdaa078a631fbbc486 : SUCCESS in 1h 43m 06s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/a48cda6e99444e8a9d052d7dfd197467 : SUCCESS in 1h 17m 42s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/13cb3ccfd7e7452b876edc15c235552b : SUCCESS in 1h 20m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f6065ffcebc847fe808c9409defa2707 : SUCCESS in 11m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/306d844ecc1042dd8a4a598a0bb9b4cf : SUCCESS in 26m 55s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c9e20bfb6ef24d70a68e7dd60020f9ca : SUCCESS in 27m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a4f6733f7794241a1478ff9e0ab2ca8 : SUCCESS in 11m 04s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/7843a72e6c374ecfad183d171a8346e0 : SUCCESS in 1h 00m 21s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/1b86362773d3441b91d8f522168b6b9c : SUCCESS in 1h 50m 51s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/ec91cfca8f4d478cb8b8f34e37b9669e : SUCCESS in 1h 28m 30s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/ee949bdf318b429198d40b07ca15de0d : SUCCESS in 51m 37s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/b7ec46046c9e4c78ad760f1c0f0c2894 : SUCCESS in 54m 04s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/5ff3caa16f624114a12b7f854ee8e832 : SUCCESS in 53m 16s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/3fdf560c77b44b318a16fc72bedb2155 : SUCCESS in 1h 08m 05s","accounts_in_message":[],"_revision_number":3},{"id":"6b275bbdcc9ac40c498046fa9420dc6ed35a0246","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-22 15:49:02.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"79e374ff3fbcce303b55ac6d75ba73f97c6e1e88","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-22 15:51:15.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6cdc5e06c8a94e9d891a592522eaef1c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a81e98f10c54e5186697e8470e8d4cf : SUCCESS in 51s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"bcf33e202f99af1546d347cac5e5339d3294278a","revisions":{"8d8aeb0acae4ac6802fb29b70d007606ff69e586":{"kind":"REWORK","_number":1,"created":"2024-01-17 21:09:01.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/06/906006/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/06/906006/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/06/906006/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/06/906006/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/06/906006/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/06/906006/1"}}},"commit":{"parents":[{"commit":"70f9dd926f7433569df59c2f309f4239b8e8e7c9","subject":"update OVS_BRANCH to be compatible with ovn main branch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/70f9dd926f7433569df59c2f309f4239b8e8e7c9"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2024-01-17 18:28:18.000000000","tz":-300},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2024-01-17 18:28:18.000000000","tz":-300},"subject":"Fix keyword-arg-before-vararg warnings","message":"Fix keyword-arg-before-vararg warnings\n\nThe correct usage when using keyword and varargs is\n\ndef func(arg1, *args, arg2\u003dNone, **kwargs):\n\nOtherwise you can end up in having multiple values passed\nfor the parameters in case the method is called with\nkeyword arguments.\n\nStart enforcing this as well.\n\nTrivialFix\n\nChange-Id: Id9d6d841133241bbc87a589117468c4e699c310a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8d8aeb0acae4ac6802fb29b70d007606ff69e586"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8d8aeb0acae4ac6802fb29b70d007606ff69e586"}]},"branch":"refs/heads/master"},"815785a0ce64423e5938b690a12d2315862a7825":{"kind":"REWORK","_number":2,"created":"2024-01-18 20:16:11.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/06/906006/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/06/906006/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/06/906006/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/06/906006/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/06/906006/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/06/906006/2"}}},"commit":{"parents":[{"commit":"70f9dd926f7433569df59c2f309f4239b8e8e7c9","subject":"update OVS_BRANCH to be compatible with ovn main branch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/70f9dd926f7433569df59c2f309f4239b8e8e7c9"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2024-01-17 18:28:18.000000000","tz":-300},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2024-01-18 17:17:23.000000000","tz":-300},"subject":"Fix keyword-arg-before-vararg warnings","message":"Fix keyword-arg-before-vararg warnings\n\nThe correct usage when using keyword and varargs is\n\ndef func(arg1, *args, arg2\u003dNone, **kwargs):\n\nOtherwise you can end up in having multiple values passed\nfor the parameters in case the method is called with\nkeyword arguments.\n\nFixed two calls to _get_ports_query() that were not\npassing the \u0027filters\u0027 argument correctly.\n\nStart enforcing this as well.\n\nTrivialFix\n\nChange-Id: Id9d6d841133241bbc87a589117468c4e699c310a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/815785a0ce64423e5938b690a12d2315862a7825"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/815785a0ce64423e5938b690a12d2315862a7825"}]},"branch":"refs/heads/master"},"bcf33e202f99af1546d347cac5e5339d3294278a":{"kind":"REWORK","_number":3,"created":"2024-02-19 21:21:44.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/06/906006/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/06/906006/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/06/906006/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/06/906006/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/06/906006/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/06/906006/3"}}},"commit":{"parents":[{"commit":"1e75d61bcf610664b65ddb544f69c7c800733c33","subject":"Merge \"Fix unidiomatic-typecheck warnings\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1e75d61bcf610664b65ddb544f69c7c800733c33"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2024-01-17 18:28:18.000000000","tz":-300},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2024-02-19 21:19:57.000000000","tz":-300},"subject":"Fix keyword-arg-before-vararg warnings","message":"Fix keyword-arg-before-vararg warnings\n\nThe correct usage when using keyword and varargs is\n\ndef func(arg1, *args, arg2\u003dNone, **kwargs):\n\nOtherwise you can end up in having multiple values passed\nfor the parameters in case the method is called with\nkeyword arguments.\n\nFixed two calls to _get_ports_query() that were not\npassing the \u0027filters\u0027 argument correctly.\n\nStart enforcing this as well.\n\nTrivialFix\n\nChange-Id: Id9d6d841133241bbc87a589117468c4e699c310a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/bcf33e202f99af1546d347cac5e5339d3294278a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/bcf33e202f99af1546d347cac5e5339d3294278a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
