)]}'
{"id":"openstack%2Fneutron~907239","triplet_id":"openstack%2Fneutron~stable%2Fussuri~I17aa53cea6aba8ea83187c99102a6f25fd33cfff","project":"openstack/neutron","branch":"stable/ussuri","attention_set":{},"removed_from_attention_set":{"14567":{"account":{"_account_id":14567,"name":"Rodrigo Barbieri","email":"rodrigo.barbieri2010@gmail.com","username":"ganso"},"last_update":"2024-02-01 20:21:56.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I17aa53cea6aba8ea83187c99102a6f25fd33cfff","subject":"[OVN] Handle OVN agents when \"Chassis\" register is deleted","status":"ABANDONED","created":"2024-01-30 16:05:36.000000000","updated":"2024-02-01 20:21:56.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"8358608a47d0c953f13e921fda08f5ff86a03475","_number":907239,"virtual_id_number":907239,"owner":{"_account_id":14567,"name":"Rodrigo Barbieri","email":"rodrigo.barbieri2010@gmail.com","username":"ganso"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":14567,"name":"Rodrigo Barbieri","email":"rodrigo.barbieri2010@gmail.com","username":"ganso"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-01-31 14:16:54.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14567,"name":"Rodrigo Barbieri","email":"rodrigo.barbieri2010@gmail.com","username":"ganso"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":14567,"name":"Rodrigo Barbieri","email":"rodrigo.barbieri2010@gmail.com","username":"ganso"},"all":[{"value":-1,"date":"2024-01-31 13:16:11.000000000","permitted_voting_range":{"min":-1,"max":0},"_account_id":14567,"name":"Rodrigo Barbieri","email":"rodrigo.barbieri2010@gmail.com","username":"ganso"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":14567,"name":"Rodrigo Barbieri","email":"rodrigo.barbieri2010@gmail.com","username":"ganso"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14567,"name":"Rodrigo Barbieri","email":"rodrigo.barbieri2010@gmail.com","username":"ganso"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-01-30 16:05:36.000000000","updated_by":{"_account_id":14567,"name":"Rodrigo Barbieri","email":"rodrigo.barbieri2010@gmail.com","username":"ganso"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"CC"},{"updated":"2024-01-30 18:24:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-02-01 20:21:56.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"CC"}],"messages":[{"id":"98e772ed9cccbeb77ab719d71d9ded3d532e47fe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14567,"name":"Rodrigo Barbieri","email":"rodrigo.barbieri2010@gmail.com","username":"ganso"},"date":"2024-01-30 16:05:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cadcf69ae9b0120a76c486c54566903f9ce7754a","author":{"_account_id":14567,"name":"Rodrigo Barbieri","email":"rodrigo.barbieri2010@gmail.com","username":"ganso"},"date":"2024-01-30 16:06:07.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"0e6403a780b4d4820c9be93b7e27dc3e49d5af5b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-30 18:24:39.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0801eebf45e64135bddfa07625cdc594\n\n- neutron-tempest-plugin-api-ussuri https://zuul.opendev.org/t/openstack/build/8b886a5386fb435b8cc7370ab4f599d7 : SUCCESS in 40m 12s\n- neutron-tempest-plugin-scenario-ovn-ussuri https://zuul.opendev.org/t/openstack/build/b7a63fde228b400eae0474648fdfa420 : FAILURE in 2h 02m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/536e3ed6a6ff4fe4b04c9af16abfc385 : SUCCESS in 32m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9723c013fd1d4fedbbe9b636c9ee49ec : FAILURE in 8m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/85c5fd8ab9c443debc33d636c451336c : SUCCESS in 27m 54s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1c283301cdaa4d0590742691bfa61c68 : SUCCESS in 32m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7c4098af70c94edcbda9f26453d44806 : SUCCESS in 23m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62b0ad725e324ae68fede126c8a32566 : SUCCESS in 9m 17s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/70f91629108a4ebda28a31f21e207222 : SUCCESS in 48m 19s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/fed6617a8c4a47238f98818f938976c9 : SUCCESS in 1h 16m 30s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/3c4c50e7fd9841f8ab034cc00cb8810f : SUCCESS in 1h 21m 37s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/5910d819df42459ca96a29e0b908a710 : SUCCESS in 1h 26m 58s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/0bbe23783a914f79a54ea4c5359af614 : FAILURE in 38m 16s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/e73d59652dc34ff88ec07689ce53d9d3 : SUCCESS in 1h 14m 10s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/0bca635677694a80a4b9ff0c82fa813e : SUCCESS in 1h 25m 19s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/01adafd4b8c44d05a7096c5fa42d44b6 : SUCCESS in 1h 34m 54s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/b92d4e5742864c54b52cc40a76e8532c : FAILURE in 53m 11s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/bf9515c4016843ac84e039fa00d6edcd : SUCCESS in 1h 11m 02s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/4917bea0f0fc4d30a2028d5efdccbb4c : SUCCESS in 1h 34m 18s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/99e5c006ffd1444793e599f0da106c46 : SUCCESS in 54m 29s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/3330aff18cb04f2c9bb27438625083ee : SUCCESS in 1h 43m 34s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ce869b43289f45faa082a1302c2ceab2 : SUCCESS in 50m 50s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/b34cc1533dcd4802a14ee03d5e8cb7f6 : SUCCESS in 1h 14m 08s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/0d469f8e83704276959d3534c2449c8e : SUCCESS in 33m 25s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/49aa3bc1d23247969f30963c2ad9d45f : POST_FAILURE in 20m 22s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/b603b94c8fba46b8babd25fbc6b9998d : SUCCESS in 53m 42s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/7397e324a0d144bd8e366fec456fb546 : FAILURE in 1h 00m 55s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"496fe28df768eb3cb617be872c2bc49da75ce6ab","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14567,"name":"Rodrigo Barbieri","email":"rodrigo.barbieri2010@gmail.com","username":"ganso"},"date":"2024-01-31 12:27:38.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":2},{"id":"b01671e5eafb8f0d2481086edc3201f938822170","author":{"_account_id":14567,"name":"Rodrigo Barbieri","email":"rodrigo.barbieri2010@gmail.com","username":"ganso"},"date":"2024-01-31 13:16:11.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"ee9433c8f2d7ce7e26af96aae5ef07b0c28fc80b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-31 14:16:54.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ae293f3e2d7b414380c350d3f1df5b73\n\n- neutron-tempest-plugin-api-ussuri https://zuul.opendev.org/t/openstack/build/4f2ccc2a97cf417db7c933220d67d65d : SUCCESS in 40m 42s\n- neutron-tempest-plugin-scenario-ovn-ussuri https://zuul.opendev.org/t/openstack/build/67e1c479cc2a4c58abbc7d958688ca82 : FAILURE in 1h 41m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e1ef778cb74d45ccb040d90e3e6d66ac : SUCCESS in 31m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9c445c41128146e8ba24ccdcc5d4ebc8 : SUCCESS in 14m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/617a0a2f00f44c9d95860f393397131a : SUCCESS in 27m 37s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f80f355b591e4f6bbad932f41dbd3c32 : SUCCESS in 19m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/660c031aff1146b4b37a63a9c6dc6cd0 : SUCCESS in 21m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/863f8a073fe1449fa3f64bbbf6523231 : SUCCESS in 9m 25s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/9c8de4d726174d009efba86736c3ebf0 : SUCCESS in 56m 34s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/dfd121bf279c4df599f70d7a0e2b1051 : FAILURE in 50m 25s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/0c2a9d1ee8d74c668300e69053ddc2f4 : SUCCESS in 1h 28m 41s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/70f4f0bc769a481f91db7ced73267bb7 : SUCCESS in 1h 25m 53s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/ec180f19e05a4eb7a68ba5bd6adb6ebf : SUCCESS in 1h 38m 10s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/c6a011ef4bce4261ae401fa1b15f48c6 : SUCCESS in 1h 17m 00s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/32abfaef00b0485fab0e44b0cdc55895 : SUCCESS in 1h 06m 47s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/5558d1a5d7aa4a0eb7429c43bbc4e4db : SUCCESS in 57m 31s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/bd8ff3f20d3f42eb8919a11e1ebb43e5 : FAILURE in 33m 25s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/4cd1f67e883e4fe4bff4d6fa1c49ccfe : SUCCESS in 1h 07m 41s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/51e55978c0da4d2181d9ce151427be0a : SUCCESS in 1h 03m 47s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/ebdaa5396ccb42e9a1acd2d4ccef4bdc : SUCCESS in 44m 12s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/683b38d84ea3429dba638818dcf1ea6e : SUCCESS in 1h 16m 08s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f2db7082686b44c2abd55dfa4d30d816 : SUCCESS in 53m 58s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/b943965944d44f269f00079704de14ae : SUCCESS in 1h 05m 32s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/8d03c955407e485f8cc795ac248bdcd3 : SUCCESS in 38m 35s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/7fac314008064eb687ee25c2dbae837d : POST_FAILURE in 17m 37s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/19c726b9a39d43a091805f543c9794bf : SUCCESS in 1h 03m 03s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/ead68620fca5457e84a13569e1c54c25 : FAILURE in 1h 17m 27s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8358608a47d0c953f13e921fda08f5ff86a03475","tag":"autogenerated:gerrit:abandon","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-02-01 20:21:56.000000000","message":"Abandoned\n\nstable/ussuri branch of openstack/neutron is about to be deleted. To be able to do that, all open patches need to be abandoned.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"f353d6f0ba9feca959922e898ec09dcffc9dc93f","revisions":{"3a2765c308341824f73718a6724e71ad3cc79e23":{"kind":"REWORK","_number":1,"created":"2024-01-30 16:05:36.000000000","uploader":{"_account_id":14567,"name":"Rodrigo Barbieri","email":"rodrigo.barbieri2010@gmail.com","username":"ganso"},"ref":"refs/changes/39/907239/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/39/907239/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/907239/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/907239/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/907239/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/39/907239/1"}}},"commit":{"parents":[{"commit":"18b820ea41e05c36a954a92cec1cd20e010b1e9f","subject":"[stable-only] Drop rally and periodic mariadb jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/18b820ea41e05c36a954a92cec1cd20e010b1e9f"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-11-17 17:33:29.000000000","tz":0},"committer":{"name":"Rodrigo Barbieri","email":"rodrigo.barbieri2010@gmail.com","date":"2024-01-30 16:05:34.000000000","tz":-180},"subject":"[OVN] Handle OVN agents when \"Chassis\" register is deleted","message":"[OVN] Handle OVN agents when \"Chassis\" register is deleted\n\nIf an \"ovn-controller\" ends not gracefully, the node \"Chassis\" and\n\"Chassis_Private\" registers will remain in the OVN SB database.\nBecause there is no a mandatory procedure to delete the \"Chassis\"\nand \"Chassis_Private\" registers, the administrator can manually\ndelete, from the OVN SB database, any register in any order.\n\nIf the \"Chassis\" register is deleted and the Neutron server restarted,\nthe updated \"Chassis_Private\" register will be read from the database.\nThat won\u0027t contain the \"Chassis\" information as this register has been\ndeleted. In this case, the ``NeutronAgent`` returns ``DeletedChassis``,\nan empty chassis register with no information.\n\nNOTE: the sequence of actions (\"Chassis\" register deletion, Neutron\nserver restart) must be follow to reproduce this issue. If the\n\"Chassis\" register is deleted, the Neutron server OVN agent local cache\nwon\u0027t update the stored information and will keep the previous value.\nIt is when the Neutron server is restarted when the OVN agent local\ncache is retrieved again; at this time the \"Chassis_Private\" register\nwon\u0027t have any related \"Chassis\" register.\n\nCloses-Bug: #1951149\n\nSmall conflicts due to lack of commitae6d06be8007e20ee7145aa7fce4943aad76451c\nConflicts:\n    neutron/plugins/ml2/drivers/ovn/agent/neutron_agent.py\n    neutron/tests/functional/plugins/ml2/drivers/ovn/mech_driver/test_mech_driver.py\n\nChange-Id: I17aa53cea6aba8ea83187c99102a6f25fd33cfff\n(cherry picked from commit f1a5511e9094d6aae7a61cd858c59c706033ca95)\n(cherry picked from commit 29998538cec9a696b01d84bd9573f2ebbbb15f90)\n(cherry picked from commit 494cd8ca5dc924dbaa3b3a185110371dfe2e59c1)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3a2765c308341824f73718a6724e71ad3cc79e23"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3a2765c308341824f73718a6724e71ad3cc79e23"}]},"branch":"refs/heads/stable/ussuri"},"f353d6f0ba9feca959922e898ec09dcffc9dc93f":{"kind":"REWORK","_number":2,"created":"2024-01-31 12:27:38.000000000","uploader":{"_account_id":14567,"name":"Rodrigo Barbieri","email":"rodrigo.barbieri2010@gmail.com","username":"ganso"},"ref":"refs/changes/39/907239/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/39/907239/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/907239/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/907239/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/907239/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/39/907239/2"}}},"commit":{"parents":[{"commit":"18b820ea41e05c36a954a92cec1cd20e010b1e9f","subject":"[stable-only] Drop rally and periodic mariadb jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/18b820ea41e05c36a954a92cec1cd20e010b1e9f"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-11-17 17:33:29.000000000","tz":0},"committer":{"name":"Rodrigo Barbieri","email":"rodrigo.barbieri2010@gmail.com","date":"2024-01-31 12:27:27.000000000","tz":-180},"subject":"[OVN] Handle OVN agents when \"Chassis\" register is deleted","message":"[OVN] Handle OVN agents when \"Chassis\" register is deleted\n\nIf an \"ovn-controller\" ends not gracefully, the node \"Chassis\" and\n\"Chassis_Private\" registers will remain in the OVN SB database.\nBecause there is no a mandatory procedure to delete the \"Chassis\"\nand \"Chassis_Private\" registers, the administrator can manually\ndelete, from the OVN SB database, any register in any order.\n\nIf the \"Chassis\" register is deleted and the Neutron server restarted,\nthe updated \"Chassis_Private\" register will be read from the database.\nThat won\u0027t contain the \"Chassis\" information as this register has been\ndeleted. In this case, the ``NeutronAgent`` returns ``DeletedChassis``,\nan empty chassis register with no information.\n\nNOTE: the sequence of actions (\"Chassis\" register deletion, Neutron\nserver restart) must be follow to reproduce this issue. If the\n\"Chassis\" register is deleted, the Neutron server OVN agent local cache\nwon\u0027t update the stored information and will keep the previous value.\nIt is when the Neutron server is restarted when the OVN agent local\ncache is retrieved again; at this time the \"Chassis_Private\" register\nwon\u0027t have any related \"Chassis\" register.\n\nCloses-Bug: #1951149\n\nSmall conflicts due to lack of commitae6d06be8007e20ee7145aa7fce4943aad76451c\nConflicts:\n    neutron/plugins/ml2/drivers/ovn/agent/neutron_agent.py\n    neutron/tests/functional/plugins/ml2/drivers/ovn/mech_driver/test_mech_driver.py\n\nChange-Id: I17aa53cea6aba8ea83187c99102a6f25fd33cfff\n(cherry picked from commit f1a5511e9094d6aae7a61cd858c59c706033ca95)\n(cherry picked from commit 29998538cec9a696b01d84bd9573f2ebbbb15f90)\n(cherry picked from commit 494cd8ca5dc924dbaa3b3a185110371dfe2e59c1)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f353d6f0ba9feca959922e898ec09dcffc9dc93f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f353d6f0ba9feca959922e898ec09dcffc9dc93f"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Workflow\u003dMIN"],"failing_atoms":["label:Workflow\u003dMAX"],"atom_explanations":{}}}]}
