)]}'
{"id":"openstack%2Fneutron~907489","triplet_id":"openstack%2Fneutron~master~I4226024cb4cfd85bdf4f717d42d48150cab22442","project":"openstack/neutron","branch":"master","topic":"bug/2027742","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2024-08-27 06:46:56.000000000","reason":"\u003cGERRIT_ACCOUNT_16688\u003e replied on the change","reason_account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},"35432":{"account":{"_account_id":35432,"name":"Roberto Acosta","display_name":"rbartzen","email":"rbartzen@gmail.com","username":"rbartzen"},"last_update":"2024-08-28 23:23:34.000000000","reason":"Change was submitted"},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2024-08-28 18:56:58.000000000","reason":"\u003cGERRIT_ACCOUNT_1131\u003e replied on the change","reason_account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},"5756":{"account":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"last_update":"2024-08-28 23:23:34.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I4226024cb4cfd85bdf4f717d42d48150cab22442","subject":"[ML2/OVN] Add external_ids.neutron:is_static_route key for Static Routes","status":"MERGED","created":"2024-02-01 19:26:27.000000000","updated":"2024-08-28 23:24:41.000000000","submitted":"2024-08-28 23:23:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":48,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"907489-bug/2027742","meta_rev_id":"5c073f64f35279053e7bce4d449b91febdab8f7a","_number":907489,"virtual_id_number":907489,"owner":{"_account_id":35432,"name":"Roberto Acosta","display_name":"rbartzen","email":"rbartzen@gmail.com","username":"rbartzen"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2024-08-21 22:43:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-08-28 23:23:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2024-08-27 06:46:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2024-08-28 18:56:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":1,"date":"2024-08-28 18:56:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"}],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-01 19:52:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-02-01 19:58:10.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-02-01 21:01:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-02-06 12:40:55.000000000","updated_by":{"_account_id":35432,"name":"Roberto Acosta","display_name":"rbartzen","email":"rbartzen@gmail.com","username":"rbartzen"},"reviewer":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"state":"REVIEWER"},{"updated":"2024-02-15 03:11:59.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2024-02-23 16:31:57.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"CC"},{"updated":"2024-06-10 13:46:29.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2024-08-28 18:56:58.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"}],"messages":[{"id":"e4c2c5f34e154dab3f81aa876031bfc3bbe3e5b2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35432,"name":"Roberto Acosta","display_name":"rbartzen","email":"rbartzen@gmail.com","username":"rbartzen"},"date":"2024-02-01 19:26:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cf8bdd55d119077560c4cd9d2e5d105338610898","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-01 19:52:55.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/893b606fa1ff4ce1a1325af2c8450672\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/67e9e8367c6c47ed871a8615e56bcb0a : SUCCESS in 17m 40s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/78fd63ce17fa4560a9eea3c0799bd19b : SUCCESS in 25m 33s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/d2ce8f836e98495ba30a1dcba56881dc : SUCCESS in 16m 16s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b2a80b516ba2146d372d5fd7238131cbd6dae54f","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2024-02-01 19:58:10.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/February-2024/Arista_Tempest_Testrefs/changes/89/907489/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"c95de8686a11c233f2938a030e14c22c438db19c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-01 21:01:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c7a844ec5b614dfaad1150e3d233346e\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/bdd6928a15d5467bbcef13d1469a579c : SUCCESS in 1h 26m 16s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/b5e693d7da0a4c1e97b46b51ba48e1a2 : SUCCESS in 38m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f7672ebb5fcf442fab5327ccbe32b98f : SUCCESS in 39m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/692657df08c445278469a84f8ea02887 : SUCCESS in 10m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/79ec2f59370c49e4b98e6df56d4affba : SUCCESS in 27m 32s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/82711bfa1cf94ca89c90501ca586a693 : SUCCESS in 20m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/611ba5348e374792951a5e3479c9f9cb : SUCCESS in 12m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/42f25fa2d03346b2a1a2fca31b339f7a : SUCCESS in 11m 56s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/20559cbaa818414584aa297cdac66f21 : SUCCESS in 28m 39s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/c645fafcb5584dd2953a112c08bc25fe : SUCCESS in 1h 10m 52s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/084a095681d04b24be68df9749a903f8 : SUCCESS in 58m 29s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/8d699ae5ba88445b995fc211c8746dc8 : SUCCESS in 1h 20m 12s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/4c5000d60ffb4e08a03679d892747366 : SUCCESS in 51m 09s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/2866f42782634542bdd911997072cc07 : SUCCESS in 1h 29m 01s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/cb509a73e0234f6a915c09991d30da57 : SUCCESS in 47m 39s","accounts_in_message":[],"_revision_number":1},{"id":"d0213cc2cf16d3901f03989d06a38576b3c4f968","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-02-15 03:11:59.000000000","message":"Patch Set 1:\n\n(7 comments)","accounts_in_message":[],"_revision_number":1},{"id":"cb7c9013cd7316814839d028444e7a607b322a6f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35432,"name":"Roberto Acosta","display_name":"rbartzen","email":"rbartzen@gmail.com","username":"rbartzen"},"date":"2024-02-20 15:35:50.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"e2af49a9d3f1a03da4456831a81cc3a7b28fd1df","author":{"_account_id":35432,"name":"Roberto Acosta","display_name":"rbartzen","email":"rbartzen@gmail.com","username":"rbartzen"},"date":"2024-02-20 15:36:46.000000000","message":"Patch Set 2:\n\n(8 comments)","accounts_in_message":[],"_revision_number":2},{"id":"0bf9f34e4b81237341670d1c7cba10b4ccb61f84","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2024-02-20 16:09:48.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/February-2024/Arista_Tempest_Testrefs/changes/89/907489/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"591becd663808f2c27f24a13069bac6b52c72d35","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-20 16:13:05.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/be28d02663eb45f0a143c46688d015b9\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/686010965f9948889f77242202526ced : SUCCESS in 28m 24s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/22a68d5e23ad4f12ba9ca53c29e294a5 : SUCCESS in 24m 13s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/1559796360b04982add37a4091668495 : SUCCESS in 35m 25s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"991a1b3386b0fcb8a6044519dd97b3529e254562","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35432,"name":"Roberto Acosta","display_name":"rbartzen","email":"rbartzen@gmail.com","username":"rbartzen"},"date":"2024-02-20 17:00:38.000000000","message":"Uploaded patch set 3: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"474569bbe2850d980a43a89c26bee1ba9f990b71","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2024-02-20 17:31:51.000000000","message":"Patch Set 3: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/February-2024/Arista_Tempest_Testrefs/changes/89/907489/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"57334fb606daee32101f44e77dc77c79fb60caeb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-20 17:33:17.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2291c9ef20e348328caec5ed2afc1750\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/972bb3e3de7c4e72a23e5eb15e079d9e : SUCCESS in 31m 05s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/91c2414ae74b44d39d22b923e9789adc : SUCCESS in 22m 28s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/cc9d93dcedc94ecc867516bf40b8d3b7 : SUCCESS in 21m 03s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"23ccd7ac9ebccbb5306625a79cae2a7497a31a0f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-20 19:57:27.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2479c7bb389c45ab89a71e72c34cdbc2\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/4ac6dc6935fb4949a4cca97cf6c5af13 : SUCCESS in 1h 45m 59s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/218fdb90dd704439be2690d2df7eb265 : SUCCESS in 35m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/272b013195954ad98b7b254604811a19 : SUCCESS in 37m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b182e61ab68d41eaae9d9c0002a69e75 : SUCCESS in 10m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/06c027d3c0cd41d5869a71b7f91e24db : SUCCESS in 25m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/57efd189f1ca43afa62e8c6765d67e63 : SUCCESS in 24m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/92f56389537d4eeb9da292d5a4c190ce : SUCCESS in 8m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/49a7c6b04ba14d15a911a242dfb59cd8 : SUCCESS in 10m 16s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/9dfc10922be240388ff8c6baa6a17683 : SUCCESS in 26m 34s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/36a8621b5415458fbcf5f79dd1bc74a1 : SUCCESS in 1h 14m 24s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/1b40fa51fc85486bbdc343b5fe4c8ccd : FAILURE in 41m 26s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/177778c0aba7428e96b5026010260dc8 : SUCCESS in 2h 54m 24s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/d5023bbbd5bb43f488754a03e967c3d8 : SUCCESS in 1h 05m 57s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/cce4f7064f8249ac9532432398b2d057 : SUCCESS in 1h 28m 28s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/9ed5a9b1ceb14770a02fca944dc601b9 : SUCCESS in 1h 20m 17s","accounts_in_message":[],"_revision_number":3},{"id":"c115ba8312ca52a9a7d4f2a49345ac12d61e1c67","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35432,"name":"Roberto Acosta","display_name":"rbartzen","email":"rbartzen@gmail.com","username":"rbartzen"},"date":"2024-02-20 20:49:16.000000000","message":"Uploaded patch set 4: New patch set was added with same tree, parent tree, and commit message as Patch Set 3.\n\nOutdated Votes:\n* Verified+1, Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"09dac11d8d600e9c734648baf2de22c951295f64","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35432,"name":"Roberto Acosta","display_name":"rbartzen","email":"rbartzen@gmail.com","username":"rbartzen"},"date":"2024-02-20 20:54:33.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"067c66eef27e7bd475f118fc159bb323d4d3a76c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-20 21:15:05.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e24e7366a7134efa9c1cadfb21ceba25\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1d45130f93ef465994bd3537415ef063 : SUCCESS in 18m 43s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/26a80a0578294892b73fbc13d4ed5f43 : SUCCESS in 18m 22s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/d0b8d162de3045fb9413b5e1067a2870 : SUCCESS in 17m 31s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"9b9cb190a3cca994f0df6f0e4a218286656bad33","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2024-02-20 21:22:03.000000000","message":"Patch Set 4: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/February-2024/Arista_Tempest_Testrefs/changes/89/907489/4 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":4},{"id":"7a0b14ce8584567ee8cd3b318a5a3275e936007a","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2024-02-20 22:25:28.000000000","message":"Patch Set 5: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/February-2024/Arista_Tempest_Testrefs/changes/89/907489/5 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":5},{"id":"cb37e4661148f80374de975e92396283b3af6293","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-20 23:26:28.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fd13e8564f5f4e58af5b413887b8ecf4\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/3d2ad2c3a4d144a5a1cd4c1b085557cb : SUCCESS in 1h 30m 59s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/f9f89459a2b146bb992468acb416a34d : SUCCESS in 47m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a18a9336239e48df8678453be899dad4 : SUCCESS in 39m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a753927c76b74990a24aedd7a0d3ddf6 : SUCCESS in 10m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3a716c1dc3eb4251866c2fd1a5315542 : SUCCESS in 25m 44s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0690f94be14c44b28cf44040180727ea : SUCCESS in 31m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b6d4b69e5fe46c4b410352582c28c8d : SUCCESS in 11m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ad3b7bba1f32496c8c26505e239d16de : SUCCESS in 16m 24s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/674e2cc65fa74a84baa520bd78ba415c : SUCCESS in 25m 54s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/1d7d75083f64470f9c05254cbd0f3803 : SUCCESS in 58m 19s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/2fd7842d0c7b4c47924cca450088c858 : SUCCESS in 51m 34s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/4566c3bbb91140409a19a44057a1908e : SUCCESS in 2h 29m 32s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/8ca543c674b24eb1ab151a3295aca791 : SUCCESS in 59m 27s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/13733f9adcd44e56b68e7e9f1ef94308 : SUCCESS in 1h 15m 05s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/f1168e8dabb74fb38c21e83597fe71c6 : SUCCESS in 1h 01m 34s","accounts_in_message":[],"_revision_number":5},{"id":"4648876482ceaf54490df01f622692d21a098b1e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35432,"name":"Roberto Acosta","display_name":"rbartzen","email":"rbartzen@gmail.com","username":"rbartzen"},"date":"2024-02-21 00:56:53.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"15e4f181aeafef893d6f1c7955c64158acb7b5a9","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2024-02-21 01:27:44.000000000","message":"Patch Set 6: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/February-2024/Arista_Tempest_Testrefs/changes/89/907489/6 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":6},{"id":"92e18503b46a2a39a7d18af1d83070e7b4527ce7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-21 01:41:39.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3ae1d2425a244740bdcb095444e5c1c3\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b0ed05c0d3f146afb675399607925ad6 : SUCCESS in 18m 26s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/9694cf7aad3e4545af522d1eca7b58e2 : SUCCESS in 42m 58s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/4b03ac9bb19b4f1ca6984e4838717e8c : SUCCESS in 27m 38s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"a8aa2dc8073ca3437c6f9084710ce9a7076a6406","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-21 02:56:31.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2645d52a39bf42a2bc2ad346bd346ca4\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/97af5c99a7dd4d04985a77426e12e84b : SUCCESS in 1h 41m 39s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/5b4bf6ae72b24415a13ed9afae6ab439 : SUCCESS in 33m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1284f5c348ab4a9ba216629555b1cb6c : SUCCESS in 31m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf1cf2f3e137436a9042eb2abad7499c : SUCCESS in 9m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ba0170a455a24de987c96d6931739c4c : SUCCESS in 19m 54s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/891e8e5e9c36421c955443d05624b6ae : SUCCESS in 22m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b23110f1fa649139d3ba0dc7663d68c : SUCCESS in 10m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/54fcd50ffce7413d84207451d0aa771a : SUCCESS in 12m 29s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/db15a4b135804fd7a8c64b909cf85337 : SUCCESS in 21m 00s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/71d189f60aba48808141e91b442ec69c : SUCCESS in 47m 33s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/fc97c601583949f3b937e382cb7a427f : SUCCESS in 56m 08s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/655ed909101347d7a93f2189f9e9878a : SUCCESS in 1h 53m 25s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/6c5a378b20d748e3af348722a1d6971c : SUCCESS in 48m 56s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/f8173bd86bed44fdb43c247322377a0b : SUCCESS in 1h 14m 54s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/fd60c4d2c8b3493da01f4338e5e81f0a : SUCCESS in 1h 23m 40s","accounts_in_message":[],"_revision_number":6},{"id":"cc40e038a312b0ee6200d1a4b613166dc4d86b74","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-02-22 17:09:12.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"568bc0efe1bc803faa6864d7c9b4bfe08d856d92","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-02-22 17:25:59.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"e3ab1584e323b023143852a728c9b4f40c3a71e9","author":{"_account_id":35432,"name":"Roberto Acosta","display_name":"rbartzen","email":"rbartzen@gmail.com","username":"rbartzen"},"date":"2024-02-22 18:08:07.000000000","message":"Patch Set 6:\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"bdcfcc41c4d407371f07b78c13e881404234118a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35432,"name":"Roberto Acosta","display_name":"rbartzen","email":"rbartzen@gmail.com","username":"rbartzen"},"date":"2024-02-22 18:31:20.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"a687604ef7074c889fa2cfbc3397984c55a2b265","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2024-02-22 19:14:17.000000000","message":"Patch Set 7: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/February-2024/Arista_Tempest_Testrefs/changes/89/907489/7 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":7},{"id":"8f1ddb942e27f38c0c143278a52f5a456300f2a6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-22 19:16:01.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7b21dc96d9864b1d9da1373014349e34\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a75f63b3a8be4f8e827713c36b297569 : SUCCESS in 40m 27s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c2473ab1a69d48138ff02ac5ef3dad47 : SUCCESS in 23m 31s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/c018998b52da4f76a9394bfeeca93d43 : SUCCESS in 35m 50s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"5504e852283b9ec9809c69bddc25f7df91204ace","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-22 21:04:54.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7d7497d60a1e4ca6a0ef81a513da41a1\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/23124aa46bfb461484ffcc4d7758721b : SUCCESS in 2h 07m 54s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/6f363313760a47808b9e2133c431d3fb : SUCCESS in 39m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/995e6c195e6646088d90d22ec212a1e6 : SUCCESS in 41m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/576275a6e9bd47e197715a8a7bdcba4f : SUCCESS in 10m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e01a5acb58fb49ae8dda3b81d8cb9d31 : SUCCESS in 24m 53s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4e1052f517c449f19bd97c250200ff10 : SUCCESS in 19m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a8b301fb5bb4581899b13717de734b2 : SUCCESS in 10m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2653c08582074c278e2e36aa8b1efe73 : SUCCESS in 12m 56s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/1f747bd9baab45afa7e0e225a324eecd : FAILURE in 29m 09s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/33504c2103594b1391f10c22d5a985a5 : SUCCESS in 1h 17m 35s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/50566ec86b7447ffa84cf670a1f2f29c : SUCCESS in 59m 48s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/2cf9922827e74968b82d19e8157b1f31 : SUCCESS in 2h 32m 10s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/c8adc814572240659a8a129fdd659032 : SUCCESS in 43m 06s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/0e79ca34ffc441378359768c1983d6cb : SUCCESS in 1h 22m 32s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/342d20f7598f465db5391c6a416245ea : SUCCESS in 55m 21s","accounts_in_message":[],"_revision_number":7},{"id":"d7fab026e3d075d2d87edb197a843c8cc048ca5f","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-02-22 23:14:24.000000000","message":"Patch Set 7:\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"69d8d50a276678bfbc9459d9f63a332561d9347c","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-02-23 16:31:57.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"58956dd1064a4a7669d905d6fa8830aec11d0441","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-23 17:10:43.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fc52fa6b05c74d8d839de4279f962c64\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/77ad7dfbeac44c3e8fc6d00baa11ca01 : SUCCESS in 29m 43s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/8ba5102373fb4993bd0c83e421cb797c : SUCCESS in 23m 14s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/e92ed9f1fe0d414ab21e1ca32aec547d : SUCCESS in 36m 16s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"30e43fbd0115870cd02c292a46ba4a5e0e43ea95","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-23 18:19:43.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/49a4a5e1ff1a46bfadf400e10681d327\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/f31305f8ee2c44c6a4b44b613595f043 : SUCCESS in 1h 19m 42s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/ed1cec07dbe640c886cf394b721eaf76 : SUCCESS in 31m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/aa4beba8192949a0a9a1282f033d44ad : SUCCESS in 35m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/809e63cb6f54434fad282b718f5dbf6e : SUCCESS in 9m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/14fe9abe4ed14c59b095e5f097d664d8 : SUCCESS in 38m 23s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/eb414d38c17449b68c0232eba9ebf3e5 : SUCCESS in 22m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28fd1b1127634a01ac0cdd8c75ae8635 : SUCCESS in 10m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/997ece6a034b4093b28c3c917508652a : SUCCESS in 8m 50s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/f16c5527a82040d792169cea90478651 : SUCCESS in 37m 29s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/52f748b0c9f4466888672ad40fd1b683 : SUCCESS in 1h 13m 45s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/ecdd45b2e6a9480498a227d2476294fc : SUCCESS in 54m 26s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/bdf94ae0296e4a74a407374dca1913e6 : SUCCESS in 1h 41m 50s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/ee9843b0a0104be3b45ad010849847d9 : SUCCESS in 59m 39s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/055e756209794ff99e036f25efcb42b2 : SUCCESS in 1h 12m 47s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/522807ca16464c45b34754d220d5561c : SUCCESS in 1h 23m 15s","accounts_in_message":[],"_revision_number":7},{"id":"bd4e64dd32f850e03171da1a360af891225a1722","author":{"_account_id":35432,"name":"Roberto Acosta","display_name":"rbartzen","email":"rbartzen@gmail.com","username":"rbartzen"},"date":"2024-03-04 18:05:36.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"fef7cf37a42a20560d447a0e26f5d961c6c53cd1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35432,"name":"Roberto Acosta","display_name":"rbartzen","email":"rbartzen@gmail.com","username":"rbartzen"},"date":"2024-03-04 18:05:50.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"9fecccbc00797b811afab6554da3374f916302a6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-04 18:33:06.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5c93b93e411a42ba873a4c5298ce15ea\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0f1db0596d554836affaf8d575cc0bc9 : SUCCESS in 18m 58s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/95efba2646544aa2b0f4e2b32915bb21 : SUCCESS in 17m 12s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/ce5f8c4e2ae04cf18f159b07990dba68 : SUCCESS in 25m 32s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"a2bfb8ed2fef54007159b5da5d969d15b4461741","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-04 20:37:06.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6273dc2a5d7243adb99cae0463f84055\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/962483e5dab043209be14e3bc8e6e6d9 : SUCCESS in 1h 26m 07s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/988a1d5137ad48e9b4d79be3a43c8769 : SUCCESS in 31m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7748985be11c4b1ba4985028bbcdcd73 : SUCCESS in 35m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/75dd0eeb327949fca6af4d370b5730f1 : SUCCESS in 10m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b9cc3d7dbe074039b2437e651b8860a2 : SUCCESS in 26m 05s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/47f84432d5fc4afc8307990ebb7ff3b6 : SUCCESS in 25m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dced781e2f0d4e358850eae7c5513b13 : SUCCESS in 12m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/de4c2c5710d34d928b50ba58434cca44 : SUCCESS in 13m 11s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/3c21126fe0c247eaa49e4a20835ac8c7 : SUCCESS in 25m 21s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/64c2035ceb994fa2b997143e7f2337f0 : SUCCESS in 1h 00m 40s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/0d4b93da5c3448c2a5e2d394477d454b : SUCCESS in 1h 00m 25s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/4d2f8615d71b4b338fd4a12c70569159 : SUCCESS in 2h 29m 53s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/0b8609f586c94d11ab714e6a03097e95 : SUCCESS in 1h 00m 44s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/cb9715eb817a400c995b9b63546550ef : SUCCESS in 1h 30m 19s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/f0eff5ebb9ed4350af897d8f7d1ec058 : SUCCESS in 1h 16m 46s","accounts_in_message":[],"_revision_number":8},{"id":"e16acbf00a477cce64257fe6ff4bd671b364541d","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-03-25 00:29:44.000000000","message":"Patch Set 8:\n\n(2 comments)","accounts_in_message":[],"_revision_number":8},{"id":"81fa713896fa31b76317ba520b167309fa88463a","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-03-25 00:32:36.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"064f0b87e0aaffabeee1a8a7cd8e38a4f45e48af","author":{"_account_id":35432,"name":"Roberto Acosta","display_name":"rbartzen","email":"rbartzen@gmail.com","username":"rbartzen"},"date":"2024-04-15 14:58:22.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"d5cb40189a050c8bd5125c064d6776599c0379a4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35432,"name":"Roberto Acosta","display_name":"rbartzen","email":"rbartzen@gmail.com","username":"rbartzen"},"date":"2024-04-15 14:58:50.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":9},{"id":"8e96463623bbe86128598c647e42f99f34075075","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2024-04-15 15:38:57.000000000","message":"Patch Set 9: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/April-2024/Arista_Tempest_Testrefs/changes/89/907489/9 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":9},{"id":"45dbedee451aaccf06f89660cf849487698fed25","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-15 15:48:59.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bc11a37730234707958cd85bbdb1537e\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f21791be4ada4477a99e485f9e6b2811 : SUCCESS in 38m 02s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/d94d063a24a143cb885044db14d09247 : SUCCESS in 23m 25s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"f30a09a85ccf3041cb4907924237719130c4add4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-15 16:36:39.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7c87ff48ae664d9bab02b4299d27d7f2\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/0765c556a3a94ba9b81d7ac03cd535fa : SUCCESS in 1h 20m 26s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/8d9304f836bb4ceb84783be8715b0194 : SUCCESS in 30m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/64731aed10c944c1be6310dab1f92831 : SUCCESS in 39m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6fa38d8b8e4f4f18ad7a8479ca1534fb : SUCCESS in 10m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bdf41aca96af49ab8c5895119ad9271f : SUCCESS in 31m 23s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0273a66869ae4355ba79b79ba812cc30 : SUCCESS in 22m 27s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/912b8efbc51f41408fea66c6b51e2dd2 : SUCCESS in 29m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fbd8c74dce4d4b79873646478ee789a3 : SUCCESS in 10m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c2efe0d71278495e988fb26d461bc0a1 : SUCCESS in 13m 45s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/af1be2cf663242cba477a1c6f39b5b85 : SUCCESS in 22m 31s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/3e20a866d8754fd182a7a0594d81c8a4 : SUCCESS in 55m 10s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/4a8510ee55c14e2091ca390a4e43ed69 : SUCCESS in 1h 35m 34s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/b3f569e015d84941ab2af04c94b97d26 : SUCCESS in 1h 00m 15s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/24e642379d4741e680e33eeac93509da : SUCCESS in 1h 08m 22s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/c095efa83fea453bb5b0751281547e06 : SUCCESS in 1h 09m 02s","accounts_in_message":[],"_revision_number":9},{"id":"264f1bd17baacbc93e18251f320b1b1e89c47ec9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35432,"name":"Roberto Acosta","display_name":"rbartzen","email":"rbartzen@gmail.com","username":"rbartzen"},"date":"2024-05-27 13:10:14.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":10},{"id":"7bde737462640abc3bb16b0004cb85a87075ed8a","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2024-05-27 13:43:14.000000000","message":"Patch Set 10: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/May-2024/Arista_Tempest_Testrefs/changes/89/907489/10 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":10},{"id":"620b7321af9853a1bdea20d619809ac8efdb23c7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-27 13:52:53.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/48202a628a4847a4b0de3068ca561792\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/fc9056278a3142a9a46dbfaddfdcb162 : FAILURE in 22m 04s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/b0ea61e0c145417f989c0ccc7e751996 : FAILURE in 40m 47s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"aeb0ab39d96d72b1b5d4246484eb633e7eb5cd7d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-27 15:05:12.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/27be8e087a484c63992424e05a435948\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/71b9e19457594b4397f206c25e837907 : SUCCESS in 1h 41m 32s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/3c681f9e9b124cdb8d768da69da096b8 : SUCCESS in 33m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/54e73844aaf54cc690eee6a757589cb5 : FAILURE in 41m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c58d292564f94124bec57dddbd9515d0 : SUCCESS in 10m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5d5f839e17f34ccf9cd50fe8caa21234 : FAILURE in 25m 55s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c3d9c465b21b46ffa4fc13bef65abe95 : FAILURE in 23m 32s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0a7c53a3b39e4ad4a9b0abb879fb9f0f : FAILURE in 31m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/88c6834465924405a01306e3afd23bb7 : SUCCESS in 12m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/317bb2b8287a47819ac0102b1098654d : SUCCESS in 18m 26s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/a88e99dfea35451b89f2da13f2556412 : FAILURE in 27m 37s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/2dd133fb6eff4c7bbe254437ded5b98c : SUCCESS in 52m 04s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/01786753c6f4474485ca0dddda0bae53 : SUCCESS in 1h 48m 41s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/7059df78c7fd42bd994919144cbfe944 : SUCCESS in 56m 09s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/e64a06c1f81b44898480747302138010 : SUCCESS in 1h 21m 05s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/6a6053629ec94cbca893ac0355717c24 : SUCCESS in 1h 25m 16s","accounts_in_message":[],"_revision_number":10},{"id":"77874df96d2aeb3ab405d3c358ea72f3620dca8f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35432,"name":"Roberto Acosta","display_name":"rbartzen","email":"rbartzen@gmail.com","username":"rbartzen"},"date":"2024-05-28 15:33:22.000000000","message":"Uploaded patch set 11.\n\nOutdated Votes:\n* Verified+1, Verified-1\n","accounts_in_message":[],"_revision_number":11},{"id":"85ee9d729e419a8cfe622d54a56f5316715c9a30","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35432,"name":"Roberto Acosta","display_name":"rbartzen","email":"rbartzen@gmail.com","username":"rbartzen"},"date":"2024-05-28 15:36:00.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"efae910379c877aed0a049a46a0da44bc5bf82c0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-28 16:06:55.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/48c19c9309124f95bb454cdfb56ef3d8\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/db866169c2cc4911b7cbede9c8ade43a : SUCCESS in 27m 13s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/c6bd931880fc4253bae30a6aa7479a19 : SUCCESS in 29m 53s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"6a6f15891940fbf73ccbfca128d28cd430b96322","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2024-05-28 16:35:42.000000000","message":"Patch Set 11: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/May-2024/Arista_Tempest_Testrefs/changes/89/907489/11 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":11},{"id":"dca861870b30e5b710b228aab505edb9e930cf08","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2024-05-28 17:09:21.000000000","message":"Patch Set 12: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/May-2024/Arista_Tempest_Testrefs/changes/89/907489/12 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":12},{"id":"19392a5f4045ef5b914fcb21b9d169b267eff81c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-28 18:00:50.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d0255d6b72434718b6788c0c003a0174\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/c9dc9c4d19d547dfb6a4bf9380043019 : SUCCESS in 2h 22m 47s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/9e737e4921df4418957b43e384133f35 : SUCCESS in 46m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2a6a6d902934475aa5eebe9bd14310d0 : SUCCESS in 38m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/707fe16585314d4884132a29bedc79cd : SUCCESS in 9m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/94882f5a08434c558afddcf918516fbe : SUCCESS in 27m 08s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3ca57dddacb841f080228d1d477c0a39 : SUCCESS in 24m 43s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1e4f3fe9895f4cf6921bbe554d4381ad : SUCCESS in 32m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9cb55b6c9215408180a49876759be815 : SUCCESS in 11m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b2f28f8bbf7346c1b1aee124adbd132c : SUCCESS in 18m 12s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/eb5dff6cefcc473da1f9b1749183d447 : SUCCESS in 27m 42s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/237c792f37c54f858a49e4aec6a14867 : SUCCESS in 56m 49s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/04d4bdc147684f6f9484adafcc714afa : SUCCESS in 1h 39m 41s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/a908ceb7cd894bb8b712bef2500c402c : SUCCESS in 57m 37s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/e9a0f3ee076345a4bf91bc7b0c5aa21c : SUCCESS in 1h 31m 44s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/9ca97e748f96448bbda32dbf72ab9607 : SUCCESS in 1h 29m 53s","accounts_in_message":[],"_revision_number":12},{"id":"14322cf1835e7fb4b879ea56413a5b6513ac1e84","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-06-10 13:46:29.000000000","message":"Patch Set 12: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":12},{"id":"1294e0dcf7a45d43ae16921b4b70a534ce9ed8fd","author":{"_account_id":35432,"name":"Roberto Acosta","display_name":"rbartzen","email":"rbartzen@gmail.com","username":"rbartzen"},"date":"2024-06-10 17:55:20.000000000","message":"Patch Set 12:\n\n(2 comments)","accounts_in_message":[],"_revision_number":12},{"id":"3d0475abf33751c7138c6c0af1553c7aa0510654","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-07-08 15:02:48.000000000","message":"Patch Set 12:\n\n(8 comments)","accounts_in_message":[],"_revision_number":12},{"id":"b9bd0d8e1082a1f948c53bb424c1717e6a0c9498","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-07-08 15:09:02.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"01e73a40f63a80841a5fe5934edfc894274e4734","author":{"_account_id":35432,"name":"Roberto Acosta","display_name":"rbartzen","email":"rbartzen@gmail.com","username":"rbartzen"},"date":"2024-08-21 22:00:59.000000000","message":"Patch Set 12:\n\n(8 comments)","accounts_in_message":[],"_revision_number":12},{"id":"3fa62d96a199d2ad12429781507a2149f9209bfc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35432,"name":"Roberto Acosta","display_name":"rbartzen","email":"rbartzen@gmail.com","username":"rbartzen"},"date":"2024-08-21 22:11:53.000000000","message":"Uploaded patch set 13.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":13},{"id":"3380d568e236fa20aec81ae0f434637a98388b67","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2024-08-21 22:43:12.000000000","message":"Patch Set 13: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/August-2024/Arista_Tempest_Testrefs/changes/89/907489/13 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":13},{"id":"206488431d7c057bc10fac03faf5295c4742d1ab","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-21 23:42:43.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7e7c43e5e3f54f0290986c3ff70a3521\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/20510ecd96894691943a932a7ebe85ab : SUCCESS in 35m 02s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/22cfc27a0b4743d5b97fa84127c1c997 : SUCCESS in 31m 46s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"319be05d54af571704f81278a3355f33ada827c1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-22 00:04:36.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/94f522ea755b46a4ab0308601df91c46\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/3058a410d4f643828d4224f16458eb64 : SUCCESS in 1h 32m 21s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/b092e4aac295472c9fcbb42075b25b20 : SUCCESS in 33m 16s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bde8cad6abf34faabceb8ffc9a964c4e : SUCCESS in 33m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5dc25c0e4820446f972c082865cacea6 : SUCCESS in 10m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c132b667ebb341b48636da631ac429f3 : SUCCESS in 23m 21s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e5deebdae0384af5b9a140b2065f8e92 : SUCCESS in 18m 34s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d2e3f0f16d7549a6a7bca43f173c6390 : SUCCESS in 21m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c1bc55983309452683078a7694cc2d54 : SUCCESS in 9m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bf27c07d74614755bb0b226b907aa27c : SUCCESS in 21m 10s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/f2dfe50397e147debaaddbb280955115 : SUCCESS in 24m 55s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/91deca52c4f546d38853dece908f4adc : SUCCESS in 1h 01m 17s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/1adbe9adeed940dca70399b5a20adbe6 : SUCCESS in 1h 51m 09s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/295417929d5744f5a37dd3662924ab6a : SUCCESS in 1h 02m 56s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/11af4d30531d4bdeb8bba4032d781b6c : SUCCESS in 52m 24s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/7b8114f8441c442d8b975f6f1ccba19c : SUCCESS in 1h 07m 19s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/772f5d967ce04752bdb3521ae4697a60 : SUCCESS in 1h 08m 18s","accounts_in_message":[],"_revision_number":13},{"id":"8c82bbdcdce43974512ae015f4f949f6cd27142d","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-08-27 06:46:51.000000000","message":"Topic set to bug/2027742","accounts_in_message":[],"_revision_number":13},{"id":"d1409a6e31b87a8fad91cf13127e41f8b41e3ffd","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-08-27 06:46:56.000000000","message":"Patch Set 13: Code-Review+2","accounts_in_message":[],"_revision_number":13},{"id":"dd9d653044fb8dbaf7182a505e502c40675525ad","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-08-28 18:56:58.000000000","message":"Patch Set 13: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":13},{"id":"df378017a774fcb81f861c1d0ee705befd52aab5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-28 18:57:59.000000000","message":"Patch Set 13: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":13},{"id":"22d20304258f9f536f2999d4b2d28d0eee1f9c17","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-28 23:23:33.000000000","message":"Patch Set 13: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3a6857e3b84c40da964e8671dc6fa886\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/cde85652bb194bcd90f1cc17df70be56 : SUCCESS in 2h 13m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e956bf4129024f7f968e78d2fbf2d39a : SUCCESS in 10m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fd63f3b3429b497b990209947bb05eef : SUCCESS in 21m 19s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a5107ebe1a974767ae0d30d3288205a0 : SUCCESS in 20m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9152a44d08dd4a91bcbf791236b735c4 : SUCCESS in 11m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/901a03cae9814367ad3d19ad89e03d78 : SUCCESS in 15m 52s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/178dab24d278412da7791ba0305b3eda : SUCCESS in 55m 23s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/c83ea4c1072641afb6b818a45e8d267c : SUCCESS in 1h 42m 46s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/a1d3b59c5fe94d82b8773661358e2373 : SUCCESS in 1h 06m 02s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/c85f66883c914781aa9ff7f83e45584c : SUCCESS in 1h 04m 20s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/00d6161d7dec48579d5392828eb717b3 : SUCCESS in 1h 12m 07s","accounts_in_message":[],"_revision_number":13},{"id":"ab91b115e973526b90e551d08987ccfd045ce5e0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-28 23:23:34.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":13},{"id":"5c073f64f35279053e7bce4d449b91febdab8f7a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-28 23:24:41.000000000","message":"Patch Set 13:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ff5c78c2dd4c45b28b1048ef983a1e4e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a85ef991940f4d1888af68b7523b41d6 : SUCCESS in 48s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bc5fe986e84a43d1becb75e40a3d7ccd : SUCCESS in 48s","accounts_in_message":[],"_revision_number":13}],"current_revision_number":13,"current_revision":"89835e43e8e60b5fdeacdd2a48abacaa9b9778bc","revisions":{"cf25d760ce797ca616314c926b73ba2f8a98688b":{"kind":"REWORK","_number":1,"created":"2024-02-01 19:26:27.000000000","uploader":{"_account_id":35432,"name":"Roberto Acosta","display_name":"rbartzen","email":"rbartzen@gmail.com","username":"rbartzen"},"ref":"refs/changes/89/907489/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/89/907489/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/89/907489/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/89/907489/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/89/907489/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/89/907489/1"}}},"commit":{"parents":[{"commit":"b85b19e3846fa74975ba5d703336b6e7cd8af433","subject":"Merge \"dhcp: rename get_process_uuid as private\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b85b19e3846fa74975ba5d703336b6e7cd8af433"}]}],"author":{"name":"Roberto Bartzen Acosta","email":"rbartzen@gmail.com","date":"2024-02-01 19:20:04.000000000","tz":-180},"committer":{"name":"Roberto Bartzen Acosta","email":"rbartzen@gmail.com","date":"2024-02-01 19:25:56.000000000","tz":-180},"subject":"[ML2/OVN] Add external_ids key for Static Routes","message":"[ML2/OVN] Add external_ids key for Static Routes\n\nThis patch updates the method for creating static routes in OVN backend and updates previously created static routes rules to include the Neutron\u0027s owner key. This static route update task runs only once during the maintenance task, and if all entries are already configured, no action is performed.\n\nRelated-bug: #2027742\nChange-Id: I4226024cb4cfd85bdf4f717d42d48150cab22442\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/cf25d760ce797ca616314c926b73ba2f8a98688b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/cf25d760ce797ca616314c926b73ba2f8a98688b"}]},"branch":"refs/heads/master"},"1592e465c31f834cddc29a924f4a433a9191f8b0":{"kind":"REWORK","_number":2,"created":"2024-02-20 15:35:50.000000000","uploader":{"_account_id":35432,"name":"Roberto Acosta","display_name":"rbartzen","email":"rbartzen@gmail.com","username":"rbartzen"},"ref":"refs/changes/89/907489/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/89/907489/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/89/907489/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/89/907489/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/89/907489/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/89/907489/2"}}},"commit":{"parents":[{"commit":"1e75d61bcf610664b65ddb544f69c7c800733c33","subject":"Merge \"Fix unidiomatic-typecheck warnings\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1e75d61bcf610664b65ddb544f69c7c800733c33"}]}],"author":{"name":"Roberto Bartzen Acosta","email":"rbartzen@gmail.com","date":"2024-02-01 19:20:04.000000000","tz":-180},"committer":{"name":"Roberto Bartzen Acosta","email":"rbartzen@gmail.com","date":"2024-02-20 15:34:28.000000000","tz":-180},"subject":"[ML2/OVN] Add external_ids key for Static Routes","message":"[ML2/OVN] Add external_ids key for Static Routes\n\nThis patch updates the method for creating static routes in OVN backend and\nupdates previously created static routes rules to include the Neutron\u0027s owner\nkey. This static route update task runs only once during the maintenance task,\nand if all entries are already configured, no action is performed.\n\nRelated-bug: #2027742\nChange-Id: I4226024cb4cfd85bdf4f717d42d48150cab22442\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1592e465c31f834cddc29a924f4a433a9191f8b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1592e465c31f834cddc29a924f4a433a9191f8b0"}]},"branch":"refs/heads/master"},"e294fd8b8ba2cf792ec274f65e4d466f75089bbd":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2024-02-20 17:00:38.000000000","uploader":{"_account_id":35432,"name":"Roberto Acosta","display_name":"rbartzen","email":"rbartzen@gmail.com","username":"rbartzen"},"ref":"refs/changes/89/907489/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/89/907489/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/89/907489/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/89/907489/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/89/907489/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/89/907489/3"}}},"commit":{"parents":[{"commit":"1e75d61bcf610664b65ddb544f69c7c800733c33","subject":"Merge \"Fix unidiomatic-typecheck warnings\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1e75d61bcf610664b65ddb544f69c7c800733c33"}]}],"author":{"name":"Roberto Bartzen Acosta","email":"rbartzen@gmail.com","date":"2024-02-01 19:20:04.000000000","tz":-180},"committer":{"name":"Roberto Bartzen Acosta","email":"rbartzen@gmail.com","date":"2024-02-20 16:59:30.000000000","tz":-180},"subject":"[ML2/OVN] Add external_ids key for Static Routes","message":"[ML2/OVN] Add external_ids key for Static Routes\n\nThis patch updates the method for creating static routes in OVN backend\nand updates previously created static routes rules to include the\nNeutron\u0027s owner key. This static route update task runs only once\nduring the maintenance task, and if all entries are already configured,\nno action is performed.\n\nRelated-bug: #2027742\nChange-Id: I4226024cb4cfd85bdf4f717d42d48150cab22442\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e294fd8b8ba2cf792ec274f65e4d466f75089bbd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e294fd8b8ba2cf792ec274f65e4d466f75089bbd"}]},"branch":"refs/heads/master"},"cd2e4f2d20b3f5130342062b9524e5462743f15a":{"kind":"NO_CHANGE","_number":4,"created":"2024-02-20 20:49:16.000000000","uploader":{"_account_id":35432,"name":"Roberto Acosta","display_name":"rbartzen","email":"rbartzen@gmail.com","username":"rbartzen"},"ref":"refs/changes/89/907489/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/89/907489/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/89/907489/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/89/907489/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/89/907489/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/89/907489/4"}}},"commit":{"parents":[{"commit":"1e75d61bcf610664b65ddb544f69c7c800733c33","subject":"Merge \"Fix unidiomatic-typecheck warnings\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1e75d61bcf610664b65ddb544f69c7c800733c33"}]}],"author":{"name":"Roberto Bartzen Acosta","email":"rbartzen@gmail.com","date":"2024-02-01 19:20:04.000000000","tz":-180},"committer":{"name":"Roberto Bartzen Acosta","email":"rbartzen@gmail.com","date":"2024-02-20 20:41:24.000000000","tz":-180},"subject":"[ML2/OVN] Add external_ids key for Static Routes","message":"[ML2/OVN] Add external_ids key for Static Routes\n\nThis patch updates the method for creating static routes in OVN backend\nand updates previously created static routes rules to include the\nNeutron\u0027s owner key. This static route update task runs only once\nduring the maintenance task, and if all entries are already configured,\nno action is performed.\n\nRelated-bug: #2027742\nChange-Id: I4226024cb4cfd85bdf4f717d42d48150cab22442\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/cd2e4f2d20b3f5130342062b9524e5462743f15a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/cd2e4f2d20b3f5130342062b9524e5462743f15a"}]},"branch":"refs/heads/master"},"6006ba83d35e562c51dcca223bd14673b1804e91":{"kind":"REWORK","_number":5,"created":"2024-02-20 20:54:33.000000000","uploader":{"_account_id":35432,"name":"Roberto Acosta","display_name":"rbartzen","email":"rbartzen@gmail.com","username":"rbartzen"},"ref":"refs/changes/89/907489/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/89/907489/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/89/907489/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/89/907489/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/89/907489/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/89/907489/5"}}},"commit":{"parents":[{"commit":"1e75d61bcf610664b65ddb544f69c7c800733c33","subject":"Merge \"Fix unidiomatic-typecheck warnings\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1e75d61bcf610664b65ddb544f69c7c800733c33"}]}],"author":{"name":"Roberto Bartzen Acosta","email":"rbartzen@gmail.com","date":"2024-02-01 19:20:04.000000000","tz":-180},"committer":{"name":"Roberto Bartzen Acosta","email":"rbartzen@gmail.com","date":"2024-02-20 20:54:15.000000000","tz":-180},"subject":"[ML2/OVN] Add external_ids key for Static Routes","message":"[ML2/OVN] Add external_ids key for Static Routes\n\nThis patch updates the method for creating static routes in OVN backend\nand updates previously created static routes rules to include the\nNeutron\u0027s owner key. This static route update task runs only once\nduring the maintenance task, and if all entries are already configured,\nno action is performed.\n\nRelated-bug: #2027742\nChange-Id: I4226024cb4cfd85bdf4f717d42d48150cab22442\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6006ba83d35e562c51dcca223bd14673b1804e91"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6006ba83d35e562c51dcca223bd14673b1804e91"}]},"branch":"refs/heads/master"},"9f2ca21461bd12bf4be5a297e25ec2b354bf0e5f":{"kind":"REWORK","_number":6,"created":"2024-02-21 00:56:53.000000000","uploader":{"_account_id":35432,"name":"Roberto Acosta","display_name":"rbartzen","email":"rbartzen@gmail.com","username":"rbartzen"},"ref":"refs/changes/89/907489/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/89/907489/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/89/907489/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/89/907489/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/89/907489/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/89/907489/6"}}},"commit":{"parents":[{"commit":"1e75d61bcf610664b65ddb544f69c7c800733c33","subject":"Merge \"Fix unidiomatic-typecheck warnings\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1e75d61bcf610664b65ddb544f69c7c800733c33"}]}],"author":{"name":"Roberto Bartzen Acosta","email":"rbartzen@gmail.com","date":"2024-02-01 19:20:04.000000000","tz":-180},"committer":{"name":"Roberto Bartzen Acosta","email":"rbartzen@gmail.com","date":"2024-02-21 00:56:40.000000000","tz":-180},"subject":"[ML2/OVN] Add external_ids key for Static Routes","message":"[ML2/OVN] Add external_ids key for Static Routes\n\nThis patch updates the method for creating static routes in OVN backend\nand updates previously created static routes rules to include the\nNeutron\u0027s owner key. This static route update task runs only once\nduring the maintenance task, and if all entries are already configured,\nno action is performed.\n\nRelated-bug: #2027742\nChange-Id: I4226024cb4cfd85bdf4f717d42d48150cab22442\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9f2ca21461bd12bf4be5a297e25ec2b354bf0e5f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9f2ca21461bd12bf4be5a297e25ec2b354bf0e5f"}]},"branch":"refs/heads/master"},"ca4d3212ba3b8757de5a35388ffd8d95e10e9a10":{"kind":"REWORK","_number":7,"created":"2024-02-22 18:31:20.000000000","uploader":{"_account_id":35432,"name":"Roberto Acosta","display_name":"rbartzen","email":"rbartzen@gmail.com","username":"rbartzen"},"ref":"refs/changes/89/907489/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/89/907489/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/89/907489/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/89/907489/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/89/907489/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/89/907489/7"}}},"commit":{"parents":[{"commit":"1e75d61bcf610664b65ddb544f69c7c800733c33","subject":"Merge \"Fix unidiomatic-typecheck warnings\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1e75d61bcf610664b65ddb544f69c7c800733c33"}]}],"author":{"name":"Roberto Bartzen Acosta","email":"rbartzen@gmail.com","date":"2024-02-01 19:20:04.000000000","tz":-180},"committer":{"name":"Roberto Bartzen Acosta","email":"rbartzen@gmail.com","date":"2024-02-22 18:31:06.000000000","tz":-180},"subject":"[ML2/OVN] Add external_ids key for Static Routes","message":"[ML2/OVN] Add external_ids key for Static Routes\n\nThis patch updates the method for creating static routes in OVN backend\nand updates previously created static routes rules to include the\nNeutron\u0027s owner key. This static route update task runs only once\nduring the maintenance task, and if all entries are already configured,\nno action is performed.\n\nRelated-bug: #2027742\nChange-Id: I4226024cb4cfd85bdf4f717d42d48150cab22442\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ca4d3212ba3b8757de5a35388ffd8d95e10e9a10"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ca4d3212ba3b8757de5a35388ffd8d95e10e9a10"}]},"branch":"refs/heads/master"},"5b5d081c6913119819fcb7e511ce9ec5b2c92feb":{"kind":"REWORK","_number":8,"created":"2024-03-04 18:05:50.000000000","uploader":{"_account_id":35432,"name":"Roberto Acosta","display_name":"rbartzen","email":"rbartzen@gmail.com","username":"rbartzen"},"ref":"refs/changes/89/907489/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/89/907489/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/89/907489/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/89/907489/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/89/907489/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/89/907489/8"}}},"commit":{"parents":[{"commit":"729937f6d2c3445b5be0c5a26582e419e048319c","subject":"Merge \"[OVN] Make mandatory the router name in the LRP.external_ids\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/729937f6d2c3445b5be0c5a26582e419e048319c"}]}],"author":{"name":"Roberto Bartzen Acosta","email":"rbartzen@gmail.com","date":"2024-02-01 19:20:04.000000000","tz":-180},"committer":{"name":"Roberto Bartzen Acosta","email":"rbartzen@gmail.com","date":"2024-03-04 18:05:07.000000000","tz":-180},"subject":"[ML2/OVN] Add external_ids key for Static Routes","message":"[ML2/OVN] Add external_ids key for Static Routes\n\nThis patch updates the method for creating static routes in OVN backend\nand updates previously created static routes rules to include the\nNeutron\u0027s owner key. This static route update task runs only once\nduring the maintenance task, and if all entries are already configured,\nno action is performed.\n\nRelated-bug: #2027742\nChange-Id: I4226024cb4cfd85bdf4f717d42d48150cab22442\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5b5d081c6913119819fcb7e511ce9ec5b2c92feb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5b5d081c6913119819fcb7e511ce9ec5b2c92feb"}]},"branch":"refs/heads/master"},"1166f5676aa9f91afddd3b23dc31f29b29d7ea6a":{"kind":"REWORK","_number":9,"created":"2024-04-15 14:58:50.000000000","uploader":{"_account_id":35432,"name":"Roberto Acosta","display_name":"rbartzen","email":"rbartzen@gmail.com","username":"rbartzen"},"ref":"refs/changes/89/907489/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/89/907489/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/89/907489/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/89/907489/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/89/907489/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/89/907489/9"}}},"commit":{"parents":[{"commit":"019294c71d94b788c14b23dc1da3c21f51bcdb0b","subject":"Merge \"[OVN] Provide HA functionality to \"Logical_Router\" chassis pinning\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/019294c71d94b788c14b23dc1da3c21f51bcdb0b"}]}],"author":{"name":"Roberto Bartzen Acosta","email":"rbartzen@gmail.com","date":"2024-02-01 19:20:04.000000000","tz":-180},"committer":{"name":"Roberto Bartzen Acosta","email":"rbartzen@gmail.com","date":"2024-04-15 14:37:23.000000000","tz":-420},"subject":"[ML2/OVN] Add external_ids key for Static Routes","message":"[ML2/OVN] Add external_ids key for Static Routes\n\nThis patch updates the method for creating static routes in OVN backend\nand updates previously created static routes rules to include the\nNeutron\u0027s owner key. This static route update task runs only once\nduring the maintenance task, and if all entries are already configured,\nno action is performed.\n\nRelated-bug: #2027742\nChange-Id: I4226024cb4cfd85bdf4f717d42d48150cab22442\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1166f5676aa9f91afddd3b23dc31f29b29d7ea6a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1166f5676aa9f91afddd3b23dc31f29b29d7ea6a"}]},"branch":"refs/heads/master"},"5c4107dc8c8fce3f1fe2c7ab31204f836a0565da":{"kind":"REWORK","_number":10,"created":"2024-05-27 13:10:14.000000000","uploader":{"_account_id":35432,"name":"Roberto Acosta","display_name":"rbartzen","email":"rbartzen@gmail.com","username":"rbartzen"},"ref":"refs/changes/89/907489/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/89/907489/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/89/907489/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/89/907489/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/89/907489/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/89/907489/10"}}},"commit":{"parents":[{"commit":"06c1972a4866c81d8dbb6131c8ea747588855e01","subject":"Merge \"Return both project_id when validating auto allocate network\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/06c1972a4866c81d8dbb6131c8ea747588855e01"}]}],"author":{"name":"Roberto Bartzen Acosta","email":"rbartzen@gmail.com","date":"2024-02-01 19:20:04.000000000","tz":-180},"committer":{"name":"Roberto Bartzen Acosta","email":"rbartzen@gmail.com","date":"2024-05-27 13:09:54.000000000","tz":-360},"subject":"[ML2/OVN] Add external_ids key for Static Routes","message":"[ML2/OVN] Add external_ids key for Static Routes\n\nThis patch updates the method for creating static routes in OVN backend\nand updates previously created static routes rules to include the\nNeutron\u0027s owner key. This static route update task runs only once\nduring the maintenance task, and if all entries are already configured,\nno action is performed.\n\nRelated-bug: #2027742\nChange-Id: I4226024cb4cfd85bdf4f717d42d48150cab22442\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5c4107dc8c8fce3f1fe2c7ab31204f836a0565da"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5c4107dc8c8fce3f1fe2c7ab31204f836a0565da"}]},"branch":"refs/heads/master"},"a0c7336a5588d67d3b2ce7bc676377817f9c10f0":{"kind":"REWORK","_number":11,"created":"2024-05-28 15:33:22.000000000","uploader":{"_account_id":35432,"name":"Roberto Acosta","display_name":"rbartzen","email":"rbartzen@gmail.com","username":"rbartzen"},"ref":"refs/changes/89/907489/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/89/907489/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/89/907489/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/89/907489/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/89/907489/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/89/907489/11"}}},"commit":{"parents":[{"commit":"06c1972a4866c81d8dbb6131c8ea747588855e01","subject":"Merge \"Return both project_id when validating auto allocate network\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/06c1972a4866c81d8dbb6131c8ea747588855e01"}]}],"author":{"name":"Roberto Bartzen Acosta","email":"rbartzen@gmail.com","date":"2024-02-01 19:20:04.000000000","tz":-180},"committer":{"name":"Roberto Bartzen Acosta","email":"rbartzen@gmail.com","date":"2024-05-28 15:33:07.000000000","tz":-360},"subject":"[ML2/OVN] Add external_ids key for Static Routes","message":"[ML2/OVN] Add external_ids key for Static Routes\n\nThis patch updates the method for creating static routes in OVN backend\nand updates previously created static routes rules to include the\nNeutron\u0027s owner key. This static route update task runs only once\nduring the maintenance task, and if all entries are already configured,\nno action is performed.\n\nRelated-bug: #2027742\nChange-Id: I4226024cb4cfd85bdf4f717d42d48150cab22442\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a0c7336a5588d67d3b2ce7bc676377817f9c10f0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a0c7336a5588d67d3b2ce7bc676377817f9c10f0"}]},"branch":"refs/heads/master"},"d4f83ca6e2437216cfbdac7fe3f801bbc520f8f1":{"kind":"REWORK","_number":12,"created":"2024-05-28 15:36:00.000000000","uploader":{"_account_id":35432,"name":"Roberto Acosta","display_name":"rbartzen","email":"rbartzen@gmail.com","username":"rbartzen"},"ref":"refs/changes/89/907489/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/89/907489/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/89/907489/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/89/907489/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/89/907489/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/89/907489/12"}}},"commit":{"parents":[{"commit":"06c1972a4866c81d8dbb6131c8ea747588855e01","subject":"Merge \"Return both project_id when validating auto allocate network\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/06c1972a4866c81d8dbb6131c8ea747588855e01"}]}],"author":{"name":"Roberto Bartzen Acosta","email":"rbartzen@gmail.com","date":"2024-02-01 19:20:04.000000000","tz":-180},"committer":{"name":"Roberto Bartzen Acosta","email":"rbartzen@gmail.com","date":"2024-05-28 15:35:50.000000000","tz":-360},"subject":"[ML2/OVN] Add external_ids key for Static Routes","message":"[ML2/OVN] Add external_ids key for Static Routes\n\nThis patch updates the method for creating static routes in OVN backend\nand updates previously created static routes rules to include the\nNeutron\u0027s owner key. This static route update task runs only once\nduring the maintenance task, and if all entries are already configured,\nno action is performed.\n\nRelated-bug: #2027742\nChange-Id: I4226024cb4cfd85bdf4f717d42d48150cab22442\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d4f83ca6e2437216cfbdac7fe3f801bbc520f8f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d4f83ca6e2437216cfbdac7fe3f801bbc520f8f1"}]},"branch":"refs/heads/master"},"89835e43e8e60b5fdeacdd2a48abacaa9b9778bc":{"kind":"REWORK","_number":13,"created":"2024-08-21 22:11:53.000000000","uploader":{"_account_id":35432,"name":"Roberto Acosta","display_name":"rbartzen","email":"rbartzen@gmail.com","username":"rbartzen"},"ref":"refs/changes/89/907489/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/89/907489/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/89/907489/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/89/907489/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/89/907489/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/89/907489/13"}}},"commit":{"parents":[{"commit":"b847d89ac1f922362945ad610c9787bc28f37457","subject":"Merge \"Revert \"Temporary mark ovs-rally job as non-voting\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b847d89ac1f922362945ad610c9787bc28f37457"}]}],"author":{"name":"Roberto Bartzen Acosta","email":"rbartzen@gmail.com","date":"2024-02-01 19:20:04.000000000","tz":-180},"committer":{"name":"Roberto Bartzen Acosta","email":"rbartzen@gmail.com","date":"2024-08-21 22:11:07.000000000","tz":-180},"subject":"[ML2/OVN] Add external_ids.neutron:is_static_route key for Static Routes","message":"[ML2/OVN] Add external_ids.neutron:is_static_route key for Static Routes\n\nThe Neutron owned Static Routes created in the OVN database are marked\nwith the external_ids key neutron:is_static_route. This mark will be used\nby the OVN DB sync tool to distinguish the Neutron created Static Routes\nfrom those added externally; the second ones should not be altered during\nthe sync process. The static route update task runs only once during the\nmaintenance task, and if all entries are already configured, no action is\nperformed.\n\nRelated-bug: #2027742\nChange-Id: I4226024cb4cfd85bdf4f717d42d48150cab22442\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/89835e43e8e60b5fdeacdd2a48abacaa9b9778bc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/89835e43e8e60b5fdeacdd2a48abacaa9b9778bc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
