)]}'
{"id":"openstack%2Fneutron~907508","triplet_id":"openstack%2Fneutron~master~I433cade500f55c215b4effa38be57cbbe8f8bce6","project":"openstack/neutron","branch":"master","topic":"link_state","attention_set":{},"removed_from_attention_set":{"7130":{"account":{"_account_id":7130,"name":"David Hill","email":"davidchill@hotmail.com","username":"dhill"},"last_update":"2024-04-04 09:01:33.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I433cade500f55c215b4effa38be57cbbe8f8bce6","subject":"In recent ip link set, state needs to be aphabetical","status":"ABANDONED","created":"2024-02-02 00:45:16.000000000","updated":"2024-04-04 09:01:33.000000000","total_comment_count":2,"unresolved_comment_count":2,"has_review_started":true,"meta_rev_id":"6b76005e1ba7af632112ee929a8ca168beac3b12","_number":907508,"virtual_id_number":907508,"owner":{"_account_id":7130,"name":"David Hill","email":"davidchill@hotmail.com","username":"dhill"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2024-02-02 01:47:40.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-02-02 01:38:05.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":-1,"date":"2024-02-06 16:29:44.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-02 01:15:59.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-02-02 01:34:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-02-02 01:38:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-02-06 16:29:44.000000000","updated_by":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"reviewer":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"state":"REVIEWER"},{"updated":"2024-04-04 09:01:33.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"CC"}],"messages":[{"id":"ea24a857e39ae820c2e0f98d29c27190e92b98cb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7130,"name":"David Hill","email":"davidchill@hotmail.com","username":"dhill"},"date":"2024-02-02 00:45:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3d49fd50a622dbbc30be383f04fea6ce912a0dc7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7130,"name":"David Hill","email":"davidchill@hotmail.com","username":"dhill"},"date":"2024-02-02 00:54:25.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"9e64c93050d4370dc8e4e6173f09335721153227","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2024-02-02 01:15:59.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/February-2024/Arista_Tempest_Testrefs/changes/08/907508/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"d9833c4ceda6fdaafddc2ece8251fbbd49f01cd1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-02 01:34:25.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a7e18daee96646f18cf85d2e63bdc218\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/814e974deb1e42d3ac8f7d601d5aaf6c : FAILURE in 38m 08s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/06fa46a45be644b89f55d4a6c0353ccc : FAILURE in 17m 49s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/770c9b5f76094548987504816ea8b5f6 : FAILURE in 28m 59s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8a210e7411e17076d71f1ed997f4d1310102e749","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-02 01:38:05.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0ffcd2a069c545ff9ec592c6cea1d0c6\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/71d6696bd1914725b1fcd4f549b014a2 : FAILURE in 37m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4e3f4ee0475646fdb0a31db1c5b488c3 : SUCCESS in 11m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ba8dd7c0c0f9437e84ebcd1dcea33679 : FAILURE in 28m 40s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6a5ab2ff77f348deb1c2096df906a038 : FAILURE in 35m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2425e2dec02046438090939dd3d46209 : SUCCESS in 10m 44s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/03c10c2d9e394632a995bef3633152c6 : FAILURE in 27m 44s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/421d000407cb41cea01e6c764e7b5cfc : FAILURE in 40m 24s","accounts_in_message":[],"_revision_number":2},{"id":"ce540f8270863aa6a61e392ca0f6947de59d1bd0","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2024-02-02 01:47:40.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/February-2024/Arista_Tempest_Testrefs/changes/08/907508/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"ce4615aa52f7b29615c429a686f80d938b0bedf3","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2024-02-06 16:29:44.000000000","message":"Patch Set 2: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"6b76005e1ba7af632112ee929a8ca168beac3b12","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2024-04-04 09:01:33.000000000","message":"Abandoned\n\nThis review is \u003e 4 weeks without comment, and failed Zuul jobs the last time it was checked. We are abandoning this for now. Feel free to reactivate the review by pressing the restore button and leaving a \u0027recheck\u0027 comment to get fresh test results.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"c4d8892d8c9b12d1d1406b80de722b583b99b69c","revisions":{"cfa9ea6d57bc03f49c3757934c16b1e374a408d4":{"kind":"REWORK","_number":1,"created":"2024-02-02 00:45:16.000000000","uploader":{"_account_id":7130,"name":"David Hill","email":"davidchill@hotmail.com","username":"dhill"},"ref":"refs/changes/08/907508/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/08/907508/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/08/907508/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/08/907508/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/08/907508/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/08/907508/1"}}},"commit":{"parents":[{"commit":"b85b19e3846fa74975ba5d703336b6e7cd8af433","subject":"Merge \"dhcp: rename get_process_uuid as private\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b85b19e3846fa74975ba5d703336b6e7cd8af433"}]}],"author":{"name":"David Hill","email":"dhill@redhat.com","date":"2024-02-02 00:44:30.000000000","tz":-300},"committer":{"name":"David Hill","email":"dhill@redhat.com","date":"2024-02-02 00:44:30.000000000","tz":-300},"subject":"In recent ip link set, state needs to be aphabetical","message":"In recent ip link set, state needs to be aphabetical\n\nIn recent ip link set, state needs to be aphabetical and the current\ncode sends an integer .   If the nic default in the kernel changes from\nauto to enable for some reasons, this breaks.\n\nChange-Id: I433cade500f55c215b4effa38be57cbbe8f8bce6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/cfa9ea6d57bc03f49c3757934c16b1e374a408d4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/cfa9ea6d57bc03f49c3757934c16b1e374a408d4"}]},"branch":"refs/heads/master"},"c4d8892d8c9b12d1d1406b80de722b583b99b69c":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2024-02-02 00:54:25.000000000","uploader":{"_account_id":7130,"name":"David Hill","email":"davidchill@hotmail.com","username":"dhill"},"ref":"refs/changes/08/907508/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/08/907508/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/08/907508/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/08/907508/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/08/907508/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/08/907508/2"}}},"commit":{"parents":[{"commit":"b85b19e3846fa74975ba5d703336b6e7cd8af433","subject":"Merge \"dhcp: rename get_process_uuid as private\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b85b19e3846fa74975ba5d703336b6e7cd8af433"}]}],"author":{"name":"David Hill","email":"dhill@redhat.com","date":"2024-02-02 00:44:30.000000000","tz":-300},"committer":{"name":"David Hill","email":"dhill@redhat.com","date":"2024-02-02 00:54:18.000000000","tz":-300},"subject":"In recent ip link set, state needs to be aphabetical","message":"In recent ip link set, state needs to be aphabetical\n\nIn recent ip link set, state needs to be aphabetical and the current\ncode sends an integer .   If the nic default in the kernel changes from\nauto to enable for some reasons, this breaks.   The current call fails :\n```\n[root@compute001 ~]# ip link set sriov-1-1 vf 18 state 1\nError: argument \"1\" is wrong: Invalid \"state\" value\n```\n\nChange-Id: I433cade500f55c215b4effa38be57cbbe8f8bce6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c4d8892d8c9b12d1d1406b80de722b583b99b69c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c4d8892d8c9b12d1d1406b80de722b583b99b69c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
