)]}'
{"id":"openstack%2Fneutron~908588","triplet_id":"openstack%2Fneutron~stable%2Fzed~I300464472f2348d148f2a3ddac384c883d9d815b","project":"openstack/neutron","branch":"stable/zed","topic":"bug/2052419","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2024-02-20 14:33:22.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I300464472f2348d148f2a3ddac384c883d9d815b","subject":"Change SG rules backref load method to \"joined\"","status":"ABANDONED","created":"2024-02-09 14:45:36.000000000","updated":"2024-02-20 14:33:22.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"c1c0c753ad8132f223ad6ec9480086fb8bb0fc81","_number":908588,"virtual_id_number":908588,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-02-20 13:47:49.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2024-02-20 11:59:03.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-09 15:24:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-02-09 16:56:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-02-12 11:24:03.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"CC"},{"updated":"2024-02-12 13:54:39.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"}],"messages":[{"id":"524f02dd15f20d72bc916f72654f0e228ac5bf10","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-02-09 14:45:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2e6ed53ccbb2287e86927e80773dcf752f617184","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-02-09 14:45:41.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"4a856911ad8caf60dff185f3cb5199a03dbc6e7f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-09 15:24:45.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/79a13ec49b29430d8ad77262bcb10552\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ac83c4be87fc458bbe86795befee8a78 : SUCCESS in 37m 28s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b8b66fe262164107b46f3acd5ef27dc1 : SUCCESS in 23m 24s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"251e842cae50382f7adf4f711fc2f31daa2f00e9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-09 16:56:19.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c8ead50722584e20891a6e659e7b07e9\n\n- neutron-tempest-plugin-linuxbridge-zed https://zuul.opendev.org/t/openstack/build/9b6de6dcd54c4110866660e247d305f7 : SUCCESS in 1h 45m 36s\n- neutron-tempest-plugin-openvswitch-zed https://zuul.opendev.org/t/openstack/build/bf7b42f59676448ea97b756f4a3f1fee : SUCCESS in 1h 33m 59s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-zed https://zuul.opendev.org/t/openstack/build/56c086174d5a4ed796eea419e91ecefa : SUCCESS in 1h 41m 52s\n- neutron-tempest-plugin-ovn-zed https://zuul.opendev.org/t/openstack/build/edbc490aefa649749e3c966aae3e3ac8 : SUCCESS in 1h 28m 01s\n- neutron-tempest-plugin-designate-scenario-zed https://zuul.opendev.org/t/openstack/build/a7c521ac22cc44cfaa3d2b90fbf33fe6 : SUCCESS in 40m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/43d1c803acd0417bad2e4f44508a958c : FAILURE in 38m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/998f7a9a909d41f098cf7b7fb84ffea2 : SUCCESS in 12m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d806f855813441e2930281e064aa92e7 : SUCCESS in 22m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3bb834edbe2346248f0dd68bdbd01765 : FAILURE in 24m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bc4363c316904fe6aa8d0e2d67bb84fe : FAILURE in 35m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/096576f459574c5dba89677a3dfc0a22 : SUCCESS in 11m 49s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/8f94835e62e74f268c54e2f35a60227f : FAILURE in 35m 54s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/924ed240f4ed4ceaa5535c49806008ab : SUCCESS in 1h 35m 00s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/9441857b0af0490c90794de38fcb14a9 : SUCCESS in 1h 17m 23s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/98d8ad00f6c24d66acd19c2b5a6d7999 : FAILURE in 31m 34s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/a15dc8d6152f406ba20ee95a06571697 : FAILURE in 44m 53s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/560ebea32e8d4f2989433570383c38c4 : SUCCESS in 1h 20m 24s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/b9d2e90eae414cf882af495f395e9325 : SUCCESS in 1h 16m 14s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fdfaa8abaefc4147ad57eca9a7ba4902 : SUCCESS in 51m 51s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/ca13363eef804034ae0335248b6a227b : SUCCESS in 1h 13m 30s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"09bc572d264f9b2af16918f470487b7c4ecb367d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-02-10 14:05:54.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":2},{"id":"fcd485f7eeea706d40241d1d6db727726b5e1795","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-10 14:44:30.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7e146151f06c4257b3eb096f87248d6f\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/77857b91d245405096acc45b0a491b9a : SUCCESS in 22m 57s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f582950cfe5c40018c354b2c945da6bc : SUCCESS in 37m 01s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"bce074efb85112f3ba97c15255b3e2dfebeb9003","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-10 16:03:20.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/52993e1df8b24ea1a11ef1c5e7275bae\n\n- neutron-tempest-plugin-linuxbridge-zed https://zuul.opendev.org/t/openstack/build/e397de42ec8c48e0bd9ef53bf5de2f90 : SUCCESS in 1h 38m 28s\n- neutron-tempest-plugin-openvswitch-zed https://zuul.opendev.org/t/openstack/build/619d478fb6ad43c8bc9f69af708ca9b2 : SUCCESS in 1h 06m 44s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-zed https://zuul.opendev.org/t/openstack/build/43056141dc3c40c0aca087cb2ca76be7 : SUCCESS in 1h 13m 46s\n- neutron-tempest-plugin-ovn-zed https://zuul.opendev.org/t/openstack/build/e0c3cbf747734886bf79b248cd9d46a5 : SUCCESS in 1h 36m 04s\n- neutron-tempest-plugin-designate-scenario-zed https://zuul.opendev.org/t/openstack/build/9e2021f58c9a4a30ba702a7771626e48 : SUCCESS in 30m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f8968bda1f084fd2880d8012c7f7a6f3 : SUCCESS in 26m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0de410093efc40dda0e94600a5c27e61 : SUCCESS in 11m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/766c4fef1f8743488210203edea3ce18 : SUCCESS in 22m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/97c0d555e8c545df9f8c2a221259d97b : SUCCESS in 19m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c13450622ae942e6b42a479675998417 : SUCCESS in 19m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f1824e374d942398cf379d3cb909de5 : SUCCESS in 10m 09s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/5ad355924b47425d8eafd93cafb61f08 : SUCCESS in 41m 10s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/b5ae8839998e4d33b679f4ba76bc828d : SUCCESS in 1h 26m 32s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/ed43d9bd9c354abc9c24abd1f2c39a50 : SUCCESS in 50m 40s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/f2ebb49749fb40b5907651f1c2e410b4 : FAILURE in 25m 46s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/670b5fa6dfcc481f97061a006be4802e : FAILURE in 41m 57s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/3025fe9264824fd286d9814807d2de21 : SUCCESS in 1h 21m 49s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/19c3cccf64994b5cbcb4e7a6f2bd487b : SUCCESS in 1h 16m 38s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e0a7a9c73b734e9a9e106a2e822b1cc7 : SUCCESS in 47m 05s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/1019ba0a65a84cff920e4f6c3e03bf2e : SUCCESS in 1h 05m 09s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"4a2723fe35c0a4f83090182579fdc40e28fb2658","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2024-02-12 11:24:03.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"03b497556aa2209fbf08d86542c981c24c439a56","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-12 12:01:21.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/353ad4004bca4610990fa7e68d90d7aa\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/76ef22138551461cacb297c72c096260 : SUCCESS in 21m 06s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/194b9dcf124e4aeda360fdc1f924716e : SUCCESS in 35m 40s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f4024bee64a8ec197b2f741835faca5d38e091d2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-12 13:16:54.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e0bd101ced804d15bb8391d5020c41b0\n\n- neutron-tempest-plugin-linuxbridge-zed https://zuul.opendev.org/t/openstack/build/ec576f7c618b4dbcb1b94b5058e4a603 : SUCCESS in 1h 33m 23s\n- neutron-tempest-plugin-openvswitch-zed https://zuul.opendev.org/t/openstack/build/827c45d598524991a742fb61f94a1b59 : SUCCESS in 1h 23m 04s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-zed https://zuul.opendev.org/t/openstack/build/83cb4ada7da24c9b83d9be5d77314691 : SUCCESS in 1h 26m 07s\n- neutron-tempest-plugin-ovn-zed https://zuul.opendev.org/t/openstack/build/ca5ce11b18894a86906112e7ecde9374 : SUCCESS in 1h 22m 35s\n- neutron-tempest-plugin-designate-scenario-zed https://zuul.opendev.org/t/openstack/build/2ffe3dbf8228425b92a8cf1f9245080f : SUCCESS in 29m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/98694fb5ca9e4252a070bb746695b1ef : SUCCESS in 32m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/99b8087d272442008f541a82bc34e241 : SUCCESS in 10m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/316836b40f4a4510bb8c0307804bd56b : SUCCESS in 17m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8b9213e84dd04561ab148cf1fbc5fa8b : SUCCESS in 27m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d9da41b8a17b450f88927a0ccb9860ed : SUCCESS in 26m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dba9c0a08b0842329fe7167fb8815735 : SUCCESS in 10m 08s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/253bfaf84fd94170ae13033f74c2cd3b : FAILURE in 44m 46s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/bba1fccb09f540d9972f0b635f18be93 : SUCCESS in 1h 21m 00s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/74815694da8a42f18f7160dbff83382f : SUCCESS in 1h 08m 20s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/a8046febe5504eaf949ae40bc32fb7e8 : FAILURE in 38m 13s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/6dd60bde41c34bb897aa8fdc7cdae2fc : FAILURE in 25m 27s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/af7e32c7f4934e95a63dd7f434df6f93 : SUCCESS in 1h 11m 05s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/2886e5d1d9094b2fbc8c6eefd7b8cd73 : SUCCESS in 1h 16m 37s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2411e56592b746b9b8a0da8a8292c0f6 : SUCCESS in 47m 46s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/89cc7c0afbbd45be9f1e2725f425f276 : SUCCESS in 1h 13m 18s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"11fa8f5371e3efe41e49615ee83ae286f8182643","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2024-02-12 13:54:39.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"fce5a40d943fe6d75284b121043caf4c88d48f55","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-02-20 11:59:03.000000000","message":"Patch Set 3: Patch Set 2 was rebased\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"8a9f3d67733272ed87680fb0241776cc3953150e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-20 12:25:36.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/47a029765d9e4e568c13fbc8f82ef06d\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/07cbf646e7ef4d1eb04975b28dee9066 : SUCCESS in 16m 16s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8c4222437ee84226b36f68c0bc30ffa4 : SUCCESS in 24m 45s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"615c3adbb2e437311f404d9ce556eabdc1d65fed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-20 13:47:49.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/65eaac7b45134cf0a6c196e8903ad363\n\n- neutron-tempest-plugin-linuxbridge-zed https://zuul.opendev.org/t/openstack/build/0260154083e14c53b3eb73304e3b3f64 : SUCCESS in 1h 34m 51s\n- neutron-tempest-plugin-openvswitch-zed https://zuul.opendev.org/t/openstack/build/461edb5327254967802a5dd264c07df3 : SUCCESS in 1h 29m 13s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-zed https://zuul.opendev.org/t/openstack/build/206e5c6939724463aa8001e89ef72951 : SUCCESS in 1h 47m 05s\n- neutron-tempest-plugin-ovn-zed https://zuul.opendev.org/t/openstack/build/bec8300e1c8b4adeb484092f981fd8b7 : SUCCESS in 1h 33m 59s\n- neutron-tempest-plugin-designate-scenario-zed https://zuul.opendev.org/t/openstack/build/f22200808ea0454f87f7a0e759d08b42 : SUCCESS in 36m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b4b49d2132ab4f82aac2ea541f27bc9e : SUCCESS in 32m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0a58616c376349a89dc3c1f0d4dd61d3 : SUCCESS in 12m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e6c36e4529904a9685d628144113bee3 : SUCCESS in 23m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4408a5b628ef462ea2be9509fb499502 : SUCCESS in 21m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/112db48f110a4477bb615211effe212f : SUCCESS in 25m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/719500af900d4e1d9049e1bf7586da95 : SUCCESS in 10m 27s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/e81a1e61f81a47cc8606ce88464f2de2 : SUCCESS in 57m 40s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/801438c7929d4e61b60e1cf06b1e4ec5 : SUCCESS in 1h 23m 32s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/187fdfc747724098b4c31820904ff336 : SUCCESS in 1h 05m 57s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/1969202dbaa2457abc7ee0bb24d5238f : FAILURE in 36m 34s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/c823b41486964ce899bd7bfdfa5f6914 : FAILURE in 40m 38s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/9c40eb12340b4555af4b5f760d5d8224 : SUCCESS in 1h 16m 22s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/5e0c500bd24b4a3d94acbfc1553c777e : SUCCESS in 1h 19m 34s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7b91262121314c0fa59a571fe066c953 : SUCCESS in 47m 16s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/369507251bd0411493c53cf61a512b81 : SUCCESS in 1h 03m 08s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"c1c0c753ad8132f223ad6ec9480086fb8bb0fc81","tag":"autogenerated:gerrit:abandon","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-02-20 14:33:22.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"91fac94ec3d34697fda866fb22e6487ac4eb5200","revisions":{"8eba0be089843d9c0aa1e9bcff7e91de75a88461":{"kind":"REWORK","_number":1,"created":"2024-02-09 14:45:36.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/88/908588/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/88/908588/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/908588/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/908588/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/908588/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/88/908588/1"}}},"commit":{"parents":[{"commit":"f83e1fd4f057578e1a0b30edae3cc3161f28f5e9","subject":"Merge \"[OVN][FT] Retry in case of timeout when executing \"ovsdb-client\".\" into stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f83e1fd4f057578e1a0b30edae3cc3161f28f5e9"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-02-08 21:56:06.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-02-08 23:39:50.000000000","tz":0},"subject":"Change SG rules backref load method to \"joined\"","message":"Change SG rules backref load method to \"joined\"\n\nSince [1], we introduced a way to skip the load of the OVO synthetic\nfields depending on the resource fields retrieved. In the case of the\nsecurity groups (SG), the SG rules are child objects to the SGs. The SG\nrules are retrieved when a SG OVO is created.\n\nThe improvement done in [1] is to make the SG rules load dynamically,\nthat means using the load mode \"lazy\u003d\u0027dynamic\u0027\". That will issue a SQL\nquery only if the SG rules are read; if not, the query is never issued.\n\nHowever since [2] (and this is previous to the [1] optimization), we\nalways add the field \"shared\" to the filters and thus to the fields to\nretrieve, because it is a policy required field. Because \"shared\" is a\nsynthetic field [3], that will force the SG \"synthetic_fields\" load and\nthe retrieval of the SG rules always. This is undoing any performance\nimprovement.\n\nThis patch is changing the loading method to \"joined\"; that will request\nthe SG rules in the same SQL query, instead of issuing separate queries\nfor the SG rules. Until a method to load the SG \"shared\" field,\nindependently of the synthetic OVO fields is implemented, this change\nwill improve the SG retrieval performance. In a testing environment\nwith around 5500K SG and 4 rules (default ones) per SG:\n* lazy\u003d\u0027dynamic\u0027: 38 seconds\n* lazy\u003d\u0027select\u0027: 20 seconds\n* lazy\u003d\u0027joined\u0027: 12 seconds\n\n[1]https://review.opendev.org/q/topic:%22bug/1810563%22\n[2]https://review.opendev.org/c/openstack/neutron/+/328313\n[3]https://github.com/openstack/neutron/blob/b85b19e3846fa74975ba5d703336b6e7cd8af433/neutron/objects/rbac_db.py#L349\n\nRelated-Bug: #2052419\nChange-Id: I300464472f2348d148f2a3ddac384c883d9d815b\n(cherry picked from commit be38f9a69a893410e6ca08e116af034b801fc078)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8eba0be089843d9c0aa1e9bcff7e91de75a88461"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8eba0be089843d9c0aa1e9bcff7e91de75a88461"}]},"branch":"refs/heads/stable/zed"},"01496cac948e4d936b01e5607291dc61dd24ce6f":{"kind":"REWORK","_number":2,"created":"2024-02-10 14:05:54.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/88/908588/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/88/908588/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/908588/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/908588/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/908588/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/88/908588/2"}}},"commit":{"parents":[{"commit":"f83e1fd4f057578e1a0b30edae3cc3161f28f5e9","subject":"Merge \"[OVN][FT] Retry in case of timeout when executing \"ovsdb-client\".\" into stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f83e1fd4f057578e1a0b30edae3cc3161f28f5e9"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-02-08 21:56:06.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-02-09 23:00:09.000000000","tz":0},"subject":"Change SG rules backref load method to \"joined\"","message":"Change SG rules backref load method to \"joined\"\n\nSince [1], we introduced a way to skip the load of the OVO synthetic\nfields depending on the resource fields retrieved. In the case of the\nsecurity groups (SG), the SG rules are child objects to the SGs. The SG\nrules are retrieved when a SG OVO is created.\n\nThe improvement done in [1] is to make the SG rules load dynamically,\nthat means using the load mode \"lazy\u003d\u0027dynamic\u0027\". That will issue a SQL\nquery only if the SG rules are read; if not, the query is never issued.\n\nHowever since [2] (and this is previous to the [1] optimization), we\nalways add the field \"shared\" to the filters and thus to the fields to\nretrieve, because it is a policy required field. Because \"shared\" is a\nsynthetic field [3], that will force the SG \"synthetic_fields\" load and\nthe retrieval of the SG rules always. This is undoing any performance\nimprovement.\n\nThis patch is changing the loading method to \"joined\"; that will request\nthe SG rules in the same SQL query, instead of issuing separate queries\nfor the SG rules. Until a method to load the SG \"shared\" field,\nindependently of the synthetic OVO fields is implemented, this change\nwill improve the SG retrieval performance. In a testing environment\nwith around 5500K SG and 4 rules (default ones) per SG:\n* lazy\u003d\u0027dynamic\u0027: 38 seconds\n* lazy\u003d\u0027select\u0027: 20 seconds\n* lazy\u003d\u0027joined\u0027: 12 seconds\n\n[1]https://review.opendev.org/q/topic:%22bug/1810563%22\n[2]https://review.opendev.org/c/openstack/neutron/+/328313\n[3]https://github.com/openstack/neutron/blob/b85b19e3846fa74975ba5d703336b6e7cd8af433/neutron/objects/rbac_db.py#L349\n\nRelated-Bug: #2052419\nChange-Id: I300464472f2348d148f2a3ddac384c883d9d815b\n(cherry picked from commit 52662cad7a07f9bb3b954366eeccd9795ae21503)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/01496cac948e4d936b01e5607291dc61dd24ce6f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/01496cac948e4d936b01e5607291dc61dd24ce6f"}]},"branch":"refs/heads/stable/zed"},"91fac94ec3d34697fda866fb22e6487ac4eb5200":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2024-02-20 11:59:03.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/88/908588/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/88/908588/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/908588/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/908588/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/908588/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/88/908588/3"}}},"commit":{"parents":[{"commit":"b9e3818b6e8905f5717d9888014f7e42141aacf0","subject":"[OVN] Add ``subnetpool-prefix-ops`` extension to ML2/OVN mech driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b9e3818b6e8905f5717d9888014f7e42141aacf0"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-02-08 21:56:06.000000000","tz":0},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2024-02-20 11:59:03.000000000","tz":0},"subject":"Change SG rules backref load method to \"joined\"","message":"Change SG rules backref load method to \"joined\"\n\nSince [1], we introduced a way to skip the load of the OVO synthetic\nfields depending on the resource fields retrieved. In the case of the\nsecurity groups (SG), the SG rules are child objects to the SGs. The SG\nrules are retrieved when a SG OVO is created.\n\nThe improvement done in [1] is to make the SG rules load dynamically,\nthat means using the load mode \"lazy\u003d\u0027dynamic\u0027\". That will issue a SQL\nquery only if the SG rules are read; if not, the query is never issued.\n\nHowever since [2] (and this is previous to the [1] optimization), we\nalways add the field \"shared\" to the filters and thus to the fields to\nretrieve, because it is a policy required field. Because \"shared\" is a\nsynthetic field [3], that will force the SG \"synthetic_fields\" load and\nthe retrieval of the SG rules always. This is undoing any performance\nimprovement.\n\nThis patch is changing the loading method to \"joined\"; that will request\nthe SG rules in the same SQL query, instead of issuing separate queries\nfor the SG rules. Until a method to load the SG \"shared\" field,\nindependently of the synthetic OVO fields is implemented, this change\nwill improve the SG retrieval performance. In a testing environment\nwith around 5500K SG and 4 rules (default ones) per SG:\n* lazy\u003d\u0027dynamic\u0027: 38 seconds\n* lazy\u003d\u0027select\u0027: 20 seconds\n* lazy\u003d\u0027joined\u0027: 12 seconds\n\n[1]https://review.opendev.org/q/topic:%22bug/1810563%22\n[2]https://review.opendev.org/c/openstack/neutron/+/328313\n[3]https://github.com/openstack/neutron/blob/b85b19e3846fa74975ba5d703336b6e7cd8af433/neutron/objects/rbac_db.py#L349\n\nRelated-Bug: #2052419\nChange-Id: I300464472f2348d148f2a3ddac384c883d9d815b\n(cherry picked from commit 52662cad7a07f9bb3b954366eeccd9795ae21503)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/91fac94ec3d34697fda866fb22e6487ac4eb5200"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/91fac94ec3d34697fda866fb22e6487ac4eb5200"}]},"branch":"refs/heads/stable/zed","description":"Rebase"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
