)]}'
{"id":"openstack%2Fneutron~908799","triplet_id":"openstack%2Fneutron~stable%2F2023.2~I101128bdb421ec83df5cdcb0d486cbafbbca2ce5","project":"openstack/neutron","branch":"stable/2023.2","topic":"bug/2028846-stable/2023.2","attention_set":{},"removed_from_attention_set":{"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2024-02-15 10:31:14.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I101128bdb421ec83df5cdcb0d486cbafbbca2ce5","subject":"Log warning about port forwardings that won\u0027t work properly","status":"MERGED","created":"2024-02-12 14:53:56.000000000","updated":"2024-02-15 10:31:14.000000000","submitted":"2024-02-15 10:31:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"908799-bug/2028846-stable/2023.2","meta_rev_id":"86cb49d65f27b4cfe24add92cdad5707f5f772f1","_number":908799,"virtual_id_number":908799,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-02-15 10:31:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2024-02-14 07:59:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2024-02-14 13:12:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2024-02-14 13:12:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-12 15:01:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-02-12 17:02:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-02-14 07:59:03.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2024-02-14 13:12:58.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"}],"messages":[{"id":"ab621070d5672e4c8217ff06654022bf367e7447","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2024-02-12 14:53:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8b78ca4fa1e5962d6746a5aef89d6fe490fdbc68","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2024-02-12 14:59:19.000000000","message":"Topic set to bug/2028846-stable/2023.2","accounts_in_message":[],"_revision_number":1},{"id":"cdadbd5270b71b45d34738149e8b84de5bf258b2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-12 15:01:44.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f2efeb149bd34a778f1f643a3531ad7a\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/85017be495f94d02aea2d1997c646489 : FAILURE in 5m 29s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4545e5f805b34c8e948e9304050063c0 : FAILURE in 5m 58s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/dc7bdbdd2ac14928a006f81bc707ba24 : FAILURE in 4m 46s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/c313c51006584fcf872f3e4865835e08 : FAILURE in 5m 56s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"166103ef9dd174dc11a102196a30f22ea9cf7df8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-12 17:02:11.000000000","message":"Patch Set 1: Verified-1\n\n(8 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/56b378284272425a9f25079444355532\n\n- neutron-tempest-plugin-ovn-2023-2 https://zuul.opendev.org/t/openstack/build/f25986157d3f4c2e8db2692986727402 : FAILURE in 16m 19s\n- neutron-tempest-plugin-designate-scenario-2023-2 https://zuul.opendev.org/t/openstack/build/f478d465c5fe46d9908eecb8cf54e416 : FAILURE in 16m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/82895fabb3454e5d9d2813484f68e904 : FAILURE in 5m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/03e98b154c624037a74194da69ef3506 : FAILURE in 5m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/86706023435440caaa76cbe9165be5b1 : FAILURE in 4m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e3f3437d5c594d5f8aafbab13cf39042 : FAILURE in 4m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/722daa0647814c0eb4b2c81917750718 : FAILURE in 4m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e0d01002d64c4afe935e934dfec454ff : FAILURE in 4m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b6b76345bbda4086a1ffc3ef50078524 : SUCCESS in 12m 02s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/7714e6f7c1e143c194bafacedf2fb38d : FAILURE in 26m 06s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/0a48e5a0faf74fb6ae98dd61b10bd2f4 : SUCCESS in 1h 51m 49s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/1ecc2e012b6a46e690d829e44450b61e : FAILURE in 31m 41s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/d77160890ec9405aa8db9b7f2215fb5b : FAILURE in 32m 25s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/1a474e67ff204e39be32dfbd26dcf629 : FAILURE in 29m 57s\n\nWarning:\n  Comments left for invalid file neutron/cmd/upgrade_checks/checks.py","accounts_in_message":[],"_revision_number":1},{"id":"29d564492aa1336304cb741ac82caa3fd010b2bc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2024-02-13 16:03:25.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"42ef76e87f70428ff6c0ea126dcc35557af462fa","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-13 16:47:25.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0f8560c7514c423f89dcc6728a04f4d5\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/eb1ad45896be4787b37aeaf13e65daf3 : SUCCESS in 41m 06s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/62a18349fb36459b97dbce67bca1fbae : SUCCESS in 20m 57s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/67a5484f56c240f6a4672bc66d83db81 : SUCCESS in 41m 50s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/2608a025f13b4d21912a442e02f71a6f : SUCCESS in 20m 14s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c3017c6277a7298187407a64dbd2792df4578300","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-13 17:43:23.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7f3fcb66597a453a8ea3de30c9f47976\n\n- neutron-tempest-plugin-ovn-2023-2 https://zuul.opendev.org/t/openstack/build/9a4007f870e64b76abb8523ba8c79311 : SUCCESS in 1h 34m 22s\n- neutron-tempest-plugin-designate-scenario-2023-2 https://zuul.opendev.org/t/openstack/build/99d64b665e874fe7853f940aeae90e05 : SUCCESS in 30m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8c082763a2654952866f92db9d340fc7 : SUCCESS in 39m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3a918779b24c4547b6fe04a0081336bf : SUCCESS in 14m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d8e41cb9956f4ca081cdf34e3f078e2d : SUCCESS in 32m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9806e5abea7a45c186f1f0585760577e : SUCCESS in 29m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9d5bd9cf26b74bb4a17fc98d0572b481 : SUCCESS in 24m 33s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8d2cb09df2744424810ddfed4307001e : SUCCESS in 25m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f8d07f8ef8da49d796986253a170187a : SUCCESS in 12m 09s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/a2c3aee9ba3c4311bbf37cac4394a29d : SUCCESS in 47m 58s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/ba60a63c1c4d4eac9417261d2bfcd24c : SUCCESS in 1h 33m 22s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/54bee5e8ab3a4e1fa7adf287c6958a97 : SUCCESS in 54m 50s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/f40d84f33dba4174bba3dacd2e3d5f79 : SUCCESS in 1h 15m 33s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/73c09a11236047c387a88b7191ea95f5 : SUCCESS in 1h 29m 25s","accounts_in_message":[],"_revision_number":2},{"id":"dbfb9127bb40bfd5a310aae35a11137f7344a43c","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-02-14 07:59:03.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"9a323b27bcae111a504590b662869d5d5224b9fb","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-02-14 13:12:58.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"e2fc9a406bbc21e9ba31b0ad9abd25f3095b1145","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-15 08:11:11.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"e6af0b53de8f3107bc82873f8f14024b703f3714","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-15 10:31:14.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7d4ed5fcddbd48a2bd498bc586b91b18\n\n- neutron-tempest-plugin-ovn-2023-2 https://zuul.opendev.org/t/openstack/build/45a923b331704f3b900fa8ec78c5bee4 : SUCCESS in 1h 22m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ed064a5cc61847ceb0f4b401717df7dc : SUCCESS in 11m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/224c9abb53564cc5a489d31db11f19d4 : SUCCESS in 25m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/802cac1060bd4fb180f4eaa7c9751805 : SUCCESS in 24m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/48b28e2e85664d859de8e6b3c5f159c3 : SUCCESS in 24m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/71f98967d8af46d9825a5b69bd0f58f5 : SUCCESS in 12m 17s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/5dc0623762624286beb0c9c5c375b3c9 : SUCCESS in 51m 25s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/0149a6dcc3ea47ee9191c81ede285556 : SUCCESS in 1h 46m 49s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/c82ea4ece1d64204b97df6013fcd74f9 : SUCCESS in 1h 05m 23s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/81edd0e14b2d4833b68d53bdf163e65b : SUCCESS in 1h 04m 32s","accounts_in_message":[],"_revision_number":2},{"id":"86cb49d65f27b4cfe24add92cdad5707f5f772f1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-15 10:31:14.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"3eaa326a7a86d46f5139d6f3c017e6128b391881","revisions":{"07764b3b89eecdabae0a7c5fdc013bc7cb63ce02":{"kind":"REWORK","_number":1,"created":"2024-02-12 14:53:56.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/99/908799/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/99/908799/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/99/908799/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/99/908799/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/99/908799/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/99/908799/1"}}},"commit":{"parents":[{"commit":"1c6d3d448d419d6176647dc0e50760c19d3fefcb","subject":"[OVN] Warn about invalid OVN and FIP PF config during start of Neutron","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1c6d3d448d419d6176647dc0e50760c19d3fefcb"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2024-02-01 10:36:29.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2024-02-12 14:51:43.000000000","tz":60},"subject":"Log warning about port forwardings that won\u0027t work properly","message":"Log warning about port forwardings that won\u0027t work properly\n\nThis is follow up patch to [1] in which was added warning about\nincompatible configuration of the vlan/flat networks allowed as tenant\nnetworks, distributed routing and port forwardings.\n\nIn this new patch similar warning is logged every time when port\nforwarding is created using router which have actually connected vlan or\nflat networks as \"internal networks\" (external gateway network is fine)\nand when distributed routing is enabled in the Neutron config.\n\nThis patch additionally adds \"neutron:is_ext_gw\" flag to the\nLogical_Router_Port\u0027s external_ids. With that it\u0027s easier to check if\nnetwork is used as gateway network (no checks needed) or not (checks are\nperfomed and warning may be logged).\n\n[1] https://review.opendev.org/c/openstack/neutron/+/892542\n\nConflicts:\n    neutron/plugins/ml2/drivers/ovn/mech_driver/ovsdb/maintenance.py\n    neutron/tests/unit/services/ovn_l3/test_plugin.py\n\nRelated-Bug: #2028846\nChange-Id: I101128bdb421ec83df5cdcb0d486cbafbbca2ce5\n(cherry picked from commit a355d2a0d55b62e4337d5c71dc4c4049883c59c3)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/07764b3b89eecdabae0a7c5fdc013bc7cb63ce02"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/07764b3b89eecdabae0a7c5fdc013bc7cb63ce02"}]},"branch":"refs/heads/stable/2023.2"},"3eaa326a7a86d46f5139d6f3c017e6128b391881":{"kind":"REWORK","_number":2,"created":"2024-02-13 16:03:25.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/99/908799/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/99/908799/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/99/908799/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/99/908799/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/99/908799/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/99/908799/2"}}},"commit":{"parents":[{"commit":"37c53085ff20c0ec9ddce844b7566912e61d0d8c","subject":"[OVN] Warn about invalid OVN and FIP PF config during start of Neutron","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/37c53085ff20c0ec9ddce844b7566912e61d0d8c"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2024-02-01 10:36:29.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2024-02-13 16:02:30.000000000","tz":60},"subject":"Log warning about port forwardings that won\u0027t work properly","message":"Log warning about port forwardings that won\u0027t work properly\n\nThis is follow up patch to [1] in which was added warning about\nincompatible configuration of the vlan/flat networks allowed as tenant\nnetworks, distributed routing and port forwardings.\n\nIn this new patch similar warning is logged every time when port\nforwarding is created using router which have actually connected vlan or\nflat networks as \"internal networks\" (external gateway network is fine)\nand when distributed routing is enabled in the Neutron config.\n\nThis patch additionally adds \"neutron:is_ext_gw\" flag to the\nLogical_Router_Port\u0027s external_ids. With that it\u0027s easier to check if\nnetwork is used as gateway network (no checks needed) or not (checks are\nperfomed and warning may be logged).\n\n[1] https://review.opendev.org/c/openstack/neutron/+/892542\n\nConflicts:\n    neutron/plugins/ml2/drivers/ovn/mech_driver/ovsdb/maintenance.py\n    neutron/tests/unit/services/ovn_l3/test_plugin.py\n\nRelated-Bug: #2028846\nChange-Id: I101128bdb421ec83df5cdcb0d486cbafbbca2ce5\n(cherry picked from commit a355d2a0d55b62e4337d5c71dc4c4049883c59c3)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3eaa326a7a86d46f5139d6f3c017e6128b391881"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3eaa326a7a86d46f5139d6f3c017e6128b391881"}]},"branch":"refs/heads/stable/2023.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
