)]}'
{"id":"openstack%2Fneutron~910308","triplet_id":"openstack%2Fneutron~stable%2F2023.2~Ic58640d89952fa03bd1059608ee6c9072fbaabf5","project":"openstack/neutron","branch":"stable/2023.2","topic":"bug/2052787-stable/2023.2","attention_set":{},"removed_from_attention_set":{"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2024-02-28 00:42:00.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ic58640d89952fa03bd1059608ee6c9072fbaabf5","subject":"Ensure that haproxy spawned by the metadata agents is active","status":"MERGED","created":"2024-02-27 11:56:16.000000000","updated":"2024-02-28 00:44:27.000000000","submitted":"2024-02-28 00:42:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"910308-bug/2052787-stable/2023.2","cherry_pick_of_change":909848,"cherry_pick_of_patch_set":1,"meta_rev_id":"08f7c9085fe023a625d5e5cbfa1d68bff6b1144a","_number":910308,"virtual_id_number":910308,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-02-28 00:42:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2024-02-27 12:22:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2024-02-27 20:54:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2024-02-27 20:54:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-27 12:22:49.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2024-02-27 12:28:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-02-27 14:11:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-02-27 20:54:36.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"}],"messages":[{"id":"852a7d9a11f4ee18a1104aa381cb506987feda6c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2024-02-27 11:56:16.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"0282c87fc925c97f1dbb97efe9174ceaec51c4a4","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-02-27 12:22:49.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d6f0ae245388b1cd5f3c031387378c957cd198ba","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-27 12:28:01.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f02d844fc1ac46dda1a7f835c60d57f7\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/67981cc784d94ba1996181f03e7854ff : SUCCESS in 22m 02s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/54f707deb13a4badb38a4fec252c976a : SUCCESS in 23m 02s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/24dec239b476480ba5bbc87f8041b653 : SUCCESS in 22m 30s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/7ca2a490c855459b901475ff234dd642 : SUCCESS in 29m 57s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"22016356dafc135e4523e827d511945fd1c24096","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-27 14:11:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9bc44b988f6b484c98b7a1c035deb428\n\n- neutron-tempest-plugin-openvswitch-2023-2 https://zuul.opendev.org/t/openstack/build/1c4d0e307c64419bbf70a5203a572b35 : SUCCESS in 1h 45m 22s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2023-2 https://zuul.opendev.org/t/openstack/build/0b8f24dbf8324d409bd12cd1f7c9f712 : SUCCESS in 2h 11m 47s\n- neutron-tempest-plugin-ovn-2023-2 https://zuul.opendev.org/t/openstack/build/b4dc5e3e57ea4c68871ec5514911c3fc : SUCCESS in 1h 53m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/508ef80178b7480c8ddaa981f9efdfbe : SUCCESS in 32m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7c473e684b1d4907b47bf59c42f3c0cc : SUCCESS in 11m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/932e2071c23e4498b96caabc86521b61 : SUCCESS in 20m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/95e0ad1c30b0493b8b4b9ca4c8ec6e1a : SUCCESS in 22m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/27fc588440ef435e9321ab103c494ba1 : SUCCESS in 23m 57s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/407f06809e7646f6b44d1b669e0e36b0 : SUCCESS in 23m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/00110c22b87049b082f4d95a40284d48 : SUCCESS in 12m 24s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/304c876d28fe4eabb1b3bc8fab122627 : SUCCESS in 52m 54s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/64ea3449441442118b86562c82cdcb0e : SUCCESS in 1h 50m 44s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/00f86814215640d09b4caeaef166aba0 : SUCCESS in 58m 46s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/b73391a88a9948539b1a4f9a86ae3fa0 : SUCCESS in 52m 56s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/8dcd2aace2dd41e5a99bd1692a598f72 : SUCCESS in 1h 15m 19s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/048a22c8abd54c299648ffc6136be889 : SUCCESS in 1h 14m 13s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/50e3c17598f94f4190a025c95ed820a3 : SUCCESS in 1h 00m 25s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/d71680dd08e54fe0b805a1346232f007 : SUCCESS in 1h 11m 37s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/7ad4cfb6418446528c6e447008a945a9 : SUCCESS in 1h 21m 31s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/64f62ee8bc5241f3b51c990478f39394 : SUCCESS in 1h 01m 52s","accounts_in_message":[],"_revision_number":1},{"id":"813d96c621ae70ab2fee2985adc5921de4ca1a8d","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-02-27 20:54:36.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"245d78979f70722cefb861fd2acf8b15af911314","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-27 20:55:25.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"14c3cb407a4b6b85cd9037cfdf8cb07b9d3940c9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-28 00:42:00.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c49e81249db8473e9144a75097bda740\n\n- neutron-tempest-plugin-ovn-2023-2 https://zuul.opendev.org/t/openstack/build/53f5102d19394668ac0516b5447ba98c : SUCCESS in 2h 03m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/144526e5f741452aac8fe9ecbc425ab4 : SUCCESS in 10m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0ef98ce9307c4351877b42abcab41f21 : SUCCESS in 25m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4d550321c2594f0da2587afc8e33e488 : SUCCESS in 24m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e97c8bb8e0fd47af82a5dfc9dfc4da9a : SUCCESS in 25m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/42cfdeeaaaf64b4a906f96eb188f71df : SUCCESS in 13m 07s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/d079f7114bc34e08aaf7f8c2b704a4fe : SUCCESS in 54m 39s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/14648cfad52142bc9a15db0ab27ba806 : SUCCESS in 1h 41m 06s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/0b62425109ca4c9b8080e6a69ddebd88 : SUCCESS in 1h 22m 16s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/59be964c8473449fa81d6f3e1950d4e1 : SUCCESS in 58m 28s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/9e60cd3eb87644ddb67ca7a1d8481786 : SUCCESS in 52m 47s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/1a3085c46a164028abd20b084a08a0a6 : SUCCESS in 46m 50s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/57f65d27ec77494f84c8f711f8a8dfef : SUCCESS in 1h 07m 02s","accounts_in_message":[],"_revision_number":1},{"id":"30c6463e1d6e346bad586d9de575975f80abd144","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-28 00:42:00.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"08f7c9085fe023a625d5e5cbfa1d68bff6b1144a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-28 00:44:27.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6bca6a5b0f6541eeb6bc319d31b9f6c7\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4cd1e5fddd78430186b5e9470a732bbb : SUCCESS in 57s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"32af674783c69c87d0feed622434c6839938a141","revisions":{"32af674783c69c87d0feed622434c6839938a141":{"kind":"REWORK","_number":1,"created":"2024-02-27 11:56:16.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/08/910308/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/08/910308/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/08/910308/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/08/910308/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/08/910308/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/08/910308/1"}}},"commit":{"parents":[{"commit":"6e578f9d64e92b32e9eb152d279f53675253d53b","subject":"Merge \"Disallow subnet cidr of :: without PD\" into stable/2023.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6e578f9d64e92b32e9eb152d279f53675253d53b"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2024-02-22 09:06:58.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2024-02-27 11:56:16.000000000","tz":0},"subject":"Ensure that haproxy spawned by the metadata agents is active","message":"Ensure that haproxy spawned by the metadata agents is active\n\nIn both neutron-metadata and neutron-ovn-metadata agents we should\nensure that haproxy service spawned for network/router is actually\nactive before moving on.\nThis patch adds that check and this is similar to what was already\nimplemented some time ago for the dnsmasq process spawned by the dhcp\nagent.\n\nRelated-Bug: #2052787\nChange-Id: Ic58640d89952fa03bd1059608ee6c9072fbaabf5\n(cherry picked from commit 2f7f7c2fc29d0ac26b5ff9d82867952a40f0fa1b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/32af674783c69c87d0feed622434c6839938a141"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/32af674783c69c87d0feed622434c6839938a141"}]},"branch":"refs/heads/stable/2023.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
