)]}'
{"id":"openstack%2Fneutron~910913","triplet_id":"openstack%2Fneutron~stable%2Fzed~Icc84b54be07d39059723d6c233c03aa130102423","project":"openstack/neutron","branch":"stable/zed","topic":"bug/2054324-stable/zed","attention_set":{},"removed_from_attention_set":{"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2024-04-24 10:44:06.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Icc84b54be07d39059723d6c233c03aa130102423","subject":"Fix iptables mapping of \u0027ipip\u0027 protocol","status":"MERGED","created":"2024-03-04 16:19:03.000000000","updated":"2024-04-24 10:45:28.000000000","submitted":"2024-04-24 10:44:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"910913-bug/2054324-stable/zed","cherry_pick_of_change":909943,"cherry_pick_of_patch_set":3,"meta_rev_id":"468c17c50f43c4b6d36fde1def8d99d0cf636ad2","_number":910913,"virtual_id_number":910913,"owner":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-04-24 10:44:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":2,"date":"2024-03-18 12:42:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2024-04-24 08:23:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":1,"date":"2024-04-24 08:23:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-03-04 16:47:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-03-04 18:31:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-03-18 12:42:33.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2024-04-24 08:23:33.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"1e6c16f9865b740877366d2dac25db6b36bcad76","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-03-04 16:19:03.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"ba616bc3d3dcbdc0af432fa2103d5f86d64d6cd1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-04 16:47:40.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7feb008d99144bfc822d5926635ac73a\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9dd2eb58f7564ab0ae6124eaf897476b : SUCCESS in 27m 29s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1f2c1597390047cf8bc4bda0edf7d212 : SUCCESS in 24m 52s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5f4c7094b0c4f502b23528dbf022228749b769f5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-04 18:31:12.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/67257ea9d11543fb99857f4e29a02814\n\n- neutron-tempest-plugin-linuxbridge-zed https://zuul.opendev.org/t/openstack/build/e658f4e089bf4b1da6b9bdc9c8dfd6b9 : SUCCESS in 2h 03m 51s\n- neutron-tempest-plugin-openvswitch-zed https://zuul.opendev.org/t/openstack/build/cf8f9c382e3e477085a40a175b178792 : SUCCESS in 1h 43m 54s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-zed https://zuul.opendev.org/t/openstack/build/9e34176c4b494bf5846269a68d7f8ab1 : SUCCESS in 1h 24m 17s\n- neutron-tempest-plugin-ovn-zed https://zuul.opendev.org/t/openstack/build/7f71a736edbe49b8a63ad4915035b3b1 : SUCCESS in 2h 06m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/31f59ecdc1b246e98a845ad8e429f715 : SUCCESS in 36m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d3ac784ccdef43418adcfb9c2cc595c2 : SUCCESS in 13m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/83eab12e25294a6f967451bc243b1312 : SUCCESS in 22m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ee549e4384df48a5880327ca64787999 : SUCCESS in 24m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/af3a59ec17d949ebbaa5d6d8fe93af7a : SUCCESS in 23m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5bb93c980dae46f1b7d6247259a7c6ce : SUCCESS in 9m 09s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/a335644b60264caa9a14ffe7fd97b2ab : FAILURE in 47m 25s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/ef757c687825438f99b01435210e74c0 : SUCCESS in 1h 21m 13s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/506a61886ff94070bc17dcda2b1f71ae : SUCCESS in 59m 41s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/a80d162bb73f43bfb34cc9171b7538c8 : SUCCESS in 1h 16m 05s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/388828a0481346989d0882610f5b9610 : SUCCESS in 1h 20m 02s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7d8c5f3e17334113906a076a96741178 : SUCCESS in 1h 07m 58s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/b97279d6f2d64bcf89864e3d77bef305 : SUCCESS in 1h 07m 03s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"69792313cc91d0d0b0c7ca61c6c1baae9910a472","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-03-04 18:32:14.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"04329cfa3b10c8e4d543fe8c8838fdea4df5306f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-04 19:05:21.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/af00258fc1404dfbb980ddc106b5a072\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/51f7c2d150114c89b24777574b0675e2 : SUCCESS in 32m 22s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6a02b9ba26194d709fd93ada9ec16f2b : SUCCESS in 27m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7c727e528136c788dd5e77586eaeed11dfc0989d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-04 20:34:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/989e3c5acadd4ef7b8864d68363767b8\n\n- neutron-tempest-plugin-linuxbridge-zed https://zuul.opendev.org/t/openstack/build/974b4608d4e74249a93479e5f2da3834 : SUCCESS in 2h 00m 46s\n- neutron-tempest-plugin-openvswitch-zed https://zuul.opendev.org/t/openstack/build/662569774c4d4adc85f6d4aef529f757 : SUCCESS in 1h 20m 00s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-zed https://zuul.opendev.org/t/openstack/build/187dcff8a89f48f1a4c7f24f58369a69 : SUCCESS in 1h 44m 43s\n- neutron-tempest-plugin-ovn-zed https://zuul.opendev.org/t/openstack/build/03f74991bd64442ea890ba6c0f7c8b1c : SUCCESS in 1h 33m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/63fb186c89be40319504605cdd4beda9 : SUCCESS in 30m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e22a433aa48b4535a0c014896f8ac489 : SUCCESS in 11m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/144799185cb34a3eaa05ca1abdd32929 : SUCCESS in 21m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f81d1c87ee024251a91b3327018a96ee : SUCCESS in 22m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/25b9ddf5d4824669b582c7f62f55fb43 : SUCCESS in 23m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5105271a15f14dcdabd6acac64d86406 : SUCCESS in 9m 44s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/12a701a1d5f841209db6861a7ed8142b : SUCCESS in 52m 52s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/0b977088248845e1a75fd6ee7b37bcda : SUCCESS in 1h 24m 34s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/aefabcece5c24bedabe83d4de821338f : SUCCESS in 1h 04m 38s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/9716da88105244519fc04281f74aa92f : SUCCESS in 1h 17m 44s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/bf2502b0fe164aa298a94b8065d85cb8 : SUCCESS in 1h 15m 59s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0269d6e689b9487e9df47fdd6d3ad63c : SUCCESS in 50m 38s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/7796aa55b1f941aea82f2993990beb17 : SUCCESS in 1h 05m 39s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c5af16837325bf80ef0ef99c7fb1ab4dbb6fd40a","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2024-03-18 12:42:33.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"a6c9613e446dce7e0d31e64d4871133c7a9f97c5","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2024-04-24 08:23:33.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7a0712af91cd8b1c1f5ac35be86e782c93d85fbd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-24 08:24:05.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9fee332ee99defdebcc8acaa95786bcae9be2553","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-24 10:44:06.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4fec710f13e24de8be51e898c348db73\n\n- neutron-tempest-plugin-ovn-zed https://zuul.opendev.org/t/openstack/build/a35b7f9342444fd68dc82924d55e7a80 : SUCCESS in 2h 14m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b9560a7c44874fb4b0becb81d0d361f5 : SUCCESS in 10m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/08780b2913934560841173799b073c19 : SUCCESS in 24m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/89faa83aaaaa4dc4bac0cfcc9f8f6ece : SUCCESS in 17m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bcc93c611c7649e9a92a3a8e1c8ab455 : SUCCESS in 11m 21s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/845818936129494fb4b3a449f3798052 : SUCCESS in 44m 31s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/0cadd62520c3465a9e16801f0134fd45 : SUCCESS in 1h 24m 42s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/027508992ff543659498167d0e6b3f3c : SUCCESS in 56m 07s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2a04c0cfd7504e139703a341d74a0fca : SUCCESS in 47m 52s","accounts_in_message":[],"_revision_number":1},{"id":"fb5782c46276c31e4d59add8943a40b84e5f1318","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-24 10:44:06.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"468c17c50f43c4b6d36fde1def8d99d0cf636ad2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-24 10:45:28.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1ff5b35862124824ba75dff4f6411e11\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a3a090b984c44cca9450484433770a3 : SUCCESS in 1m 00s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7c7b020abe15fb7086dcca9095cddb964f03eab6","revisions":{"7c7b020abe15fb7086dcca9095cddb964f03eab6":{"kind":"REWORK","_number":1,"created":"2024-03-04 16:19:03.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/13/910913/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/13/910913/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/910913/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/910913/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/910913/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/13/910913/1"}}},"commit":{"parents":[{"commit":"d253552ad26f4920ea9f1f53e5435f291f391790","subject":"[stable-only][OVN] Set VETH interface MAC address before up","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d253552ad26f4920ea9f1f53e5435f291f391790"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2024-02-22 23:33:14.000000000","tz":-300},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2024-03-04 16:19:03.000000000","tz":0},"subject":"Fix iptables mapping of \u0027ipip\u0027 protocol","message":"Fix iptables mapping of \u0027ipip\u0027 protocol\n\nMap \u0027ipip\u0027 to use the string \u0027ipencap\u0027 so the\nIptablesFirewallDriver class in neutron works correctly.\nOnce neutron-lib is bumped this can be removed.\n\nAdd tests for IP protocol \u0027ipip\u0027, \u00274\u0027 and \u002794\u0027 to make\nsure the IptablesFirewallDriver class in neutron treats\nthem correctly.\n\nLong description below.\n\nThis is one of those confusing edge cases and I think\nLinux is conspiring against us. Let me explain.\n\n1) neutron-lib does correctly define the protocol name \u0027ipip\u0027 as 4.\n\n2) The linux kernel uses the same in in.h:\n\n IPPROTO_IPIP \u003d 4\n IPPROTO_BEETPH \u003d 94 (?)\n\n3) iptables maps \u0027ipip\u0027 to 94 and \u0027ipencap\u0027 to 4.\n\n # for num in {0..255}; do iptables -A INPUT -p $num; done\n # iptables-save | grep -E \u0027ipip|ipencap\u0027\n -A INPUT -p ipencap\n -A INPUT -p ipip\n\n4) /etc/protocols does the same as iptables:\n\n grep -E \u0027ipencap|ipip\u0027 /etc/protocols\n ipencap 4 IP-ENCAP # IP encapsulated in IP (officially ``IP\u0027\u0027)\n ipip 94 IPIP # IP-within-IP Encapsulation Protocol\n\n5) getprotoby{name|number} does what /etc/protocols does:\n\n $ getprotobyname ipip\n struct protoent: (0x7fbbbcca9c60)\n   p_name ipip\n   p_aliases IPIP\n   p_proto 94\n\n $ getprotobynumber 4\n struct protoent: (0x7fc51ad86be0)\n   p_name ipencap\n   p_aliases IP-ENCAP\n   p_proto 4\n\nNeutron actually builds a mapping based on the getprotoby*\ncalls, so in the iptables case it winds-up doing the wrong\nthing.\n\nPartial-bug: #2054324\nChange-Id: Icc84b54be07d39059723d6c233c03aa130102423\n(cherry picked from commit 793dfb04d087c96470dc4913eb4f17aeff777534)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7c7b020abe15fb7086dcca9095cddb964f03eab6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7c7b020abe15fb7086dcca9095cddb964f03eab6"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
