)]}'
{"id":"openstack%2Fneutron~912072","triplet_id":"openstack%2Fneutron~stable%2F2023.2~I2e703ab4b63c90d7a14f0dc41d37b0a98163bce0","project":"openstack/neutron","branch":"stable/2023.2","attention_set":{},"removed_from_attention_set":{"34118":{"account":{"_account_id":34118,"name":"Arnau Verdaguer","email":"averdagu@redhat.com","username":"averdaguer"},"last_update":"2024-03-13 04:19:36.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I2e703ab4b63c90d7a14f0dc41d37b0a98163bce0","subject":"Fix TestOVNMechanismDriver ipv6 tests","status":"MERGED","created":"2024-03-12 11:19:55.000000000","updated":"2024-03-13 04:21:04.000000000","submitted":"2024-03-13 04:19:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"912072","cherry_pick_of_change":912475,"cherry_pick_of_patch_set":1,"meta_rev_id":"2457f50b11f3ea03cc425f808b1b3cdb6de6f4df","_number":912072,"virtual_id_number":912072,"owner":{"_account_id":34118,"name":"Arnau Verdaguer","email":"averdagu@redhat.com","username":"averdaguer"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-03-13 04:19:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":2,"date":"2024-03-12 13:21:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2024-03-13 03:49:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":1,"date":"2024-03-13 03:49:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-03-12 12:03:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-03-12 12:09:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-03-12 13:21:47.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2024-03-13 03:49:51.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"}],"messages":[{"id":"ac88ca5b42a726040a1dcdf3147109558afa133c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34118,"name":"Arnau Verdaguer","email":"averdagu@redhat.com","username":"averdaguer"},"date":"2024-03-12 11:19:55.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"02d9601da649ea53fb3bc6c0dc7444967d49d28b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-12 12:03:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3eec4ebe0dc0422ebf4ef512e643d84e\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c771c42019634c909ce25629809bf970 : SUCCESS in 35m 18s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2c07df23a9894814a753c6b934da83db : SUCCESS in 22m 35s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/661bba3152f44baeb8aef35bc97e66c3 : SUCCESS in 21m 43s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/c20d0904481f4a2d953c792818a1cf90 : SUCCESS in 19m 52s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ad6414b411db33e9678ac473d0a8ce7a63e9055b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-12 12:09:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2f07d982e777427ca5a22754bd126865\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/eb00ec00c9954975a464be5967ac666e : SUCCESS in 38m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0cb45364117f424c9bce4c913bb0f3a5 : SUCCESS in 11m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0799f55f3c594bf1ad78ce19e75eed5c : SUCCESS in 22m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4fddbadab8084c2bb373442414ef1008 : SUCCESS in 32m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/26c6fc163cf74966a432d83cf6ec9da2 : SUCCESS in 22m 48s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d0c5004dca704cfb85bdcac4468929b7 : SUCCESS in 19m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/69a8d77df0ee439f8d9848f878682bb6 : SUCCESS in 13m 17s","accounts_in_message":[],"_revision_number":1},{"id":"e84782b3d202d32bc639e10698726afcfc1f2b2d","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2024-03-12 13:21:47.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"cddf89ff1b2c09e3c0c238f6096c4c609362e743","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-03-13 03:49:51.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"49c5758264f3d95ab835a5f403e187d005f80be2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-13 03:50:12.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"71bf6f80e14106680187e9c0c48330237270ec7d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-13 04:19:36.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/011b6db1c1644dee92738385d4cf6be0\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dfb324f05a07427ba60ffcf6eea1ff77 : SUCCESS in 10m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1c91a800cac643f6842ab822439f9c2f : SUCCESS in 22m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/53da7c2647534809865c33058456b3d1 : SUCCESS in 28m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a70bb1225503472d8fbde7dd3c88ef2f : SUCCESS in 23m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0db6fb52f5744d25b3ef321ad80294b9 : SUCCESS in 10m 26s","accounts_in_message":[],"_revision_number":1},{"id":"939003304f2a8ced1dde5fb211919bd39eb44873","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-13 04:19:36.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"2457f50b11f3ea03cc425f808b1b3cdb6de6f4df","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-13 04:21:04.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2ee23e38123f46efb22a1eddca1ff651\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/35cf272736c14ee49061be9050d5b673 : SUCCESS in 1m 07s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8db03d65e851033d3830a6b502c839c03111f16d","revisions":{"8db03d65e851033d3830a6b502c839c03111f16d":{"kind":"REWORK","_number":1,"created":"2024-03-12 11:19:55.000000000","uploader":{"_account_id":34118,"name":"Arnau Verdaguer","email":"averdagu@redhat.com","username":"averdaguer"},"ref":"refs/changes/72/912072/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/72/912072/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/72/912072/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/72/912072/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/72/912072/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/72/912072/1"}}},"commit":{"parents":[{"commit":"101898fde869a5f530580d1c69c52cdc4a33cd98","subject":"[OVN] Set MTU of the VETH interfaces between OVS and metadata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/101898fde869a5f530580d1c69c52cdc4a33cd98"}]}],"author":{"name":"Arnau Verdaguer","email":"averdagu@redhat.com","date":"2024-03-11 16:30:11.000000000","tz":60},"committer":{"name":"Arnau Verdaguer","email":"averdagu@redhat.com","date":"2024-03-12 11:19:55.000000000","tz":0},"subject":"Fix TestOVNMechanismDriver ipv6 tests","message":"Fix TestOVNMechanismDriver ipv6 tests\n\n- test_update_subnet_dhcp_options_in_ovn_ipv6_not_change\n- test_enable_subnet_dhcp_options_in_ovn_ipv6\nThis tests will fail if host where unit tests has ipv6 dns_servers\nconfigured. This patch mocks get_system_dns_servers to avoid tests\nto look at the host configuration.\n\nCloses-Bug: #2056778\nChange-Id: I2e703ab4b63c90d7a14f0dc41d37b0a98163bce0\n(cherry picked from commit 2a196fefd4a8a0c15176fe87d88b05d1690ab5fa)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8db03d65e851033d3830a6b502c839c03111f16d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8db03d65e851033d3830a6b502c839c03111f16d"}]},"branch":"refs/heads/stable/2023.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
