)]}'
{"id":"openstack%2Fneutron~912349","triplet_id":"openstack%2Fneutron~master~I13b4c804ea6a9f8a89d3796c1cec88ffa1de6ded","project":"openstack/neutron","branch":"master","topic":"bug/2056558","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2024-03-15 10:30:43.000000000","reason":"Change was submitted"},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2024-03-15 10:30:43.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I13b4c804ea6a9f8a89d3796c1cec88ffa1de6ded","subject":"[OVN] Use the LSP update event to update the LRP","status":"MERGED","created":"2024-03-10 17:21:49.000000000","updated":"2024-03-15 10:32:51.000000000","submitted":"2024-03-15 10:30:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"912349-bug/2056558","meta_rev_id":"171ecfb94930ed07fd910590e946e1b8390460c5","_number":912349,"virtual_id_number":912349,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},{"value":1,"date":"2024-03-13 11:55:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-03-15 10:30:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"recommended":{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},"all":[{"value":1,"date":"2024-03-13 14:43:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2024-03-15 08:30:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":2,"date":"2024-03-15 08:23:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"all":[{"value":0,"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":1,"date":"2024-03-15 08:30:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}],"REVIEWER":[{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-03-10 17:49:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-03-10 17:55:52.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-03-10 19:12:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-03-11 16:51:22.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2024-03-13 14:43:09.000000000","updated_by":{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},"reviewer":{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},"state":"REVIEWER"},{"updated":"2024-03-15 08:23:11.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2024-03-15 08:30:40.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"}],"messages":[{"id":"fd56b748ccd74f9f6066887471c67e381d696b02","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-03-10 17:21:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b722f5f68cedfb61f485739fdf9fb4b5b7b59211","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-10 17:49:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3abc6586659b4988a46d5598e7d74a7f\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/13bd7321f98c4bb1b32fe5e4579aa4ad : FAILURE in 17m 18s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/faaab4660108496aa7297fb4a2a84104 : FAILURE in 17m 14s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/0e5acf1a5ee046b094899f8d79e8abc1 : FAILURE in 25m 45s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"84b2fc25f1daf00412a9339ef6096e5181c85f64","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2024-03-10 17:55:52.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2024/Arista_Tempest_Testrefs/changes/49/912349/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"dcbde469b21bf37cd1ecde2753c69caa029056ba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-10 19:12:03.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ff2dfd54fce747d9b286a26e2d9809db\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/d0ae6a2e1a9c4ae5a61bdc576011bd69 : SUCCESS in 1h 43m 41s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/e76b5e1ea3db479aa65c63bad7be7509 : SUCCESS in 33m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/70bd1901ccaa48ee8889fe9a35b7ee23 : FAILURE in 36m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f93477a2c39d4fc7bfe485614d6a44b9 : SUCCESS in 8m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c275e0dc74ed40b8a8cdf00234cfca82 : FAILURE in 20m 49s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/15a3048983034eceb6eccb1cdb5e3abd : FAILURE in 17m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f5b10818a7e04959a27c6db68b033ea4 : SUCCESS in 9m 13s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/7d4728fbf76f47e5a216c9a65bdb8903 : FAILURE in 18m 46s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/f262cc1270a6485cadfa6ab4d89d3852 : SUCCESS in 44m 47s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/52072600d77d47c29a1fce74a0a33835 : FAILURE in 46m 50s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/f974d8331ed146a38c03de99070b56c7 : SUCCESS in 1h 48m 57s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/bb7675da37414f34b0ec78bb6cc6e7cc : SUCCESS in 1h 09m 17s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/6b31f508084d413482260890a2a1a35e : SUCCESS in 1h 32m 00s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/962e1c00e6754aeabe98c2570fd21235 : SUCCESS in 50m 59s","accounts_in_message":[],"_revision_number":1},{"id":"9b986c4b690aeb235373be80fecb2e40515d0bca","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-03-11 13:21:08.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1, Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"bf1153e5c2850099a99a71c4bea334dd2b08171e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-11 13:48:37.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a0959eecfe3049508f8ee5b209e9df39\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3144905fab4a49cd989e360d9afdc984 : FAILURE in 23m 20s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c4b30ef165824e2b8afd5682ffbdf3a6 : FAILURE in 22m 57s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/f73b47dc7bf64e11b3134cf4a75bccfb : FAILURE in 22m 15s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"5c6f4cf447063b5bc1dd28db2bf8b9701753f5f9","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2024-03-11 13:55:15.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2024/Arista_Tempest_Testrefs/changes/49/912349/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"37aababc0f3e7ff4ba40674c7d6fd1cbb63d0883","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-03-11 14:56:33.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"e8bcd5bb04f13993af98e85d1768ea08a0adbf97","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-03-11 15:16:51.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"b67f6e0a450222c933d5dac3986449221fc0e466","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2024-03-11 15:30:54.000000000","message":"Patch Set 3: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2024/Arista_Tempest_Testrefs/changes/49/912349/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"27e2a4505a06c693baa8bdb040dd066d558205d1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-11 15:39:39.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e4d911ec8bd741d7a92210009e2e2992\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b48044efbc6c4a76985d6cac005d7b13 : FAILURE in 21m 39s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/7073e95b0bfa49fea812e5670970a8d1 : FAILURE in 21m 35s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/d4183219f9f344ec99ebad1add1835e4 : FAILURE in 19m 59s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"f3e9c18e0c72ef08c4c5f72fefce39f5adffc7b3","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2024-03-11 16:05:11.000000000","message":"Patch Set 4: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2024/Arista_Tempest_Testrefs/changes/49/912349/4 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":4},{"id":"6da08aaa0b56db9129a9d9ff6da7b28bbab4437c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-03-11 16:44:39.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"e3f6b400bc915917979f7e95d03e4af5c7ed61b4","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-03-11 16:51:22.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"07585640a526359d924bfe8805ba3f3c55526973","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-11 17:14:24.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e5db94971e5e4814a8b0c87eafe1477f\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f7fb8aa41d7742318cb1f941dc7b0982 : SUCCESS in 20m 44s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/ca9a9b002e7646dd9da2a728bdb78c88 : SUCCESS in 20m 36s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/d440a00ccf4c4386a05940c8d7201cd9 : SUCCESS in 28m 00s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"a7770fa3a8ad711ad566deb3b6bffaf5e7f2fda5","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2024-03-11 17:49:42.000000000","message":"Patch Set 5: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2024/Arista_Tempest_Testrefs/changes/49/912349/5 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":5},{"id":"851f10b47e7e735c214159de9cb7caa7da49cd3a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-11 18:31:31.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7bddd1b84aee4062a95ef7f26846be55\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/38d66a69eb93405caedc9689b4e70154 : SUCCESS in 1h 35m 03s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/1ec19fec13454c16bb8c77781e7c7478 : SUCCESS in 45m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/85e7495dc83040a49b3f5aae08fde0ed : SUCCESS in 43m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/336d964cb2314ae3bd7faaadeb4383cd : SUCCESS in 10m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bf1c5af0de30441297f1a07d94f468c9 : SUCCESS in 25m 01s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/247aa9b442f54ef3ba657dcfec5bbab1 : SUCCESS in 21m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e6cff9e1f0014091ae824a141c5d3681 : SUCCESS in 10m 51s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/6f8f9c34412b472c9352b9a3f49a495a : SUCCESS in 27m 17s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/633e84cee988479ea0ab45a8e78fd301 : SUCCESS in 55m 56s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/d672c44357534009a49973725b67f0d2 : SUCCESS in 51m 35s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/6fabbc0b427a46f78473bf4a11794fe0 : SUCCESS in 1h 40m 47s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/b59ffcd7e89e444ebb5fc6ad37555eb7 : SUCCESS in 1h 07m 24s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/b11e98a23e3a47c6af31ce9f2d74fd6d : SUCCESS in 1h 13m 44s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/5808c11f4762492783873a0a07781791 : SUCCESS in 57m 19s","accounts_in_message":[],"_revision_number":5},{"id":"8854f60e885b3b250d45f7f7accc98f5b02e6f5f","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-03-13 09:39:20.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"19302479f35c4f0397ba35e788b3818fc28b3155","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-03-13 10:44:09.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"425990c16c79e41cf5fbbbb3fe1f311f7755adf9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-13 11:09:30.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a22ff9e0450f45aa8153fc57647845e8\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/565efb8fbfff451e9cb88448934f10ff : SUCCESS in 22m 01s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/a0b6bee5c7c54341ac3fcedaed9a0f85 : SUCCESS in 20m 05s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/40ef51f382b7495fb8cd786969c71661 : SUCCESS in 23m 32s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"f0000c889dd1786702951434821a1329738d8838","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2024-03-13 11:55:30.000000000","message":"Patch Set 6: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2024/Arista_Tempest_Testrefs/changes/49/912349/6 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":6},{"id":"4777715a34838f0bad3b343191133f858e73f0bb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-13 12:37:01.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3a4090d03a864587ab5b349cca229f35\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/17c69ed8c4404d1f94ce01f4eefc2e20 : SUCCESS in 1h 32m 48s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/bc95d5da6cf841d485cbcbb757e2e4c1 : SUCCESS in 46m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/39f6f4b4e1b741eea85cf23412130d76 : SUCCESS in 35m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/268ec40c814a424a9f0aefdd2e286d07 : SUCCESS in 10m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/761d8c2582394bb892e7d37401df57ed : SUCCESS in 27m 03s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4d0f461133434c5a9cc5c04abd4388a8 : SUCCESS in 19m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d036670006e0420da22f12b45b07ec79 : SUCCESS in 11m 01s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/0e000c54ebf5463b9770d05ff76dcbe3 : SUCCESS in 31m 46s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/ea29e08a296048318056c0d801ccf128 : SUCCESS in 1h 04m 31s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/ffc07e6082194b5caa82aebbe9449b08 : SUCCESS in 55m 17s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/023f7fb71646405caa5d7fd130ccd34f : SUCCESS in 1h 36m 38s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/3a2bbbfe13314b93ab29f8b86f2e38ae : SUCCESS in 1h 01m 16s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/9b2dd33e9fc8407ba800f6eb593e3a7f : SUCCESS in 1h 31m 57s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/b84da3bae849461a874d58a77e0ce825 : SUCCESS in 1h 17m 56s","accounts_in_message":[],"_revision_number":6},{"id":"3fee9537fba91ec25ee8870d87dc91cebd51ffed","author":{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},"date":"2024-03-13 14:43:09.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"f1d8746a4d8907273c07daec8288acf83ee88d4b","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2024-03-15 08:23:11.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"d04408b361f465fdbb440496fb9624f08241e3b9","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2024-03-15 08:30:40.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"ddfa391e5235af9a93c656995a0c98cca7897fb1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-15 08:32:37.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"c596f2d2e584b99edc97bcbf5c5a33aa1236a700","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-15 10:30:43.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/db04f0e9eafd4cd2a391ca4edb8978d1\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/7d1818ed519f4784bd9ccb0a48d0ea12 : SUCCESS in 1h 50m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/451b30be64e24ee092dd64dcf2ba0963 : SUCCESS in 10m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b0c5dbe1b733431182e44aed695db03d : SUCCESS in 28m 59s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2a5fbc2000c144acabe9ef4fd7602f4e : SUCCESS in 28m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f4f4e5e8070744ccb5eeee5398122588 : SUCCESS in 10m 53s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/83e84195a7854deeb261e02117998ef2 : SUCCESS in 59m 14s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/bf78ee65bc334c7580ca7e41811420d0 : SUCCESS in 1h 33m 59s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/072c0702364c4546ad4398450c53d78c : SUCCESS in 56m 02s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/8f126d51de8149feab10358b00c689ce : SUCCESS in 1h 05m 04s","accounts_in_message":[],"_revision_number":6},{"id":"904e9f701860e283df3c4904256caec4329f01fd","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-15 10:30:43.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"171ecfb94930ed07fd910590e946e1b8390460c5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-15 10:32:51.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8dedaf27c4514dbdbac06e194a18989b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28a40938811b447f9f14985b24e47963 : SUCCESS in 49s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"0fd654f5929fc6a4f3a36d6b9020d6622564128b","revisions":{"65eca37fadf8721a1186664ba2b8abc42d6089aa":{"kind":"REWORK","_number":1,"created":"2024-03-10 17:21:49.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/49/912349/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/49/912349/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/49/912349/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/49/912349/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/49/912349/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/49/912349/1"}}},"commit":{"parents":[{"commit":"c996ab5b51f2386c064c79f818be3a138c6249e9","subject":"[OVN] Add the network type to the ``Logical_Switch`` register","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c996ab5b51f2386c064c79f818be3a138c6249e9"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-03-10 17:20:39.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-03-10 17:21:15.000000000","tz":0},"subject":"WIP \u003d\u003d [OVN] Use the LSP update event to update the LRP","message":"WIP \u003d\u003d [OVN] Use the LSP update event to update the LRP\n\nTo be defined...\n\nbug #2056558\n\nChange-Id: I13b4c804ea6a9f8a89d3796c1cec88ffa1de6ded\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/65eca37fadf8721a1186664ba2b8abc42d6089aa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/65eca37fadf8721a1186664ba2b8abc42d6089aa"}]},"branch":"refs/heads/master"},"0cba7999918f1de21822aaab5bb29daf0859742e":{"kind":"REWORK","_number":2,"created":"2024-03-11 13:21:08.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/49/912349/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/49/912349/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/49/912349/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/49/912349/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/49/912349/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/49/912349/2"}}},"commit":{"parents":[{"commit":"1bd72dc8c0822906694bd7e95fcb8f11f74b0c76","subject":"[OVN] Add the network type to the ``Logical_Switch`` register","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1bd72dc8c0822906694bd7e95fcb8f11f74b0c76"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-03-10 17:20:39.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-03-08 07:16:30.000000000","tz":0},"subject":"[OVN] Use the LSP update event to update the LRP","message":"[OVN] Use the LSP update event to update the LRP\n\nNow the \"Logical_Router_Port\" is updated when a \"Logical_Switch_Port\"\nevent is received. When the event is received, it is first checked that\nthe \"Logical_Switch_Port\" belongs to a router; if that check is\npossitive, the router port update method is called.\n\nCloses-Bug: #2056558\nChange-Id: I13b4c804ea6a9f8a89d3796c1cec88ffa1de6ded\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0cba7999918f1de21822aaab5bb29daf0859742e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0cba7999918f1de21822aaab5bb29daf0859742e"}]},"branch":"refs/heads/master"},"f7eac7bfde1290ba73b0b184b2714fe18d73a9a4":{"kind":"REWORK","_number":3,"created":"2024-03-11 14:56:33.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/49/912349/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/49/912349/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/49/912349/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/49/912349/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/49/912349/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/49/912349/3"}}},"commit":{"parents":[{"commit":"1bd72dc8c0822906694bd7e95fcb8f11f74b0c76","subject":"[OVN] Add the network type to the ``Logical_Switch`` register","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1bd72dc8c0822906694bd7e95fcb8f11f74b0c76"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-03-10 17:20:39.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-03-08 08:52:01.000000000","tz":0},"subject":"[OVN] Use the LSP update event to update the LRP","message":"[OVN] Use the LSP update event to update the LRP\n\nNow the \"Logical_Router_Port\" is updated when a \"Logical_Switch_Port\"\nevent is received. When the event is received, it is first checked that\nthe \"Logical_Switch_Port\" belongs to a router; if that check is\npossitive, the router port update method is called.\n\nCloses-Bug: #2056558\nChange-Id: I13b4c804ea6a9f8a89d3796c1cec88ffa1de6ded\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f7eac7bfde1290ba73b0b184b2714fe18d73a9a4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f7eac7bfde1290ba73b0b184b2714fe18d73a9a4"}]},"branch":"refs/heads/master"},"7b227dad24bd319ed28c6ffdbe3c81f326cc4ae0":{"kind":"REWORK","_number":4,"created":"2024-03-11 15:16:51.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/49/912349/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/49/912349/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/49/912349/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/49/912349/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/49/912349/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/49/912349/4"}}},"commit":{"parents":[{"commit":"1bd72dc8c0822906694bd7e95fcb8f11f74b0c76","subject":"[OVN] Add the network type to the ``Logical_Switch`` register","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1bd72dc8c0822906694bd7e95fcb8f11f74b0c76"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-03-10 17:20:39.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-03-08 09:12:28.000000000","tz":0},"subject":"[OVN] Use the LSP update event to update the LRP","message":"[OVN] Use the LSP update event to update the LRP\n\nNow the \"Logical_Router_Port\" is updated when a \"Logical_Switch_Port\"\nevent is received. When the event is received, it is first checked that\nthe \"Logical_Switch_Port\" belongs to a router; if that check is\npossitive, the router port update method is called.\n\nCloses-Bug: #2056558\nChange-Id: I13b4c804ea6a9f8a89d3796c1cec88ffa1de6ded\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7b227dad24bd319ed28c6ffdbe3c81f326cc4ae0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7b227dad24bd319ed28c6ffdbe3c81f326cc4ae0"}]},"branch":"refs/heads/master"},"f33ebfebf3fc0b69acf8b2259161cb699db55cf8":{"kind":"REWORK","_number":5,"created":"2024-03-11 16:44:39.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/49/912349/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/49/912349/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/49/912349/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/49/912349/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/49/912349/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/49/912349/5"}}},"commit":{"parents":[{"commit":"1bd72dc8c0822906694bd7e95fcb8f11f74b0c76","subject":"[OVN] Add the network type to the ``Logical_Switch`` register","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1bd72dc8c0822906694bd7e95fcb8f11f74b0c76"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-03-10 17:20:39.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-03-08 10:39:47.000000000","tz":0},"subject":"[OVN] Use the LSP update event to update the LRP","message":"[OVN] Use the LSP update event to update the LRP\n\nNow the \"Logical_Router_Port\" is updated when a \"Logical_Switch_Port\"\nevent is received. When the event is received, it is first checked that\nthe \"Logical_Switch_Port\" belongs to a router; if that check is\npossitive, the router port update method is called.\n\nCloses-Bug: #2056558\nChange-Id: I13b4c804ea6a9f8a89d3796c1cec88ffa1de6ded\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f33ebfebf3fc0b69acf8b2259161cb699db55cf8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f33ebfebf3fc0b69acf8b2259161cb699db55cf8"}]},"branch":"refs/heads/master"},"0fd654f5929fc6a4f3a36d6b9020d6622564128b":{"kind":"REWORK","_number":6,"created":"2024-03-13 10:44:09.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/49/912349/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/49/912349/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/49/912349/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/49/912349/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/49/912349/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/49/912349/6"}}},"commit":{"parents":[{"commit":"f82c650c8c1bfb0b4283b8624d8b6f32f1f8d188","subject":"[OVN] Add the network type to the ``Logical_Switch`` register","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f82c650c8c1bfb0b4283b8624d8b6f32f1f8d188"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-03-10 17:20:39.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-03-10 04:39:40.000000000","tz":0},"subject":"[OVN] Use the LSP update event to update the LRP","message":"[OVN] Use the LSP update event to update the LRP\n\nNow the \"Logical_Router_Port\" is updated when a \"Logical_Switch_Port\"\nevent is received. When the event is received, it is first checked that\nthe \"Logical_Switch_Port\" belongs to a router; if that check is\npossitive, the router port update method is called.\n\nCloses-Bug: #2056558\nChange-Id: I13b4c804ea6a9f8a89d3796c1cec88ffa1de6ded\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0fd654f5929fc6a4f3a36d6b9020d6622564128b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0fd654f5929fc6a4f3a36d6b9020d6622564128b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
