)]}'
{"id":"openstack%2Fneutron~915053","triplet_id":"openstack%2Fneutron~master~If7626543f9095196ef1515039f153b16282c27a4","project":"openstack/neutron","branch":"master","topic":"abandon_unmaintained","attention_set":{},"removed_from_attention_set":{"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2024-08-06 00:33:31.000000000","reason":"Change was submitted"},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2024-08-06 00:33:31.000000000","reason":"Change was submitted"},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2024-08-05 13:44:24.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"If7626543f9095196ef1515039f153b16282c27a4","subject":"Skip abandoning of the old patches in unmaintained branches","status":"MERGED","created":"2024-04-04 10:28:46.000000000","updated":"2024-08-06 00:34:40.000000000","submitted":"2024-08-06 00:33:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"915053-abandon_unmaintained","meta_rev_id":"821d41b63fbc924d1725be8cb1a623d16e3751ab","_number":915053,"virtual_id_number":915053,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-08-06 00:33:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"all":[{"value":2,"date":"2024-08-05 22:53:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"value":2,"date":"2024-08-05 13:44:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"all":[{"value":1,"date":"2024-08-05 22:53:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-04-04 13:20:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-04-04 14:27:57.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2024-04-08 08:40:35.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"CC"},{"updated":"2024-05-16 16:17:16.000000000","updated_by":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"reviewer":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"state":"REVIEWER"},{"updated":"2024-05-19 00:16:10.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2024-08-05 22:53:36.000000000","updated_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"REVIEWER"}],"messages":[{"id":"7d54328c02c43df5deec14859e8fc6099d7b4154","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2024-04-04 10:28:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0279a6011ff8068b4b5343cd499f9ee19d4be5db","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-04 13:20:30.000000000","message":"Patch Set 1: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d50743a0a1434eb7b65bd8b35e5f532d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5dd4c49b56b54079925baae6496bac98 : FAILURE in 10m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9396b3547aa84b878e0bb8421f404177 : SUCCESS in 10m 18s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/30d969a9788b40558bcff3c497261e06 : SUCCESS in 57m 57s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/9c6193e989f24504adaad7394f1b703d : SUCCESS in 2h 50m 47s","accounts_in_message":[],"_revision_number":1},{"id":"b443df9295a5e7ec92049131c87685442dcea094","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-04-04 14:27:57.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5a3ac32dd16562e66e2067ebc7f287df62ecace5","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2024-04-04 15:02:37.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0d737d73f1224017c92c2f8916ab3abf07851d4e","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-04-08 08:40:35.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0e5abf103cafac845547b5484e3ce2afd5e5379b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2024-04-26 07:51:21.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"bc9a39a2973f2445e1cd25f7504973250d003a37","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2024-04-26 07:52:18.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b4ffbf61918e5421574023a308d7432dea5d4225","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-26 09:39:20.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3f560766054f46d48645d2c6e7fb050a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5507d7fb1b504b9e814fe6397f48d0ce : SUCCESS in 11m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/04eaa1092b90451cb569591a37750d52 : SUCCESS in 11m 07s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/94bfca505fc54a389616f3061949c24a : SUCCESS in 51m 04s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/bd0994ccd501491e9ccb50a86dd12b73 : SUCCESS in 1h 42m 07s","accounts_in_message":[],"_revision_number":2},{"id":"d950adcafef65984a92fee756de9f6a136f7fa2f","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2024-05-16 16:17:16.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"fb88c87234ca1d4a93b291c88c65a5edb0c9113c","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2024-05-16 16:22:33.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"abcab6b4c76e83b1703710aff08cffd26736b75c","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-05-19 00:16:10.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1d6bff3194fd85c54bb91effb7e35e6529a10fad","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2024-05-20 12:42:03.000000000","message":"Patch Set 2: -Code-Review","accounts_in_message":[],"_revision_number":2},{"id":"10cd7f53e5a6fcf35b705cafb3fd62ef7179d808","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2024-08-04 16:52:01.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f1f0d780c568832cefa237352a261e1dc6142935","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2024-08-04 16:53:19.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"18ad98e24467c9a59e656d4375d58b2f71210de5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-04 18:31:15.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e5d2ae93e48d4f53a03a0a71f22ac70d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d4e69086f2c64fbf9466099c76874fd3 : SUCCESS in 6m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/862a19227853473bbb0b1986229ef8f3 : SUCCESS in 8m 31s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/1fb454564f3140bfba17ca097fa4dc98 : SUCCESS in 38m 02s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/e2883c0635c04ec793ddfc71a659eb10 : SUCCESS in 1h 37m 12s","accounts_in_message":[],"_revision_number":3},{"id":"d925a9616c0f8127e3ef6b8e9b99cbe40430795f","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-08-05 13:44:24.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"84cbaaeb9da593ff4b0afa9f69a07bddd18e8163","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2024-08-05 22:53:36.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"213d86c6f1101f01ff89bbd4641f303c853cc5fe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-05 22:53:58.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"0b719b2eb2c5980c0ff1753c5e9878b95c82beba","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-06 00:33:31.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d6b830f2cb16486aa9b046f948fac7ac\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f91a1adf2a2940629ce2a68bb62a9722 : SUCCESS in 7m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2eec0ab5afba4eddbb6a8c4388cb3bb9 : SUCCESS in 8m 35s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/15b1ca3f3543401eb6bce78a83fed0a9 : SUCCESS in 42m 07s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/2ea1cc446ba44de7a1ef02c042186d0a : SUCCESS in 1h 39m 29s","accounts_in_message":[],"_revision_number":3},{"id":"6a13f5bdc568ca6243f09bb03c9dac88da27db57","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-06 00:33:31.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"821d41b63fbc924d1725be8cb1a623d16e3751ab","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-06 00:34:40.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/191800a78f6c42b6ba59035abd6daf93\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b4f8edf8012544618abaf41dfe1ab85e : SUCCESS in 49s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"5e583d32402b9a8f51a688806817370d83590c42","revisions":{"477b372e3354c3db1a087a17c64bf83f3476228e":{"kind":"REWORK","_number":1,"created":"2024-04-04 10:28:46.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/53/915053/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/53/915053/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/915053/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/915053/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/915053/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/53/915053/1"}}},"commit":{"parents":[{"commit":"2db8620523fef3d78e8a5c5411576e5cd4f395da","subject":"Merge \"[OVN] Create an OVN DB transaction context decorator\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2db8620523fef3d78e8a5c5411576e5cd4f395da"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2024-04-04 10:25:24.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2024-04-04 10:25:24.000000000","tz":120},"subject":"Skip abandoning of the old patches in unmaintained branches","message":"Skip abandoning of the old patches in unmaintained branches\n\nBy default abandon_old_reviews.sh script will now ignore\npatches from any unmaintained branches.\nCleanup of those patches can be done by passing \"--unmaintained\"\nparameter to the script\n\nChange-Id: If7626543f9095196ef1515039f153b16282c27a4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/477b372e3354c3db1a087a17c64bf83f3476228e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/477b372e3354c3db1a087a17c64bf83f3476228e"}]},"branch":"refs/heads/master"},"2452dbab77694445880e5fcdcac8261ab64423ea":{"kind":"REWORK","_number":2,"created":"2024-04-26 07:51:21.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/53/915053/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/53/915053/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/915053/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/915053/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/915053/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/53/915053/2"}}},"commit":{"parents":[{"commit":"2db8620523fef3d78e8a5c5411576e5cd4f395da","subject":"Merge \"[OVN] Create an OVN DB transaction context decorator\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2db8620523fef3d78e8a5c5411576e5cd4f395da"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2024-04-04 10:25:24.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2024-04-26 07:50:56.000000000","tz":120},"subject":"Skip abandoning of the old patches in unmaintained branches","message":"Skip abandoning of the old patches in unmaintained branches\n\nBy default abandon_old_reviews.sh script will now ignore\npatches from any unmaintained branches.\nCleanup of those patches can be done by passing \"--unmaintained\"\nparameter to the script\n\nChange-Id: If7626543f9095196ef1515039f153b16282c27a4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2452dbab77694445880e5fcdcac8261ab64423ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2452dbab77694445880e5fcdcac8261ab64423ea"}]},"branch":"refs/heads/master"},"5e583d32402b9a8f51a688806817370d83590c42":{"kind":"REWORK","_number":3,"created":"2024-08-04 16:53:19.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/53/915053/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/53/915053/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/915053/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/915053/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/915053/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/53/915053/3"}}},"commit":{"parents":[{"commit":"2db8620523fef3d78e8a5c5411576e5cd4f395da","subject":"Merge \"[OVN] Create an OVN DB transaction context decorator\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2db8620523fef3d78e8a5c5411576e5cd4f395da"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2024-04-04 10:25:24.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2024-08-04 16:52:50.000000000","tz":120},"subject":"Skip abandoning of the old patches in unmaintained branches","message":"Skip abandoning of the old patches in unmaintained branches\n\nBy default abandon_old_reviews.sh script will now ignore\npatches from any unmaintained branches.\nCleanup of those patches can be done by passing \"--unmaintained\"\nparameter to the script\n\nChange-Id: If7626543f9095196ef1515039f153b16282c27a4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5e583d32402b9a8f51a688806817370d83590c42"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5e583d32402b9a8f51a688806817370d83590c42"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
