)]}'
{"id":"openstack%2Fneutron~917002","triplet_id":"openstack%2Fneutron~master~I3f4538e1e63fb316425ef7c6ab5c53a0a74f739f","project":"openstack/neutron","branch":"master","topic":"bug/2042647","attention_set":{},"removed_from_attention_set":{"36049":{"account":{"_account_id":36049,"name":"Jang YeHoon","display_name":"Jang YeHoon","email":"jkjk128jkjk@gmail.com","username":"JangYeHoon"},"last_update":"2024-05-17 18:21:47.000000000","reason":"Change was submitted"},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2024-05-02 08:45:14.000000000","reason":"\u003cGERRIT_ACCOUNT_8313\u003e replied on the change","reason_account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}}},"hashtags":[],"change_id":"I3f4538e1e63fb316425ef7c6ab5c53a0a74f739f","subject":"Remove and modify outdated options in DHCP HA documentation","status":"MERGED","created":"2024-04-25 08:21:12.000000000","updated":"2024-05-17 18:23:17.000000000","submitted":"2024-05-17 18:21:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"917002-bug/2042647","meta_rev_id":"4a7ed0d85debc93e58426537c9446f9b9475f163","_number":917002,"virtual_id_number":917002,"owner":{"_account_id":36049,"name":"Jang YeHoon","display_name":"Jang YeHoon","email":"jkjk128jkjk@gmail.com","username":"JangYeHoon"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-05-17 18:21:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"all":[{"value":2,"date":"2024-05-17 16:04:31.000000000","_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"value":2,"date":"2024-05-02 08:45:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"all":[{"value":1,"date":"2024-05-17 16:04:31.000000000","_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-04-25 08:51:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-04-26 09:28:02.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2024-05-17 16:04:31.000000000","updated_by":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"reviewer":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"state":"REVIEWER"}],"messages":[{"id":"491b680f4675eeb7db9785e6c7316e2d6ca8d6a1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36049,"name":"Jang YeHoon","display_name":"Jang YeHoon","email":"jkjk128jkjk@gmail.com","username":"JangYeHoon"},"date":"2024-04-25 08:21:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8b0339001d4af4f665a663f52a7447b50e237e40","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-25 08:51:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6637b548d1db4cf2b97a731d20984950\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/16f7f1c9c6ae48909f5a6692dbce44a4 : SUCCESS in 11m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/827a74cb8fc549beb7b9db626f87b520 : SUCCESS in 9m 47s","accounts_in_message":[],"_revision_number":1},{"id":"4cc9480423403f43e301c976801904299bf8a7ad","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-04-26 09:28:02.000000000","message":"Patch Set 1: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"096e0dcce560f588411d6520a1bfe4b6408d4c2f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36049,"name":"Jang YeHoon","display_name":"Jang YeHoon","email":"jkjk128jkjk@gmail.com","username":"JangYeHoon"},"date":"2024-04-29 04:02:13.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"e59ad4648347be736a79b0a84976846fcefebde0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-29 04:25:19.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1eb569a3dd0744e4888fc4a5980a80c5\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0629e466837e44ab936eee860886dcd9 : SUCCESS in 9m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b055e86b3b5e4bd0ac51cfe12e922be9 : SUCCESS in 10m 59s","accounts_in_message":[],"_revision_number":2},{"id":"a5cfcb8fa08ec5605b629fe4219ce1b610cb20d9","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-05-02 08:45:14.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"44acad3991b80acba4b796658bb2232a32bd4f2d","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2024-05-17 16:04:31.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"dc4a434fa9298c4e9e295d6445c41cbab465da15","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-17 16:05:10.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"e264c3e63860e8f35f6193331cd94f68a9f27290","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-17 18:21:47.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ef98e6ab019a4025a8db21984e041f76\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/91a7c3e5297b47d28d7cc82a3d07fae1 : SUCCESS in 11m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/928e9b2bca104a65bed1595cbddbe21a : SUCCESS in 12m 29s","accounts_in_message":[],"_revision_number":2},{"id":"f0bea2d203b70210363683c016d7eb7b83eb57f8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-17 18:21:47.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"4a7ed0d85debc93e58426537c9446f9b9475f163","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-17 18:23:17.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e9ab78a6ffaf42b19141062f4197278b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ef8fb677ff2452eb5fa17c53677ae10 : SUCCESS in 46s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"5cee4b34faa511a73a88a40a4e8994f053f3cc6c","revisions":{"d74013a0c2f41301900b0b7d95a0f35231356a4e":{"kind":"REWORK","_number":1,"created":"2024-04-25 08:21:12.000000000","uploader":{"_account_id":36049,"name":"Jang YeHoon","display_name":"Jang YeHoon","email":"jkjk128jkjk@gmail.com","username":"JangYeHoon"},"ref":"refs/changes/02/917002/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/02/917002/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/02/917002/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/02/917002/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/02/917002/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/02/917002/1"}}},"commit":{"parents":[{"commit":"076d9ad47f42c18cdb7fc8d9a0a0f3c43a66f945","subject":"Merge \"Enable HA for OVN router flavors\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/076d9ad47f42c18cdb7fc8d9a0a0f3c43a66f945"}]}],"author":{"name":"JangYeHoon","email":"jkjk128jkjk@gmail.com","date":"2024-04-25 08:17:20.000000000","tz":0},"committer":{"name":"JangYeHoon","email":"jkjk128jkjk@gmail.com","date":"2024-04-25 08:21:06.000000000","tz":0},"subject":"Remove and modify outdated options in DHCP HA documentation","message":"Remove and modify outdated options in DHCP HA documentation\n\nA database option has been added to neutron.conf. And removed the old options rabbit_host and rabbit_password.\n\nRemoved outdated options use_neutron and firewall_driver from linuxbridge_conf.ini. And the options using admin_ in the neutron part have been updated to the latest.\n\nCloses-Bug: #2042647\nChange-Id: I3f4538e1e63fb316425ef7c6ab5c53a0a74f739f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d74013a0c2f41301900b0b7d95a0f35231356a4e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d74013a0c2f41301900b0b7d95a0f35231356a4e"}]},"branch":"refs/heads/master"},"5cee4b34faa511a73a88a40a4e8994f053f3cc6c":{"kind":"REWORK","_number":2,"created":"2024-04-29 04:02:13.000000000","uploader":{"_account_id":36049,"name":"Jang YeHoon","display_name":"Jang YeHoon","email":"jkjk128jkjk@gmail.com","username":"JangYeHoon"},"ref":"refs/changes/02/917002/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/02/917002/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/02/917002/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/02/917002/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/02/917002/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/02/917002/2"}}},"commit":{"parents":[{"commit":"076d9ad47f42c18cdb7fc8d9a0a0f3c43a66f945","subject":"Merge \"Enable HA for OVN router flavors\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/076d9ad47f42c18cdb7fc8d9a0a0f3c43a66f945"}]}],"author":{"name":"JangYeHoon","email":"jkjk128jkjk@gmail.com","date":"2024-04-25 08:17:20.000000000","tz":0},"committer":{"name":"JangYeHoon","email":"jkjk128jkjk@gmail.com","date":"2024-04-29 03:58:31.000000000","tz":0},"subject":"Remove and modify outdated options in DHCP HA documentation","message":"Remove and modify outdated options in DHCP HA documentation\n\nA database option has been added to neutron.conf.\nRemoved the old options rabbit_host and rabbit_password in neutron.conf.\n\nRemoved outdated options use_neutron and firewall_driver from linuxbridge_conf.ini.\nThe admin_ option used in the neutron part of linuxbridge_conf.ini has been updated to the latest.\nModify tenant_name to project_name in the neutron part of linuxbridge_conf.ini\n\nCloses-Bug: #2042647\nChange-Id: I3f4538e1e63fb316425ef7c6ab5c53a0a74f739f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5cee4b34faa511a73a88a40a4e8994f053f3cc6c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5cee4b34faa511a73a88a40a4e8994f053f3cc6c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
