)]}'
{"id":"openstack%2Fneutron~917262","triplet_id":"openstack%2Fneutron~master~Ic96b5ce6576c31075602e0f033fd74acbae333ee","project":"openstack/neutron","branch":"master","topic":"bug/2054435","attention_set":{},"removed_from_attention_set":{"9656":{"account":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"last_update":"2024-05-20 11:27:06.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ic96b5ce6576c31075602e0f033fd74acbae333ee","subject":"Rename create_flows_for_ip_address to reflect it accepts (ip, mac)","status":"MERGED","created":"2024-04-27 02:50:41.000000000","updated":"2024-05-20 11:29:17.000000000","submitted":"2024-05-20 11:27:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"917262-bug/2054435","meta_rev_id":"4c77e965b3b700e5c3ccb0d334ea5d9ca5ea31a8","_number":917262,"virtual_id_number":917262,"owner":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2024-05-15 16:27:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-05-20 11:27:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2024-05-20 09:11:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"value":2,"date":"2024-05-16 07:20:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":1,"date":"2024-05-20 09:11:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-04-27 03:15:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-04-27 03:22:59.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-04-27 06:19:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-04-29 20:05:23.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2024-05-16 07:20:07.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2024-05-20 09:11:52.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"d2bfb87bae2d225f6636a01decb61b951dbe0ccb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2024-04-27 02:50:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3714d801f7b1253803e9c12111fd1808659ee1c4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-27 03:15:37.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8e19d46d9dce4550bb2359d5483efbd3\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6882839e586d43ad9275fa21c2506de1 : FAILURE in 18m 42s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/7f9481f7d4304e1f9d776b0449364a62 : FAILURE in 23m 00s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ad8540b8fc56f9060c3f1f2339bcd3f220a0711f","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2024-04-27 03:22:59.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/April-2024/Arista_Tempest_Testrefs/changes/62/917262/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"50aebba240685a67ff913744647252d8bad0ade0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-27 06:19:37.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4dcdefc229d54be09a96f4b7b3d5cece\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/fb09dea8768943e28bf2f6e50106c7d2 : FAILURE in 2h 50m 39s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/4f8f48c0656242b99180fbe69f1e02b8 : TIMED_OUT in 2h 50m 26s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/aaa0a23a99b64814be25b0b3851d51b9 : SUCCESS in 2h 05m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/440e6c3b49614022be1ca8b03708d1f0 : FAILURE in 36m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8e3cc3626ca04463b8dffa47f79a9f83 : SUCCESS in 11m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b2f6d672efc645c79fe7abd6ba855cca : FAILURE in 24m 56s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/bd9f394af7744d3ea5dd117becbca79d : FAILURE in 28m 21s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9085324aacc746ad8833d309ff484f7e : FAILURE in 26m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b4a5e49b8cc240edb8a634a1fc6414f3 : SUCCESS in 10m 42s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/5e71813cf37d499ea820aa1980be3d66 : FAILURE in 25m 50s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/eb5acc2ff2db470e9122c4df94dcfafe : SUCCESS in 48m 58s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/ef77d696c95d4598ab3f8a2ac6e269e8 : FAILURE in 1h 28m 33s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/0a83ca3fe8234447baffc728853a302e : SUCCESS in 57m 33s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/e444c060ba6241f6972516c5cce66ddc : SUCCESS in 1h 04m 26s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/104cf6546cf14fa8a226982e5a41efb8 : TIMED_OUT in 3h 05m 30s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/c01c552d1787426699ab4c060433a3dc : FAILURE in 3h 00m 30s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/d6ce00c6d280425a931ac81788a1cdb7 : SUCCESS in 48m 18s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/ecc373649bc34ebd8286406fd3f44f5a : SUCCESS in 1h 19m 39s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/633a29c360ba401787fc77d0e6094858 : SUCCESS in 1h 24m 27s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/8bc4fd5566484714a989f128263732a5 : SUCCESS in 1h 09m 14s","accounts_in_message":[],"_revision_number":1},{"id":"c7df3402dcbab15f3b6d67d287c0cb9c592dbe62","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2024-04-29 13:27:48.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"415271acec151ab575a185ef9d6bd6ae1589f3e5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2024-05-03 19:22:36.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1, Verified-1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":2},{"id":"2e7d3910abd09131915f14cc6aa92afc932ecb82","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-03 19:41:58.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4c5dc1892aa94b988111fdaa603fc718\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ae32f9a08c574d83873775893fe33aaa : SUCCESS in 18m 29s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/f805bee47922478c9b76ca6a4758fa4a : SUCCESS in 16m 52s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"53561d79e48c5fb4a16dc5bba561a0f69ab48a06","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2024-05-03 19:56:35.000000000","message":"Patch Set 2:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/May-2024/Arista_Tempest_Testrefs/changes/62/917262/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"25cd2b5be825d429ee350245744e66452f617b60","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2024-05-03 20:29:15.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"338f23a7f7528b6134fd6b8e74f79e37483e6d0a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-03 21:17:31.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/69bdb5881fbd4437828357423c192cae\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/666fc67b11914e6d90bad2127137f658 : SUCCESS in 1h 31m 31s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/228e0954057447cb84fb7bc934d50f9e : SUCCESS in 1h 43m 55s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/f0899163bdf544e095097cd457bb48d6 : SUCCESS in 1h 52m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/743c2a18dd1d400680a383bad636f580 : SUCCESS in 46m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b8110e958d064367a49bbd5b5729f8f4 : SUCCESS in 10m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/90545aebdd7f4a97a6d3f071f91ac5dd : SUCCESS in 27m 20s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3d535c42a07e493485a3489d5d820692 : SUCCESS in 25m 32s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/18ebc589780645fc8eedf0f76500950c : SUCCESS in 26m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0aa5bb1358bf458fa6fb6c28fd324b1c : SUCCESS in 9m 21s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/a0433ece5430477a9ebc68fb380cb108 : SUCCESS in 27m 23s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/ddf9d36f4e594c7abc1ac38c71fdb6bc : SUCCESS in 51m 58s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/d5786b7ba2ec4e3b9b935a8c6f87e99c : SUCCESS in 1h 16m 40s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/722770afa88d464ebb62ce900a6a93a8 : SUCCESS in 47m 09s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/466b79e4679941e291cfe9d65f5580a5 : SUCCESS in 1h 07m 08s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/a21ea6a42be8492dbd178254a37540b5 : FAILURE in 1h 08m 03s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/761fa72592114c7c83013d5396e8ea36 : SUCCESS in 1h 19m 17s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/1b83a99780fa4e35bc2937ffb2000a35 : SUCCESS in 50m 30s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/ff540ac3baf54ebe9a087f9b908482fa : SUCCESS in 1h 13m 56s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/91a06dde99a34ca4a9685f28aafc907a : SUCCESS in 1h 04m 39s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/2c7ca060d9984f879b6d5a34f68fe90d : SUCCESS in 1h 19m 30s","accounts_in_message":[],"_revision_number":2},{"id":"d8815eb64c4d00b93090bd5de5d27b7c9c185b72","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2024-05-14 16:48:15.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"69fbb81961fcabb6d883f57a041033a9410f6d19","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-14 17:21:17.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ab2b5e6e537748c28b94185c85095346\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2b19d3079f5a48188347a10bd98db76c : SUCCESS in 18m 41s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/ca1baf220ed44892b89ad008d1e436b8 : SUCCESS in 31m 19s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"691d3bab25d09b322bd20223189f9a578fd81127","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2024-05-14 17:33:36.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/May-2024/Arista_Tempest_Testrefs/changes/62/917262/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"3f7865e7f45ab4e8676c602cbbe177687e7ea0a1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-14 18:45:02.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d8f8619e3e7b4717b505de8896fad238\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/f5eceb1ca2a44b3bb19d6e05a8e3ed40 : SUCCESS in 1h 32m 29s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/e1f40afefef3419d985e32c49066db12 : SUCCESS in 1h 28m 41s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/7debbe74f8004c839a6edd14b02f3007 : SUCCESS in 1h 51m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/411beb01b9c547fca55681ab354ec7a8 : FAILURE in 37m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/25e5421b64e042b5b310fa413e48a026 : TIMED_OUT in 30m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d8a3a22cb3d14b63add421593b173e21 : SUCCESS in 31m 56s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a43d43a205004129b636132a622a15ae : SUCCESS in 25m 02s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0cc87fcf721147219b9ae86a47e70199 : SUCCESS in 36m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/520d94fab20c4ba193e84029d3003cc4 : SUCCESS in 11m 37s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/27f51166610340da932969331c6a1dff : SUCCESS in 26m 28s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/464c4f4c65a64381a105e0ff489cc8ef : SUCCESS in 58m 47s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/4ed106664e794f0e8fa46394bc8f6fc4 : FAILURE in 1h 13m 06s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/8a660e29788d44d4a3c23245f998528d : SUCCESS in 54m 55s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/d1ab0960d9e8472cb2ae154a811cf366 : SUCCESS in 41m 22s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/ce4d4c97258e438babc64dc7508a7635 : SUCCESS in 1h 26m 23s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/12aa4ced894f414a8901f0d53c942aaa : SUCCESS in 1h 17m 17s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/b614470e3a1949219f6cf6c92eec3a55 : SUCCESS in 1h 02m 26s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/c9ee327a2c704d82bbc7d8eb56a3b2af : SUCCESS in 1h 10m 02s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/2474ecdfd2ec480ca2709abdb4fa4ce0 : SUCCESS in 1h 28m 42s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/f7b8a38ebdc846389165187d7f083cdd : SUCCESS in 53m 16s","accounts_in_message":[],"_revision_number":2},{"id":"1c32816625d430de68c1ff5ad8a8a54831a9181a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2024-05-15 15:10:39.000000000","message":"Patch Set 3: Patch Set 2 was rebased\n\nOutdated Votes:\n* Verified+1, Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"41e0efcc810abd28988528016eb309d82cedbb80","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-15 15:38:53.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/153de1bf1fd0494caee635725efc32b9\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ad1ba82158ba432196c2fe8059c3a4ae : SUCCESS in 26m 29s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/4e1fff9b8b104c0a8f9eac56dded45c9 : SUCCESS in 24m 55s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"0ac2f27be4af5f8560fb97bdc2a8aa4a53046f39","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2024-05-15 16:27:30.000000000","message":"Patch Set 3: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/May-2024/Arista_Tempest_Testrefs/changes/62/917262/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"a783bc76f66cc1a13da914d153c5d4ac0749f6dc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-15 17:07:04.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8673304ffbb9482e85222149bc377b78\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/79168c2cbe0d4dd5ae7da6fb6fb6807d : SUCCESS in 1h 52m 35s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/8419e9750ea74a31b4f7d25993d2d312 : SUCCESS in 1h 46m 58s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/ee35ea5490234d88a07d735b95073b31 : SUCCESS in 1h 48m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2180cc3a3a364506aa8a657b5663634d : FAILURE in 42m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/13313e8155214c6db1acf8db5c53d31d : SUCCESS in 10m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/08cd3d306e264728b439138e74813df6 : SUCCESS in 32m 29s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/93f5192bf0bc4b4587bcf247e6e3207c : SUCCESS in 19m 53s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e69179e3a8e040839bd08dcad410ce6c : SUCCESS in 33m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cfabdcb040ba4a45ae0af54e0a858916 : SUCCESS in 11m 01s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/3aa1709a23804dca8bc2a6981c767c10 : SUCCESS in 32m 39s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/9237a6c7fb6e4d97bc0a7df5437a8eef : SUCCESS in 57m 22s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/2e4fbbb05bf447cdbf8550032d966f29 : SUCCESS in 1h 39m 25s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/e648e305a8594aadaee6457634a84fc8 : SUCCESS in 1h 11m 52s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/54289593e023462c83818e593b435967 : SUCCESS in 1h 27m 05s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/ecc2580dac9b41869215c93e4eef3ade : SUCCESS in 1h 32m 03s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/d6053b9a4d924d3cb4659513c1f3b47e : SUCCESS in 1h 41m 09s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/337ff78c90a9413d9c0a77cb4824a687 : SUCCESS in 55m 06s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/501fa8e7c3794548b376c43bf08aefe4 : SUCCESS in 1h 19m 39s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/b3c86669a62c42cb8eae62907c12535c : SUCCESS in 1h 17m 18s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/a11c02dd09f345d197691096b6b015c4 : SUCCESS in 1h 08m 24s","accounts_in_message":[],"_revision_number":3},{"id":"b14e34043b349ddd6b9af4580f846ddc931f64f5","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2024-05-15 20:53:10.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"9c9666459e1e9311abfed5ebe4460f48810842fe","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-15 21:16:51.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5d49a1442eea443ea4d90008e29b2f77\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/26cdff41f67143b0b6706cb74ccd89eb : SUCCESS in 22m 33s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/b664db2a5d124d4eae9cf639297d06ad : SUCCESS in 16m 09s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"80ca2508e9e51dbc41f8a84072bcd2be9ce18a6c","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2024-05-15 21:26:29.000000000","message":"Patch Set 3:\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/May-2024/Arista_Tempest_Testrefs/changes/62/917262/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"6af9817e74c78124fc94ac71dba67ad2555a68a9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-15 23:20:13.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/98cbe738b25c42dbb98fbb2fdb2fd438\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/8fbcea2ae7f349d5bf80e5342e47b003 : SUCCESS in 1h 55m 04s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/ee313ba87c3f4220ac1be24b2c83c743 : SUCCESS in 1h 28m 54s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/2e90f41d04e044778b1dbececb4c08cb : SUCCESS in 2h 24m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4184b1296c54407f9fe8556929dadbac : FAILURE in 27m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4ee58aaffcf14e25b7c540bef02db4e7 : SUCCESS in 10m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/04be8c4ef0194b1a8ecd52332e640db8 : SUCCESS in 26m 47s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4ad06c9a95114a3db1dc9a21a0affb2b : SUCCESS in 25m 49s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ec18a655d1f9434bacf4afc0fbfa7ca7 : SUCCESS in 29m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b3e4dc29c62447b9ebaafa281fe46fe : SUCCESS in 9m 27s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/48f26fdffed64483b7c3550c51bbb60c : SUCCESS in 20m 11s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/713cd1b784dc462d9903765266ac34ee : SUCCESS in 56m 11s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/4c485a3066074f7fadee18542cf2030e : SUCCESS in 1h 39m 47s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/400da81c26364f918e928533df85cdc9 : SUCCESS in 55m 44s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/7c395f3c603d48488d6ec7999e8a8c85 : SUCCESS in 41m 49s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/44d0072fa61949ff8cc2e38cca114089 : SUCCESS in 1h 07m 52s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/8312d9a02d674735b7e1bc4f89943912 : SUCCESS in 1h 12m 47s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/0231d59191c9446c9e8a7b582d4936f9 : SUCCESS in 1h 03m 34s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/e26853abe1ba403bbc027f901074dfed : SUCCESS in 1h 20m 31s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/d77d9e3924a844c8b36c5382d765a597 : SUCCESS in 1h 14m 28s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/372ac6ab11014dea8f929f7f11f17ba0 : SUCCESS in 1h 10m 23s","accounts_in_message":[],"_revision_number":3},{"id":"27daf994151c073f7e20529efc2012e9e0cc8bbd","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2024-05-16 00:15:12.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"3ea85c1a2a78290a9e1aa9ad2280483cac618d19","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2024-05-16 00:48:08.000000000","message":"Patch Set 3:\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/May-2024/Arista_Tempest_Testrefs/changes/62/917262/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"2d411732ce48d44d8f59d5679ffb3bc1c70aad13","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-16 00:52:16.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c73cc5fc6f934b6999e9281b959fdff3\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/99d7c5d8dec844c39c8f4067e9857d37 : SUCCESS in 35m 27s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/a0f00f0769014ecead158fed417e711a : SUCCESS in 33m 09s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"066f60ceb3f639672572c8d2e4a165e214c55da2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-16 02:15:11.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a034d6e8fdcc464cb884f5032e715dfd\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/62860fe7c22e491c8825404304daf819 : SUCCESS in 1h 32m 16s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/c395f304d1e24191a7b588aeedf4a7d0 : SUCCESS in 1h 29m 47s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/949eb49b226e405994f52f30c436516d : SUCCESS in 1h 57m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f794de59ffba4cdbb1fc0f02da38d8e0 : SUCCESS in 30m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d6ff6ee916ba41a3902fed30c42ae1e2 : SUCCESS in 8m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d21d3c231cf644ab8d2e4dcf3eb3c409 : SUCCESS in 28m 58s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5a829c80018f4141bdc35b6ceeaf5b60 : SUCCESS in 28m 24s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/74550dee5adf4bd7859da82090af0af3 : SUCCESS in 31m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/95a48addfab24007a0000076dc79eb04 : SUCCESS in 10m 42s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/a8b98918b03b4b21bf4f642236f9bf44 : SUCCESS in 24m 45s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/73c8060b65564e6bb7563aeb1b755b9a : SUCCESS in 46m 46s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/ad32bfa4f94142d4b30dd016b210e4f3 : SUCCESS in 1h 19m 09s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/8d9bc02a497149819062aae746d7a7bd : SUCCESS in 40m 16s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/c2e57edd0ee44c8f98aa0ee62356d2b7 : SUCCESS in 44m 49s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/6a24be4de2524eba9c884d6d6baec528 : SUCCESS in 1h 07m 04s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/7639885d5ae943929c2c8b485e7d91db : SUCCESS in 1h 02m 59s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/dc26b0b01eb24f84ab14d8309c9d7a69 : SUCCESS in 47m 43s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/f85313cf39924b3ab02243629ad078cb : SUCCESS in 1h 24m 07s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/929525d9978247ecbdcfeae65ab454d6 : SUCCESS in 1h 06m 05s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/014d1c77dbff4df281e095ada04c5832 : SUCCESS in 1h 17m 30s","accounts_in_message":[],"_revision_number":3},{"id":"73b51816e4bc408e22fcef34c8c7e2b98220a232","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-05-16 07:20:07.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"a2e8dafba4ac5fb51024c475aa7c6420aab236c1","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2024-05-20 09:11:52.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"86daafe2c947aafc07a81b0efe384fc3ce89335c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-20 09:13:01.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"17f677f2f01914cee3684c51f8e506ac98aa3aed","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-20 11:27:06.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4415f7efb1af4dfd96eaeef2ada5e84f\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/566c73bae4b246c3b20a39f51f83bfe6 : SUCCESS in 1h 33m 07s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/0f00029a0bda465da5633bdc0c25b9b2 : SUCCESS in 1h 40m 37s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/6ad4d77c9fae4e6c9829149e7a0bde36 : SUCCESS in 1h 37m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f54346f641b94bb9bc1f471e2b0521d8 : SUCCESS in 10m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a24c4be7bcff4e568a180d2e10fec932 : SUCCESS in 27m 32s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4fc9f75febe1488895abd30ca3d585c5 : SUCCESS in 24m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e850a2e55e3645a8821cf6a0c39b57b1 : SUCCESS in 11m 30s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/8f479d1a84184974a7e36991066cd239 : SUCCESS in 45m 17s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/1b5f801929154e909cfa5bc2a57fdbc1 : SUCCESS in 2h 13m 26s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/9531ec36375443a494bc59c7d67415e8 : SUCCESS in 1h 16m 44s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/3446f4cbb91c4cb299df7632027eaa02 : SUCCESS in 56m 29s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/5ca7e53b2520444d9c2624c78b437940 : SUCCESS in 57m 42s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/28b1f3e7123246448e8aec1cc0f4ea46 : SUCCESS in 1h 07m 11s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/c3baa921041f4c3792c08fe8018a03bb : SUCCESS in 1h 13m 27s","accounts_in_message":[],"_revision_number":3},{"id":"60b3770d3f424a0ac243588171a2b49b61138204","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-20 11:27:06.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"4c77e965b3b700e5c3ccb0d334ea5d9ca5ea31a8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-20 11:29:17.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7383eecd07804a6f84c4300e51a9b123\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c87d7403b9a6420093c506b86832bf5b : SUCCESS in 1m 48s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"d040a38f499d7d3d9dea9192a98583c1cc081ecd","revisions":{"b54b64805e6b4467950138a2ee0d53eab77d8f7e":{"kind":"REWORK","_number":1,"created":"2024-04-27 02:50:41.000000000","uploader":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"ref":"refs/changes/62/917262/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/62/917262/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/62/917262/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/62/917262/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/62/917262/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/62/917262/1"}}},"commit":{"parents":[{"commit":"a69bd0dd6040adbd0b6c0383b670c4c6ea79c1ec","subject":"Make it explicit ovs_firewall operates with ip, mac pairs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a69bd0dd6040adbd0b6c0383b670c4c6ea79c1ec"}]}],"author":{"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","date":"2024-04-27 02:48:09.000000000","tz":-240},"committer":{"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","date":"2024-04-27 02:50:29.000000000","tz":-240},"subject":"Rename create_flows_for_ip_address to reflect it accepts (ip, mac)","message":"Rename create_flows_for_ip_address to reflect it accepts (ip, mac)\n\nThe semantics changed since I2e3aa7c400d7bb17cc117b65faaa160b41013dde\nbut the name was not updated.\n\nChange-Id: Ic96b5ce6576c31075602e0f033fd74acbae333ee\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b54b64805e6b4467950138a2ee0d53eab77d8f7e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b54b64805e6b4467950138a2ee0d53eab77d8f7e"}]},"branch":"refs/heads/master"},"f71558c6fe92749ee95b53c54aa220e565198f9d":{"kind":"REWORK","_number":2,"created":"2024-05-03 19:22:36.000000000","uploader":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"ref":"refs/changes/62/917262/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/62/917262/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/62/917262/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/62/917262/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/62/917262/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/62/917262/2"}}},"commit":{"parents":[{"commit":"7f56084c8844be70d4b69f85e7fb18ab06483800","subject":"Merge \"Start trunk plugin RPC via service framework\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7f56084c8844be70d4b69f85e7fb18ab06483800"}]}],"author":{"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","date":"2024-04-27 02:48:09.000000000","tz":-240},"committer":{"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","date":"2024-05-03 19:21:41.000000000","tz":-240},"subject":"Rename create_flows_for_ip_address to reflect it accepts (ip, mac)","message":"Rename create_flows_for_ip_address to reflect it accepts (ip, mac)\n\nThe semantics changed since I2e3aa7c400d7bb17cc117b65faaa160b41013dde\nbut the name was not updated.\n\nThis patch also revealed a number of issues in unit tests where\nmock flow_states were using the old key format (bare ip) instead of the\nnew (ip, mac) tuple. These are also now fixed.\n\nChange-Id: Ic96b5ce6576c31075602e0f033fd74acbae333ee\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f71558c6fe92749ee95b53c54aa220e565198f9d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f71558c6fe92749ee95b53c54aa220e565198f9d"}]},"branch":"refs/heads/master"},"d040a38f499d7d3d9dea9192a98583c1cc081ecd":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2024-05-15 15:10:39.000000000","uploader":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"ref":"refs/changes/62/917262/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/62/917262/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/62/917262/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/62/917262/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/62/917262/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/62/917262/3"}}},"commit":{"parents":[{"commit":"11255ede9715800e2945e0b318c93a5ac9f1e968","subject":"Fix the broken neutron gate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/11255ede9715800e2945e0b318c93a5ac9f1e968"}]}],"author":{"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","date":"2024-04-27 02:48:09.000000000","tz":-240},"committer":{"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","date":"2024-05-15 15:10:39.000000000","tz":0},"subject":"Rename create_flows_for_ip_address to reflect it accepts (ip, mac)","message":"Rename create_flows_for_ip_address to reflect it accepts (ip, mac)\n\nThe semantics changed since I2e3aa7c400d7bb17cc117b65faaa160b41013dde\nbut the name was not updated.\n\nThis patch also revealed a number of issues in unit tests where\nmock flow_states were using the old key format (bare ip) instead of the\nnew (ip, mac) tuple. These are also now fixed.\n\nChange-Id: Ic96b5ce6576c31075602e0f033fd74acbae333ee\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d040a38f499d7d3d9dea9192a98583c1cc081ecd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d040a38f499d7d3d9dea9192a98583c1cc081ecd"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
