)]}'
{"id":"openstack%2Fneutron~920608","triplet_id":"openstack%2Fneutron~stable%2F2024.1~I447990509cdea9830228d3bc92a97062cc57a472","project":"openstack/neutron","branch":"stable/2024.1","topic":"bug/2065701","attention_set":{},"removed_from_attention_set":{"13861":{"account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"last_update":"2024-05-29 10:27:45.000000000","reason":"Change was submitted"},"6773":{"account":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"last_update":"2024-05-29 10:27:45.000000000","reason":"Change was submitted"},"21798":{"account":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"last_update":"2024-05-29 10:27:45.000000000","reason":"Change was submitted"},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2024-05-29 08:20:32.000000000","reason":"\u003cGERRIT_ACCOUNT_11975\u003e replied on the change","reason_account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2024-05-29 08:23:39.000000000","reason":"\u003cGERRIT_ACCOUNT_8313\u003e replied on the change","reason_account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},"5756":{"account":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"last_update":"2024-05-29 10:27:45.000000000","reason":"Change was submitted"},"8655":{"account":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"last_update":"2024-05-29 10:27:45.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I447990509cdea9830228d3bc92a97062cc57a472","subject":"[stable only] Do not fail on missing logical router ports","status":"MERGED","created":"2024-05-28 07:53:15.000000000","updated":"2024-05-29 10:30:15.000000000","submitted":"2024-05-29 10:27:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"920608-bug/2065701","meta_rev_id":"7cba84494b95cb81245e0599dc3c56fa46c5e445","_number":920608,"virtual_id_number":920608,"owner":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-05-29 10:27:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":2,"date":"2024-05-29 08:20:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2024-05-29 08:23:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":1,"date":"2024-05-29 08:23:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"reviewers":{"REVIEWER":[{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-05-28 08:31:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-05-28 08:47:41.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2024-05-28 09:37:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-05-28 15:24:27.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"state":"REVIEWER"},{"updated":"2024-05-28 15:24:27.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"state":"REVIEWER"},{"updated":"2024-05-28 15:24:27.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"state":"REVIEWER"},{"updated":"2024-05-28 15:24:27.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2024-05-28 15:24:27.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"4fad8c7fc1299a7c8512b635e00855ebbbd1af27","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2024-05-28 07:53:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d8a283f85dcd61b3557d92ed31cdb7a097226a68","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-28 08:31:57.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1e9831d1fc2748cd8b509a18508c77a3\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ca2ed25207bb4febb8de02127f8142ab : SUCCESS in 36m 55s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/bef4ca3bc824468fa329858281a10a3c : SUCCESS in 22m 34s\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/0b2b39b4d3064ddf89d45194b581c359 : SUCCESS in 29m 34s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"fa2a8d9706cca228629c7383460d3fc0f0661f0f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-28 09:37:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4974349f53014d04b68bfa7bc8a6aadb\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/504b18376b6d499f91aec26845213fb1 : SUCCESS in 1h 42m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d638c9ab6f5d42cea4854e057b4e2ee1 : SUCCESS in 36m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/80685431cc314eeea7590a5a4417c832 : SUCCESS in 10m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/579846eafe364029baf33518d85345ac : SUCCESS in 24m 33s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d39bb391a41e498387126839f411b71a : SUCCESS in 17m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57ec39b5f2ad46caaa615e360050d7f1 : SUCCESS in 11m 48s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/fe3d53927fa44d90acf9883f9ac6606f : SUCCESS in 57m 08s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/736253ec32264a1dbcc20734383e1c62 : SUCCESS in 55m 23s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/9216da1d5b5c400eb3d91bf4b2036d4e : SUCCESS in 45m 19s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/e730ed2c2b8e4d67adf005a511492bad : SUCCESS in 1h 38m 10s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/99089ef2f01e474184ab9d686107b56c : SUCCESS in 1h 06m 58s","accounts_in_message":[],"_revision_number":1},{"id":"1ab9288869c58f06e9883a2766b95a322740e1cf","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2024-05-29 08:20:32.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"a03fd7ec818e0f1fa76b22c097261e8ded987530","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-05-29 08:23:39.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a10d25390681fc039d9657e9f3ff160458e56eea","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-29 08:24:08.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"fb738b3bd48f765bece07bf33d276fe09926f4d4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-29 10:27:45.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/01ca14b2fb614b6b8de35bdbd16aa3ba\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/61f3b7a091854c2e8f29cf2f285e826c : SUCCESS in 2h 02m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eeebf8d9fb9b47b0b5ab091fdaca8cc3 : SUCCESS in 9m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d464ca4a17c74778b969cfc93c0312c0 : SUCCESS in 26m 27s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1aa291f6c596484ea7eb12457e1adda5 : SUCCESS in 28m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/41cc7f8b2d5a46b1b452a7b9a264cccb : SUCCESS in 8m 30s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/dfd26dc57fdf4604b3b1b11007a2d000 : SUCCESS in 1h 00m 12s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/d756b109723244ccaaa5dc82566b6caf : SUCCESS in 55m 15s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/b03af406ab974c5a85b53c45a03a8f18 : SUCCESS in 1h 13m 35s","accounts_in_message":[],"_revision_number":1},{"id":"e87c9bd8f49afa3913c891ef8e656e465f85dfa4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-29 10:27:45.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"7cba84494b95cb81245e0599dc3c56fa46c5e445","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-29 10:30:15.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bf502af15ac144dea727b8baf8bbf96c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e3d4bc614d04dbf8475fa643dc22eeb : SUCCESS in 1m 04s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5bdd0efb3970a52c60043f166bc728778ac3f395","revisions":{"5bdd0efb3970a52c60043f166bc728778ac3f395":{"kind":"REWORK","_number":1,"created":"2024-05-28 07:53:15.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/08/920608/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/08/920608/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/08/920608/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/08/920608/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/08/920608/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/08/920608/1"}}},"commit":{"parents":[{"commit":"d8208fc51482737a2aeed2a1c5e61737a2808d94","subject":"Return both project_id when validating auto allocate network","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d8208fc51482737a2aeed2a1c5e61737a2808d94"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2024-05-28 07:41:58.000000000","tz":330},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2024-05-28 07:46:39.000000000","tz":330},"subject":"[stable only] Do not fail on missing logical router ports","message":"[stable only] Do not fail on missing logical router ports\n\nset_gateway_mtu runs for all the gateway ports for a network\nand if one of the ports get\u0027s deleted in meanwhile\nwhole transaction fails.\n\nTo handle this we need to add if_exists\u003dTrue to the transaction\nbut for that it needs to be supported in ovsdbapp. It\u0027s fixed\nin ovsdbapp with [1] but would require to bump ovsdbapp\nminimal version in requirements.txt which we normally don\u0027t\ndo for stable branches.\n\nSo using \"update_lrouter_port\" instead as that have the\nrequired option available. Before [2] that was only used\nbut during the switch if_exists part was missed.\n\n[1] https://review.opendev.org/q/I56685478214aae7b6d3a2a3187297ad4eb1869a3\n[2] https://review.opendev.org/c/openstack/neutron/+/762695\n\nCloses-Bug: #2065701\nRelated-Bug: #2060163\nChange-Id: I447990509cdea9830228d3bc92a97062cc57a472\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5bdd0efb3970a52c60043f166bc728778ac3f395"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5bdd0efb3970a52c60043f166bc728778ac3f395"}]},"branch":"refs/heads/stable/2024.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
