)]}'
{"id":"openstack%2Fneutron~921567","triplet_id":"openstack%2Fneutron~stable%2F2024.1~I83ec349e49e1f343da8996cab149d76443120873","project":"openstack/neutron","branch":"stable/2024.1","topic":"sql-profiling-revert","attention_set":{},"removed_from_attention_set":{"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2024-06-14 16:20:59.000000000","reason":"Change was submitted"},"9656":{"account":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"last_update":"2024-06-14 16:20:59.000000000","reason":"Change was submitted"},"34271":{"account":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"last_update":"2024-06-14 16:20:59.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I83ec349e49e1f343da8996cab149d76443120873","subject":"Revert \"Use HasStandardAttributes as parent class for Tags DB model\"","status":"MERGED","created":"2024-06-07 19:26:50.000000000","updated":"2024-06-14 16:22:21.000000000","submitted":"2024-06-14 16:20:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"921567-sql-profiling-revert","cherry_pick_of_change":921566,"cherry_pick_of_patch_set":1,"meta_rev_id":"ed0b163f39286de8decb033a6a32c274cc2793f7","_number":921567,"virtual_id_number":921567,"owner":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-06-14 16:20:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2024-06-12 05:31:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2024-06-14 14:37:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2024-06-14 14:37:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"}],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-06-07 19:49:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-06-07 21:39:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-06-11 18:34:29.000000000","updated_by":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2024-06-11 18:34:29.000000000","updated_by":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"reviewer":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"state":"REVIEWER"},{"updated":"2024-06-12 05:31:17.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2024-06-14 14:37:06.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"}],"messages":[{"id":"150be655d79821a7776c3be6edd094cb462fb236","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2024-06-07 19:26:50.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"7086c1980233b1356dff896ea95f86b57187fdc0","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2024-06-07 19:37:01.000000000","message":"Topic sql-profiling-stable/2024.1 removed","accounts_in_message":[],"_revision_number":1},{"id":"f8c319d960b168a0b88bebe37863d4285c14b989","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2024-06-07 19:37:04.000000000","message":"Topic set to sql-profiling-revert","accounts_in_message":[],"_revision_number":1},{"id":"12bcec02de152424dbb18c3a2409539622ef3d29","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-07 19:49:06.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e37f8896d0af432db74987b9d4189509\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/924a2780152f47e391e4f6ed50b06617 : SUCCESS in 21m 08s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e83bf1591c6d489396a206cf52026600 : SUCCESS in 20m 02s\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/20f5704c0588476c84f692f49187aae5 : SUCCESS in 19m 37s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d413f707ffc88c0a009a8d76ca65af15a74f4dad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-07 21:39:03.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d920ca0e137e422c9bf0debeb2d8f81d\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/bc4a7bd829a74ec8a257cb3763af27ee : SUCCESS in 1h 50m 42s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/df5276d81f5249cc8435f2f6ea3ebf18 : SUCCESS in 1h 53m 09s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/a0c42cfb32bb44379d8210d2cb820c8a : SUCCESS in 1h 51m 02s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/a3fc765b9a9a48e0bab3ec2ca61206ef : FAILURE in 38m 17s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/885ad5cd5ea54dfe919eb9dd26059815 : SUCCESS in 1h 01m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7487a56faee94c0cba8338f4d905bcf5 : SUCCESS in 40m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/69eb186b84cb46ad8458ba21f1a5f26f : POST_FAILURE in 16m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/95359cf393ea42c3ba196ac2cd90f253 : SUCCESS in 28m 44s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f37881fe1bfa45c6af9a267fd8b2a88f : SUCCESS in 27m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/763f32a8ffdd4d7885bec79064404d51 : SUCCESS in 11m 10s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/093e38227f28410abefba34c006301a1 : SUCCESS in 57m 58s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/42980185bbe74f5990d0e5e8210d1e14 : SUCCESS in 1h 07m 31s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/4369a1c704594e71b646637b8c70abe6 : SUCCESS in 57m 53s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/0c9af94dc42340ac8731ecfcbf2efc5a : SUCCESS in 1h 29m 24s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/618ed930194e48d2b314057cd4a07e01 : SUCCESS in 59m 53s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/a521e13f4ee34147951d0431d6f0a9bd : SUCCESS in 1h 00m 42s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/caba7e8f75d145d7b9d0cd8b55ac7c54 : SUCCESS in 1h 37m 23s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/2e546cdbbe9248968ad6c21012197df9 : SUCCESS in 1h 14m 24s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/cc2e178a205c459abace060c846c198a : SUCCESS in 1h 02m 47s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/e45c862f31ef4beb807760ef104f3d7c : SUCCESS in 1h 23m 41s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/236701417e46440bba3b8ae76ba875b1 : SUCCESS in 1h 29m 11s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/893ca8cea69e441087d50bd379acbdae : SUCCESS in 1h 19m 25s","accounts_in_message":[],"_revision_number":1},{"id":"118f7a25653d103a493e53c017e78bd981decfe4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2024-06-11 16:31:04.000000000","message":"Patch Set 2: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"39cf4026af7477b9d76fb9bea2b545130fd4e2d8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-11 16:53:56.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/34b623ffd76e4933bc3dce8c5efe9c70\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6d9b0fb3babc4a599f1e3798b72d94b5 : SUCCESS in 21m 19s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/33372a28404845ebb91f24ad52f9e0a7 : SUCCESS in 20m 11s\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/49d8020339ed4536bb4c9b2d2fe1d880 : SUCCESS in 21m 43s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"13b45593361da9e3d56b8213d88456acd3804f2c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-11 18:33:23.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ec38dca56e4f4414a10b5f0fe475b30c\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/da8f2149652845bc8b9542e0f573ce4e : SUCCESS in 1h 17m 34s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/85edbbefa3644d1c9ff98735a51da88b : SUCCESS in 1h 43m 14s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/09ba8d2a93db4ecab20c4bbcdf795959 : SUCCESS in 1h 16m 23s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/2c5a229413de46e5bccb908fe23c7b3e : SUCCESS in 1h 28m 17s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/5c8e0e522224449e9f07fa22a911272a : SUCCESS in 47m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/69a762e76d154e6086d435878ff21584 : SUCCESS in 43m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c8420c07ca62481496ade7a87f8ed585 : SUCCESS in 9m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ac789dd872344bd4914062c8fea0e37a : SUCCESS in 30m 08s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/31ebed08e2f2499eabe427e3e9f5403c : SUCCESS in 26m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a8b0e8786cc421986397acda4c7cad9 : SUCCESS in 10m 48s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/3ed9bfe4f290470ba9dd1ec3070723bd : SUCCESS in 1h 00m 27s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/49cf4690e12f44fdac105562ff70873f : FAILURE in 52m 00s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/0f6e06484902426495c3b17954f7073f : SUCCESS in 56m 22s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/ba0332be47bf4ffebec1f5bfe99199ee : SUCCESS in 1h 56m 07s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/41b2ed2c85f84b1db9739be81afa1e78 : SUCCESS in 1h 02m 32s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/c935e7b0667e4214a893a8ccdc623b76 : SUCCESS in 1h 00m 59s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/6a52936ec1804606b9b94149ccaf7b3a : SUCCESS in 1h 34m 09s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/9549877318994bb09c50d1de02375bd7 : SUCCESS in 1h 30m 07s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/ecb38c2e63e049bc98c06f58e19fe61f : SUCCESS in 1h 05m 47s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/b271fd773dc9430aa59e9cbad73fa427 : SUCCESS in 1h 42m 13s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/e71ebf261b2141918a3160472bb57338 : SUCCESS in 1h 25m 01s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/d8f0aa5e8d13425a9cdd52c2b2f27f7e : SUCCESS in 1h 16m 05s","accounts_in_message":[],"_revision_number":2},{"id":"4433894013bf8bf8ceb0c86b038663bdff20d781","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-06-12 05:31:17.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"be46bd98ae87ee749f1053eb6d8ee0f5afe0861c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-12 06:13:19.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5081df03214641cdb9776d1ece2378f2\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9c9b59367dec467eb1058289c1680b07 : SUCCESS in 20m 39s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/f28e8ab64ae1402fb5742e388e331b60 : SUCCESS in 40m 01s\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/75346114693848568b1dbb7fe4d6cdfb : SUCCESS in 35m 43s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"521970a91d81d75a19b669ad4947ff7ebd6a929a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-12 07:44:16.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d180ca86f99743f5bec4654642f1d4d5\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/e96e10209ca540f1b67c683d7614f371 : SUCCESS in 1h 27m 54s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/c9b5dcf0941748deb07be10f53a24bbd : SUCCESS in 1h 47m 12s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/f6fe8ccb054e400d9ad39794cf6c640d : SUCCESS in 1h 44m 35s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/052ebaf115614abf9b00b74ce9a23b63 : SUCCESS in 2h 10m 43s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/aed7da25e38d41f8bb3bd1ef98a4e21a : FAILURE in 37m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a45cd7323f2648a880c605785ee746d8 : SUCCESS in 41m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aa14514750c24098a303947ebe8d2fd5 : SUCCESS in 10m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f9afcccf89b34cc29998a8f180be6ca6 : SUCCESS in 28m 34s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2088bf528cf04e8a986b3dabc44efd1b : SUCCESS in 26m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57f2bc75acca45eea1bb3137b4f038b3 : SUCCESS in 11m 11s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/a755943452834ccaa4e7e38ae3184d60 : SUCCESS in 56m 47s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/16f14dbb10d34d29a4ec549525d5d1ae : SUCCESS in 1h 20m 04s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/4ab44b89961a4a5e9a1098c83922220d : FAILURE in 50m 08s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/72486efeeda44f09a1ace9da95180ddb : SUCCESS in 1h 37m 34s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/eda7be95ede44c4fa00927856829d3b9 : SUCCESS in 59m 12s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/8eb1eacc48544506a4c5a47451d3327e : SUCCESS in 1h 02m 18s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/1218640b56864d42bb73f152ce2452a3 : SUCCESS in 1h 27m 32s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/67d9d0cf386a4c049522c19960bb17cc : SUCCESS in 1h 24m 49s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/4a8c5204213d411384c5f503a657528d : SUCCESS in 1h 06m 31s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/22c3455cb342456fb8719166495cb32a : SUCCESS in 1h 12m 50s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/bceccf55e5294dcda0070e37765865cb : SUCCESS in 1h 28m 58s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/f952e72e79504c0884830f9b2a112efe : SUCCESS in 1h 09m 01s","accounts_in_message":[],"_revision_number":2},{"id":"5342dab3830dfb2941e29a347df94df7f169979d","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2024-06-13 01:36:18.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9096b1943540cc503b52de55b1885b2fbb94cb7c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-13 01:59:15.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0b9274360a534253a206910adbdecf7c\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/896fdaac050044f893ae4452ab94267f : SUCCESS in 20m 07s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/ac2504a2d14b425cb8cdd84775ada565 : SUCCESS in 19m 47s\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/8fa9841c3bdd4a3d89eda87a13bee4b3 : SUCCESS in 22m 03s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"354083fc20f392d49101787a3e5202a258dbcd84","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-13 04:29:15.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/23026be78f2c479e92688efa78287aae\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/c422242157254b9c9d42cf0441750aaf : SUCCESS in 1h 49m 54s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/ac0f2dcef0624a78a22cb18a6fdfea57 : SUCCESS in 2h 30m 49s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/5335f5436f474a35a319e19cba4b6257 : SUCCESS in 1h 36m 20s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/bcd93a7291b542f0a1a0c20fcf4df277 : SUCCESS in 2h 02m 34s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/c13722bb1dd24ebdb4c280a42fd2ffd5 : SUCCESS in 46m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/83819bcebc5342c5af396743b73ae606 : SUCCESS in 38m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2f2987ebd6a743c1aceefed03b51a5de : SUCCESS in 7m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/aa2b944069ca4883a62fac623d7fef86 : SUCCESS in 27m 00s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/fd73ba386ac84b79b98b4193ec09e53e : SUCCESS in 19m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f83ed6214644cb3a9d14cd84331eeff : SUCCESS in 10m 04s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/51432640daef492cba3ac33451cf570a : SUCCESS in 55m 14s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/928b3444b3e64e69a2415b4c093f5f9f : SUCCESS in 1h 08m 41s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/8e392ad74947417c8f7c8acd8c57b23e : SUCCESS in 48m 36s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/e1eb3c5f4e724bc2b0c4217c2272eb51 : SUCCESS in 1h 37m 03s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/54f3168658a2429a99970d0c030427ef : SUCCESS in 59m 36s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/31e8b9e69d884199b8fd334830a8ce21 : SUCCESS in 1h 03m 47s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/bf232f5e09434b739db6c9cfe7fea188 : SUCCESS in 1h 20m 47s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/b72db914e9fc45c69c7b40b9338b0e0c : SUCCESS in 1h 21m 07s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/f01cf11218bb4bdd80686cddbf3358b9 : SUCCESS in 1h 02m 01s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/f76446e7546147a8b5a56097a6616179 : SUCCESS in 1h 27m 46s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/759b5a1490b742f6b897ff2e99910818 : SUCCESS in 1h 36m 36s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/91da252c5f3d4dacbea924c34740062c : SUCCESS in 1h 23m 33s","accounts_in_message":[],"_revision_number":2},{"id":"8dcb68b6c3fd02c2db5728d75c3643ea95aa15d5","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-06-14 14:37:06.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"e0326c0daf600f2c05215a157a3cf42ce5b27921","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-14 14:37:39.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"d23b43e85e2ed0b48a0a070cb8f39a1120276af7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-14 16:20:59.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8b28fb7e956b4e32bf14f7a4996ee37d\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/727bd1bac2ca401c8fb2422d4da7c233 : SUCCESS in 1h 14m 56s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/aeb9056abd3747848510392871d30d26 : SUCCESS in 1h 31m 54s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/c7bb68f7955a4779aa1c057da280e9b0 : SUCCESS in 1h 42m 44s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/1be1d919137b419da133496a992367a9 : SUCCESS in 1h 15m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/232b036f130d46b0a1af6cc82dd4eb7b : SUCCESS in 11m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2f5d4475892c473b8a328d434a23e2ba : SUCCESS in 30m 49s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/58350dbb5e7149329e82c45b3b0e209f : SUCCESS in 24m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d298920764d448fa13e8d0231e9de27 : SUCCESS in 9m 53s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/34194656b0d64d89984b42f8788e37bc : SUCCESS in 48m 08s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/f88ff747841e4a4fb95015049c17a843 : SUCCESS in 1h 32m 15s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/7550260b4fdc4b0d8e54347ff341e8d5 : SUCCESS in 1h 16m 37s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/d970c87a774542a1b49a7041bf1e61f8 : SUCCESS in 54m 13s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/48c585154f384154b227630b3d4c029e : SUCCESS in 57m 31s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/c5ef88017c72465c85c46917a24aedd8 : SUCCESS in 1h 00m 50s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/4d260e6042124355b0af8a256e367d2b : SUCCESS in 1h 21m 58s","accounts_in_message":[],"_revision_number":2},{"id":"0e40e930875944a1bf3e946ddf919496ae75a54d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-14 16:20:59.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"ed0b163f39286de8decb033a6a32c274cc2793f7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-14 16:22:21.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1f1a2f7d1d5f46069e1b9dc059ec6fee\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1112fab5e1f34f0cb83f1fb76620cdca : SUCCESS in 1m 02s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"f7e9b2e5b31f9275036e644e13365ad78bb17b2b","revisions":{"2fb624367715a6beae8ffa3944b4b6d322d18c2d":{"kind":"REWORK","_number":1,"created":"2024-06-07 19:26:50.000000000","uploader":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"ref":"refs/changes/67/921567/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/67/921567/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/921567/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/921567/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/921567/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/67/921567/1"}}},"commit":{"parents":[{"commit":"0d8cc09c4a5f652503c4ba9d968959b522aca980","subject":"[stable only] Fix KeyError in set_gateway_mtu","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0d8cc09c4a5f652503c4ba9d968959b522aca980"}]}],"author":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2024-06-07 19:14:00.000000000","tz":0},"committer":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2024-06-07 19:26:50.000000000","tz":0},"subject":"Revert \"Use HasStandardAttributes as parent class for Tags DB model\"","message":"Revert \"Use HasStandardAttributes as parent class for Tags DB model\"\n\nThis reverts commit 85d3fff97e55ba85f72cda4365ad0441c10bd9f6.\n\nReason for revert: Using \"joined\" relation can cause very long query times when dealing with many(hundreds) of tags on a large deployments. The initial reason for this change was to reduce amount of SELECT logs during the testing, but given that this change can have negative side effects on a large deployments(in some corner cases), we will favor the original `subquery` option. See closes-bug link for further details.\n\nChange-Id: I83ec349e49e1f343da8996cab149d76443120873\nCloses-bug: #2068761\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2fb624367715a6beae8ffa3944b4b6d322d18c2d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2fb624367715a6beae8ffa3944b4b6d322d18c2d"}]},"branch":"refs/heads/stable/2024.1"},"f7e9b2e5b31f9275036e644e13365ad78bb17b2b":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2024-06-11 16:31:04.000000000","uploader":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"ref":"refs/changes/67/921567/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/67/921567/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/921567/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/921567/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/921567/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/67/921567/2"}}},"commit":{"parents":[{"commit":"0d8cc09c4a5f652503c4ba9d968959b522aca980","subject":"[stable only] Fix KeyError in set_gateway_mtu","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0d8cc09c4a5f652503c4ba9d968959b522aca980"}]}],"author":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2024-06-07 19:14:00.000000000","tz":0},"committer":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2024-06-11 16:31:04.000000000","tz":0},"subject":"Revert \"Use HasStandardAttributes as parent class for Tags DB model\"","message":"Revert \"Use HasStandardAttributes as parent class for Tags DB model\"\n\nThis reverts commit 85d3fff97e55ba85f72cda4365ad0441c10bd9f6.\n\nReason for revert:\nThe original change was made as a “cheap win” to optimize the number\nof queries the neutron server makes during testing. This did\nimprove the number of queries made but introduced regression in\nreal world deployments where some customers(through automation)\nwould define hundreds of tags per port across a large deployment.\nI am proposing to revert this change in favor of the old “subquery”\nrelation in order to fix this regression. In addition, I filed the\nRelated-Bug #2069061 to investigate using `selectin` as the more\nappropriate long term solution.\n\nChange-Id: I83ec349e49e1f343da8996cab149d76443120873\nCloses-bug: #2068761\nRelated-Bug: #2069061\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f7e9b2e5b31f9275036e644e13365ad78bb17b2b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f7e9b2e5b31f9275036e644e13365ad78bb17b2b"}]},"branch":"refs/heads/stable/2024.1","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
