)]}'
{"id":"openstack%2Fneutron~921839","triplet_id":"openstack%2Fneutron~stable%2F2023.1~I3666cf0509747863ca2a416c8bfc065582573734","project":"openstack/neutron","branch":"stable/2023.1","topic":"bug/2038413","attention_set":{},"removed_from_attention_set":{"34451":{"account":{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},"last_update":"2024-06-14 12:32:59.000000000","reason":"Change was submitted"},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2024-06-13 07:12:43.000000000","reason":"\u003cGERRIT_ACCOUNT_11975\u003e replied on the change","reason_account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2024-06-14 12:32:59.000000000","reason":"Change was submitted"},"25468":{"account":{"_account_id":25468,"name":"Michel Nederlof","email":"michel@nederlof.info","username":"pellucid"},"last_update":"2024-06-14 12:32:59.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I3666cf0509747863ca2a416c8bfc065582573734","subject":"[OVN] Fix virtual parent match for PortBindingUpdateVirtualPortsEvent","status":"MERGED","created":"2024-06-12 08:57:31.000000000","updated":"2024-06-14 12:32:59.000000000","submitted":"2024-06-14 12:32:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"921839-bug/2038413","meta_rev_id":"d061a5c1d879d91fd3d828188c22866e5b467660","_number":921839,"virtual_id_number":921839,"owner":{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":25468,"name":"Michel Nederlof","email":"michel@nederlof.info","username":"pellucid"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-06-14 12:32:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2024-06-14 10:17:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2024-06-13 07:12:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":25468,"name":"Michel Nederlof","email":"michel@nederlof.info","username":"pellucid"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2024-06-14 10:17:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":25468,"name":"Michel Nederlof","email":"michel@nederlof.info","username":"pellucid"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":25468,"name":"Michel Nederlof","email":"michel@nederlof.info","username":"pellucid"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":25468,"name":"Michel Nederlof","email":"michel@nederlof.info","username":"pellucid"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-06-12 08:57:31.000000000","updated_by":{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},"reviewer":{"_account_id":25468,"name":"Michel Nederlof","email":"michel@nederlof.info","username":"pellucid"},"state":"CC"},{"updated":"2024-06-12 09:18:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-06-12 09:19:17.000000000","updated_by":{"_account_id":25468,"name":"Michel Nederlof","email":"michel@nederlof.info","username":"pellucid"},"reviewer":{"_account_id":25468,"name":"Michel Nederlof","email":"michel@nederlof.info","username":"pellucid"},"state":"REVIEWER"},{"updated":"2024-06-12 09:30:40.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2024-06-12 10:48:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-06-12 16:09:52.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2024-06-12 17:45:56.000000000","updated_by":{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},"reviewer":{"_account_id":25468,"name":"Michel Nederlof","email":"michel@nederlof.info","username":"pellucid"},"state":"CC"},{"updated":"2024-06-14 10:17:58.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"}],"messages":[{"id":"fb63ac3ec6f851904d60d17d5bc61c338fa27f89","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},"date":"2024-06-12 08:57:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"07072945c57d7abad4d1ef2a0653e12b5b275cb6","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},"date":"2024-06-12 09:11:33.000000000","message":"Topic set to bug/2038413","accounts_in_message":[],"_revision_number":1},{"id":"a0bbb8cef5e36e2b584e94e6c9e80bba655c51f5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-12 09:18:31.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d57d0a9c372b4b0bbf50674565ad2525\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1fe77bac3a6249dbadd073b4dea522a1 : SUCCESS in 17m 53s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/87d1b4a3b5cf4465a210de5f76e18e56 : SUCCESS in 20m 04s","accounts_in_message":[],"_revision_number":1},{"id":"9c1011ac6f895e81271d28b78c8fc9ded89ee993","author":{"_account_id":25468,"name":"Michel Nederlof","email":"michel@nederlof.info","username":"pellucid"},"date":"2024-06-12 09:19:17.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"77067ee275b0fce65a88e6777a13aec55a872ee1","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2024-06-12 09:30:40.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"47caa9c2482918bea4ec53f669cddee20db0c3e1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-12 10:48:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9056c705b8a346db8a3b30e6db7b8935\n\n- neutron-tempest-plugin-ovn-2023-1 https://zuul.opendev.org/t/openstack/build/5e5c46b3b3084618962353b2809762fa : SUCCESS in 1h 49m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/de2b7dbe1c6648aeac05b3302a1d2f2a : SUCCESS in 30m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0d99af33cb824a7d851ca2fedab18adb : SUCCESS in 12m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1f19108a24db4d6aaab99a27500b712f : SUCCESS in 24m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/627b64486cbb4b5299818bfe7b46dc3e : SUCCESS in 23m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a5d5749062646f7886a353d495de667 : SUCCESS in 10m 27s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/580a4b1d3f3f4c33ba5ff001b1086d28 : SUCCESS in 21m 16s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/19e6fe8de54a44128ca03acc4cb74316 : SUCCESS in 49m 43s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/6f4c23773dc9461095005a5fe2b9342b : SUCCESS in 1h 05m 05s (non-voting)\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/298f28f9ae3f4cc0a777df8dcfcd3af9 : SUCCESS in 57m 55s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/bbd69ab4b3974279a86dd4da5df1f6c2 : SUCCESS in 1h 16m 12s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/65d440297e6140ecb3d512fcb9b71386 : SUCCESS in 1h 07m 04s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/53df17c2cc1f4831836cda337eb45f49 : SUCCESS in 1h 08m 31s","accounts_in_message":[],"_revision_number":1},{"id":"ea71df485c796eb5de19a4cf8c1caf42effcef9d","author":{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},"date":"2024-06-12 13:03:58.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fa07f72470066bc120fbadc1b8645d0c41055a63","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-12 13:24:36.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d424fb6c2c5e4e3bae08908584a6040c\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e5b1cda273604c03b69b0eaf00ee7ee0 : SUCCESS in 19m 39s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/aca69feb50ab4257bca072d8f95b2a70 : SUCCESS in 17m 16s","accounts_in_message":[],"_revision_number":1},{"id":"62abe6eba99aa432ed749e68c5dbe5e8a8e80cf5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-12 14:46:15.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ff16857b4c23426c9cd5ec2fe8f1a056\n\n- neutron-tempest-plugin-ovn-2023-1 https://zuul.opendev.org/t/openstack/build/48d9995452c84dc694fd8bdde5758cc5 : SUCCESS in 1h 41m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5e0977ada68f43fd99043102abb4c720 : SUCCESS in 33m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d07f126713e64a66850bd14a6db9cef9 : SUCCESS in 14m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/da373767a2c345d5bc0f8aed3af7cf5f : SUCCESS in 22m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/518bfb2c552e451ca0ae4a29b77001d3 : SUCCESS in 22m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a850424fa29d4b0e867af4adc3badfd5 : SUCCESS in 10m 21s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/306f7bd72df045548b4ac635f39368f2 : SUCCESS in 19m 56s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/1c51f170b23c48fcb45d16706706e6b0 : SUCCESS in 50m 38s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/5aab197e0bae49b0b2dd763ecf31b2cf : SUCCESS in 1h 08m 33s (non-voting)\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/ab62cf9e26f74241b449091bad7ec03d : SUCCESS in 52m 06s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/55c7d7ba29ba4a55b4b2baded6d2a8a0 : SUCCESS in 1h 13m 20s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/60ccaeef55cb4f9ca047ee2b1c57b19b : SUCCESS in 47m 02s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/e34ef1bf99b54f13b284d7432f5646f4 : SUCCESS in 1h 11m 46s","accounts_in_message":[],"_revision_number":1},{"id":"73ea09b3cfae06e31e35b9c48f7974b9dc0eab41","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-06-12 16:09:52.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"aaf8427ad54a4b14e8efa866410163da65b806ca","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},"date":"2024-06-12 17:27:52.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nCopied Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n* Workflow+1\n","accounts_in_message":[],"_revision_number":2},{"id":"5f91dbd8016441607181e6be842461016e799e97","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-12 17:42:11.000000000","message":"Patch Set 2:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/921836 is needed.","accounts_in_message":[],"_revision_number":2},{"id":"8c02b610021f5d34c938a00ca2d9354938054fde","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-12 17:42:12.000000000","message":"Patch Set 2: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/921836 is needed.","accounts_in_message":[],"_revision_number":2},{"id":"30360526d23b586dd93c0ddca21731743f4a1454","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},"date":"2024-06-12 17:42:13.000000000","message":"Patch Set 3: Patch Set 2 was rebased\n\nCopied Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"b7d76cce770a7c1addd8ecec1430fdbf1f179373","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},"date":"2024-06-12 17:45:56.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":4},{"id":"401b33585a1fd79851a2346fa83ceae570d83a8d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-12 18:16:09.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dad590835796490c8ac5efba70f4914e\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1b96e2c51ead47b8a303378de3da16e9 : SUCCESS in 28m 39s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/71d3212d854745eb98bd1962474b56cd : SUCCESS in 15m 15s","accounts_in_message":[],"_revision_number":4},{"id":"3c5bf404cf0430d87f34da49dde54e56a003f367","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-12 19:30:16.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/21868d2dcc9b4cf281dfb79e4dcd68a6\n\n- neutron-tempest-plugin-ovn-2023-1 https://zuul.opendev.org/t/openstack/build/ea925059612e4797875fcf0a3071d9fb : SUCCESS in 1h 32m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ec075f95626e4d74b9f6086df8a8ad9d : SUCCESS in 36m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ae89c6f828ad4805a564127af7ad8304 : SUCCESS in 11m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3d6db7085f544662884102747b83285b : SUCCESS in 22m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d830082b1b674e44a33097825d5f616c : SUCCESS in 24m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2240dd658e9f4db6b22a9c5ad176c951 : SUCCESS in 10m 10s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/6126ee410908473d9fc64ab9ce6a0f2a : SUCCESS in 20m 54s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/37d6a06d4fbc41c0882cb01fae101db4 : FAILURE in 24m 30s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/4534dfa5bcc647368efa7a611249dd8f : FAILURE in 40m 39s (non-voting)\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/a1cadd8a4cee46d4b69a851eac85d4a2 : SUCCESS in 56m 32s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/31080db6b78a4216b8e9422378233ccc : SUCCESS in 1h 37m 51s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/f7e56aa8748a43829181b33c85a08840 : SUCCESS in 59m 06s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/53d737c2a8694e10b80867ceb6bba460 : SUCCESS in 56m 03s","accounts_in_message":[],"_revision_number":4},{"id":"1b43659ec381513e6f60b129d23a20747bf1d370","author":{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},"date":"2024-06-12 19:48:13.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"13e46edfaf95c38831b2ea82979270934a5e8253","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-12 20:34:22.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4b8f86241b8c4a47a943609c51b4d8ea\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7c6c2f75b19a49178484dbbf33e3d992 : SUCCESS in 17m 47s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/9486204ff2c3420788338c1af000161d : SUCCESS in 44m 30s","accounts_in_message":[],"_revision_number":4},{"id":"123c73d940ddd8e110b778f446ea5586f72f012e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-12 21:12:50.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4f5f9751f45c4bd7a501089373524a22\n\n- neutron-tempest-plugin-ovn-2023-1 https://zuul.opendev.org/t/openstack/build/d6235f5c89024e75b4f636b568ec9fa3 : SUCCESS in 1h 22m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/792ba2ef24b0437eb3c868e9e5ff2e9d : SUCCESS in 33m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cf77dde32d8d48de8e38377748c029d9 : SUCCESS in 13m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/33fd083056cd4741b573f7cb38443295 : SUCCESS in 33m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/21e5042ffe894e55879b8b7c4e607f5d : SUCCESS in 21m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8cb2c543aac64caca3d86b040f3f20b9 : SUCCESS in 10m 33s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/a2cd323f74964bddbc817fb90c941b6d : SUCCESS in 21m 57s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/dd7c413fb50044ba9cc8871b9c26a47a : SUCCESS in 52m 02s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/cc6e3318814d4a67a31aae87f2d47593 : SUCCESS in 1h 04m 41s (non-voting)\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/6d86f5d5330f492d9cb3717d413e60d2 : SUCCESS in 1h 02m 52s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/a0366aca893c418c9f9b3d7375a98d24 : SUCCESS in 1h 19m 01s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/24730e9b65374c98a176e765000fce98 : SUCCESS in 1h 01m 41s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/18aa21846071439e8aa6385af935217a : SUCCESS in 1h 14m 39s","accounts_in_message":[],"_revision_number":4},{"id":"2d4a6b951bcea84e1ae069351dc375b3e0d1479c","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2024-06-13 07:12:43.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"7982e0662cb2298ed89e99101ed2beb780c69acc","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-06-14 10:17:58.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"80df82d75848541a7dd04c6b41b35275925aed73","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-14 10:18:28.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"17eaeb9a23b267ef34fb6a7e68e81cf61f24cff1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-14 12:32:59.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/66eb61787f914139b22ffa9101d52014\n\n- neutron-tempest-plugin-ovn-2023-1 https://zuul.opendev.org/t/openstack/build/161c2d8d3f6444f3836d3f1b7e9b7af6 : SUCCESS in 2h 12m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/17ef7f514acc4623aeaa6bbf2b6b492e : SUCCESS in 12m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/801a3eab8d2a4b2dbc5c75e579ab8b49 : SUCCESS in 32m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b677165606a04b58818d12529face191 : SUCCESS in 27m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce1d656a6005434fbd3512692dd34b89 : SUCCESS in 9m 54s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/45a213a68dc943db90cd2b0c68ed3128 : SUCCESS in 45m 58s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/c499f6eb0445471cb4b3bb5cc16f018a : SUCCESS in 59m 04s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/40a0d794d21b4bc9b0274465cc34a9e5 : SUCCESS in 1h 03m 33s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/eddb8953fa6a466ba16af989e0df8881 : SUCCESS in 1h 07m 55s","accounts_in_message":[],"_revision_number":4},{"id":"d061a5c1d879d91fd3d828188c22866e5b467660","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-14 12:32:59.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"cbcfa692fcaa272f1355c9d2f599f27d5c01ef60","revisions":{"95e1338796b481a7ff664f9a3b839950cdfa9b25":{"kind":"REWORK","_number":1,"created":"2024-06-12 08:57:31.000000000","uploader":{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},"ref":"refs/changes/39/921839/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/39/921839/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/921839/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/921839/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/921839/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/39/921839/1"}}},"commit":{"parents":[{"commit":"43ac7bc13bde0e955a30f032c6f86d9b3594c359","subject":"[OVN] Update lsp host id when virtual parent moves","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/43ac7bc13bde0e955a30f032c6f86d9b3594c359"}]}],"author":{"name":"Michel Nederlof","email":"mnederlof@cloudvps.com","date":"2024-06-10 10:18:46.000000000","tz":120},"committer":{"name":"Fernando Royo","email":"froyo@redhat.com","date":"2024-06-12 08:57:06.000000000","tz":120},"subject":"[OVN] Fix virtual parent match for PortBindingUpdateVirtualPortsEvent","message":"[OVN] Fix virtual parent match for PortBindingUpdateVirtualPortsEvent\n\nAs mentioned in change [1], the condition should be a `is None` as per inline comment.\n\n[1] https://review.opendev.org/c/openstack/neutron/+/896883\n\nRelated-Bug: #2038413\nChange-Id: I3666cf0509747863ca2a416c8bfc065582573734\n(cherry picked from commit 170d99f2d53f77d4c66e505f310fd9d8f3481149)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/95e1338796b481a7ff664f9a3b839950cdfa9b25"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/95e1338796b481a7ff664f9a3b839950cdfa9b25"}]},"branch":"refs/heads/stable/2023.1"},"780a954c79e2274fac45ce951862b1fc1a0634ef":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-06-12 17:27:52.000000000","uploader":{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},"ref":"refs/changes/39/921839/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/39/921839/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/921839/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/921839/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/921839/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/39/921839/2"}}},"commit":{"parents":[{"commit":"c1b4ce8f4f5d834a9c165604bfacc9f4de5329bc","subject":"[OVN] Update lsp host id when virtual parent moves","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c1b4ce8f4f5d834a9c165604bfacc9f4de5329bc"}]}],"author":{"name":"Michel Nederlof","email":"mnederlof@cloudvps.com","date":"2024-06-10 10:18:46.000000000","tz":120},"committer":{"name":"Fernando Royo","email":"froyo@redhat.com","date":"2024-06-12 17:27:52.000000000","tz":0},"subject":"[OVN] Fix virtual parent match for PortBindingUpdateVirtualPortsEvent","message":"[OVN] Fix virtual parent match for PortBindingUpdateVirtualPortsEvent\n\nAs mentioned in change [1], the condition should be a `is None` as per inline comment.\n\n[1] https://review.opendev.org/c/openstack/neutron/+/896883\n\nRelated-Bug: #2038413\nChange-Id: I3666cf0509747863ca2a416c8bfc065582573734\n(cherry picked from commit 170d99f2d53f77d4c66e505f310fd9d8f3481149)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/780a954c79e2274fac45ce951862b1fc1a0634ef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/780a954c79e2274fac45ce951862b1fc1a0634ef"}]},"branch":"refs/heads/stable/2023.1","description":"Rebase"},"6910f8c28b4fcf152cba9548e40f00659efb02f1":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2024-06-12 17:42:13.000000000","uploader":{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},"ref":"refs/changes/39/921839/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/39/921839/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/921839/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/921839/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/921839/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/39/921839/3"}}},"commit":{"parents":[{"commit":"ebd52b0df31ccd68cf32d214ef98a53be118af38","subject":"[OVN] Update lsp host id when virtual parent moves","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ebd52b0df31ccd68cf32d214ef98a53be118af38"}]}],"author":{"name":"Michel Nederlof","email":"mnederlof@cloudvps.com","date":"2024-06-10 10:18:46.000000000","tz":120},"committer":{"name":"Fernando Royo","email":"froyo@redhat.com","date":"2024-06-12 17:42:13.000000000","tz":0},"subject":"[OVN] Fix virtual parent match for PortBindingUpdateVirtualPortsEvent","message":"[OVN] Fix virtual parent match for PortBindingUpdateVirtualPortsEvent\n\nAs mentioned in change [1], the condition should be a `is None` as per inline comment.\n\n[1] https://review.opendev.org/c/openstack/neutron/+/896883\n\nRelated-Bug: #2038413\nChange-Id: I3666cf0509747863ca2a416c8bfc065582573734\n(cherry picked from commit 170d99f2d53f77d4c66e505f310fd9d8f3481149)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6910f8c28b4fcf152cba9548e40f00659efb02f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6910f8c28b4fcf152cba9548e40f00659efb02f1"}]},"branch":"refs/heads/stable/2023.1","description":"Rebase"},"cbcfa692fcaa272f1355c9d2f599f27d5c01ef60":{"kind":"REWORK","_number":4,"created":"2024-06-12 17:45:56.000000000","uploader":{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},"ref":"refs/changes/39/921839/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/39/921839/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/921839/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/921839/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/921839/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/39/921839/4"}}},"commit":{"parents":[{"commit":"ebd52b0df31ccd68cf32d214ef98a53be118af38","subject":"[OVN] Update lsp host id when virtual parent moves","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ebd52b0df31ccd68cf32d214ef98a53be118af38"}]}],"author":{"name":"Michel Nederlof","email":"mnederlof@cloudvps.com","date":"2024-06-10 10:18:46.000000000","tz":120},"committer":{"name":"Fernando Royo","email":"froyo@redhat.com","date":"2024-06-12 17:45:38.000000000","tz":120},"subject":"[OVN] Fix virtual parent match for PortBindingUpdateVirtualPortsEvent","message":"[OVN] Fix virtual parent match for PortBindingUpdateVirtualPortsEvent\n\nAs mentioned in change [1], the condition should be a `is None` as per\ninline comment.\n\n[1] https://review.opendev.org/c/openstack/neutron/+/896883\n\n(NOTE: This patch retrieves a test that was skipped in a previous\nbackport.)\n\nRelated-Bug: #2038413\nChange-Id: I3666cf0509747863ca2a416c8bfc065582573734\n(cherry picked from commit 170d99f2d53f77d4c66e505f310fd9d8f3481149)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/cbcfa692fcaa272f1355c9d2f599f27d5c01ef60"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/cbcfa692fcaa272f1355c9d2f599f27d5c01ef60"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
