)]}'
{"id":"openstack%2Fneutron~921940","triplet_id":"openstack%2Fneutron~stable%2F2024.1~Ia61784174ee610b338f26660b2954330abc131a1","project":"openstack/neutron","branch":"stable/2024.1","topic":"bug/2051351","attention_set":{},"removed_from_attention_set":{"15554":{"account":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"last_update":"2024-06-13 21:28:28.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia61784174ee610b338f26660b2954330abc131a1","subject":"Add a default goto table\u003d94 for openvswitch fw","status":"MERGED","created":"2024-06-13 10:56:01.000000000","updated":"2024-06-13 21:29:37.000000000","submitted":"2024-06-13 21:28:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"921940-bug/2051351","meta_rev_id":"13b7e11e3322821aca569d0f44c05dd707bd9733","_number":921940,"virtual_id_number":921940,"owner":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-06-13 21:28:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":2,"date":"2024-06-13 14:22:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":2,"date":"2024-06-13 13:21:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":1,"date":"2024-06-13 14:22:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-06-13 10:56:01.000000000","updated_by":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"reviewer":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"state":"CC"},{"updated":"2024-06-13 11:32:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-06-13 12:58:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-06-13 13:21:02.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2024-06-13 14:22:11.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"}],"messages":[{"id":"13b103441443ec4a9f71fef5cdef47a647a294a4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"date":"2024-06-13 10:56:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"868d1f3d9b5e0b7d4f35aac634bb448a03b3a524","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-13 11:32:32.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/37760ea5e840444f9d9a6c577d5dd5ad\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d9f830003e6a4655bded262b51b69c40 : SUCCESS in 34m 06s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/aaadf343ff4e4bfc856a902a65abcf30 : SUCCESS in 35m 31s\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/7ff3aa1dfa554302b8c865222baa0f5d : SUCCESS in 22m 39s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"98cc78fef6aa3bb75fe6aea8a31be877fcec0e35","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-13 12:58:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2ed64c522ae941bfb05afe4f837a4bde\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/3d0814a6925043b49bd347d7646da26d : SUCCESS in 1h 44m 55s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/263c3c3e7d0e4f4da63ce83ff16b5275 : SUCCESS in 1h 52m 26s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/28b672f616914c618632469add8d5402 : SUCCESS in 1h 12m 41s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/ca3a59e7667d48688b1144618e27d296 : SUCCESS in 2h 00m 13s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d13d2a1ef8984482bfffc5877c70526d : SUCCESS in 49m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dee4ab52b7974f338c7f026e7eef7f3a : SUCCESS in 38m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1772aa22713e46578862575c08d912fe : SUCCESS in 10m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/20717377c0a7478f973dbe9ef037ecaf : SUCCESS in 24m 42s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/23f62e4d537843a29c248920de799188 : SUCCESS in 23m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d466a2e996f748879754de48495d068f : SUCCESS in 11m 13s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/8d36aa4564314a7bbfadad9fd6bb40da : SUCCESS in 59m 46s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/0d3fbc7299b1404f9247a0bf07b89a61 : SUCCESS in 1h 06m 31s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/a1662c35ff59483da19ff031906c5ddc : SUCCESS in 53m 16s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/efb54858c90c41748ae76eb6b3507654 : SUCCESS in 1h 34m 55s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/03216485be67449fa118681d5629f157 : SUCCESS in 58m 52s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/a6b48e88c1824d07bf5e0103c5e0fdd0 : SUCCESS in 1h 03m 09s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/ddf83fdab63847d7a8e4dda4a61528a1 : SUCCESS in 1h 27m 43s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/03ad798e997b43928b630502ed5ad165 : SUCCESS in 1h 18m 00s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/a06f3a81ef874d96a8168b1996b2d69a : SUCCESS in 53m 36s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/28e750b66020472baeba44ea5a89c6b9 : SUCCESS in 1h 11m 11s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/9eae5e667d1a4d5ea9992d91c37ead8a : SUCCESS in 1h 24m 53s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/d3409d4639a84717b04aeb6257a195e9 : SUCCESS in 1h 17m 20s","accounts_in_message":[],"_revision_number":1},{"id":"1758cd396e4c8195b3a24ce04e838ce7ada97d01","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-06-13 13:21:02.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"42b01b1139502b2c02a128d4b8d9a3cefd890143","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-06-13 14:22:11.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"19ddbca9bf11a51714dc0ed3b7aff11b797065b0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-13 14:22:49.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6db5a4be6051aaa796e6df945c7c9e6bceace696","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-13 17:15:54.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f1b8f2337be046209c77b24192d2e65b\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/439b59b8a3a940a99f93f493445b1125 : SUCCESS in 1h 34m 58s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/1a71486463014fe891b06dacb464fe6c : SUCCESS in 1h 56m 51s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/3eeae67ae2b344338572782ef63ea3c2 : FAILURE in 2h 00m 19s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/ef2ba8f2d65e42b6b5c82c4131af560e : SUCCESS in 1h 49m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e561f5e75e8a4b08a2a74c491527c14d : SUCCESS in 11m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0859ae67ca96427e9754bab87f94062c : SUCCESS in 23m 45s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8ce513863b5f4b14afe15299236fae1f : SUCCESS in 27m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/78fc146dd63847f0907653975ecb03db : SUCCESS in 11m 11s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/43e19c3524e541fab2f3966c50412e66 : SUCCESS in 51m 11s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/25dc8baefc9345c699ed5a08e1f9a418 : SUCCESS in 1h 31m 19s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/e41e01634c9549e083bb0cdec7ce36c2 : SUCCESS in 1h 34m 55s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/dc7d51c632e1464e85cf07f5d7398d3f : SUCCESS in 54m 49s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/31211a84c8ca41cb96b17258d013b6da : SUCCESS in 1h 03m 24s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/87ace6f7c9fb49a7a38aaaae8d98e3d2 : SUCCESS in 59m 09s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/f788757792eb426f909b1b2444185993 : SUCCESS in 1h 06m 51s","accounts_in_message":[],"_revision_number":1},{"id":"ad2f583c9e45933d893a91b02e1a90155521e6eb","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-06-13 17:24:48.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"09b45ef0e3910609c816ab07d42f075ec47b6068","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-13 18:16:37.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/37dcb984da364df9947a4df2d2aef89e\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7a423a8dc81341a5b8d26ec0e0d780bf : SUCCESS in 49m 47s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/1b64ef34815a4a929de89f21ff77179e : SUCCESS in 29m 42s\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/d9fd95a102b743b2b495052d80167c2e : SUCCESS in 28m 14s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9d96a73c7d450eb043fec1901a969bf2ec517a6a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-13 19:31:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eb3af5e36b8846e2b4dd231c20a9ea48\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/8ffaf5db57264c47bb5ac94f6319a8b1 : SUCCESS in 1h 47m 37s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/1418696df4494bbf80342cc9d62d3c82 : SUCCESS in 1h 56m 35s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/ef7c0a268f954332801dca03f609d967 : SUCCESS in 1h 13m 50s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/e9b1526c44e4480bbc9d836a8ebb8131 : SUCCESS in 2h 03m 57s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/f1db9eb15ce74771bb86559150c6adea : SUCCESS in 46m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5157ca1d85a64fba8e46b144eca38036 : SUCCESS in 36m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f227e42b443941eeb3a5b66d05284cf9 : SUCCESS in 11m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/43bd2014b79d4229a43e3b32e2692933 : SUCCESS in 26m 18s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/748da7a32bfa457dbf118bd14f82b50b : SUCCESS in 23m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5edf90d2b8a44c08abc751dee5ae0c1b : SUCCESS in 10m 25s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/e0141800545946d98db955e21396a0f5 : SUCCESS in 1h 09m 35s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/8299297e8b6442ada53d41118fff5c57 : SUCCESS in 1h 07m 21s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/ebd35ce06e854a8f8d5ad762540c381c : SUCCESS in 56m 55s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/df2230025d1b46ce8f5f10f43ab046f7 : SUCCESS in 1h 20m 42s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/99074e9eb6c142ddbcd6e322602bd33d : SUCCESS in 1h 04m 29s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/808cc55e1fb44b998e1cbba719293d96 : SUCCESS in 1h 06m 30s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/5b50e9ef5b9c4114bd2a9f02081c683f : SUCCESS in 1h 11m 08s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/3b18bcbd78124bd3bbfb74c2abf188e0 : SUCCESS in 1h 16m 19s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/1559ea490fad4b7ea01d3828fe0c147b : SUCCESS in 57m 48s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/1358bd5d4ee2471bbd45a85651e675c7 : SUCCESS in 1h 13m 35s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/1846d69dac434adb903f9b2105eceb92 : SUCCESS in 1h 20m 36s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/832ed0185be94eb9a0a53d65eda9688d : SUCCESS in 1h 01m 16s","accounts_in_message":[],"_revision_number":1},{"id":"80b1fefa7d287fbc0eaa9bb83cc67811133b8ab0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-13 19:32:25.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"bf92ceef7130ecaf28a6ea078fd76fa37dc67664","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-13 21:28:28.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/653e99af5dea4e73aec9cebdae8c6b54\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/bc6399e5d4c148b58e29396a05647b15 : SUCCESS in 1h 47m 37s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/d1f51833c6ac4dd08aab91e2068a0fcb : SUCCESS in 1h 33m 00s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/5daf01841d9042799176418efc75dfdd : SUCCESS in 1h 31m 49s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/dd2fdca93cdc4d79b195659d9303f40e : SUCCESS in 1h 15m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b06fd59a309e4266af2e527e9032693b : SUCCESS in 10m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fcd5984dac37455c9fec0cb33e279250 : SUCCESS in 26m 37s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/20518c8d183742c89dd06f2fe7037650 : SUCCESS in 24m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/abafa6aab9d240c8b50cc694f10f73f4 : SUCCESS in 10m 56s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/fc42fc8924d74ff68cff69451a40a1ee : SUCCESS in 52m 30s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/e09bf74d66c444d79f246197256e8b25 : SUCCESS in 1h 50m 06s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/11aa5e48deb4474989e5f2de04eaa1ad : SUCCESS in 1h 30m 03s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/84f14b9d78c54c88ab73817c7a7440e8 : SUCCESS in 1h 04m 46s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/bb7050baea254698ac345ed81afd2ffe : SUCCESS in 1h 02m 45s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/730b2f6448394a9c98bfd6c13d2290b8 : SUCCESS in 1h 06m 38s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/d087ba4aa9264b42a38670acfa609f79 : SUCCESS in 1h 27m 25s","accounts_in_message":[],"_revision_number":1},{"id":"d474870c043b945dd30a7a396fa0eeede75e31ae","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-13 21:28:28.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"13b7e11e3322821aca569d0f44c05dd707bd9733","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-13 21:29:37.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a42d7ae0002a465fa16c2664dba56d2b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6067077e77774149a5cb8b0500e09894 : SUCCESS in 49s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f94f8b63842099904ff88906e8221a366a809f99","revisions":{"f94f8b63842099904ff88906e8221a366a809f99":{"kind":"REWORK","_number":1,"created":"2024-06-13 10:56:01.000000000","uploader":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"ref":"refs/changes/40/921940/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/40/921940/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/40/921940/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/40/921940/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/40/921940/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/40/921940/1"}}},"commit":{"parents":[{"commit":"f0901c75e2a47043101fa77aca6672ecc1213520","subject":"[OVN] Bump revision number after update_virtual_port_host","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f0901c75e2a47043101fa77aca6672ecc1213520"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2022-01-27 09:01:43.000000000","tz":480},"committer":{"name":"Bence Romsics","email":"bence.romsics@gmail.com","date":"2024-06-13 08:57:41.000000000","tz":120},"subject":"Add a default goto table\u003d94 for openvswitch fw","message":"Add a default goto table\u003d94 for openvswitch fw\n\nIf enable explicitly_egress_direct\u003dTrue and set port as\nno security group and port_security\u003dFalse, the ingress\nflood will reappear. The pipleline is:\nIngress\ntable_0 -\u003e table_60 -\u003e NORMAL -\u003e VM\nEgress\ntable_0 -\u003e ... -\u003e table_94 -\u003e output\n\nBecause ingress final action is normal, the br-int will learn the\nsource MAC, but egress final action is output. So VM\u0027s mac will\nnever be learnt by the br-int. Then ingress flood comes again.\n\nThis patch adds a default direct flow to table 94 during the\nopenflow security group init and explicitly_egress_direct\u003dTrue, then\nthe pipleline will be:\nIngress\ntable_0 -\u003e table_60 -\u003e table_94 -\u003e output VM\nEgress\ntable_0 -\u003e ... -\u003e table_94 -\u003e output\n\nAnd this patch adds the flows coming from patch port which will\nmatch local vlan then go to table 94 do the same direct actions.\n\nAbove flood issue will be addressed by these flows.\n\nCloses-Bug: #2051351\nChange-Id: Ia61784174ee610b338f26660b2954330abc131a1\n(cherry picked from commit d6f56c5f96c42e1682f3d1723a65253429778c20)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f94f8b63842099904ff88906e8221a366a809f99"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f94f8b63842099904ff88906e8221a366a809f99"}]},"branch":"refs/heads/stable/2024.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
