)]}'
{"id":"openstack%2Fneutron~922264","triplet_id":"openstack%2Fneutron~master~Ie15c3344161ad521bf10b98303c7bb730351e2d8","project":"openstack/neutron","branch":"master","topic":"metadata-on-ipv6","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2024-08-20 14:26:53.000000000","reason":"\u003cGERRIT_ACCOUNT_16688\u003e replied on the change","reason_account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},"4694":{"account":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"last_update":"2024-08-20 23:16:32.000000000","reason":"Change was submitted"},"21798":{"account":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"last_update":"2024-08-20 23:16:32.000000000","reason":"Change was submitted"},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2024-08-14 18:46:49.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"Ie15c3344161ad521bf10b98303c7bb730351e2d8","subject":"Fix support of IPv6 only networks in OVN metadata agent","status":"MERGED","created":"2024-06-19 00:39:14.000000000","updated":"2024-08-20 23:17:45.000000000","submitted":"2024-08-20 23:16:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"922264-metadata-on-ipv6","meta_rev_id":"c389f71421b684177aa1d996e2b8bdd87f8fc95d","_number":922264,"virtual_id_number":922264,"owner":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2024-08-14 01:38:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-08-20 23:16:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2024-08-20 14:26:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":2,"date":"2024-08-14 18:46:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":1,"date":"2024-08-20 21:34:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":1,"date":"2024-08-14 00:28:58.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":33341,"name":"Maor Blaustein","email":"mblue@redhat.com","username":"blue"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-06-19 01:13:31.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-06-19 01:13:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-06-19 03:02:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-06-19 05:24:01.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2024-06-19 07:11:03.000000000","updated_by":{"_account_id":33341,"name":"Maor Blaustein","email":"mblue@redhat.com","username":"blue"},"reviewer":{"_account_id":33341,"name":"Maor Blaustein","email":"mblue@redhat.com","username":"blue"},"state":"CC"},{"updated":"2024-06-19 13:29:59.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2024-06-20 15:46:56.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2024-08-14 18:46:49.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"}],"messages":[{"id":"eb425e0ec17df237ed6438983468e3c71b0874e8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2024-06-19 00:39:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8ff993288bb31da3ec4159d0cfeb31a1f3d3ca24","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2024-06-19 00:41:10.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"4de7bf295b3909783ba89631f752949d389d3a7f","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2024-06-19 01:13:31.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2024/Arista_Tempest_Testrefs/changes/64/922264/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"b12ed6d5b2c851aa4926f3cef57df38ce105ccdf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-19 01:13:55.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fa1748688c334df29025f0f337f1b254\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9262ae34659e441c8fe550f474781d91 : SUCCESS in 32m 25s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/8881d53b76fa4f0a83a35f1d8b0b6f40 : SUCCESS in 30m 11s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"7bb7ede84320492340218582e8705b4abdfa9be2","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2024-06-19 01:48:07.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2024/Arista_Tempest_Testrefs/changes/64/922264/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"4f93c587c2a74f7cc8af3607a493ca44f5089eb6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-19 03:02:29.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/30b7e48772f44072994e535dae121c61\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/e29f5a96c6104aa88f0c10bb15c26617 : SUCCESS in 2h 19m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7fa68e9a14aa42a28c437243ec71e2f9 : SUCCESS in 34m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a0ac3ccd01c74680b8fed29ef4099da8 : SUCCESS in 8m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/308b2d43ea2e47c5876a1781cf4e0187 : SUCCESS in 19m 06s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/abb3142d749f436983df5fa594789155 : SUCCESS in 24m 33s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1534694e887042ab8e4ef525b1e729e5 : SUCCESS in 25m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bcf1951fdf624d2584e72b556d1e177a : SUCCESS in 9m 55s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/aac75d29c79842c19762ca1e2935bd7b : SUCCESS in 23m 05s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/3baee2af29f24967bbd65c076a35db84 : SUCCESS in 43m 18s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/f449d56b0cc74971a17cf9951f24489a : SUCCESS in 1h 05m 57s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/a9eb1abac73f4b279b9989af5b86559f : SUCCESS in 1h 03m 47s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/3e7c15a1503b405bae10f0dd2112a34c : SUCCESS in 1h 21m 15s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/35630c838e01465f95500799c9f66479 : SUCCESS in 1h 14m 45s","accounts_in_message":[],"_revision_number":2},{"id":"9cbf88213307f17e04c2cec1b7057b9946018a8e","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-06-19 13:29:59.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"0ee2c622604e80e25c311cc648f21512e916edfd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2024-07-11 23:43:00.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"c5c397751ccc8371cc4fdae7038e87cb1590dd5e","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2024-07-11 23:43:32.000000000","message":"Patch Set 3: Workflow-1","accounts_in_message":[],"_revision_number":3},{"id":"38cbb5ea80a9e3d3a67c71e0164093de87adf3d2","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2024-07-12 00:14:35.000000000","message":"Patch Set 3: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/July-2024/Arista_Tempest_Testrefs/changes/64/922264/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"7068f038424928a252d2c89224b19920078000f0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-12 00:15:15.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ac95ff64d02441409538a8ea7fba825b\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0f1b37aa05e54bdcb95a28f4a1a36f68 : FAILURE in 4m 54s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/05d904fa08e54f05939f2ffef718ad2b : SUCCESS in 30m 37s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"2def4b0ae80b63cbee80bf0a789d080865bc44bd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-12 02:10:44.000000000","message":"Patch Set 3: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/fa80b692d3084844902375f250fbf183\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/0be59d8e7598498a9721c5c60c3b1bff : SUCCESS in 2h 20m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/977856d8f9db414a8210e67741180c87 : SUCCESS in 24m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3be4d4be5189447d867d4e08bb9e788b : FAILURE in 3m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/aa4222e0521c457a9ec72b11a9d83ad5 : SUCCESS in 16m 39s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/80d2aa19f7c54efba3d9ba4068bde2aa : SUCCESS in 26m 43s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/efd3ab501f6e4e1cbcd488c7b4172136 : SUCCESS in 26m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8510d159ec854fe89c5227b41531150a : SUCCESS in 10m 17s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/af9be9f9ffb1444dbb7b89ab0127a22a : SUCCESS in 22m 25s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/19fd5e54065f42c0b5d917c710643c5d : SUCCESS in 55m 03s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/400b920440134b558010d35debbbb9d3 : SUCCESS in 58m 10s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/c10dc5b0d34e49cbbef114fb322c29f4 : SUCCESS in 56m 33s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/f801600dbdf8457694ee8321ddb21743 : SUCCESS in 1h 09m 18s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/0aac29142d984d5eb8b65f2fd330fd73 : SUCCESS in 1h 19m 48s","accounts_in_message":[],"_revision_number":3},{"id":"eabb667c987842731fc2bef5340a292693d17640","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2024-07-18 00:09:44.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified+1, Verified-1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":4},{"id":"ec116dfa052af7b49b51bfbf2c9b1406e405c0c4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2024-07-18 00:11:35.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"08a168b5a8e1c9ecdeeffd0ef57c7136fe35c3c2","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2024-07-18 00:13:16.000000000","message":"Patch Set 5: Workflow-1","accounts_in_message":[],"_revision_number":5},{"id":"b266c6b8f45e2ff37ce29ef06955202b4739919f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2024-07-18 00:18:09.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Workflow-1\n","accounts_in_message":[],"_revision_number":6},{"id":"57bdc2bc11cb75526f5dc5e4d11db2a34f5caf77","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2024-07-18 00:21:58.000000000","message":"Patch Set 6: Workflow-1","accounts_in_message":[],"_revision_number":6},{"id":"52c1e9471953af4cc65c37179412f1d8e6bc6f91","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-18 00:36:14.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/676be769715545b6859192cdaa2337e1\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b823ef295c204bceb6c02358e7d1e39c : SUCCESS in 15m 54s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/55a6cb6bcbf34cf98995a33df3151e92 : SUCCESS in 17m 14s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"dc05cab47b83ef17bea12e4960f681426501b057","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-18 03:16:09.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b849656cb800444c9dfd453d7df31654\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/3b446385aa594395b177d439ee5e4f34 : SUCCESS in 2h 47m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ca0ed1ed97f74de4b96f87209d35a831 : SUCCESS in 22m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/802737229f044318867f5b02d7fb9d12 : SUCCESS in 7m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/995ffd690287475bb3d2f73488e7d389 : SUCCESS in 24m 04s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/bdc4ec6734464a37bb3b397f2d382546 : SUCCESS in 21m 17s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d5329054118d475c864c3cbe6a467d47 : SUCCESS in 28m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4afb2073b0a149f0ba3cfe23bfab6456 : SUCCESS in 11m 20s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/0941c1cb1f8d448b953d316629030d36 : SUCCESS in 20m 48s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/0a6ec58ad2704a8c80b04fa696cbc499 : SUCCESS in 44m 11s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/8302a56b7e344eb2be31a5101ca57d0f : SUCCESS in 1h 07m 18s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/02bbd278a8c54a98a395a2482a237e7f : SUCCESS in 40m 47s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/6a98e5de82f44c06a7ce6a22ab45cf34 : SUCCESS in 42m 54s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/e4048b98d7a741baa9f471ac9ca595c3 : SUCCESS in 54m 23s","accounts_in_message":[],"_revision_number":6},{"id":"7151bde4fc752915e47b2788e452ca211d1c8940","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2024-07-27 04:19:19.000000000","message":"Patch Set 4: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/July-2024/Arista_Tempest_Testrefs/changes/64/922264/4 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":4},{"id":"6bd15dbd69bff05985fda6be13d801e3142b176b","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2024-07-27 04:51:36.000000000","message":"Patch Set 5: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/July-2024/Arista_Tempest_Testrefs/changes/64/922264/5 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":5},{"id":"00ecc16354077852ca52420b69f9a6bfd96dfdad","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2024-07-27 05:24:56.000000000","message":"Patch Set 6: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/July-2024/Arista_Tempest_Testrefs/changes/64/922264/6 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":6},{"id":"030faef3c1e2334c53ed349cfb71ad7548a0bd28","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2024-08-13 00:38:07.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"ac17a65475c05ffb33ff0ee3f0e6ef5346c9f16f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2024-08-13 00:41:01.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified+1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":7},{"id":"852620762d6ef5ce76a252f77d550614cbb00650","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-13 01:08:36.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9680c66ab41e4517980a02be84781463\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7ddfbe40f537405d8234dcacc0d0604b : SUCCESS in 23m 45s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/dd5478efcd964f43ac3ade5b6c154ccb : SUCCESS in 25m 12s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"8b4d9e72cef2e8543907ea7c7a80890a202e39a2","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2024-08-13 01:16:14.000000000","message":"Patch Set 7: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/August-2024/Arista_Tempest_Testrefs/changes/64/922264/7 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":7},{"id":"3811a6ea7557c19cb142e0cc271b6cb6160e5363","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-13 02:54:56.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4765f31fcdae47eb8a773ea93dd7a0f7\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/f399fd68c48344b5af0640832eca26ab : SUCCESS in 2h 12m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/573be452b7a249c6b3b96a51441256c1 : SUCCESS in 24m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/29948fe5c6534c07b7999fbb78817891 : SUCCESS in 7m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/61f183645d7c48ef9d211fb4f70f5490 : SUCCESS in 16m 51s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3c74a86197864aabafdbcabb308facc9 : SUCCESS in 14m 48s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/10c911ad36044fa7910fe26f54534da0 : SUCCESS in 18m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b594b17e54f4d16ad7b3f1105036fab : SUCCESS in 10m 26s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/819486aa7a5446079bdb9d5cb35d9cb0 : SUCCESS in 16m 04s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/ea809dad982f415490c7970872eef622 : SUCCESS in 56m 14s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/49d77a26ec18445492d3acd4d4433364 : SUCCESS in 46m 03s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/1014f4bf8ad34a85b2277f0782da0d9e : SUCCESS in 1h 07m 22s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/1140bd92e0bd4c85ae184b33bdd07fd6 : SUCCESS in 1h 05m 48s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/5f6031b9ca2c4ad0b6b0ced321a54168 : SUCCESS in 52m 45s","accounts_in_message":[],"_revision_number":7},{"id":"80c9601cbb0dd4c2fbdf0a01344f73e62289ee2f","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-08-13 19:44:11.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"a684e48e6f62afa12f96e0209a1b2f397efb358e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2024-08-14 00:23:54.000000000","message":"Uploaded patch set 8: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"8aa5b171f68d47957432921ba41216c811412660","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2024-08-14 00:24:52.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"2977398200607236a128cf14069d2c8df871a90e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2024-08-14 00:26:28.000000000","message":"Uploaded patch set 9: Commit message was updated.","accounts_in_message":[],"_revision_number":9},{"id":"0e794be42e83ae4ff6fb2406306123aecd08d1bb","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2024-08-14 00:28:58.000000000","message":"Patch Set 9: Review-Priority+1","accounts_in_message":[],"_revision_number":9},{"id":"d7871f4b3f4f79f6bfc318a51de5469d224cb995","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2024-08-14 00:57:05.000000000","message":"Patch Set 8: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/August-2024/Arista_Tempest_Testrefs/changes/64/922264/8 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":8},{"id":"6534ef556b440f085c91d03b92d3ef4aa2ecb884","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-14 01:06:45.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/802defb0b4f14ecc883045713674db20\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7c939678c9aa42bab7f5f979f0067152 : SUCCESS in 37m 46s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/8b7696419f4a48c58000929cf1e65140 : SUCCESS in 24m 19s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"7c4b2a3f7d1234dbfd0262b4701bf2327cc1d507","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2024-08-14 01:38:11.000000000","message":"Patch Set 9: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/August-2024/Arista_Tempest_Testrefs/changes/64/922264/9 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":9},{"id":"85ff87f0ce76434ed7d36879856ed67ebc8ebf56","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-14 03:07:23.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/585644a10f314790a8f27f8f3ed99ace\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/2fef7c91b5ce492bba6f2f9a1c8903ce : FAILURE in 2h 32m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e1cbc0e459e745829a70bd502c722354 : SUCCESS in 30m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4a390030f32543298bbace622a98a04b : SUCCESS in 12m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e46589e5766645e6a5e3fd581372d588 : SUCCESS in 21m 58s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/28de5c783a8f49dd8038dec075a558c3 : SUCCESS in 21m 53s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e5ee49f3e5264c6e80e3fe9f4e652bf8 : SUCCESS in 26m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/137fbb033e14455db72094bd970ca6a3 : SUCCESS in 8m 22s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/211a114c2e384964a99322bd07c35e8a : SUCCESS in 16m 09s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/b5ef98e1ac4b43e89a4425c33c488c22 : SUCCESS in 52m 02s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/c9afbecf658646f8a6726f0870da72c9 : SUCCESS in 1h 06m 27s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/a31ef440e1184c119431680d68921c7e : SUCCESS in 55m 59s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/9d7566c03ffa4e339d7ab440c10deda2 : SUCCESS in 42m 06s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/868e03c81b2446079bf4d51414122de6 : SUCCESS in 46m 46s","accounts_in_message":[],"_revision_number":9},{"id":"757fcb78265cb6d7fd6f837175c81706e5a6d493","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2024-08-14 12:37:01.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"b5b01288e3c7a4f48a6d799296186af9aec45806","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-14 13:13:23.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/716bc8787a0c4cc0bc8ea1cf261834bf\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/fb371ce8d1d040c9a08b56962fdc9221 : SUCCESS in 34m 29s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/d1a7424d5d6047f0ae2b68575addb277 : SUCCESS in 20m 32s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"2f14b3f3f5eb142cb15dbc76cb6d0425c3e06d5f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-14 14:49:03.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/99ac9b64a98e425e9d768fde2a7a6ccb\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/67432af6e0a9440caf727ed6c4d1f417 : SUCCESS in 2h 10m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2783b7e19fca44889d4ef0a8787156ed : SUCCESS in 26m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2daa7140013b4489be300019706671f0 : SUCCESS in 8m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a38331e3fe08430281d53c5c9ecb18ec : SUCCESS in 18m 25s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/fed2ee7fd5814bbe96cc32f80c7635ce : SUCCESS in 15m 53s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/48e03d198a23424382920e8842d136a1 : SUCCESS in 27m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a2e5c2996c04effac06d29c71076e8f : SUCCESS in 10m 53s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/39bd411e89c247779da8470c51098e33 : SUCCESS in 21m 31s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/cabc9206aaf94b6bb353a9f81c4a9ce8 : SUCCESS in 55m 41s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/39e367dfc013422d84c2c841521cc284 : SUCCESS in 48m 56s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/e5f1d5353f484b8bb52de1442a32f5ea : SUCCESS in 54m 16s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/828e44160c864abbbc9754f5945dd249 : SUCCESS in 59m 42s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/46fb06c34ac74b52a6e3995949957ab0 : SUCCESS in 1h 00m 16s","accounts_in_message":[],"_revision_number":9},{"id":"78b657d3980d67ea0d6fadc8f94bf1264ca48115","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-08-14 18:46:49.000000000","message":"Patch Set 9: Code-Review+2","accounts_in_message":[],"_revision_number":9},{"id":"a5f886f7ae9f8f44899635e2f9e5a4a5ffba6cc3","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-08-20 14:26:53.000000000","message":"Patch Set 9: Code-Review+2","accounts_in_message":[],"_revision_number":9},{"id":"607b1b7f08757013d62801ffb469a3c61a9666a5","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-08-20 21:34:43.000000000","message":"Patch Set 9: Workflow+1","accounts_in_message":[],"_revision_number":9},{"id":"996616132d45d00f48e2d8d2bfaee441a229743c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-20 21:35:11.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"9515051771d895fea25fa40b8e1c1d22e38e16c6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-20 23:16:32.000000000","message":"Patch Set 9: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e8bd523557df47d0ada8058141c7b280\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/e5f52d5fd48b4444bf5588cc49aec396 : SUCCESS in 1h 40m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b7a39d76831e4bb480874e3da360f6c4 : SUCCESS in 9m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0a3fa83b139d4f3cb1916fac7b5d9b82 : SUCCESS in 25m 09s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/fec94fc63814415e90be3f9937f7dc52 : SUCCESS in 20m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5923a12b05c2425cb6fc5a999c2e5223 : SUCCESS in 10m 16s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/2e4e37fefe114be2ac6508b3cd699747 : SUCCESS in 55m 27s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/ada0187751e04570929ac311956b372d : SUCCESS in 1h 12m 34s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/3a1e7676ac74468cb7104d8bef4bdad2 : SUCCESS in 1h 00m 39s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/8be3b32242e54ba6b0ff249599900e65 : SUCCESS in 1h 19m 04s","accounts_in_message":[],"_revision_number":9},{"id":"8c2b97e6e1f61c3cbb12391d68757f71f254b8fe","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-20 23:16:32.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":9},{"id":"c389f71421b684177aa1d996e2b8bdd87f8fc95d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-20 23:17:45.000000000","message":"Patch Set 9:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/32f86298f4c3479bae4c3c8c298a6930\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2772f0a064044691afa592f898832c8a : SUCCESS in 55s","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"f7000f3d57bc59732522c4943d6ff2e9dfcf7d31","revisions":{"c8600816e0cac70197702bc326e33b7ddc1bec51":{"kind":"REWORK","_number":1,"created":"2024-06-19 00:39:14.000000000","uploader":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"ref":"refs/changes/64/922264/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/64/922264/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/64/922264/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/64/922264/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/64/922264/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/64/922264/1"}}},"commit":{"parents":[{"commit":"1431a084402e000e5e0b9e8b1984051f8fa4e99c","subject":"Merge \"Use transient systemd units in Process fixture\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1431a084402e000e5e0b9e8b1984051f8fa4e99c"}]}],"author":{"name":"Miguel Lavalle","email":"mlavalle@redhat.com","date":"2024-06-19 00:36:13.000000000","tz":-300},"committer":{"name":"Miguel Lavalle","email":"mlavalle@redhat.com","date":"2024-06-19 00:36:13.000000000","tz":-300},"subject":"[DNM] Fix support of IPv6 only networks in OVN metadata agent","message":"[DNM] Fix support of IPv6 only networks in OVN metadata agent\n\nFix support of IPv6 only networks in OVN metadata agent\n\nChange-Id: Ie15c3344161ad521bf10b98303c7bb730351e2d8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c8600816e0cac70197702bc326e33b7ddc1bec51"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c8600816e0cac70197702bc326e33b7ddc1bec51"}]},"branch":"refs/heads/master"},"5a14057768fbee58424db1f2cd9ad98a88d085fc":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2024-06-19 00:41:10.000000000","uploader":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"ref":"refs/changes/64/922264/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/64/922264/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/64/922264/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/64/922264/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/64/922264/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/64/922264/2"}}},"commit":{"parents":[{"commit":"1431a084402e000e5e0b9e8b1984051f8fa4e99c","subject":"Merge \"Use transient systemd units in Process fixture\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1431a084402e000e5e0b9e8b1984051f8fa4e99c"}]}],"author":{"name":"Miguel Lavalle","email":"mlavalle@redhat.com","date":"2024-06-19 00:36:13.000000000","tz":-300},"committer":{"name":"Miguel Lavalle","email":"mlavalle@redhat.com","date":"2024-06-19 00:39:47.000000000","tz":-300},"subject":"[DNM] Fix support of IPv6 only networks in OVN metadata agent","message":"[DNM] Fix support of IPv6 only networks in OVN metadata agent\n\nFix support of IPv6 only networks in OVN metadata agent\n\nCloses-Bug: #2069482\n\nChange-Id: Ie15c3344161ad521bf10b98303c7bb730351e2d8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5a14057768fbee58424db1f2cd9ad98a88d085fc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5a14057768fbee58424db1f2cd9ad98a88d085fc"}]},"branch":"refs/heads/master"},"380b84a6b6e82d1a805c9f94d5e1d5ae2891d6db":{"kind":"REWORK","_number":3,"created":"2024-07-11 23:43:00.000000000","uploader":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"ref":"refs/changes/64/922264/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/64/922264/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/64/922264/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/64/922264/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/64/922264/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/64/922264/3"}}},"commit":{"parents":[{"commit":"6450d3066727ad25be1380b31ff335cbafdf7131","subject":"Merge \"[DHCP] Reduce to one single thread the event processing\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6450d3066727ad25be1380b31ff335cbafdf7131"}]}],"author":{"name":"Miguel Lavalle","email":"mlavalle@redhat.com","date":"2024-06-19 00:36:13.000000000","tz":-300},"committer":{"name":"Miguel Lavalle","email":"mlavalle@redhat.com","date":"2024-07-11 23:42:47.000000000","tz":-300},"subject":"[DNM] Fix support of IPv6 only networks in OVN metadata agent","message":"[DNM] Fix support of IPv6 only networks in OVN metadata agent\n\nFix support of IPv6 only networks in OVN metadata agent\n\nCloses-Bug: #2069482\n\nChange-Id: Ie15c3344161ad521bf10b98303c7bb730351e2d8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/380b84a6b6e82d1a805c9f94d5e1d5ae2891d6db"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/380b84a6b6e82d1a805c9f94d5e1d5ae2891d6db"}]},"branch":"refs/heads/master"},"e467f35ade4176a2ee45167ea746b1c3c88af09e":{"kind":"REWORK","_number":4,"created":"2024-07-18 00:09:44.000000000","uploader":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"ref":"refs/changes/64/922264/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/64/922264/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/64/922264/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/64/922264/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/64/922264/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/64/922264/4"}}},"commit":{"parents":[{"commit":"6450d3066727ad25be1380b31ff335cbafdf7131","subject":"Merge \"[DHCP] Reduce to one single thread the event processing\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6450d3066727ad25be1380b31ff335cbafdf7131"}]}],"author":{"name":"Miguel Lavalle","email":"mlavalle@redhat.com","date":"2024-06-19 00:36:13.000000000","tz":-300},"committer":{"name":"Miguel Lavalle","email":"mlavalle@redhat.com","date":"2024-07-18 00:09:15.000000000","tz":-300},"subject":"[DNM] Fix support of IPv6 only networks in OVN metadata agent","message":"[DNM] Fix support of IPv6 only networks in OVN metadata agent\n\nFix support of IPv6 only networks in OVN metadata agent\n\nCloses-Bug: #2069482\n\nChange-Id: Ie15c3344161ad521bf10b98303c7bb730351e2d8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e467f35ade4176a2ee45167ea746b1c3c88af09e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e467f35ade4176a2ee45167ea746b1c3c88af09e"}]},"branch":"refs/heads/master"},"70149a91d414e6b291e67e9db892ba25bfd5df5d":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2024-07-18 00:11:35.000000000","uploader":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"ref":"refs/changes/64/922264/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/64/922264/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/64/922264/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/64/922264/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/64/922264/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/64/922264/5"}}},"commit":{"parents":[{"commit":"5a6b916d956e39b3194011e1d28203a41255bfd7","subject":"Merge \"[FT] Make LR name random\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5a6b916d956e39b3194011e1d28203a41255bfd7"}]}],"author":{"name":"Miguel Lavalle","email":"mlavalle@redhat.com","date":"2024-06-19 00:36:13.000000000","tz":-300},"committer":{"name":"Miguel Lavalle","email":"mlavalle@redhat.com","date":"2024-07-18 00:11:11.000000000","tz":-300},"subject":"[DNM] Fix support of IPv6 only networks in OVN metadata agent","message":"[DNM] Fix support of IPv6 only networks in OVN metadata agent\n\nFix support of IPv6 only networks in OVN metadata agent\n\nCloses-Bug: #2069482\n\nChange-Id: Ie15c3344161ad521bf10b98303c7bb730351e2d8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/70149a91d414e6b291e67e9db892ba25bfd5df5d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/70149a91d414e6b291e67e9db892ba25bfd5df5d"}]},"branch":"refs/heads/master"},"7670f44663ef7aeeef7e47ae7136d6da6cde16a2":{"kind":"REWORK","_number":6,"created":"2024-07-18 00:18:09.000000000","uploader":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"ref":"refs/changes/64/922264/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/64/922264/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/64/922264/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/64/922264/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/64/922264/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/64/922264/6"}}},"commit":{"parents":[{"commit":"5a6b916d956e39b3194011e1d28203a41255bfd7","subject":"Merge \"[FT] Make LR name random\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5a6b916d956e39b3194011e1d28203a41255bfd7"}]}],"author":{"name":"Miguel Lavalle","email":"mlavalle@redhat.com","date":"2024-06-19 00:36:13.000000000","tz":-300},"committer":{"name":"Miguel Lavalle","email":"mlavalle@redhat.com","date":"2024-07-18 00:16:23.000000000","tz":-300},"subject":"[DNM] Fix support of IPv6 only networks in OVN metadata agent","message":"[DNM] Fix support of IPv6 only networks in OVN metadata agent\n\nFix support of IPv6 only networks in OVN metadata agent\n\nCloses-Bug: #2069482\n\nChange-Id: Ie15c3344161ad521bf10b98303c7bb730351e2d8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7670f44663ef7aeeef7e47ae7136d6da6cde16a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7670f44663ef7aeeef7e47ae7136d6da6cde16a2"}]},"branch":"refs/heads/master"},"e30872d5d2db30f59655bc23d0c4d4ca2f4eb58e":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":7,"created":"2024-08-13 00:41:01.000000000","uploader":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"ref":"refs/changes/64/922264/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/64/922264/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/64/922264/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/64/922264/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/64/922264/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/64/922264/7"}}},"commit":{"parents":[{"commit":"37b616a62df8e0a85459285807ac9410b8b67569","subject":"Merge \"Use the RBAC actions field for \"network\" and \"subnet\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/37b616a62df8e0a85459285807ac9410b8b67569"}]}],"author":{"name":"Miguel Lavalle","email":"mlavalle@redhat.com","date":"2024-06-19 00:36:13.000000000","tz":-300},"committer":{"name":"Miguel Lavalle","email":"mlavalle@redhat.com","date":"2024-08-13 00:40:33.000000000","tz":-300},"subject":"Fix support of IPv6 only networks in OVN metadata agent","message":"Fix support of IPv6 only networks in OVN metadata agent\n\nFix support of IPv6 only networks in OVN metadata agent\n\nCloses-Bug: #2069482\n\nChange-Id: Ie15c3344161ad521bf10b98303c7bb730351e2d8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e30872d5d2db30f59655bc23d0c4d4ca2f4eb58e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e30872d5d2db30f59655bc23d0c4d4ca2f4eb58e"}]},"branch":"refs/heads/master"},"85fbbda13ecbac0e8889297f8c821cfa7f60e635":{"kind":"NO_CODE_CHANGE","_number":8,"created":"2024-08-14 00:23:54.000000000","uploader":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"ref":"refs/changes/64/922264/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/64/922264/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/64/922264/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/64/922264/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/64/922264/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/64/922264/8"}}},"commit":{"parents":[{"commit":"37b616a62df8e0a85459285807ac9410b8b67569","subject":"Merge \"Use the RBAC actions field for \"network\" and \"subnet\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/37b616a62df8e0a85459285807ac9410b8b67569"}]}],"author":{"name":"Miguel Lavalle","email":"mlavalle@redhat.com","date":"2024-06-19 00:36:13.000000000","tz":-300},"committer":{"name":"Miguel Lavalle","email":"mlavalle@redhat.com","date":"2024-08-14 00:14:43.000000000","tz":-300},"subject":"Fix support of IPv6 only networks in OVN metadata agent","message":"Fix support of IPv6 only networks in OVN metadata agent\n\nWhen an IPv6 only network is used as the sole network for a VM and\nthere are no other bound ports on the same network in the same chassis,\nthe OVN metadata agent concludes that the associated namespace is not\nneeded and deletes it. As a consequence, the VM cannot access the\nmetadata service. With this change, the namespace is preserved if there\nat least one bound port on the chassis with either IPv4 or IPv6\naddresses.\n\nCloses-Bug: #2069482\n\nChange-Id: Ie15c3344161ad521bf10b98303c7bb730351e2d8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/85fbbda13ecbac0e8889297f8c821cfa7f60e635"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/85fbbda13ecbac0e8889297f8c821cfa7f60e635"}]},"branch":"refs/heads/master"},"f7000f3d57bc59732522c4943d6ff2e9dfcf7d31":{"kind":"NO_CODE_CHANGE","_number":9,"created":"2024-08-14 00:26:28.000000000","uploader":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"ref":"refs/changes/64/922264/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/64/922264/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/64/922264/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/64/922264/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/64/922264/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/64/922264/9"}}},"commit":{"parents":[{"commit":"37b616a62df8e0a85459285807ac9410b8b67569","subject":"Merge \"Use the RBAC actions field for \"network\" and \"subnet\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/37b616a62df8e0a85459285807ac9410b8b67569"}]}],"author":{"name":"Miguel Lavalle","email":"mlavalle@redhat.com","date":"2024-06-19 00:36:13.000000000","tz":-300},"committer":{"name":"Miguel Lavalle","email":"mlavalle@redhat.com","date":"2024-08-14 00:25:36.000000000","tz":-300},"subject":"Fix support of IPv6 only networks in OVN metadata agent","message":"Fix support of IPv6 only networks in OVN metadata agent\n\nWhen an IPv6 only network is used as the sole network for a VM and\nthere are no other bound ports on the same network in the same chassis,\nthe OVN metadata agent concludes that the associated namespace is not\nneeded and deletes it. As a consequence, the VM cannot access the\nmetadata service. With this change, the namespace is preserved if there\nis at least one bound port on the chassis with either IPv4 or IPv6\naddresses.\n\nCloses-Bug: #2069482\n\nChange-Id: Ie15c3344161ad521bf10b98303c7bb730351e2d8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f7000f3d57bc59732522c4943d6ff2e9dfcf7d31"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f7000f3d57bc59732522c4943d6ff2e9dfcf7d31"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
