)]}'
{"id":"openstack%2Fneutron~922613","triplet_id":"openstack%2Fneutron~master~I691cda35117c32e0ca20eca03166f58ba13a9895","project":"openstack/neutron","branch":"master","topic":"bug/1912359","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2024-07-03 09:07:20.000000000","reason":"\u003cGERRIT_ACCOUNT_16688\u003e replied on the change","reason_account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}}},"hashtags":[],"change_id":"I691cda35117c32e0ca20eca03166f58ba13a9895","subject":"DNM \u003d\u003d Test devstack patch 922611","status":"ABANDONED","created":"2024-06-24 11:35:19.000000000","updated":"2024-07-04 06:42:38.000000000","total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"bbcce50584a2ec4fe14f8c9c3449ad848fa3029a","_number":922613,"virtual_id_number":922613,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2024-07-03 08:37:06.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-07-03 09:06:48.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-06-24 12:34:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-06-24 12:35:28.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-06-24 12:40:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ebf442715c6d39f999c3ef30c2c319239f4b9e16","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-06-24 11:35:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"43013bc3f2cb8bf7eb1e79dbb04e7df0e00bdb96","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-06-24 11:35:32.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6470704840469cc7f1d79c3feea726a686627344","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-24 12:34:59.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4cb850f7ac1c4269adff6af0e34cd117\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5a9a7b5ae0f14af2a2f10607fb97a5a1 : SUCCESS in 38m 05s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/64bc50620602492992775082213cded7 : SUCCESS in 20m 46s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e28b8716daa73239a2021c40955cf118ac595f2e","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2024-06-24 12:35:28.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2024/Arista_Tempest_Testrefs/changes/13/922613/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"0778cb2378aea07c7a19f2c8fd58ee687e90667a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-06-24 12:40:45.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"f0ee7273ad4ad6ebf47812a756b13eb2d55735d6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-24 12:40:46.000000000","message":"Patch Set 1: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/922611 is needed.","accounts_in_message":[],"_revision_number":1},{"id":"a987700a6f2bf5b0f912c1a9d94f1b1e16fdc703","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-24 12:40:46.000000000","message":"Patch Set 1:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/922611 is needed.","accounts_in_message":[],"_revision_number":1},{"id":"662ddd950c64a87cdd267d256bc343dcd1ea55ff","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-24 13:19:35.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/763fec6f5b144026bac7b6832dedb21e\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/23c94ff3d51a46eeab993f4b4cbfeed1 : SUCCESS in 36m 26s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/0cba20a0ff70442ca0acde2b838b6f8a : SUCCESS in 16m 55s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b221fdd2e8eed7f63056c7ba97e50a2a6c20340b","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2024-06-24 14:14:33.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2024/Arista_Tempest_Testrefs/changes/13/922613/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"06ce7d72e30ecdc67cb32232b6ea72e8456e7434","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-24 14:45:26.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/69b9e11ece0c423db5ee0541282df412\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/6bc04b3214514c748029894fa8ef4a92 : SUCCESS in 1h 41m 58s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/84a338e5ed0a41cda2d6b4f90a16c093 : SUCCESS in 2h 03m 06s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/4c7b97c6eebb4450939a1b3c236e5a85 : SUCCESS in 1h 39m 15s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/bb8ce8d48b064c699605fb5baf0ad2ce : SUCCESS in 1h 53m 27s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/f4d4a7e316d94c62a5dec31d4cb9c059 : SUCCESS in 33m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0a3447ac6134417ea47cc29d2a4c0625 : SUCCESS in 35m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9938d98e570044fdb702ed3629559328 : SUCCESS in 11m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cc4369e02749460c9b00578ae85f9dde : SUCCESS in 43m 21s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a76ba914ef1540aa9596a71e37f7df01 : SUCCESS in 39m 15s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c9945c3967df411a82481374363de722 : SUCCESS in 27m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b73f10efe9fc4ffaa9348bfdb1582ad9 : SUCCESS in 10m 41s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/f9c07655aa8e49ddaf6e23f7c2fc59d5 : SUCCESS in 29m 38s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/737d2ef1b35b459cac3cb14a757d2f48 : SUCCESS in 56m 46s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/e6386b4c02c84cc2899f1c3def37d2df : SUCCESS in 1h 25m 21s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/ece19da1af334820aef8c249be51a8ac : SUCCESS in 1h 08m 47s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/9b0d5bce5d0e4d079a4522f833e329ce : SUCCESS in 1h 23m 40s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/7de1b5f320d94702beff0a0dadce8168 : SUCCESS in 1h 49m 00s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/e70802ed15e74c659a409c2f53865b46 : SUCCESS in 1h 36m 01s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/53c39e5ecb0d4f4ea13d77360c719632 : SUCCESS in 1h 10m 13s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/bce8a87eb7d04f22889c2cee8d082cdb : SUCCESS in 1h 25m 13s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/721b02cb8b8d4935af8bca125ebcba86 : SUCCESS in 1h 21m 05s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/2464480c5c624b52b404d2a240bcec2c : SUCCESS in 1h 21m 22s","accounts_in_message":[],"_revision_number":2},{"id":"a9d94f4696dae63d125c42eeef298aa4390b8159","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-06-25 05:49:16.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"97a55402a6c392ecebddb52ffac6f9ae7641d726","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-06-27 06:19:35.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"06fdb5c351949f088963c6e93adc23a998cff82c","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-27 08:50:13.000000000","message":"Patch Set 2:\n\nBuild failed (experimental pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/34e624a419e5434b9e666acb1e7f9f28\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/f09bf8a5e4004f8ab2649a96c7e0cc33 : SUCCESS in 2h 12m 06s\n- neutron-tempest-plugin-openvswitch-distributed-dhcp https://zuul.opendev.org/t/openstack/build/5b5d9cb744d948cb901017f274f7bef7 : SUCCESS in 1h 44m 06s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-distributed-dhcp https://zuul.opendev.org/t/openstack/build/57bcf87d6dc64ce990707f2ba58e75af : FAILURE in 2h 27m 31s\n- neutron-tempest-plugin-api-ovs-wsgi https://zuul.opendev.org/t/openstack/build/beceb07f272f44d0ab2995d4b20ecf72 : FAILURE in 1h 49m 49s (non-voting)\n- neutron-tempest-plugin-api-ovn-wsgi https://zuul.opendev.org/t/openstack/build/69547dee8130414eafd83c8be736e622 : FAILURE in 18m 16s (non-voting)\n- neutron-functional-with-uwsgi-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/43c01c75a96f4c988aecd8ff0ceeba54 : SUCCESS in 57m 43s\n- neutron-fullstack-with-uwsgi-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/274611bb2bd0477faa746772dadbf0fe : SUCCESS in 1h 19m 21s\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/c292c9599e8141f99fed7d2474717e25 : SUCCESS in 1h 19m 40s\n- neutron-ovn-tempest-ovs-master https://zuul.opendev.org/t/openstack/build/a3cbc9a87d0a4be9b628fe3961bb6a55 : SUCCESS in 1h 06m 56s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/1a8712408c374a4d98c38526fe9e4868 : SUCCESS in 1h 10m 10s\n- neutron-ovs-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/5d345b5f0bf4446cb378ad3fe1f9ca45 : SUCCESS in 1h 11m 02s\n- neutron-ovn-tempest-with-uwsgi-loki https://zuul.opendev.org/t/openstack/build/da5e9964c2204a40822aa27dba4f9e4b : SUCCESS in 1h 15m 05s\n- openstack-tox-py311-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/39cdc0460322459ebd5642bc012a1a6e : SUCCESS in 20m 56s\n- openstack-tox-py311-with-oslo-master https://zuul.opendev.org/t/openstack/build/1cf864e52f144aef8d3020c507b9fd33 : SUCCESS in 23m 39s\n- openstack-tox-py311-with-ovsdbapp-master https://zuul.opendev.org/t/openstack/build/237ae7655e954ed09183eb498c502e40 : SUCCESS in 20m 09s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/098d2e8b18de40aa9f094858cf8a3b76 : SUCCESS in 47m 21s\n- neutron-functional-with-uwsgi-fips https://zuul.opendev.org/t/openstack/build/74349f4da69e467eae687790d013fa66 : SUCCESS in 56m 31s\n- neutron-functional-with-pyroute2-master https://zuul.opendev.org/t/openstack/build/80b51a2101b542968aab62ac985983b9 : FAILURE in 39m 13s\n- neutron-functional-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/011d3ce0cbcf4d23b54b47857d38d620 : SUCCESS in 49m 21s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/3b132a607ce646e99aa634a79d08f22a : SUCCESS in 1h 36m 54s\n- neutron-fullstack-with-uwsgi-fips https://zuul.opendev.org/t/openstack/build/c58e917761524dfd8588384405e77f18 : SUCCESS in 1h 18m 49s\n- neutron-fullstack-with-pyroute2-master https://zuul.opendev.org/t/openstack/build/0d43531f24ed4d56adee054377a1e194 : SUCCESS in 1h 39m 03s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/2494603bf480401aa9128da2ff49bbb8 : SUCCESS in 56m 41s\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/e074ab8e70d040eb9543406525259836 : SUCCESS in 1h 02m 57s\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/0d51a62890714f158c995ada99ccc959 : SUCCESS in 2h 04m 10s\n- neutron-ovs-tempest-with-os-ken-master https://zuul.opendev.org/t/openstack/build/98ae4f404233466d9dc3c4aac68e1964 : SUCCESS in 1h 04m 41s\n- neutron-ovn-tempest-postgres-full https://zuul.opendev.org/t/openstack/build/2875938a3f7d41558a2cae0bac014405 : SUCCESS in 51m 12s\n- neutron-ovn-tempest-mariadb-full https://zuul.opendev.org/t/openstack/build/4e5ab1d93c754169ab4fe164af6288db : SUCCESS in 1h 04m 06s\n- neutron-ovn-tempest-ipv6-only-ovs-master https://zuul.opendev.org/t/openstack/build/d85f8f86992f438eb31a11432bb81968 : SUCCESS in 1h 13m 33s\n- neutron-ovn-tempest-ovs-master-centos-9-stream https://zuul.opendev.org/t/openstack/build/5d35e2080cc54045b1c367fdac2b8ba1 : SUCCESS in 1h 12m 17s\n- neutron-ovn-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/4c602a2d3bb849e18041347f89443a35 : SUCCESS in 1h 05m 34s\n- neutron-ovn-tempest-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/0595420553ec4f0b9c2d98452cf24d2d : SUCCESS in 1h 10m 19s\n- neutron-ovs-tempest-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/201a8ee3e9e645a28adfe8bd9d44c5ba : SUCCESS in 1h 16m 51s\n- neutron-ovs-tempest-fips https://zuul.opendev.org/t/openstack/build/0317a41d2fe2490295d885052fed3563 : SUCCESS in 1h 18m 58s\n- neutron-ovn-tempest-ovs-release-fips https://zuul.opendev.org/t/openstack/build/3f529801e90c44cf9be305503a58cdd9 : SUCCESS in 1h 00m 26s\n- devstack-tobiko-neutron https://zuul.opendev.org/t/openstack/build/bc8aa46775ca4a7997b53f051b8244ea : SUCCESS in 48m 54s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/839a18feec1d49b09f42152d38ccf2e1 : SUCCESS in 53m 22s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/15e7cd4d419741bc9361314d31723695 : FAILURE in 1h 26m 20s\n- neutron-linuxbridge-tempest-plugin-nftables https://zuul.opendev.org/t/openstack/build/7579c9eecdc54276a8eab2ad51963435 : SUCCESS in 2h 06m 01s\n- neutron-ovs-tempest-plugin-iptables_hybrid-nftables https://zuul.opendev.org/t/openstack/build/a68975eee4ec4bc5980507eebea86384 : SUCCESS in 1h 51m 12s\n- devstack-enforce-scope https://zuul.opendev.org/t/openstack/build/93e19ee8e51f425b96473dfdd1eba0e8 : SUCCESS in 22m 41s\n- openstack-tox-py39-with-oslo-master https://zuul.opendev.org/t/openstack/build/c307a4ff318e4dcaa59d364a1e4a947f : SUCCESS in 30m 51s\n- neutron-functional-with-oslo-master https://zuul.opendev.org/t/openstack/build/fa2c418fead34c949997334f1b229f2c : SUCCESS in 55m 19s\n- neutron-ovs-tempest-with-oslo-master https://zuul.opendev.org/t/openstack/build/114990fc0b5a4fdab60c2f1c5c0bd6e6 : SUCCESS in 1h 17m 23s\n- neutron-ovn-tempest-ovs-release-with-oslo-master https://zuul.opendev.org/t/openstack/build/ab4460bb6984464bb6fcb2dab820bacc : SUCCESS in 1h 17m 52s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/b9aefe8e75c944ffaa280c34baeab0c4 : SUCCESS in 53m 21s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/3c2c7182562b44c89115c6c8fdaa2a2a : SUCCESS in 1h 10m 47s","accounts_in_message":[],"_revision_number":2},{"id":"6ce23b8bf4c8033950f76e7921854e95e0211a87","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-07-03 07:33:42.000000000","message":"Patch Set 3: Patch Set 2 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"815fba6bcce919186f4cebe16474b39f6fea7ee1","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-07-03 07:34:40.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"b6415bfc3acf404906639239a28fd8fe2bd06a46","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-03 07:56:16.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/10643d4df8724ab3a4c1fff70f745b26\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f9745eda07be4b0382ca022e451123be : SUCCESS in 17m 21s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/e90f57d8927b4ac5bcf85ab0468725fd : SUCCESS in 21m 06s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"c4f6df0238fa5a11b8b528f19045c2015055ada4","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2024-07-03 08:37:06.000000000","message":"Patch Set 3: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/July-2024/Arista_Tempest_Testrefs/changes/13/922613/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"001ef2534111cf0eedb916afc81a58b3b127c02d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-03 09:06:48.000000000","message":"Patch Set 3: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/922611 is needed.","accounts_in_message":[],"_revision_number":3},{"id":"99a5643427b467c5287fe12429a50adadc6306a5","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-03 09:06:49.000000000","message":"Patch Set 3:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/922611 is needed.","accounts_in_message":[],"_revision_number":3},{"id":"3b5a6e97ee98177129e2dcd499b2963ee9c6cd03","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-07-03 09:07:20.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"0862b48991fd904e5585e61f55637d50b949e612","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-07-03 09:13:31.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"81dd2a97639ae1b8eec63994ee9f7d4983cb096e","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-03 11:27:29.000000000","message":"Patch Set 3:\n\nBuild failed (experimental pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ac0beed9087b49149a059954d044c872\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/7bc942dd3d534844ad7e21fd3953272e : SUCCESS in 2h 05m 13s\n- neutron-tempest-plugin-openvswitch-distributed-dhcp https://zuul.opendev.org/t/openstack/build/3b6520c4cb8140319f20f68579de0d47 : SUCCESS in 1h 20m 40s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-distributed-dhcp https://zuul.opendev.org/t/openstack/build/7f186a7444804b649fb7685fe70439bd : FAILURE in 1h 39m 18s\n- neutron-tempest-plugin-api-ovs-wsgi https://zuul.opendev.org/t/openstack/build/44e900b8b3274c9ea2c47dcc862f7b25 : FAILURE in 1h 40m 11s (non-voting)\n- neutron-tempest-plugin-api-ovn-wsgi https://zuul.opendev.org/t/openstack/build/6e9bb1fe0d974cfca83588fdd7d1ed16 : SUCCESS in 1h 55m 26s (non-voting)\n- neutron-functional-with-uwsgi-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/43e057e6b2474cca94fa9244789a6600 : SUCCESS in 48m 46s\n- neutron-fullstack-with-uwsgi-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/cb89d1ece1c94b64bd55f3a83cb40539 : SUCCESS in 1h 41m 35s\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/c3abdbf2e1b94a9181a41c7f8400341e : SUCCESS in 1h 18m 50s\n- neutron-ovn-tempest-ovs-master https://zuul.opendev.org/t/openstack/build/8830d09ccdeb4667b1884af676e169f5 : SUCCESS in 1h 29m 04s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/936a5fa81c144cd88f5b60958ee8fe77 : SUCCESS in 1h 11m 57s\n- neutron-ovs-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/47cf13a1f9ee42e381d8c3c99afdad75 : SUCCESS in 1h 17m 43s\n- neutron-ovn-tempest-with-uwsgi-loki https://zuul.opendev.org/t/openstack/build/05ee9fc5be0340a2b4bff6ff721ec0ea : SUCCESS in 1h 28m 42s\n- openstack-tox-py311-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/51efb7b94ba64ce3b623849dd574d6aa : SUCCESS in 20m 55s\n- openstack-tox-py311-with-oslo-master https://zuul.opendev.org/t/openstack/build/1a3ca8ccf19541039baad601c262cb3a : SUCCESS in 35m 19s\n- openstack-tox-py311-with-ovsdbapp-master https://zuul.opendev.org/t/openstack/build/c680bec2c74044aabd5691fb92be8e91 : SUCCESS in 19m 52s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/e989463604314c99b4518df54746ef86 : SUCCESS in 53m 37s\n- neutron-functional-with-uwsgi-fips https://zuul.opendev.org/t/openstack/build/0196e916b29043668ef8aa85a8834288 : FAILURE in 51m 45s\n- neutron-functional-with-pyroute2-master https://zuul.opendev.org/t/openstack/build/24fc133718454d54acffaf8f3b659c43 : SUCCESS in 48m 30s\n- neutron-functional-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/c651628caad34b149ca952c95549e164 : SUCCESS in 47m 12s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/f7b765327c89434f87f0b30dfd6dea71 : SUCCESS in 1h 38m 33s\n- neutron-fullstack-with-uwsgi-fips https://zuul.opendev.org/t/openstack/build/5ba3d47c394e4cc19d3b24c65b3f5efe : SUCCESS in 1h 30m 26s\n- neutron-fullstack-with-pyroute2-master https://zuul.opendev.org/t/openstack/build/1a90330b0c08417f96f52f98a63c178e : SUCCESS in 1h 43m 51s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/d55aa8eab7da455db1f0660571683847 : SUCCESS in 59m 49s\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/198fb70130f0493f84d93f292ff75292 : SUCCESS in 1h 26m 08s\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/d146cfdcd1114dcb88b6fa8329bd8da2 : SUCCESS in 1h 58m 51s\n- neutron-ovs-tempest-with-os-ken-master https://zuul.opendev.org/t/openstack/build/2227d7b08927497db29fe10ab124b203 : SUCCESS in 1h 21m 35s\n- neutron-ovn-tempest-postgres-full https://zuul.opendev.org/t/openstack/build/9b1e878755ad438ca324d8fad5366eb9 : SUCCESS in 1h 22m 29s\n- neutron-ovn-tempest-mariadb-full https://zuul.opendev.org/t/openstack/build/1c90a35193934553934912343adc077d : SUCCESS in 1h 06m 10s\n- neutron-ovn-tempest-ipv6-only-ovs-master https://zuul.opendev.org/t/openstack/build/88ea33af0339460cbe27c178acf969df : SUCCESS in 1h 11m 36s\n- neutron-ovn-tempest-ovs-master-centos-9-stream https://zuul.opendev.org/t/openstack/build/803082f6358942dc9f971f1c940dfcff : SUCCESS in 1h 11m 47s\n- neutron-ovn-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/f262b8834dba464db45e4824b96153fd : SUCCESS in 1h 06m 20s\n- neutron-ovn-tempest-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/c6bb361fbf0c47a1a301ac9f5ff83145 : SUCCESS in 1h 07m 15s\n- neutron-ovs-tempest-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/65c6912145ac461c80b798925168c3da : SUCCESS in 1h 09m 24s\n- neutron-ovs-tempest-fips https://zuul.opendev.org/t/openstack/build/aab4365ace544b1685212e8ecbb0e2df : SUCCESS in 1h 07m 41s\n- neutron-ovn-tempest-ovs-release-fips https://zuul.opendev.org/t/openstack/build/22c211ce93ee4c6a84393fcdaea3eccc : SUCCESS in 1h 21m 06s\n- devstack-tobiko-neutron https://zuul.opendev.org/t/openstack/build/380aa098b14e44cea477497cce271338 : SUCCESS in 48m 56s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/ab3700bd6c934ed6902d1456672c74db : SUCCESS in 57m 00s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/9f459dd0a224428f9952631c1003c8aa : SUCCESS in 1h 23m 08s\n- neutron-linuxbridge-tempest-plugin-nftables https://zuul.opendev.org/t/openstack/build/f1e7a55c45ca47ce8f97ea0132ec878a : SUCCESS in 1h 51m 29s\n- neutron-ovs-tempest-plugin-iptables_hybrid-nftables https://zuul.opendev.org/t/openstack/build/22871683fe5c4c70b65377d2e5d31601 : SUCCESS in 1h 50m 14s\n- devstack-enforce-scope https://zuul.opendev.org/t/openstack/build/3cacaaecb047487ab69759a9c2da76c0 : SUCCESS in 17m 51s\n- openstack-tox-py39-with-oslo-master https://zuul.opendev.org/t/openstack/build/ef88e7255485402e9daa276406e1e86c : SUCCESS in 32m 01s\n- neutron-functional-with-oslo-master https://zuul.opendev.org/t/openstack/build/58f93182a9fa41568e068665e4cfc2b7 : SUCCESS in 1h 17m 18s\n- neutron-ovs-tempest-with-oslo-master https://zuul.opendev.org/t/openstack/build/425eb8ba014a402fb1e3c57e1570a3ff : SUCCESS in 1h 29m 36s\n- neutron-ovn-tempest-ovs-release-with-oslo-master https://zuul.opendev.org/t/openstack/build/f9b7f03af8924e43b0b9eafe41417c01 : SUCCESS in 1h 18m 57s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/21c3370709eb47ff932ffd590bea558a : SUCCESS in 54m 46s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/a77be0b76014469faf5f3a13575f6b43 : SUCCESS in 1h 08m 58s","accounts_in_message":[],"_revision_number":3},{"id":"bbcce50584a2ec4fe14f8c9c3449ad848fa3029a","tag":"autogenerated:gerrit:abandon","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-07-04 06:42:38.000000000","message":"Abandoned\n\nneutron patch merged, devstack one under review","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"c7f876c2f57b6de9c2ac834f33d08f26a0f16a1d","revisions":{"1eb8b3646bc62a94e96794688716a44b2a6dbde1":{"kind":"REWORK","_number":1,"created":"2024-06-24 11:35:19.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/13/922613/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/13/922613/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/922613/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/922613/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/922613/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/13/922613/1"}}},"commit":{"parents":[{"commit":"b02d428c7d7c984a32ead8b94e856db233988092","subject":"[OVN] Enable the WSGI module for the OVN mechanism driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b02d428c7d7c984a32ead8b94e856db233988092"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-06-24 11:33:51.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-06-24 11:33:51.000000000","tz":0},"subject":"DNM \u003d\u003d Test devstack patch 922611","message":"DNM \u003d\u003d Test devstack patch 922611\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/922611\nRelated-Bug: #1912359\nChange-Id: I691cda35117c32e0ca20eca03166f58ba13a9895\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1eb8b3646bc62a94e96794688716a44b2a6dbde1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1eb8b3646bc62a94e96794688716a44b2a6dbde1"}]},"branch":"refs/heads/master"},"051eb8cc73569fc1f21e257500c58fa01b2f1ec5":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-06-24 12:40:45.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/13/922613/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/13/922613/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/922613/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/922613/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/922613/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/13/922613/2"}}},"commit":{"parents":[{"commit":"02908885b2e4183a779844e0e763b7985eb8d48d","subject":"[OVN] Enable the WSGI module for the OVN mechanism driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/02908885b2e4183a779844e0e763b7985eb8d48d"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-06-24 11:33:51.000000000","tz":0},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2024-06-24 12:40:45.000000000","tz":0},"subject":"DNM \u003d\u003d Test devstack patch 922611","message":"DNM \u003d\u003d Test devstack patch 922611\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/922611\nRelated-Bug: #1912359\nChange-Id: I691cda35117c32e0ca20eca03166f58ba13a9895\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/051eb8cc73569fc1f21e257500c58fa01b2f1ec5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/051eb8cc73569fc1f21e257500c58fa01b2f1ec5"}]},"branch":"refs/heads/master","description":"Rebase"},"c7f876c2f57b6de9c2ac834f33d08f26a0f16a1d":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2024-07-03 07:33:42.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/13/922613/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/13/922613/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/922613/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/922613/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/922613/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/13/922613/3"}}},"commit":{"parents":[{"commit":"cfab008eef3d11055ad16b8dd0d49aaa1801bc95","subject":"[OVN] Enable the WSGI module for the OVN mechanism driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/cfab008eef3d11055ad16b8dd0d49aaa1801bc95"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-06-24 11:33:51.000000000","tz":0},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2024-07-03 07:33:42.000000000","tz":0},"subject":"DNM \u003d\u003d Test devstack patch 922611","message":"DNM \u003d\u003d Test devstack patch 922611\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/922611\nRelated-Bug: #1912359\nChange-Id: I691cda35117c32e0ca20eca03166f58ba13a9895\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c7f876c2f57b6de9c2ac834f33d08f26a0f16a1d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c7f876c2f57b6de9c2ac834f33d08f26a0f16a1d"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
