)]}'
{"id":"openstack%2Fneutron~923035","triplet_id":"openstack%2Fneutron~master~I7cc931d637f0679e5881e6ffcad49ac06867e275","project":"openstack/neutron","branch":"master","topic":"bug/2069543","attention_set":{},"removed_from_attention_set":{"15554":{"account":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"last_update":"2024-10-04 08:02:27.000000000","reason":"Change was abandoned"},"11604":{"account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"last_update":"2024-07-01 12:16:44.000000000","reason":"\u003cGERRIT_ACCOUNT_11604\u003e replied on the change","reason_account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}},"16137":{"account":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"last_update":"2024-07-02 09:23:08.000000000","reason":"Change was abandoned"},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2024-07-02 09:23:08.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I7cc931d637f0679e5881e6ffcad49ac06867e275","subject":"Put trunk tpi/spi ports in the dead vlan","status":"ABANDONED","created":"2024-06-28 09:44:58.000000000","updated":"2024-10-04 08:02:27.000000000","total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"6f92a3cc49705bffb1903c2680fb4281a48415a0","_number":923035,"virtual_id_number":923035,"owner":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":1,"date":"2024-07-01 12:20:35.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-07-01 14:16:34.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"all":[{"value":-1,"date":"2024-07-01 12:16:44.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-06-28 10:02:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-06-28 10:15:31.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-06-28 10:18:04.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2024-06-28 11:34:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-06-28 11:39:16.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2024-06-28 13:44:09.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2024-06-28 13:55:19.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2024-08-04 16:45:30.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"CC"},{"updated":"2024-08-27 13:52:53.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"CC"}],"messages":[{"id":"86f5d41c33ee6e666cc197960025ac96be9b193c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"date":"2024-06-28 09:44:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"909d5d1a596d7f376875bc707c4f6a2af931086c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-28 10:02:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5fa2169ff282425c9f8b0c6457ec0527\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/96b10a0033764b4eb1747067c357bb01 : SUCCESS in 16m 17s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/a3c27a9d9686448982060e6b8b249313 : SUCCESS in 17m 07s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bd2eb7a843f60faef9df97aa7674a56a4e682b95","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2024-06-28 10:15:31.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2024/Arista_Tempest_Testrefs/changes/35/923035/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"68a461f9480c5575bf38e064601c398e786c7302","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2024-06-28 10:18:04.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2a6e9b9866bf87e634db963f86469c4a3fa79232","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-28 11:34:25.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/fac144a3cc75449b99a2f4d0d270498f\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/ff6ef93719d34973babf8337d868516a : FAILURE in 1h 40m 29s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/ac316d70726144209eebd45c925050b5 : FAILURE in 1h 43m 23s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/f7d0a63351ae444da1fc6a3a579fcd5b : FAILURE in 1h 43m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5d59b4ee8a4544f0bc6089696471926c : SUCCESS in 46m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/df68757280484a74adc908fba47b651a : SUCCESS in 10m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/20c6876512d04c7692c1506c85031c58 : SUCCESS in 21m 25s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/70946db0a20c4a1987c90addc3816eff : SUCCESS in 18m 15s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/22d5669096bc4f7fa042f3beea823b0d : SUCCESS in 26m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a3907a5c022f445c90969ee284b448c3 : SUCCESS in 9m 52s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/8103b1869d0b431e91a643b4331657fe : SUCCESS in 19m 38s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/32910bd62f1c427688ab5261ba1af36c : FAILURE in 55m 23s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/06319d035fef4928b133b4f6289a8cf3 : SUCCESS in 1h 43m 29s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/e05182e50a8147c185069ad3bef3ac83 : SUCCESS in 1h 03m 06s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/0a8436b2ff974bf0b81355d936ca4c23 : SUCCESS in 1h 43m 22s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/1295772132734a929c030d87ca6cab78 : FAILURE in 1h 17m 20s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/b2e20232be324a74b1dfe587bfa76a92 : SUCCESS in 1h 13m 34s","accounts_in_message":[],"_revision_number":1},{"id":"085349a6c492752a6bcfa214d4f0810985089850","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2024-06-28 13:55:19.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7535cdb376d87e3ef2caf84dccbe5febcc79e31a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"date":"2024-07-01 11:49:19.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1, Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"0806c3a8461482adf916fbe3fdda7a27f2cd7316","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-01 11:53:25.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c063a31003b64db78e14705afc647293\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/592a669f789443e59ba795178cd13a09 : FAILURE in 3m 14s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/6e494e47e3bb47ba83707d0dd6396b8e : FAILURE in 2m 41s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"88ce1677ff75ae07c2d06ca3fcd6755e6fdbc94b","author":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"date":"2024-07-01 11:54:03.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"aab1fca556314c57aa25e70d1b1c1fd2cff59b89","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2024-07-01 12:16:44.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7f3729b8dd224dea12d3a2c797998b1758784d70","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2024-07-01 12:20:35.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/July-2024/Arista_Tempest_Testrefs/changes/35/923035/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"af06e5bcea5da8f1f19d1be315a130c1902b24b3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-01 14:16:34.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b16dbab14ac34676add21ce215e46a5e\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/a7bb27c3fe60492cba75ab942593ea62 : SUCCESS in 1h 54m 16s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/94545acfec80449c9183ff653737be62 : SUCCESS in 2h 14m 01s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/340fa758ed5a4e4bac1e7e8c8556732e : SUCCESS in 2h 25m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d53be4d9c10b44808ad59d8606028636 : SUCCESS in 48m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8638cdc04e7043079d0cfdce1bec581c : SUCCESS in 9m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/49836c17dbc04b449036f6df7e4ff1eb : SUCCESS in 24m 35s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c866746d88f749fcb34ef4fb2056ec01 : SUCCESS in 20m 46s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c766fc666b664d5694b751a55700432a : SUCCESS in 26m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a9281450a608480ab9c4b52947fe3608 : SUCCESS in 10m 55s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/5971c860722a42a59f0a94ceb8423cae : SUCCESS in 23m 16s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/73b8badd54fc4d7dac1d87f32ee6c986 : FAILURE in 48m 12s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/757d74dc92e24dcd9f79d175d17b03af : SUCCESS in 1h 50m 14s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/6b0f30415daf4ad5b504f570a4e57480 : SUCCESS in 59m 45s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/c83cc4dfeb09474aa2e0a0cd4718f4c3 : SUCCESS in 1h 47m 54s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/25c0654f1b9d483bb66af8b8d4db87dd : FAILURE in 1h 14m 07s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/f2bc22a864894b8eb97bafab6c29156a : SUCCESS in 1h 14m 18s","accounts_in_message":[],"_revision_number":2},{"id":"e985e4f9702144c0ef4574b7c5d23756c51f8ac7","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"date":"2024-07-02 09:23:08.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2},{"id":"92bc98c6ec1dfb04e24bdbd5519713ead858dcb2","tag":"autogenerated:gerrit:restore","author":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"date":"2024-07-03 14:06:13.000000000","message":"Restored","accounts_in_message":[],"_revision_number":2},{"id":"8a0b9286dacd01586070b3284801ac68a198a732","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-03 14:45:11.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7d2cced267ce4c4f9dd22b5ac6a8f438\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/88a6aa5b173d43b79de01922fa8d0279 : SUCCESS in 26m 31s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/1616e0bc1de24e7cbfe2a4ff9abb5266 : SUCCESS in 36m 09s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c0e5c9dded1cfdc7bb2850c3d29acea37e2489f4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-03 16:25:11.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8eebc056b098483592e43ea268bd06fe\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/32d8cdd8721f4911a3d9a06132f484a1 : SUCCESS in 1h 42m 34s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/4a054871534e4aa78a19aa8d72b80e02 : SUCCESS in 1h 37m 56s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/3746e79b96754b508e3290c06abe0640 : SUCCESS in 1h 32m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/73b8831c88994d888bb0458e7bfe81ee : SUCCESS in 33m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b5b6fd65aafe469fa6874c765564916c : SUCCESS in 10m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/389888899b4e439ea9fe67af0a5f9af8 : SUCCESS in 21m 58s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/38afd1cb6afc46b6a0994e390a2854ed : SUCCESS in 20m 43s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a3585d89d2554eb79f277d10239f0822 : SUCCESS in 30m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf7ed8fb1e254962a1468ca2ccf0d962 : SUCCESS in 11m 20s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/db7b0a9d230449e48246fe3612ac7bfe : SUCCESS in 23m 14s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/dcf2d424228448aa883cd19e5895b90e : FAILURE in 52m 42s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/1864435e5c7448219ba239e79669bca7 : SUCCESS in 1h 45m 54s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/f4b434e5581f49f8975d47e40431f02d : SUCCESS in 49m 53s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/a053523a52274d48837436d970eb0801 : SUCCESS in 1h 44m 29s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/a58bc68f150d4ca6811877d2758e77aa : SUCCESS in 1h 31m 59s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/a5f8f472bb6c40f6bc2710abb9506088 : SUCCESS in 1h 17m 01s","accounts_in_message":[],"_revision_number":2},{"id":"051f648c2ec57b482a53f9bb97c0d5133514dead","author":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"date":"2024-07-04 10:41:59.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9576ba1bc06617f5ca4914890666c2ec8011fce2","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2024-08-04 16:45:30.000000000","message":"Abandoned\n\nThis review is \u003e 4 weeks without comment, and failed Zuul jobs the last time it was checked. We are abandoning this for now. Feel free to reactivate the review by pressing the restore button and leaving a \u0027recheck\u0027 comment to get fresh test results.","accounts_in_message":[],"_revision_number":2},{"id":"986a5c3520cd523023a08f9d75d4e9644f823285","tag":"autogenerated:gerrit:restore","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-08-27 12:05:16.000000000","message":"Restored","accounts_in_message":[],"_revision_number":2},{"id":"7d9620b7a11c5a9e071a22ad1e7f08f11c5a8a10","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-27 14:05:56.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/39a85c707a6d47aa885870dc41f484c5\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/48d3bcf82b044155905d95758d0e1d2a : SUCCESS in 1h 58m 06s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/4ee425731c654c008de9ad155a85beee : SUCCESS in 1h 34m 18s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/bacf08eb4672406fb85e043489c9f2ec : SUCCESS in 1h 21m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d78419d4810b4d5b8b5b82c37094818e : SUCCESS in 29m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9e7d1c17544f42319b9dd13919672d5a : SUCCESS in 10m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/32d09b0739a5454b96995cf16b8ee925 : SUCCESS in 30m 42s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/337e1b3d2c204e4ba5994b60d71fab94 : SUCCESS in 21m 26s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/cdfe11215c404ee99e9b0928234f06d7 : SUCCESS in 24m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2fd6b61d869f47e69a32132e084ff752 : SUCCESS in 11m 32s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/196d3cd9a51149f89337d821ea7e7f18 : SUCCESS in 19m 55s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/81ed71bf42e241d79f642cfedd0cd8e5 : FAILURE in 49m 58s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/5c06290d5ae54b61a5c2849d0bbd224d : SUCCESS in 1h 35m 44s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/38ee4418f2174511b1c0455c9f64a719 : SUCCESS in 1h 01m 07s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/a80299c556ca4cac8635deec028bfe62 : SUCCESS in 1h 29m 14s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/406f26c0778840409a98dcd00b523a1f : SUCCESS in 1h 49m 43s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/5151de6c0c1e4534b07c8a1c120cf4e5 : SUCCESS in 53m 47s","accounts_in_message":[],"_revision_number":2},{"id":"e544cee0f4962a2261e5c714ef769c1072b92257","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-27 14:37:00.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bc178bc426b940cf946cd113003db7b7\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7ed28d7ff53443ca949c93f9ca5cadd3 : SUCCESS in 26m 31s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/97cce9bceedb404fa6c6af9d6e0ae740 : SUCCESS in 29m 47s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"6f92a3cc49705bffb1903c2680fb4281a48415a0","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2024-10-04 08:02:27.000000000","message":"Abandoned\n\nThis review is \u003e 4 weeks without comment, and failed Zuul jobs the last time it was checked. We are abandoning this for now. Feel free to reactivate the review by pressing the restore button and leaving a \u0027recheck\u0027 comment to get fresh test results.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"b6604d809f9a4efb54eeece53ba43cf1d4a4c2c9","revisions":{"20b1f79ab33c5f2271618260a7e95ce505529aa0":{"kind":"REWORK","_number":1,"created":"2024-06-28 09:44:58.000000000","uploader":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"ref":"refs/changes/35/923035/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/35/923035/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/35/923035/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/35/923035/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/35/923035/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/35/923035/1"}}},"commit":{"parents":[{"commit":"466f5e6f1afc12f61d073a0c4e3e4774feb4a7cb","subject":"Merge \"[FT] Run test_periodic_sync_routers_task tests serially\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/466f5e6f1afc12f61d073a0c4e3e4774feb4a7cb"}]}],"author":{"name":"Bence Romsics","email":"bence.romsics@gmail.com","date":"2024-06-28 09:10:57.000000000","tz":120},"committer":{"name":"Bence Romsics","email":"bence.romsics@gmail.com","date":"2024-06-28 09:42:59.000000000","tz":120},"subject":"Remove taps in trunk bridges from the dead vlan","message":"Remove taps in trunk bridges from the dead vlan\n\nWhen [os_vif_ovs].isolate_vif is True, os-vif puts the tap in a trunk\nbridge into the dead vlan but neutron did not remove it, since it only\nset the vlan of the tpi port.\n\nHere we remove the tap in a trunk bridge from the dead vlan.\n\nThis patch has a counterpart in os-vif, please see the related change.\n\nChange-Id: I7cc931d637f0679e5881e6ffcad49ac06867e275\nCloses-Bug: #2069543\nRelated-Bug: #1734320\nRelated-Change: I4f80d94bc1f346fd0a7ad0b9d684792f1451cf24\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/20b1f79ab33c5f2271618260a7e95ce505529aa0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/20b1f79ab33c5f2271618260a7e95ce505529aa0"}]},"branch":"refs/heads/master"},"b6604d809f9a4efb54eeece53ba43cf1d4a4c2c9":{"kind":"REWORK","_number":2,"created":"2024-07-01 11:49:19.000000000","uploader":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"ref":"refs/changes/35/923035/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/35/923035/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/35/923035/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/35/923035/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/35/923035/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/35/923035/2"}}},"commit":{"parents":[{"commit":"466f5e6f1afc12f61d073a0c4e3e4774feb4a7cb","subject":"Merge \"[FT] Run test_periodic_sync_routers_task tests serially\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/466f5e6f1afc12f61d073a0c4e3e4774feb4a7cb"}]}],"author":{"name":"Bence Romsics","email":"bence.romsics@gmail.com","date":"2024-07-01 10:32:13.000000000","tz":120},"committer":{"name":"Bence Romsics","email":"bence.romsics@gmail.com","date":"2024-07-01 11:43:28.000000000","tz":120},"subject":"Put trunk tpi/spi ports in the dead vlan","message":"Put trunk tpi/spi ports in the dead vlan\n\nWhile working on a fix for bug #2069543 I realized that the tap in a\ntrunk bridge should not be put in the dead vlan, however the trunk tpi\nand spi ports should be. However we never did that before. Start doing\nit.\n\nFor the background please see related os-vif change I4f80d94bc1f346fd0a7ad0b9d684792f1451cf24\n\nChange-Id: I7cc931d637f0679e5881e6ffcad49ac06867e275\nRelated-Bug: #2069543\nRelated-Change: I4f80d94bc1f346fd0a7ad0b9d684792f1451cf24\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b6604d809f9a4efb54eeece53ba43cf1d4a4c2c9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b6604d809f9a4efb54eeece53ba43cf1d4a4c2c9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
