)]}'
{"id":"openstack%2Fneutron~923040","triplet_id":"openstack%2Fneutron~master~I7ad7df72807c95571ef3156c99072852d1c4f494","project":"openstack/neutron","branch":"master","topic":"bug/2071451","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2024-07-02 13:13:38.000000000","reason":"\u003cGERRIT_ACCOUNT_16688\u003e replied on the change","reason_account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2024-07-08 11:36:10.000000000","reason":"\u003cGERRIT_ACCOUNT_8313\u003e replied on the change","reason_account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2024-07-11 07:37:36.000000000","reason":"Change was submitted"},"9531":{"account":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"last_update":"2024-07-11 02:46:00.000000000","reason":"\u003cGERRIT_ACCOUNT_9531\u003e replied on the change","reason_account":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"}}},"hashtags":[],"change_id":"I7ad7df72807c95571ef3156c99072852d1c4f494","subject":"Always get local vlan from port other_config","status":"MERGED","created":"2024-06-28 10:09:32.000000000","updated":"2024-07-11 07:39:05.000000000","submitted":"2024-07-11 07:37:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":17,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"923040-bug/2071451","meta_rev_id":"fdfd60f563625e1096f3e26ae3bb8a50c4e45336","_number":923040,"virtual_id_number":923040,"owner":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2024-07-02 14:38:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-07-11 07:37:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2024-07-02 13:13:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2024-07-08 11:36:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":1,"date":"2024-07-08 11:36:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-06-28 10:43:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-06-28 10:47:09.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-06-28 12:21:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-06-28 19:44:38.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2024-07-01 07:26:35.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"CC"},{"updated":"2024-07-01 11:53:58.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2024-07-02 09:51:14.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"}],"messages":[{"id":"cbd7674d1ea34d30a479aa9c764f8cd362d85dab","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2024-06-28 10:09:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"908fcf3a6e96281a7c578c525971bbddc38d0ad7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2024-06-28 10:10:47.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"1e87b01234638fe2ccec7eef319d7ff00fe67336","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-28 10:43:52.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/473c677848974d0fb65bd29a7f897fb7\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9f40629fc35c41459446471c4c22d664 : FAILURE in 2m 59s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/9eb8287d0d5b4a88b6e280394420b55a : SUCCESS in 32m 39s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"3afd3deb0c01df2e1ab12ff3b839fc8513c95b76","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2024-06-28 10:47:09.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2024/Arista_Tempest_Testrefs/changes/40/923040/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"651113307151972ab989159cabcb07bfbedc0dde","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2024-06-28 11:18:01.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2024/Arista_Tempest_Testrefs/changes/40/923040/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"d3248f0d49616b53c8c54de65cab4a3e8245c29d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-28 12:21:30.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/575f91c2b9b04497aafc5efd0c8493f1\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/9badad77aa5147bd882cb1b18e3d78f9 : SUCCESS in 1h 33m 16s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/692677680c8d4636af596652912da5f1 : SUCCESS in 1h 49m 36s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/0c591569b17e444a8211090cc5448d68 : SUCCESS in 2h 08m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2d5cd11ce2a44f48ae4142347e479aa0 : SUCCESS in 37m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/61d1ec8eea5a4aecbd5ceee333c98fd6 : SUCCESS in 8m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/39a5e6187bb94b6093e4f436c6fe37a6 : SUCCESS in 22m 56s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f073ce8954d54582ac696fc0d1701587 : SUCCESS in 21m 08s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/eb6b36ef315f4d69850e462e16316580 : SUCCESS in 26m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c62c2ba37cb940f4a40a93ca9137ab3c : SUCCESS in 10m 52s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/8538f1d9147749e1a25bab5c64f92513 : SUCCESS in 24m 50s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/c4b6122c973144e1863822c791f138b2 : SUCCESS in 53m 40s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/a31e689b433248b9b67dabb1fcc736f7 : SUCCESS in 1h 41m 21s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/d1eb53000bab46fe943bd2bb67847cc4 : SUCCESS in 1h 01m 01s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/2f89b87ab4d74c4aad1fe9f94442bbd4 : SUCCESS in 1h 04m 06s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/882da3441d644e6d8d4c0689f1ec39bd : SUCCESS in 1h 44m 21s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/eb073d080fda472b85d0331f6dd856f4 : SUCCESS in 1h 27m 51s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/171558a72c2b4879946737b3c008c093 : SUCCESS in 1h 06m 03s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/3f0e117ca82644fd8580975c0849abbd : SUCCESS in 1h 12m 02s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/7cb5212adbc84c20a73f5b9d22726d79 : SUCCESS in 1h 13m 45s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/67c64eff0e26459fb9dd601df0791f91 : SUCCESS in 1h 20m 59s","accounts_in_message":[],"_revision_number":2},{"id":"61b9c11b5b5de3a73dd7a7be1aa26c21102a78fc","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-06-28 19:44:38.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"5dbb17055e9f9f28bc8034d174b664666b849f41","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2024-07-01 03:36:17.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4800e71dcc3edef05769e450a390a653d616c2d9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2024-07-01 03:45:11.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"3f08e4a0e4cb9197faab812e1345efef05dc22d1","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2024-07-01 03:46:12.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"b0e3fce166f784846ab730766e23174a1cdcf944","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-01 04:04:23.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/42d30eca85344e9ea98e3c0ad5f63b31\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/58d4aa89319342f89a0633afad02ef60 : SUCCESS in 16m 45s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/f0bd057113be4e1eac16964a27722569 : SUCCESS in 18m 21s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"1fafe016b3682c73d15420b46f2767e05f63efac","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2024-07-01 04:17:27.000000000","message":"Patch Set 3: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/July-2024/Arista_Tempest_Testrefs/changes/40/923040/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"7b362bfa6c7f7b088923f6de4a90d4a9294527a9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-01 05:34:39.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/fb98b0ea521347b4bc4f2916ab93863c\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/845a7c5ee35b4857abfec8829378c75d : FAILURE in 1h 28m 45s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/2a8fdf50f0974e5c83b6a33e9c118b5c : SUCCESS in 1h 42m 47s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/038faf0c176d447fb1c977ac3a4a11a1 : SUCCESS in 1h 47m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b277d2ee3dc74eb3847313ffc5c56ef6 : SUCCESS in 29m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e1fdd1da366f4bb6bb3941647f757416 : SUCCESS in 8m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5efc2c9e9fab4ce495f56588fd32bed7 : SUCCESS in 26m 37s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1b38516c3ade4a008254f9a91a4e3e2b : SUCCESS in 15m 59s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8333f34e01e8423bb1ade5709b99b1df : SUCCESS in 24m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c1a36e14824a4748a86a4311818d99f6 : SUCCESS in 10m 15s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/5b22c8075f3b438a932a2d716c128f8e : SUCCESS in 25m 12s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/75d24298091644f49bee5dd6bd6a47c8 : SUCCESS in 37m 30s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/9453fff48c5a4e39b63ffed341d0d607 : SUCCESS in 1h 33m 05s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/618479898c71495a932784b25bb83322 : SUCCESS in 47m 00s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/45c61436fef7439ba3e4c7c626441fb9 : SUCCESS in 1h 01m 28s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/b3eae926a0094594b27e1f4b0ce5e99f : SUCCESS in 1h 18m 46s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/513563f9056e49a7ae576afbc7e13656 : SUCCESS in 1h 22m 40s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/12da429800ae4a37bb721433dc99f9a7 : SUCCESS in 49m 34s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/2ce0469ca0314d96a790fee3a64527bd : SUCCESS in 1h 13m 15s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/84f150ef39524a258a6ce819dbdd4ec2 : SUCCESS in 1h 13m 58s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/7e922dccdb9247d6980b9282c233afab : SUCCESS in 1h 16m 00s","accounts_in_message":[],"_revision_number":3},{"id":"cb3e9f15d899eb5602f17754889bdbda56959c5e","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2024-07-01 07:02:19.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"89147150015b8a7cf1df83ebe1cb07d28c114f56","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-01 07:22:30.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ce72a9c820f840ee89491677f9caae29\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/fae7ea9c382941129fd4fbad4cc00237 : FAILURE in 4m 12s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/174bf761105b46ae9c17d7e27c8adfeb : SUCCESS in 19m 26s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"41300b22c354809c71ada712cfa71c3b0f4247a9","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-07-01 07:26:35.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"e2af0c3e9aae7dc42d25ba49ebb7bce2baa5e538","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-07-01 07:27:31.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"ca315328991d87a7ae6f4b9c4ecb162c7c2dae12","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-01 07:44:33.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8a39aef660484eb3b77f2644fa53a606\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/61de85bc17b247b889bf8f69420bcd87 : SUCCESS in 17m 07s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/2b83781c57784b54a1d4d63add4837aa : SUCCESS in 14m 41s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"6475249983b558bd1d556cd9249b2fef9e0b1d01","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-01 09:40:45.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/20e88d118250404d9ad98cefdd8e3d2f\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/37b8c1faa06746fbadd996d73fed1066 : SUCCESS in 1h 59m 56s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/e3d22f4bc113461b9d659a8670819a86 : SUCCESS in 1h 50m 30s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/aae372503d1841a99a9bb020513a05ec : SUCCESS in 2h 17m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bc3152a5f70b42ea931cccaa4431cafc : SUCCESS in 47m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0749f4af9eba4664afcd48b637cc8634 : SUCCESS in 9m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3d7af0631a054eb39f0012673af31e00 : SUCCESS in 22m 24s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/bb24b38bc2174e5281f7f01b5296a95c : SUCCESS in 21m 10s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/04ad8e7604cf4475b0fbd2a92b95da08 : SUCCESS in 26m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/20873b8c819947c6a07d15315a1eb795 : SUCCESS in 9m 53s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/f0a7452a36d843b6b8bfdaac13b7e3cd : SUCCESS in 27m 38s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/4ad670b343bb48eca15a858f2fa2edf7 : SUCCESS in 42m 33s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/7817b5f08093450b8abc1032e636c2e6 : SUCCESS in 1h 48m 08s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/8afc1414484e46c592a0f613368df509 : SUCCESS in 59m 40s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/73dcaafb0a6b4e8f834e134dca15c0cb : SUCCESS in 1h 09m 42s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/d2600373ce15417ab63a18d44f1274b4 : SUCCESS in 1h 42m 59s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/bc78d41a11534c40b9af6f9644486bce : SUCCESS in 1h 37m 58s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/cb0a539e36e548898fe60719174dea60 : SUCCESS in 1h 01m 36s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/5189c60dd4be4415876150bc81a30c9f : SUCCESS in 1h 16m 26s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/72c35ae2713442faadf129c6fecc799f : SUCCESS in 1h 15m 42s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/4ec893cfcf5349b0a6a8a0ec91774101 : SUCCESS in 1h 13m 40s","accounts_in_message":[],"_revision_number":3},{"id":"501b8290a9dc9f866ce9551feb51e435957d229a","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-07-01 11:53:58.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"b33fc54ff9daf9560faf27330e7b90e16e3f354a","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-07-02 09:51:14.000000000","message":"Patch Set 3: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"bc357f06ec23f133dc49db2fffec97bdf4040546","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2024-07-02 13:11:25.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"8b737b21c0bb17420166f83374f1fc71849127f5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2024-07-02 13:12:46.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"ada982513a35a1d2eec8d7afe52c357e3cf53a66","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-07-02 13:13:38.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"8ad8cf5b28b677968054b7069acac53b65118c7f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-02 13:38:15.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4a8800e7054944dd96d481b7a771a314\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ca06a3d8bd524d949e754c0ac79d2b8a : FAILURE in 3m 10s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/d5843ee676664fc0bb882e6d61002558 : SUCCESS in 24m 37s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"3bc5b23ba2097d6c92870dfad8604aa53987103c","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2024-07-02 14:38:05.000000000","message":"Patch Set 4: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/July-2024/Arista_Tempest_Testrefs/changes/40/923040/4 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":4},{"id":"60c1f97d8016d283f4f4c5f7a1c82384f5f316ce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-02 15:52:13.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/249bcea279964c7d987a7b609a702b61\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/3de247538fc04772b01fdc2e84eb28a8 : SUCCESS in 2h 01m 13s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/6c4c25c9fc884fe286243047552bbbeb : FAILURE in 2h 36m 03s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/95a582fd3cf74e6bbce112aa627765f2 : SUCCESS in 2h 36m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2132ba532eab496b99064a37860c6cbd : SUCCESS in 42m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/238cc488ca2e4dd4afab1603cb727942 : SUCCESS in 9m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5c1fc4fcab3d45b1b460e0aba8c6f6ed : SUCCESS in 22m 14s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d844554238b64bd49f9d326ff617ee2d : SUCCESS in 23m 14s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ac5c9c16ec384afda044d26a882e9bd6 : SUCCESS in 36m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e82ead961d1d4e9785a0a9b9e014017f : SUCCESS in 9m 29s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/700aca3521a94dd7baefb3413ae2d74c : SUCCESS in 22m 21s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/4878f5aba1f342aaa48e5cbbb588b118 : SUCCESS in 37m 15s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/3c91fc3fd9a54496a299e9defa29546a : SUCCESS in 1h 39m 38s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/043f01bf8e1e4ddbb96ad7f62ce861fc : SUCCESS in 59m 45s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/181735228ea74871ab29afd14eff7c7d : SUCCESS in 1h 09m 26s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/2d589466e94b4fad8e4ff9a5146d6660 : SUCCESS in 1h 48m 24s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/62ce1c0337ae4a6ba6edeec23d1bed62 : SUCCESS in 1h 33m 28s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/e7126a92512e475086ee813b5ceef2a9 : SUCCESS in 59m 34s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/7747bc784a58411c842f50a25bd1a460 : SUCCESS in 1h 18m 55s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/1f5b2659bf1a480caf9f89848c5d931d : SUCCESS in 1h 20m 30s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/0532e95e0a1640b78bb84a5df2af740b : SUCCESS in 1h 25m 29s","accounts_in_message":[],"_revision_number":4},{"id":"7b0a33fa1059b3a134d4f82b65177d70b9fd4068","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2024-07-03 00:36:20.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"36e8b223ed0b695b6884254c2cea06393dba482c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-03 00:54:27.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6356a1ba048d45b4a4c1df5d1336e5ee\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6d2b75dd6ec941b4b0b8a221e0ddcce6 : SUCCESS in 16m 49s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/6ff1aa1a13424a5c874e28f99e2861aa : SUCCESS in 17m 19s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"f250273f0b4aa374c9ffccb5663ffaff7917ad5f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-03 02:43:00.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/431d3dd0bc3c42649f87d156a62586cb\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/6699c672358540549e1ba55fee50276c : SUCCESS in 1h 54m 12s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/152a68617beb4a9facf5a89ecda3437a : SUCCESS in 1h 45m 16s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/0381ac951e884e01a134d4a74d750b51 : SUCCESS in 2h 05m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8791259d85724164bfbe687705db7ede : SUCCESS in 39m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/90e9630dd3464e068ae0934812670e51 : SUCCESS in 9m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3e9020257d7e4015961fd6da46a2ec1c : SUCCESS in 21m 34s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4afe2642613f4395a477a1486f3b0f80 : SUCCESS in 21m 00s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/cf05e6143d9e43268329a0871cb52f3b : SUCCESS in 29m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce2475daab73418d831d67d5d9fa8cc2 : SUCCESS in 9m 40s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/bfacc7d2c8f548ebb462f66e0d6308fd : SUCCESS in 20m 57s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/eb64f7c1a9574c3fbd3f1ed9f5fb458e : SUCCESS in 51m 31s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/2406198b4cd44d548cc75d401c3c3cee : SUCCESS in 1h 46m 12s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/cee98dca831d43b2b6461fa81009f032 : SUCCESS in 58m 59s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/edcb466bf5eb47e0962ae5cd52e33125 : SUCCESS in 1h 12m 52s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/484e222ebbd94fdeafca520e0250b69c : SUCCESS in 1h 39m 19s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/4436a00445cd420d8fa9ad9fa3ebbe49 : SUCCESS in 1h 25m 50s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/ddd41f1514b54cdab249dc98c12c9182 : SUCCESS in 1h 06m 51s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/b9a9a03f23304ae289f4d2bdcbfdaaf7 : SUCCESS in 1h 12m 54s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/6f23ae6db22342119fffc08b7b57938d : SUCCESS in 1h 17m 31s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/feb2a453ee964238818f489d4a03623d : SUCCESS in 1h 21m 05s","accounts_in_message":[],"_revision_number":4},{"id":"cdecaf03c3633d895d9c350aa9d10c370e0c1c9d","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-07-08 11:36:10.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"556b0dd407a9eb4b6279b8acf726ebf221a0bb9d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-08 11:36:54.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"f7253657d7b5636127a0321663309d14dfa9ce05","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-08 14:04:44.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7a8559d3a8ff43919b2aeb5c3140e474\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/5c30850080db49f6836601008a88a09c : SUCCESS in 1h 29m 35s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/3f1b30b7a7df452198f2637a1c30c45e : SUCCESS in 1h 41m 13s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/81c4e993d6b54584abc96347c57ba89d : SUCCESS in 1h 25m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1d691218a322429a9feca25f12586d10 : SUCCESS in 9m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/867511d186e3464fa5fdbca8300394f8 : SUCCESS in 24m 29s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/65f3f40ee9ce4c38a918a60d919053e5 : SUCCESS in 20m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9bec430a8cac4605a4aaffc1d269e535 : SUCCESS in 11m 54s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/1762cd8b119e4887a31c5355d67c35de : SUCCESS in 55m 28s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/dd5e81eadce6441aaebc5d5814d72df3 : FAILURE in 1h 26m 06s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/b64787f732314071aa75acad5881104f : SUCCESS in 1h 40m 30s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/caab9acd2bcb404a9a1744bd086c705c : SUCCESS in 1h 03m 05s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/8c20e965fb1040668998ff5fc47195bb : SUCCESS in 59m 39s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/6937ad6afd5441e69f72eb60a4f0ac41 : SUCCESS in 56m 33s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/e5754d48dd6f419aa57b0e6e525f5249 : SUCCESS in 1h 10m 48s","accounts_in_message":[],"_revision_number":4},{"id":"4b49e2ba47b64adb639d7b13039db20522165b93","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2024-07-09 00:56:13.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"bea16b905100431e1479d2b0934a44f62e14a29c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-09 01:00:41.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/34d33fc22a6643eca0c4e3706e9fcb65\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1a12309fa47b497f9426845698e3c9ea : FAILURE in 3m 36s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/1c1e88f5c76b457fa91057a4a0ebf075 : FAILURE in 2m 59s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"0d38c497782aaa4ab5efec0ccd40e6ca3d32e445","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-09 02:51:54.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/15b320c2700046f086467ccd6cf70c14\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/7ee88590afd04ef1acf55f12ba776d5e : SUCCESS in 1h 43m 09s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/e405869a46ad4bd0a86f0844e1c1b748 : SUCCESS in 1h 49m 19s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/ff45e8f041cd4490ad7687d42ab06b16 : SUCCESS in 1h 54m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/20790035017149bbb9ade129bdbe33db : SUCCESS in 23m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/11e649cd67e3432583d2bdd364a7d383 : SUCCESS in 7m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/674638a8561e4140a6529c8760ced6d1 : SUCCESS in 15m 39s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d98bc1c2215641f6ae273b24fc99184d : SUCCESS in 14m 44s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0b0dd602039142aa8ee17f5ae4dcaa44 : SUCCESS in 18m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3f0200d8d84d4e3e95de1383bc5e164f : SUCCESS in 8m 02s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/e62b3847ffff42938c5b98a3d0c40228 : SUCCESS in 14m 56s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/8b6ccbcced5f4ee0ab08b0e8c4a8c933 : FAILURE in 50m 34s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/215b1b496d1d467d9080bf84324b0912 : SUCCESS in 1h 47m 01s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/c15c93098ea04ce292be2d165bfe2703 : SUCCESS in 58m 57s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/7ce792501e4642efbbbeda5ae2f1638e : SUCCESS in 1h 10m 17s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/be20e3a60d054946b57e4aac0b77e71f : SUCCESS in 1h 33m 28s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/a0198367bcb1434b9d6a94b02a7a2ff3 : SUCCESS in 1h 28m 05s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/c5e887a990ac4ea881bc9d6b915cdbc9 : SUCCESS in 1h 02m 01s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/532e835fcc6d4d80bad89c060195d59e : SUCCESS in 1h 09m 28s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/043c53916d8c478b834bc448ea95352f : SUCCESS in 51m 30s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/c63f6adb5c70470188bded485e262da4 : SUCCESS in 1h 05m 34s","accounts_in_message":[],"_revision_number":4},{"id":"972d3cc54c2a688945b57f23f1a7f2557355844d","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2024-07-09 06:56:15.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"ccd13ac5aa67aca31e255177e7c2320cc21437ad","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-09 07:37:14.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5220b2a965644494930285e5c0072ba4\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/58ca32c73282469884c4e5e2e0dcf1b3 : SUCCESS in 39m 01s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/f969d31ef409480abb149bd37f9a0013 : SUCCESS in 32m 54s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"6795de39cc3d59af1fd0ca6e62fe206c68e5602b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-09 09:03:04.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0bb91177a05e4f70bf335f64fdeb4cde\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/2c55bbf864ff4ddf9d7a2cf543e13712 : SUCCESS in 1h 20m 35s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/9a1b8a9d9e8248b6a8fb12e7d0d6741f : SUCCESS in 1h 33m 41s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/18b337a74b50472181da794f1a8911e6 : SUCCESS in 2h 03m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3edad1a742924899bd8a416adf091de4 : SUCCESS in 45m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/17df5b9cef1b44d3abb07573b40e0940 : SUCCESS in 9m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fd6578ec9a0b4018aace00bbd3d1e830 : SUCCESS in 25m 51s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/16dfdc452c5d4d14bea9565fac023154 : SUCCESS in 18m 15s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/311ea31d3c114550b1e748360f51ac66 : SUCCESS in 25m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b38a96ed8ab6497a98e4fa4ffb8291ae : SUCCESS in 9m 57s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/3f02e3cb7cd14feabf5642e8c2132661 : SUCCESS in 20m 20s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/ec46c877e40949dcae727bc2224077c4 : FAILURE in 44m 06s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/c0a6d5c356ed44ccb135af1a0dd898e1 : SUCCESS in 1h 28m 35s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/c5729518f72e4dd6a03cd6f4765835c5 : SUCCESS in 53m 17s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/389902da5ef6459ca5c7a002c693f2a2 : SUCCESS in 1h 03m 19s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/2f9fe0812de040339d8f0bac22231602 : SUCCESS in 1h 27m 01s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/b15866fd22904920a34dfdafae657fd1 : SUCCESS in 1h 27m 42s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/3a6cf89471cb4982b6ab820e7d547a70 : SUCCESS in 49m 33s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/0127a20583d040c6bca9e33a61a7bedd : SUCCESS in 1h 04m 06s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/dbdc74a3bdb540a6ad6e7a95e24ed0c3 : SUCCESS in 1h 24m 35s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/ffd9ecaf1fd04b55962afd1a4b23a4f7 : SUCCESS in 1h 09m 05s","accounts_in_message":[],"_revision_number":4},{"id":"648e85fc74708a883f7358f82ab144fca7fcc060","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2024-07-11 02:46:00.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"cd302594e05e7e05104fe83320d1096261650a42","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-11 03:07:29.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d81592e6acb94ab1a33f947ddfa89161\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0a91f314107b4521be20d509addd1610 : SUCCESS in 18m 27s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/0af394c3ccf14b8cb0dee3a241e4386d : SUCCESS in 20m 36s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"214d1e8fec2b0184bf5df9ba05acedf2d7bc0be4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-11 05:11:09.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/91b1c7811df24f3290fb5ccf56ca2bd6\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/8a7588154ecf4af4a4019b9b9558479c : SUCCESS in 1h 49m 38s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/b3d99bf664b54df2bb8cfce63a3db5c2 : SUCCESS in 1h 59m 28s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/4aa096c2c26e4c48a0653dd824a4d292 : SUCCESS in 2h 22m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/43185782e9d6473993e6a9d87137122e : SUCCESS in 31m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/073fe31ac8ab49909421a35651044828 : SUCCESS in 11m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a9b3891ff7e24ad9946b493d1c4bcf55 : SUCCESS in 25m 19s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c52094ed7ec44bb29a51ba2348622d18 : SUCCESS in 17m 51s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/31aeedaab752462b937760c1155b5557 : SUCCESS in 24m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e01e88961908409999d78a9be3142cdf : SUCCESS in 9m 42s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/6e34a26701d94f2c99f89d3f3cc10adc : SUCCESS in 24m 40s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/01a798d065124ad3a25782c191fb2c69 : SUCCESS in 53m 21s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/d397bf3c47224eb7be9d41406d81c944 : SUCCESS in 1h 39m 31s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/bce08745213441f08f73287033decd85 : SUCCESS in 48m 48s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/a1301e3c474c4a7c9503e635bee6d88a : SUCCESS in 1h 07m 14s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/08d59a542a8d47a6b7ef314d69b19d26 : SUCCESS in 1h 39m 24s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/55b30b40405248c6ba8abe56c3d9139b : SUCCESS in 1h 15m 39s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/a08894446d41437c9489c378cda90482 : SUCCESS in 1h 02m 34s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/b54d72fddde140f4be11033d7a602fc1 : SUCCESS in 1h 00m 53s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/a2fe8caf16d0420094e0d23f8e651bfc : SUCCESS in 1h 08m 42s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/6e9489620c5f44309aed93caed71a0f9 : SUCCESS in 1h 16m 13s","accounts_in_message":[],"_revision_number":4},{"id":"acdd501deb494d2d69d9f2938b1c4bc00575c944","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-11 05:11:50.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"05599c46001e2e70ec14dcaaf111bbcdd4dab081","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-11 07:37:36.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/08021cb8ceee44959d4d7936dc19d105\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/2a29805e7a1e40bd9c2ac3810a8011bb : SUCCESS in 2h 14m 40s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/1fb3d0be7edd41719c9e97c232817d51 : SUCCESS in 2h 24m 27s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/87f0a6a3a1f0412f9345348044069c94 : SUCCESS in 2h 00m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/201e08fa503c40f6997d024647e28fb5 : SUCCESS in 10m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0032302a82bf4fd38cb180d2f1f60477 : SUCCESS in 24m 05s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e99e6688214145df938163e381d2e0dc : SUCCESS in 15m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f61c0ec01c0b43ecb1b4906afaf6e220 : SUCCESS in 9m 12s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/68945aa184c6460082db99694d02b6b4 : SUCCESS in 58m 29s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/ec61de2c042b4d968c53d3f8a41fbea5 : SUCCESS in 1h 49m 41s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/dee7688db2b340eb9e01c2a9af054ec9 : SUCCESS in 1h 27m 22s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/3e5f492e1c724207a4fe2027de4bcc3e : SUCCESS in 50m 41s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/02e98732423a4a9d9511ee45d376f778 : SUCCESS in 1h 00m 50s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/37bd5083625049278d30b511e82728e1 : SUCCESS in 39m 10s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/8d591f623d1a48e0a6fec1edbcf67085 : SUCCESS in 1h 01m 34s","accounts_in_message":[],"_revision_number":4},{"id":"eb4d7d48c516b13d905238d811cb540298a17fdf","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-11 07:37:36.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"fdfd60f563625e1096f3e26ae3bb8a50c4e45336","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-11 07:39:05.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/36436fff3357405896db249509f1e183\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b3622dbd4184b5db9303244ee662fbf : SUCCESS in 1m 10s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"ae587c34ab59a5717630eded2fab84413f3c1742","revisions":{"fa339d860d6df80d287aa55cce98e6bdc7281b68":{"kind":"REWORK","_number":1,"created":"2024-06-28 10:09:32.000000000","uploader":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"ref":"refs/changes/40/923040/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/40/923040/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/40/923040/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/40/923040/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/40/923040/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/40/923040/1"}}},"commit":{"parents":[{"commit":"2ce9ca6cdf065b28aa0c4f7fac6dcdffc1c52760","subject":"Merge \"Improve Process fixture service restart handling\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2ce9ca6cdf065b28aa0c4f7fac6dcdffc1c52760"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2024-06-28 10:08:39.000000000","tz":480},"committer":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2024-06-28 10:08:39.000000000","tz":480},"subject":"Always get local vlan from port other_config","message":"Always get local vlan from port other_config\n\nFor openvswitch security group, due to some extreme\ncase, if ofport is processed once, the openvswitch\nsecurity driver will cache some old ofport informations\nwith different local vlan from current assignment.\nSo this patch changes the local_vlan get method\nto the port other_config, this value should be\nmanaged by ovs_agent properly, we can rely on\nthat.\n\nCloses-Bug: #2071451\n\nChange-Id: I7ad7df72807c95571ef3156c99072852d1c4f494\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/fa339d860d6df80d287aa55cce98e6bdc7281b68"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/fa339d860d6df80d287aa55cce98e6bdc7281b68"}]},"branch":"refs/heads/master"},"8a7506139fa75b4140d4b050d56d0c57b4db3765":{"kind":"REWORK","_number":2,"created":"2024-06-28 10:10:47.000000000","uploader":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"ref":"refs/changes/40/923040/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/40/923040/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/40/923040/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/40/923040/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/40/923040/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/40/923040/2"}}},"commit":{"parents":[{"commit":"2ce9ca6cdf065b28aa0c4f7fac6dcdffc1c52760","subject":"Merge \"Improve Process fixture service restart handling\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2ce9ca6cdf065b28aa0c4f7fac6dcdffc1c52760"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2024-06-28 10:08:39.000000000","tz":480},"committer":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2024-06-28 10:10:27.000000000","tz":480},"subject":"Always get local vlan from port other_config","message":"Always get local vlan from port other_config\n\nFor openvswitch security group, due to some extreme\ncase, if ofport is processed once, the openvswitch\nsecurity driver will cache some old ofport informations\nwith different local vlan from current assignment.\nSo this patch changes the local_vlan get method\nto the port other_config, this value should be\nmanaged by ovs_agent properly, we can rely on\nthat.\n\nCloses-Bug: #2071451\n\nChange-Id: I7ad7df72807c95571ef3156c99072852d1c4f494\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8a7506139fa75b4140d4b050d56d0c57b4db3765"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8a7506139fa75b4140d4b050d56d0c57b4db3765"}]},"branch":"refs/heads/master"},"e4606e0c5490efd480662291ec4c680282a47dd6":{"kind":"REWORK","_number":3,"created":"2024-07-01 03:45:11.000000000","uploader":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"ref":"refs/changes/40/923040/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/40/923040/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/40/923040/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/40/923040/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/40/923040/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/40/923040/3"}}},"commit":{"parents":[{"commit":"2ce9ca6cdf065b28aa0c4f7fac6dcdffc1c52760","subject":"Merge \"Improve Process fixture service restart handling\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2ce9ca6cdf065b28aa0c4f7fac6dcdffc1c52760"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2024-06-28 10:08:39.000000000","tz":480},"committer":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2024-07-01 03:42:22.000000000","tz":480},"subject":"Always get local vlan from port other_config","message":"Always get local vlan from port other_config\n\nFor openvswitch security group, due to some extreme\ncase, if ofport is processed once, the openvswitch\nsecurity driver will cache some old ofport informations\nwith different local vlan from current assignment.\nSo this patch changes the local_vlan get method\nto the port other_config, this value should be\nmanaged by ovs_agent properly, we can rely on\nthat.\n\nCloses-Bug: #2071451\n\nChange-Id: I7ad7df72807c95571ef3156c99072852d1c4f494\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e4606e0c5490efd480662291ec4c680282a47dd6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e4606e0c5490efd480662291ec4c680282a47dd6"}]},"branch":"refs/heads/master"},"ae587c34ab59a5717630eded2fab84413f3c1742":{"kind":"REWORK","_number":4,"created":"2024-07-02 13:12:46.000000000","uploader":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"ref":"refs/changes/40/923040/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/40/923040/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/40/923040/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/40/923040/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/40/923040/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/40/923040/4"}}},"commit":{"parents":[{"commit":"2ce9ca6cdf065b28aa0c4f7fac6dcdffc1c52760","subject":"Merge \"Improve Process fixture service restart handling\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2ce9ca6cdf065b28aa0c4f7fac6dcdffc1c52760"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2024-06-28 10:08:39.000000000","tz":480},"committer":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2024-07-02 13:04:47.000000000","tz":480},"subject":"Always get local vlan from port other_config","message":"Always get local vlan from port other_config\n\nFor openvswitch security group, due to some extreme\ncase, if ofport is processed once, the openvswitch\nsecurity driver will cache some old ofport informations\nwith different local vlan from current assignment.\nSo this patch changes the local_vlan get method\nto the port other_config, this value should be\nmanaged by ovs_agent properly, we can rely on\nthat.\n\nCloses-Bug: #2071451\n\nChange-Id: I7ad7df72807c95571ef3156c99072852d1c4f494\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ae587c34ab59a5717630eded2fab84413f3c1742"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ae587c34ab59a5717630eded2fab84413f3c1742"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
