)]}'
{"id":"openstack%2Fneutron~923391","triplet_id":"openstack%2Fneutron~master~Id16b16113420e9f024e1936f9427824d711b6d30","project":"openstack/neutron","branch":"master","topic":"broadcast-arps-to-all-routers","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2024-07-15 13:47:52.000000000","reason":"\u003cGERRIT_ACCOUNT_16688\u003e replied on the change","reason_account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},"6773":{"account":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"last_update":"2024-07-19 14:30:32.000000000","reason":"\u003cGERRIT_ACCOUNT_6773\u003e replied on the change","reason_account":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"}},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2024-07-19 20:44:01.000000000","reason":"Change was submitted"},"28056":{"account":{"_account_id":28056,"name":"Liushy","email":"liuxie_11@163.com","username":"liuxie_11"},"last_update":"2024-07-19 20:44:01.000000000","reason":"Change was submitted"},"9656":{"account":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"last_update":"2024-07-15 15:50:07.000000000","reason":"\u003cGERRIT_ACCOUNT_9656\u003e replied on the change","reason_account":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"}},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2024-07-19 20:44:01.000000000","reason":"Change was submitted"},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2024-07-19 20:44:01.000000000","reason":"Change was submitted"},"23804":{"account":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"last_update":"2024-07-19 20:44:01.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id16b16113420e9f024e1936f9427824d711b6d30","subject":"[OVN] Add support for broadcast_arps_to_all_routers","status":"MERGED","created":"2024-07-03 12:42:16.000000000","updated":"2024-07-26 17:29:07.000000000","submitted":"2024-07-19 20:44:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":49,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"923391-broadcast-arps-to-all-routers","meta_rev_id":"f615e13a2cbfa52e28ccf03da419ef880d3097ee","_number":923391,"virtual_id_number":923391,"owner":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2024-07-26 17:29:07.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-07-19 20:43:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2024-07-15 13:47:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2024-07-15 15:50:07.000000000","_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":1,"date":"2024-07-15 15:50:07.000000000","_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":28056,"name":"Liushy","email":"liuxie_11@163.com","username":"liuxie_11"}],"reviewers":{"CC":[{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"}],"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"_account_id":28056,"name":"Liushy","email":"liuxie_11@163.com","username":"liuxie_11"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-07-03 13:15:41.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-07-03 13:34:04.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"CC"},{"updated":"2024-07-03 13:55:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-07-03 16:33:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-07-04 06:09:11.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2024-07-05 06:04:22.000000000","updated_by":{"_account_id":28056,"name":"Liushy","email":"liuxie_11@163.com","username":"liuxie_11"},"reviewer":{"_account_id":28056,"name":"Liushy","email":"liuxie_11@163.com","username":"liuxie_11"},"state":"REVIEWER"},{"updated":"2024-07-08 07:26:48.000000000","updated_by":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2024-07-08 07:56:09.000000000","updated_by":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"reviewer":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"state":"REVIEWER"},{"updated":"2024-07-09 10:40:22.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2024-07-09 19:17:32.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2024-07-10 20:43:27.000000000","updated_by":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"reviewer":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"state":"CC"},{"updated":"2024-07-15 15:50:07.000000000","updated_by":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"reviewer":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"state":"REVIEWER"},{"updated":"2024-07-19 09:14:22.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"CC"}],"messages":[{"id":"3a4547cba0e720cb1abe1428624fd12b40c0a469","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2024-07-03 12:42:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"976a007d58a6807ec76c1de1026e00ebb6824c01","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2024-07-03 13:15:41.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/July-2024/Arista_Tempest_Testrefs/changes/91/923391/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"e24b2fd2983d0a66bac2a67ac96493e5d296b499","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2024-07-03 13:32:03.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"49a38b9bb549bc15e79d5054ca8ed32813d1fdb8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2024-07-03 13:34:28.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"d8bcf43734bbc1c55aa6e328c1183b12153ab350","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-03 13:55:38.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a31ce344c54247f09cabc95d6ac41987\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d3b2d38108d549a4ae32d8c16eb047f5 : SUCCESS in 19m 38s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/daefd6aa215148cda82567f6d207f7a1 : SUCCESS in 17m 10s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"f692a6e86f1686b57193950816790122c7753bb3","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2024-07-03 14:03:32.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/July-2024/Arista_Tempest_Testrefs/changes/91/923391/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"2d151fae995a1c5363a31032149216f458e44e06","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2024-07-03 14:06:41.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"d85bcb4ab50f42eda3a665c5611c6907b082d3f4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-03 14:27:09.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4811d7cb0c9d463c9c758a15f999e3d0\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/047fb733bb554e729d23d0abe98ee109 : SUCCESS in 19m 25s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/6c109726f06a43cfb936ca2d6cbca71e : SUCCESS in 16m 58s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"93b91df7beb0531883fabc147c6fd7509c02adae","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2024-07-03 14:36:14.000000000","message":"Patch Set 3: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/July-2024/Arista_Tempest_Testrefs/changes/91/923391/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"897b7d86254fac5c010ac6301a5f1a1f35955b6b","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2024-07-03 15:40:46.000000000","message":"Patch Set 4: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/July-2024/Arista_Tempest_Testrefs/changes/91/923391/4 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":4},{"id":"762fe86e316577a54de697d4ecdd37ee280e2175","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-03 16:33:41.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/595e5e769d8b4fbbb573b1688df702b2\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/8be115578fbe4c719faa49d3ff09186e : SUCCESS in 1h 29m 32s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/c3c94a201f4a46348b606523e641630f : SUCCESS in 1h 44m 02s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/339a2549d90b4032aa22810a80d27031 : SUCCESS in 1h 34m 28s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/f0cc8091960945ad81c1f557d009f66f : SUCCESS in 2h 13m 09s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/7cab7d76f1c8408a8a6d82b8f8f6319e : SUCCESS in 34m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3f1a8574e06c4216a817419d772f2f6a : SUCCESS in 44m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/65ac4e0283444e63b7663727e1b2fd9f : SUCCESS in 9m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b183a4e3deab4b1a9bf0d7370e52221e : SUCCESS in 24m 59s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/85459fee14194d9c9e14e3d821097b1f : SUCCESS in 20m 30s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/33d9e9b0ee034cbb85d0725e711707e2 : SUCCESS in 29m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea84b517e80c439a8740489de58890db : SUCCESS in 9m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7437db1443e644bfa4ea43cb1b5f9c29 : SUCCESS in 15m 29s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/0cbc3f948aa4403c832a3cc570b4fa0f : SUCCESS in 20m 17s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/4cb9cc9eb7d8425393da2ff3c1391342 : SUCCESS in 45m 56s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/2bd2925bdb19431d97f8abfb1cb1ff90 : SUCCESS in 1h 45m 44s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/810d6731666c4906acb61ab71de47e2d : SUCCESS in 56m 22s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/d6bbf68169b5410c9f7583a4dc54eaa7 : SUCCESS in 1h 10m 53s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/87181b9a5a0146548a107e160ce76dd6 : SUCCESS in 1h 39m 59s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/4a82ddabf0994b88ab8f6f44db327e87 : SUCCESS in 1h 28m 51s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/a3e68ff919f240f196a98396b5c28517 : SUCCESS in 57m 49s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/4a14d0c0219941d7ab86cccbd5d4ac66 : SUCCESS in 1h 14m 21s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/a2547db9941d451a8560eed0e24cc8ad : SUCCESS in 1h 19m 51s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/60f1a64737a14b9da0062b4f6a3848b4 : SUCCESS in 1h 14m 28s","accounts_in_message":[],"_revision_number":4},{"id":"e27b263a256c9b1e17bab1c3d94225888a349fef","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-07-04 06:09:11.000000000","message":"Patch Set 4: Code-Review-1\n\n(5 comments)","accounts_in_message":[],"_revision_number":4},{"id":"1668b3bc4c8ef7dc5f903ca2a39e8c08f40920d8","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2024-07-04 08:13:28.000000000","message":"Patch Set 4:\n\n(4 comments)","accounts_in_message":[],"_revision_number":4},{"id":"14ef19bc8d17e1c62dc50f2da045b3fd824ad559","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2024-07-04 09:09:04.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"c96c6f395464b850ae9c3d764cc94b4744ec6602","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2024-07-04 09:30:44.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"4be1bab7cbf70aad789c602c969430ff5880d01f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-04 09:39:55.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/52454b35272a4624ba3a7735cf7b7d3d\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9418ad4c027b47aba41f3991d91cf573 : SUCCESS in 19m 58s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/5b6b81b10d694d72b441ad2d9bb54240 : SUCCESS in 28m 54s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"2b3db6180a91f71e7fb26f7065ae3c68a3ad8476","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-07-04 09:40:24.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"2e983046d8cf92a976baae7a582067d3664f878b","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2024-07-04 09:41:57.000000000","message":"Patch Set 5: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/July-2024/Arista_Tempest_Testrefs/changes/91/923391/5 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":5},{"id":"a77ef6e2334a8cc212f9023be6742551c1679309","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-04 11:49:53.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b4490f4bb2484f649cbe623d4c4e0eec\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/992a8a35526c40e0b575c678d6d1cd5b : SUCCESS in 2h 07m 50s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/4534b8d7653a48d79ac6c86b05e696f7 : SUCCESS in 2h 28m 15s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/34ce9a2dd6d14e11bc8997679a4fbe68 : SUCCESS in 1h 32m 24s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/2c85376757cf497f920e95bce2cf2723 : FAILURE in 2h 37m 36s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/16319925a511498093f239a78dbc1ab0 : SUCCESS in 47m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fffba3a282ca405ab72cee9298dc3ea4 : SUCCESS in 41m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1440883d582d4f1a8369d763c4cc0e29 : SUCCESS in 10m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5c9d11d9989a4288978137f4609226dc : SUCCESS in 20m 56s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8800319c92af4f0ab5d64e319702287a : SUCCESS in 21m 31s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1ce232432dce402d91b5e8b32ba086cc : SUCCESS in 24m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6bcc93ec65d04be4977ee9a0ffb6064f : SUCCESS in 10m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8e5fb7dc62f7473b9db6fe79dc5f8e58 : SUCCESS in 16m 48s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/6e0f2f64188049548327ab5064d54eaa : SUCCESS in 24m 39s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/b0bbcf27439b42f4b822d2501df71f40 : SUCCESS in 47m 58s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/c14c8da903714b30a78ae12b97f921a7 : SUCCESS in 1h 37m 41s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/bd442b153cb24adca57c7faf1b76b558 : SUCCESS in 56m 36s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/d14d19df3fce48c3a99ced997a27708d : SUCCESS in 1h 05m 29s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/a3f6a1d9af53423480e4bcf0fb549273 : SUCCESS in 1h 42m 17s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/1d68ea8638044cbdbdc00fd5b18776ab : SUCCESS in 1h 28m 37s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/5afd2644b41f4acbb98074829d89873d : SUCCESS in 1h 01m 48s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/fc84871467b4422ca16e31aeb9a9e221 : SUCCESS in 1h 15m 27s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/5ce32e29ce6142d8b198189414e5e8df : SUCCESS in 1h 15m 59s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/d43d90cb54504cbaa1845c827423e8d5 : SUCCESS in 1h 16m 14s","accounts_in_message":[],"_revision_number":5},{"id":"6c7fc24fe7a04fd637a46fe368402b62f30fe7c9","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2024-07-04 12:33:35.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"8ac544bb5785ea1597b3f5d19c83a7ce65d85bbc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-04 12:53:35.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b463a5178faa41a1ab6f19e4cb5c7748\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2a5c044f62d54f22942ce7a3876fb4ef : SUCCESS in 15m 56s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/958699ee7cc04235a458af4948ed7e3d : SUCCESS in 19m 09s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"fc423285df40d6a0813d3465ac62648b96ce2b92","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-04 15:15:28.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1ea560b1b7ad43a7bb45ef5df982448a\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/575fad48d6f147cb87632d843be993a8 : SUCCESS in 1h 44m 26s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/4e8b2c797a3441308ef0ed32ed70c57f : SUCCESS in 2h 37m 55s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/e229489c1e7948cc839b8d2aeb2f9e0d : SUCCESS in 2h 08m 30s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/a7374dca4e25443da3e9862ae0443726 : SUCCESS in 1h 49m 45s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/88b9929aa8d545fdbc75d37b1693e494 : SUCCESS in 45m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8388e3ca908146e6b3c4a9d5e68c641d : SUCCESS in 46m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d97d262e0ac34b6a9aa01d0db9889f05 : SUCCESS in 9m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/56197f9b8bde44e39ffbfd1d5da33792 : SUCCESS in 21m 06s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/588bba56c21843869c1d9d6c6c1d0935 : SUCCESS in 20m 51s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/86c12ab11b62494e8b51eb15c2d0b6f7 : SUCCESS in 24m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c2a67f4d4b74f8aa35830e1691169d7 : SUCCESS in 10m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/064ba9f7de1942fcb05fd91fa8a9c9aa : SUCCESS in 17m 23s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/4e0250fabde84c489f56e989ff0ed19e : SUCCESS in 22m 02s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/3d673b2882c240cd92fc667e04e0711c : SUCCESS in 50m 29s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/ad2bbd2d401a432dbeebb0c8052d8320 : FAILURE in 1h 23m 14s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/c82bd806a89f4f94a23269c6e41db9fd : SUCCESS in 55m 46s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/cd224a0c5af049f2944a7dad23f94926 : SUCCESS in 1h 06m 08s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/ae196ea407174fcf888043430d375361 : SUCCESS in 1h 40m 43s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/9daf663eda8c4b249a59c8a6f2fe160f : SUCCESS in 1h 29m 57s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/58ba1fd215284c90a3fff66f4d2c1da2 : SUCCESS in 1h 03m 19s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/120787c0af1e4d80b646a064d5fe5a35 : SUCCESS in 1h 11m 28s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/a063d09ab1974e24a83849a059a066c8 : SUCCESS in 1h 27m 00s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/b0615c5382f546888fb426bce8bf624f : FAILURE in 59m 10s","accounts_in_message":[],"_revision_number":5},{"id":"58539789571cf24c2313928cb71ca0ab540551f8","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2024-07-04 15:26:45.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"e46920c7efe1f7e0a06bc5e38b4feef266cc1642","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-04 16:10:11.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1afface8084b4672a5b7af98e09e1464\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4132b09dc2224000a8897ab7aa708fcb : SUCCESS in 41m 40s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/8ffb850f4aab40c59724fc550fa554fd : SUCCESS in 21m 40s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"ed200ecbb0963d686d7d5d113de61c1dadf1f9f8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-04 17:34:31.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4a3309c0868343e89d7e0456e0e0882c\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/a5108789e4d845e8a466f6bede41ff80 : SUCCESS in 1h 58m 44s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/4178cc1f7d03476f8fb10d9e14697f5b : SUCCESS in 2h 06m 34s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/d95d7ad1f52d49cfbb9b2a1e066e680b : SUCCESS in 1h 25m 18s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/0be7bc8053bb43bdb9ef6d1d9d24f59a : SUCCESS in 1h 49m 22s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/136158c4485a44ec92743caf34952af2 : SUCCESS in 32m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3e25a3bc48a9433b96b847a599d4cd0f : SUCCESS in 50m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7322be67f7514171ae4e2837b3d6df92 : SUCCESS in 9m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5a7355ddcebb4136ba4f3aba2da7cc62 : SUCCESS in 25m 47s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/cb7723bd73484c76a039402456bb7430 : SUCCESS in 22m 35s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0642fa27470043bba01c5a769566f095 : SUCCESS in 31m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/59c357901aec4acc89f210e2a1e56d45 : SUCCESS in 14m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2b92138aba5d48cf989d271cfd5598aa : SUCCESS in 19m 41s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/04263656b7eb4e9bb5be41842171a35f : SUCCESS in 20m 29s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/0b987b93f8814956a5f7842acb12c398 : FAILURE in 43m 59s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/15804895bf904b25baa08d87cfd08b96 : SUCCESS in 1h 51m 00s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/e4543601dd5c4012abf2a324c817a7a4 : SUCCESS in 58m 01s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/a2aa1c44e614486db81797e88e3d87a3 : SUCCESS in 1h 08m 28s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/e71112dcd3a743a1aa2a7ee813fc477c : SUCCESS in 1h 41m 29s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/076a2da485ec4b21bff8fda76e8afcc5 : SUCCESS in 1h 31m 19s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/4f3d13d028ee49a78acbd4183971c923 : SUCCESS in 56m 55s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/43cd9f0b18794e36bd0bafd8150099ba : SUCCESS in 1h 03m 50s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/c31c5363dd824dc381131157d7c72aae : SUCCESS in 1h 17m 42s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/6b7a27c813ae4078a5c456fe783f10be : FAILURE in 1h 00m 15s","accounts_in_message":[],"_revision_number":5},{"id":"12c489220a32ae0cef75a3eefff0c8fb065f9251","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-07-05 06:06:45.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"29b9c41b9246dc5629d55b3e043eb6feaa2f6cd9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-05 06:11:00.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/616c45f466394828bb8b8aa3c46f2ace\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f94e58a9ff5c49f79ea4d196abdfdefa : FAILURE in 3m 20s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/f13a8b76156b4a07841e06900b6e141d : FAILURE in 3m 20s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"ab722ad488ec09a0a214164854f89a96f74e7691","author":{"_account_id":28056,"name":"Liushy","email":"liuxie_11@163.com","username":"liuxie_11"},"date":"2024-07-05 06:12:21.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"7796440b607d94ddf7e9b72d0f952f0494f1728d","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-07-05 06:38:31.000000000","message":"Patch Set 5: -Code-Review\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"6bc68dd10dc29b3336e98d808fa4ae07d071e070","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-05 08:08:55.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/58607f87b493409b8073cb4073321a1f\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/f2b8ab548c074f36bd6faa1537858302 : SUCCESS in 1h 45m 41s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/a84f7dfcf9dc4710ba5cd718b4a915fb : SUCCESS in 1h 54m 00s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/7a07b910c38b431eb8cce17dc467d75d : SUCCESS in 1h 42m 54s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/a39b49ee94874824ac2be8742f2dc39d : SUCCESS in 1h 57m 18s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/53abf1dbf0e24c069f0c1afa58d0bf0f : SUCCESS in 37m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6412ae2bfb124d75acc66186f6a0addc : SUCCESS in 48m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/659382887eef47bfbdfa243cc4c92e07 : SUCCESS in 9m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/44332c236b494983ab0b7bd80a121f77 : SUCCESS in 26m 45s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/51a9112e3b824b4488579f79cf509f20 : SUCCESS in 22m 33s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a18a84f6da3048cfbeb9312847edf453 : SUCCESS in 31m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c4098d8ce78d476a8188da629f030a59 : SUCCESS in 10m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/235288e387724d0dba1201e762b6cf98 : SUCCESS in 20m 20s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/13de2cc6d0c74cd9a778b4e6095cb5be : SUCCESS in 22m 33s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/437cde5c3d864e47bde53959f6832064 : SUCCESS in 51m 25s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/30c29eb0d059432b916b0ee8955e25fa : SUCCESS in 1h 48m 48s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/d0fe089508da49469b5026db38ef2e15 : SUCCESS in 1h 00m 38s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/02d9f075ad9d40d79f33a39850041378 : SUCCESS in 1h 19m 29s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/6507c2b3509e4a0992a2029f29aabc01 : SUCCESS in 1h 55m 18s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/72a058aff4f648b4ba9c4ecfa3786463 : SUCCESS in 1h 35m 50s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/f53ae9598ceb441e9659fdc21c8b53f4 : SUCCESS in 1h 03m 03s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/c874594661104f07b6da965ea9fd4701 : SUCCESS in 1h 01m 54s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/96e541e5f18c4f9594f5e8e30f72f987 : SUCCESS in 1h 17m 43s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/f83529d5e3714c53a3e5e731b20129b4 : SUCCESS in 59m 08s","accounts_in_message":[],"_revision_number":5},{"id":"29808ce270c7a27415a5b51edd03d7e6473f813e","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2024-07-05 08:17:41.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"8e940ff4fe91987a6687510f1217105cd36460c4","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2024-07-05 08:17:52.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"d4018ca51e1205a6beed592d315b51a382bdf9d2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2024-07-05 08:20:47.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"0d3769c7275f1ae3e4c49a43fe53762d1b1a3e2b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-05 08:45:57.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/776dc345987b41e89916df0eb21c72ed\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d89b786050c8460eb0c61a6d37478f3c : SUCCESS in 24m 19s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/28421a6e1fb44323ab20a6a8c9c903b4 : FAILURE in 3m 01s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"a4fc327e01767f14501922981ee97c736eabd938","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2024-07-05 08:54:04.000000000","message":"Patch Set 6: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/July-2024/Arista_Tempest_Testrefs/changes/91/923391/6 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":6},{"id":"7c932b7102b4251b9bf65c2be7fdc7dc166b6893","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-07-05 10:06:23.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"c55427dc25c50094a3b2bd557810968c19f93244","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-05 10:39:30.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c68d12b3830f47bf83ef038a7ead8f00\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/7fe61d408dec4f1fa02f42079cf78a78 : SUCCESS in 1h 34m 46s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/a606de4a348d44a98bbfc80bd6793eb3 : SUCCESS in 2h 16m 53s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/ed81da216f084d508af79530aefe9ba2 : SUCCESS in 1h 35m 00s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/993a8a0e849943a7ada065201bee4968 : SUCCESS in 2h 04m 28s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/82b7ab06fefa4d2b84d7c7d5ef85a49f : SUCCESS in 36m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b4678565722841f7ba5b15e4bc97884b : SUCCESS in 42m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9bd2ff0dbec54207aa80c5768d2cad6d : SUCCESS in 9m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/456c8d2e328144b092842608a34484c2 : SUCCESS in 21m 27s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f0452f912877439b9e81a21ce432b9e4 : SUCCESS in 19m 00s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9fd9633ce3644f0e9c54af1bbaefd545 : SUCCESS in 27m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac190a7a5a6f4c858f70d07bdcc038f7 : SUCCESS in 10m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8ccb6de5a7724db78285f8db29c77dd8 : SUCCESS in 18m 30s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/7e0a13f3c61e422eab27170322bad4da : SUCCESS in 22m 00s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/4f314f7a0d6a4eae85c9e9210392bcf2 : SUCCESS in 53m 04s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/a25f6e55875042e3b7209b7650082c66 : SUCCESS in 1h 44m 29s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/31292c5de55742b4b5f4336f13977942 : SUCCESS in 1h 05m 29s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/7ee4ce0b168b4bff921754c2ccb485f4 : SUCCESS in 1h 08m 46s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/e94508197da6455498149aa3f710a376 : SUCCESS in 1h 44m 05s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/c6c5ac5b277d46d79fe22d75658e396e : SUCCESS in 1h 39m 20s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/8fa5e99cae9b492581e4d135125a53aa : SUCCESS in 1h 06m 34s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/f4835e7503ac4f3cbc3289a59a60b1ae : SUCCESS in 56m 17s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/068e38ac96064b70b5426a7013527a66 : SUCCESS in 1h 21m 31s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/78edfe4f7030466c83ce8e483d6cadf7 : SUCCESS in 1h 02m 37s","accounts_in_message":[],"_revision_number":6},{"id":"5fab03aa6e437299dc507f0ca312eaa9e1fa3a63","author":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"date":"2024-07-08 07:56:09.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"03c8853d7536192532e94cbbae9d4923fab0ea48","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-07-09 10:40:22.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"9953ccc98273ee10d5ca24571087e1dd3f333d83","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-09 10:41:01.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"8edc448aaf3074bac72c917a2d9ab50d6464d5b5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-09 13:53:12.000000000","message":"Patch Set 6: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/114bffae31214b50a1573eb88ba15eb0\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/60208e6ab3f14a44b6e92f91d283bff2 : SUCCESS in 1h 55m 38s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/0599ecbfae64451dae8a30dc091391ce : SUCCESS in 2h 26m 40s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/b2b3cc4aac1d4e4e87cf59bf1c957dbb : SUCCESS in 1h 56m 14s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/89b88e470e284cde805404b5791aba7a : SUCCESS in 1h 42m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/57ebbb7dfca048c89fb9778d443c4d51 : SUCCESS in 10m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3b4f16ebbc4747c88b7a78cba76aa723 : SUCCESS in 20m 31s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c7bd2310d9cc40e29131d43bca632d43 : SUCCESS in 21m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be0821e2bfaa442491407ea9267ef740 : SUCCESS in 10m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fff2357fee69443d86f2e0d1ec618a8b : SUCCESS in 16m 51s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/12bc215500d244ed8bc1134b3618fa14 : FAILURE in 41m 40s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/d7664d9c8d7740ee8e736156869ad2c4 : SUCCESS in 1h 51m 20s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/443957bee4de4d0d811e7ed55f08f1b0 : SUCCESS in 1h 28m 09s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/43d5849895e94c1c8746680f105962eb : SUCCESS in 1h 01m 23s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/a3c774a3e7ac4811acb7ce2c4a8faf74 : SUCCESS in 44m 40s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/97238e48d3514f94954141fea77bc127 : SUCCESS in 1h 16m 11s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/ee1d3a60f46240a5a14680aa3aea53ca : SUCCESS in 1h 05m 20s","accounts_in_message":[],"_revision_number":6},{"id":"67264719d91ef3bfbbc25267ebd17467ad9179c1","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2024-07-09 13:58:30.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"cf64c20aef29f467f878edf1ea641f45e501199b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-09 14:06:01.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ceb4583c6c5546e5948f6258319a430b\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b1e23c26558a40c6a927ab9e9af525f9 : FAILURE in 3m 44s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/8a0f06c7a5df48f881c6a6cbf428bd2e : FAILURE in 2m 45s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"0f1831711bb996b8bef6cdad6b161fcae48cb69a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-09 16:44:10.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4ac6b4106cb346e2b4d36c3223a7f20d\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/6bd6f38f09a64fd1879b2a26bbad2787 : SUCCESS in 1h 46m 32s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/0aa67f1df9b14bc99249b5c3592c5851 : SUCCESS in 2h 13m 26s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/6e4a08b4ff2f40b1ad5193dd376689c7 : SUCCESS in 1h 57m 30s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/3700801f9799461097bc6fcde506e555 : SUCCESS in 2h 11m 51s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/c7fed74aa2364f5293e2e0ad504d3d74 : FAILURE in 57m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5f6bbfcb444a435e9b8fc499d73c42ce : SUCCESS in 43m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/16e2dfe13eed412faee7da47b2e0d3b7 : SUCCESS in 10m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/390261bee5714546a2e718b9c4421db3 : SUCCESS in 24m 40s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/01b3bd73a03a41acb265efc018c24784 : SUCCESS in 21m 57s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/59b2a896a83c46798b2ab5bab2902539 : SUCCESS in 26m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ed0807bb7f04471a9aa11c4289a50ac : SUCCESS in 11m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1c91489ef8a94b0b9835a6de88a55e03 : SUCCESS in 13m 48s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/8f5a01ff9d284ee29639627bb9315e1c : SUCCESS in 20m 24s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/a23ad8b08f9943759007aac57c03d7e3 : FAILURE in 48m 18s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/4697533766ab4cec8cc221a30b47555c : SUCCESS in 1h 44m 18s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/9d36fe8ce3164d81bf68d25b4e65f70a : SUCCESS in 59m 44s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/878453e1232449ffb64b8a200c869133 : SUCCESS in 58m 53s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/c66d07d095294114879a9c64fc0d64f3 : SUCCESS in 1h 25m 22s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/7429144d781340cfb933b770a53cfa9b : SUCCESS in 1h 13m 01s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/577c20744d374f089ead8232780d2363 : SUCCESS in 45m 22s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/80f83ac7cd3043e6a6f82fea2b184494 : SUCCESS in 1h 01m 37s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/0856a083133c4412a64573859bdddf01 : SUCCESS in 46m 22s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/095c03c1b1634c94ae759e41198ee61f : SUCCESS in 1h 18m 31s","accounts_in_message":[],"_revision_number":6},{"id":"3ac2d0aa5783102a153107b24330f60fcb4d36cf","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-07-09 19:17:32.000000000","message":"Patch Set 6: Code-Review-1\n\n(6 comments)","accounts_in_message":[],"_revision_number":6},{"id":"09edbe3897d67f39b7c94cc569952ab5f22d35b6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-07-09 19:21:25.000000000","message":"Patch Set 7: Published edit on patch set 6.\n\nOutdated Votes:\n* Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1, Verified-1\n* Workflow+1\n","accounts_in_message":[],"_revision_number":7},{"id":"8805e182ad1c976e05fcd38eed66ac2a3a1dfe8b","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-07-09 19:21:48.000000000","message":"Patch Set 7: Code-Review+2\n\n(6 comments)","accounts_in_message":[],"_revision_number":7},{"id":"eee5d63eba25a17b5a71b10d04d1ac388a1257ef","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-09 19:26:18.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/874a06496a954a648a45a75be6ad28e2\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7ab7eb2e4e0d4ded86734fddce508a4f : FAILURE in 3m 03s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/deae5903b35f490988a9270258d0ef92 : FAILURE in 4m 08s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"77a4469626cd56ee145e91674268474b1695a5ab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-09 21:21:53.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3fab60ae76de456d890cfd1d7e785031\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/aad61daf2ce44a269987c27f09983967 : SUCCESS in 1h 58m 40s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/7bafa0c63dd7478cba1abd1f6c725891 : SUCCESS in 1h 46m 03s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/02f3536b43bd4fd4bd8b576e5a2be168 : SUCCESS in 1h 47m 13s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/f2a8fa4f89fc4ee698ffb47250c73f03 : SUCCESS in 1h 42m 45s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/c763fa4f13924ba291f589a8449415c6 : SUCCESS in 35m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/73e3f8904fd3443eb7d608f1784b8b90 : SUCCESS in 28m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/da40aedac850414d9424f73568f39e6f : SUCCESS in 10m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f63ba081ee914e52976c370726c45745 : SUCCESS in 24m 04s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/25f212e602c74b22879b126bf6c871e3 : SUCCESS in 21m 37s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/212b33afae084c4ea78ac4998db3b01a : SUCCESS in 26m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2dde8244915a4d69af3a15406d4048b5 : SUCCESS in 10m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5fa0406edd7c40ff8849c694a82585f5 : SUCCESS in 19m 45s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/c7415178575a4cbca62467df0fa8f74c : SUCCESS in 18m 06s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/157db12c38ec449e9cbc7406893bd434 : SUCCESS in 54m 30s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/c089c689daa64bdd8e4f51a33d684c28 : SUCCESS in 1h 42m 52s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/e37ccc912ba74cabb6cb76cbcb584300 : SUCCESS in 1h 02m 30s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/79076160bbea4bf98aa8afe531c2a73f : SUCCESS in 1h 08m 46s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/cb05b16ae23a41ecb7645faa78b03d5f : SUCCESS in 1h 44m 59s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/a4d2a0d53a4041359dadc68a044dfcfe : SUCCESS in 1h 29m 20s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/2a0c8a68aba949548cf3cca2d7a8b1e4 : SUCCESS in 54m 25s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/95fba6a3d7334f2e94a5fd537e57659a : SUCCESS in 1h 03m 21s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/4151c30fc53141f7b61e8c03d0880a78 : SUCCESS in 1h 02m 44s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/aa9842813b0d4fa7a62b69bb1c75652d : SUCCESS in 1h 16m 26s","accounts_in_message":[],"_revision_number":7},{"id":"5d8e5e7ec09a898c3a79abce82f5f4307838226e","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2024-07-10 01:22:18.000000000","message":"Patch Set 7: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/July-2024/Arista_Tempest_Testrefs/changes/91/923391/7 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":7},{"id":"f3ce0c963a215db43ba3b9f073a4f05c9535ff3d","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2024-07-10 08:19:58.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"e3afe9c8160b5ed1eb908e5e4bbe251b8c8d3567","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2024-07-10 20:43:27.000000000","message":"Patch Set 7:\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"6a8c9fffa290c60f9561554df839a8866c856ccc","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2024-07-15 13:26:19.000000000","message":"Patch Set 7:\n\n(3 comments)","accounts_in_message":[],"_revision_number":7},{"id":"0066deba3794a4303b5b026f0c9d4cd1bfc7d738","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2024-07-15 13:26:52.000000000","message":"Patch Set 8: Published edit on patch set 7.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"572567388748a443a8fd427c8f40a3e40349de80","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-07-15 13:47:52.000000000","message":"Patch Set 8: Code-Review+2","accounts_in_message":[],"_revision_number":8},{"id":"62fdeabd1737ac48ed5602ff6898ec56127b2c58","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-15 14:23:50.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5c06ddbba0174c9a87add751c08a05b6\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/bd1752d54cc84de9a813431e9a63e997 : SUCCESS in 17m 25s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/4cf6244fc7814dbd8b0c46b5438915b1 : SUCCESS in 26m 20s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"b248bd66bcacde34312d476e85986eea5eacdbec","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2024-07-15 15:50:07.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"534dedfca5c2f23515a88579a485ad733f7bb1a1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-15 17:17:06.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/bc5b177d17fb41df81b780656cd33763\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/bf358a6d76a54c70818e204a70ae31ac : SUCCESS in 2h 13m 40s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/6078de6ec37a4bd5959b93b3ebe8801d : FAILURE in 2h 10m 53s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/85052b211d2b4d32a10dfb1bc83eb1ab : SUCCESS in 1h 48m 01s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/08e370ab69c649ec9b110c0fdba65ec0 : SUCCESS in 2h 04m 04s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/4ed5cf06f7e44fc19eefdc5a28b2d59a : SUCCESS in 42m 33s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7abda8cfa64844209b5c6826cc832da7 : SUCCESS in 55m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1c8de5b8f31c466e8c094f772a0c5730 : SUCCESS in 10m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a8d76c37d2d042f8ab86c0531a923bd2 : SUCCESS in 22m 20s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e0f81c901af9432786015397e3f2c070 : SUCCESS in 15m 35s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b25b80f9af184f28a030cf9915d463b3 : SUCCESS in 20m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bdd53ee6bd7c402387e2c9194eb58d3a : SUCCESS in 11m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/09c7fb27915548079b0b2ea5c4dbaef1 : SUCCESS in 15m 59s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/4419cce2e6b34fdcb906477d07e1bfe1 : SUCCESS in 19m 04s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/a1a2f208f6c84baeac049353ff96f522 : SUCCESS in 55m 57s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/7eeeba9083394a5f949fa88dbe384a17 : SUCCESS in 1h 44m 22s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/94531efbfd224c57bfc9020177cc1af7 : SUCCESS in 42m 24s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/fe5d2f21f14748c5872d4aeb613be353 : SUCCESS in 1h 09m 00s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/72f2dfec2b524380b7502613df5151f1 : SUCCESS in 1h 43m 21s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/c96f1f4a118c4e94b7a562c7027bc0a6 : SUCCESS in 1h 34m 30s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/2dd90d35fbee4a3b827e1b6ade4b3cb5 : SUCCESS in 1h 01m 58s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/7f5427d93f3c4c6cb67a90c2599af786 : SUCCESS in 49m 35s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/2ed99f3c5a21460bad94eef1a495c9cd : SUCCESS in 49m 43s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/4b8be068dfdf494296d41a14a5bcc13e : SUCCESS in 59m 10s","accounts_in_message":[],"_revision_number":8},{"id":"6f53d0426f24a112d7bf2f262184b6af4a17bbdc","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2024-07-16 07:01:22.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"c2d4fce23c06c272846c5a3d4babc0bc3de701c5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-16 07:19:15.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/916cf308882e4ea2bd900fa80eb3664f\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2cfbd34f31db4adfa1bcbfecaead883e : SUCCESS in 17m 07s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/f8add4f166de4a24ade947e25c586ffe : SUCCESS in 13m 53s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"52ab1edff13f3dc6272d99c288e78c0f50908056","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-16 09:29:54.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4cc7aa8b59fc4b3e912853d08845e8f2\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/4ac0e32ed2f74de68c324bf6360d1d4f : SUCCESS in 1h 34m 23s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/eb3b43f61025470f905c6868962ceee6 : SUCCESS in 2h 26m 18s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/5a2a25d0254b454683dfedd474c6d00f : FAILURE in 1h 46m 54s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/f2e6af1923b648b588993c8e4da1e3d5 : SUCCESS in 2h 07m 21s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/18f4e76715704ca3ae4f187a98326d0a : SUCCESS in 38m 35s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e75d3c262e9f4e328238fdc5ced21439 : SUCCESS in 35m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4f5ef639e6744c79af6ead5ce9555fa2 : SUCCESS in 12m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b6d1341459c94d9286acec0230af0ff1 : SUCCESS in 18m 59s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c127b6c787d1471abe22d861c1fa428f : SUCCESS in 16m 24s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c4c4842925864346b2cfbcc714406325 : SUCCESS in 29m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e503df018afd48168e562f13f8f2c17a : SUCCESS in 10m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/846995fbca6d4256bf27e45d1b9c6174 : SUCCESS in 18m 00s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/0134b0f092d24ba4accb9a9b32200961 : SUCCESS in 17m 25s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/b4cbb9ed3fae4a048c5227712eca7f20 : SUCCESS in 44m 52s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/302401174c2342318fc4af631d6b437e : SUCCESS in 1h 32m 50s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/9531210b5ae947c7aeac33737ebebcb7 : SUCCESS in 59m 43s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/ebd7dcd1bfe64d3091ec736a14babb9c : SUCCESS in 1h 10m 22s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/5119dfe2cdea44608d3164b52184f04e : SUCCESS in 1h 31m 01s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/762653804d0f47b7b465b3f3189c3c5a : SUCCESS in 1h 21m 35s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/812d3542f02646709a27ce9d13b483c5 : SUCCESS in 1h 04m 55s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/4c9a352ffa5043409a3f7da79516b771 : SUCCESS in 1h 00m 07s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/d5a2831ec5b344b6b245518ef939e5e2 : SUCCESS in 1h 12m 13s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/5fbd4ca563ba422c91f6c81ec25f830e : SUCCESS in 1h 12m 04s","accounts_in_message":[],"_revision_number":8},{"id":"5a070e2a6863ee5c533a50106ba5c20dcbcab0b2","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2024-07-16 09:32:15.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"5ec0ee5a629e638638e54caa95da774e1d7e310a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-16 10:06:28.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e7aeb5d1311c4716836a8c30fc306714\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/229f406177a44e95a5beca543a98a141 : SUCCESS in 31m 57s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/751d045c8feb4f519498d586251b7f48 : SUCCESS in 27m 11s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"5873a8b0d7e082c3df18d4ed2329ca7f26cd3247","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-16 11:54:05.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d2138b34ac584858b7dd1030f2366546\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/5fb6c8abc61f49eebb22ed76577d35f8 : FAILURE in 1h 40m 52s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/0bfe054456474a0ab8d4cd0b0d9fd77d : FAILURE in 2h 19m 35s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/22cb54b8507f4fff98de28b1996823eb : FAILURE in 1h 34m 23s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/94fc8ec99487434198da335ca9d26014 : SUCCESS in 1h 53m 59s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/bd2aee07fd99438d874a9e4c2ff49884 : SUCCESS in 33m 12s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a2e178a018b04ac0be778c405eaca6a5 : SUCCESS in 47m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/08835f34a6244f80a20aebf9ff6c39ce : SUCCESS in 9m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/adb37b0f0b0f4ef7ac56650c3b2c8d28 : SUCCESS in 20m 41s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/927c1340c7294d8d90cbcc6bdd0fa659 : SUCCESS in 20m 44s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/927357ad1a9141c49deb0a7f8a0d7ae8 : SUCCESS in 23m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1cff3bcf332c499cba105cf3be0e1b10 : SUCCESS in 9m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e148d94b0a2e4ff580fb7fc9a577f65b : SUCCESS in 20m 03s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/a324e1f3ed4b4f6f824ae20330ba4996 : SUCCESS in 21m 39s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/e0feeadeb1f5453e98ebf682c666fd12 : SUCCESS in 48m 38s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/92ff53e498824362b6445a70a8ff8f0f : SUCCESS in 1h 32m 17s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/72e5676ed61349ab8bf0500511fa14a0 : SUCCESS in 59m 27s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/3444988f5952454f9a54e927f27d06a4 : SUCCESS in 1h 30m 31s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/ea3f3e597df94ec589b77a89c3524a5b : SUCCESS in 1h 21m 42s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/5f0d87c227fe4bbb94153036bc442198 : SUCCESS in 1h 17m 45s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/574d822fb474414894758b01052e3d96 : SUCCESS in 59m 19s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/980f96a96f5845a7ac1a581170e4c853 : SUCCESS in 59m 15s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/ea6e98d1f1e74099ad71c3ca87e08bd4 : SUCCESS in 52m 22s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/c57184e4a91e45c7979e81799a5fae70 : SUCCESS in 1h 07m 25s","accounts_in_message":[],"_revision_number":8},{"id":"c84f5881f4d0df98f02521baa1ff43ba6e2f8823","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2024-07-17 07:34:52.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"fac030b9ba3cdee377229585f12bcb85d8fb8f5a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-17 08:02:36.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/74e4155129d049a0b739c6358e8b06cb\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1faa9147cfcb4d4faa0cc4aec7c40121 : SUCCESS in 15m 42s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/31a1301effad41e286e9a1009eb1c81d : SUCCESS in 26m 50s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"0ec83f3481afcc61818ebface1cdf320202045e6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-17 10:00:48.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/53f921bbdda04b88b516f2fa564bf52e\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/7388481f9b114d88969ea255b0135114 : FAILURE in 2h 02m 21s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/46b1da2c1d2344b19b03261511aa5f0a : FAILURE in 2h 10m 08s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/90735127818844d5b1221f863db07eda : FAILURE in 1h 56m 41s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/46c63c83ee9e42a49685c38cbb1ed2eb : SUCCESS in 2h 24m 00s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/8904e6fad6324123bf03b2146c7af2cc : SUCCESS in 37m 51s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1543bb3cf53d446ab4e897fb8a3836cb : SUCCESS in 35m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/24c127248c77434fb8846c6b27db9173 : SUCCESS in 9m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/02b6884bb4714cb88166f915c1560e84 : SUCCESS in 20m 42s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/823012d362ca41158ca3d33971ba898a : SUCCESS in 22m 45s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/cd16594a57ee4f74b02e9571f619cfbc : SUCCESS in 29m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a69bbd9f21864c99981755be5ff111d4 : SUCCESS in 10m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/52506a71e27d433eae39b870a0f10f9f : SUCCESS in 13m 49s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/2dce5004f83148baacf9817a4190d805 : SUCCESS in 21m 44s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/5ed0da9ec092455e8c24b8f70a48d5c6 : SUCCESS in 55m 17s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/1fc429b506bc4d1d8dd8d2b941931039 : SUCCESS in 1h 46m 34s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/81e7e14e3d7c447e980dffe2ee705696 : SUCCESS in 51m 21s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/49e9490c1bbd4614aebda0015f1e1981 : SUCCESS in 1h 10m 15s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/c522b7ce9d8140c9b9545d31aa32c9c0 : SUCCESS in 1h 37m 06s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/91337d32412f4c5e94d093dc5441f12a : SUCCESS in 1h 25m 33s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/c8ac1948191b4053857695587d53afff : SUCCESS in 51m 48s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/05f98a28df254b6da173879f9abffac7 : SUCCESS in 55m 02s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/bd43ec0751b0471797e58ec8d0d8285e : SUCCESS in 1h 00m 38s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/fe3f5c60ddb74bfa9b571543924c6dff : SUCCESS in 1h 05m 03s","accounts_in_message":[],"_revision_number":8},{"id":"99f0fbc57430863079fd1ddf5a93ebf22265aaca","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2024-07-17 14:59:27.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"9565f51195950636a99527de34bd2706a82908f7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-17 15:17:14.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0ceffe00bb874fccb7f87a0eb6652da5\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9597747499d64b41b34210517dbc8dd7 : SUCCESS in 16m 59s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/7fe79b68e935404da4a04a73880b5cba : SUCCESS in 16m 25s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"f66bfdf473afa536252f8b98ca3a5f856d95d9f8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-17 17:02:45.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7c8e54926b514463a1d9ff86bc77e504\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/44f6509522fd45eaa26e35078443a6e3 : SUCCESS in 2h 01m 22s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/295017eb7981406eb85b73d9d93da899 : SUCCESS in 1h 55m 06s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/04394b23b292448385414c007d9a57b1 : SUCCESS in 1h 48m 02s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/dd475a4b68a84d0bb682c3c918754e4c : SUCCESS in 1h 55m 37s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/2be9050442cf4da1a7efd7826b5b45c4 : SUCCESS in 38m 21s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dc94076824e54b7bbe09114beb14d245 : SUCCESS in 25m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6aa207dea67d46c890096afede601775 : SUCCESS in 7m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c7c123e3cd3348a98b2fbdeac43829d7 : SUCCESS in 16m 30s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9f279e172ffd449c966fd7282023fc4a : SUCCESS in 16m 41s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1b2d38c89f2044d6ae3e4fd563e603f1 : SUCCESS in 29m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4018814b2bbe448ea4031b8706be5c5b : SUCCESS in 10m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/51636aa7e1484fb295b1e17a4dc889f5 : SUCCESS in 13m 39s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/9e0bef29501745f1b6b8391f8c9aa316 : SUCCESS in 19m 28s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/6b0ac48123484c87bda4becb496d0f14 : FAILURE in 39m 22s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/7c8873668f2e4664aa9af63eb4a9e46c : SUCCESS in 1h 40m 32s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/5635355c5ad444948ff72f4e0f653048 : SUCCESS in 1h 00m 22s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/492139337dbd4e72b1d49e6f92d8fd10 : SUCCESS in 1h 17m 53s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/a38c366eedab4e8a896359f84b23bdea : SUCCESS in 1h 33m 11s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/9ace408c33b8468087481d08799a1a7e : SUCCESS in 1h 33m 37s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/d753df5fdd184b61a068bfb20b117662 : SUCCESS in 47m 01s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/38a86a2d11c8439883b8d2d659e85ee6 : SUCCESS in 1h 02m 20s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/7b17b76ddd8f4b889e1b7c298c7541aa : SUCCESS in 58m 42s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/ec6f8a0ed7c8443caba0e917968d8782 : SUCCESS in 56m 09s","accounts_in_message":[],"_revision_number":8},{"id":"fce371a55e55660cf272f9da5ca28d00fa0ccf1a","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2024-07-18 07:46:36.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"3b93d5aca876e743a0aa2a24c7330be38b2ff22b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-18 08:09:55.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f21f316250b54ea9acf7a69ae60210e3\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/41fbebb4aa774fc7a0c4edca23df65b2 : SUCCESS in 22m 29s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/f58d7f1391a54705b6578b081654551a : SUCCESS in 15m 12s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"ace39346c752764ee1fc8cfd99d17607375ce674","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-18 09:57:43.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6b8250b8b5ca45a4a366518c36f81c48\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/3caf3cee08e9448d988bb07b43607f79 : SUCCESS in 1h 45m 07s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/378cdd044c634e6ba556b2ac5526984b : FAILURE in 2h 09m 54s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/41bc4ed9739949d591353841dbef5254 : SUCCESS in 1h 22m 50s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/f4d3598d1e8f41009852e56054eec80f : SUCCESS in 2h 01m 21s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/a3d5c3cef10f418a95aa1f11a06fb714 : SUCCESS in 35m 57s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4dd3c9618ae34b049771cca74727c11f : SUCCESS in 28m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c104f3e852164cfb881cf4b13cb54b75 : SUCCESS in 8m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fafaea1ddbde486ea6c166069b59d8ee : SUCCESS in 20m 03s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/cdf78073ebf44c6b877a5e1fd7af2750 : SUCCESS in 16m 06s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c378fbc86182466b82576836ccc55b76 : SUCCESS in 27m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/04accd7ec00245a28face3c5d9d13f37 : SUCCESS in 11m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4cea53be418d489d93a152fe2c1725e3 : SUCCESS in 18m 17s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/35072447c9b54683a1788e4830a2389a : SUCCESS in 23m 20s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/c477c3aba9694ac181f6ef8aa9dee440 : SUCCESS in 51m 57s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/f7306ee222e24b37a8dce02be35668f4 : SUCCESS in 1h 48m 16s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/a56dd22603fd4f81b30d10befe6b7b8f : SUCCESS in 57m 07s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/29fdfec57d8a4a4299873b9240db8f6a : SUCCESS in 1h 04m 11s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/7dd13916d0a24086b7d1bea71ec1ed0c : SUCCESS in 1h 39m 04s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/667db86fe9b44cd3ac1634f8ddd920b0 : SUCCESS in 1h 22m 01s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/bfce6062fa594e5fb045c86bb2e5eed5 : SUCCESS in 1h 02m 03s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/2140dbf8e4e34a549337577b1374f1a0 : SUCCESS in 54m 52s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/febb50e9ca1244439c1fa23fe3376d1b : SUCCESS in 58m 52s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/053c11c0da7b45aba5fabffc987dd033 : SUCCESS in 1h 04m 30s","accounts_in_message":[],"_revision_number":8},{"id":"48b3c069dbfc19abdec2f5a9074c6fe65ab62340","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2024-07-18 12:17:49.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"b41908909a7ea389ed07f989316522f0c85090c2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-18 12:34:14.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/61dae2fd3e634278b5b90306f5fdcb12\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b55b2b115ba649e09f4660acfcd06d2c : SUCCESS in 15m 35s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/7e2d91a29a0543959992609097af360d : SUCCESS in 14m 19s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"7a34921017c75113cb20e0e5521e3e4e328330c7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-18 14:32:55.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7a539df081b54520b99965e57d46938d\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/ade4d0d454ea4ea7a23de23aa4102c87 : SUCCESS in 1h 38m 48s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/cbc18823e9964e36b1064d770a9b0ca2 : SUCCESS in 2h 12m 31s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/c9f5e829b1ba4404956fe8db26009528 : SUCCESS in 1h 31m 56s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/0dc761e9b13e45aeb2ec320666d8acd8 : SUCCESS in 2h 07m 34s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/dfc6cb842e1c496f87478ee295af87ff : SUCCESS in 32m 02s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/08218fec65c94be6a004550a01111cde : SUCCESS in 33m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1b15438393c045f9a9cda1945f95c3f7 : SUCCESS in 9m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/78fba96db5e546d79adee3b3aceaa056 : SUCCESS in 23m 48s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b2ebdc06620b4deca4078b83ea453e57 : SUCCESS in 20m 09s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/74c5b221c3424226b8b80a496be98058 : SUCCESS in 27m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/20f01354ffc547d49597cf0a306557d2 : SUCCESS in 9m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b895aa4fb45b4a85a9dfd0e0d43b06a7 : SUCCESS in 19m 12s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/0080a39f18124c6ba917ae0b6db2d7ef : SUCCESS in 23m 14s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/6111a03219b84569bce167680021d89e : SUCCESS in 43m 23s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/c3e9e676efc440f48906db132b95b801 : SUCCESS in 1h 48m 03s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/43997c6991dc443d923d6345d0447c0b : SUCCESS in 46m 49s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/444779e5dca8431b9f95354364764390 : SUCCESS in 1h 06m 22s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/cc60d34504094d1183055b76e20d72b6 : SUCCESS in 1h 24m 36s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/c520aed9a33944c88cae6840ac7a2cc3 : SUCCESS in 1h 19m 36s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/82403ffaaf3a47c18b08b0f5b32de3bb : SUCCESS in 1h 03m 34s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/d276e17ee293485882401eaece4dedf9 : SUCCESS in 59m 47s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/c908dc51f7204a3fa7bce659a5bd55f0 : SUCCESS in 1h 05m 04s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/3752732fd2424d76a8cccf70b5da7039 : SUCCESS in 1h 13m 30s","accounts_in_message":[],"_revision_number":8},{"id":"511d5f1c3d2616ff43a2cc9e6ed7e433dfc6b9f5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-18 14:33:23.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"1f797d9ec5aa712600ea824ff90905b03bb571c2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-18 16:54:07.000000000","message":"Patch Set 8: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f16fe2e7e0e947fe99aeeb0b488145fa\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/1fd4eacaa6e04f5d92070a733275837b : SUCCESS in 1h 32m 58s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/8ca4318cac3d4ad1a79660771a09c86a : SUCCESS in 2h 19m 47s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/5416875b8c2c4f128505bbb5a0a9f0ec : FAILURE in 1h 44m 39s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/6713ca4ceb3d40e7b5d24f990d52df9d : FAILURE in 1h 37m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/99135206e63d431cb854342c759061b6 : SUCCESS in 10m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/52270bd4f02d4b8c88069d8f7e0b2148 : SUCCESS in 22m 16s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d532edc3289746058f6324856b9d2a97 : SUCCESS in 14m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/24f99584e4594112becb990d98260f4e : SUCCESS in 10m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9b94abb1c9e04db88177a225719cd815 : SUCCESS in 18m 00s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/1787e26bdda94475bb5848e8a581f9dd : SUCCESS in 56m 40s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/c61b7e458934486d8b1a574fb6871285 : SUCCESS in 1h 55m 24s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/20af4b3d2999463cb232977c658f0772 : SUCCESS in 1h 26m 17s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/46ee8a14a3b14d65a7009da97e4539d2 : SUCCESS in 1h 01m 18s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/b2b7bf1d809647bab122300315fce35b : SUCCESS in 59m 54s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/1cba01fcfc6a4eee9472f517ff78ea66 : SUCCESS in 59m 22s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/2b36368e8928434ab94264f73e321049 : SUCCESS in 1h 26m 55s","accounts_in_message":[],"_revision_number":8},{"id":"4b1f4f1ceb181988c0ca9ea6075c24a2bb72eb00","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2024-07-19 08:05:59.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"a63a3de8c47a3c7dbebf93194686d419383b7f22","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-19 08:24:42.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1f09590c67964ac3ac4fa1a1dbb32419\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/96dca3370f8543e6b481a75a0c817c87 : RETRY_LIMIT in 2m 30s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/8b7f6eb5516f4f09949eb1529ad882bf : SUCCESS in 13m 39s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"66f002dd9bc4b5331da04bbb62bbbb1dfe7c9a84","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2024-07-19 09:14:22.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"0a45d88df458695903f77454d2b075d071eeb54c","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2024-07-19 10:03:52.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"b482631529e4b46ac349acd38a3937c856887c3f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-19 10:18:31.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ac103d7020394b19a1faa2f9b8246122\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/a6c2a5ea89ad413981e2c9af1e563596 : SUCCESS in 1h 33m 00s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/f874e5dd4356405cbc92064d32b1ff46 : SUCCESS in 2h 01m 24s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/920a6a6050c4461f895a63edf3efcb98 : SUCCESS in 1h 42m 39s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/ae038ebc1d5641e48b18395ceef4c1a1 : SUCCESS in 2h 09m 58s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/650e6e8230a34019a88a16ee41685838 : SUCCESS in 36m 31s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5403ae442b3246bbbb88dee23546621f : SUCCESS in 44m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/42e6d4b2ff364d11bfcac90849380a75 : SUCCESS in 11m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4dfc5d91877d4ae281e3bb59f378c973 : SUCCESS in 27m 20s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/39e54629756c4ac29a31dbaa18763241 : SUCCESS in 20m 03s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/cf58a8f0a61e42688a6a02739cede360 : SUCCESS in 28m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b10d70d6282462db43ee987826df54d : SUCCESS in 11m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7abefa9fb63943348c2e780b9a9102f5 : SUCCESS in 19m 33s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/827c32d937c04df595ad7c7047da38ac : SUCCESS in 16m 54s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/d874c34484fb4e0eb9b4a63bb5e205ad : SUCCESS in 1h 10m 12s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/fdb13522e0164d769689a9c6df0b7dc0 : SUCCESS in 1h 46m 27s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/2306f54ac88b4fedaace20a76a66f3f9 : SUCCESS in 1h 15m 56s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/aca6f8e5bf704f79ac1b80aa7cbc4221 : SUCCESS in 1h 11m 20s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/574d4d0160d04199abb7c539739978e5 : SUCCESS in 1h 19m 11s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/b51a95c43d5a49a7a5dea74015fc62c4 : SUCCESS in 1h 20m 36s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/e85ab8f1caf54f84acb9c4c43c54f2fa : SUCCESS in 43m 57s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/961f9b7733d0422e93a11fa65047d926 : SUCCESS in 59m 58s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/f6ab6da8c3dd4a128ede85d5b375c4d4 : SUCCESS in 59m 49s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/50e985bcfcf24292b776ddbf264b0340 : SUCCESS in 1h 12m 05s","accounts_in_message":[],"_revision_number":8},{"id":"5bad79f2a649fb11a0a60e9d10749c7dd22689d2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-19 10:18:59.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"2dc8d6aa9327f01840eacd2e78b04209a1af3e1e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-19 12:09:48.000000000","message":"Patch Set 8: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2d4bda37ac744ad7ad5b95b75cccb554\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/04a88cb1cd0a4a4888158bb7dec82d6c : SUCCESS in 1h 34m 25s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/0c264d6e7ab44de6b5f144b5d97b7e97 : FAILURE in 1h 46m 20s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/8f695458030a4290bdce8d1eebea5be7 : SUCCESS in 1h 37m 44s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/c2afd50b01bd46019b5a91b5d0b46fbf : SUCCESS in 1h 37m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e7e7d6b2b83d49cd95664c1db5b166b4 : SUCCESS in 10m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c89480c7fd224470a4927ae91b151b5e : SUCCESS in 23m 36s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/dce5236157a04e9c974725e42b582dd9 : SUCCESS in 18m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0c3be24c8b6c4c1094d1f8da6d2c6abd : SUCCESS in 10m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8ce7da13a4b04e3da18664b89e2ae1ce : SUCCESS in 18m 47s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/82aef7d1978246e8a59eb2129885e8c7 : SUCCESS in 52m 30s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/e53a6718681846e5882812a8ad8acf9a : SUCCESS in 1h 45m 58s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/de85311cc4a1423b815731449af77ce0 : SUCCESS in 1h 37m 47s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/a7d5eafc362b4cedb651a95c2430569c : SUCCESS in 56m 26s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/17c40fedb14d4f849f297f4dff92ca8f : SUCCESS in 59m 49s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/f9af7064ff6943aa89f5ee523237c315 : SUCCESS in 51m 20s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/c9e74cd400cd40fdbf9b52463cf4eb9c : SUCCESS in 1h 18m 32s","accounts_in_message":[],"_revision_number":8},{"id":"a34c531fa2da5532f5883a5034aab6e82d3c9a97","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2024-07-19 14:30:32.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"2f198a47c709bbb2892adfbcc8f420787d5b1260","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-19 14:56:17.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1fd3ce5e995a45489cb1cb30167c94bb\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/260051664e3a4dae9d6bffcfe3234341 : SUCCESS in 17m 31s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/16e31d9ec1af4ae5af19ee608bd32ca2 : RETRY_LIMIT in 2m 41s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"d4d668fa4e89023e02af17d6a0a53eb0f0511e85","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-19 16:58:55.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/55babeae34ba486987936d9b4b3ce1d5\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/dbcd3c6e9a77449ba1f42cc2a8712dd9 : SUCCESS in 1h 32m 42s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/5f754f90ec3f4b77a1dbb93b14a188fe : SUCCESS in 1h 38m 51s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/766cd6febc4a43b18f5c378aa1b2d111 : SUCCESS in 1h 59m 13s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/baedde8aca9240d6a28e495178ea4a2a : SUCCESS in 2h 25m 50s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/fccc784674754776aa64adb88e958e6b : SUCCESS in 36m 58s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/aeb746cc001449639ea15581429cd8f8 : SUCCESS in 47m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/56e4ee49da2b4df48b3b9de88e26d4e2 : SUCCESS in 10m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f8b819b90d27439fbaa367434fe69d11 : SUCCESS in 28m 41s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/62b869c02d244737a0fe2fed17030f01 : SUCCESS in 18m 44s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0b816d8e589a4624a474ce9645560e1e : SUCCESS in 27m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f6ec06504274152ad9d5092a482f943 : SUCCESS in 11m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a2336227ec18419f882e75e522070960 : SUCCESS in 20m 09s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/be242009d3664b259456878a43b7195e : SUCCESS in 22m 32s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/5fd8efa037db405e9c963f36ae8ce2b7 : SUCCESS in 52m 45s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/dc099cbbd3004bbeab2c7c12b2fc5972 : SUCCESS in 1h 38m 49s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/1cd0b9db9a5f4a1ca8b0b85d0106a68c : SUCCESS in 55m 33s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/8eeed727771c4e1c85e9f276c3eae61e : SUCCESS in 53m 56s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/22079aa9e8a64601bbd9e161f4c31fa6 : SUCCESS in 1h 37m 57s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/1a26dd066fcc4b48a944644f35c2b238 : SUCCESS in 1h 27m 22s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/0edad2dbaa8e42cba12771951f526577 : SUCCESS in 53m 59s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/6d1224105b7245b59cf219261fc851eb : SUCCESS in 42m 45s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/d09bbe68766145899d231c065219a433 : SUCCESS in 52m 57s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/19b1df5a32f240bc85c2861fee6add9a : SUCCESS in 1h 25m 56s","accounts_in_message":[],"_revision_number":8},{"id":"8b0f6d5fce270b32831da96487f368a6d7b52e13","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-19 16:59:26.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"a4dfb165c8cce78aae98bfe44d162183bf9268d9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-19 20:43:59.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/345aad0130f243f4ae45839c2b338f8f\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/e4e9a30772054cc1b46b46c1cd123b35 : SUCCESS in 1h 44m 10s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/24a90ff4976849429f33e668e1dbd5f0 : SUCCESS in 1h 46m 35s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/b0dd03a45af04b6abe504c30381f9300 : SUCCESS in 1h 57m 45s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/46567c9b0dcb41728f3e76ccc791ae7b : SUCCESS in 1h 39m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8d7b01dac4c04896a916eebce7ae5b2e : SUCCESS in 9m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/00829d60354b4de7b6c401cb02cdf55a : SUCCESS in 26m 06s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a54e91cdb4b549f09c396b06ff3a60e3 : SUCCESS in 15m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/95a2878e92544e079d65031830ba6bf4 : SUCCESS in 10m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d550509d9a9c495da74ac9b0b79c0b1c : SUCCESS in 21m 22s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/0a4e0ebea1c343829fa793b26b3901f6 : SUCCESS in 59m 14s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/efbe77d3dac54b11bac9075767878aef : SUCCESS in 1h 23m 40s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/cd7ad606e5754a149838dfac7e6991a8 : SUCCESS in 1h 41m 33s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/3512d89c85314b54963749af48f344a5 : SUCCESS in 53m 43s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/410234fcffc24afaa091964a9acf4f64 : SUCCESS in 43m 08s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/f02b90d00476427483371c96d18de2cc : SUCCESS in 48m 00s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/7767fefe3e6c4ab59df0db38d2fdea34 : SUCCESS in 57m 39s","accounts_in_message":[],"_revision_number":8},{"id":"1f0339b2a2c932dc513e2fff87f5b84fe9f15d9d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-19 20:44:01.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":8},{"id":"e196bc46f4e3ef3538a46d03ae15f5d34e21dfca","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-19 20:45:56.000000000","message":"Patch Set 8:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d117ad86b070461689ea8420bf0e2ffa\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b928c836c4754ddb87b0be809bbaff86 : SUCCESS in 1m 36s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e4389f1e39d64eff8da8edbfa3c9fca4 : SUCCESS in 1m 21s","accounts_in_message":[],"_revision_number":8},{"id":"f615e13a2cbfa52e28ccf03da419ef880d3097ee","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2024-07-26 17:29:07.000000000","message":"Patch Set 8: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/July-2024/Arista_Tempest_Testrefs/changes/91/923391/8 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"f651b28c4234ab7d20ecae7d12b47cc090b92ea9","revisions":{"e2b58c2c9d0c78090e48a409e49a85e84f23c86c":{"kind":"REWORK","_number":1,"created":"2024-07-03 12:42:16.000000000","uploader":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"ref":"refs/changes/91/923391/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/91/923391/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/91/923391/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/91/923391/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/91/923391/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/91/923391/1"}}},"commit":{"parents":[{"commit":"9649251f4db3e74a4337326f8295fff3554bb216","subject":"Merge \"Add the port \"fixed_ips\" information in the DHCP RPC\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9649251f4db3e74a4337326f8295fff3554bb216"}]}],"author":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2024-07-03 12:39:18.000000000","tz":60},"committer":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2024-07-03 12:41:17.000000000","tz":60},"subject":"WIP: [OVN] Add support for broadcast_arps_to_all_routers","message":"WIP: [OVN] Add support for broadcast_arps_to_all_routers\n\nThis patch allows ML2/OVN to configura the OVN\nbroadcast-arps-to-all-routers setting on provider networks.\n\nChange-Id: Id16b16113420e9f024e1936f9427824d711b6d30\nSigned-off-by: Lucas Alvares Gomes \u003clucasagomes@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e2b58c2c9d0c78090e48a409e49a85e84f23c86c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e2b58c2c9d0c78090e48a409e49a85e84f23c86c"}]},"branch":"refs/heads/master"},"a4171b09bfd690f8887c320118fd783cf4268358":{"kind":"REWORK","_number":2,"created":"2024-07-03 13:32:03.000000000","uploader":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"ref":"refs/changes/91/923391/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/91/923391/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/91/923391/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/91/923391/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/91/923391/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/91/923391/2"}}},"commit":{"parents":[{"commit":"9649251f4db3e74a4337326f8295fff3554bb216","subject":"Merge \"Add the port \"fixed_ips\" information in the DHCP RPC\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9649251f4db3e74a4337326f8295fff3554bb216"}]}],"author":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2024-07-03 12:39:18.000000000","tz":60},"committer":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2024-07-03 13:31:55.000000000","tz":60},"subject":"WIP: [OVN] Add support for broadcast_arps_to_all_routers","message":"WIP: [OVN] Add support for broadcast_arps_to_all_routers\n\nThis patch allows ML2/OVN to configura the OVN\nbroadcast-arps-to-all-routers setting on provider networks.\n\nChange-Id: Id16b16113420e9f024e1936f9427824d711b6d30\nSigned-off-by: Lucas Alvares Gomes \u003clucasagomes@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a4171b09bfd690f8887c320118fd783cf4268358"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a4171b09bfd690f8887c320118fd783cf4268358"}]},"branch":"refs/heads/master"},"d651be8b9e286e2daf8f2a26a81dd567ed9da32f":{"kind":"REWORK","_number":3,"created":"2024-07-03 13:34:28.000000000","uploader":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"ref":"refs/changes/91/923391/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/91/923391/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/91/923391/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/91/923391/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/91/923391/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/91/923391/3"}}},"commit":{"parents":[{"commit":"9649251f4db3e74a4337326f8295fff3554bb216","subject":"Merge \"Add the port \"fixed_ips\" information in the DHCP RPC\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9649251f4db3e74a4337326f8295fff3554bb216"}]}],"author":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2024-07-03 12:39:18.000000000","tz":60},"committer":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2024-07-03 13:34:21.000000000","tz":60},"subject":"WIP: [OVN] Add support for broadcast_arps_to_all_routers","message":"WIP: [OVN] Add support for broadcast_arps_to_all_routers\n\nThis patch allows ML2/OVN to configura the OVN\nbroadcast-arps-to-all-routers setting on provider networks.\n\nChange-Id: Id16b16113420e9f024e1936f9427824d711b6d30\nSigned-off-by: Lucas Alvares Gomes \u003clucasagomes@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d651be8b9e286e2daf8f2a26a81dd567ed9da32f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d651be8b9e286e2daf8f2a26a81dd567ed9da32f"}]},"branch":"refs/heads/master"},"1d9351d6510e7aad312512a88b326f317aa4326e":{"kind":"REWORK","_number":4,"created":"2024-07-03 14:06:41.000000000","uploader":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"ref":"refs/changes/91/923391/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/91/923391/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/91/923391/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/91/923391/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/91/923391/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/91/923391/4"}}},"commit":{"parents":[{"commit":"9649251f4db3e74a4337326f8295fff3554bb216","subject":"Merge \"Add the port \"fixed_ips\" information in the DHCP RPC\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9649251f4db3e74a4337326f8295fff3554bb216"}]}],"author":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2024-07-03 12:39:18.000000000","tz":60},"committer":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2024-07-03 14:06:32.000000000","tz":60},"subject":"[OVN] Add support for broadcast_arps_to_all_routers","message":"[OVN] Add support for broadcast_arps_to_all_routers\n\nThis patch introduces a new configuration option for OVN\ncalled \"broadcast_arps_to_all_routers\". This option is\nresponsible for configuring the provider networks with the\n\u0027broadcast-arps-to-all-routers\u0027 configuration from OVN 23.06 and newer. By\nenabling this option (default) OVN will flood ARP requests to all attached\nports on a network. If disabled, ARP requests are only sent to routers\non that network if the target MAC address matches. ARP requests that\ndoes not match a router will only be forwarded to non-router ports.\n\nCloses-Bug: #2071818\nChange-Id: Id16b16113420e9f024e1936f9427824d711b6d30\nSigned-off-by: Lucas Alvares Gomes \u003clucasagomes@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1d9351d6510e7aad312512a88b326f317aa4326e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1d9351d6510e7aad312512a88b326f317aa4326e"}]},"branch":"refs/heads/master"},"96d7f8084231f518a61fcd24256996ab996a6fa6":{"kind":"REWORK","_number":5,"created":"2024-07-04 09:09:04.000000000","uploader":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"ref":"refs/changes/91/923391/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/91/923391/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/91/923391/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/91/923391/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/91/923391/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/91/923391/5"}}},"commit":{"parents":[{"commit":"9649251f4db3e74a4337326f8295fff3554bb216","subject":"Merge \"Add the port \"fixed_ips\" information in the DHCP RPC\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9649251f4db3e74a4337326f8295fff3554bb216"}]}],"author":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2024-07-03 12:39:18.000000000","tz":60},"committer":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2024-07-04 09:03:42.000000000","tz":60},"subject":"[OVN] Add support for broadcast_arps_to_all_routers","message":"[OVN] Add support for broadcast_arps_to_all_routers\n\nThis patch introduces a new configuration option for OVN called\n\"broadcast_arps_to_all_routers\". This option is responsible for\nconfiguring the external networks with the \u0027broadcast-arps-to-all-routers\u0027\nconfiguration from OVN 23.06 and newer. By enabling this option (default)\nOVN will flood ARP requests to all attached ports on a network. If\ndisabled, ARP requests are only sent to routers on that network if the\ntarget MAC address matches. ARP requests that does not match a router\nwill only be forwarded to non-router ports.\n\nCloses-Bug: #2071818\nChange-Id: Id16b16113420e9f024e1936f9427824d711b6d30\nSigned-off-by: Lucas Alvares Gomes \u003clucasagomes@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/96d7f8084231f518a61fcd24256996ab996a6fa6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/96d7f8084231f518a61fcd24256996ab996a6fa6"}]},"branch":"refs/heads/master"},"b995862e5b38a9944f9b743afe3ae9c98f08d082":{"kind":"REWORK","_number":6,"created":"2024-07-05 08:20:47.000000000","uploader":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"ref":"refs/changes/91/923391/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/91/923391/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/91/923391/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/91/923391/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/91/923391/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/91/923391/6"}}},"commit":{"parents":[{"commit":"9649251f4db3e74a4337326f8295fff3554bb216","subject":"Merge \"Add the port \"fixed_ips\" information in the DHCP RPC\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9649251f4db3e74a4337326f8295fff3554bb216"}]}],"author":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2024-07-03 12:39:18.000000000","tz":60},"committer":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2024-07-05 08:19:28.000000000","tz":60},"subject":"[OVN] Add support for broadcast_arps_to_all_routers","message":"[OVN] Add support for broadcast_arps_to_all_routers\n\nThis patch introduces a new configuration option for OVN called\n\"broadcast_arps_to_all_routers\". This option is responsible for\nconfiguring the external networks with the \u0027broadcast-arps-to-all-routers\u0027\nconfiguration from OVN 23.06 and newer. By enabling this option (default)\nOVN will flood ARP requests to all attached ports on a network. If\ndisabled, ARP requests are only sent to routers on that network if the\ntarget MAC address matches. ARP requests that does not match a router\nwill only be forwarded to non-router ports.\n\nCloses-Bug: #2071818\nChange-Id: Id16b16113420e9f024e1936f9427824d711b6d30\nSigned-off-by: Lucas Alvares Gomes \u003clucasagomes@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b995862e5b38a9944f9b743afe3ae9c98f08d082"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b995862e5b38a9944f9b743afe3ae9c98f08d082"}]},"branch":"refs/heads/master"},"f72abb2a8d09fb68d2e811fc3c77013caa7f08c3":{"kind":"REWORK","_number":7,"created":"2024-07-09 19:21:25.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/91/923391/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/91/923391/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/91/923391/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/91/923391/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/91/923391/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/91/923391/7"}}},"commit":{"parents":[{"commit":"9649251f4db3e74a4337326f8295fff3554bb216","subject":"Merge \"Add the port \"fixed_ips\" information in the DHCP RPC\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9649251f4db3e74a4337326f8295fff3554bb216"}]}],"author":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2024-07-03 12:39:18.000000000","tz":60},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2024-07-09 19:21:14.000000000","tz":0},"subject":"[OVN] Add support for broadcast_arps_to_all_routers","message":"[OVN] Add support for broadcast_arps_to_all_routers\n\nThis patch introduces a new configuration option for OVN called\n\"broadcast_arps_to_all_routers\". This option is responsible for\nconfiguring the external networks with the \u0027broadcast-arps-to-all-routers\u0027\nconfiguration from OVN 23.06 and newer. By enabling this option (default)\nOVN will flood ARP requests to all attached ports on a network. If\ndisabled, ARP requests are only sent to routers on that network if the\ntarget MAC address matches. ARP requests that does not match a router\nwill only be forwarded to non-router ports.\n\nCloses-Bug: #2071818\nChange-Id: Id16b16113420e9f024e1936f9427824d711b6d30\nSigned-off-by: Lucas Alvares Gomes \u003clucasagomes@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f72abb2a8d09fb68d2e811fc3c77013caa7f08c3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f72abb2a8d09fb68d2e811fc3c77013caa7f08c3"}]},"branch":"refs/heads/master"},"f651b28c4234ab7d20ecae7d12b47cc090b92ea9":{"kind":"REWORK","_number":8,"created":"2024-07-15 13:26:52.000000000","uploader":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"ref":"refs/changes/91/923391/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/91/923391/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/91/923391/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/91/923391/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/91/923391/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/91/923391/8"}}},"commit":{"parents":[{"commit":"9649251f4db3e74a4337326f8295fff3554bb216","subject":"Merge \"Add the port \"fixed_ips\" information in the DHCP RPC\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9649251f4db3e74a4337326f8295fff3554bb216"}]}],"author":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2024-07-03 12:39:18.000000000","tz":60},"committer":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2024-07-15 13:26:52.000000000","tz":0},"subject":"[OVN] Add support for broadcast_arps_to_all_routers","message":"[OVN] Add support for broadcast_arps_to_all_routers\n\nThis patch introduces a new configuration option for OVN called\n\"broadcast_arps_to_all_routers\". This option is responsible for\nconfiguring the external networks with the \u0027broadcast-arps-to-all-routers\u0027\nconfiguration from OVN 23.06 and newer. By enabling this option (default)\nOVN will flood ARP requests to all attached ports on a network. If\ndisabled, ARP requests are only sent to routers on that network if the\ntarget MAC address matches. ARP requests that does not match a router\nwill only be forwarded to non-router ports.\n\nCloses-Bug: #2071818\nChange-Id: Id16b16113420e9f024e1936f9427824d711b6d30\nSigned-off-by: Lucas Alvares Gomes \u003clucasagomes@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f651b28c4234ab7d20ecae7d12b47cc090b92ea9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f651b28c4234ab7d20ecae7d12b47cc090b92ea9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
