)]}'
{"id":"openstack%2Fneutron~923711","triplet_id":"openstack%2Fneutron~master~I35d700885e6de0f1b92537789ae9819f8b21db0c","project":"openstack/neutron","branch":"master","topic":"bug/2069581","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2024-07-17 11:45:50.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I35d700885e6de0f1b92537789ae9819f8b21db0c","subject":"[WSGI] Move all OVS jobs to use WSGI API module","status":"MERGED","created":"2024-07-09 07:02:44.000000000","updated":"2024-07-17 11:48:29.000000000","submitted":"2024-07-17 11:45:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"923711-bug/2069581","meta_rev_id":"5b41276af0adb77293dc3469fe64b6c88c82c979","_number":923711,"virtual_id_number":923711,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-07-17 11:45:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"all":[{"value":2,"date":"2024-07-15 17:36:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":2,"date":"2024-07-17 07:10:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":2,"date":"2024-07-15 13:19:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"all":[{"value":1,"date":"2024-07-15 17:36:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-07-09 09:37:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-07-09 14:31:12.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2024-07-12 07:04:29.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"CC"},{"updated":"2024-07-12 07:04:29.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"CC"},{"updated":"2024-07-12 07:04:29.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"CC"},{"updated":"2024-07-15 13:19:02.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2024-07-15 17:36:10.000000000","updated_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"REVIEWER"},{"updated":"2024-07-17 07:10:54.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"}],"messages":[{"id":"2916f95632eab739b982e80f7e704f5891f8cd0b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-07-09 07:02:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2796f96aa8d9e0f908444cb67e85ad16c626ba39","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-07-09 07:07:12.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e54e403b1927988c83b2f7decacebfcb5cf13b3d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-09 09:37:25.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/708caac58589438eb18bc328b6f8623e\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/dd75da8e1d744442af87415d5dba31c5 : SUCCESS in 1h 35m 56s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/dd981391979441c0b33eaee8c4f0c652 : SUCCESS in 1h 40m 19s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/ced4f9cda9704c59abc42bd0b0c9473f : SUCCESS in 1h 28m 12s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/0fe8bea7b040402a95a13e06d5080e48 : SUCCESS in 2h 22m 47s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/90d209ecb22c4013927494cef09f657c : FAILURE in 1h 03m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/43430e40a97d41cbba3beff0f515eab6 : SUCCESS in 9m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca8e731302ca46eaa2991ce513887cdc : SUCCESS in 11m 12s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/0814ba1e52e344eaa30c0f5953437a5c : RETRY_LIMIT in 7m 40s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/f795b773d7074f568c35d88c420ba90a : RETRY_LIMIT in 17m 49s","accounts_in_message":[],"_revision_number":1},{"id":"3855f3774f10f41aba839b9ba57a13d002a58e67","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-09 09:43:12.000000000","message":"Patch Set 1:\n\nBuild failed (experimental pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e4100ca76e0a412db1237dafaa0ee9a6\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/3dd90f1ac4484ce69512e59b47f0fc4c : SUCCESS in 2h 02m 54s\n- neutron-tempest-plugin-openvswitch-distributed-dhcp https://zuul.opendev.org/t/openstack/build/33d9a29223694deca604c64e378aafed : SUCCESS in 2h 01m 45s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-distributed-dhcp https://zuul.opendev.org/t/openstack/build/d04df3e6f0f240f48118fa2867715d92 : FAILURE in 2h 05m 28s\n- neutron-tempest-plugin-api-ovs-wsgi https://zuul.opendev.org/t/openstack/build/6fc4586f77b949c09e5f40f65fac73d9 : SUCCESS in 1h 44m 25s (non-voting)\n- neutron-tempest-plugin-api-ovn-wsgi https://zuul.opendev.org/t/openstack/build/4ce584948db34d27b8b7cfc294c749d0 : FAILURE in 2h 23m 36s (non-voting)\n- neutron-ovs-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/9114719b15e14c9ba52f24700649abb5 : SUCCESS in 59m 52s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/49f4368a92924289811423c5164fda90 : RETRY_LIMIT in 6m 45s\n- neutron-ovs-tempest-with-os-ken-master https://zuul.opendev.org/t/openstack/build/8d41750c58324f41813ff881b43fa012 : SUCCESS in 1h 21m 09s\n- neutron-ovn-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/07adb4e8a0074856b7ac07166919ab81 : SUCCESS in 1h 02m 40s\n- neutron-ovn-tempest-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/329f3ca181df44e7b19a9c6301454452 : SUCCESS in 55m 44s\n- neutron-ovs-tempest-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/d073e101de35456ab6fc620cacff1318 : SUCCESS in 1h 00m 41s\n- neutron-ovs-tempest-fips https://zuul.opendev.org/t/openstack/build/500c31cc22ae4fae92df9644253835f4 : SUCCESS in 1h 06m 43s\n- devstack-tobiko-neutron https://zuul.opendev.org/t/openstack/build/2e093b9a06d549f6b1d269e60fed1301 : SUCCESS in 1h 02m 29s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/6352e1ba999e4a71a63675c97ae50391 : SUCCESS in 55m 49s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/f7476892d5d24cdfbf569a9507e9d347 : SUCCESS in 1h 29m 43s\n- neutron-linuxbridge-tempest-plugin-nftables https://zuul.opendev.org/t/openstack/build/e987f7931ebf45fdbe4505ad133c611e : SUCCESS in 2h 23m 11s\n- neutron-ovs-tempest-plugin-iptables_hybrid-nftables https://zuul.opendev.org/t/openstack/build/cb37199b24aa499ca42992cbe0fa9475 : SUCCESS in 2h 17m 24s\n- devstack-enforce-scope https://zuul.opendev.org/t/openstack/build/98bc73edf12a4b1b934075414ff916ba : SUCCESS in 31m 05s\n- neutron-ovs-tempest-with-oslo-master https://zuul.opendev.org/t/openstack/build/0184ba3d2a7742e09d283570d0a251f5 : SUCCESS in 1h 05m 39s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/4ba64b1b11c04490b7851e8a105826ba : RETRY_LIMIT in 20m 59s","accounts_in_message":[],"_revision_number":1},{"id":"c6a938f1af8a7e95aeda2fdd1ee79c5da2c8fc05","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-07-09 10:06:18.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"4e05d6bf740c36e4a8f9bd7faf96c49735be9c60","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-07-09 10:33:49.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"663cd95f2798484128487f263e1f1bb269899a1d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-07-09 12:16:03.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"4dcc0a92ab84d59f7b91c0f70259637538b5bbf6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-09 14:08:26.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/03a7e195599e498fbbe0754512d7215d\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/3c255c96529d4675b26e2b43abf81082 : SUCCESS in 1h 29m 18s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/be508f10ece34878a41055e5084d0143 : FAILURE in 1h 40m 49s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/a985bc4bf1c24a7babbf309de02df07f : SUCCESS in 1h 31m 13s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/099e120bcac74a4ea3f55d0dae756cb6 : SUCCESS in 1h 50m 55s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/899c94525f0f4433b757bcd932e122e1 : FAILURE in 36m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d6204c64c69f447f875f8f4d479f9a97 : SUCCESS in 8m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b978c74a5ab347f8a98545b3b48c2ab4 : SUCCESS in 10m 36s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/5ca59a8ac44b42b69cb15684619b9dc1 : SUCCESS in 1h 27m 11s","accounts_in_message":[],"_revision_number":4},{"id":"0e1463819c9c0e101daba236be664c70b7be3ff8","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-07-10 05:45:22.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"92bd9d6739518ac831b7a89a4ba89c5de0853a33","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-10 08:01:22.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9fc5b78c92004fbeae56edae536667d4\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/62b963b6bf27466995eb4e1b205fbde3 : SUCCESS in 1h 47m 00s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/3156ff6b30244cb48ed4983b79bf6528 : SUCCESS in 2h 08m 52s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/5f63963f9b444d4ba5fbe7f9c1113bbc : SUCCESS in 1h 56m 57s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/25d2bb5fab6d48e8a33352b2bb9ef704 : SUCCESS in 1h 34m 09s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/f3faa5bc4ba34ef69e6b5210094ac6ee : FAILURE in 56m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eeb9ceab209544369ff9697590593dda : SUCCESS in 10m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d1b123490c846ad872252b6b352a1ad : SUCCESS in 10m 00s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/353cde4b10fc47beb2a242d6ef5fb53b : SUCCESS in 1h 31m 14s","accounts_in_message":[],"_revision_number":4},{"id":"368c4cd08366c728c4948b310ac8c484ac1be5a4","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-07-11 12:52:17.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"d8577d59eee2bdb0c2b237374aa240deee5220b3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-11 15:10:19.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/672b4457822348b6a7e0c5acce2dd8c8\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/95ac76b82e764deb820fa77b039b0b71 : SUCCESS in 1h 49m 13s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/939146ecd4bd4b60b945cd2c29bcdf3c : SUCCESS in 1h 43m 46s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/85b281e8bed84970941b6db772a1b0f0 : SUCCESS in 1h 37m 51s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/0fe7079dbd064d0cada638c9ade15f72 : SUCCESS in 2h 15m 32s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/27155ddc5bd44df398227aac2d137266 : SUCCESS in 31m 10s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/46b8af92c05e4c9191f2e076e0e561e6 : SUCCESS in 10m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/83ec0881bbc94b1fb562b0abfd8665f5 : SUCCESS in 10m 46s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/6decaf71b05342ee98fa86e8aa0fd058 : SUCCESS in 1h 30m 06s","accounts_in_message":[],"_revision_number":4},{"id":"db4f6ffc1cb132be3d69d44e645b4250231a0b87","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-07-12 07:03:53.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"902b9543825bcaea39291587f85cb1d828074484","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-07-15 13:19:02.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"f76be8be25bafe6035b9aa797662d2e3797d135d","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2024-07-15 17:36:10.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"05e62cf60cbbf6f2b9c66e2061b3810e07a88214","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-15 17:36:58.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"9a30dc1eb5de62d6fce63032d58416d88ddcadb1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-15 23:39:08.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b7acbbf07bfd469fa7d10ae7fb044720\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/3a6603a6f71043dd9ba568ae452d3275 : SUCCESS in 1h 49m 43s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/4de9ccc8182e4c50bef27e6ecbc39e55 : SUCCESS in 2h 25m 10s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/f8ae1d0c82fd472ebfaba06fd3187da9 : FAILURE in 2h 07m 07s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/4624dc8b89504e75810ccb53687dcc34 : FAILURE in 2h 27m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e8ba4190f12649d5935695a677b33e9a : SUCCESS in 9m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d98bac982dc54e179190320a8ae6874f : SUCCESS in 10m 04s","accounts_in_message":[],"_revision_number":4},{"id":"8188d11f5cbd1106acdc4428c17ac32f0e41dbdb","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-07-17 07:10:54.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"1ee5e7e4fd70797b1dfaa7f050994d367dd092f4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-17 09:19:29.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e4ab2764b0c4472493919aaf884f8d88\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/203a2e81b3884b3ba9ffc58c86cc85be : SUCCESS in 1h 40m 37s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/5192bbf70b4f461d9ca6b25ee2b8f0fb : SUCCESS in 2h 06m 34s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/288dee62961f4b72aa53344bf572986e : SUCCESS in 2h 00m 16s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/055d1d7ddab64f21b258e6715a30f546 : SUCCESS in 2h 00m 50s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/c97ec0faca1b4dcc95fa47c67077a904 : SUCCESS in 39m 19s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a378d64434324dd5ab98de32528825c0 : SUCCESS in 10m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/00f55d1098b0407c9f68fabf69f2cd46 : SUCCESS in 8m 45s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/077bd0a547c44422b1dd9b21a1d4e0d8 : SUCCESS in 1h 13m 42s","accounts_in_message":[],"_revision_number":4},{"id":"36aaa6f31f9026873d35a5d97adf8b8cbf24c6bb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-17 09:20:09.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"1b70edd96aa6ad59b1d18fb0adb510fa4b6e19d6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-17 11:45:50.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f887fd77627c4a8ca7ee7c6eee232c08\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/401a21d8e324458486b5706eb906c288 : SUCCESS in 1h 54m 33s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/fbb2bd5161d44f4a8d188a7a83f84b7b : SUCCESS in 2h 23m 50s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/38613e4ef74b46bba84fe5f6f49c3088 : SUCCESS in 2h 22m 49s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/e84e4a8d994b458bbcf9d4e68ad8ef23 : SUCCESS in 1h 56m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/00f157936e84424c8a952833fb1b9bbd : SUCCESS in 12m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7942074d546c45f28bddc829e5aa683e : SUCCESS in 12m 16s","accounts_in_message":[],"_revision_number":4},{"id":"b6199c9280af4de96960d58e6cc84c66881935eb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-17 11:45:50.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"5b41276af0adb77293dc3469fe64b6c88c82c979","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-17 11:48:29.000000000","message":"Patch Set 4:\n\nBuild failed (promote pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/fe82c77565fa4edb8ef0623651aaa3ca\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b99f21eb82c45f699f2d55c5ad4b195 : FAILURE in 59s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"19208ebdae49da690c4250358962f990bc707644","revisions":{"9266f2bcc86acaaf1e1f9b65c35f1818828e32d2":{"kind":"REWORK","_number":1,"created":"2024-07-09 07:02:44.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/11/923711/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/11/923711/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/11/923711/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/11/923711/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/11/923711/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/11/923711/1"}}},"commit":{"parents":[{"commit":"3445d84f92debe21ebaf70bc50d4bf78de0ba0ce","subject":"Merge \"Checking pci_slot to avoid changing staus to BUILD forever\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3445d84f92debe21ebaf70bc50d4bf78de0ba0ce"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-07-08 16:14:31.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-07-08 16:14:31.000000000","tz":0},"subject":"[WSGI] Move all OVS jobs to use WSGI API module","message":"[WSGI] Move all OVS jobs to use WSGI API module\n\nAll the ML2/OVS jobs (grenade, tempest) are now using the WSGI API\nmodule, instead of the eventlet server.\n\nRelated-Bug: #2069581\nTopic: eventlet-deprecation\nChange-Id: I35d700885e6de0f1b92537789ae9819f8b21db0c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9266f2bcc86acaaf1e1f9b65c35f1818828e32d2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9266f2bcc86acaaf1e1f9b65c35f1818828e32d2"}]},"branch":"refs/heads/master"},"10302aba008ec80aa231f47e359d2a7cf1c1a544":{"kind":"REWORK","_number":2,"created":"2024-07-09 10:06:18.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/11/923711/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/11/923711/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/11/923711/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/11/923711/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/11/923711/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/11/923711/2"}}},"commit":{"parents":[{"commit":"3445d84f92debe21ebaf70bc50d4bf78de0ba0ce","subject":"Merge \"Checking pci_slot to avoid changing staus to BUILD forever\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3445d84f92debe21ebaf70bc50d4bf78de0ba0ce"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-07-08 16:14:31.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-07-08 19:19:56.000000000","tz":0},"subject":"[WSGI] Move all OVS jobs to use WSGI API module","message":"[WSGI] Move all OVS jobs to use WSGI API module\n\nAll the ML2/OVS jobs (grenade, tempest) are now using the WSGI API\nmodule, instead of the eventlet server.\n\nRelated-Bug: #2069581\nTopic: eventlet-deprecation\nChange-Id: I35d700885e6de0f1b92537789ae9819f8b21db0c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/10302aba008ec80aa231f47e359d2a7cf1c1a544"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/10302aba008ec80aa231f47e359d2a7cf1c1a544"}]},"branch":"refs/heads/master"},"fe81bf17487f174ed45e42d280086ac738c8b933":{"kind":"REWORK","_number":3,"created":"2024-07-09 10:33:49.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/11/923711/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/11/923711/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/11/923711/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/11/923711/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/11/923711/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/11/923711/3"}}},"commit":{"parents":[{"commit":"3445d84f92debe21ebaf70bc50d4bf78de0ba0ce","subject":"Merge \"Checking pci_slot to avoid changing staus to BUILD forever\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3445d84f92debe21ebaf70bc50d4bf78de0ba0ce"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-07-08 16:14:31.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-07-08 19:48:03.000000000","tz":0},"subject":"[WSGI] Move all OVS jobs to use WSGI API module","message":"[WSGI] Move all OVS jobs to use WSGI API module\n\nAll the ML2/OVS jobs (grenade, tempest) are now using the WSGI API\nmodule, instead of the eventlet server.\n\nRelated-Bug: #2069581\nTopic: eventlet-deprecation\nChange-Id: I35d700885e6de0f1b92537789ae9819f8b21db0c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/fe81bf17487f174ed45e42d280086ac738c8b933"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/fe81bf17487f174ed45e42d280086ac738c8b933"}]},"branch":"refs/heads/master"},"19208ebdae49da690c4250358962f990bc707644":{"kind":"REWORK","_number":4,"created":"2024-07-09 12:16:03.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/11/923711/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/11/923711/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/11/923711/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/11/923711/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/11/923711/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/11/923711/4"}}},"commit":{"parents":[{"commit":"3445d84f92debe21ebaf70bc50d4bf78de0ba0ce","subject":"Merge \"Checking pci_slot to avoid changing staus to BUILD forever\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3445d84f92debe21ebaf70bc50d4bf78de0ba0ce"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-07-08 16:14:31.000000000","tz":0},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2024-07-09 12:16:03.000000000","tz":0},"subject":"[WSGI] Move all OVS jobs to use WSGI API module","message":"[WSGI] Move all OVS jobs to use WSGI API module\n\nAll the ML2/OVS tempest jobs are now using the WSGI API\nmodule, instead of the eventlet server.\n\nRelated-Bug: #2069581\nTopic: eventlet-deprecation\nChange-Id: I35d700885e6de0f1b92537789ae9819f8b21db0c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/19208ebdae49da690c4250358962f990bc707644"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/19208ebdae49da690c4250358962f990bc707644"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
