)]}'
{"id":"openstack%2Fneutron~923719","triplet_id":"openstack%2Fneutron~stable%2F2024.1~I2d7702e17c75c96ca2f29749dccab77cb2f4bcf4","project":"openstack/neutron","branch":"stable/2024.1","attention_set":{},"removed_from_attention_set":{"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2024-07-09 19:05:01.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I2d7702e17c75c96ca2f29749dccab77cb2f4bcf4","subject":"Checking pci_slot to avoid changing staus to BUILD forever","status":"MERGED","created":"2024-07-09 09:05:38.000000000","updated":"2024-07-09 19:06:22.000000000","submitted":"2024-07-09 19:05:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"923719","cherry_pick_of_change":923467,"cherry_pick_of_patch_set":10,"meta_rev_id":"05e55ff3993d43e7e4b67e3380f3116bd796094c","_number":923719,"virtual_id_number":923719,"owner":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-07-09 19:05:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2024-07-09 09:07:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2024-07-09 18:21:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2024-07-09 18:21:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":5112,"name":"Seyeong Kim","email":"seyeong.kim@canonical.com","username":"xtrusia"}],"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-07-09 09:05:38.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":5112,"name":"Seyeong Kim","email":"seyeong.kim@canonical.com","username":"xtrusia"},"state":"CC"},{"updated":"2024-07-09 09:07:08.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2024-07-09 09:53:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-07-09 09:56:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-07-09 18:21:03.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"}],"messages":[{"id":"d71b897d0710fa46a1e14587ea61c5ecd1fdf3f2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-07-09 09:05:38.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"fdaa5275b827104ce9807b23ffe06f55c104bde5","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-07-09 09:07:08.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"2ab9f7281f3dc423ca4483d47cc346cde7998e1a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-09 09:53:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/58a3729a422b4fecad14f375a46df0ce\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a8769b6c5c774f67b3c7a47777140c89 : FAILURE in 3m 03s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d97982b8355342b3b71145703a94b818 : SUCCESS in 46m 04s\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/e417e777c04b404fa25a1148308bdeca : FAILURE in 2m 49s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2ce9d594d1c3322f08cb32efa9896a22db3e11f1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-09 09:56:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7692f23997f745f491cc9192a0f3410a\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d08a421eb69f4bb8b41ce9647eac9368 : SUCCESS in 40m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2fb99c5a09ff4cf182c531491d6bbd32 : SUCCESS in 10m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ef9f77a37d5e43baa55ca36c486fa206 : SUCCESS in 23m 57s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/cc57ba19194e42a78ef631e5fe42288f : SUCCESS in 27m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a96d1fab95f4d5ab70179ab8ec5f6b1 : SUCCESS in 10m 04s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/80f4c569acde41a0822b6749b0af53d8 : SUCCESS in 48m 25s","accounts_in_message":[],"_revision_number":1},{"id":"fae39ae85f590bc6971c4b9707f5703c1334cd24","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-07-09 18:21:03.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"54010e43934beb229c920abaffafb3b264cd5a84","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-09 18:21:27.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"20e70f09bbb0dfb6d1e00dbc829eed2aa2b0a94c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-09 19:05:01.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b78f8d398ce4454bbbe09923c10355d2\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cea331c13d1e4db5bd4f71b8f8d3b48c : SUCCESS in 10m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/945a5ae225494d1c9462ebb0ae34026f : SUCCESS in 28m 41s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/416b9f6420e84ba49f123a25b8ff5c3e : SUCCESS in 24m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ba4be98515b34033bbde75b036c3c6de : SUCCESS in 11m 25s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/abe790db088a4e3d80ea30baeba9f1b6 : SUCCESS in 42m 27s","accounts_in_message":[],"_revision_number":1},{"id":"68a102e074ce0bb81cdd7ff07dc52e0d33f7af91","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-09 19:05:01.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"05e55ff3993d43e7e4b67e3380f3116bd796094c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-09 19:06:22.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ca5a9e0318fd40a7b7655e05fc29fbbd\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc69976e64f34b2eaa7143724ffe096b : SUCCESS in 1m 03s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"59bc8e476f17dee323973a0e6ea4cd4c343f77b6","revisions":{"59bc8e476f17dee323973a0e6ea4cd4c343f77b6":{"kind":"REWORK","_number":1,"created":"2024-07-09 09:05:38.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/19/923719/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/19/923719/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/19/923719/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/19/923719/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/19/923719/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/19/923719/1"}}},"commit":{"parents":[{"commit":"df02d0dfecc057e021b8e9c19270645c07f8cb79","subject":"Merge \"Add the port \"fixed_ips\" information in the DHCP RPC\" into stable/2024.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/df02d0dfecc057e021b8e9c19270645c07f8cb79"}]}],"author":{"name":"Seyeong Kim","email":"seyeong.kim@canonical.com","date":"2024-07-04 06:23:59.000000000","tz":0},"committer":{"name":"Lajos Katona","email":"katonalala@gmail.com","date":"2024-07-09 09:05:38.000000000","tz":0},"subject":"Checking pci_slot to avoid changing staus to BUILD forever","message":"Checking pci_slot to avoid changing staus to BUILD forever\n\nCurrently when sriov agent is enabled and migrating a non-sriov\ninstance, non-sriov port status is frequently set to BUILD\ninstead of ACTIVE.\nThis is because the \u0027binding_activate\u0027 function in sriov-nic-agent sets it\nBUILD with get_device_details_from_port_id(as it calls _get_new_status).\n\nThis patch checks network_ports in binding_activate and\nskip binding port if it is not sriov port\n\nCloses-Bug: #2072154\nChange-Id: I2d7702e17c75c96ca2f29749dccab77cb2f4bcf4\n(cherry picked from commit a311606fcdae488e76c29e0e5e4035f8da621a34)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/59bc8e476f17dee323973a0e6ea4cd4c343f77b6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/59bc8e476f17dee323973a0e6ea4cd4c343f77b6"}]},"branch":"refs/heads/stable/2024.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
