)]}'
{"id":"openstack%2Fneutron~923849","triplet_id":"openstack%2Fneutron~stable%2F2023.2~I8d185cdc807e94098c137314bcaa2317a2f85ebe","project":"openstack/neutron","branch":"stable/2023.2","topic":"bug/2062009-stable/2024.1-stable/2023.2","attention_set":{},"removed_from_attention_set":{"1004":{"account":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"last_update":"2025-04-30 09:00:29.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I8d185cdc807e94098c137314bcaa2317a2f85ebe","subject":"Start logging plugin RPC via service framework","status":"ABANDONED","created":"2024-07-10 13:52:57.000000000","updated":"2025-04-30 09:00:29.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"cherry_pick_of_change":923848,"cherry_pick_of_patch_set":1,"meta_rev_id":"d8d221f992e8d3c69d7e7b21772c4e0f5d7db80d","_number":923849,"virtual_id_number":923849,"owner":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:check","value":1,"date":"2024-07-10 15:48:00.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"date":"2024-07-11 08:00:45.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":-1,"date":"2025-04-26 20:08:10.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":30314,"name":"Sebastian Lohff","email":"sebastian.lohff@sap.com","username":"seba"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-07-10 13:52:57.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":30314,"name":"Sebastian Lohff","email":"sebastian.lohff@sap.com","username":"seba"},"state":"CC"},{"updated":"2024-07-10 14:54:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-07-10 15:48:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-07-11 07:59:11.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2025-04-26 20:08:10.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2025-04-30 09:00:29.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"CC"}],"messages":[{"id":"19e86267996cf57ae812ca858d9f0d5853fefdcf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2024-07-10 13:52:57.000000000","message":"Patch Set 1: Cherry Picked from branch stable/2024.1.","accounts_in_message":[],"_revision_number":1},{"id":"699d67174b70342a05d94f7266ab665cfcfc7ffe","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-10 14:54:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/328b6202c20d43eba58bd863d1fca778\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1e9fe95230d840ce92a1992eab555ea8 : FAILURE in 4m 04s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2b3d84f224724122a53a0cbbdace783e : SUCCESS in 55m 47s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/76116a3d05ff4d0a888a988337e96d79 : SUCCESS in 58m 53s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/a6e408a5ac6a4c2caf983883a27fc1b0 : SUCCESS in 50m 30s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"da03bec2fabb0408844ad03fecf13976b541507b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-10 15:48:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8076e9e4aad0424198fd2748ce36a6a9\n\n- neutron-tempest-plugin-openvswitch-2023-2 https://zuul.opendev.org/t/openstack/build/beff2f1083234e81a15dc07075cc9cda : SUCCESS in 1h 16m 55s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2023-2 https://zuul.opendev.org/t/openstack/build/6e8f2dff72c34a57b300b1c76e4a71b8 : SUCCESS in 1h 19m 01s\n- neutron-tempest-plugin-ovn-2023-2 https://zuul.opendev.org/t/openstack/build/bf59717319714875b6812b0fe98f8319 : SUCCESS in 1h 53m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/425563cdd8fa48509e770183500d0899 : SUCCESS in 41m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4171fb43eb65487eb29dd84eb1d03c80 : SUCCESS in 11m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f8f5976ec4234f89a413494c2456c346 : SUCCESS in 33m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f5f0785e8dd54ba79bc3541a1d9d5536 : SUCCESS in 25m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/edcd10df423044508592ec4e721ee0c6 : SUCCESS in 26m 24s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/16a50350f4d5486f870b80142b0e4adc : SUCCESS in 23m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e987c360673640fdbe2d7abd384f6650 : SUCCESS in 11m 11s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/e96b35d45e614edf939ed2e1bd5142e7 : SUCCESS in 46m 11s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/3c4f3c0bf45b4e05a55536ea3e6d0ce5 : SUCCESS in 1h 46m 37s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/b9ed77baafa44ab1acb13176b6c6ebf7 : SUCCESS in 1h 09m 01s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/c9cc0ca090514b68a06757f9b8133462 : SUCCESS in 1h 07m 10s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/5cd372bd38854510ab474d9759aa9e0b : SUCCESS in 1h 19m 54s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/1711a977d4da41b988fe0b76d8b15b3c : SUCCESS in 1h 17m 41s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/6546a65b89d7445c86357ef4f55c3f73 : SUCCESS in 54m 19s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/d5d4a4b0ebe64c358115fee14f63ab0b : SUCCESS in 1h 23m 08s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/522d390d270e45418f1fccf1a2baee97 : SUCCESS in 1h 17m 50s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/9c0c67ab9fa540389636ed859865f30b : SUCCESS in 1h 14m 53s","accounts_in_message":[],"_revision_number":1},{"id":"c8203c50769f4326e8b4c2715b474266dd319edf","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-07-11 07:59:11.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"27bcdb59d6d057e3df69c4e9adf2d50e77232c3c","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-07-11 08:00:21.000000000","message":"Patch Set 1: -Code-Review","accounts_in_message":[],"_revision_number":1},{"id":"c86d60dc2548d0c2cf97524e086718db682ad682","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-07-11 08:00:45.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"cf867724a475636084beb4c6372f03c343f88f48","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-04-26 20:08:10.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d8d221f992e8d3c69d7e7b21772c4e0f5d7db80d","tag":"autogenerated:gerrit:abandon","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-04-30 09:00:29.000000000","message":"Abandoned\n\nstable/2023.2 branch of openstack/neutron transitioned to End of Life and is about to be deleted. To be able to do that, all open patches need to be abandoned.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"82b75ced9d4be1caf8c5313ba1bf0d206fba0300","revisions":{"82b75ced9d4be1caf8c5313ba1bf0d206fba0300":{"kind":"REWORK","_number":1,"created":"2024-07-10 13:52:57.000000000","uploader":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"ref":"refs/changes/49/923849/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/49/923849/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/49/923849/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/49/923849/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/49/923849/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/49/923849/1"}}},"commit":{"parents":[{"commit":"1da102e4024a1c7398179b29bd63e2ba42b19000","subject":"Checking pci_slot to avoid changing staus to BUILD forever","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1da102e4024a1c7398179b29bd63e2ba42b19000"}]}],"author":{"name":"Sebastian Lohff","email":"sebastian.lohff@sap.com","date":"2024-04-15 13:39:27.000000000","tz":120},"committer":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2024-07-10 13:52:57.000000000","tz":0},"subject":"Start logging plugin RPC via service framework","message":"Start logging plugin RPC via service framework\n\nInstead of the LoggingServiceDriverManager starting the RPC if any\ndriver needs it, we now only start it when this is requested by neutron\nvia start_rpc_listeners(). This is required when running neutron-server\nand neutron-rpc-server separately to run RPC only in neutron-rpc-server.\n\nChange-Id: I8d185cdc807e94098c137314bcaa2317a2f85ebe\nPartial-Bug: #2062009\n(cherry picked from commit 6170309157d6922db2bc6014f17487b74f37e08b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/82b75ced9d4be1caf8c5313ba1bf0d206fba0300"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/82b75ced9d4be1caf8c5313ba1bf0d206fba0300"}]},"branch":"refs/heads/stable/2023.2"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
