)]}'
{"id":"openstack%2Fneutron~923971","triplet_id":"openstack%2Fneutron~master~Icc27c2b8825d7f96c9dac87dec8bbb55d493d942","project":"openstack/neutron","branch":"master","topic":"ls_use_persist_uuid","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2025-04-04 08:35:42.000000000","reason":"\u003cGERRIT_ACCOUNT_16688\u003e replied on the change","reason_account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},"26471":{"account":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"last_update":"2025-04-04 22:16:56.000000000","reason":"Change was submitted"},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2025-04-04 19:42:35.000000000","reason":"\u003cGERRIT_ACCOUNT_1131\u003e replied on the change","reason_account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},"5756":{"account":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"last_update":"2025-04-04 22:16:56.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Icc27c2b8825d7f96c9dac87dec8bbb55d493d942","subject":"Avoid race condition by using persist_uuid","status":"MERGED","created":"2024-07-11 16:11:54.000000000","updated":"2025-04-04 22:18:27.000000000","submitted":"2025-04-04 22:16:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":45,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"923971-ls_use_persist_uuid","meta_rev_id":"34c53a9e29b9cf021b49e85b1f6b8dc19d4a3d2e","_number":923971,"virtual_id_number":923971,"owner":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2025-03-19 20:59:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-04-04 22:16:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2025-04-04 08:35:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":2,"date":"2025-04-04 19:42:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":1,"date":"2025-04-04 19:42:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":28056,"name":"Liushy","email":"liuxie_11@163.com","username":"liuxie_11"}],"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-07-11 16:16:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-07-11 16:44:00.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-07-11 20:09:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-07-12 05:59:31.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2024-07-15 21:19:24.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2024-10-04 08:03:53.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"CC"},{"updated":"2025-02-05 15:27:56.000000000","updated_by":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"reviewer":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"state":"REVIEWER"},{"updated":"2025-02-06 07:27:10.000000000","updated_by":{"_account_id":28056,"name":"Liushy","email":"liuxie_11@163.com","username":"liuxie_11"},"reviewer":{"_account_id":28056,"name":"Liushy","email":"liuxie_11@163.com","username":"liuxie_11"},"state":"CC"},{"updated":"2025-04-04 19:42:35.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"}],"messages":[{"id":"db6521b1fa756cb372358574ccb52738d0331b90","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2024-07-11 16:11:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f578dc0d19391530ba52454f2553d29366eb6e8a","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2024-07-11 16:13:09.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ec891beaa7ee7f7a2fb14f51ab28cb483f8121af","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-11 16:16:03.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5d2a2e54f4294d1f9383776f8b706e2a\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/62fd492a617c462b843d49cc30cec3e7 : FAILURE in 2m 57s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/b76a1f52ba294ecfa9879ca26151bf0b : FAILURE in 3m 08s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8ed8505fe6895c994ea654a2877dd62ae359e18a","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2024-07-11 16:44:00.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/July-2024/Arista_Tempest_Testrefs/changes/71/923971/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"ba5458e09f6e5d9ef3d31037268725cf091635da","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-11 20:09:02.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e2f5a88016ae4180aebbd1b051e9932c\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/6fcd96b7a6c24a8cb50e84bb3a5243b4 : FAILURE in 17m 40s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/628d7a8876854a41bc768e502b500d13 : FAILURE in 19m 07s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3e1f337fd4ed467585e8786c6d6cc827 : FAILURE in 29m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eef40e98c8f4471e87df660d990a54e9 : SUCCESS in 9m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/571e743a60ba48fb9335958a37a15a88 : FAILURE in 27m 06s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6a0dfd48f63f4341b2a380c286af2aef : FAILURE in 20m 45s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/732633d25c2143ea9f69ccb8afe79baa : FAILURE in 29m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b9c2d82dc166425e8064183d4a086fc0 : SUCCESS in 10m 18s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/139bc8e397484714b71b325bccff17b4 : FAILURE in 20m 04s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/913c0deae2ec4c00b9de1e315b3430bd : FAILURE in 47m 24s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/e5e158bdeb3d4a7284683b16daa0921e : FAILURE in 1h 35m 29s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/6594eb143fbb4affbe09dfbfd51c24fd : RETRY_LIMIT Host unreachable in 1h 20m 48s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/b134ce16064d406c862e50ad94a866d1 : FAILURE in 19m 27s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/5a578b1c3f0246cfa17347048145b644 : FAILURE in 19m 12s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/a36651cf502742768fa97ed2310b3182 : FAILURE in 16m 17s","accounts_in_message":[],"_revision_number":1},{"id":"10b10637049a1ae044218b6da41a3f3ce808a1e9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2024-07-12 00:38:41.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1, Verified-1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":2},{"id":"fa1e35f482b92f89b8cd1daa30b6642474feea68","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2024-07-12 00:41:34.000000000","message":"Patch Set 2: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3b29eab5f079bcc4f984a687408077e56995672f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-12 00:45:56.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0639db8b457f4ed6b5d803777fe5aa0b\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/db128b86069d448ea55850064485bb95 : FAILURE in 3m 11s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/d41817df1d5b4e1080bace023ab69412 : FAILURE in 2m 42s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"837ccca30ab92733b481bc30022a6da23746783c","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2024-07-12 01:10:52.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/July-2024/Arista_Tempest_Testrefs/changes/71/923971/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"300bb59989e18a55b1eccc8669431310f0dc2002","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-12 03:48:58.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9c1622951e4e4acdb0c1a63e9ede9832\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/37301b1f336a408c96af735479442e0e : TIMED_OUT in 3h 03m 26s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/101b1c892ac849139318b3972df8dd18 : TIMED_OUT in 1h 02m 18s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f77fcfee9c134b6da87a4734615bc7ba : FAILURE in 25m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b9a5a72352994c76ad46a01cbeaf633a : SUCCESS in 7m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e307e0e91f1a4e11a792b4fae9aac513 : FAILURE in 18m 22s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/bb06764d00dd4632b56a3b378c54f6fa : FAILURE in 19m 44s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c028c6dc0f654a52ab34f9b3a3d324ec : FAILURE in 21m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d17949ccfcca4de89616f8e200c57828 : SUCCESS in 9m 44s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/25743fc335454ed291b261e137590063 : FAILURE in 20m 24s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/dd00ddc06ea54f6fbb88f7e7954b56c4 : FAILURE in 31m 51s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/23cf9b2d39384fdd8545d6014e59a969 : SUCCESS in 1h 46m 23s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/b497305a13914bf28eec517c315640a6 : SUCCESS in 1h 09m 46s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/69d730314a064790bbc56c6e05aa8aaf : FAILURE in 1h 10m 51s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/420d361b4500434eafe55db0358d9255 : SUCCESS in 41m 08s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/34ec310b784d4573b80e1f6e890fb30f : FAILURE in 57m 07s","accounts_in_message":[],"_revision_number":2},{"id":"54b572284d2dac5fcbab956fb26237170ca413f9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2024-07-12 05:48:11.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1, Verified-1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":3},{"id":"3c04eecff237e1715cf464a459c7f06348bc1082","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2024-07-12 05:49:09.000000000","message":"Patch Set 3: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"15192f965a0f58f8056e1a8272d14123833c603d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-12 05:51:48.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/67052dc33feb46ccb87747e753809688\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4dccca08b8ad480585993563aa0410f9 : FAILURE in 2m 46s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/cbb9a0731fc24a5da4e957a59dcd1820 : FAILURE in 2m 51s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"e67dc75231de9e71e1386a2ba141c45cf3f094d2","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2024-07-12 06:20:52.000000000","message":"Patch Set 3: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/July-2024/Arista_Tempest_Testrefs/changes/71/923971/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"9da427daf238535a3cbc4be350e17cce09fb0f35","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-12 08:52:48.000000000","message":"Patch Set 3: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/84204c66c75248a590771b7a0944b97f\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/faf449d6fa0b4c6ebe63a562841ad6b0 : TIMED_OUT in 3h 02m 57s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/c4372f4237794892926299acb87904c3 : FAILURE in 53m 15s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ed7efacc2b9444838e4476c8af19b3db : SUCCESS in 24m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c56313f9ed9c4a36a36029b08886f038 : FAILURE in 4m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b7be0735dc9a4b3ebc3eee2bacddb889 : SUCCESS in 23m 10s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a3d9ae653e12498ea93e0f0e4de60dbc : SUCCESS in 20m 20s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/29f89072eb7d4626b3564a5535459a9c : SUCCESS in 27m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0db5ceddb5c74ea9a2da9724ff9d45d5 : SUCCESS in 10m 36s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/3f3fb70545794e2ca4bfd6e82cc4cd26 : SUCCESS in 21m 41s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/d4027260c7034056808f2649f439d76c : FAILURE in 45m 42s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/119eb9088215438d958a491258e149e7 : SUCCESS in 1h 47m 40s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/d250cba75c0a419eaf6f10ac1623a14f : SUCCESS in 48m 20s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/36c20635ae68454e82c684de0ec573e1 : FAILURE in 58m 54s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/9bb14c398395485596d7ab4806ad4dbf : SUCCESS in 45m 37s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/88f7efbf499c465598b5dc285de4f69c : FAILURE in 1h 11m 18s","accounts_in_message":[],"_revision_number":3},{"id":"44fe35197966a21af4b747abac384d212bff3964","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-07-12 16:04:27.000000000","message":"Patch Set 3: Code-Review-1\n\n(5 comments)","accounts_in_message":[],"_revision_number":3},{"id":"fe4746b88c0c1427abe049f822433ad4370525b3","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2024-07-12 18:42:15.000000000","message":"Patch Set 3:\n\n(4 comments)","accounts_in_message":[],"_revision_number":3},{"id":"c9fe762a872f425809c96fa4329d99f912b14ba8","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2024-10-04 08:03:53.000000000","message":"Abandoned\n\nThis review is \u003e 4 weeks without comment, and failed Zuul jobs the last time it was checked. We are abandoning this for now. Feel free to reactivate the review by pressing the restore button and leaving a \u0027recheck\u0027 comment to get fresh test results.","accounts_in_message":[],"_revision_number":3},{"id":"a5d62cd2b05f123179870c77e6b755d194ae39e9","tag":"autogenerated:gerrit:restore","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2025-02-05 15:27:34.000000000","message":"Restored","accounts_in_message":[],"_revision_number":3},{"id":"540599b9dd00e3a538c228a62b85a49272b4fa3c","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2025-02-05 15:27:56.000000000","message":"Patch Set 3: Code-Review-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging opendev.org/openstack/neutron for 923971,3","accounts_in_message":[],"_revision_number":3},{"id":"6a81063d89755d7381978039a103b5f234dd88b8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-05 15:27:56.000000000","message":"Patch Set 3:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/neutron for 923971,3","accounts_in_message":[],"_revision_number":3},{"id":"933da67d19d02121bfc769211d1325b5efa197da","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-05 15:27:57.000000000","message":"Patch Set 3:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/neutron for 923971,3","accounts_in_message":[],"_revision_number":3},{"id":"0c7afefa7d4e75c4336bbbbfaf2c8c99c829101a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2025-02-05 15:48:13.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1, Verified-1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":4},{"id":"96400803b71367024fa15bdf58e79cb6149c5dba","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2025-02-05 15:54:25.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"73f55e97ae58e3ff4c993b2ae3559ee03fd0d05d","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2025-02-05 15:54:34.000000000","message":"Patch Set 3: Workflow-1\n\n(5 comments)","accounts_in_message":[],"_revision_number":3},{"id":"e216de7d0e1bdfb6fbd0679475578e332addfe02","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-02-05 16:32:23.000000000","message":"Patch Set 4:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/February-2025/Arista_Tempest_Testrefs/changes/71/923971/4 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":4},{"id":"a2119987cf50ba958ef9ea74646aa7d1a5fe70ca","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-05 16:32:53.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a3d1da2fd105491783e6af6dae092efc\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3d5d4aa1d5f54053b889042aa2584d2a : FAILURE in 32m 07s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/fca44f30d8224bdcab441cd4782ff325 : FAILURE in 36m 01s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"e953bd688cd27f7cbf5451bb5e9631aeda00d4e9","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-02-05 16:50:50.000000000","message":"Patch Set 5:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/February-2025/Arista_Tempest_Testrefs/changes/71/923971/5 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":5},{"id":"53ce7f36f934192a09d11905cb3a789e9bc4ce0e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-05 18:59:56.000000000","message":"Patch Set 5: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3a69eb0994bf402d9f166eeba577a3c1\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/c7f3ffab18a44fa6b9d040fccbd35e56 : TIMED_OUT in 3h 04m 27s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/4c0ca359318c4f95bc1e735558772298 : SUCCESS in 32m 44s (non-voting)\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dafe6f1e6bc745da9323d8a9c227ddad : TIMED_OUT in 3h 03m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/eb871e3d78c24a1d8a4ebbdd9b6e47ea : FAILURE in 50m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/63e6825927d8472582a04112eacf0f00 : FAILURE in 8m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e344044365a8436da23385316a97f097 : FAILURE in 22m 28s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/dc523852f6234301aee585fca76565ed : FAILURE in 18m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c34ca8ec09f34a05903ce9d848e46da5 : SUCCESS in 9m 31s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/06cf7be4834a4bfbbdbe97e39232092c : SUCCESS in 1h 11m 01s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/f4252df5196d4e379e67a359f4d34324 : FAILURE in 34m 48s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/d733dd3c7e4d465ebef8009f26144194 : SUCCESS in 1h 53m 38s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/7265fd220594494ba1222547fd63c78b : SUCCESS in 1h 04m 31s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/19256845190546a4864183f9ecc76990 : FAILURE in 1h 10m 52s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/9671d463358a4aaa92c109d232336535 : SUCCESS in 1h 16m 04s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/12cff71b551a4b0d9dd8bb88bd7fd303 : FAILURE in 1h 01m 32s","accounts_in_message":[],"_revision_number":5},{"id":"f20d30681bd10e1433c8095fbb985350f9809ca1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2025-02-05 23:11:35.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"0c7d6a9d5181ef265b17cc34452bfce61efb215f","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-02-05 23:32:22.000000000","message":"Patch Set 6:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/February-2025/Arista_Tempest_Testrefs/changes/71/923971/6 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":6},{"id":"0d1c12c829dbda10509ad91d2905e8ee1a92ada1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-05 23:46:17.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1a93791ff3d440bdab96a1f0df326692\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c79e2c3ec857405597e32d83d26486e2 : SUCCESS in 30m 43s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/4a0045f273974b92ab1c8bc92dbb1642 : SUCCESS in 32m 55s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"efb69af9e7ce46130f5f7f8821c7c2ed03517303","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-02-06 01:04:43.000000000","message":"Patch Set 6:\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"8be39cb8f4a591645e521ce277b95a1e3cb951e2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-06 02:21:34.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7de6c1821d81424f858f62128a479250\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/4f1429ec2ce3479f83984fcf0ebf8e73 : TIMED_OUT in 3h 05m 01s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/b1368a94c0ce4067b24182e3c3030b1c : SUCCESS in 46m 59s (non-voting)\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fa089ac5fc9f47518565ba06b3f2c8c8 : TIMED_OUT in 3h 03m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ee00485874604f469b278a6be7f99f12 : FAILURE in 47m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7196e899fba54faeb3613dde7e8411e6 : SUCCESS in 11m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9adeb201c0b34c1a92cb0618705c64f2 : SUCCESS in 22m 16s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/db15f19465a5418ab464f01dbaa1a8e6 : SUCCESS in 21m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2755be4cb7d14749a1b9f2263bfee624 : SUCCESS in 10m 50s\n- requirements-check https://zuul.opendev.org/t/openstack/build/a0a41ac8694d4f9c943a3484d56820b9 : FAILURE in 3m 31s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/48e8a399e0bf41a881960f0b4fbb98c8 : SUCCESS in 45m 46s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/bffe4790a6134bff8a479c10b59ac8d1 : SUCCESS in 1h 06m 59s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/3113da1b62f5490b87b8add2383c4946 : FAILURE in 42m 23s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/886ac01c117f4625a98c77313c81d65c : SUCCESS in 2h 15m 22s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/ae07ec1e0b0a41a3b6df5e6a8756a5ec : SUCCESS in 1h 10m 48s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/eb23e6f1695e4109b96bc0332ef8f86d : SUCCESS in 1h 27m 23s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/3c399061b33b44a3b2ddb64606158a4c : SUCCESS in 1h 52m 41s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/c7a17878e796422697bdea9f3c1cd021 : SUCCESS in 1h 09m 06s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/8c121884b9254d05a477eee540f586c6 : FAILURE in 50m 30s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/690339b75bc3457baf4917bf221c3192 : SUCCESS in 51m 25s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/aa70ff3744d44b47b39d11699075e918 : SUCCESS in 49m 57s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/67d38d1880dc432fb87bc733c809dce2 : FAILURE in 54m 47s","accounts_in_message":[],"_revision_number":6},{"id":"e3d20681b3c4aa365f2bdd5af04ac191b2a486fe","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2025-02-06 02:47:45.000000000","message":"Patch Set 6:\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"b1c197833b291e9de799e076ac1487952e1ffa89","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-02-06 17:44:11.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"7779f1b54be9dace9047af09d0cd9ddebb45d0b1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2025-02-13 20:31:29.000000000","message":"Patch Set 7: Published edit on patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"287644640c956ff6b84182884fb0acc741b74c26","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-13 21:05:27.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/96111bd74b374971a60302e32e9c0a7a\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a4feef4d7dbf4489a52ac4f2d60238e8 : SUCCESS in 19m 07s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/346476b28cb946b88f0750a59743b15f : SUCCESS in 31m 49s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"6d15804b4c6ec898d8a8d7aeddd968fb1883967d","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-02-13 21:10:51.000000000","message":"Patch Set 7:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/February-2025/Arista_Tempest_Testrefs/changes/71/923971/7 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":7},{"id":"5586bfb49a10425ab4201ee25bfc5bcf5ea32ca6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-13 23:37:41.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/86739702a7114c00a1712b2e9690f39b\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/c4f71af2be064b1395a9c9a890904eb0 : TIMED_OUT in 3h 04m 23s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/8bd08cafef8f4ef68450a19a22a62527 : SUCCESS in 43m 05s (non-voting)\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a00e029c8ce5487585b851f3dae51675 : TIMED_OUT in 3h 03m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/96d89472377c4cecad10ad95734063cd : SUCCESS in 32m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b9e23793556644e7853d4d04ac0c34c9 : SUCCESS in 11m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3efbb342190c4903a4d34b5cf3f59c3f : SUCCESS in 21m 52s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3919f011028f4d75944ba60d8fce822e : SUCCESS in 17m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/07f1b17d2cd44616ba20ddce1d776b88 : SUCCESS in 9m 38s\n- requirements-check https://zuul.opendev.org/t/openstack/build/5c448c2ac56d407195a55a709baf3574 : SUCCESS in 2m 58s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/84a57a01753946149f1804314b68112a : SUCCESS in 1h 01m 35s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/cc3e3d53f39443cda15a62612bdb27aa : SUCCESS in 1h 11m 50s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/1b01541b452b4295b454591619e3f2e4 : FAILURE in 29m 18s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/975e3e0866784801bfb625a623a8b1b3 : SUCCESS in 1h 57m 42s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/b43431c8d27a4c96b7e893fb199829e6 : SUCCESS in 49m 08s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/4effc18612e747e4893001566d732cae : SUCCESS in 56m 50s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/bf77aa71585d4729a16ab7923722554a : SUCCESS in 1h 38m 54s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/34b0c31380a6439fa749ae3ec38ae72a : SUCCESS in 1h 38m 27s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/1af642021fa947c8b2001f070122aea1 : FAILURE in 1h 18m 58s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/d93a34ec8bf544008b517d7a145991d1 : SUCCESS in 1h 06m 44s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/518e34eedbb44c5caa098f9d694c0b3f : SUCCESS in 1h 01m 40s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/c85942522ed4466ab6f58283fbc30696 : FAILURE in 1h 12m 13s","accounts_in_message":[],"_revision_number":7},{"id":"b418dda9901fd3e29e9b91ca20f68ee073e89022","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2025-02-18 20:56:13.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"7f414de5be5a88ddb4dd9330c521cafffa333b60","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2025-02-18 20:59:00.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"3fede6a4d47c871c6f2f943cf840ef4b00ce72bf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2025-02-18 21:16:43.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"b7a0db29782cc92dcea395b79ab5330afff7132e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2025-02-18 21:29:49.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"08cfdb4c3e8b32156dd3ed94d3410ae54785bf31","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-02-18 22:01:52.000000000","message":"Patch Set 8: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/February-2025/Arista_Tempest_Testrefs/changes/71/923971/8 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":8},{"id":"45ce774a4dd45a4b8612c18792d107da5ffaad97","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-18 22:04:01.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1d1f986e648f4c4ea926d435c40db905\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/dd2cb3e4b895455f8449adcb5f7abd07 : SUCCESS in 20m 36s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/500eb595f2d54ecdb536bbba6abdbdae : SUCCESS in 32m 13s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"4f022f034dd0e6653cad3cc672195bd08fea40a7","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-02-18 22:35:44.000000000","message":"Patch Set 9: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/February-2025/Arista_Tempest_Testrefs/changes/71/923971/9 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":9},{"id":"6ad09faf0c37eed881b9f36746be19acaf746800","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-02-18 23:09:21.000000000","message":"Patch Set 10: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/February-2025/Arista_Tempest_Testrefs/changes/71/923971/10 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":10},{"id":"4997ce08059e13d5d324205a4a56d099534618cc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-18 23:50:25.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8e7efa4921f147189b2e95d4930f1a24\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/53ad0b547b6a418c8d805004107b025d : SUCCESS in 2h 04m 29s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/ba9b70f02f5d4dc1bd5e3d3326506188 : SUCCESS in 35m 14s (non-voting)\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e9997b5a29db42e0b26d3f608ea19b3a : FAILURE in 2h 17m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/463212e1400a42c093698a2fd76646ef : SUCCESS in 53m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4bdd9c5e97d846a49c5cbae9895388c8 : SUCCESS in 15m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b8909a44a3224349855f8a262be00990 : SUCCESS in 21m 43s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4f332b7d300243368ef4ac35c8d0f90f : SUCCESS in 20m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ece9cf635b14eebaa4f90fc27c65f5f : SUCCESS in 9m 37s\n- requirements-check https://zuul.opendev.org/t/openstack/build/b54e137e69914bcbab680be6d1b9f8b2 : SUCCESS in 2m 43s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/bd04c38014a241d5b5d3ac9ef579271d : SUCCESS in 45m 39s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/fe2864d4cb9243f285ccfd9eb66df55e : SUCCESS in 49m 30s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/ffea92a193564aac8d699dbeafe766ef : SUCCESS in 1h 12m 28s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/e828333cff5a43b6b3b9dce429248484 : SUCCESS in 2h 02m 01s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/ad2c7b32215d4b48afef4f295f713a4d : SUCCESS in 45m 32s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/0ea1e8b3f0b542579c8b8f38f31792ee : SUCCESS in 1h 12m 05s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/b5cf2b688013485f93d4b557394d9f5f : SUCCESS in 1h 47m 29s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/5c24d17365ff427cad10619675d265b2 : SUCCESS in 1h 28m 24s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/16494c0be80e4280b8a3209fe8fc33ac : SUCCESS in 1h 02m 39s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/564f798e5ac44f1bbde75671144f7be4 : SUCCESS in 1h 01m 10s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/28cbe0b87f0a4e45b6b4bd26cf4c67cb : SUCCESS in 1h 15m 12s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/57eb709fe0ca4f72a7dc03251d82a7b6 : SUCCESS in 1h 21m 29s","accounts_in_message":[],"_revision_number":10},{"id":"b25df7f45790d537eab19cda91c62de3d2cda121","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2025-02-19 03:06:08.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"1e504d361fc24a1c01bcd2fa018daec275d8c521","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2025-02-19 03:24:54.000000000","message":"Patch Set 10:\n\n(2 comments)","accounts_in_message":[],"_revision_number":10},{"id":"727234c264d96d101e3dd4ad1acfd3cafeb830b5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-19 04:10:50.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5fcab6dcc5424e118c259fabe20acafd\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/bb2bcb981abf4ba7b607f829692bd681 : SUCCESS in 25m 26s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/267852d3e2c44b56818e37e780f73f76 : SUCCESS in 40m 38s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"117588c27988d21c9391d499aaa62ec99762e7b9","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-02-19 04:38:17.000000000","message":"Patch Set 10:\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/February-2025/Arista_Tempest_Testrefs/changes/71/923971/10 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":10},{"id":"083c42ffd18ba4d0aefc42ca83bd5c8a6bae8c4a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-19 05:16:07.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4ad0e95e301a401e82d77aa15db87b0b\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/e02f322c8937431d8bb279883401bb52 : SUCCESS in 1h 49m 50s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d2ffa5e12d27425eb0cd20763a4227c0 : SUCCESS in 41m 37s (non-voting)\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f3f5c1c5675a4f1595974e3c4708dbb9 : SUCCESS in 2h 06m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3b881f04f70f4c8bafb02c33217f61d4 : SUCCESS in 48m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b3b98895c9cb481a8121b6e8a3ca1409 : SUCCESS in 11m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/241b736acf87459487983dbeb6f4d275 : SUCCESS in 21m 07s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5a7d7d801326486782312eac39a602e6 : SUCCESS in 22m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce72d9771cd44a96ad8f4932b91b75f2 : SUCCESS in 9m 52s\n- requirements-check https://zuul.opendev.org/t/openstack/build/553e820b2bc04c3e8d6a28e2016c038a : SUCCESS in 2m 48s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/8143fc209f494ecba97976c336e00757 : SUCCESS in 59m 48s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/1c535e665a324ff3a32e44bbb196cc53 : SUCCESS in 45m 55s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/b48a4bba9d924d928cd33d9c468e796a : SUCCESS in 1h 04m 55s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/8bcf853a13f146c1b6a9341c7ff6e869 : SUCCESS in 1h 29m 09s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/6aff3fd7a88b46178543d145e22be61d : SUCCESS in 49m 11s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/0e1c312a97574255ad0b9d1873c4eacf : SUCCESS in 1h 15m 25s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/000c9550851340b48f07048a43443308 : SUCCESS in 1h 26m 08s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/06b9838c7b4240659f75582d4428e154 : SUCCESS in 1h 15m 51s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/24c576e388e24991a0d651e2853ca88e : SUCCESS in 1h 01m 44s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/65f9f8c5bced4e4ab22df49bd6f9c31f : SUCCESS in 1h 18m 49s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/e4567a4a3b014220b2ce018bad56e4ec : SUCCESS in 48m 11s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/4a51c7ce6bd04da4829a7f28574cc911 : SUCCESS in 1h 19m 18s","accounts_in_message":[],"_revision_number":10},{"id":"3f90dbb3d6445504087b0698fc31aed21c25f6ac","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2025-02-19 20:17:05.000000000","message":"Patch Set 10: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"3d9f1dfb252a53a2ba9688cda2a0610c5207d2ad","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2025-03-06 17:00:44.000000000","message":"Uploaded patch set 11.\n\nOutdated Votes:\n* Verified+1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":11},{"id":"13a4f29af42b580365cb1ea0af8c84776abaddd2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-06 17:29:49.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d960e0fb7a4d4e8baad6a77eb86c94fe\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1bddd916b1ee44e48a574b82924ac1ca : SUCCESS in 25m 09s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/7b1ccc37f23a45c2bd5544b21ec0a1e4 : SUCCESS in 24m 59s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"3ac77b4145d4d1c3afb7e22f8627517dece95770","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-03-06 17:54:50.000000000","message":"Patch Set 11: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2025/Arista_Tempest_Testrefs/changes/71/923971/11 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":11},{"id":"38954974c6ed259697de87e0570cc9a460d6d072","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-06 19:26:52.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/17fdcf2c4de74e0493d764c34bc5614e\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/9dd0bff285144e4199de02768fc3f4e5 : SUCCESS in 2h 08m 13s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/b8ba48ea45094ce38b43558028f3a390 : SUCCESS in 36m 35s (non-voting)\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/084d659cf4474c42b28c5f970832ade1 : SUCCESS in 1h 46m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8a611f5ea34747aa8a5323989abe4ee1 : SUCCESS in 52m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f6de024f54d24eea9046e18e30b239f7 : SUCCESS in 12m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5009f764a3d14cb38e99d68af527e5f5 : SUCCESS in 20m 33s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1f726d4b702a4e60b20ca13b76d2764b : SUCCESS in 19m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e575316f9b184722a7eb1de60f53fd70 : SUCCESS in 11m 18s\n- requirements-check https://zuul.opendev.org/t/openstack/build/2a6647e64e474bff816e54e4e55a70bf : SUCCESS in 5m 39s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/34747bce08bb4abe89f7abc2009bebde : SUCCESS in 1h 01m 21s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/bf6c45cc3c6b4ba89c600b2d3b9413fe : SUCCESS in 1h 10m 27s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/b9da34cac70240218e29ed1cb4d05b5e : FAILURE in 24m 17s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/a078203515c54593bf56fbca43a45417 : SUCCESS in 1h 43m 00s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/9f85435ff57e43bbbc06e4d1f72b15d4 : SUCCESS in 1h 11m 00s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/87aafe60e9a74e0b834d43baac8936b1 : SUCCESS in 1h 03m 18s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/6816c7ce360f460d90e6855cddaaeb39 : SUCCESS in 1h 39m 34s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/b244c49b2bb144618a8d72a3ada83014 : SUCCESS in 1h 26m 33s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/4f845efc9b3e46bdba5281a429a29e96 : SUCCESS in 53m 03s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/abbe4b670d08414e837228dee44d9827 : SUCCESS in 49m 59s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/b605cab73adc4c19a9ae8fffffd6f17a : SUCCESS in 1h 11m 42s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/c55703041d6148acb4868b9520d7a37b : SUCCESS in 1h 17m 47s","accounts_in_message":[],"_revision_number":11},{"id":"cdade2f5f24807d485ac2de5db22c8214ee8ca85","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2025-03-06 19:33:33.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"94a33da364fbd37a660032e30c30ec4de28591d9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-06 20:11:11.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0a27b5f8d2ed4698b33194d4819d2f68\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d42db16e5ff440b8a5d46d9068cd437e : SUCCESS in 35m 32s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/596789cf49084026bb075fd65bc7b7dd : SUCCESS in 32m 56s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"e0de667fe451647091533793e7859c00fbbc4058","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-03-06 20:13:51.000000000","message":"Patch Set 11:\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2025/Arista_Tempest_Testrefs/changes/71/923971/11 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":11},{"id":"dcf64d24005bfa3124a622f4164587f42e5c2671","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-06 22:30:14.000000000","message":"Patch Set 11:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/87ac936cdfdd4f8b8f245242d14cb97f\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/fbf0ffd690964f7991bd47d0aa5e1663 : SUCCESS in 2h 25m 25s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/3dd9ae616620465d96c135737d3bb718 : SUCCESS in 39m 17s (non-voting)\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/640abfbfaeac4ccfb64bb94713307bf2 : FAILURE in 2h 55m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0fddbf5b6f114b1c9b1d6c3849e3c0c8 : SUCCESS in 49m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f0c7e7af3af74b71a49a24a4b6fd78be : SUCCESS in 11m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dd58677ea571459a8c9780e95ac1e825 : SUCCESS in 23m 17s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4dfc6c392d55464cbb491cec4aacf479 : SUCCESS in 21m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a4375696f7db49e88737ae0225663f7f : SUCCESS in 9m 25s\n- requirements-check https://zuul.opendev.org/t/openstack/build/fac2d315cc6840a695fb62e8f9ea1de6 : SUCCESS in 2m 51s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/2e53db76611e478982711ec24fcb3e80 : SUCCESS in 46m 15s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/56e48637ad1d43a690d4f214aeb0229e : SUCCESS in 52m 35s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/ec505c16bda94b568dd167e83a841770 : SUCCESS in 1h 16m 05s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/38bc064b905d4c4a9de6ffa44b04987c : SUCCESS in 2h 09m 47s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/1c589be4227e4902a3b60695060ff964 : SUCCESS in 58m 58s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/7f861943915a4f48a6565da686269028 : SUCCESS in 1h 16m 17s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/d41d9014422b4fbbb6521f30bafff178 : SUCCESS in 1h 30m 22s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/ccb6c417e08649a8a234111088a93eb7 : SUCCESS in 1h 24m 00s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/a0d97bedfcf54500a970dc7184075ab4 : SUCCESS in 1h 08m 45s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/4ade8dbe007246f6a148a5ac28879f49 : SUCCESS in 1h 14m 22s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/ed97d096aad9449aa2ff7443982ec25e : FAILURE in 1h 47m 03s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/bd36c6469d0d4e159f80f8bb193b8851 : SUCCESS in 1h 02m 50s","accounts_in_message":[],"_revision_number":11},{"id":"078bc21e156f1984b8b23a4521967612ce21ea53","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2025-03-06 23:58:41.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"e2dafae232ee982067615b69a4d2f9a5bdd6d938","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-07 00:56:38.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a8a53ed5f9cd4fd99eeb3dc419b56829\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/129897772031401a96c958c14c561d6a : SUCCESS in 23m 56s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/7d8794705fd540068a4a8b79a3e23106 : SUCCESS in 28m 59s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"a591203b3f2612132fbf3170a6857974ca4c8241","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-03-07 01:12:17.000000000","message":"Patch Set 11:\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2025/Arista_Tempest_Testrefs/changes/71/923971/11 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":11},{"id":"adfa0668b8e07c53d34f0012b4251ac389de4392","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-07 02:16:31.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dcf1a76e3a534f9a9b8e5091f25f36e6\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/66b177ca38eb4cc1ac8406ee919e9268 : SUCCESS in 1h 26m 57s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/a959bab395be4c7ca36b2080b5cfb71d : SUCCESS in 19m 42s (non-voting)\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8caae5514d2e46f5a67e3260415c1c47 : SUCCESS in 1h 25m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/88a528bb5d524b358d2ec8f2c46788af : SUCCESS in 26m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dd4f4c3fd4ff45b2b6b898943772d35b : SUCCESS in 4m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cd6a4e4011fd45cdba6b1df99f5c00f7 : SUCCESS in 14m 21s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8a28d358bcde428a9f5f46082e4db3db : SUCCESS in 16m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bab59dfc348b462b93b1d0a2b7d92f66 : SUCCESS in 4m 33s\n- requirements-check https://zuul.opendev.org/t/openstack/build/ae49294806984dfd80a65c9c9392adbb : SUCCESS in 2m 24s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/ee742945f3a240c391b1e7c9610ebf61 : SUCCESS in 43m 01s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/2b9e2804330c4440aeb558dfa663842c : SUCCESS in 48m 48s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/0f8c967b16d5418fb2cd2437071b181b : SUCCESS in 1h 19m 55s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/3f7249f3ce7248eaa3cdbbfa18321719 : SUCCESS in 2h 16m 44s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/067bb922c1cf4dc2bdeab78b2e1cca41 : SUCCESS in 51m 07s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/9e6793b2caea412e92b0421336349194 : SUCCESS in 58m 29s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/b2253935f7364017b8fe373c541ea147 : SUCCESS in 2h 00m 29s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/2c295f3cfeef4341b40d2bbd35760184 : SUCCESS in 54m 48s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/754b89bf1eb7420a8c7416995b481ce7 : SUCCESS in 1h 08m 21s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/78ae60840056459ebc0a994ef9cdd3db : SUCCESS in 27m 35s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/fc912d672e0d40d3af16cee91743cb01 : SUCCESS in 1h 13m 41s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/61dc533c623246ba8dfd8f7000fbda00 : SUCCESS in 1h 31m 36s","accounts_in_message":[],"_revision_number":11},{"id":"cd71e23f3e81abae7048945f877591ccc0c89691","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2025-03-10 13:09:14.000000000","message":"Uploaded patch set 12.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":12},{"id":"ed6376e580e95716e69dc514ee4152719488a9fd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-10 13:41:53.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b951844594474fa39d5d33195b1e03e6\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d5f20a2784fc499db59fa41e03526c05 : SUCCESS in 23m 15s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b4633a0da5ad4453a6bd11031f59b9c5 : SUCCESS in 29m 54s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"396393f4351f538ee19d1906d6f4b65d7a88b5fd","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-03-10 15:03:57.000000000","message":"Patch Set 12: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2025/Arista_Tempest_Testrefs/changes/71/923971/12 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":12},{"id":"edcfdeab174c8f69a40315975f15e6ec16aac11a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-10 15:32:16.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ca1c91518ec94c75b78b3e85df8e43ef\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/5212857a9a63475091aaeccc795eba08 : SUCCESS in 1h 35m 14s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/e8d26573190f4da682f116e16c3b9214 : SUCCESS in 21m 29s (non-voting)\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8918b0c6228f4e049dd6406e64f1e29a : SUCCESS in 1h 42m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ea3dfcce1c7d4ad881e37f40f4ac1a05 : SUCCESS in 54m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/077b42d72af34717bfdbcc9a85470b20 : SUCCESS in 11m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dbbf314b71df49b99e319d7c3b998e8b : SUCCESS in 21m 55s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a3ee5aa8bf474bf8b4bfefbda74a6b79 : SUCCESS in 25m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9db5c96f1d444ce7aaf40d1788eb06b4 : SUCCESS in 9m 09s\n- requirements-check https://zuul.opendev.org/t/openstack/build/8f01f4e57391485f983c37ee490be587 : SUCCESS in 2m 34s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/189b25a829474b2a9f92db0db518e2a6 : SUCCESS in 1h 02m 33s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/e3f3d70d388a4f86b073c1ec42668574 : SUCCESS in 33m 17s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/f6d88eba01244f81848ce1eb1dcf89c4 : SUCCESS in 1h 21m 25s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/6af810e2bbb54836a336277c45ea17f8 : SUCCESS in 2h 08m 32s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/db0c078c41c24c819e3dfa162babb5c0 : SUCCESS in 1h 03m 12s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/75f5733c15af4e468a9c16980c059f83 : SUCCESS in 46m 21s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/6983b88e8ad0473689a5b13416824674 : SUCCESS in 1h 07m 08s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/ae9597564c5e4f648234ba2ea2807027 : SUCCESS in 1h 41m 09s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/1542567d3a5a43fab2c4cf075e412227 : SUCCESS in 1h 05m 21s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/fc8f5f2500ba4d778ef6a0b586e62ca1 : SUCCESS in 38m 05s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/2bcb54d5625d49a08a89ec1562199806 : SUCCESS in 33m 39s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/02ae23762d7f428580b87595d9d817e3 : SUCCESS in 1h 29m 56s","accounts_in_message":[],"_revision_number":12},{"id":"538a260ff440d89334cf1969e8435763e9459eea","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2025-03-14 18:57:44.000000000","message":"Patch Set 12: Review-Priority+1","accounts_in_message":[],"_revision_number":12},{"id":"555f50edc829d40e9e7e37f68049294186511081","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-03-17 12:11:07.000000000","message":"Patch Set 12: Code-Review+2\n\n(5 comments)","accounts_in_message":[],"_revision_number":12},{"id":"32345dbdb9e421347209a55dce96c4842becc071","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-03-17 19:08:18.000000000","message":"Patch Set 12:\n\n(3 comments)","accounts_in_message":[],"_revision_number":12},{"id":"b99ec653828bfe09e3378eb4fa7f5bfcf4fb54f4","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2025-03-19 19:45:26.000000000","message":"Patch Set 12:\n\n(4 comments)","accounts_in_message":[],"_revision_number":12},{"id":"23f7dc7853b39726d51acc91cc525532e6fcfe7f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2025-03-19 19:46:03.000000000","message":"Uploaded patch set 13.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE OR is:MAX OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":13},{"id":"b555bb9cf091bacbdc7cd7c33a8e7719b3d451ef","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2025-03-19 20:06:52.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"ba8a4079377ede6e5787445db4f2ac9b691dc1fd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-19 20:11:58.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/07752e169bf74cbebc3c7e386e063df2\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/98ce04eddc664f04a8becf1e633e136b : SUCCESS in 18m 36s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e752e032b756470b82cb72daec3bfab8 : SUCCESS in 24m 02s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"4b239f651b057c1e15075e93339f666b6c307c75","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-03-19 20:59:27.000000000","message":"Patch Set 13: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2025/Arista_Tempest_Testrefs/changes/71/923971/13 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":13},{"id":"190b8a3ed155901486fd1f7c486e6a239dabccb4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-19 21:56:35.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8c842fb1673f4542902bde5e8cdf6d02\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/16815650073b42dfaf6491a2bebe915e : SUCCESS in 1h 23m 03s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/f064540aa2244b45903b80e263fe37e7 : SUCCESS in 43m 21s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4a71030702044d7db91c7a7736363fde : FAILURE in 2h 07m 47s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/baa723cbd63d43c49dddf67b45a6d97b : SUCCESS in 25m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/84bbd9c05cb1470b80edcd4de989f046 : SUCCESS in 13m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1067ce79a8e741cfb8b9535e1ae83354 : SUCCESS in 21m 37s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1a7411fd799743779ff49f742f3ec0ad : SUCCESS in 22m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df8efa54e22847879c9e40c4a5b7a5cd : SUCCESS in 5m 36s\n- requirements-check https://zuul.opendev.org/t/openstack/build/1d6a15c1af6e49cdaa94ac480d7c6617 : SUCCESS in 3m 00s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/e42bc2db25794fa398186a10003d8804 : SUCCESS in 1h 20m 24s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/ec62eb3902a2401997f8390c9987325b : FAILURE in 1h 43m 39s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/a9c5d58cc6374f5ba6514bbad1810638 : SUCCESS in 47m 42s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/8733aaea99064fbeab4e8c13986cc5b2 : SUCCESS in 59m 58s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/05496e280bb34a71a1acdd3dd029c58d : SUCCESS in 1h 35m 51s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/d014266f0abc4dac8f2b3811c964e67b : SUCCESS in 1h 23m 42s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/eca95132277c49cfb42e6fb46e3a55c3 : SUCCESS in 59m 14s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/9f935188bb164e46a4fe42034af2e2d5 : SUCCESS in 1h 02m 00s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/dd4f7086949d4fe2864d42bb0cca9b1c : SUCCESS in 1h 20m 20s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/1f0b1df84f1e4af2b7fdcab41080974b : SUCCESS in 1h 00m 23s","accounts_in_message":[],"_revision_number":13},{"id":"4b0b6f9f2835cf654f1ec7eb3cba6ae1fdaa5479","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2025-03-20 20:21:15.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"c1c061575e2099fdf0dfd4a98e7f939ce0a4f3e8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-20 20:54:02.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fedd00a9807b45078e8417af8d938df6\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/817fdac4455c45edaad82f1dff8adc6c : SUCCESS in 16m 47s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/8372dec9ce2a433c91335f5d185800c6 : SUCCESS in 30m 09s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"4ced00ad217ae9cd5597d5333ce2c08f82248c29","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-03-20 20:59:23.000000000","message":"Patch Set 13:\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2025/Arista_Tempest_Testrefs/changes/71/923971/13 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":13},{"id":"9804f8789c855a8b346b8d4b3da3ff2e0407662f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-20 23:13:43.000000000","message":"Patch Set 13:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/374f90ba998542e08c3c0995a5b78f80\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/4be1015962be4d14a91f76fb48a3a286 : SUCCESS in 1h 41m 51s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/65fa378a8c9d4fa5ad16c24b54af2e15 : SUCCESS in 34m 38s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/52ddbc85e1344d488ed3a2f999220d14 : FAILURE in 1h 50m 37s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6403bc25938d4597bc8e8aeb8b4f1e41 : SUCCESS in 59m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9d218dc7a75e4c2a836cfcecd4fcffad : SUCCESS in 11m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/629cb07b624f4300bdccb8623fa29c99 : SUCCESS in 20m 33s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/98ab0ca9be9e43afbb04ff06cf6ffc9f : SUCCESS in 14m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b82cc96b9c54495a034cded21d84190 : SUCCESS in 13m 05s\n- requirements-check https://zuul.opendev.org/t/openstack/build/baa2b3f9d6684fa78372929ea60fdeb6 : SUCCESS in 3m 02s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/7d4049d97d5c44f8a24067da424b9302 : FAILURE in 24m 18s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/9b8aea53a306486bb224de0b42d7287a : SUCCESS in 2h 30m 51s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/0f97d6ae87694f749a443b692f814995 : SUCCESS in 33m 50s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/ccc3298b390b4ee6ad5c0bedb79c7bec : SUCCESS in 1h 00m 47s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/ece0c483c1ad4f60b2439e4b192d2cc6 : SUCCESS in 2h 44m 55s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/927432b55bc349c1994754346c82b841 : SUCCESS in 1h 43m 18s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/5a45c275a75b4b238c452e91880d73e1 : SUCCESS in 50m 12s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/6d1bec8ebb144c02a85a114c0a19ea6b : SUCCESS in 1h 45m 33s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/e96a2219c6254017ac44d443642f484e : SUCCESS in 1h 07m 53s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/2000d6964120445e9a64063da44ff8a6 : SUCCESS in 1h 19m 14s","accounts_in_message":[],"_revision_number":13},{"id":"1da100fe81bfc65acd5a62f541994b51482907b2","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2025-03-24 14:23:12.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"941b437f4880b02d2aae3d91893864f2acddd979","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-24 14:56:18.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/351c455724484777a4629ca5957c3d55\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f1c46cdf08554dd7b12fd8d0e6d34c56 : SUCCESS in 31m 23s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/2b0fea4d82384df7969e36ebd161fad8 : SUCCESS in 19m 08s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"a6d4df0995c369e35dbc96780f9468587a0e790a","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-03-24 15:26:08.000000000","message":"Patch Set 13:\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2025/Arista_Tempest_Testrefs/changes/71/923971/13 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":13},{"id":"acba4dabfe93e9306b390b12259deab17938d4eb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-24 16:56:35.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a78e2ea2d58f45b1a2572c95d8a73763\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/251529be10204a8f873d439889df4f3d : SUCCESS in 2h 31m 44s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/24b35bb36a5747469bc5ce709b668f2e : SUCCESS in 37m 17s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b00b2ee93f2644968278621703eda022 : SUCCESS in 1h 59m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b0e0b23b837342de96725d875d537363 : SUCCESS in 32m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/32d6196ad26f4c13a2474bae7eb41a4d : SUCCESS in 11m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/868c1a11d3884339a9d62e5d1aaede94 : SUCCESS in 23m 55s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/547977fdc3074d41a54cb1b9fa266641 : SUCCESS in 22m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3cbed9b02df841969b1c55b7283e94f1 : SUCCESS in 13m 25s\n- requirements-check https://zuul.opendev.org/t/openstack/build/cf53e332043c49a0b78e1cf21ee613aa : SUCCESS in 2m 49s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/9c8904e677bd4f4a9917d3317ee6021d : SUCCESS in 53m 56s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/c2bd1038571440f7a49b279cbf4239a2 : SUCCESS in 1h 41m 49s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/80738412d141469991c05164507806ef : SUCCESS in 54m 03s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/7078972195204e69810adbf732ac32d2 : SUCCESS in 1h 02m 40s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/f121e400f729432c9df90f051590301d : SUCCESS in 59m 56s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/89449803b3ff4b768ad27a1e8a3c610b : SUCCESS in 1h 23m 30s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/e8c96817eb1f4e1d98f38f41babe33b3 : SUCCESS in 51m 34s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/5728070bf02643cdac640009c70b4cd6 : SUCCESS in 54m 24s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/35fcbe76105c4b3f9b2a9c3e67bf3f04 : SUCCESS in 1h 05m 15s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/744aac5f8bef4f509692538d61be5a43 : SUCCESS in 1h 14m 10s","accounts_in_message":[],"_revision_number":13},{"id":"4effd48c7992edffe6278215ffa584b5ff6f23ab","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-04-04 08:35:42.000000000","message":"Patch Set 13: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"8c8aa87770c5f4fcdb8be5cec6c4d7babb0ea0b1","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-04-04 19:42:35.000000000","message":"Patch Set 13: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":13},{"id":"734f51a7adf00de746f4fe57e16fc5ecbfdd6cf0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-04 19:43:18.000000000","message":"Patch Set 13: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":13},{"id":"f70fc6002a54e30d371b75a82a304e1d1de0d059","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-04 22:16:56.000000000","message":"Patch Set 13: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d52c35ae8484489f98b0cf41a98ac915\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/48eccad907f6413ea4ad0dc6da7ad7c7 : SUCCESS in 2h 32m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/db19c604b4a146c7a72bf927b493834f : SUCCESS in 9m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d81e25e0af5142258c718f699783ef36 : SUCCESS in 18m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/74a656d847ef4882884779fdcecab30e : SUCCESS in 16m 13s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ce9c59ca69484a09a8d87377d3f862ff : SUCCESS in 14m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3194da30f7ac477ea4f2dacf51e3fe7f : SUCCESS in 9m 41s\n- requirements-check https://zuul.opendev.org/t/openstack/build/fcddf53551c44611885c74a1f344c786 : SUCCESS in 2m 25s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/4f7f8b27024249798e5d0d9fe952e57c : SUCCESS in 56m 44s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/01ad2fc3992c4f8fa24530370d364b1e : SUCCESS in 1h 13m 24s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/a1245ecf36254a07a9528b17551a1339 : SUCCESS in 1h 05m 55s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/4d95ebb764604c01a15b6eea51b524d3 : SUCCESS in 32m 15s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/83ec6633db77460e83d1a7d0595afc4a : SUCCESS in 30m 10s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/7cc5b3f5dab64802ab5f8b2f44a0629f : SUCCESS in 1h 00m 14s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/877e14b5adfe4745b254641409c574f4 : SUCCESS in 1h 09m 52s","accounts_in_message":[],"_revision_number":13},{"id":"5e0d3c05300788f07af81df1785fcaa6ed11861c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-04 22:16:56.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":13},{"id":"34c53a9e29b9cf021b49e85b1f6b8dc19d4a3d2e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-04 22:18:27.000000000","message":"Patch Set 13:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a5d0ee16fbb44356b99207921c4053a2\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9da8e73837bf416aaf3056cd20628b95 : SUCCESS in 1m 11s","accounts_in_message":[],"_revision_number":13}],"current_revision_number":13,"current_revision":"71ff8ee881a2c387ff318df92c61e7b8333e7014","revisions":{"cf8bce4ebf9ea7745800f07d4266f7eea4f0ebcb":{"kind":"REWORK","_number":1,"created":"2024-07-11 16:11:54.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/71/923971/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/71/923971/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/71/923971/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/71/923971/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/71/923971/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/71/923971/1"}}},"commit":{"parents":[{"commit":"cfab008eef3d11055ad16b8dd0d49aaa1801bc95","subject":"[OVN] Enable the WSGI module for the OVN mechanism driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/cfab008eef3d11055ad16b8dd0d49aaa1801bc95"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2024-06-11 16:37:11.000000000","tz":-300},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2024-07-11 16:11:52.000000000","tz":0},"subject":"Avoid race condition by using persist_uuid","message":"Avoid race condition by using persist_uuid\n\nWhen creating a network and adding a port, the API requests will\noften be handled by different workers. Since each worker maintains\nits own internal copy of the OVN DBs, it is possible that the\nrequest for creating the port arrives at a worker that has not yet\nreceived notification that the Logical Switch for the network has\nbeen created. When the code for adding the Logical Switch Port is\ncalled, it tries to look up the LS, and fails.\n\nCurrently, a retry-and-wait solution is in place for the LS/LSP\nissue in create_port(), but a new feature introduced in OVS 3.1\nallows specifying the UUID of the LS, so that we can use a value\nknown to the other worker, and add the LSP to the LS by that UUID\neven if the LS isn\u0027t in our local DB cache. This allows the\novsdb-server to be the source of truth on the existance of the LS.\n\nThere are lots of other interactions that could result in this\nkind of race condition, any create/update of objects that happen\nrapidly on systems under high load for example. This method of\ncreating and referencing objects should be generalizable.\n\nIn this patch, the UUID for the LS is defined as the UUID for\nthe neutron network. The name of the LS continues to be\nneutron-$UUID to match existing usage and to keep lookups by\nname working.\n\nTODO: Investigate upgrade scenarios where a LSP is added to\nan LS that was created in a version that did not have this\npatch. It may be temporarily necessary to first do a lookup by\nname of the LS locally and use that if it exists before specifying\nthe \"known\" UUID of the LS when adding an LSP.\n\nChange-Id: Icc27c2b8825d7f96c9dac87dec8bbb55d493d942\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/cf8bce4ebf9ea7745800f07d4266f7eea4f0ebcb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/cf8bce4ebf9ea7745800f07d4266f7eea4f0ebcb"}]},"branch":"refs/heads/master"},"ebf353a69955bdc4ef14e66b30ab7696959be74a":{"kind":"REWORK","_number":2,"created":"2024-07-12 00:38:41.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/71/923971/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/71/923971/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/71/923971/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/71/923971/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/71/923971/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/71/923971/2"}}},"commit":{"parents":[{"commit":"cfab008eef3d11055ad16b8dd0d49aaa1801bc95","subject":"[OVN] Enable the WSGI module for the OVN mechanism driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/cfab008eef3d11055ad16b8dd0d49aaa1801bc95"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2024-06-11 16:37:11.000000000","tz":-300},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2024-07-12 00:33:04.000000000","tz":0},"subject":"Avoid race condition by using persist_uuid","message":"Avoid race condition by using persist_uuid\n\nWhen creating a network and adding a port, the API requests will\noften be handled by different workers. Since each worker maintains\nits own internal copy of the OVN DBs, it is possible that the\nrequest for creating the port arrives at a worker that has not yet\nreceived notification that the Logical Switch for the network has\nbeen created. When the code for adding the Logical Switch Port is\ncalled, it tries to look up the LS, and fails.\n\nCurrently, a retry-and-wait solution is in place for the LS/LSP\nissue in create_port(), but a new feature introduced in OVS 3.1\nallows specifying the UUID of the LS, so that we can use a value\nknown to the other worker, and add the LSP to the LS by that UUID\neven if the LS isn\u0027t in our local DB cache. This allows the\novsdb-server to be the source of truth on the existance of the LS.\n\nThere are lots of other interactions that could result in this\nkind of race condition, any create/update of objects that happen\nrapidly on systems under high load for example. This method of\ncreating and referencing objects should be generalizable.\n\nIn this patch, the UUID for the LS is defined as the UUID for\nthe neutron network. The name of the LS continues to be\nneutron-$UUID to match existing usage and to keep lookups by\nname working.\n\nChange-Id: Icc27c2b8825d7f96c9dac87dec8bbb55d493d942\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ebf353a69955bdc4ef14e66b30ab7696959be74a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ebf353a69955bdc4ef14e66b30ab7696959be74a"}]},"branch":"refs/heads/master"},"7f0de5fbf1b658983c20a15e52e79e0560344d5e":{"kind":"REWORK","_number":3,"created":"2024-07-12 05:48:11.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/71/923971/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/71/923971/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/71/923971/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/71/923971/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/71/923971/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/71/923971/3"}}},"commit":{"parents":[{"commit":"cfab008eef3d11055ad16b8dd0d49aaa1801bc95","subject":"[OVN] Enable the WSGI module for the OVN mechanism driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/cfab008eef3d11055ad16b8dd0d49aaa1801bc95"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2024-06-11 16:37:11.000000000","tz":-300},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2024-07-12 03:58:18.000000000","tz":0},"subject":"Avoid race condition by using persist_uuid","message":"Avoid race condition by using persist_uuid\n\nWhen creating a network and adding a port, the API requests will\noften be handled by different workers. Since each worker maintains\nits own internal copy of the OVN DBs, it is possible that the\nrequest for creating the port arrives at a worker that has not yet\nreceived notification that the Logical Switch for the network has\nbeen created. When the code for adding the Logical Switch Port is\ncalled, it tries to look up the LS, and fails.\n\nCurrently, a retry-and-wait solution is in place for the LS/LSP\nissue in create_port(), but a new feature introduced in OVS 3.1\nallows specifying the UUID of the LS, so that we can use a value\nknown to the other worker, and add the LSP to the LS by that UUID\neven if the LS isn\u0027t in our local DB cache. This allows the\novsdb-server to be the source of truth on the existance of the LS.\n\nThere are lots of other interactions that could result in this\nkind of race condition, any create/update of objects that happen\nrapidly on systems under high load for example. This method of\ncreating and referencing objects should be generalizable.\n\nIn this patch, the UUID for the LS is defined as the UUID for\nthe neutron network. The name of the LS continues to be\nneutron-$UUID to match existing usage and to keep lookups by\nname working.\n\nChange-Id: Icc27c2b8825d7f96c9dac87dec8bbb55d493d942\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7f0de5fbf1b658983c20a15e52e79e0560344d5e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7f0de5fbf1b658983c20a15e52e79e0560344d5e"}]},"branch":"refs/heads/master"},"32cc022bb2cec2d826c160a61213257836c31db5":{"kind":"REWORK","_number":4,"created":"2025-02-05 15:48:13.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/71/923971/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/71/923971/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/71/923971/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/71/923971/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/71/923971/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/71/923971/4"}}},"commit":{"parents":[{"commit":"f44977d09477a6199c7d234f84163e3a7f99e6e0","subject":"Fix typo in linuxbridge removal note","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f44977d09477a6199c7d234f84163e3a7f99e6e0"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2024-06-11 16:37:11.000000000","tz":-300},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2025-02-05 15:48:11.000000000","tz":0},"subject":"Avoid race condition by using persist_uuid","message":"Avoid race condition by using persist_uuid\n\nWhen creating a network and adding a port, the API requests will\noften be handled by different workers. Since each worker maintains\nits own internal copy of the OVN DBs, it is possible that the\nrequest for creating the port arrives at a worker that has not yet\nreceived notification that the Logical Switch for the network has\nbeen created. When the code for adding the Logical Switch Port is\ncalled, it tries to look up the LS, and fails.\n\nCurrently, a retry-and-wait solution is in place for the LS/LSP\nissue in create_port(), but a new feature introduced in OVS 3.1\nallows specifying the UUID of the LS, so that we can use a value\nknown to the other worker, and add the LSP to the LS by that UUID\neven if the LS isn\u0027t in our local DB cache. This allows the\novsdb-server to be the source of truth on the existance of the LS.\n\nThere are lots of other interactions that could result in this\nkind of race condition, any create/update of objects that happen\nrapidly on systems under high load for example. This method of\ncreating and referencing objects should be generalizable.\n\nIn this patch, the UUID for the LS is defined as the UUID for\nthe neutron network. The name of the LS continues to be\nneutron-$UUID to match existing usage and to keep lookups by\nname working.\n\nChange-Id: Icc27c2b8825d7f96c9dac87dec8bbb55d493d942\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/32cc022bb2cec2d826c160a61213257836c31db5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/32cc022bb2cec2d826c160a61213257836c31db5"}]},"branch":"refs/heads/master"},"1a03b50da90fbc8937b16f8de201c567af87533b":{"kind":"REWORK","_number":5,"created":"2025-02-05 15:54:25.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/71/923971/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/71/923971/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/71/923971/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/71/923971/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/71/923971/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/71/923971/5"}}},"commit":{"parents":[{"commit":"f44977d09477a6199c7d234f84163e3a7f99e6e0","subject":"Fix typo in linuxbridge removal note","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f44977d09477a6199c7d234f84163e3a7f99e6e0"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2024-06-11 16:37:11.000000000","tz":-300},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2025-02-05 15:54:18.000000000","tz":0},"subject":"Avoid race condition by using persist_uuid","message":"Avoid race condition by using persist_uuid\n\nWhen creating a network and adding a port, the API requests will\noften be handled by different workers. Since each worker maintains\nits own internal copy of the OVN DBs, it is possible that the\nrequest for creating the port arrives at a worker that has not yet\nreceived notification that the Logical Switch for the network has\nbeen created. When the code for adding the Logical Switch Port is\ncalled, it tries to look up the LS, and fails.\n\nCurrently, a retry-and-wait solution is in place for the LS/LSP\nissue in create_port(), but a new feature introduced in OVS 3.1\nallows specifying the UUID of the LS, so that we can use a value\nknown to the other worker, and add the LSP to the LS by that UUID\neven if the LS isn\u0027t in our local DB cache. This allows the\novsdb-server to be the source of truth on the existance of the LS.\n\nThere are lots of other interactions that could result in this\nkind of race condition, any create/update of objects that happen\nrapidly on systems under high load for example. This method of\ncreating and referencing objects should be generalizable.\n\nIn this patch, the UUID for the LS is defined as the UUID for\nthe neutron network. The name of the LS continues to be\nneutron-$UUID to match existing usage and to keep lookups by\nname working.\n\nChange-Id: Icc27c2b8825d7f96c9dac87dec8bbb55d493d942\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1a03b50da90fbc8937b16f8de201c567af87533b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1a03b50da90fbc8937b16f8de201c567af87533b"}]},"branch":"refs/heads/master"},"383f2662f538530200e4baef142fbbd6cafd4d06":{"kind":"REWORK","_number":6,"created":"2025-02-05 23:11:35.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/71/923971/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/71/923971/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/71/923971/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/71/923971/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/71/923971/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/71/923971/6"}}},"commit":{"parents":[{"commit":"0aa5d2ab56e9aa1ccb1adf9649f6db9da3fcf275","subject":"Merge \"QinQ implementation for the ML2/OVN backend\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0aa5d2ab56e9aa1ccb1adf9649f6db9da3fcf275"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2024-06-11 16:37:11.000000000","tz":-300},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2025-02-05 23:11:29.000000000","tz":0},"subject":"Avoid race condition by using persist_uuid","message":"Avoid race condition by using persist_uuid\n\nWhen creating a network and adding a port, the API requests will\noften be handled by different workers. Since each worker maintains\nits own internal copy of the OVN DBs, it is possible that the\nrequest for creating the port arrives at a worker that has not yet\nreceived notification that the Logical Switch for the network has\nbeen created. When the code for adding the Logical Switch Port is\ncalled, it tries to look up the LS, and fails.\n\nCurrently, a retry-and-wait solution is in place for the LS/LSP\nissue in create_port(), but a new feature introduced in OVS 3.1\nallows specifying the UUID of the LS, so that we can use a value\nknown to the other worker, and add the LSP to the LS by that UUID\neven if the LS isn\u0027t in our local DB cache. This allows the\novsdb-server to be the source of truth on the existance of the LS.\n\nThere are lots of other interactions that could result in this\nkind of race condition, any create/update of objects that happen\nrapidly on systems under high load for example. This method of\ncreating and referencing objects should be generalizable.\n\nIn this patch, the UUID for the LS is defined as the UUID for\nthe neutron network. The name of the LS continues to be\nneutron-$UUID to match existing usage and to keep lookups by\nname working.\n\nChange-Id: Icc27c2b8825d7f96c9dac87dec8bbb55d493d942\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/383f2662f538530200e4baef142fbbd6cafd4d06"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/383f2662f538530200e4baef142fbbd6cafd4d06"}]},"branch":"refs/heads/master"},"4387599dfd9842fe40edb0b5ebdbdca01398c300":{"kind":"REWORK","_number":7,"created":"2025-02-13 20:31:29.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/71/923971/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/71/923971/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/71/923971/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/71/923971/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/71/923971/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/71/923971/7"}}},"commit":{"parents":[{"commit":"0aa5d2ab56e9aa1ccb1adf9649f6db9da3fcf275","subject":"Merge \"QinQ implementation for the ML2/OVN backend\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0aa5d2ab56e9aa1ccb1adf9649f6db9da3fcf275"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2024-06-11 16:37:11.000000000","tz":-300},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2025-02-13 20:31:29.000000000","tz":0},"subject":"Avoid race condition by using persist_uuid","message":"Avoid race condition by using persist_uuid\n\nWhen creating a network and adding a port, the API requests will\noften be handled by different workers. Since each worker maintains\nits own internal copy of the OVN DBs, it is possible that the\nrequest for creating the port arrives at a worker that has not yet\nreceived notification that the Logical Switch for the network has\nbeen created. When the code for adding the Logical Switch Port is\ncalled, it tries to look up the LS, and fails.\n\nCurrently, a retry-and-wait solution is in place for the LS/LSP\nissue in create_port(), but a new feature introduced in OVS 3.1\nallows specifying the UUID of the LS, so that we can use a value\nknown to the other worker, and add the LSP to the LS by that UUID\neven if the LS isn\u0027t in our local DB cache. This allows the\novsdb-server to be the source of truth on the existance of the LS.\n\nThere are lots of other interactions that could result in this\nkind of race condition, any create/update of objects that happen\nrapidly on systems under high load for example. This method of\ncreating and referencing objects should be generalizable.\n\nIn this patch, the UUID for the LS is defined as the UUID for\nthe neutron network. The name of the LS continues to be\nneutron-$UUID to match existing usage and to keep lookups by\nname working.\n\nChange-Id: Icc27c2b8825d7f96c9dac87dec8bbb55d493d942\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4387599dfd9842fe40edb0b5ebdbdca01398c300"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4387599dfd9842fe40edb0b5ebdbdca01398c300"}]},"branch":"refs/heads/master"},"54fbcb0475b7bd61f2b06fd925e578eb8920f6c7":{"kind":"REWORK","_number":8,"created":"2025-02-18 20:56:13.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/71/923971/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/71/923971/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/71/923971/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/71/923971/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/71/923971/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/71/923971/8"}}},"commit":{"parents":[{"commit":"0aa5d2ab56e9aa1ccb1adf9649f6db9da3fcf275","subject":"Merge \"QinQ implementation for the ML2/OVN backend\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0aa5d2ab56e9aa1ccb1adf9649f6db9da3fcf275"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2024-06-11 16:37:11.000000000","tz":-300},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2025-02-18 20:55:51.000000000","tz":0},"subject":"Avoid race condition by using persist_uuid","message":"Avoid race condition by using persist_uuid\n\nWhen creating a network and adding a port, the API requests will\noften be handled by different workers. Since each worker maintains\nits own internal copy of the OVN DBs, it is possible that the\nrequest for creating the port arrives at a worker that has not yet\nreceived notification that the Logical Switch for the network has\nbeen created. When the code for adding the Logical Switch Port is\ncalled, it tries to look up the LS, and fails.\n\nCurrently, a retry-and-wait solution is in place for the LS/LSP\nissue in create_port(), but a new feature introduced in OVS 3.1\nallows specifying the UUID of the LS, so that we can use a value\nknown to the other worker, and add the LSP to the LS by that UUID\neven if the LS isn\u0027t in our local DB cache. This allows the\novsdb-server to be the source of truth on the existance of the LS.\n\nThere are lots of other interactions that could result in this\nkind of race condition, any create/update of objects that happen\nrapidly on systems under high load for example. This method of\ncreating and referencing objects should be generalizable.\n\nIn this patch, the UUID for the LS is defined as the UUID for\nthe neutron network. The name of the LS continues to be\nneutron-$UUID to match existing usage and to keep lookups by\nname working.\n\nChange-Id: Icc27c2b8825d7f96c9dac87dec8bbb55d493d942\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/54fbcb0475b7bd61f2b06fd925e578eb8920f6c7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/54fbcb0475b7bd61f2b06fd925e578eb8920f6c7"}]},"branch":"refs/heads/master"},"7f4a0f790e5836a4b6222e9e33818c566fd7f2e9":{"kind":"REWORK","_number":9,"created":"2025-02-18 21:16:43.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/71/923971/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/71/923971/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/71/923971/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/71/923971/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/71/923971/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/71/923971/9"}}},"commit":{"parents":[{"commit":"0aa5d2ab56e9aa1ccb1adf9649f6db9da3fcf275","subject":"Merge \"QinQ implementation for the ML2/OVN backend\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0aa5d2ab56e9aa1ccb1adf9649f6db9da3fcf275"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2024-06-11 16:37:11.000000000","tz":-300},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2025-02-18 21:16:34.000000000","tz":0},"subject":"Avoid race condition by using persist_uuid","message":"Avoid race condition by using persist_uuid\n\nWhen creating a network and adding a port, the API requests will\noften be handled by different workers. Since each worker maintains\nits own internal copy of the OVN DBs, it is possible that the\nrequest for creating the port arrives at a worker that has not yet\nreceived notification that the Logical Switch for the network has\nbeen created. When the code for adding the Logical Switch Port is\ncalled, it tries to look up the LS, and fails.\n\nCurrently, a retry-and-wait solution is in place for the LS/LSP\nissue in create_port(), but a new feature introduced in OVS 3.1\nallows specifying the UUID of the LS, so that we can use a value\nknown to the other worker, and add the LSP to the LS by that UUID\neven if the LS isn\u0027t in our local DB cache. This allows the\novsdb-server to be the source of truth on the existance of the LS.\n\nThere are lots of other interactions that could result in this\nkind of race condition, any create/update of objects that happen\nrapidly on systems under high load for example. This method of\ncreating and referencing objects should be generalizable.\n\nIn this patch, the UUID for the LS is defined as the UUID for\nthe neutron network. The name of the LS continues to be\nneutron-$UUID to match existing usage and to keep lookups by\nname working.\n\nChange-Id: Icc27c2b8825d7f96c9dac87dec8bbb55d493d942\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7f4a0f790e5836a4b6222e9e33818c566fd7f2e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7f4a0f790e5836a4b6222e9e33818c566fd7f2e9"}]},"branch":"refs/heads/master"},"6ff4eeab037c641d04358c9cc87528e6868859c5":{"kind":"REWORK","_number":10,"created":"2025-02-18 21:29:49.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/71/923971/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/71/923971/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/71/923971/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/71/923971/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/71/923971/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/71/923971/10"}}},"commit":{"parents":[{"commit":"0aa5d2ab56e9aa1ccb1adf9649f6db9da3fcf275","subject":"Merge \"QinQ implementation for the ML2/OVN backend\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0aa5d2ab56e9aa1ccb1adf9649f6db9da3fcf275"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2024-06-11 16:37:11.000000000","tz":-300},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2025-02-18 21:23:04.000000000","tz":0},"subject":"Avoid race condition by using persist_uuid","message":"Avoid race condition by using persist_uuid\n\nWhen creating a network and adding a port, the API requests will\noften be handled by different workers. Since each worker maintains\nits own internal copy of the OVN DBs, it is possible that the\nrequest for creating the port arrives at a worker that has not yet\nreceived notification that the Logical Switch for the network has\nbeen created. When the code for adding the Logical Switch Port is\ncalled, it tries to look up the LS, and fails.\n\nCurrently, a retry-and-wait solution is in place for the LS/LSP\nissue in create_port(), but a new feature introduced in OVS 3.1\nallows specifying the UUID of the LS, so that we can use a value\nknown to the other worker, and add the LSP to the LS by that UUID\neven if the LS isn\u0027t in our local DB cache. This allows the\novsdb-server to be the source of truth on the existance of the LS.\n\nThere are lots of other interactions that could result in this\nkind of race condition, any create/update of objects that happen\nrapidly on systems under high load for example. This method of\ncreating and referencing objects should be generalizable.\n\nIn this patch, the UUID for the LS is defined as the UUID for\nthe neutron network. The name of the LS continues to be\nneutron-$UUID to match existing usage and to keep lookups by\nname working.\n\nChange-Id: Icc27c2b8825d7f96c9dac87dec8bbb55d493d942\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6ff4eeab037c641d04358c9cc87528e6868859c5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6ff4eeab037c641d04358c9cc87528e6868859c5"}]},"branch":"refs/heads/master"},"fb041f0997e3ad5c01f134a3b8ec0c89300f458a":{"kind":"REWORK","_number":11,"created":"2025-03-06 17:00:44.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/71/923971/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/71/923971/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/71/923971/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/71/923971/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/71/923971/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/71/923971/11"}}},"commit":{"parents":[{"commit":"5c22bcca010e5bef285362bbca465b548c7ecd14","subject":"Merge \"ml2/sriov: bind port on segment that subnet belongs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5c22bcca010e5bef285362bbca465b548c7ecd14"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2024-06-11 16:37:11.000000000","tz":-300},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2025-03-06 17:00:11.000000000","tz":0},"subject":"Avoid race condition by using persist_uuid","message":"Avoid race condition by using persist_uuid\n\nWhen creating a network and adding a port, the API requests will\noften be handled by different workers. Since each worker maintains\nits own internal copy of the OVN DBs, it is possible that the\nrequest for creating the port arrives at a worker that has not yet\nreceived notification that the Logical Switch for the network has\nbeen created. When the code for adding the Logical Switch Port is\ncalled, it tries to look up the LS, and fails.\n\nCurrently, a retry-and-wait solution is in place for the LS/LSP\nissue in create_port(), but a new feature introduced in OVS 3.1\nallows specifying the UUID of the LS, so that we can use a value\nknown to the other worker, and add the LSP to the LS by that UUID\neven if the LS isn\u0027t in our local DB cache. This allows the\novsdb-server to be the source of truth on the existance of the LS.\n\nThere are lots of other interactions that could result in this\nkind of race condition, any create/update of objects that happen\nrapidly on systems under high load for example. This method of\ncreating and referencing objects should be generalizable.\n\nIn this patch, the UUID for the LS is defined as the UUID for\nthe neutron network. The name of the LS continues to be\nneutron-$UUID to match existing usage and to keep lookups by\nname working.\n\nChange-Id: Icc27c2b8825d7f96c9dac87dec8bbb55d493d942\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/fb041f0997e3ad5c01f134a3b8ec0c89300f458a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/fb041f0997e3ad5c01f134a3b8ec0c89300f458a"}]},"branch":"refs/heads/master"},"ba2690f788e9ce8bf680944cb751cbd839582fe3":{"kind":"REWORK","_number":12,"created":"2025-03-10 13:09:14.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/71/923971/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/71/923971/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/71/923971/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/71/923971/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/71/923971/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/71/923971/12"}}},"commit":{"parents":[{"commit":"5c22bcca010e5bef285362bbca465b548c7ecd14","subject":"Merge \"ml2/sriov: bind port on segment that subnet belongs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5c22bcca010e5bef285362bbca465b548c7ecd14"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2024-06-11 16:37:11.000000000","tz":-300},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2025-03-10 13:08:57.000000000","tz":-300},"subject":"Avoid race condition by using persist_uuid","message":"Avoid race condition by using persist_uuid\n\nWhen creating a network and adding a port, the API requests will\noften be handled by different workers. Since each worker maintains\nits own internal copy of the OVN DBs, it is possible that the\nrequest for creating the port arrives at a worker that has not yet\nreceived notification that the Logical Switch for the network has\nbeen created. When the code for adding the Logical Switch Port is\ncalled, it tries to look up the LS, and fails.\n\nCurrently, a retry-and-wait solution is in place for the LS/LSP\nissue in create_port(), but a new feature introduced in OVS 3.1\nallows specifying the UUID of the LS, so that we can use a value\nknown to the other worker, and add the LSP to the LS by that UUID\neven if the LS isn\u0027t in our local DB cache. This allows the\novsdb-server to be the source of truth on the existance of the LS.\n\nThere are lots of other interactions that could result in this\nkind of race condition, any create/update of objects that happen\nrapidly on systems under high load for example. This method of\ncreating and referencing objects should be generalizable.\n\nIn this patch, the UUID for the LS is defined as the UUID for\nthe neutron network. The name of the LS continues to be\nneutron-$UUID to match existing usage and to keep lookups by\nname working.\n\nChange-Id: Icc27c2b8825d7f96c9dac87dec8bbb55d493d942\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ba2690f788e9ce8bf680944cb751cbd839582fe3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ba2690f788e9ce8bf680944cb751cbd839582fe3"}]},"branch":"refs/heads/master"},"71ff8ee881a2c387ff318df92c61e7b8333e7014":{"kind":"REWORK","_number":13,"created":"2025-03-19 19:46:03.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/71/923971/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/71/923971/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/71/923971/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/71/923971/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/71/923971/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/71/923971/13"}}},"commit":{"parents":[{"commit":"5c22bcca010e5bef285362bbca465b548c7ecd14","subject":"Merge \"ml2/sriov: bind port on segment that subnet belongs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5c22bcca010e5bef285362bbca465b548c7ecd14"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2024-06-11 16:37:11.000000000","tz":-300},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2025-03-19 19:45:56.000000000","tz":0},"subject":"Avoid race condition by using persist_uuid","message":"Avoid race condition by using persist_uuid\n\nWhen creating a network and adding a port, the API requests will\noften be handled by different workers. Since each worker maintains\nits own internal copy of the OVN DBs, it is possible that the\nrequest for creating the port arrives at a worker that has not yet\nreceived notification that the Logical Switch for the network has\nbeen created. When the code for adding the Logical Switch Port is\ncalled, it tries to look up the LS, and fails.\n\nCurrently, a retry-and-wait solution is in place for the LS/LSP\nissue in create_port(), but a new feature introduced in OVS 3.1\nallows specifying the UUID of the LS, so that we can use a value\nknown to the other worker, and add the LSP to the LS by that UUID\neven if the LS isn\u0027t in our local DB cache. This allows the\novsdb-server to be the source of truth on the existance of the LS.\n\nThere are lots of other interactions that could result in this\nkind of race condition, any create/update of objects that happen\nrapidly on systems under high load for example. This method of\ncreating and referencing objects should be generalizable.\n\nIn this patch, the UUID for the LS is defined as the UUID for\nthe neutron network. The name of the LS continues to be\nneutron-$UUID to match existing usage and to keep lookups by\nname working.\n\nChange-Id: Icc27c2b8825d7f96c9dac87dec8bbb55d493d942\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/71ff8ee881a2c387ff318df92c61e7b8333e7014"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/71ff8ee881a2c387ff318df92c61e7b8333e7014"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
