)]}'
{"id":"openstack%2Fneutron~925039","triplet_id":"openstack%2Fneutron~master~I54d027f7cb5014d296a99029cfa0a13a7800da0a","project":"openstack/neutron","branch":"master","topic":"lp2073567","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2024-07-31 06:41:37.000000000","reason":"\u003cGERRIT_ACCOUNT_16688\u003e replied on the change","reason_account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},"13861":{"account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"last_update":"2024-07-31 05:38:54.000000000","reason":"\u003cGERRIT_ACCOUNT_13861\u003e replied on the change","reason_account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"}},"34271":{"account":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"last_update":"2024-07-31 08:17:55.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I54d027f7cb5014d296a99029cfa0a13a7800da0a","subject":"Only query for port do not get directly","status":"MERGED","created":"2024-07-26 18:28:43.000000000","updated":"2024-07-31 08:19:14.000000000","submitted":"2024-07-31 08:17:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"925039-lp2073567","meta_rev_id":"88c517b23f4230e9c4d33accc2ad91165f76c9b6","_number":925039,"virtual_id_number":925039,"owner":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2024-07-30 20:04:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-07-31 08:17:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2024-07-31 06:41:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2024-07-31 05:38:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2024-07-31 06:41:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}],"REVIEWER":[{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-07-26 18:35:16.000000000","updated_by":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2024-07-26 18:35:16.000000000","updated_by":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2024-07-26 18:46:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-07-26 20:22:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-07-27 23:22:40.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-07-28 20:16:43.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"}],"messages":[{"id":"d0c97eaa111e42fe36628550d50630999c79d235","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2024-07-26 18:28:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"36c8e8e0d741d43fb300aa3781553c1057a9f15c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-26 18:46:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dc1a811afafa445fa4e2883b7e927393\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a97cb62a13bd46978b3f56748b681329 : SUCCESS in 17m 16s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/454c693583234112811947387cf438eb : SUCCESS in 13m 36s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e84e549080640199e7654368aad1ea971adfaca7","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-07-26 19:24:13.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"851a8f717af912bbffd14cc7f9572f3c12c9f8ed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-26 20:22:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0e3ec6fe832b438ab56e2e3832134016\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/2e0d1914f36e4372a60962241ceeada5 : SUCCESS in 1h 51m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a280f33b7b294d159a8c9a87115dffa8 : SUCCESS in 42m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f18e9268f5804c8085dfebbfa25a5678 : SUCCESS in 8m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/efb3fb0fb46640bab030f5ee94b31aab : SUCCESS in 17m 48s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1d811710e29b476eaf7bcad03d80b7f4 : SUCCESS in 21m 16s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/815587881d2846faaf7e9472d5cdb292 : SUCCESS in 21m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/04201bed63bc476393276523211f0ecb : SUCCESS in 9m 17s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/cb0028df13fc47259fb8f0a4b3592542 : SUCCESS in 23m 02s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/8ba34796a4b84bcea25c681d7c834c26 : SUCCESS in 36m 45s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/ebb5ad249ce2497d9b9585f48d9e16cb : SUCCESS in 48m 20s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/c63980093f2a4c4bb08a91c322281c48 : SUCCESS in 1h 03m 40s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/e23d50b5e31f4a75ace8c2c818020467 : SUCCESS in 47m 04s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/7d217a33b9d34f27bfbc863f84331a66 : SUCCESS in 1h 02m 05s","accounts_in_message":[],"_revision_number":1},{"id":"ec5ab77bf147bfba066be3e42012e955c8d6b764","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2024-07-27 23:22:40.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/July-2024/Arista_Tempest_Testrefs/changes/39/925039/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"5e766dfc53fa33051dc13803d66d1f66e5f4ee84","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-07-28 20:16:43.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"165d5a968021d5215f5c69e17248263230b1d2c1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2024-07-30 19:30:29.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"69681ae9e883389df524334ff3228b2cd3525d36","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-30 19:57:33.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/35703e51be184b74add9cd9f3886845a\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0b9c99d23d974f6fbf9b5a2d2d0b2399 : SUCCESS in 25m 17s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/644ab0c9b3474a788f03fed3a61a54b1 : SUCCESS in 23m 31s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b970df96c9a772766c5d503099dc04e53767c5bc","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2024-07-30 20:04:24.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/July-2024/Arista_Tempest_Testrefs/changes/39/925039/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"03b1b80713d4b0764f1dd4e4e039b242fc3d14ea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-30 21:35:58.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/adac271d264d47ac9a819d6caa7c3407\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/48bf811c56e948d8a3a6cbd6b37cd90c : SUCCESS in 2h 03m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9c165092736e4b4a98fb549be574d7e5 : SUCCESS in 29m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9384959149944981a0ce0530b45a97eb : SUCCESS in 8m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/caf482f82e3b48d4abb82e7f4916bf05 : SUCCESS in 24m 19s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/776778649a8f4b918029a95b43a91338 : SUCCESS in 20m 58s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/af03c696fcc6408ca6626ea1cec548f1 : SUCCESS in 20m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e284fa7b3df245909f66e0678076fb98 : SUCCESS in 11m 17s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/fee9d75d45aa415eaec2ecdf6b559849 : SUCCESS in 17m 53s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/8f52e0e6dd7c4c50a5d972ddcd425b2e : SUCCESS in 43m 05s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/4fa072b3d2904e069a78d9cb3ae461bc : SUCCESS in 49m 19s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/aaf6a36571044a5c858c3482748cb87c : SUCCESS in 45m 27s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/0819bf50e6084e4db5ba014c99c63ec2 : SUCCESS in 1h 02m 34s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/ec1ee9b851764fc1b7cb6f17333b2fc9 : SUCCESS in 50m 46s","accounts_in_message":[],"_revision_number":2},{"id":"2aac61f86fa5c686002caedee3888300885e8840","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2024-07-31 05:38:54.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"6c39e4e7f03c97f8da0f2372b65730e9f9d91382","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-07-31 06:41:37.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"797203c49a8e3eff9a523a90f61b00d13344d4d2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-31 06:42:08.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"b66f940356c33cf8697f768a59ca71ebdf569eb7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-31 08:17:55.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bb0abfd6f7164e4fb882e886f3a10b21\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/e4a7277acce944a3ade36b721992c350 : SUCCESS in 1h 35m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/759b45646ed04c87a199e6ef13843d61 : SUCCESS in 8m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/15a658691a6a497489f9f5c3f5eb86fe : SUCCESS in 30m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e2ace5601eca46fca5235a534ccb2c10 : SUCCESS in 22m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ac63cee4ebd44f6a9335f32ed518a59 : SUCCESS in 9m 59s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/118e4e44bd2040fe9c5fa4be922ab82f : SUCCESS in 1h 06m 03s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/6f5961599c0d4d81ad05e5c1e8a733b5 : SUCCESS in 1h 10m 34s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/d5365c93d93e49bf930fc0364eed73d9 : SUCCESS in 1h 09m 07s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/844ee72e76d1487eb71f9824108dd0a5 : SUCCESS in 1h 24m 44s","accounts_in_message":[],"_revision_number":2},{"id":"c29a4cc1b80260a95cc640eadba925b0ae893834","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-31 08:17:55.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"88c517b23f4230e9c4d33accc2ad91165f76c9b6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-31 08:19:14.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/03b1fb2fecce49678bb742d4b5e8a002\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c3df71c9ad94d9788018269a0568795 : SUCCESS in 59s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"c1b88fc5f52283380261f4fdc1562ff56ea06a29","revisions":{"62b06816c66acb1137c43a2590e38f6468e3f3d1":{"kind":"REWORK","_number":1,"created":"2024-07-26 18:28:43.000000000","uploader":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"ref":"refs/changes/39/925039/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/39/925039/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/925039/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/925039/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/925039/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/39/925039/1"}}},"commit":{"parents":[{"commit":"725f3cbd8e9dfab18cccdcb2073f49473e634346","subject":"Merge \"Actualy set global \"removal limit\" options\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/725f3cbd8e9dfab18cccdcb2073f49473e634346"}]}],"author":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2024-07-26 14:50:40.000000000","tz":-240},"committer":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2024-07-26 18:27:09.000000000","tz":-240},"subject":"Only query for port do not get directly","message":"Only query for port do not get directly\n\nIt was observed in the tempest tests that the port could be already\ndeleted by some other concurrent event when the `run` is called.\nThis caused a flood of exception logs. Thus, with this patch we only\nquery for the port and only update_router_port when the port was\nfound.\n\nCloses-Bug: #2073567\nChange-Id: I54d027f7cb5014d296a99029cfa0a13a7800da0a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/62b06816c66acb1137c43a2590e38f6468e3f3d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/62b06816c66acb1137c43a2590e38f6468e3f3d1"}]},"branch":"refs/heads/master"},"c1b88fc5f52283380261f4fdc1562ff56ea06a29":{"kind":"REWORK","_number":2,"created":"2024-07-30 19:30:29.000000000","uploader":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"ref":"refs/changes/39/925039/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/39/925039/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/925039/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/925039/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/925039/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/39/925039/2"}}},"commit":{"parents":[{"commit":"725f3cbd8e9dfab18cccdcb2073f49473e634346","subject":"Merge \"Actualy set global \"removal limit\" options\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/725f3cbd8e9dfab18cccdcb2073f49473e634346"}]}],"author":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2024-07-26 14:50:40.000000000","tz":-240},"committer":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2024-07-30 19:30:19.000000000","tz":-240},"subject":"Only query for port do not get directly","message":"Only query for port do not get directly\n\nIt was observed in the tempest tests that the port could be already\ndeleted by some other concurrent event when the `run` is called.\nThis caused a flood of exception logs. Thus, with this patch we only\nquery for the port and only update_router_port when the port was\nfound.\n\nCloses-Bug: #2073567\nChange-Id: I54d027f7cb5014d296a99029cfa0a13a7800da0a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c1b88fc5f52283380261f4fdc1562ff56ea06a29"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c1b88fc5f52283380261f4fdc1562ff56ea06a29"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
