)]}'
{"id":"openstack%2Fneutron~925077","triplet_id":"openstack%2Fneutron~stable%2F2024.1~Ia35c51fb81cfc0a55c5a2436fc5c55f2b4c9bd01","project":"openstack/neutron","branch":"stable/2024.1","topic":"bug/2073782","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2024-08-02 11:28:04.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia35c51fb81cfc0a55c5a2436fc5c55f2b4c9bd01","subject":"Initialize the policy enforcer for the \"tagging\" service plugin","status":"MERGED","created":"2024-07-29 06:37:25.000000000","updated":"2024-08-02 11:29:10.000000000","submitted":"2024-08-02 11:28:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"925077-bug/2073782","meta_rev_id":"cf473b8db3ab873597be0fed1342481fd98d6f78","_number":925077,"virtual_id_number":925077,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-08-02 11:28:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":2,"date":"2024-08-01 12:22:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2024-08-02 09:17:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":1,"date":"2024-08-02 09:17:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-07-29 07:23:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-07-29 08:43:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-08-01 12:22:30.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2024-08-02 06:55:01.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2024-08-02 09:17:10.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"3cf2141d3a949fb32cfcb8eeee09fde63f3debc7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-07-29 06:37:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8f52aff616029a8a1530c8b1d41cc1aeadcaaac9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-29 07:23:15.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/14753e16c5924c89950fb2ff7a27812d\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c18696baf37649409a6bf199f39c0e1a : SUCCESS in 43m 58s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/5c119348cd174798af6191e389ffeeb5 : SUCCESS in 43m 21s\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/5e62e92431084b86a93827e5221fa1af : SUCCESS in 23m 46s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a8417ff59c3c7e7411931509b7a20ff7c50a862f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-29 08:43:36.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d0387da98f1e4a4094ae1e9386942aa7\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/1c1004035ff44c3f9b375815e3a84254 : FAILURE in 14m 39s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/38730798be4b4d6087345ba34d53b3f0 : FAILURE in 18m 18s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/4b36d78ad210447592e4f72c9585f338 : FAILURE in 18m 12s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/a39e7af68ca44f1bac018785db94ca20 : SUCCESS in 2h 04m 08s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d51709eeb0c44168b9bc463d69a63355 : SUCCESS in 51m 35s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/661e7bdc4b664d5b81f26b9e0d261e6a : SUCCESS in 41m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e96217aa01974f6f80e63fee9d173baa : SUCCESS in 10m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c7e3e2ee48c8470080a522e7c055aa8e : SUCCESS in 22m 22s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/47a6a8498c5843fcb4676b8f1f5f94d4 : SUCCESS in 25m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b28bbd1a576b455f9785e69452cd418c : SUCCESS in 9m 51s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/8fb459e30d8346bbbcc4d9112559ab10 : SUCCESS in 45m 20s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/9e14ad305add47a397fc1e8cd15c6ba5 : SUCCESS in 54m 33s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/3fa8c66338424986b647ca1394ecac3f : SUCCESS in 47m 50s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/840d2fa5441e48d095a641dc7ba748dd : SUCCESS in 1h 20m 25s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/6d2b77b398c3462eb497a18d33d89924 : SUCCESS in 54m 19s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/db452537a17d4e3ea9c573a414228e20 : SUCCESS in 58m 08s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/697f244a4bd547a59092371d0898be82 : FAILURE in 1h 08m 01s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/da047ee927a84f2e85a8a3eb13734819 : FAILURE in 1h 13m 23s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/1d06364a266b40ce8cfa2ce2eea5f5c4 : FAILURE in 55m 42s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/6a336a7ad67c403d9303de046b61ef59 : SUCCESS in 1h 12m 15s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/1e833116665c4e51ac27b676ffd9c818 : SUCCESS in 1h 28m 38s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/968a1509fd5d41aca788eb3879dd8f4a : SUCCESS in 1h 12m 39s","accounts_in_message":[],"_revision_number":1},{"id":"69fd377023689be256cb6f5fd6920854f60e859c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-07-30 05:36:12.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"e62d9d1772fa547a68d8b130b926798c8f62b4a4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-30 06:28:17.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5c2dfc0ca81b4383bb18d9e4e2b35d75\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/99d8ea27b16b4ddea690876c0cefe065 : SUCCESS in 31m 41s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c996ec824a4a4ef18233c679292cdcf3 : SUCCESS in 49m 50s\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/e150c120ea9243e494a5f338e5a057a5 : SUCCESS in 43m 38s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"84152b7c2bff437cf475a5c15dc60aecd0a6af74","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-30 07:35:27.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e82ef5ec299d4b09b8b27b18e26cb298\n\n- neutron-tempest-plugin-openvswitch-2024-1 https://zuul.opendev.org/t/openstack/build/133d724006254273bbd0b625cb0fc60e : SUCCESS in 1h 43m 17s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-1 https://zuul.opendev.org/t/openstack/build/2e585b31d8d54ba192e1981c9f1f90c8 : SUCCESS in 1h 46m 46s\n- neutron-tempest-plugin-ovn-2024-1 https://zuul.opendev.org/t/openstack/build/ec13cc02c4414e79a5f3ac0798d8c078 : SUCCESS in 1h 56m 23s\n- neutron-tempest-plugin-designate-scenario-2024-1 https://zuul.opendev.org/t/openstack/build/ac223420344447ae98453ff776a4ca00 : SUCCESS in 40m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/638b1c31933447c6971f0e877efb48cd : SUCCESS in 33m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1dd7c212b8d54685b072f76a11b7b7d3 : SUCCESS in 9m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/abe09a0dfa5241e6aa844c268db2e7ec : SUCCESS in 25m 00s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d67e0ac43b08461e8bd5a57b0fb2596f : SUCCESS in 22m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2db329136e914661a8690a99549990e3 : SUCCESS in 10m 29s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/6f290c97207449ffb03d2a3250775842 : SUCCESS in 55m 52s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/16267ac12a584c108ed2d758c0ab9da7 : SUCCESS in 1h 00m 14s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/935b1aa08ae54a6289341c618fb37a34 : SUCCESS in 54m 56s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/da4746d6fd3b449e80d6f01b7f9aa5f5 : SUCCESS in 1h 18m 25s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/90911c6cdae24f32b50d1c67bcb55fc3 : SUCCESS in 59m 24s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/8656b172d5b54df2b526324f9107de60 : SUCCESS in 41m 08s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/37239cc39b114213b518c125b86e8b5f : SUCCESS in 1h 21m 45s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/e27bb21f56824c0f91a06c6e8e3bb2f9 : SUCCESS in 1h 18m 34s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/94b8dcb4fae54d55a9df672db024afbc : SUCCESS in 51m 49s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/b579cb852ad042f1953b45a8bb504632 : SUCCESS in 1h 28m 28s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/c0f556e9767742d08a8d3aa390760f1f : SUCCESS in 1h 26m 55s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/7bec89deb46046838ed2c2a37226ce71 : SUCCESS in 1h 04m 10s","accounts_in_message":[],"_revision_number":2},{"id":"5770485160fe7b87f45d675f6487ca3885dbe964","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2024-08-01 12:22:30.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b2cfd2646e66f970656155f1e74dea67b916c2ff","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2024-08-02 09:17:10.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"e5354aabfc4b0e3a9da3876b4c32f0f4132f2baf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-02 09:17:51.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"6c94ea19cc2b0193fffe82d1a09c3cf316e8a87b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-02 11:28:04.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c9cf097467474dad9e564fd89294aeac\n\n- neutron-tempest-plugin-ovn-2024-1 https://zuul.opendev.org/t/openstack/build/6194a2475440453eb97056ea63d4cd07 : SUCCESS in 2h 08m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2631b89de2054a9d99e0558c6165e6ac : SUCCESS in 9m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/48e8ed5e8f1c496eadb1c207b00f41fa : SUCCESS in 31m 10s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/30af4e7ca1c54427ab41ce3bc2dc7cd6 : SUCCESS in 21m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad3615b9dba54e379808aab45d4ca5ab : SUCCESS in 11m 46s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/a33be71807af49e5be9d43135206c472 : SUCCESS in 58m 45s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/56e202862dd8478ba9c06fa3fabe010a : SUCCESS in 2h 04m 49s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/d2b3da3b0aa042c9bab5302daeb0fc65 : SUCCESS in 1h 52m 46s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/3382af26e5ae4de38678a23688e45bd4 : SUCCESS in 59m 27s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/c322aa1fd0f648fcb8acf819b66ab468 : SUCCESS in 53m 12s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/ae812866f50e400aacb9e821ccd00199 : SUCCESS in 1h 09m 57s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/fb62734970b0492fbd138c7593fb3164 : SUCCESS in 1h 20m 31s","accounts_in_message":[],"_revision_number":2},{"id":"54edb903a8457f14fd7e2be7dcda1c14c4e25625","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-02 11:28:04.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"cf473b8db3ab873597be0fed1342481fd98d6f78","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-02 11:29:10.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/17d689174fb5425e80bdfe15c3d1d823\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2de803d4bc744b3b9ab8cf9e450b4b52 : SUCCESS in 48s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"2526d0fdc77b85e2473eacf4091fd1ce628ff1fc","revisions":{"da8cbaf2daaa2fa3ccc4ff0c6600a7cc67b40813":{"kind":"REWORK","_number":1,"created":"2024-07-29 06:37:25.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/77/925077/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/77/925077/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/77/925077/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/77/925077/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/77/925077/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/77/925077/1"}}},"commit":{"parents":[{"commit":"1f761c27f2b47212eadf0e6e192f7e8201ee2aa6","subject":"Don\u0027t print traceback if standard attr is missing on update","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1f761c27f2b47212eadf0e6e192f7e8201ee2aa6"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-07-20 00:46:04.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-07-24 00:34:31.000000000","tz":0},"subject":"Initialize the policy enforcer for the \"tagging\" service plugin","message":"Initialize the policy enforcer for the \"tagging\" service plugin\n\nThe \"tagging\" service plugin API extension does use the policy enforcer\nsince [1]. If a tag API call is done just after the Neutron server has\nbeen initialized and the policy enforcer, that is a global variable per\nAPI worker, has not been initialized, the API call will fail.\n\nThis patch initializes the policy enforcer as is done in the\n``PolicyHook``, that is called by many other API resources that inherit\nfrom the ``APIExtensionDescriptor`` class.\n\n[1]https://review.opendev.org/q/I9f3e032739824f268db74c5a1b4f04d353742dbd\n\nCloses-Bug: #2073782\nChange-Id: Ia35c51fb81cfc0a55c5a2436fc5c55f2b4c9bd01\n(cherry picked from commit 776178e90763d004ccb595b131cdd4dd617cd34f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/da8cbaf2daaa2fa3ccc4ff0c6600a7cc67b40813"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/da8cbaf2daaa2fa3ccc4ff0c6600a7cc67b40813"}]},"branch":"refs/heads/stable/2024.1"},"2526d0fdc77b85e2473eacf4091fd1ce628ff1fc":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-07-30 05:36:12.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/77/925077/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/77/925077/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/77/925077/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/77/925077/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/77/925077/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/77/925077/2"}}},"commit":{"parents":[{"commit":"28466f849c48b1a506afba0d3b6e870dbbef485e","subject":"[2024.1 only] Switch to 2024.1 neutron-tempest-plugin jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/28466f849c48b1a506afba0d3b6e870dbbef485e"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-07-20 00:46:04.000000000","tz":0},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2024-07-30 05:36:12.000000000","tz":0},"subject":"Initialize the policy enforcer for the \"tagging\" service plugin","message":"Initialize the policy enforcer for the \"tagging\" service plugin\n\nThe \"tagging\" service plugin API extension does use the policy enforcer\nsince [1]. If a tag API call is done just after the Neutron server has\nbeen initialized and the policy enforcer, that is a global variable per\nAPI worker, has not been initialized, the API call will fail.\n\nThis patch initializes the policy enforcer as is done in the\n``PolicyHook``, that is called by many other API resources that inherit\nfrom the ``APIExtensionDescriptor`` class.\n\n[1]https://review.opendev.org/q/I9f3e032739824f268db74c5a1b4f04d353742dbd\n\nCloses-Bug: #2073782\nChange-Id: Ia35c51fb81cfc0a55c5a2436fc5c55f2b4c9bd01\n(cherry picked from commit 776178e90763d004ccb595b131cdd4dd617cd34f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2526d0fdc77b85e2473eacf4091fd1ce628ff1fc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2526d0fdc77b85e2473eacf4091fd1ce628ff1fc"}]},"branch":"refs/heads/stable/2024.1","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
