)]}'
{"id":"openstack%2Fneutron~925187","triplet_id":"openstack%2Fneutron~master~If658f45e12f7f86de941ec997d5f8177994543a4","project":"openstack/neutron","branch":"master","topic":"dnm","attention_set":{},"removed_from_attention_set":{"13861":{"account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"last_update":"2024-12-03 05:13:15.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"If658f45e12f7f86de941ec997d5f8177994543a4","subject":"[DNM] check ovn/ovs source job ovs_branch\u003d3.4","status":"ABANDONED","created":"2024-07-30 11:06:08.000000000","updated":"2024-12-03 05:13:15.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"40cf78907d46892615f65f05dc93d4e94a5abd97","_number":925187,"virtual_id_number":925187,"owner":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2024-07-30 12:43:16.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-07-30 11:19:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1d46a47d7f6c688471d8655971e75fa96863e09e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2024-07-30 11:06:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9b973dea139bb6034f0c159241b25213de0742d7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-30 11:19:12.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e84b442f982b407baa121a64a0499a57\n\n- neutron-ovn-tempest-ipv6-only-ovs-master https://zuul.opendev.org/t/openstack/build/0a8238777d944e14baf2c94574a589bf : FAILURE in 11m 05s","accounts_in_message":[],"_revision_number":1},{"id":"46577e4fde215754c875b127779de11f09b10cb3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2024-07-30 11:38:57.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"f1d6ea3ff6e199c2c6f5fcae2d6f44e4caf123f4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-30 12:43:16.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/00db858b5031463582204e30ba0dd996\n\n- neutron-ovn-tempest-ipv6-only-ovs-master https://zuul.opendev.org/t/openstack/build/1aea8a0871af41109d224007462d30d1 : SUCCESS in 1h 03m 08s","accounts_in_message":[],"_revision_number":2},{"id":"40cf78907d46892615f65f05dc93d4e94a5abd97","tag":"autogenerated:gerrit:abandon","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2024-12-03 05:13:15.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"1f5c1cf090a848202a571b204a2648fbfff4c369","revisions":{"d50e1f90620f83b7409ace53e457e1c9acb258ef":{"kind":"REWORK","_number":1,"created":"2024-07-30 11:06:08.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/87/925187/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/87/925187/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/87/925187/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/87/925187/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/87/925187/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/87/925187/1"}}},"commit":{"parents":[{"commit":"72ca14cefe4c17c739a9ee594d5a6f8b1380fa89","subject":"[ovs-master jobs] Move OVS_BRANCH to branch-3.4","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/72ca14cefe4c17c739a9ee594d5a6f8b1380fa89"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2024-07-30 11:04:25.000000000","tz":330},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2024-07-30 11:04:28.000000000","tz":330},"subject":"[DNM] check ovn/ovs source job ovs_branch\u003d3.4","message":"[DNM] check ovn/ovs source job ovs_branch\u003d3.4\n\nsomehow the job didn\u0027t got triggered in experimental\nqueue.\n\nChange-Id: If658f45e12f7f86de941ec997d5f8177994543a4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d50e1f90620f83b7409ace53e457e1c9acb258ef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d50e1f90620f83b7409ace53e457e1c9acb258ef"}]},"branch":"refs/heads/master"},"1f5c1cf090a848202a571b204a2648fbfff4c369":{"kind":"REWORK","_number":2,"created":"2024-07-30 11:38:57.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/87/925187/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/87/925187/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/87/925187/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/87/925187/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/87/925187/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/87/925187/2"}}},"commit":{"parents":[{"commit":"72ca14cefe4c17c739a9ee594d5a6f8b1380fa89","subject":"[ovs-master jobs] Move OVS_BRANCH to branch-3.4","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/72ca14cefe4c17c739a9ee594d5a6f8b1380fa89"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2024-07-30 11:04:25.000000000","tz":330},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2024-07-30 11:38:36.000000000","tz":330},"subject":"[DNM] check ovn/ovs source job ovs_branch\u003d3.4","message":"[DNM] check ovn/ovs source job ovs_branch\u003d3.4\n\nsomehow the job didn\u0027t got triggered in experimental\nqueue.\n\nChange-Id: If658f45e12f7f86de941ec997d5f8177994543a4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1f5c1cf090a848202a571b204a2648fbfff4c369"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1f5c1cf090a848202a571b204a2648fbfff4c369"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
