)]}'
{"id":"openstack%2Fneutron~925197","triplet_id":"openstack%2Fneutron~master~I2de3b5d7ba5783dd82acacda89ab4b64c2d29149","project":"openstack/neutron","branch":"master","topic":"bug/2074209","attention_set":{},"removed_from_attention_set":{"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2024-07-30 19:56:06.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I2de3b5d7ba5783dd82acacda89ab4b64c2d29149","subject":"Use has_lock_periodic decorator for the tasks which require ovn db lock","status":"MERGED","created":"2024-07-30 12:53:01.000000000","updated":"2024-07-30 19:57:17.000000000","submitted":"2024-07-30 19:56:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"925197-bug/2074209","meta_rev_id":"d78947d41a6ff501d103e87ef9394dc8a01e060d","_number":925197,"virtual_id_number":925197,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2024-07-30 13:38:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-07-30 19:56:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2024-07-30 18:15:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":2,"date":"2024-07-30 17:40:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":1,"date":"2024-07-30 18:15:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-07-30 13:23:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-07-30 13:38:34.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-07-30 14:29:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-07-30 17:40:56.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2024-07-30 18:15:01.000000000","updated_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"REVIEWER"}],"messages":[{"id":"30fe79aafcf6ba3130e703f415c2d7471cf2546f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2024-07-30 12:53:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f5a4f38fa23cb24c7e79a23352cd95e18242f67b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-30 13:23:12.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6c6bd36cac25403d99fd2370d0b906c3\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8dc9db0d8f1c436f91fc4269b72c6835 : SUCCESS in 25m 16s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/889cd14fbecb4f20856386adb7f8c5a2 : SUCCESS in 27m 50s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4501d44980d0c14f2c8d200d3a90a7d2f6782bd6","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2024-07-30 13:38:34.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/July-2024/Arista_Tempest_Testrefs/changes/97/925197/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"9ca3430a07547d9f11cb207d477e9186e58944c7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-30 14:29:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ca01564827c7439689fabd340737dd58\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/ff03c28c5af44387ba0ce46071f43a68 : SUCCESS in 1h 35m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1ac8e6fa46fb4f428792f92fa64b23df : SUCCESS in 30m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c00c7822df244181a4caebc26bc1538c : SUCCESS in 9m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1b93c59346f247fd8319f582d279e76f : SUCCESS in 21m 13s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9fe1823a36fb468988f295d8f02c229e : SUCCESS in 16m 32s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a86f410442eb43158a6a500a5b5d4ef7 : SUCCESS in 23m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/938c2ea80cea49b28c945b4cf76118e5 : SUCCESS in 11m 49s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/243d38d5a2df4df8a09fb6c6241116e6 : SUCCESS in 20m 19s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/bc11a1d68b49461ab3100a8fbd2e4124 : SUCCESS in 54m 26s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/9e119c2aaa074062a9b06e4921e2c8d3 : SUCCESS in 1h 20m 39s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/f7fc1ad10e4f42478a3079b98c540902 : SUCCESS in 1h 01m 07s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/1197cc7ae060451b8e63b2ee458564a7 : SUCCESS in 59m 59s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/69c68061d8654bb1b44aed8f6d5bf26d : SUCCESS in 1h 16m 11s","accounts_in_message":[],"_revision_number":1},{"id":"be5be9c61a1b0471b40eb10b73dd0b00c17129eb","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-07-30 17:40:56.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"fdefc3bc944b1ad90e8c81b9f9fbfafe3f83df1b","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2024-07-30 18:15:01.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"817c42aebe3a6811fd0efe93fd2fdc7161aecdd4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-30 18:16:38.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2570bd8ea9cddfe0cc847699f333e59abd9ba127","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-30 19:56:06.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7933c6cf0ff641219c026e237f55ffe4\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/de8d63882aae4696b03fe64ab9e40653 : SUCCESS in 1h 28m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6bdf5a151983404b9ea466e6ef8ed5fc : SUCCESS in 7m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/87dfb697e93f4e8a88f72f16d89ee28d : SUCCESS in 16m 56s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/958c71cb33a94c62be4f531c8532dc4d : SUCCESS in 17m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a5480d5b3ac494c83567ef833b8f7e7 : SUCCESS in 9m 00s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/bc215af561a94665aa3de1f508dbae3c : SUCCESS in 43m 56s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/cbde500cde9b429c8edbd76d608c9a58 : SUCCESS in 1h 09m 32s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/801dade8a3224c0ebb12a5c00fb621a8 : SUCCESS in 1h 03m 51s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/f0e1582bb2d843ffbc6f89f5807d8592 : SUCCESS in 1h 12m 33s","accounts_in_message":[],"_revision_number":1},{"id":"b8938904e85eeba4eeda86cf68aa01731121d980","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-30 19:56:06.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"d78947d41a6ff501d103e87ef9394dc8a01e060d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-30 19:57:17.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/70737d2eaebf4330a73ee5134a666d6d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6502e385887a437db70c7a08ca28ba5f : SUCCESS in 51s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2a6bc5db237d28ddfdda16aea7c1b3416f3e14a4","revisions":{"2a6bc5db237d28ddfdda16aea7c1b3416f3e14a4":{"kind":"REWORK","_number":1,"created":"2024-07-30 12:53:01.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/97/925197/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/97/925197/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/925197/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/925197/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/925197/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/97/925197/1"}}},"commit":{"parents":[{"commit":"2b6abf0fc08e0b1d5beae4f1536a7f49247146e7","subject":"Merge \"Remove the tenant_id/project_id parameter from QoS rule commands\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2b6abf0fc08e0b1d5beae4f1536a7f49247146e7"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2024-07-30 12:50:36.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2024-07-30 12:52:49.000000000","tz":120},"subject":"Use has_lock_periodic decorator for the tasks which require ovn db lock","message":"Use has_lock_periodic decorator for the tasks which require ovn db lock\n\nThis is follow up patch for the [1] which introduced this new decorator.\n\n[1] https://review.opendev.org/c/openstack/neutron/+/896544\n\nChange-Id: I2de3b5d7ba5783dd82acacda89ab4b64c2d29149\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2a6bc5db237d28ddfdda16aea7c1b3416f3e14a4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2a6bc5db237d28ddfdda16aea7c1b3416f3e14a4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
