)]}'
{"id":"openstack%2Fneutron~925376","triplet_id":"openstack%2Fneutron~master~If2aa37b2a510a85172da833ca20564810817d246","project":"openstack/neutron","branch":"master","topic":"bug/2075147","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2024-08-30 09:48:32.000000000","reason":"\u003cGERRIT_ACCOUNT_16688\u003e replied on the change","reason_account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},"13861":{"account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"last_update":"2024-08-30 14:39:09.000000000","reason":"Change was submitted"},"4694":{"account":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"last_update":"2024-08-30 14:39:09.000000000","reason":"Change was submitted"},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2024-08-28 19:01:03.000000000","reason":"\u003cGERRIT_ACCOUNT_1131\u003e replied on the change","reason_account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}}},"hashtags":[],"change_id":"If2aa37b2a510a85172da833ca20564810817d246","subject":"Monkey patch the system libraries before calling them","status":"MERGED","created":"2024-07-31 10:57:00.000000000","updated":"2024-08-30 14:41:14.000000000","submitted":"2024-08-30 14:39:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":23,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"925376-bug/2075147","meta_rev_id":"02afb9c8758b84804c60be2999e05db5bf06796f","_number":925376,"virtual_id_number":925376,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2024-08-28 19:35:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-08-30 14:39:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":2,"date":"2024-08-28 19:00:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":2,"date":"2024-08-28 19:00:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":2,"date":"2024-08-28 19:01:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":1,"date":"2024-08-28 19:01:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":1,"date":"2024-08-28 19:01:03.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-07-31 11:43:18.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-07-31 12:19:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-07-31 13:58:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-07-31 14:21:59.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2024-07-31 14:55:21.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"CC"},{"updated":"2024-07-31 14:55:21.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"CC"},{"updated":"2024-08-06 11:53:50.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"CC"},{"updated":"2024-08-06 14:50:15.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2024-08-06 18:09:58.000000000","updated_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"REVIEWER"},{"updated":"2024-08-08 13:01:59.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2024-08-26 12:31:32.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"}],"messages":[{"id":"369081c1652021d5e72683a94e38193dfa2b474d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-07-31 10:57:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3879051aeb9bfd02c49158cc0e4f9c2aa7c00da2","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-07-31 11:02:53.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c32fec65513c24be17925893124e35e98aa6264b","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-07-31 11:31:14.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"e1371b3662ad46b572d372808cef23a6a093b10d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-07-31 11:40:16.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"55b57cc211a308f58c51881562ecc23115f7609d","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2024-07-31 11:43:18.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/July-2024/Arista_Tempest_Testrefs/changes/76/925376/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"ed96609aad3cdcc6c9cbcf825c8b65f45eca8c25","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2024-07-31 12:17:15.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/July-2024/Arista_Tempest_Testrefs/changes/76/925376/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"3051223641d957c879c1e05d702068b92a16a4af","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-31 12:19:43.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0c3e8b3e903f449cb530a3a18f38e43f\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6702851b4b654d34bd8c149443b6a6f7 : SUCCESS in 32m 44s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/456a344fab2043c5b739e4a2c3ea0ba5 : SUCCESS in 37m 52s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f829dc41b50de57d963ba582df3b7f1ff274869b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-31 13:58:19.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9883e900b1c341818c256b886dafeca6\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/8056f721f6914b8880bac0c5a2e7f48f : FAILURE in 1h 35m 31s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/ba2794264d4f49aa855cfdca7edabca5 : FAILURE in 1h 49m 39s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/56e495bf84814b64b9153883db916408 : SUCCESS in 2h 15m 49s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/488984981ef74231a04ccda342e7bcee : SUCCESS in 1h 40m 14s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/c4d359c018e94dcbbcea4cac97c7fae2 : SUCCESS in 39m 02s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fc50574af52d4709b03d57d38264ef14 : SUCCESS in 32m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7ab76c2d455f41d9978630acb49a9c4b : SUCCESS in 8m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/aa75030a0df44503b2905f85d684b04d : SUCCESS in 19m 04s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/02cbdf87d5654831adac16250ad19a1c : SUCCESS in 20m 32s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8d734aa166bc470badf061f5bfba21c4 : SUCCESS in 24m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/987054fa789942d08783a1b8ffb656f0 : SUCCESS in 11m 21s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/8c304b936e3f416bbf546b714150a040 : SUCCESS in 20m 59s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/bdf8d488099d4f919bb1069e9be64fae : SUCCESS in 58m 02s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/8c31c7a2648f4f5c9cdca1e94e4d35e9 : SUCCESS in 1h 52m 17s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/ccf932f9e25944ccb46228131fd5358a : SUCCESS in 59m 36s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/58daef0f9ddd49589e60c8815383c188 : SUCCESS in 1h 08m 48s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/3a2efa8406ce49dda981016a475e8bc6 : SUCCESS in 1h 36m 32s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/3a8c7f50b8e547118a7c25bc475b8b5e : SUCCESS in 1h 24m 46s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/e631cf55feae4a97a946d0edaf72b835 : SUCCESS in 53m 39s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/7bcc85e90b0e42559f775f85c2dbe8aa : SUCCESS in 52m 49s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/234a344bcc134517a7dc5ce9e72877e0 : SUCCESS in 1h 06m 28s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/22d6ce46d6c042ffac8b3c4fb77bc74b : SUCCESS in 1h 15m 22s","accounts_in_message":[],"_revision_number":2},{"id":"9f2facd2855be1b6a0145b6f4492875a1e6af86b","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-07-31 14:53:53.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5624366d8be8fcc9b90f7c283dcaa40667cedcc9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-31 16:38:11.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/46e69e8184e74032b377eb310cdb90ac\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/20a6e2a4e403427191d75067fae9d9a0 : SUCCESS in 36m 00s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/e9c0c9bc6c134903b3df4bf0191f87e7 : SUCCESS in 29m 11s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"58f957ddf61395324e00a52fed5bdbc2f42e655d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-31 17:26:17.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ac3af81cb37343018f7b8955042c0659\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/20808a946142454e97dadb951e46631a : FAILURE in 2h 07m 26s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/b21fabadc29c41b9a8aad432fa8297a1 : SUCCESS in 2h 29m 59s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/e85e4ef3f7e44482aad2fc087327f050 : SUCCESS in 1h 52m 32s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/dade4e14a11b45099c503b40f61a0dd6 : SUCCESS in 2h 07m 38s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/ea831ab924f34fb3af47069e2905a9ec : SUCCESS in 46m 56s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/22f51b48379146699f563da5c05903d4 : SUCCESS in 49m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d8fcd1f29e9a4f449fdd34583438f9df : SUCCESS in 9m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a05c40f445ba446091c6037529e4ce22 : SUCCESS in 25m 48s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a4ccc81bc53746458770c452dd0a3504 : SUCCESS in 21m 52s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/095d8925ff40403298c49ba747b06d20 : SUCCESS in 33m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d5d09bfedffb4acd91afa3236c827c7c : SUCCESS in 9m 02s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/1eeca11534e44f27815a39ed4d697bcb : SUCCESS in 20m 26s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/d9d9de8852d04f7d8633869a371a807d : SUCCESS in 51m 49s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/47a5495371a1408798563d67cd137922 : SUCCESS in 1h 28m 28s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/4b4f2787debf46128e53ee616910a05b : SUCCESS in 1h 32m 19s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/befbe58b6aa1494cb34026bd63273035 : SUCCESS in 58m 09s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/3d1604646c414fd09002f6b2dc2b2f8c : SUCCESS in 1h 34m 34s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/270af1e8f12247da8950e6bd4a28e6f5 : SUCCESS in 1h 28m 24s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/ef188cdb4fd44849bd4c1a54d818d840 : SUCCESS in 54m 13s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/d0b4f1513ed74ddd8cd24d2f740ddafa : SUCCESS in 56m 16s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/21446c6378794a1b8b882f58e5da26b7 : SUCCESS in 1h 03m 18s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/df78c0c10102474496bad51d813ec56b : SUCCESS in 1h 05m 03s","accounts_in_message":[],"_revision_number":2},{"id":"73081f6bbd6614a09ff1af1a22ce34b42fb18521","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-08-02 06:00:16.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f5250f7065d73ae120307b9d23d61ffdbdc929d8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-02 06:47:02.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3af6426c2cb54c1ca1b0429cd6d171d3\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/748dae12a71246d9a369a440692599ba : SUCCESS in 44m 26s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/9da78e3e7e5c4945a085c4ca9dd79691 : SUCCESS in 36m 14s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"027f37d9813d0b44b225cb7c8ed41c815aaf6711","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-02 08:21:48.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c8f8f86b7d61495092fdf632c3d62bc7\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/343e36b1b9104efe82094b4e0c476f6f : FAILURE in 1h 29m 16s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/5d99e99739f744258ef100af6870cec5 : SUCCESS in 2h 06m 17s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/c646e63445964d7b994bb1d1145de60a : FAILURE in 2h 14m 24s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/1901598e8fa64cef8cb9de42aa15f52e : SUCCESS in 2h 05m 37s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/46a89fcea6944cda8663b98bbf41c469 : SUCCESS in 48m 29s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b2f9b5c03d8b44efa6b8faf7a957af47 : SUCCESS in 46m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/665d65e9492c4f28a2f51f84d1ac492a : SUCCESS in 11m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/18d53966f7ed40a08e810f2be0f2445b : SUCCESS in 28m 12s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9f87ea91e2384be1baab66716a6bc768 : SUCCESS in 24m 07s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/48860fe44cc54ff4aca62481685c06b8 : SUCCESS in 25m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c205a075f2434e34a88cb1049ebc835b : SUCCESS in 10m 26s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/cabef19723ce4af99831ae5e5b1f79e7 : SUCCESS in 19m 14s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/71188ccca1fb425f992742ff4a6704a3 : SUCCESS in 1h 01m 15s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/a75bac913017495e8ae6ba279fb86f10 : SUCCESS in 1h 31m 28s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/a6685774543248848a06a5838319b7e0 : SUCCESS in 1h 00m 41s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/7be4f77ce8e84614a3a44d29226fcbc5 : SUCCESS in 1h 34m 48s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/a91649e2a14d4f59b71c4d03e846000f : SUCCESS in 1h 40m 58s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/f86dcad01a6e46ce9a1d901409fb27df : SUCCESS in 1h 24m 51s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/57a2d360812f49bba00dddec908c4996 : SUCCESS in 43m 52s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/03310906a97b4a2483edb2a96863dc19 : SUCCESS in 45m 48s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/02d78b19bb8842fc83438ba7d62d7d55 : SUCCESS in 49m 02s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/b04c650c791645be80424ea611d062f7 : SUCCESS in 1h 07m 28s","accounts_in_message":[],"_revision_number":2},{"id":"ec15e44886846caa43e531be2e1849c5c7b3d9ef","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-08-02 08:57:17.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"70431daba04577318e7b6d1d572422500864b405","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-02 09:26:51.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/128aa93c35e64b508dbe30e555be5368\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/dfac2c4591084e8392aef4abfc5b7f27 : SUCCESS in 27m 48s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/71021497b1554fcc96d75fcab10993e3 : SUCCESS in 20m 30s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"2a582efa4fa4ef59b855624d7b68c59e3fa1d2c9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-02 10:50:42.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5064e89c53894d4989a427261f2451af\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/0e1bd10f3e164d54abd9f3299b32212a : FAILURE in 1h 45m 38s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/a368197057314922bbb4a059f5f68780 : SUCCESS in 1h 38m 49s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/5f1075ec9b66471ba8cb6ddf3d171217 : FAILURE in 1h 28m 34s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/2e55947e97694e90995b518841ab1255 : SUCCESS in 1h 46m 32s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/a8069f6d5814459e990de2d52b3f8f19 : SUCCESS in 33m 27s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/99f3ce59a0c74266abe22bd09dbd96ee : SUCCESS in 33m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/54bf99d73b064be7ac8c75f98a0ed1fc : SUCCESS in 9m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a118454414494d4ea4dde22f748686f5 : SUCCESS in 19m 42s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/dbcb812e08834848a566c561315d6f8c : SUCCESS in 18m 12s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0678c704e3d5456b99ae85a2d952e184 : SUCCESS in 25m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/787268f45aaa4b27948463cd8c065bd7 : SUCCESS in 9m 55s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/d762e5575bfe4a6fa638d247ca7eb460 : SUCCESS in 18m 01s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/51ee7b12b6b248ff88e5acb99c8c2a78 : SUCCESS in 52m 14s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/1c07f872006b47258bc69e3b549ffe04 : SUCCESS in 1h 47m 48s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/14540c132a2f427c8e402eef5660f2a7 : SUCCESS in 1h 00m 15s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/2169308bc1564090916d916837c94b6d : SUCCESS in 1h 13m 54s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/a62ca8881d6042daa4c120d2145c475e : SUCCESS in 1h 44m 51s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/eed29ed3925841bcb274a015de397be1 : SUCCESS in 1h 14m 41s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/9eaa2bd390ab43bda0d93cdd90919acf : SUCCESS in 46m 00s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/4ac57152fcf34c2faad5a6e1646f1cad : SUCCESS in 56m 18s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/f6bd604b26ab4a18ba8297ade87e8f27 : SUCCESS in 1h 03m 17s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/924c0602fb1047d4b6435e45d03adbd2 : SUCCESS in 55m 08s","accounts_in_message":[],"_revision_number":2},{"id":"95aae46777567d4c286b5a56f5f7a97f18a247e2","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-08-02 11:25:20.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a83e97f4ea3030b25e568db7556a2b2145f96342","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-02 12:05:28.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6d4a54c84a974ec59422f9fede5d7202\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3951cd2ba08a4f84a4f41c2ca4dcee38 : SUCCESS in 38m 18s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/ad9564bd72014f4aaef6ddaf9de6f030 : SUCCESS in 32m 17s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"87a702d876baa8388b2898d8c6815f758f90e2de","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-02 13:18:25.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/89787c79ad424d7396996201673957c4\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/ba6d08dd9d864ceeabdbf8dde48d2018 : FAILURE in 1h 34m 14s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/c85f77a993204080a4001c0042eb0599 : FAILURE in 1h 46m 00s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/c04b1a6a969646879d7618709bf029d4 : FAILURE in 1h 37m 07s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/916d2f97b66f4dedaebd3cc2e4063f98 : SUCCESS in 1h 31m 22s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/73946d5a4d54411fa9e132e6954bbabc : SUCCESS in 36m 11s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8462ff9a9c7e4f16a51823ea711d649c : SUCCESS in 41m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d45ad6c7e6b044fba659ee4f56c67546 : SUCCESS in 9m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9ce358b06795491ba947402767ffdbfd : SUCCESS in 23m 29s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/fbfef57a6a884437ad96ed6c31ccb502 : SUCCESS in 21m 18s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/866a1248e4c8476394c209881beacc96 : SUCCESS in 21m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/658d7097e6274b2eb98e7130098ada8a : SUCCESS in 12m 36s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/e891c4e185f34fe7b9277cfecf72d05b : SUCCESS in 30m 37s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/3a563d5784424aa486a7f66b4b0aeada : FAILURE in 39m 18s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/7b9621657f1a4dc3b541b0ecd8cb2b2b : SUCCESS in 1h 46m 20s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/3289b59fe5b049e78b43a637138dabe3 : SUCCESS in 1h 05m 00s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/2218cb9e8efc456cbaa03578d9b16ad5 : SUCCESS in 56m 17s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/0f793fe00e9e4f9cad686e76771430ef : FAILURE in 1h 27m 34s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/19fbabd1b557401aa27e490ea45802a3 : SUCCESS in 1h 26m 16s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/46f3c22ba337407099bf97462bbf8547 : SUCCESS in 53m 45s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/e73460ee8bed46b7ba491c6d609660ec : SUCCESS in 44m 45s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/22ba590dd4274821b75713a5308a44f9 : SUCCESS in 49m 52s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/e74217de578a4657a76a70076d13569c : SUCCESS in 1h 08m 38s","accounts_in_message":[],"_revision_number":2},{"id":"19b1220395b65bf61924e7c342ed15fbde674bac","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2024-08-06 11:53:50.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a54a9a75f7daa3f1ff73ebdf7940cac2e0835420","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-06 12:32:17.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e3d1699339094cb887209b22efe1279d\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/79378aead7d946a2863899d761a743c1 : SUCCESS in 36m 48s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/6ebd4dfb21fa4fbfa3c973fc13684b04 : SUCCESS in 32m 32s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8f199b7021f6b705feaa7cb2c96b89ec5fa6165c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-06 14:13:44.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/44e15337f2114d0a91ff3b4f87b36926\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/31ad51dcf6864ee4a14540c82198eb63 : SUCCESS in 2h 14m 27s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/99b85bfefa514b66916f23d35c3b6c7e : SUCCESS in 2h 04m 26s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/a164d0c6bed54ae0aa5bcd77ad006701 : SUCCESS in 1h 37m 43s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/fe71f8e5b7d24b2ca8687872d462ac3d : SUCCESS in 2h 09m 35s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/e685a7931b4b406093137045bef6b70b : SUCCESS in 37m 27s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/72df2fb00c45485cb9f22915a0290b18 : SUCCESS in 50m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fc23734b1cc54d43a6c31576cb67875b : SUCCESS in 10m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/604c9757d39148dcb1f576101fd20457 : SUCCESS in 21m 13s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/bc52ead0736741f38b0a83b9ba4e1d08 : SUCCESS in 19m 22s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f96aa8b42e4e473b9b33a3ee6586f3c6 : SUCCESS in 30m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e0b1873e4e349c4afbec79230e72d24 : SUCCESS in 9m 49s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/760758855a8f489282236ea3c1f03103 : SUCCESS in 24m 10s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/027dd3e5d9dc421b827b29ce4deb9fdf : SUCCESS in 51m 21s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/b951f9daa78d49c9a3fa1dc918bc4cfa : SUCCESS in 1h 33m 08s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/d38a7c5c0075475884824502ef6ebc7d : SUCCESS in 1h 05m 03s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/4187f8acb8924ad092e9bcade148f480 : SUCCESS in 1h 09m 16s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/e8f9ee922ddd4a618fc7e7b6e0267f7d : SUCCESS in 1h 22m 22s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/f937e01a4fb2451bb30d6d88e71e139a : SUCCESS in 1h 28m 29s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/c27ddae3bba94fa89784311012646d14 : SUCCESS in 1h 01m 44s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/85f1f2e606e5406999af54eff41ae55c : SUCCESS in 54m 57s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/ce39d781035941fbbb3dc5fb56ec1ffb : SUCCESS in 52m 43s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/c2231264508a4919b91884c146f1f7db : SUCCESS in 1h 07m 32s","accounts_in_message":[],"_revision_number":2},{"id":"022c989127e4f71342db77abf7524f6b903538b9","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-08-06 14:50:15.000000000","message":"Patch Set 2: Review-Priority+1","accounts_in_message":[],"_revision_number":2},{"id":"ead6a48f4377d80f9b4ec90f10070bc5228c44e9","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-08-06 15:15:25.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"55549cc929fc33f83507fc030f38483a5b899796","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2024-08-06 18:09:58.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"df0cf13429380bc1fd11e4a7d66640c004db1e7a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-06 18:10:33.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"9478094dfaee5e7a2759a908488a38e5579b4b01","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-06 21:19:14.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7333aa67855949a49a8d4ff0a7e481bb\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/f527ae4441e24e4bb8dde2b4e6ae3fc0 : SUCCESS in 1h 52m 27s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/e0515b4b1e7b48618afd680dfb826929 : TIMED_OUT in 3h 07m 02s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/7a99ebd36fc345f2a01b30358097aa0b : TIMED_OUT in 2h 53m 03s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/533091d240e3403b933c0a2d9559a882 : FAILURE in 2h 37m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0b669c881187428bb86066caf8ef2ff6 : SUCCESS in 10m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/09d756c6de3e4cb8b8da9f29b0d4ebd1 : SUCCESS in 27m 42s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b8952f8d319c437190eb9ac6a953fb91 : SUCCESS in 19m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/14b564bf3f8d44e29cd740ec64ebadfd : SUCCESS in 11m 17s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/e4a83d1ce016450dbd839e0e6327c23b : SUCCESS in 53m 49s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/2b21b0bd98c14f2396fe9aba8f3302aa : SUCCESS in 1h 48m 43s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/7b17e7c8900e4b2685bbc92a10d397b7 : SUCCESS in 1h 37m 42s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/3fd193c191234419b1db90e043d652bf : SUCCESS in 1h 11m 33s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/78ad3286e46a4f17bb363266e0db9653 : SUCCESS in 53m 31s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/f89dfdb8e5c34d7d857a6717946dfcef : SUCCESS in 58m 21s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/91fa993de1284ca7af47039676447ec6 : SUCCESS in 1h 04m 45s","accounts_in_message":[],"_revision_number":2},{"id":"13854d9e99a016641d9e781e7123ef9836c99236","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-08-07 01:31:56.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"26e938e1d1fecb1d78acbdd6c4f7f7159158ee33","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-07 01:58:08.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f9112074187b48b3bafa26e5cc18fb39\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a91a186e3e514f4bb99a4dcd3ed97019 : SUCCESS in 24m 13s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/4292f2ae9b374283b768801096c91fb3 : SUCCESS in 22m 33s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"9b3a8bcd838919cddbce725c746b19cc9db65f75","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-07 04:04:48.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3a6dab02e11d4a9080fc2c0cbdbc0fc5\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/96c49e584d9043eb8f3b4fb542fcd340 : FAILURE in 1h 43m 21s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/7d93a5f937384bee95337918ca891bf3 : SUCCESS in 2h 12m 17s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/679294d7720a41a3af4f2be9252f8ec1 : FAILURE in 1h 34m 28s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/6d393b401dd64956b94d80e5e20c02e8 : SUCCESS in 2h 30m 51s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/2f597c3697c4489faa4d75394e5f0836 : SUCCESS in 35m 27s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5717d6eeaaed46238bf792fe74d7ec75 : SUCCESS in 44m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0d729236251c4bdaa80f42e74a4b6ec3 : SUCCESS in 10m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8557cd32b3024133a4cbd5565ee10d3d : SUCCESS in 17m 30s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e79c3edea9e343d3ab0981da2e10c7fb : SUCCESS in 19m 57s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/bc97ef78c5af4842910206fd686a89d1 : SUCCESS in 23m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/25157b83b0734379884d4d707e51dc58 : SUCCESS in 10m 25s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/5d286c5066da4862ab7cfa7cfa60a45c : SUCCESS in 18m 11s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/caba84dd82804f89a6fc1143619b4888 : SUCCESS in 56m 50s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/3b1b7b7a2b484fd688ba9eab3c181f80 : SUCCESS in 2h 08m 02s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/130ae18d4eac4fe6ac1e88660d23f85f : SUCCESS in 45m 26s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/94393cae6a9b487f84136ddfe7bf6cf6 : SUCCESS in 1h 09m 30s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/cdb308c6b7ed43c89ce5d7f2e808aa1c : SUCCESS in 1h 21m 28s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/f9cb0f5e90794a32bd066205a50cd7a6 : SUCCESS in 1h 36m 13s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/fc1a0cb1e49549bc88ad71a9d1a8998b : SUCCESS in 48m 57s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/1366053815b64904a61afa0be93b2902 : SUCCESS in 48m 23s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/e0a9977af0b046279e152d9573d23b50 : SUCCESS in 1h 06m 40s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/af1f2de16d024422a04052f2c6ce9510 : SUCCESS in 1h 09m 03s","accounts_in_message":[],"_revision_number":2},{"id":"37a79d41bed32b52b414ffbd6adf30f132668a24","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-08-07 11:59:44.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"78be338d90df66f25cbd22c1729cf28e145c0221","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-07 12:34:34.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1869f9b993ac49c0a5c6541e87ed00f5\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6be88fd5829443b387f474f71810fed5 : SUCCESS in 32m 57s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/f511cdd873bd48e1b5a7e4ffbe9d283e : SUCCESS in 28m 26s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"39fea7de2789c82edeb9ae8e9d47763e234591fe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-07 14:23:02.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c78a1d3dabe6411f844e098d2069f850\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/206bd38140f44f009b51a1f7933c0420 : SUCCESS in 1h 41m 38s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/5208ae717e9544e29e8a701d5762bba0 : FAILURE in 2h 21m 01s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/0c4e31d11ede4376a24becd0ba511e02 : SUCCESS in 1h 45m 05s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/20cd5c5247394cc29f81b08df0f8f750 : SUCCESS in 1h 49m 11s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/83c38bae921b4ddeaff3aabdb583b1ab : SUCCESS in 43m 34s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6d87a9f5b738460495e7ec0d5c1ed616 : SUCCESS in 26m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9ff6813a3e854507985c39b1ad516d9d : SUCCESS in 9m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/662df80ae85f479bad9ac203056b40bb : SUCCESS in 20m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/935b9d440b52494fa6f5fb0913622ac8 : SUCCESS in 21m 07s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/09ab51874e6843e2972cad5a65ca14ab : SUCCESS in 25m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e6dec11ca8bb44ab874dac755fa9b88a : SUCCESS in 9m 38s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/4bd346c874944c609f8327ce45d8c723 : SUCCESS in 22m 08s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/e5af696d04844521ac2680f94078f6f5 : SUCCESS in 55m 07s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/ad819d48aa1b4f6dbb7b5b827459696f : SUCCESS in 1h 36m 14s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/5b763f0ba7684e01b2f2b5823913c068 : SUCCESS in 52m 13s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/7e2e427a595c47b49467b95a00a8eca0 : SUCCESS in 1h 00m 22s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/8133c0fe7b714c4d82621affad4ce7ca : SUCCESS in 1h 39m 10s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/1c719dafab064eb580839131f1ef62a1 : SUCCESS in 1h 27m 33s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/1b5a7557a8b842f590751cb00d2e4392 : SUCCESS in 50m 06s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/602a32542748495094b50be103d3d32e : SUCCESS in 57m 11s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/7ec54d9bc70c4ea99929347ed2cf8ef6 : SUCCESS in 47m 19s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/ba767929a201414b8dd7cb30ea6e3958 : SUCCESS in 1h 13m 40s","accounts_in_message":[],"_revision_number":2},{"id":"4dc9e361928a77f5b019f87d4111305933049952","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-08-07 16:45:51.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"dcc3f943bf0b7babe39d241d16e9e7b2924c2b02","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-07 18:56:51.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/218f30ac66ce494ab1a899b84a018df5\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/567f1343797744a1a551ccbc2917e2d8 : SUCCESS in 1h 56m 42s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/5bcd57d9936440dd820b12f3a2e24f8a : SUCCESS in 2h 08m 56s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/97037260b86c4b00959a6994cc3e3b4b : FAILURE in 1h 47m 00s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/7033f2cecc3c4944881367f4e19e9c40 : SUCCESS in 1h 55m 56s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/f757d28d56484bd8b5a2997ecc4b31f2 : SUCCESS in 47m 01s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e484f004154441168ca005833f65611e : SUCCESS in 52m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4532a76950a945f8a3419306351776e8 : SUCCESS in 10m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4e8259f31abc4fcdbe964daef8b3b70a : SUCCESS in 26m 52s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7dcc4d9695634610ad54af84e4ee2c57 : SUCCESS in 23m 42s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/379a29b19cd4484a96648b635fbe79ac : SUCCESS in 21m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/381af9b8aa5c42f2bfb8f3d187ea994b : SUCCESS in 12m 02s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/cce0c61f1923413dbd9f2426fc07a3a1 : SUCCESS in 22m 13s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/3568676d99ee4559ad5181cca0213276 : FAILURE in 47m 57s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/3aaacaa598a9477e9ace0e4ed1b67588 : SUCCESS in 1h 29m 55s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/a6812a14bf6e4cc2bb1396d71019b109 : SUCCESS in 58m 52s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/2cbb82b65b054a00ad316ed383f311d5 : SUCCESS in 1h 15m 49s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/fff1e8ae74984d0ab7d224771da715cf : SUCCESS in 1h 22m 54s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/47395e4bc2a746d984ac9d7570a510cc : SUCCESS in 1h 16m 30s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/c31a265da1db4033aa45187cf2b0f01c : SUCCESS in 45m 47s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/fb03014485d14f0398616724e35013e5 : FAILURE in 23m 28s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/3817dcc788b74972805cfe1d91e15980 : SUCCESS in 53m 26s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/46940f97c34b4d65b67d329103ae64dc : SUCCESS in 1h 02m 08s","accounts_in_message":[],"_revision_number":2},{"id":"0f72191942de6a9ab3199518cba0ee9f9f6da9d8","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-08-07 19:01:56.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7d4d553f2becea5aa60d2c95fc1d90ea18326ee4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-07 19:47:01.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ff6ad203ef6141fab9fee1ca3d296e67\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/89158e74b06c4c8387a1e7adb2e20350 : SUCCESS in 34m 18s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/c3caaa787e3548909995c908dedd240a : SUCCESS in 26m 18s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"7fe6ef3583de3982dec4a35e323cd00f741dbaab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-07 21:27:19.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/71eae178503447ca99f4720cc08a024d\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/c102edb4bde04d649674811e4c2fc440 : SUCCESS in 1h 40m 41s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/6a198d6c86014a298f3c285b1e97a116 : SUCCESS in 2h 05m 06s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/86b4ba7299314887b9cf6c3acc8a2d70 : FAILURE in 1h 49m 44s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/3b713ee084734e829b19c57a0121700c : SUCCESS in 2h 22m 36s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/7eaadbf83e5246649fd90ae44c6a28f2 : SUCCESS in 38m 19s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/110b10c588f74d4b937d7e0d5cf20bd9 : SUCCESS in 45m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7d4734300a01462d86d7fe2635958ab6 : SUCCESS in 10m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dfeeb4aee2244bf3bf73d92a71b13cf1 : SUCCESS in 22m 02s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/154fd94288af48f79acd36c483f4c8de : SUCCESS in 17m 25s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6bb563b357d843b5a93c485e70a96fae : SUCCESS in 24m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d795b56cf34745fcbc4a6e05edf7ebb4 : SUCCESS in 10m 34s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/619f7d5124c04151ab61036fafe4a89a : SUCCESS in 18m 36s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/ce6da9b7aede4cb5b9fb91f4aba7a498 : SUCCESS in 56m 56s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/77e044b3747c4779ba2d6959bc7436a6 : SUCCESS in 1h 46m 20s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/49cc28dbb2cc411da6c87f7270eca933 : SUCCESS in 1h 01m 14s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/314c44a9c589426baef2dd63e0e987bf : SUCCESS in 1h 07m 51s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/d27a4f6d76f44a2e99ca0d2cdbc9d516 : FAILURE in 1h 02m 54s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/c912875ac26b47a88123e76ae83a28f3 : SUCCESS in 1h 05m 24s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/f53518fecd294b3e9d467408ce71e116 : SUCCESS in 50m 15s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/7535e4f12eee40c889d444e985e32f0b : FAILURE in 19m 20s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/78f2d7eac2f54417ad7b34f2b65867e6 : SUCCESS in 59m 17s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/21d28fd7b06d4948b2e3138385caad8d : SUCCESS in 1h 13m 11s","accounts_in_message":[],"_revision_number":2},{"id":"17dd51278d9484b3e939326253a46910b2279f32","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-08-07 22:10:58.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"88c67bc58269e4f910f616d0ea1f81363205713f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-07 22:41:49.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c434b48f85864bb794d73fb8e0767221\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c6c0adc1a56c4a9f9d0d49c194ebcc71 : SUCCESS in 24m 21s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/8a5a3a5ecbcf43b59579d7b4999bc789 : SUCCESS in 29m 08s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d48f15ccec52ae6747fdee128f1fb35e02e6e7ad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-08 00:22:45.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/75191a79124e481481ee1aa1a2303385\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/f361574cb92744879011745e53093e57 : SUCCESS in 1h 23m 50s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/399b54c4ce6a4aacb7b185c165b08b7d : FAILURE in 2h 03m 12s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/cac023b605d04ebd9ee096b53682c517 : SUCCESS in 1h 41m 40s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/8f80bd7cf6dc4e1fb3beec1e9dc78a27 : SUCCESS in 2h 09m 44s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/93e0f216d4f74bdfa0d9b8eb7a1a0798 : SUCCESS in 40m 06s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/512733cc92594642bd4c1fb8214e03fc : SUCCESS in 47m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/87e8950906af47e9a78dc30ab6a9645d : SUCCESS in 10m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6a3ab4095fbf40b6b50ddb3b482ea62f : SUCCESS in 23m 52s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f9073361621340fbaa72e72dcb326eb3 : SUCCESS in 19m 53s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/719fe6d546854b7a88babf6a460afca1 : SUCCESS in 28m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e2bf0c9c9544e01a2f16444da8d13b7 : SUCCESS in 10m 06s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/9b641ba0b1df44469d9e13f6d913a6ae : SUCCESS in 26m 22s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/af805945e1eb4e41ad0c8153a05244fd : SUCCESS in 47m 36s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/dfc67392bdba441594b95720d7808a97 : SUCCESS in 1h 30m 10s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/9882114a62ea46209fab594566f4d688 : SUCCESS in 39m 44s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/c443dea9686941ada2359ac14251cf85 : SUCCESS in 1h 00m 41s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/3766da5a95a54ef3893dad39e6a90c31 : SUCCESS in 1h 19m 15s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/e4c8370319f64f6285973ddbf5552276 : SUCCESS in 1h 13m 47s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/284f8ce2ec134f6c92ba04a8a4ff4249 : SUCCESS in 42m 45s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/3c8b08df56a447d393ba88d73c2ac36e : FAILURE in 22m 31s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/3aee79a828e04003bbcd37fc708d0680 : SUCCESS in 49m 55s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/798c1a6c23b44e06a49a1942493198c6 : SUCCESS in 58m 16s","accounts_in_message":[],"_revision_number":2},{"id":"55bc7adca1ae9a079aacd3414dba8fa93bb01677","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2024-08-08 13:01:59.000000000","message":"Patch Set 2: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f00efc45db7fba00b5ad46d653c13d9a02b22a62","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-08-13 14:27:53.000000000","message":"Patch Set 2: -Code-Review\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c5e0a6c3978a6d4b08d8e4c8eb3d92a06e643f30","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-08-13 14:28:09.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"c44310e5d8f63fa715d3a7135b0a898663a7d4ea","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-08-13 15:04:18.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"47a0845396f6036a96194025ffa776e47275ff4f","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-08-15 17:26:04.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5278a248766e698d13310c297af57f407f67ec1c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-15 17:48:03.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d32fffb4addf4dbfbb6ffabce40b2cdf\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d3d54ee777f749a39d411768720827c8 : SUCCESS in 19m 40s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/b182135ea8c14f0b8949b65d2077217b : SUCCESS in 19m 10s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"5e1778e22aec921d73c9d5de71d6ccf81ffaa90e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-15 19:39:05.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a3396fb5c8bd44db89122ed4e58183e8\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/52768274ab624d09beda2b63a2b5df63 : SUCCESS in 1h 13m 40s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/a0dd7db0dd3b4acb9aeb41c9828c391c : SUCCESS in 2h 09m 24s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/e6ac3874019c42b787906c18bb8d7188 : SUCCESS in 1h 36m 30s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/9c88d6469eba4a3a99507d8fdbe361ac : SUCCESS in 1h 37m 01s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/557835bb281e42b0a11c0525e08a8001 : SUCCESS in 35m 03s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/73b00f1302f14fad9d4940ff472fbb1f : SUCCESS in 48m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cbfa20dcb93e429fa59f69f90afa6570 : SUCCESS in 10m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/eb409ccc29474ce19d81ed861ba2fe84 : SUCCESS in 23m 59s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/19d055c51dea4293b1ac712ce19d76a1 : SUCCESS in 21m 19s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/40c697ec79a44e59931ebbe10bbb32c1 : SUCCESS in 22m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e736e3b6f6c14bb385cb53b125eb7f72 : SUCCESS in 9m 41s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/693fc1581776446eb48669ee1ffbfd5a : SUCCESS in 20m 37s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/9b6f08b6e5a5459a8f34aaa06b9d3836 : FAILURE in 35m 25s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/5dd999a51cf243499741c5946f0de12e : SUCCESS in 1h 37m 06s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/d93453bda54b487f950af7a1afb46598 : SUCCESS in 55m 17s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/6995f293f838427c9165abcc068d8dcb : SUCCESS in 49m 53s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/493170a7e0d04979b888bd33203fc33a : SUCCESS in 1h 29m 02s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/3eb29f07d9cc462d843cb59fce921d04 : SUCCESS in 1h 03m 52s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/31fe24e770c343eb84d47a2b82118d93 : SUCCESS in 53m 35s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/6680fbe04a304ba2bcc7efd71cf94f2c : FAILURE in 22m 23s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/7fcc3d6630854aa8b6a943ef4467997b : SUCCESS in 1h 28m 41s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/e1b99b0067e9492cb18e1fbf7bc02086 : SUCCESS in 1h 19m 29s","accounts_in_message":[],"_revision_number":2},{"id":"cebcba2154c5189264d987f59226c682904023f0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-08-19 10:59:55.000000000","message":"Patch Set 3: Patch Set 2 was rebased\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **changekind:TRIVIAL_REBASE** OR is:MAX OR is:MIN\")\n\nOutdated Votes:\n* Verified+1, Verified-1\n* Workflow+1, Workflow-1\n","accounts_in_message":[],"_revision_number":3},{"id":"8f7384d609b60c0fdd0f1f9443b979c38f415eb0","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-08-19 11:01:08.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"1da88228c172e2b897e31aeb2a4275caf92788d1","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2024-08-19 11:34:55.000000000","message":"Patch Set 3: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/August-2024/Arista_Tempest_Testrefs/changes/76/925376/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"e43ee0af6749994f2c43166379f0ba85746d05bb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-19 12:18:04.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/46763757de3241dea40e6f9a95cfb2b9\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4339b0737afa45cfbba17b1c2ff71adb : SUCCESS in 22m 48s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/249d9a64a7534efeb9a789d0ce06cd6a : SUCCESS in 33m 38s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"205f77dba38a96e35ed4e1a1d0630afe167e01b6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-19 13:12:28.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/27f6bb531e654295bb3a627713ad56cc\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/3bae9f7623d84109b70da43c156fcf20 : FAILURE in 1h 31m 12s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/ef73e0ba202c4649b697fba507e2d853 : FAILURE in 1h 57m 57s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/07ed54c571ad4d6fbb78e9f2a3d828a8 : SUCCESS in 1h 26m 22s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/25a290f17ccc4b88915e29f6057c6e9d : SUCCESS in 1h 49m 19s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/b565ddb0ed9f40c4b6615d8aa4733fda : SUCCESS in 44m 17s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5cdb733aba7448a887d5ae50b59e2d16 : SUCCESS in 32m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/32c6d316922b4c70a35456cf8f1634d5 : SUCCESS in 10m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/90d00cc0d51c402c849b1121e390e5ba : SUCCESS in 21m 57s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c5cf7e0674ea4fb087028a9b04e89ff0 : SUCCESS in 19m 57s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5bc79f05302149bba59862d67a8e9e19 : SUCCESS in 21m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7739b1ea46cc44a9a324a5b977fb1bed : SUCCESS in 11m 50s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/ca1e9685e62c464784525e4dea7bdd12 : SUCCESS in 22m 52s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/2a2f257f5165471fa0966cc67e916396 : SUCCESS in 48m 32s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/8eb74653cb7348b98905c90230a1e834 : SUCCESS in 2h 10m 49s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/8b76d8eae03f4536a91ca06e7e1720f6 : SUCCESS in 1h 00m 41s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/1c4da84da5c940d3b3347a68723a823f : SUCCESS in 1h 13m 45s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/5d95633c076f44af81c9acd84e1a425e : SUCCESS in 1h 36m 36s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/0a1d7d409fd8444eb4e5926e19c14d38 : SUCCESS in 1h 19m 28s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/02f233de6f8348dcac3a26ebff6b4c5e : SUCCESS in 55m 24s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/d706076b56df481490c7d006c9485458 : FAILURE in 31m 33s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/1cba249634d74494914a412d8e18edfe : SUCCESS in 1h 14m 52s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/49660c2b339e4dc6b8e3cbc5cd69775a : SUCCESS in 1h 13m 01s","accounts_in_message":[],"_revision_number":3},{"id":"aeed4956b0579f846d8f1178290c9c13a786eece","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-08-19 14:14:31.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"1ad7f28ef014eb681c067a3f1ca822c3665b60c1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-19 14:58:24.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/44a0e2154e7d4670a4cd9432464c93f1\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1040257185214b11a8abd8e6796d2d82 : SUCCESS in 39m 05s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/f4b60cb3a40343e39f54523ce1217c2e : SUCCESS in 31m 34s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"07270e5dee4d19e7cbc57a11ecfdbf110782fe5b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-19 16:12:45.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e222c1d75fac408497276037694bb70c\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/217c63060f044e7ea4208187d448c638 : SUCCESS in 1h 53m 28s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/3b64f7dc553c46dbbc1ab35a141bc35e : FAILURE in 1h 47m 41s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/9ba0ee01e92441a8b8cc961e7983946f : FAILURE in 1h 54m 07s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/18a3ff85d5394d5aacc2bf7dd5dd37ba : SUCCESS in 1h 56m 32s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/2a3b5dcb4fe44a3fb6ed62fae968d585 : SUCCESS in 35m 22s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dadf30dbc8fb4676b5f1bec9a16391f8 : SUCCESS in 45m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3397bfdd793a4b3abbb6b59153b2a090 : SUCCESS in 10m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/38d95317382f4a199c3de59241c905ae : SUCCESS in 25m 24s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6902c43d7ac642d9b2846b7a65adc773 : SUCCESS in 21m 17s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/94246c912f564210a09b28ad36f33dcf : SUCCESS in 20m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/05295c880cc24a698b4cda1fc9d0b51f : SUCCESS in 10m 34s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/073c9f6d093846e9aff41f11568e08aa : SUCCESS in 24m 31s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/e31a757697fe404f99414c38e33e9f94 : SUCCESS in 56m 44s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/bb378c4c8036478f9e336512fd3c913e : SUCCESS in 1h 41m 38s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/6cc04e8a8a644dbb8704c32ca26eb2b7 : SUCCESS in 59m 59s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/88f267755557483cb38181a45e149cf3 : SUCCESS in 1h 06m 40s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/2dc324baddb14ec5801e4764dad0de78 : SUCCESS in 1h 36m 23s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/9c4ff09bb2c841c1ba0190924133ba88 : SUCCESS in 1h 28m 23s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/1559f0b0bf204f558fe22e1d80d054d2 : SUCCESS in 52m 22s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/e60304d14bab4b86a408dbe7c5f83fef : FAILURE in 27m 19s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/1cc91f905e244436875d2d0b0b010fa5 : SUCCESS in 57m 13s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/242368cd9d44467aa5b60715f378d257 : SUCCESS in 1h 04m 30s","accounts_in_message":[],"_revision_number":3},{"id":"94a2fcf0ddddf9d3e5f2c13ac714d8c3d2a4feff","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-08-20 11:15:55.000000000","message":"Patch Set 4: Commit message was updated.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"**changekind:NO_CODE_CHANGE** OR changekind:TRIVIAL_REBASE OR is:MAX OR is:MIN\")\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1, Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"e949fbb1fa23a8adb055228afa0b8cb0d7291e64","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-20 11:23:29.000000000","message":"Patch Set 4:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/926621 is needed.","accounts_in_message":[],"_revision_number":4},{"id":"ac208a37a8e0e7edbfcc67ad8a234ce9ade1cebd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-20 11:23:32.000000000","message":"Patch Set 4: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/926621 is needed.","accounts_in_message":[],"_revision_number":4},{"id":"577add4ff1c9e73560e70afa408ce2d19b6ff7ab","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2024-08-20 11:48:15.000000000","message":"Patch Set 4: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/August-2024/Arista_Tempest_Testrefs/changes/76/925376/4 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":4},{"id":"3f7ed47988a2e63ed73eef07b209cce25564db6d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-08-21 11:33:27.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **changekind:TRIVIAL_REBASE** OR is:MAX OR is:MIN\")\n\nOutdated Votes:\n* Verified+1, Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"b021f96ae137476b713b38a5f8ed20427ce4df83","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2024-08-21 12:08:12.000000000","message":"Patch Set 5: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/August-2024/Arista_Tempest_Testrefs/changes/76/925376/5 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":5},{"id":"2047095b02861aa8a7631ee1c2a27a8325244555","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-21 13:53:20.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5939b06155984e52a79eb346b2873421\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/80dcf9becf634a03a1877a954371964e : FAILURE in 2h 08m 37s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/c9e33a2f9f644c19bf9c9eefeb06b186 : SUCCESS in 1h 29m 38s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/bba92295ddab41c1a0b4d34acde88e37 : SUCCESS in 1h 55m 26s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/4c3c794c63e843789df724be0905ef2b : SUCCESS in 2h 16m 51s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/fe87a9e2121e4ea7b34bffd926ae956f : SUCCESS in 32m 30s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a00ab85ca87c4a339c3676851d39a191 : SUCCESS in 51m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/30c91f70438d4ab0abcd37621c95ddff : SUCCESS in 10m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b0097b7df8794dd286f1b56feea0200b : SUCCESS in 22m 42s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/cabf8f626c6849189ac2c1513c047045 : SUCCESS in 22m 00s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/35ab5427b7a54b1d91a9ff60d9c36fb0 : SUCCESS in 29m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/339fabb80c8543ac8c7213ce5dd9187b : SUCCESS in 12m 17s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/8a7eccd68c5b4270938608dcf748b367 : SUCCESS in 23m 43s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/db332ec93d1c4b9c8f676c1b9a61517d : FAILURE in 45m 06s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/03699af8576848de8c3d59c64a726b57 : SUCCESS in 1h 44m 06s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/7db5e127df8d460db587bae0daebd879 : SUCCESS in 1h 01m 34s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/5157bd5bcc3c4042a49f049f65889419 : SUCCESS in 1h 08m 15s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/99f3c4775dfa441b903beee34f722e25 : SUCCESS in 1h 40m 37s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/d724ae73ac2d4c8a8662f35c14f46564 : SUCCESS in 1h 24m 39s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/fa8b77d2da2b4eaa87d577c762fc7d2f : SUCCESS in 1h 15m 25s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/18eb238a2e274e02a2919a009d8bbd18 : SUCCESS in 55m 46s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/4bc8d6d206554f6d9ca3501c064cc942 : SUCCESS in 1h 08m 32s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/92e22c631a2244ca8686220eb87235d8 : SUCCESS in 1h 16m 08s","accounts_in_message":[],"_revision_number":5},{"id":"911e87f6cf1c8e6d75d137ef3b77d606c3e59800","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-21 15:41:26.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1af8393dfafe43199dcc5bd6c833d4f3\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/33edc6013cf84787890f3d3b6c2ca0aa : SUCCESS in 42m 07s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/251a50d98c8c479ebd9b52609bcacb1e : SUCCESS in 30m 40s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"3af9928941dcf82b6fceb50e1625d0f3c66d7479","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-08-22 14:48:58.000000000","message":"Uploaded patch set 6: New patch set was added with same tree, parent tree, and commit message as Patch Set 5.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"**changekind:NO_CODE_CHANGE** OR **changekind:TRIVIAL_REBASE** OR is:MAX OR is:MIN\")\n\nOutdated Votes:\n* Verified+1, Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"a8b2969984978cd4e916af9836e64ec4966dd08e","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2024-08-22 15:21:38.000000000","message":"Patch Set 6: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/August-2024/Arista_Tempest_Testrefs/changes/76/925376/6 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":6},{"id":"80c8b4d01bbbf7673072c3446e4a16d51a39a28c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-22 17:27:05.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/654ae7e638494ef3899e970526718014\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/cd3ec6ab1109436a94ca5ad9c5bb14a9 : SUCCESS in 1h 52m 40s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/4503a18c84c349209853874bd8e62ef2 : SUCCESS in 1h 59m 07s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/5f868db719534fefba3d3d5e31d574f6 : FAILURE in 1h 57m 23s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/da946a62407145c9a229051b8b274957 : SUCCESS in 2h 35m 20s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/7a4074c5d78841e195b22746b4a12687 : SUCCESS in 42m 57s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/961da1a7f7924c7986670cec6233c72a : SUCCESS in 44m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1712f10539ac4f9fbc331aa3ed85b7e7 : SUCCESS in 10m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c1f1b8a52aad4080971c8db5b6701b31 : SUCCESS in 26m 09s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/072e818d859f4efda0fb1c7df03d08f1 : SUCCESS in 19m 26s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/fc97f534b96d45ac82763b48c5b96d79 : SUCCESS in 25m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2796eb7e05cc41b092dae12f1672e49f : SUCCESS in 16m 16s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/e689e647b97e40548cc2b64448e74077 : SUCCESS in 22m 37s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/0423240abdd147468971cdded92d968a : SUCCESS in 52m 19s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/06ee4f432c2946a892eaa86a6a62d618 : SUCCESS in 1h 43m 39s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/983f08c1379f4e87932d7ce1eb778651 : SUCCESS in 54m 49s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/83fa89cafb2f4efbaed58eef15b4ebcf : SUCCESS in 1h 04m 09s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/0319d81d1ff747838ee82c8e1582d145 : SUCCESS in 1h 43m 21s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/b4e851b930f84d2e80dbc560477c0057 : SUCCESS in 1h 27m 12s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/89a1fc35fac34cb78f42420c89bf5cc3 : SUCCESS in 1h 01m 01s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/70e685cdda474d99b044b352e53d0681 : SUCCESS in 57m 35s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/11f7d281add349958a8cbfc3d2a60c88 : SUCCESS in 1h 15m 21s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/4d79af28ecda4d0687c83f83bdd685d1 : SUCCESS in 1h 09m 13s","accounts_in_message":[],"_revision_number":6},{"id":"8a43edad34b5731c577a6825e800a542d5cd08a6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-22 17:59:34.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a8994748770c4637b2fca88d16745cee\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d40135ba1bbb4b799acebfc68fcc2c66 : SUCCESS in 33m 20s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/60315ff93f5b4fa7a54e0751fd0fe27b : SUCCESS in 28m 06s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"3dec43efe97c933a4a0d4383e5539f219e80a2cb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-08-24 11:38:25.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **changekind:TRIVIAL_REBASE** OR is:MAX OR is:MIN\")\n\nOutdated Votes:\n* Verified+1, Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"865389c90a45d890ba54ea25c0ccd2892fdeb32e","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2024-08-24 12:08:45.000000000","message":"Patch Set 7: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/August-2024/Arista_Tempest_Testrefs/changes/76/925376/7 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":7},{"id":"d63e43679c4b09a81e24b89af808628cb2a40b32","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-24 14:44:28.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/77f271203b034b3188ad43d7a8e524ce\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/f6570586a6424533a3fac53223c07948 : FAILURE in 1h 24m 53s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/a191b9716efd40f9bf3248a37f468376 : SUCCESS in 1h 35m 25s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/194b32a1422d47ecb33509be8c045d2b : SUCCESS in 1h 22m 02s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/c601e0b7a30f4a409f3ef18b2c6676e6 : SUCCESS in 2h 00m 29s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/5cb598f7d2624f55b3a2a6530a5f8ea3 : SUCCESS in 31m 10s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cfa1afb3207440a2ad85e400332b734a : SUCCESS in 25m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2be2dc05ad854b5aa29f975a245822c6 : SUCCESS in 7m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/04f37613f5ec41f992065e14a52da32c : SUCCESS in 18m 34s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3e75e38f3f8540c58061b69585a268e4 : SUCCESS in 14m 11s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4c637204e6ee4b65bfbb1c1357ed634a : SUCCESS in 18m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/94e40847de8c432ba6a147fc99cdbf3e : SUCCESS in 10m 42s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/abb7872f219e45e4849a6082c26961b2 : SUCCESS in 16m 08s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/fb445d1d6d224773929484ec99041466 : SUCCESS in 40m 50s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/66b73c6ef1ea40a4b908d6823d3c1694 : SUCCESS in 1h 49m 46s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/c30260aee4544ab79fd31f9686aefb50 : SUCCESS in 59m 29s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/da8eabadb9fe4a66a2ed5283a2930c29 : SUCCESS in 1h 07m 56s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/79b2948632cb44d1b3170b8412c575e3 : SUCCESS in 1h 45m 18s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/47ccf7bc2fc04cb6a11de14af4e24000 : SUCCESS in 1h 15m 21s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/2ba125276b0644d09a03061c3991a968 : SUCCESS in 1h 16m 47s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/e81f333073174d489f86e70542641799 : SUCCESS in 1h 01m 50s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/aee8b4faccdc4a5f9ed828b4f477b7d8 : SUCCESS in 1h 07m 58s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/645f5bcbf7b0403eabc25ee777d6e5cb : SUCCESS in 1h 13m 44s","accounts_in_message":[],"_revision_number":7},{"id":"93ecedc4b5bca7a73cd0ae344fb3994f81f2928b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-24 15:31:56.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2fb70910f3ff4ac3ba428fda3188e1db\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9420e1196ced4cb8bcd031d2072a8082 : SUCCESS in 39m 13s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/64af4c07174d4cefbbf52d22c6811e41 : SUCCESS in 28m 23s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"a8d5f28aca81d7fddae35d1a70b34d9c3b9aec41","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-08-26 07:13:16.000000000","message":"Uploaded patch set 8: Patch Set 7 was rebased.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **changekind:TRIVIAL_REBASE** OR is:MAX OR is:MIN\")\n\nOutdated Votes:\n* Verified+1, Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"c07c045863097721479badf28e0e99753e59dc4e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-08-26 07:42:50.000000000","message":"Patch Set 9: Commit message was updated.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"**changekind:NO_CODE_CHANGE** OR changekind:TRIVIAL_REBASE OR is:MAX OR is:MIN\")\n","accounts_in_message":[],"_revision_number":9},{"id":"c0437679ce75d238c7ccd36b88cdec68de6c20db","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2024-08-26 08:00:04.000000000","message":"Patch Set 8: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/August-2024/Arista_Tempest_Testrefs/changes/76/925376/8 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":8},{"id":"38f9792c15d98ea92840b905b359df52a03b5f26","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2024-08-26 09:03:35.000000000","message":"Patch Set 9: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/August-2024/Arista_Tempest_Testrefs/changes/76/925376/9 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":9},{"id":"eb5273d74121180061d810217b5885002c749c38","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-26 10:16:01.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2f08c8c4cfd647adb9a556d8819aeb6c\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/968272c2de8b4a26998749e43c657760 : SUCCESS in 1h 46m 46s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/6089b38a35e14041aa47c254e45d4200 : SUCCESS in 2h 13m 56s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/161f400d24544976926209e802aeb25b : SUCCESS in 1h 58m 16s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/37a66ba6eb4e4baabcaf13286e963540 : SUCCESS in 2h 31m 09s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/430f9c582ac644fb87ad847540ed0dd7 : SUCCESS in 40m 16s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/977fd03ffaac4ad7a95651411381b4fd : FAILURE in 37m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/23bf80089ceb437fb72f10189615bf17 : SUCCESS in 9m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9fcb386ae30148caa2c1338aed6a636c : FAILURE in 20m 30s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5d890090646144e0abcec4c129da850c : FAILURE in 19m 40s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/38f97b462a6f411e8743e7c32f4d9993 : FAILURE in 23m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a6b549d054d245caa39b465c6443f3bd : SUCCESS in 10m 22s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/003bddd998b242908cebceab72bb0643 : FAILURE in 25m 30s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/7d2a9a33503c4fb7985f4b7e901edaa1 : SUCCESS in 52m 51s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/e924dc5ffa9146b59007a0050e311441 : SUCCESS in 1h 33m 52s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/72bb474798a842ceb3e02ec6e196a8a8 : SUCCESS in 54m 47s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/d45adb42fd6e4cb5b7c1aaa1cc0162ca : SUCCESS in 1h 10m 16s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/eadf38787b2f439c98d8e8f2432a765a : SUCCESS in 1h 39m 40s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/6901fe118ebc4146b643b986ddc10093 : SUCCESS in 1h 24m 06s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/b580a088d0514bcd86a0f1a9d7a2a1cb : SUCCESS in 52m 04s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/3a96b18c37714f2dadb7d73a70bf491f : SUCCESS in 1h 05m 04s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/e8ad697727504237ae776a4e6334ee99 : SUCCESS in 1h 03m 51s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/805244500b224205ad926870420d5a63 : SUCCESS in 1h 27m 38s","accounts_in_message":[],"_revision_number":9},{"id":"1594556a57dd03ab45f766133b2d05f1e7d20266","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-08-26 10:22:12.000000000","message":"Patch Set 10: Patch Set 9 was rebased\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **changekind:TRIVIAL_REBASE** OR is:MAX OR is:MIN\")\n\nOutdated Votes:\n* Verified+1, Verified-1\n","accounts_in_message":[],"_revision_number":10},{"id":"196d463b93851ea1270af20f6511186da5a69438","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2024-08-26 11:27:50.000000000","message":"Patch Set 10: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/August-2024/Arista_Tempest_Testrefs/changes/76/925376/10 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":10},{"id":"a58e88fbe491d9c3a752b23125759a26ee767e4e","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-08-26 12:31:32.000000000","message":"Patch Set 10: Code-Review+2","accounts_in_message":[],"_revision_number":10},{"id":"04a4786bbb8c7f6f400c5e093fd4ec36ee1d7457","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-08-26 12:41:17.000000000","message":"Patch Set 11: Patch Set 10 was rebased\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **changekind:TRIVIAL_REBASE** OR is:MAX OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":11},{"id":"229f91bd0e6969f9bb9d3c246e3bbd16c2b4d073","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2024-08-26 13:46:24.000000000","message":"Patch Set 11: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/August-2024/Arista_Tempest_Testrefs/changes/76/925376/11 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":11},{"id":"26c9c569dc4edbfa3aff8273127d6628ba2e64dd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-26 14:46:03.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2ed82479712142498cd9801480d092be\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/02dce0107b5742e8851dd59a06575b05 : SUCCESS in 1h 16m 04s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/21d30560e0504d0b8047e0f66567cd66 : SUCCESS in 1h 38m 45s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/7dd8fa182ca64abe9ecfc180a11890e2 : SUCCESS in 1h 25m 30s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/0d3ba75fcf7348ed836f7b5779f00481 : SUCCESS in 2h 02m 55s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/f03c72cc80aa429a92132386a26e1a78 : SUCCESS in 32m 40s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e1e00f813dd9474d995c2420bb81720c : SUCCESS in 37m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/77d16927038e4e3f96399068eb7feaa8 : FAILURE in 4m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7891507ee45948e48b9452f66101d754 : SUCCESS in 20m 38s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5ab185acd4494c6ea0360f8d16812305 : SUCCESS in 21m 11s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8491cd899a2d434992ba3cb7d94b203f : SUCCESS in 22m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc92939bac3f4569835aebe869be31fd : SUCCESS in 12m 03s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/dd244b5bf8d946cbb8b5ce6153a4e305 : SUCCESS in 21m 56s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/55f5557279f04520846df7bae9ab87ae : SUCCESS in 55m 34s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/d2f1af0eb98b450fadc75681f50c56cd : SUCCESS in 1h 41m 52s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/f2f1d5ac54f34bd3bf63079b3728ee9d : SUCCESS in 52m 48s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/4a2b86918cd74368acd0f50b1f7c7120 : SUCCESS in 1h 01m 15s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/361e20c534db41ac906abec91123a707 : SUCCESS in 1h 32m 25s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/599e574ce01145f280e7fa8b75fa23e4 : SUCCESS in 1h 27m 14s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/77fe660f9b7743f59f6a1a776431f208 : SUCCESS in 57m 40s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/3d9b0df09994407bad3d19d710480433 : SUCCESS in 1h 00m 38s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/88bc6826606b401d8a5584f0ee392aa7 : SUCCESS in 1h 04m 44s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/ac671f76629a4e829f7ab0d70f723633 : SUCCESS in 1h 05m 55s\n\nWarning:\n  Comments left for invalid file neutron/plugins/ml2/ovo_rpc.py","accounts_in_message":[],"_revision_number":11},{"id":"8807d7e72e9555d1d65d27bdb4394e9843cb1f9a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-27 03:47:02.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/75de3994aa0d4fb68a19993545795f63\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4acc0b076eea4827ba810d603248be8e : SUCCESS in 33m 14s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/79b0e3585c2a4799977b99eb34b2d7fb : SUCCESS in 28m 01s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"503d87c926b3dc5e55c5fcada6b8a01e481af565","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-08-27 06:29:02.000000000","message":"Patch Set 12: Patch Set 11 was rebased\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **changekind:TRIVIAL_REBASE** OR is:MAX OR is:MIN\")\n\nOutdated Votes:\n* Verified+1, Verified-1\n","accounts_in_message":[],"_revision_number":12},{"id":"b3f0f2d3432977e46a6ab0bf932845e92d73cc35","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2024-08-27 07:34:21.000000000","message":"Patch Set 12: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/August-2024/Arista_Tempest_Testrefs/changes/76/925376/12 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":12},{"id":"be21000a7d88ca6c02cf153f905e0bc776318eaa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-27 09:15:12.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d78bb5f011644d219e5f8e95eb047679\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/87d78f8ca26c4484a6eca9a30ce1e9e1 : SUCCESS in 1h 18m 49s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/14498149f7594a23a5307a9bc147d88d : SUCCESS in 1h 37m 19s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/1b980272e28749e0a1e5b9dada115cc1 : SUCCESS in 1h 34m 25s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/a0b390c38f6842f2b34541cdb5daf3ac : SUCCESS in 2h 43m 12s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/02c3b9b0f26043529b2047a6eb109832 : SUCCESS in 37m 54s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0e804be7ab004434acb9208b00444009 : SUCCESS in 52m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1ecd8e0e3ebd474fb086e807dafaf8ac : SUCCESS in 10m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/980d676e99a8417dacef9895c40d7652 : SUCCESS in 25m 54s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b5f742c227e14d0d800e170c12ac4da3 : SUCCESS in 18m 55s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b9aa605a1f4e475a81ba4eccfee063f7 : SUCCESS in 19m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c6a50243ec4401a8b5b39f041961a89 : SUCCESS in 12m 05s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/2174f29d760545ea8d2d5a171ca799bf : SUCCESS in 22m 09s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/a19e46cbd00e48ea9ea3d7f28ba90bcb : SUCCESS in 55m 40s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/a7e140579a984785aa04e335e8ac4a42 : SUCCESS in 1h 50m 36s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/578811db2cc44d5ab9cade40e1db9e32 : SUCCESS in 59m 31s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/3fc378a9e7094c9f8bf8db89063982d0 : SUCCESS in 1h 14m 40s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/5ebbc32985b0489b81c5c2be3ebe4fe1 : SUCCESS in 1h 30m 40s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/b5318c1fe09148be9a79467434160ea4 : SUCCESS in 1h 16m 09s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/d35863f842b34542b579583ca105aaca : SUCCESS in 51m 39s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/bad5f5ce318244f88a7eb3324138b1fe : SUCCESS in 52m 08s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/a9d804dcf9b145839c47836baed10ff0 : SUCCESS in 55m 35s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/d363f6f8b35c4062a391eb03992e8dc7 : SUCCESS in 1h 13m 29s","accounts_in_message":[],"_revision_number":12},{"id":"c543f98955e242360f6ae982b69f0cce101a67e8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-27 09:40:39.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6bed104dfb52431a9d02048949354ea0\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ea234060fae442579434fa5a7600ca6a : SUCCESS in 24m 28s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/174edd57b4a140298adee2640a199966 : SUCCESS in 29m 51s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"f910c15f0f9b3e86a66a432fc011ad9ce1a73053","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2024-08-27 14:43:45.000000000","message":"Patch Set 12: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":12},{"id":"32e65b4e1993b9da584ab56e833d1ac63fd9c67d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"real_author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-08-28 19:00:18.000000000","message":"Patch Set 13: Patch Set 12 was rebased on behalf of \u003cGERRIT_ACCOUNT_16688\u003e\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **changekind:TRIVIAL_REBASE** OR is:MAX OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n* Workflow+1\n","accounts_in_message":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}],"_revision_number":13},{"id":"e512e7795f522d8819b72bfd153aca05aceba47c","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-08-28 19:01:03.000000000","message":"Patch Set 13: Code-Review+2 Review-Priority+1 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"d6d7b09f3701bdaa19772a0f69d85c6a3a89f19d","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2024-08-28 19:35:38.000000000","message":"Patch Set 13: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/August-2024/Arista_Tempest_Testrefs/changes/76/925376/13 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":13},{"id":"2fd703208d9bdd69403d85f8b23bd27e12b1528e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-28 19:49:48.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b6bd30d1a5c74869b2b2b16e43987b2d\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f33eb3dd878e446fb6e1b75919b05a3e : SUCCESS in 47m 16s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/2d8e4d35600d4969bdfb63ce6f66f55e : SUCCESS in 31m 04s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"bb66777132794097dce76100d021d8fb480f83d1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-28 21:39:02.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/efd9ebefb175438c961488da75691dd0\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/706ea41fbc5a48c7a30a22aa1b0d3015 : SUCCESS in 2h 22m 30s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/fb6465e375ee4a7c8bdfc20cf810b963 : SUCCESS in 2h 31m 53s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/85ea4fc8471c47a8b7200612b4895bd5 : FAILURE in 2h 14m 30s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/37962d83102c423a8fc124732dd216a1 : SUCCESS in 2h 07m 51s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/3c4aa42dd31d4d77aa4333cb22cbb6d9 : SUCCESS in 40m 46s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/41a56b76358246eaaad4fbd76e1f8edd : SUCCESS in 36m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/806c006de04b432abf57d7cd2d3f86db : SUCCESS in 10m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/84fc0707469c485b966ef0076b375c67 : SUCCESS in 27m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a9850b9fd1f840659bc7b59972d945a2 : SUCCESS in 21m 52s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7827bf99bdc1491da9781bc7c72da262 : SUCCESS in 24m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d41fc681f1b945058875859cd37cf398 : SUCCESS in 11m 02s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/4048c558abcb4ff084e4f3bffeb8a1c3 : SUCCESS in 19m 34s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/7dde7b0706e246d88c627c1c5d5c0908 : FAILURE in 45m 25s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/3795e7bc94b74e1dad53d9e105e2594c : SUCCESS in 2h 37m 01s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/9d1b3a51faa4453bba92a723b65784c0 : SUCCESS in 56m 41s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/2e84e5f0c26042659cd7db0460c9f5bb : SUCCESS in 1h 00m 45s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/35e1d9caa23548b28d44526457d1ecc2 : SUCCESS in 1h 40m 54s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/50b4d76baba54e35a0779acf89cd3832 : SUCCESS in 1h 35m 32s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/9351fb15e8554efea3da10c08123f867 : SUCCESS in 53m 55s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/fabb4c6d5c764d5b899d115bc857ec48 : SUCCESS in 56m 00s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/483044ab691946ae9c208de9d1908620 : SUCCESS in 1h 07m 20s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/539063313d0f4c228515654f02b9fe1d : SUCCESS in 1h 18m 07s","accounts_in_message":[],"_revision_number":13},{"id":"7b87293d14c9f214e1cf71c216454849a0fbddfa","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-08-29 05:49:00.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"1b03836261383eed64a03cc41d73ac3bd9d334c3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-29 06:55:38.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d149ba60340d4bdbb6c404138c592097\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9367b764a89f4a7180147b87d0114114 : SUCCESS in 45m 44s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/d111cd03a2ae41039b7378bae0a3c42a : SUCCESS in 26m 45s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"ff561af7256e7012dc1829b8a6e1924ad4136ff2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-29 08:18:51.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f62283023033494983f6747f4225f5c7\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/5d7eb98adc4c4b14b1bfb4d2ea2ee909 : SUCCESS in 2h 13m 33s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/b291895730f9472b9682e55cc8906c5e : SUCCESS in 2h 27m 03s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/e9616a087779468ca7324c87512c79b1 : SUCCESS in 1h 58m 41s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/7cf626214b454bb69b54a2ffbd8ec191 : SUCCESS in 2h 00m 57s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/1aae073e4e1642478021bccddab89a49 : SUCCESS in 35m 47s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0bd915ab3790471bb4bebb6b0a2d108f : SUCCESS in 47m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/74bfbd57e2cd4554884584a801e8ddbc : SUCCESS in 9m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a74bf34370544b1e97197941c1bcf8ce : SUCCESS in 19m 46s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b60c3c989cf141eabcedef896b6efdea : SUCCESS in 17m 55s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/658f3174b04e4324b27cb1845dcee600 : SUCCESS in 18m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eb5fc063e88d4b09bc1a211e8fef0954 : SUCCESS in 9m 29s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/02d8cc10fd444fc891c63a563ed079d9 : SUCCESS in 19m 18s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/34661d5b4cbf4602997b02be30acd0b8 : SUCCESS in 57m 57s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/dfb1f61ba7e24ba294599db915c6dc5c : SUCCESS in 1h 37m 49s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/029ede8ed1d24ef1aab6659ae2383438 : SUCCESS in 1h 06m 37s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/40004d07edf5492aa607df7c8680e37e : SUCCESS in 56m 59s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/23e8b8421b1e41ac946c488e9b735ae3 : SUCCESS in 1h 31m 13s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/7877b04c79034336bd27341cb7778be1 : SUCCESS in 1h 09m 39s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/c939d676b9404255a596ab4e34daa3c1 : SUCCESS in 1h 09m 13s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/257ecfa0ac244415b41c403a7e2f3557 : SUCCESS in 55m 37s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/e7c6d27300f84966aa7d127f638a845b : SUCCESS in 1h 00m 37s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/fb237f0243b743659384df68a56cafd3 : SUCCESS in 1h 15m 34s","accounts_in_message":[],"_revision_number":13},{"id":"2ba70e337bc75f984081e4512a6e14577af32641","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-29 08:19:52.000000000","message":"Patch Set 13: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":13},{"id":"258dc50e4be7f406151d222dfdfde0ffb7bc1e46","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-29 11:25:48.000000000","message":"Patch Set 13: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1acb91e3786e45e8916b28b3ed95fe26\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/e33e56a6294b49ba8b69c5268f0a3cc4 : SUCCESS in 1h 38m 00s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/e152625d81d247ef847466dfdfd375f3 : SUCCESS in 2h 39m 57s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/fdcded6a41c547da9bc3fe2b06942c88 : SUCCESS in 2h 13m 08s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/9e300431a5a84eccbb5f512d30b88a50 : FAILURE in 2h 10m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b5adf3099eb4d99b35e2eada68024e8 : SUCCESS in 9m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ffb131cffbfd45469c26b9ff946cc638 : SUCCESS in 23m 24s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/73666c60798b44cf95239f359b375782 : SUCCESS in 22m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/36b8ccdcdb48415d96ea229218265f14 : SUCCESS in 11m 02s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/b6a503a6cc524f94a0f97622dbbf6ae8 : SUCCESS in 51m 50s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/814593b565f14a62aeced5fa5251fe67 : SUCCESS in 1h 38m 18s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/4f6df5015bb8495baba5d766f9461e5d : SUCCESS in 1h 26m 36s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/2f49cdf39beb48fc9c2d1bc6ef858023 : SUCCESS in 1h 10m 36s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/33d1e7e65041402585e53fd499b92227 : SUCCESS in 57m 25s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/8cc692b36668412686a9707531ff4f2f : SUCCESS in 49m 43s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/4e23501b02de4d99a56e592b22b8ac54 : SUCCESS in 1h 02m 32s","accounts_in_message":[],"_revision_number":13},{"id":"860f0bcca886b48a8395a372aeaa33c2c3185a5a","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-08-29 12:56:39.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"140e234fdb05f871bb84c7ab9f4c4d71bcad01f4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-29 13:38:43.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/21347a7390634df49d945eb3370683a3\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3e090a82db814b479cb4d23f5f52bd7c : SUCCESS in 37m 27s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/bf76bc35e54d4464a16b144ba8879a4a : SUCCESS in 27m 28s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"88bd863cd0ba1f6f987f21cf164a6b4da980ce74","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-29 15:07:53.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f93eed6d5e02420ebdb8076227354b9b\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/e40e0b50a8674821bef2a4e06818c176 : SUCCESS in 1h 16m 19s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/8bae32dc0957459c9196260184421fd0 : SUCCESS in 1h 54m 15s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/537f9ee3623d4d938b4aba543a391835 : SUCCESS in 1h 23m 37s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/e15fffd4299a4245993b4eebf6eb9b4d : SUCCESS in 2h 08m 29s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d28fe3463234448d80eb4d4a54798d0b : SUCCESS in 32m 36s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3bfd2e3d05014f78abc860a0447f4394 : SUCCESS in 41m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/68d92260892f4ca1be97c199f34ddf34 : SUCCESS in 10m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8bd2192d3c8d47c5b8c83d439df81f9b : SUCCESS in 24m 45s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/12680340985e4570a694a046006e2465 : SUCCESS in 18m 19s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/589a6c2dda864e09b6cb3a099e78226f : SUCCESS in 22m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ddc24be5ea942d181b6fa1381c30165 : SUCCESS in 13m 52s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/7c735b2ec9ab43f08b0c2d423a48534c : SUCCESS in 18m 43s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/f553e79c14b447a2a4c240a565c5822a : SUCCESS in 56m 50s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/eb7aa90e07a54a9faa4fb758a15e5e7c : SUCCESS in 1h 38m 30s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/0268bb8a68a4430b8694044effb4a8ce : SUCCESS in 47m 33s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/0580d6f69616413da72e7d3a9791e3e7 : SUCCESS in 1h 07m 11s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/a980ad5862924db997fc5cba2675b646 : SUCCESS in 1h 34m 15s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/65a2fdfdbc754d24a468763bef29dcbd : SUCCESS in 1h 19m 20s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/50ee57ac7b6c4a629c33cf49fb146ad9 : SUCCESS in 1h 00m 36s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/e1f9bc28e45c473994bed7848e89779d : SUCCESS in 1h 08m 31s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/256af6e086b348da9e8a9b19ff02b8b2 : SUCCESS in 1h 01m 03s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/7906ee65e67e449aba8bc568314ea95e : SUCCESS in 1h 05m 17s","accounts_in_message":[],"_revision_number":13},{"id":"632eb62bc13fd880127096e3b828a2db6e908e96","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-29 15:08:38.000000000","message":"Patch Set 13: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":13},{"id":"b10527e251cbb1237840b26d22341ee2f5c981b0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-29 20:13:55.000000000","message":"Patch Set 13: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/47a7d4c8304e44cc934dfceda4bcd6b6\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/13f16b5aad2e4d418167850e08988088 : SUCCESS in 2h 06m 36s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/f85d8b3e3b1642de8350ea1d1b353782 : FAILURE in 1h 53m 38s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/1df81c71bc28405c918bb42ccf992f14 : SUCCESS in 2h 25m 19s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/d73a8d84c9dc4f5fb09325cb07c70d6b : SUCCESS in 1h 58m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/36b054158c1c4b5a9a1f871e40f156df : SUCCESS in 9m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/891153d74af847fa9d3856e86b9df0f5 : SUCCESS in 19m 31s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f8a7b06114da437093d09f18ff33959b : SUCCESS in 16m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/32eda197e5a542e98bfb4606b58afa6f : SUCCESS in 10m 39s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/eb1749d1b51243bb9445cd792c95483e : SUCCESS in 50m 24s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/6e8d8519a9a74c37895feb8c4f16d55d : SUCCESS in 1h 26m 12s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/6fd0547296684b12be993810e9bea2f4 : SUCCESS in 1h 33m 25s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/19804cc642d34de9bc48ab51c056449a : SUCCESS in 1h 05m 50s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/6666aaa75e764a669dd994f6a3c0d667 : SUCCESS in 39m 31s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/b233c3c12329453f9ff09ab2b4af1dbe : SUCCESS in 57m 09s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/9aa1f002ce514fb6b66321b875cef36f : SUCCESS in 1h 22m 49s","accounts_in_message":[],"_revision_number":13},{"id":"da7e3a5757529672d4f268a46f56ba492fbc4a11","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-08-30 09:48:32.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"d5b212263825ac7a22731abff9eaec48d052b0d7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-30 10:12:45.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/64a0205ba3f948f5ac7156ea0eb6d587\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a9a262ad60f04d5fa7a039ff2135aa67 : SUCCESS in 22m 33s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/a047389a92b042baa7ee918cf9a9c95e : SUCCESS in 18m 28s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"bf0f31c4cf3a473ed027a35185013557467745f8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-30 12:18:35.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7b063e19191a49258ee40077edfac572\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/bea7dda85c6842b9b0b1ce468664d2fa : SUCCESS in 2h 17m 02s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/f099e72ab8c94b7ebdcc6377e7e5b8d9 : SUCCESS in 2h 24m 47s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/8f98803e17f545ffa9c7e7ae59e419ee : SUCCESS in 2h 04m 32s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/f583c920270c4268b24f9b14a0af26f5 : SUCCESS in 2h 27m 31s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/5f4b2aed55fc495f99470ad27d03d657 : SUCCESS in 47m 21s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/eff8ef6b52434228ac46e5e4867ddc81 : SUCCESS in 47m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6e87a55e1f7d4ab5a9646aa5cd359ddf : SUCCESS in 10m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/007a758cc7b84371b8d4df0b2da59837 : SUCCESS in 20m 58s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ee32a83067d444e393b222566e3722c7 : SUCCESS in 19m 03s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1732dc0b3e034ce88caafc8ca2f4ea25 : SUCCESS in 16m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8232286e92fe432d8eb2352f55d854fd : SUCCESS in 9m 58s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/66ea194b75a242028e03072ffa8c63f4 : SUCCESS in 17m 01s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/42637b1d23714dd7b68875c986486577 : SUCCESS in 54m 24s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/680b472a594247e4ac4d38f83a1b5871 : SUCCESS in 1h 47m 05s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/8972d0ef3efe484eaacb2d3e564db51f : SUCCESS in 56m 12s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/ea5113640ca04396b3321d9733f03f05 : SUCCESS in 58m 22s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/be8694b261df4161b108e4900a44c27e : SUCCESS in 1h 32m 36s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/07044965e8424b2aac3ff1880203ef53 : SUCCESS in 1h 21m 34s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/034b4fd817f5424d8d21e79c672da245 : SUCCESS in 53m 18s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/6a4fee111251429289aaa80eadb3ccf8 : SUCCESS in 1h 00m 15s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/45ff242e52d942b38e56d1f6651fc683 : SUCCESS in 1h 07m 45s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/b1535d53017849f8b946986e0d6f82a7 : SUCCESS in 1h 03m 08s","accounts_in_message":[],"_revision_number":13},{"id":"1031f8ac7cf94898abc84dcc5fbf5366f30ef06b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-30 12:19:08.000000000","message":"Patch Set 13: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":13},{"id":"8e6ab0b51c5f6c0ccf20c2e041c1e4d4c4aef854","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-30 14:39:08.000000000","message":"Patch Set 13: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/88868f3f02914a20a4e40db1f0b685a7\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/f3968e111275448f8ee47761155af263 : SUCCESS in 1h 41m 44s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/4718e32580d94379a3437f6e8e445745 : SUCCESS in 1h 38m 14s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/d68d45086f944da4b32d35a565f7e281 : SUCCESS in 2h 09m 57s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/1b13b921662947b28b9d7966b1d88837 : SUCCESS in 1h 45m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/30fa3fe2b9af449b8eb59a3092a60673 : SUCCESS in 11m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fb1e605ece2d4c39bafde8b453e0f2c4 : SUCCESS in 19m 36s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c47c32e4caba44e3b70e4a64983b3ad3 : SUCCESS in 18m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/882ce6bd22514205901d90409a71e282 : SUCCESS in 9m 22s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/b97da586447d4d928a3e3c2c27ba4897 : SUCCESS in 45m 50s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/753ee530eecb44e7b609e8a2e927543f : SUCCESS in 1h 35m 00s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/68873e6f820044f0aa274861fb36c330 : SUCCESS in 1h 32m 54s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/da18cee506024a9e9412937b9359f4da : SUCCESS in 1h 18m 40s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/bcbfa3b78a174b90b9de082c4b82c7d1 : SUCCESS in 1h 01m 31s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/ff622147f3f343c88227a2170fd43b69 : SUCCESS in 49m 39s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/3ba2a3894bfd4363b6bd56819bddae34 : SUCCESS in 1h 05m 11s","accounts_in_message":[],"_revision_number":13},{"id":"6dd2b9d552849002ab69eeefc429b1038affa0ab","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-30 14:39:09.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":13},{"id":"02afb9c8758b84804c60be2999e05db5bf06796f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-30 14:41:14.000000000","message":"Patch Set 13:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1215c01e97d043718a641b733a195580\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff1d69dd94f34704a79054c69930ef3c : SUCCESS in 1m 29s","accounts_in_message":[],"_revision_number":13}],"current_revision_number":13,"current_revision":"76f343c5868556f12f9ee74b7ef2291cf5e2ff85","revisions":{"da057a454abad0b30f1558b988d44e0ec762037a":{"kind":"REWORK","_number":1,"created":"2024-07-31 10:57:00.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/76/925376/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/76/925376/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/76/925376/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/76/925376/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/76/925376/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/76/925376/1"}}},"commit":{"parents":[{"commit":"0488bdafa1aab877fc5907ae3cce89908cfe3803","subject":"Merge \"Only query for port do not get directly\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0488bdafa1aab877fc5907ae3cce89908cfe3803"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-07-31 10:53:14.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-07-31 10:53:14.000000000","tz":0},"subject":"Monkey patch the system libraries before calling them","message":"Monkey patch the system libraries before calling them\n\nThe Neutron API with WSGI module, and specifically when using ML2/OVN,\nwas importing some system libraries before patching them. That was\nleading to a recursion error, as reported in the related LP bug.\nBy calling ``eventlet_utils.monkey_patch()`` at the very beginning\nof the WSGI entry point [1], this issue is fixed.\n\n[1] WSGI entry point:\n  $ cat /etc/neutron/neutron-api-uwsgi.ini\n  ...\n  module \u003d neutron.wsgi.api:application\n\nCloses-Bug: #2075147\nChange-Id: If2aa37b2a510a85172da833ca20564810817d246\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/da057a454abad0b30f1558b988d44e0ec762037a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/da057a454abad0b30f1558b988d44e0ec762037a"}]},"branch":"refs/heads/master"},"d6f6548edfe29e4c97e7546c679b735bf100e858":{"kind":"REWORK","_number":2,"created":"2024-07-31 11:40:16.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/76/925376/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/76/925376/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/76/925376/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/76/925376/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/76/925376/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/76/925376/2"}}},"commit":{"parents":[{"commit":"0488bdafa1aab877fc5907ae3cce89908cfe3803","subject":"Merge \"Only query for port do not get directly\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0488bdafa1aab877fc5907ae3cce89908cfe3803"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-07-31 10:53:14.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-07-31 11:40:05.000000000","tz":0},"subject":"Monkey patch the system libraries before calling them","message":"Monkey patch the system libraries before calling them\n\nThe Neutron API with WSGI module, and specifically when using ML2/OVN,\nwas importing some system libraries before patching them. That was\nleading to a recursion error, as reported in the related LP bug.\nBy calling ``eventlet_utils.monkey_patch()`` at the very beginning\nof the WSGI entry point [1], this issue is fixed.\n\n[1] WSGI entry point:\n  $ cat /etc/neutron/neutron-api-uwsgi.ini\n  ...\n  module \u003d neutron.wsgi.api:application\n\nCloses-Bug: #2075147\nChange-Id: If2aa37b2a510a85172da833ca20564810817d246\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d6f6548edfe29e4c97e7546c679b735bf100e858"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d6f6548edfe29e4c97e7546c679b735bf100e858"}]},"branch":"refs/heads/master"},"709234e38c9f71e3030758582d6eb085d1dfa094":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2024-08-19 10:59:55.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/76/925376/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/76/925376/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/76/925376/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/76/925376/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/76/925376/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/76/925376/3"}}},"commit":{"parents":[{"commit":"f17cc24e8adb2bf18af32a45a44e68790c50dc6b","subject":"Adopt to StandardAttribute load method change to \"selectin\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f17cc24e8adb2bf18af32a45a44e68790c50dc6b"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-07-31 10:53:14.000000000","tz":0},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2024-08-19 10:59:55.000000000","tz":0},"subject":"Monkey patch the system libraries before calling them","message":"Monkey patch the system libraries before calling them\n\nThe Neutron API with WSGI module, and specifically when using ML2/OVN,\nwas importing some system libraries before patching them. That was\nleading to a recursion error, as reported in the related LP bug.\nBy calling ``eventlet_utils.monkey_patch()`` at the very beginning\nof the WSGI entry point [1], this issue is fixed.\n\n[1] WSGI entry point:\n  $ cat /etc/neutron/neutron-api-uwsgi.ini\n  ...\n  module \u003d neutron.wsgi.api:application\n\nCloses-Bug: #2075147\nChange-Id: If2aa37b2a510a85172da833ca20564810817d246\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/709234e38c9f71e3030758582d6eb085d1dfa094"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/709234e38c9f71e3030758582d6eb085d1dfa094"}]},"branch":"refs/heads/master","description":"Rebase"},"d48e4caf56e5f33ad5f912286b289383029b98d2":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2024-08-20 11:15:55.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/76/925376/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/76/925376/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/76/925376/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/76/925376/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/76/925376/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/76/925376/4"}}},"commit":{"parents":[{"commit":"f17cc24e8adb2bf18af32a45a44e68790c50dc6b","subject":"Adopt to StandardAttribute load method change to \"selectin\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f17cc24e8adb2bf18af32a45a44e68790c50dc6b"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-07-31 10:53:14.000000000","tz":0},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2024-08-20 11:15:55.000000000","tz":0},"subject":"Monkey patch the system libraries before calling them","message":"Monkey patch the system libraries before calling them\n\nThe Neutron API with WSGI module, and specifically when using ML2/OVN,\nwas importing some system libraries before patching them. That was\nleading to a recursion error, as reported in the related LP bug.\nBy calling ``eventlet_utils.monkey_patch()`` at the very beginning\nof the WSGI entry point [1], this issue is fixed.\n\n[1] WSGI entry point:\n  $ cat /etc/neutron/neutron-api-uwsgi.ini\n  ...\n  module \u003d neutron.wsgi.api:application\n\nDepends-On: https://review.opendev.org/c/openstack/neutron-tempest-plugin/+/926621\n\nCloses-Bug: #2075147\nChange-Id: If2aa37b2a510a85172da833ca20564810817d246\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d48e4caf56e5f33ad5f912286b289383029b98d2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d48e4caf56e5f33ad5f912286b289383029b98d2"}]},"branch":"refs/heads/master","description":"Edit commit message"},"849d5112a486d19b55e17f3fe4b8a6d065c5a11a":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2024-08-21 11:33:27.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/76/925376/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/76/925376/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/76/925376/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/76/925376/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/76/925376/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/76/925376/5"}}},"commit":{"parents":[{"commit":"0807c94dc9843fff318c21d1f6f7b8838f948f5f","subject":"Merge \"Fix support of IPv6 only networks in OVN metadata agent\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0807c94dc9843fff318c21d1f6f7b8838f948f5f"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-07-31 10:53:14.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-08-08 19:17:48.000000000","tz":0},"subject":"Monkey patch the system libraries before calling them","message":"Monkey patch the system libraries before calling them\n\nThe Neutron API with WSGI module, and specifically when using ML2/OVN,\nwas importing some system libraries before patching them. That was\nleading to a recursion error, as reported in the related LP bug.\nBy calling ``eventlet_utils.monkey_patch()`` at the very beginning\nof the WSGI entry point [1], this issue is fixed.\n\n[1] WSGI entry point:\n  $ cat /etc/neutron/neutron-api-uwsgi.ini\n  ...\n  module \u003d neutron.wsgi.api:application\n\nDepends-On: https://review.opendev.org/c/openstack/neutron-tempest-plugin/+/926621\n\nCloses-Bug: #2075147\nChange-Id: If2aa37b2a510a85172da833ca20564810817d246\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/849d5112a486d19b55e17f3fe4b8a6d065c5a11a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/849d5112a486d19b55e17f3fe4b8a6d065c5a11a"}]},"branch":"refs/heads/master"},"febded1e6c431c4f4271bcbc17b5c37ac024dac6":{"kind":"NO_CHANGE","_number":6,"created":"2024-08-22 14:48:58.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/76/925376/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/76/925376/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/76/925376/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/76/925376/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/76/925376/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/76/925376/6"}}},"commit":{"parents":[{"commit":"0807c94dc9843fff318c21d1f6f7b8838f948f5f","subject":"Merge \"Fix support of IPv6 only networks in OVN metadata agent\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0807c94dc9843fff318c21d1f6f7b8838f948f5f"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-07-31 10:53:14.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-08-21 13:05:59.000000000","tz":0},"subject":"Monkey patch the system libraries before calling them","message":"Monkey patch the system libraries before calling them\n\nThe Neutron API with WSGI module, and specifically when using ML2/OVN,\nwas importing some system libraries before patching them. That was\nleading to a recursion error, as reported in the related LP bug.\nBy calling ``eventlet_utils.monkey_patch()`` at the very beginning\nof the WSGI entry point [1], this issue is fixed.\n\n[1] WSGI entry point:\n  $ cat /etc/neutron/neutron-api-uwsgi.ini\n  ...\n  module \u003d neutron.wsgi.api:application\n\nDepends-On: https://review.opendev.org/c/openstack/neutron-tempest-plugin/+/926621\n\nCloses-Bug: #2075147\nChange-Id: If2aa37b2a510a85172da833ca20564810817d246\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/febded1e6c431c4f4271bcbc17b5c37ac024dac6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/febded1e6c431c4f4271bcbc17b5c37ac024dac6"}]},"branch":"refs/heads/master"},"ba122b7a6002ffd2c67f8b49d6e339a68ba4e4f0":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2024-08-24 11:38:25.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/76/925376/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/76/925376/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/76/925376/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/76/925376/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/76/925376/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/76/925376/7"}}},"commit":{"parents":[{"commit":"f19bbae345eeadd28bb54d28e7cea91315b1a49e","subject":"Merge \"docs: Clarify fullstack local run procedure\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f19bbae345eeadd28bb54d28e7cea91315b1a49e"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-07-31 10:53:14.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-08-24 10:06:03.000000000","tz":0},"subject":"Monkey patch the system libraries before calling them","message":"Monkey patch the system libraries before calling them\n\nThe Neutron API with WSGI module, and specifically when using ML2/OVN,\nwas importing some system libraries before patching them. That was\nleading to a recursion error, as reported in the related LP bug.\nBy calling ``eventlet_utils.monkey_patch()`` at the very beginning\nof the WSGI entry point [1], this issue is fixed.\n\n[1] WSGI entry point:\n  $ cat /etc/neutron/neutron-api-uwsgi.ini\n  ...\n  module \u003d neutron.wsgi.api:application\n\nDepends-On: https://review.opendev.org/c/openstack/neutron-tempest-plugin/+/926621\n\nCloses-Bug: #2075147\nChange-Id: If2aa37b2a510a85172da833ca20564810817d246\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ba122b7a6002ffd2c67f8b49d6e339a68ba4e4f0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ba122b7a6002ffd2c67f8b49d6e339a68ba4e4f0"}]},"branch":"refs/heads/master"},"cb9660039c0072b3b35d7a0231bbb553b903a9f4":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2024-08-26 07:13:16.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/76/925376/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/76/925376/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/76/925376/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/76/925376/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/76/925376/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/76/925376/8"}}},"commit":{"parents":[{"commit":"ed5cb6208fb7766bd57f40bc2a456450554fee44","subject":"DNM \u003d\u003d Make RPC event cast synchronous with the event","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ed5cb6208fb7766bd57f40bc2a456450554fee44"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-07-31 10:53:14.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-08-13 14:58:25.000000000","tz":0},"subject":"Monkey patch the system libraries before calling them","message":"Monkey patch the system libraries before calling them\n\nThe Neutron API with WSGI module, and specifically when using ML2/OVN,\nwas importing some system libraries before patching them. That was\nleading to a recursion error, as reported in the related LP bug.\nBy calling ``eventlet_utils.monkey_patch()`` at the very beginning\nof the WSGI entry point [1], this issue is fixed.\n\n[1] WSGI entry point:\n  $ cat /etc/neutron/neutron-api-uwsgi.ini\n  ...\n  module \u003d neutron.wsgi.api:application\n\nDepends-On: https://review.opendev.org/c/openstack/neutron-tempest-plugin/+/926621\n\nCloses-Bug: #2075147\nChange-Id: If2aa37b2a510a85172da833ca20564810817d246\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/cb9660039c0072b3b35d7a0231bbb553b903a9f4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/cb9660039c0072b3b35d7a0231bbb553b903a9f4"}]},"branch":"refs/heads/master"},"fef91a09a4dd7d4020c74de05f73a72ef9814393":{"kind":"NO_CODE_CHANGE","_number":9,"created":"2024-08-26 07:42:50.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/76/925376/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/76/925376/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/76/925376/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/76/925376/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/76/925376/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/76/925376/9"}}},"commit":{"parents":[{"commit":"ed5cb6208fb7766bd57f40bc2a456450554fee44","subject":"DNM \u003d\u003d Make RPC event cast synchronous with the event","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ed5cb6208fb7766bd57f40bc2a456450554fee44"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-07-31 10:53:14.000000000","tz":0},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2024-08-26 07:42:50.000000000","tz":0},"subject":"Monkey patch the system libraries before calling them","message":"Monkey patch the system libraries before calling them\n\nThe Neutron API with WSGI module, and specifically when using ML2/OVN,\nwas importing some system libraries before patching them. That was\nleading to a recursion error, as reported in the related LP bug.\nBy calling ``eventlet_utils.monkey_patch()`` at the very beginning\nof the WSGI entry point [1], this issue is fixed.\n\n[1] WSGI entry point:\n  $ cat /etc/neutron/neutron-api-uwsgi.ini\n  ...\n  module \u003d neutron.wsgi.api:application\n\nCloses-Bug: #2075147\nChange-Id: If2aa37b2a510a85172da833ca20564810817d246\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/fef91a09a4dd7d4020c74de05f73a72ef9814393"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/fef91a09a4dd7d4020c74de05f73a72ef9814393"}]},"branch":"refs/heads/master","description":"Edit commit message"},"9a425f2328c29477dfbf102e9172d6e5bb8fdf79":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2024-08-26 10:22:12.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/76/925376/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/76/925376/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/76/925376/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/76/925376/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/76/925376/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/76/925376/10"}}},"commit":{"parents":[{"commit":"43997ef28ab270b37259adb1ea1a24a385deca1d","subject":"Make RPC event cast synchronous with the event","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/43997ef28ab270b37259adb1ea1a24a385deca1d"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-07-31 10:53:14.000000000","tz":0},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2024-08-26 10:22:12.000000000","tz":0},"subject":"Monkey patch the system libraries before calling them","message":"Monkey patch the system libraries before calling them\n\nThe Neutron API with WSGI module, and specifically when using ML2/OVN,\nwas importing some system libraries before patching them. That was\nleading to a recursion error, as reported in the related LP bug.\nBy calling ``eventlet_utils.monkey_patch()`` at the very beginning\nof the WSGI entry point [1], this issue is fixed.\n\n[1] WSGI entry point:\n  $ cat /etc/neutron/neutron-api-uwsgi.ini\n  ...\n  module \u003d neutron.wsgi.api:application\n\nCloses-Bug: #2075147\nChange-Id: If2aa37b2a510a85172da833ca20564810817d246\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9a425f2328c29477dfbf102e9172d6e5bb8fdf79"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9a425f2328c29477dfbf102e9172d6e5bb8fdf79"}]},"branch":"refs/heads/master","description":"Rebase"},"293ca5f228f71685560d9e85b20e791132534118":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2024-08-26 12:41:17.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/76/925376/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/76/925376/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/76/925376/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/76/925376/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/76/925376/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/76/925376/11"}}},"commit":{"parents":[{"commit":"5b7faa1897f330879673261bbed0e819139709b2","subject":"Make RPC event cast synchronous with the event","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5b7faa1897f330879673261bbed0e819139709b2"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-07-31 10:53:14.000000000","tz":0},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2024-08-26 12:41:17.000000000","tz":0},"subject":"Monkey patch the system libraries before calling them","message":"Monkey patch the system libraries before calling them\n\nThe Neutron API with WSGI module, and specifically when using ML2/OVN,\nwas importing some system libraries before patching them. That was\nleading to a recursion error, as reported in the related LP bug.\nBy calling ``eventlet_utils.monkey_patch()`` at the very beginning\nof the WSGI entry point [1], this issue is fixed.\n\n[1] WSGI entry point:\n  $ cat /etc/neutron/neutron-api-uwsgi.ini\n  ...\n  module \u003d neutron.wsgi.api:application\n\nCloses-Bug: #2075147\nChange-Id: If2aa37b2a510a85172da833ca20564810817d246\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/293ca5f228f71685560d9e85b20e791132534118"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/293ca5f228f71685560d9e85b20e791132534118"}]},"branch":"refs/heads/master","description":"Rebase"},"db1cd8c5926429fc65d9fc1476d44c4880d04c2f":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2024-08-27 06:29:02.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/76/925376/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/76/925376/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/76/925376/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/76/925376/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/76/925376/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/76/925376/12"}}},"commit":{"parents":[{"commit":"70ba9cb17f6aecba94bb17a9dca77e6f06273b6e","subject":"Make RPC event cast synchronous with the event","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/70ba9cb17f6aecba94bb17a9dca77e6f06273b6e"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-07-31 10:53:14.000000000","tz":0},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2024-08-27 06:29:02.000000000","tz":0},"subject":"Monkey patch the system libraries before calling them","message":"Monkey patch the system libraries before calling them\n\nThe Neutron API with WSGI module, and specifically when using ML2/OVN,\nwas importing some system libraries before patching them. That was\nleading to a recursion error, as reported in the related LP bug.\nBy calling ``eventlet_utils.monkey_patch()`` at the very beginning\nof the WSGI entry point [1], this issue is fixed.\n\n[1] WSGI entry point:\n  $ cat /etc/neutron/neutron-api-uwsgi.ini\n  ...\n  module \u003d neutron.wsgi.api:application\n\nCloses-Bug: #2075147\nChange-Id: If2aa37b2a510a85172da833ca20564810817d246\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/db1cd8c5926429fc65d9fc1476d44c4880d04c2f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/db1cd8c5926429fc65d9fc1476d44c4880d04c2f"}]},"branch":"refs/heads/master","description":"Rebase"},"76f343c5868556f12f9ee74b7ef2291cf5e2ff85":{"kind":"TRIVIAL_REBASE","_number":13,"created":"2024-08-28 19:00:18.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"real_uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/76/925376/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/76/925376/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/76/925376/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/76/925376/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/76/925376/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/76/925376/13"}}},"commit":{"parents":[{"commit":"f1585775d92a2622dca006d23c8396036e8742e9","subject":"Merge \"Make RPC event cast synchronous with the event\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f1585775d92a2622dca006d23c8396036e8742e9"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-07-31 10:53:14.000000000","tz":0},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2024-08-28 19:00:18.000000000","tz":0},"subject":"Monkey patch the system libraries before calling them","message":"Monkey patch the system libraries before calling them\n\nThe Neutron API with WSGI module, and specifically when using ML2/OVN,\nwas importing some system libraries before patching them. That was\nleading to a recursion error, as reported in the related LP bug.\nBy calling ``eventlet_utils.monkey_patch()`` at the very beginning\nof the WSGI entry point [1], this issue is fixed.\n\n[1] WSGI entry point:\n  $ cat /etc/neutron/neutron-api-uwsgi.ini\n  ...\n  module \u003d neutron.wsgi.api:application\n\nCloses-Bug: #2075147\nChange-Id: If2aa37b2a510a85172da833ca20564810817d246\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/76f343c5868556f12f9ee74b7ef2291cf5e2ff85"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/76f343c5868556f12f9ee74b7ef2291cf5e2ff85"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
