)]}'
{"id":"openstack%2Fneutron~925402","triplet_id":"openstack%2Fneutron~master~I3203678c1ca6fb778c993b6084bab171a312ec28","project":"openstack/neutron","branch":"master","topic":"I56e791847c8f4f3a07f543689bf22fde8160c9b7","attention_set":{},"removed_from_attention_set":{"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2024-09-06 16:56:28.000000000","reason":"Change was submitted"},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2024-09-06 15:01:21.000000000","reason":"\u003cGERRIT_ACCOUNT_1131\u003e replied on the change","reason_account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}}},"hashtags":[],"change_id":"I3203678c1ca6fb778c993b6084bab171a312ec28","subject":"[OVN] Set reside-on-chassis-redirect also for FLAT networks","status":"MERGED","created":"2024-07-31 14:22:37.000000000","updated":"2024-09-06 16:57:37.000000000","submitted":"2024-09-06 16:56:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"925402-I56e791847c8f4f3a07f543689bf22fde8160c9b7","meta_rev_id":"8bf6378ef7425479fb1bda3008185354ee79d2d6","_number":925402,"virtual_id_number":925402,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2024-09-05 10:57:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-09-06 16:56:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2024-09-06 08:25:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":2,"date":"2024-09-06 15:01:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":1,"date":"2024-09-06 15:01:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-07-31 14:26:51.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"CC"},{"updated":"2024-07-31 14:57:49.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-07-31 15:42:13.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2024-07-31 16:16:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-09-06 08:25:20.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"}],"messages":[{"id":"2222380b4929373000c16de3a36815d851cb663a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2024-07-31 14:22:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2d2d9dfa691cfed63133d59d560e41b333c40736","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-07-31 14:26:51.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1a46cc70dd37c2ed49fb3e51c46c2dd3cba62c69","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2024-07-31 14:31:41.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"122b480747a69970e0fecc802be6e4f5e852c439","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2024-07-31 14:57:49.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/July-2024/Arista_Tempest_Testrefs/changes/02/925402/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"d265c4659f93c8b3672f35606eefdaf692770464","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-07-31 15:42:13.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0143aa8cb01141934cd59e6d75e88a131c8f8678","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2024-07-31 16:07:55.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/July-2024/Arista_Tempest_Testrefs/changes/02/925402/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"5de4c999f20e957aa85888700b93bb89425e68b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-31 16:16:06.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4571866aa4cb4d7c8a51b6c14e969ff3\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/80a10798597e4e7a8e22e03274c13bd7 : SUCCESS in 1h 41m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/64212fd3b7044a0794329bf84b71084f : SUCCESS in 38m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b19955d41bd44fca017a353f91aef5b : SUCCESS in 9m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5c5d81314212464488d02459bd3292bf : SUCCESS in 26m 52s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/fb4aa0742e7a44a1a7a48d1d1a862f70 : SUCCESS in 18m 37s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3583eb1d583840f092eb48c0c23aa3f5 : SUCCESS in 21m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a173073c28548e7b522bb7b48d96c9d : SUCCESS in 10m 14s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/e0db6d10804a41eaa545ebd77c096723 : SUCCESS in 25m 11s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/f5e57b08eff448aa823fcb035141a11f : SUCCESS in 54m 23s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/7ede68b3885940f78075f0e3a5809ea8 : SUCCESS in 1h 08m 56s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/b5c014241f414f44b735fe72b038ec44 : SUCCESS in 1h 00m 56s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/4b854a7139eb44bea14d5103387fbbfa : SUCCESS in 54m 22s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/08520582335d4c8c89ffa6cfa9dc16db : SUCCESS in 1h 09m 55s","accounts_in_message":[],"_revision_number":2},{"id":"c2c84db1c08f852ef262d716b1e3f32cc7aef363","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-31 16:26:19.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1c05c5be084f434eb7b81d49ce1a7f73\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f349887924e2437f8f36d172ed79545f : SUCCESS in 33m 58s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/568d203b5d124befb5fe931f438e17d3 : SUCCESS in 31m 12s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"97eb9d7931c181b582bed52c9224665c0a97cf91","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2024-08-05 09:22:33.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"380848c693b91e626eb29d7c41d19ccde4a9ce90","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-08-05 22:16:51.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c9d953881d252b65ad3ce6843089acd40f01a4e3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2024-09-05 09:53:30.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"9797c92e12dc87c1b14b1bf8a365137431eaccce","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2024-09-05 09:54:06.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"51535e25390002685b0b213980a5f9263e0ecd98","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2024-09-05 09:55:16.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"a59e66bf16b6504319877a0dafecf4ef5908bdc5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-05 10:19:47.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c2e036260e6b4d5181c8e941c3097090\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/152aac4d09e64b3cbf96f0ba0cd3ef78 : SUCCESS in 24m 19s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/a9fa7c09aadd491ba42e0414afcc5d64 : SUCCESS in 19m 11s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"fd199c4c593182f1abaedd12a0a84a2f325eff9d","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2024-09-05 10:25:41.000000000","message":"Patch Set 3: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/September-2024/Arista_Tempest_Testrefs/changes/02/925402/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"d2a8f09d81edc51a3894bfbfa33781adbef3b020","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2024-09-05 10:57:55.000000000","message":"Patch Set 4: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/September-2024/Arista_Tempest_Testrefs/changes/02/925402/4 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":4},{"id":"635c28e6608fc7ae324260c30d96e70181d9a8d4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-05 12:06:38.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/99f21d16177e4a00b3e9532fe5caa5a2\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/78861d50501a4c82929dee3b012ac171 : FAILURE in 2h 09m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/79fbc088b8494f2f83b3bc1f0a6713f1 : SUCCESS in 40m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f1e39539d9b7421cb83c0c0cb3227041 : SUCCESS in 9m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9c24d9d57e214ea48bf6ba761df2b27f : SUCCESS in 22m 51s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/364279f47b684ae68d7f34347cd12d32 : SUCCESS in 22m 05s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/88e746cfc0fa4e1c8669fae6ec4d8f62 : SUCCESS in 21m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/40d05f93d2234ecbabee418ca4583ac7 : SUCCESS in 10m 44s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/c8303d0f89c8470588496d7c3ef7db18 : SUCCESS in 24m 57s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/324b5694b17748e282610b9045906c59 : SUCCESS in 52m 02s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/376d435c79084e27b5c3fb180eaec97a : SUCCESS in 1h 09m 00s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/d740a7df00ad4cb287f60120a908a6bc : SUCCESS in 1h 07m 36s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/f9bcb1b59a8a48df8e81ebbe23a079fd : SUCCESS in 1h 09m 14s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/07c4b895f7b94052a52a3a0dcf12241f : SUCCESS in 1h 20m 13s","accounts_in_message":[],"_revision_number":4},{"id":"28e6217c378edb6e23dc5793d9e3caeec21f472f","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2024-09-05 13:57:59.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"0fdbd3c701d7a2355654118312489975311f4fe4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-05 14:30:56.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7a0a29c0c6f045f489096af74f4e7fee\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/cd40adf7406f427dbecee2726ed0b79c : SUCCESS in 27m 30s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/90efbca26d0c46b9a3175d74789789c6 : SUCCESS in 30m 34s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"7d346195e27a317e00745def53841bca3dcebeb3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-05 16:27:15.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ffecfd45fbf24761b4fe916d5878cb2c\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/4db1ece7a07842ee910e3d539d4691e6 : SUCCESS in 2h 27m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/373e6e78373d488fa73db62ff7aaa662 : SUCCESS in 52m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b022098f3a954c3facb19eddd9a38771 : SUCCESS in 10m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3465af375a7848a2a3bd29cf9a33a03a : SUCCESS in 21m 25s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6278c24e0bdf4809af76a1ce311c5587 : SUCCESS in 20m 55s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ff34fd44e7f74aa697bbe34207ef4f5d : SUCCESS in 21m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e06d9bd0b47d4ddf87d8087fb5999726 : SUCCESS in 10m 49s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/12ddd17ed66d4baf994a129a5b5a8608 : SUCCESS in 21m 41s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/aa821d0d05be4da4afa7f3e7d0d41f2f : SUCCESS in 57m 15s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/5031235dd06847f6b1ef47ee14bfe6fe : SUCCESS in 1h 07m 34s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/fc56f3a5b3f34a8b869fcffd648cfef0 : SUCCESS in 1h 02m 11s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/7a1e77379d4049d1b92b45cbb5b467d6 : SUCCESS in 1h 07m 41s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/f64d31af45df4650a83bb58ec73a2127 : SUCCESS in 1h 17m 23s","accounts_in_message":[],"_revision_number":4},{"id":"cd36c14a9d88e776c666bdc53da354bca698b916","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-09-06 08:25:20.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"a4bf90d62d4b8458b42cf8fbcb8dd30102bfb143","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-09-06 15:01:21.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"9bbd532fa0d48c3f913bd83ad17f543e55fa4f7d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-06 15:02:04.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"c5c50f2053d97dda26562d4e140ffdd79e024d2b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-06 16:56:28.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7196e6d7cc1c4c61a0272bfac9246543\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/c454199f689b4a989f39bfc80b21860f : SUCCESS in 1h 52m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2a39c27559ea4887a2dcfacace9593c7 : SUCCESS in 9m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0deb2a8282824b428eedec53a17b640d : SUCCESS in 19m 52s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/399b96f3037a4eb0b97bf5937bcba649 : SUCCESS in 22m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8bd7995bd48043a0bffd792801467c6d : SUCCESS in 7m 50s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/992d1fff0f8441fba0167bf89bec64b4 : SUCCESS in 1h 01m 56s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/690db62af5b441cba52d078c9740775f : SUCCESS in 1h 04m 49s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/a6decfe59179452594312d33fd9db792 : SUCCESS in 55m 16s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/7249f9dac48e45cdbab003db55475a90 : SUCCESS in 1h 17m 43s","accounts_in_message":[],"_revision_number":4},{"id":"8015f95848da4f716c01cfe3dfe9dc782cdf8a19","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-06 16:56:28.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"8bf6378ef7425479fb1bda3008185354ee79d2d6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-06 16:57:37.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/98ab19cbd55a44c38275625d057c9b17\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/46796233eb9f4b7cb7baada3077bf4b0 : SUCCESS in 50s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"9fd1f58394efd083cfe9a194601e30b827af56a5","revisions":{"0f591f450c3c4f0ccbb5265dd17edd432bb199ce":{"kind":"REWORK","_number":1,"created":"2024-07-31 14:22:37.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/02/925402/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/02/925402/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/02/925402/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/02/925402/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/02/925402/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/02/925402/1"}}},"commit":{"parents":[{"commit":"8134715cfd143c4c28ccc7cb8d51ef19f5a1e814","subject":"Fix setting correct \u0027reside-on-chassis-redirect\u0027 in the maintenance task","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8134715cfd143c4c28ccc7cb8d51ef19f5a1e814"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2024-07-31 12:58:18.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2024-07-31 14:22:23.000000000","tz":120},"subject":"[OVN] Set reside-on-chassis-redirect also for FLAT networks","message":"[OVN] Set reside-on-chassis-redirect also for FLAT networks\n\nThis ovn option for the Logical Router Port was added to be set with [1]\nbut FLAT networks are basically working in the same way and should have\nthis option set in the same way.\n\n[1] https://review.opendev.org/c/openstack/neutron/+/871252\n\nChange-Id: I3203678c1ca6fb778c993b6084bab171a312ec28\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0f591f450c3c4f0ccbb5265dd17edd432bb199ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0f591f450c3c4f0ccbb5265dd17edd432bb199ce"}]},"branch":"refs/heads/master"},"7f5db0099c6698f729313b916e5cf22af61cd644":{"kind":"NO_CHANGE","_number":2,"created":"2024-07-31 14:31:41.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/02/925402/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/02/925402/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/02/925402/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/02/925402/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/02/925402/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/02/925402/2"}}},"commit":{"parents":[{"commit":"4b1bfb93e380b8dce78935395b2cda57076e5476","subject":"Fix setting correct \u0027reside-on-chassis-redirect\u0027 in the maintenance task","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4b1bfb93e380b8dce78935395b2cda57076e5476"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2024-07-31 12:58:18.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2024-07-31 14:31:41.000000000","tz":0},"subject":"[OVN] Set reside-on-chassis-redirect also for FLAT networks","message":"[OVN] Set reside-on-chassis-redirect also for FLAT networks\n\nThis ovn option for the Logical Router Port was added to be set with [1]\nbut FLAT networks are basically working in the same way and should have\nthis option set in the same way.\n\n[1] https://review.opendev.org/c/openstack/neutron/+/871252\n\nChange-Id: I3203678c1ca6fb778c993b6084bab171a312ec28\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7f5db0099c6698f729313b916e5cf22af61cd644"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7f5db0099c6698f729313b916e5cf22af61cd644"}]},"branch":"refs/heads/master","description":"Rebase"},"b6281e8b6d895774df18d2951ed23479a7af4116":{"kind":"REWORK","_number":3,"created":"2024-09-05 09:53:30.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/02/925402/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/02/925402/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/02/925402/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/02/925402/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/02/925402/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/02/925402/3"}}},"commit":{"parents":[{"commit":"f8fce1cb05886c146a278c0e1ebfedc02f586255","subject":"Merge \"Remove default override for RBAC config options\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f8fce1cb05886c146a278c0e1ebfedc02f586255"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2024-07-31 12:58:18.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2024-09-05 09:52:53.000000000","tz":120},"subject":"[OVN] Set reside-on-chassis-redirect also for FLAT networks","message":"[OVN] Set reside-on-chassis-redirect also for FLAT networks\n\nThis ovn option for the Logical Router Port was added to be set with [1]\nbut FLAT networks are basically working in the same way and should have\nthis option set in the same way.\n\n[1] https://review.opendev.org/c/openstack/neutron/+/871252\n\nChange-Id: I3203678c1ca6fb778c993b6084bab171a312ec28\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b6281e8b6d895774df18d2951ed23479a7af4116"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b6281e8b6d895774df18d2951ed23479a7af4116"}]},"branch":"refs/heads/master"},"9fd1f58394efd083cfe9a194601e30b827af56a5":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2024-09-05 09:54:06.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/02/925402/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/02/925402/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/02/925402/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/02/925402/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/02/925402/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/02/925402/4"}}},"commit":{"parents":[{"commit":"babf19aeec4aeec8053114e6000f8fe53c8b3d2c","subject":"Merge \"nit: Fix variable name used in a comment\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/babf19aeec4aeec8053114e6000f8fe53c8b3d2c"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2024-07-31 12:58:18.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2024-09-05 09:53:52.000000000","tz":120},"subject":"[OVN] Set reside-on-chassis-redirect also for FLAT networks","message":"[OVN] Set reside-on-chassis-redirect also for FLAT networks\n\nThis ovn option for the Logical Router Port was added to be set with [1]\nbut FLAT networks are basically working in the same way and should have\nthis option set in the same way.\n\n[1] https://review.opendev.org/c/openstack/neutron/+/871252\n\nChange-Id: I3203678c1ca6fb778c993b6084bab171a312ec28\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9fd1f58394efd083cfe9a194601e30b827af56a5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9fd1f58394efd083cfe9a194601e30b827af56a5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
