)]}'
{"id":"openstack%2Fneutron~925452","triplet_id":"openstack%2Fneutron~stable%2F2024.1~I3bba89ac14414c6b7d375072ae92d2e0b5497736","project":"openstack/neutron","branch":"stable/2024.1","topic":"fix-regex-stable/2024.1","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2024-08-05 12:20:41.000000000","reason":"Change was submitted"},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2024-08-05 12:20:41.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I3bba89ac14414c6b7d375072ae92d2e0b5497736","subject":"Fix regex lines in zuul.d/* files","status":"MERGED","created":"2024-08-01 09:41:22.000000000","updated":"2024-08-05 12:22:50.000000000","submitted":"2024-08-05 12:20:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"925452-fix-regex-stable/2024.1","meta_rev_id":"1a6ccf1b18f5194d5fd615e37298e8591e6b4833","_number":925452,"virtual_id_number":925452,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-08-05 12:20:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2024-08-02 06:58:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2024-08-01 10:15:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":2,"date":"2024-08-05 10:08:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":1,"date":"2024-08-05 10:08:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-08-01 09:41:22.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2024-08-01 10:15:41.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2024-08-01 10:42:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-08-01 11:25:32.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2024-08-01 11:45:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-08-02 06:58:32.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2024-08-05 10:08:43.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"}],"messages":[{"id":"cf5a497bbd0270ebbcb8a2594f3e2c1e72b37fe1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2024-08-01 09:41:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c3650c51ba406ecccb6e9847a28645edb4b98a1a","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2024-08-01 10:15:41.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"347fe23462766e43e7712b72eb220d028f5768ee","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-01 10:42:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bf05ab497f56407da75e1f2df7f94775\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3db9bd17923f4a04acc4d110d94b0dfa : SUCCESS in 31m 06s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/ac5130b818124f90bfbfec1043c3b08e : SUCCESS in 29m 26s\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/5b6f3c7cf95e439c8b7a8d506c82db10 : SUCCESS in 38m 16s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0f8a241bb1861baddd2c97da0280f12451c751eb","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-08-01 11:25:32.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e74e02810ce3f907a3e4e11132ba20897e81201e","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2024-08-01 11:39:23.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5e017b06e50a883c128d774610c44dbbc60a732b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-01 11:45:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e7b83cd1f3bd4aefa8eb5cc6963fda51\n\n- neutron-tempest-plugin-openvswitch-2024-1 https://zuul.opendev.org/t/openstack/build/7760dc00f1cb4c0c93f0eef4b6704409 : SUCCESS in 1h 08m 09s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-1 https://zuul.opendev.org/t/openstack/build/a8a6f043d3834240bb040a482073ebd8 : SUCCESS in 2h 01m 29s\n- neutron-tempest-plugin-ovn-2024-1 https://zuul.opendev.org/t/openstack/build/14e4c2cc035148339ba97e4a7d04aa64 : SUCCESS in 1h 24m 32s\n- neutron-tempest-plugin-designate-scenario-2024-1 https://zuul.opendev.org/t/openstack/build/46be00ab38c44d55b198753a9ef3864d : SUCCESS in 39m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/604aa61f6ac14bfda8dd82e8d1c854c8 : SUCCESS in 47m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/31b409d61059419483c288641e69e5d3 : SUCCESS in 7m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/43e6438f40404c6c9164d574afd94a9f : SUCCESS in 25m 29s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/22c11fbedcc24ad2886dccc13553e705 : SUCCESS in 24m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/acd67ffe8fd545b1978cf1bf5aa075bf : SUCCESS in 9m 46s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/09a20fc1965446078cc7a1e6126a15c1 : SUCCESS in 41m 47s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/738a3960a8554f8f990dfcac197d16e6 : SUCCESS in 1h 12m 56s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/ec065729c934466abd06e9534df5aafb : SUCCESS in 55m 59s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/83b391b72c9b410ab2a0a0b693a754dd : SUCCESS in 1h 44m 22s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/c5da1fa56053425a9afca83c0d30b104 : SUCCESS in 44m 05s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/df1985afad524fb5b47e7ff182a67adb : SUCCESS in 1h 02m 30s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/40351b9b4242419a915553b4b3a3be2e : SUCCESS in 1h 31m 11s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/1c1bb9cce00f41c99226880914d57b13 : SUCCESS in 1h 29m 34s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/9320c35f7b694f7183c67ded3bbed917 : SUCCESS in 59m 32s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/5fe1b4f64ae844dba5b716d883c802c8 : SUCCESS in 1h 23m 01s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/5b8e59289c844eaba088aaef2cab46d7 : SUCCESS in 1h 46m 22s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/e4e9cd7fdfba48c988112deb4d1a2a51 : SUCCESS in 1h 15m 01s","accounts_in_message":[],"_revision_number":1},{"id":"38fc42906c15e1a2857811187462f791f4fdc2fb","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-08-02 06:58:32.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"1f245e14b0917986e3782ef2d369257a74471465","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-08-05 10:08:43.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e3b748980155dafdd810e3e7eba7eba56c71fe75","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-05 10:09:21.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b222b805fa7ec0273b45c239ec821176b4e059ae","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-05 12:20:40.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b6ab503744754616b1e5128d750950cd\n\n- neutron-tempest-plugin-ovn-2024-1 https://zuul.opendev.org/t/openstack/build/8edceb10675a4e449c59a46282ef6e17 : SUCCESS in 2h 10m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/badd3e0eda2843bab65a357f3eeccf68 : SUCCESS in 8m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5f48351cc22e457bae19bd6576ef04e4 : SUCCESS in 28m 11s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c00e5830d6284974873828b8f35efa9e : SUCCESS in 19m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/428cbbfc64ac41b296ad7e6a4909a3c0 : SUCCESS in 9m 23s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/03a9e0996adc480d8ca420cf87cfd7a8 : SUCCESS in 49m 52s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/5b88f57409a34b279799a5faaf477253 : SUCCESS in 2h 02m 13s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/26caadad1fd6413ba4dbfc0539ee065e : SUCCESS in 1h 35m 11s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/6cb797ef9a0f4d798c03dd0da04e22c2 : SUCCESS in 1h 01m 46s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/034dbbd30e8c43bfb7ce6d450ed41cdc : SUCCESS in 58m 23s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/10c5de20fdf14da8a5d32e435c843a1f : SUCCESS in 47m 14s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/6e661a6b069e4ae9b0de0289eddc2287 : SUCCESS in 1h 10m 02s","accounts_in_message":[],"_revision_number":1},{"id":"a74c818c4cdb69b8804b65f354d9ec1fa50c3a82","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-05 12:20:41.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"1a6ccf1b18f5194d5fd615e37298e8591e6b4833","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-05 12:22:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5ae51e0c8e174f20939e0b5ce6d32f3a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/61c892ef39744816b37982bfd4ae2589 : SUCCESS in 54s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0afdfb0ad5a4b2ede06624b8732a0b14965475df","revisions":{"0afdfb0ad5a4b2ede06624b8732a0b14965475df":{"kind":"REWORK","_number":1,"created":"2024-08-01 09:41:22.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/52/925452/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/52/925452/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/52/925452/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/52/925452/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/52/925452/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/52/925452/1"}}},"commit":{"parents":[{"commit":"59bda697ad9f6830575c70e06e833b31aba1335f","subject":"Use has_lock_periodic decorator for the tasks which require ovn db lock","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/59bda697ad9f6830575c70e06e833b31aba1335f"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2024-06-04 16:23:27.000000000","tz":-240},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2024-08-01 09:40:18.000000000","tz":120},"subject":"Fix regex lines in zuul.d/* files","message":"Fix regex lines in zuul.d/* files\n\nCommit 260c968118934 broke the gate by causing jobs\nto not get run when it added RE2 compatibility for\nirrelevant-files. Digging found that RE2 doesn\u0027t\nsupport negative lookahead (and won\u0027t ever), so it\u0027s\nimpossible to replace the previous pcre filter with a\nsimilar RE2 filter.\n\nInstead of reverting to the original filter, which\nis considered obsolete by zuul, this patch fixes the\nissue by explicitly listing all files under zuul.d/\nexcept the one that we actually want to trigger the\njobs: zuul.d/project.yaml.\n\nListing all the files in the directory for every job\nis not ideal, and we may revisit it later, or perhaps\neven reconsider the extensive use of irrelevant-files\nin the neutron tree. This will have to wait for when\nthe gate is in better shape though.\n\n[0] https://github.com/google/re2/issues/156\n\nConflicts:\n    zuul.d/base.yaml\n    zuul.d/grenade.yaml\n    zuul.d/job-templates.yaml\n    zuul.d/project.yaml\n    zuul.d/rally.yaml\n    zuul.d/tempest-multinode.yaml\n    zuul.d/tempest-singlenode.yaml\n\nRelated-bug: #2065821\nChange-Id: I3bba89ac14414c6b7d375072ae92d2e0b5497736\n(cherry picked from commit 11027e3e1ef9a58d5b2faa575a3764bd33cd2a08)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0afdfb0ad5a4b2ede06624b8732a0b14965475df"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0afdfb0ad5a4b2ede06624b8732a0b14965475df"}]},"branch":"refs/heads/stable/2024.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
