)]}'
{"id":"openstack%2Fneutron~929255","triplet_id":"openstack%2Fneutron~stable%2F2024.2~Ideca1f7b96eddf051fedbd880ed7e9ead7f1d587","project":"openstack/neutron","branch":"stable/2024.2","topic":"create-2024.2","hashtags":[],"change_id":"Ideca1f7b96eddf051fedbd880ed7e9ead7f1d587","subject":"Update .gitreview for stable/2024.2","status":"MERGED","created":"2024-09-13 11:13:44.000000000","updated":"2024-09-13 23:17:08.000000000","submitted":"2024-09-13 23:16:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"929255-create-2024.2","meta_rev_id":"8bfb9539fa834e79c31c6dee6df534a5a236e7c3","_number":929255,"virtual_id_number":929255,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-09-13 23:16:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":2,"date":"2024-09-13 14:56:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":1,"date":"2024-09-13 22:49:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-09-13 11:50:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-09-13 12:06:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-09-13 14:56:08.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"}],"messages":[{"id":"81ee6f97a32cf672a1bde59a57ce461e716ebb04","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2024-09-13 11:13:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a69f130d4fb3897a975504cb737e08db7acdaeb9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-13 11:50:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d58ecc0fb0f149199ce782156aef27d3\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c7890f3acae74071b09a21f4f71bc235 : SUCCESS in 24m 31s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/9e7529e736464ade85fff1c9aa66d041 : SUCCESS in 28m 22s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4075b1520db4920736c81fd00dfd4bbb5dd7d7d0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-13 12:06:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/092ec61b0a7b479cbb422f26b65db571\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1b99aaf655924941a4b7a54f53d9ca2d : SUCCESS in 40m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c8b357a745214bbea4c637a534e473c0 : SUCCESS in 10m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8da6c0f90859460b923beded8d33afb1 : SUCCESS in 20m 35s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6a378937add2466e9c25ceeedb71a601 : SUCCESS in 20m 42s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ecac92bea320445a88859915125c8455 : SUCCESS in 19m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2723f6d92bbb4b9aacdf7d987867dd69 : SUCCESS in 9m 30s","accounts_in_message":[],"_revision_number":1},{"id":"c01c2756e80fe7d857a7714b5684a02ce0f505c1","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-09-13 14:56:08.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"230471f3360841d5c927690c942827df6a3b9b89","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-09-13 22:49:11.000000000","message":"Patch Set 1: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ee256ea527b6d1f1ef77159857390d853d5af7a2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-13 22:49:32.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"cef9829d413cfe3f51669e46c44de0ea4347d601","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-13 23:16:06.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b27f96ca17684d48adc001309fbfcc8c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6ff82e77e0ef4f2980a233feb4d8500e : SUCCESS in 11m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/61b5290e4d5546678216f292b00cbd57 : SUCCESS in 18m 43s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e272cdf69c584d6e890b6732dbb5fa17 : SUCCESS in 20m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f32d960ee774024b813729173a8d706 : SUCCESS in 11m 05s","accounts_in_message":[],"_revision_number":1},{"id":"dfd1f600ea79f9b5e9b4363b5ab9176077908370","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-13 23:16:06.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"8bfb9539fa834e79c31c6dee6df534a5a236e7c3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-13 23:17:08.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/61ceb6244f054446a4a76beeae520123\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b3b386c05f1442da96d1b68a8f86b61 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"74c155b2cd3f9213f324ed016701cbf89a0ea682","revisions":{"74c155b2cd3f9213f324ed016701cbf89a0ea682":{"kind":"REWORK","_number":1,"created":"2024-09-13 11:13:44.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/55/929255/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/55/929255/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/929255/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/929255/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/929255/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/55/929255/1"}}},"commit":{"parents":[{"commit":"dd3853007bb36999d391a823505fbe0c279d2317","subject":"Merge \"Enable iptables debugging in the L3 agent functional tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/dd3853007bb36999d391a823505fbe0c279d2317"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2024-09-13 11:13:42.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2024-09-13 11:13:42.000000000","tz":0},"subject":"Update .gitreview for stable/2024.2","message":"Update .gitreview for stable/2024.2\n\nChange-Id: Ideca1f7b96eddf051fedbd880ed7e9ead7f1d587\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/74c155b2cd3f9213f324ed016701cbf89a0ea682"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/74c155b2cd3f9213f324ed016701cbf89a0ea682"}]},"branch":"refs/heads/stable/2024.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
