)]}'
{"id":"openstack%2Fneutron~930004","triplet_id":"openstack%2Fneutron~unmaintained%2Fwallaby~I6d1979dfb4d6f455ca419e64248087047fbf73d7","project":"openstack/neutron","branch":"unmaintained/wallaby","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2025-12-19 14:27:20.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I6d1979dfb4d6f455ca419e64248087047fbf73d7","subject":"Correct logic error when associating FIP with OVN LB","status":"ABANDONED","created":"2024-09-20 07:34:06.000000000","updated":"2025-12-19 14:27:20.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"cherry_pick_of_change":921663,"cherry_pick_of_patch_set":2,"meta_rev_id":"4fa61f18fd757a234425b202dd878a00f84bedb6","_number":930004,"virtual_id_number":930004,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-09-20 08:46:56.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2024-09-20 08:34:40.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-09-20 07:34:06.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"CC"},{"updated":"2024-09-20 08:34:40.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2024-09-20 08:46:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-12-19 14:27:20.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"CC"}],"messages":[{"id":"70ee646060aee73af0490c45c4350c697f6086fa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-09-20 07:34:06.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"bd9944b659df6082ab769207ca6352a438a3f985","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-09-20 08:34:40.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"146452c7a85ecce8b70869ddd4e7394c6b65f6c5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-20 08:46:56.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7b93a86b25674f93acc7b64bcc2d986d\n\n- neutron-tempest-plugin-api-wallaby https://zuul.opendev.org/t/openstack/build/1c87e4df83e748009dcd35abd3659b76 : SUCCESS in 49m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2c1bfdfebd1c4871b2d77f2a8221b638 : SUCCESS in 30m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/31d260500d3c41e386f893116f2d203f : SUCCESS in 14m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1db76c96f58846a098dcc6fc800060e0 : SUCCESS in 25m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3072089ba99f40fea91207c065d2e5ef : SUCCESS in 29m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7f2b529b1eb9406b9a61d4dd48107629 : SUCCESS in 26m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d794c66e4c074d4a96dd0a76f568f94d : SUCCESS in 10m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/512e971111ec4ce4852711b5e66c7129 : SUCCESS in 22m 57s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/abf6414c09ac45f09ae08bb5601a2d65 : SUCCESS in 1h 10m 54s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/8b60f94aad8341168e47536e0abc6938 : SUCCESS in 1h 03m 28s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/5ff0109ec695472d8033687929cf3952 : SUCCESS in 43m 11s\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/7a1264721662434a88d02bbc76a54e90 : FAILURE in 25m 50s","accounts_in_message":[],"_revision_number":1},{"id":"225b1c6fdb4e0ea2faf3d3491b418e79953b9fb0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-20 09:59:29.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/66fc7ba2f5764a68ad8e8d4e5ff200c2\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e74476cb08a04ca5b10e7c7ef3b21cc1 : SUCCESS in 40m 10s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ad94997078fc4423ba1ef77f8eb27a84 : SUCCESS in 31m 24s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4fa61f18fd757a234425b202dd878a00f84bedb6","tag":"autogenerated:gerrit:abandon","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-12-19 14:27:20.000000000","message":"Abandoned\n\nunmaintained/wallaby branch of openstack/neutron transitioned to End of Life and is about to be deleted. To be able to do that, all open patches need to be abandoned.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"158238966758fb4d10dd8aef81cfe666a641b42f","revisions":{"158238966758fb4d10dd8aef81cfe666a641b42f":{"kind":"REWORK","_number":1,"created":"2024-09-20 07:34:06.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/04/930004/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/04/930004/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/930004/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/930004/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/930004/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/04/930004/1"}}},"commit":{"parents":[{"commit":"48dbb21d3cb9bc4d67788dde0b87c39d2775be87","subject":"Fix regex lines in zuul.d/* files","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/48dbb21d3cb9bc4d67788dde0b87c39d2775be87"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-06-10 12:44:14.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-09-20 07:34:06.000000000","tz":0},"subject":"Correct logic error when associating FIP with OVN LB","message":"Correct logic error when associating FIP with OVN LB\n\nFixes a logic error which meant that we didn\u0027t iterate over all logical\nswitches when associating a FIP to an OVN loadbalancer. The symptom was\nthat the FIP would show in neutron, but would not exist in OVN.\n\nCloses-Bug: #2068644\nChange-Id: I6d1979dfb4d6f455ca419e64248087047fbf73d7\nCo-Authored-By: Brian Haley \u003chaleyb.dev@gmail.com\u003e\n(cherry picked from commit d8a4ad9167afd824a3f823d86a8fd33fb67c4abd)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/158238966758fb4d10dd8aef81cfe666a641b42f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/158238966758fb4d10dd8aef81cfe666a641b42f"}]},"branch":"refs/heads/unmaintained/wallaby"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
