)]}'
{"id":"openstack%2Fneutron~930209","triplet_id":"openstack%2Fneutron~unmaintained%2Fzed~Id3defceaeea674b735cd997131f42f2ab772fa1b","project":"openstack/neutron","branch":"unmaintained/zed","topic":"unmaintained-jobs-periodic-weekly","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2024-09-30 10:21:05.000000000","reason":"Change was submitted"},"21798":{"account":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"last_update":"2024-09-30 10:21:05.000000000","reason":"Change was submitted"},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2024-09-27 07:27:04.000000000","reason":"\u003cGERRIT_ACCOUNT_8313\u003e replied on the change","reason_account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}}},"hashtags":[],"change_id":"Id3defceaeea674b735cd997131f42f2ab772fa1b","subject":"[unmaintained-only] Periodic jobs will be weekly only","status":"MERGED","created":"2024-09-23 15:01:01.000000000","updated":"2024-09-30 10:23:40.000000000","submitted":"2024-09-30 10:21:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"930209-unmaintained-jobs-periodic-weekly","meta_rev_id":"8510c81f1698a71ccf8d65e58be972f56b43b026","_number":930209,"virtual_id_number":930209,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-09-30 10:21:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"recommended":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"all":[{"value":1,"date":"2024-09-27 05:55:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":2,"date":"2024-09-27 07:35:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":1,"date":"2024-09-26 15:58:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"value":2,"date":"2024-09-27 07:27:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":1,"date":"2024-09-30 08:25:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-09-23 15:56:11.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2024-09-23 16:24:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-09-23 16:56:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-09-24 11:50:55.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2024-09-26 15:58:21.000000000","updated_by":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"reviewer":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"state":"REVIEWER"},{"updated":"2024-09-27 05:55:21.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2024-09-27 07:35:24.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"8cb91db79ab2a7c77cf08164db3b80904fc6e241","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-09-23 15:01:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"23bb4068f4c3596d719c23f8a4758532340ca8a1","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-09-23 15:56:11.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fb2561dcb897bc99775cb3a86ea9954e1b0ff35b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-23 16:24:53.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d7216460b4a14031afc88f90a68db490\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/efa46ed5066c422ea2de253022e07ff0 : SUCCESS in 24m 14s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f5ffbf5e432149baac1251358dcec3df : SUCCESS in 23m 45s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1e29f34094863408ce64ada485012a7e58b053da","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-23 16:56:52.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5a01adc107bd4860a57fdc34c6bd51cc\n\n- neutron-tempest-plugin-linuxbridge-zed https://zuul.opendev.org/t/openstack/build/9a49d0f88cca40b8940a9c2ce743f2dc : SUCCESS in 1h 15m 50s\n- neutron-tempest-plugin-openvswitch-zed https://zuul.opendev.org/t/openstack/build/547b9fabe0e04ede94cefd2a12b327f3 : SUCCESS in 1h 22m 44s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-zed https://zuul.opendev.org/t/openstack/build/ba1e8edc4aa84fd6859c21219bf5e27a : SUCCESS in 1h 20m 56s\n- neutron-tempest-plugin-ovn-zed https://zuul.opendev.org/t/openstack/build/725e96bf1a714ff8ac8a96310b73d2b9 : FAILURE in 1h 54m 40s\n- neutron-tempest-plugin-designate-scenario-zed https://zuul.opendev.org/t/openstack/build/b54748f9b7514cf6b2fbbe40331b9b4b : SUCCESS in 27m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d95d6ae91feb477d90640831ad676d73 : SUCCESS in 39m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/64a2c7ff714a4136b79394d4daf1c227 : SUCCESS in 12m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/073d7a5e002049adb2b972ffba9c4577 : SUCCESS in 25m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b3e819c2cdcb4528952b70d131a41a33 : SUCCESS in 23m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/41db19920ba548218e3f6741b5a9ed21 : SUCCESS in 27m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/39c35e0139d74edba1ef885cb032326b : SUCCESS in 10m 09s","accounts_in_message":[],"_revision_number":1},{"id":"6a68e2b7cd52224f3b39e2d6b32b3e599d7daf21","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2024-09-24 11:50:55.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f47fb76530fc9782c0496cf0f030a6af5ea8a366","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2024-09-24 11:51:14.000000000","message":"Patch Set 1: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"33c23cc0cd3a2609a9d7cf2bab7c234b98c955ad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-24 13:50:50.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7a068d13efa742868717e9b24c68e5d8\n\n- neutron-tempest-plugin-linuxbridge-zed https://zuul.opendev.org/t/openstack/build/6d7461c882a140aab48875ebdc6be936 : SUCCESS in 1h 17m 15s\n- neutron-tempest-plugin-openvswitch-zed https://zuul.opendev.org/t/openstack/build/b292e9b5fa6343ec9f8dabb7e69bd994 : SUCCESS in 1h 21m 21s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-zed https://zuul.opendev.org/t/openstack/build/afa30e09b4864269a498c9742acad377 : SUCCESS in 1h 13m 15s\n- neutron-tempest-plugin-ovn-zed https://zuul.opendev.org/t/openstack/build/c7e56618b8ac4f45a6378d3d7c1f439b : FAILURE in 1h 30m 13s\n- neutron-tempest-plugin-designate-scenario-zed https://zuul.opendev.org/t/openstack/build/bbba89c369984c8db330c4aa9c15705b : SUCCESS in 29m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dfe8f1db98fc48d196af9ca8c6abc0c3 : SUCCESS in 33m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b915c306bce246549e11fb5d1cf7a4f6 : SUCCESS in 13m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/971d259a834c4209b2df0d5e238499ac : SUCCESS in 24m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8c8cb7903fc34eb7a845384e616b8cc2 : SUCCESS in 27m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3ba3aaf32042463381b9c4a288f20b60 : SUCCESS in 18m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b5dcd6239e4c4c1585f767bd745bb4c9 : SUCCESS in 12m 00s","accounts_in_message":[],"_revision_number":1},{"id":"a8a6a364a0a6c9afc81d611e99375aff756b9e93","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-24 14:04:08.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e83473e817aa4893bd6997269bc7f7d3\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d7845593fc0e49b282ecd6f8eb0b42ae : SUCCESS in 26m 26s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d0974ae6bf3a48109574fc011195ff4c : SUCCESS in 23m 58s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"af0a76d33fbf0e0de18ebaf4af57f033fe8d9b40","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-09-26 09:58:58.000000000","message":"Patch Set 2: Commit message was updated.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n* Workflow+1\n","accounts_in_message":[],"_revision_number":2},{"id":"706e5fa3c47a778854bc73d08f0332800d97c811","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-26 12:07:35.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/774ca1599c504982977c7136ed4b3680\n\n- neutron-tempest-plugin-linuxbridge-zed https://zuul.opendev.org/t/openstack/build/ae1855c458fd434bb9bf65473b6c0032 : SUCCESS in 1h 35m 48s\n- neutron-tempest-plugin-openvswitch-zed https://zuul.opendev.org/t/openstack/build/93e86a1bb7b141e48bdbb5c7e647bbc6 : SUCCESS in 1h 32m 02s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-zed https://zuul.opendev.org/t/openstack/build/309d4450ca82489c9a47317211705760 : SUCCESS in 1h 29m 17s\n- neutron-tempest-plugin-ovn-zed https://zuul.opendev.org/t/openstack/build/739a64647ad146138f37652c1af8d47e : FAILURE in 1h 05m 19s\n- neutron-tempest-plugin-designate-scenario-zed https://zuul.opendev.org/t/openstack/build/dd46fb02305c4754b44a41d8a8de941c : SUCCESS in 24m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8db941a42da74f8ab256367ec25cb804 : SUCCESS in 31m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dc7e0aad6fc04ea48e8b973a84772bac : SUCCESS in 14m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7db8a502d7554caa9d9838b1dd0b671c : SUCCESS in 24m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7154fa1c1a8f4220ad2afe01d997fb20 : SUCCESS in 24m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/530fffc5a40e4a8e813c1f197bec63e5 : SUCCESS in 25m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a6eccab55fc4a1889aaf510692af0c6 : SUCCESS in 9m 42s","accounts_in_message":[],"_revision_number":2},{"id":"0b61ec9d8094aa157916817937353d1005b42d3c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-26 12:24:36.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/041bff369c07483980c6496c74a9cc6d\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e58a645c70534946bb16cd022cc4702b : SUCCESS in 33m 43s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/09b22f46c21946d0acd053922f3f899a : SUCCESS in 27m 27s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c3c7cffef1aafb9a73269f3e969eca8d5029adf2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-09-26 15:03:20.000000000","message":"Patch Set 3: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"4586990e092f353187dd0f38f64579ba6a2fc73c","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2024-09-26 15:58:21.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"865a4054d4c9e8d4aa46d65a31a6cac2e7bc1787","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-26 17:52:25.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1e88ad425570432386642d8b11045d29\n\n- neutron-tempest-plugin-linuxbridge-zed https://zuul.opendev.org/t/openstack/build/c2fa8db6947347f5ac08aee34f159338 : SUCCESS in 2h 00m 40s\n- neutron-tempest-plugin-openvswitch-zed https://zuul.opendev.org/t/openstack/build/99d2ed28e0144fc085f98ccfcbe6fdd4 : SUCCESS in 1h 29m 40s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-zed https://zuul.opendev.org/t/openstack/build/87079455826e4ca4a0b4d0ca8811f95e : SUCCESS in 1h 31m 59s\n- neutron-tempest-plugin-ovn-zed https://zuul.opendev.org/t/openstack/build/c634926cdc204c5f84103772e59c2395 : SUCCESS in 2h 02m 06s\n- neutron-tempest-plugin-designate-scenario-zed https://zuul.opendev.org/t/openstack/build/1155af4c21c54a25bfa5a5bad02e6b5d : SUCCESS in 27m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4b1b6869176943d58a72f5ad48e70db0 : SUCCESS in 34m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b21c1662fbdf4addabc94ba16c73fbf4 : SUCCESS in 9m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2a0745d3a83b4e8f90049933f792f808 : SUCCESS in 27m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1ff1d600af344d759d099ebf9a275299 : SUCCESS in 22m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/026a396a5f2645ac88bfc991b3cc58fb : SUCCESS in 24m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c5bb710c45f44111a1bff1d5515e9c2b : SUCCESS in 9m 10s","accounts_in_message":[],"_revision_number":3},{"id":"78a95260233ba9c6542f5a3b96d4beff9559867f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-26 18:14:47.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fb0d5b3759114759a8ff53086258e10c\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5022225777564fe6bde257a5da5a96b0 : SUCCESS in 34m 35s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2dc432ac049e48f786408822e97de90f : SUCCESS in 45m 10s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"8f905aa2ea842e6304081febbc138c1788c23413","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2024-09-27 05:55:21.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"759e4c8ae58f4c5483976a4bde04820a15aa53ec","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-09-27 07:27:04.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"b43f45e24ffb005420ec786ec21b1b2d9bde8420","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-09-27 07:35:24.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"926312f23bfd7211f3e7776ee8eb3ac4e119688e","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-09-30 08:25:32.000000000","message":"Patch Set 3: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"688c9441fec04532cceb436caa947b13506046af","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-30 08:26:06.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"d11734523b04476406f44b37948ec25682e98553","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-30 10:21:05.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ea16d1f5be444d158ecb2895f7455038\n\n- neutron-tempest-plugin-ovn-zed https://zuul.opendev.org/t/openstack/build/333a279056234a58b3cba2fce8c9d388 : SUCCESS in 1h 48m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6c317595298c4b948c3113c3578a3bf1 : SUCCESS in 11m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0010508c84604727be6370e28b73e616 : SUCCESS in 22m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/10e3be949b584be18be130382e37e3c2 : SUCCESS in 22m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/49f14c62d9c5404695b8066133f293bd : SUCCESS in 9m 09s","accounts_in_message":[],"_revision_number":3},{"id":"7d64961dfcec13d392d62c4fc3895123173d38b1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-30 10:21:05.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"8510c81f1698a71ccf8d65e58be972f56b43b026","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-30 10:23:40.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eff2dd7c9c1f4010a9e3513c79eaf339\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e04ffde794e449daa4ab9fc115eeaec : SUCCESS in 57s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"127ce0f509cac33790a34813fc57a043972f4c30","revisions":{"0142c89d12171902bb2c0783aef7710d8f1c6890":{"kind":"REWORK","_number":1,"created":"2024-09-23 15:01:01.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/09/930209/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/09/930209/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/09/930209/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/09/930209/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/09/930209/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/09/930209/1"}}},"commit":{"parents":[{"commit":"aa68e695e1b9cca8a45e772edab471cd28212f42","subject":"Checking pci_slot to avoid changing staus to BUILD forever","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/aa68e695e1b9cca8a45e772edab471cd28212f42"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-09-23 14:58:16.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-09-23 14:58:16.000000000","tz":0},"subject":"[unmaintained-only] Periodic jobs will be weekly only","message":"[unmaintained-only] Periodic jobs will be weekly only\n\nAs agreed in the Neutron CI meeting, the unmaintained branches periodic\njobs will be weekly only [1].\n\n[1]https://meetings.opendev.org/meetings/neutron_ci/2024/neutron_ci.2024-09-23-14.01.log.html#l-97\n\nChange-Id: Id3defceaeea674b735cd997131f42f2ab772fa1b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0142c89d12171902bb2c0783aef7710d8f1c6890"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0142c89d12171902bb2c0783aef7710d8f1c6890"}]},"branch":"refs/heads/unmaintained/zed"},"5f68981fdc3ba38670334a4d96eab04dc74e827e":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2024-09-26 09:58:58.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/09/930209/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/09/930209/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/09/930209/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/09/930209/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/09/930209/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/09/930209/2"}}},"commit":{"parents":[{"commit":"aa68e695e1b9cca8a45e772edab471cd28212f42","subject":"Checking pci_slot to avoid changing staus to BUILD forever","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/aa68e695e1b9cca8a45e772edab471cd28212f42"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-09-23 14:58:16.000000000","tz":0},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2024-09-26 09:58:58.000000000","tz":0},"subject":"[unmaintained-only] Periodic jobs will be weekly only","message":"[unmaintained-only] Periodic jobs will be weekly only\n\nAs agreed in the Neutron CI meeting, the unmaintained branches periodic\njobs will be weekly only [1].\n\n[1]https://meetings.opendev.org/meetings/neutron_ci/2024/neutron_ci.2024-09-23-14.01.log.html#l-97\n\nDepends-On: https://review.opendev.org/c/openstack/neutron-tempest-plugin/+/930548\n\nChange-Id: Id3defceaeea674b735cd997131f42f2ab772fa1b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5f68981fdc3ba38670334a4d96eab04dc74e827e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5f68981fdc3ba38670334a4d96eab04dc74e827e"}]},"branch":"refs/heads/unmaintained/zed","description":"Edit commit message"},"127ce0f509cac33790a34813fc57a043972f4c30":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2024-09-26 15:03:20.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/09/930209/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/09/930209/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/09/930209/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/09/930209/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/09/930209/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/09/930209/3"}}},"commit":{"parents":[{"commit":"aa68e695e1b9cca8a45e772edab471cd28212f42","subject":"Checking pci_slot to avoid changing staus to BUILD forever","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/aa68e695e1b9cca8a45e772edab471cd28212f42"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-09-23 14:58:16.000000000","tz":0},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2024-09-26 15:03:20.000000000","tz":0},"subject":"[unmaintained-only] Periodic jobs will be weekly only","message":"[unmaintained-only] Periodic jobs will be weekly only\n\nAs agreed in the Neutron CI meeting, the unmaintained branches periodic\njobs will be weekly only [1].\n\n[1]https://meetings.opendev.org/meetings/neutron_ci/2024/neutron_ci.2024-09-23-14.01.log.html#l-97\n\nDepends-On: https://review.opendev.org/c/openstack/neutron-tempest-plugin/+/930594\n\nChange-Id: Id3defceaeea674b735cd997131f42f2ab772fa1b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/127ce0f509cac33790a34813fc57a043972f4c30"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/127ce0f509cac33790a34813fc57a043972f4c30"}]},"branch":"refs/heads/unmaintained/zed","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
