)]}'
{"id":"openstack%2Fneutron~930211","triplet_id":"openstack%2Fneutron~unmaintained%2Fyoga~I0bc95e3705bd3114c36e1ec95701da2e2d13b13f","project":"openstack/neutron","branch":"unmaintained/yoga","topic":"unmaintained-jobs-periodic-weekly","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2024-09-30 15:36:42.000000000","reason":"Change was submitted"},"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2024-09-30 15:36:42.000000000","reason":"Change was submitted"},"13861":{"account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"last_update":"2024-09-30 15:36:42.000000000","reason":"Change was submitted"},"21798":{"account":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"last_update":"2024-09-30 13:59:59.000000000","reason":"\u003cGERRIT_ACCOUNT_21798\u003e replied on the change","reason_account":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},"9656":{"account":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"last_update":"2024-09-30 15:36:42.000000000","reason":"Change was submitted"},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2024-09-30 07:43:39.000000000","reason":"\u003cGERRIT_ACCOUNT_8313\u003e replied on the change","reason_account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}}},"hashtags":[],"change_id":"I0bc95e3705bd3114c36e1ec95701da2e2d13b13f","subject":"[unmaintained-only] Periodic jobs will be weekly only","status":"MERGED","created":"2024-09-23 15:04:52.000000000","updated":"2024-09-30 15:41:10.000000000","submitted":"2024-09-30 15:36:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"930211-unmaintained-jobs-periodic-weekly","meta_rev_id":"1c5836613e7a7689722944dcc70f1013a8f4ad6e","_number":930211,"virtual_id_number":930211,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-09-30 15:36:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":2,"date":"2024-09-30 13:59:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"value":2,"date":"2024-09-30 09:41:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":1,"date":"2024-09-30 13:59:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-09-23 15:56:21.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2024-09-23 17:01:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-09-24 11:51:45.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2024-09-26 15:58:14.000000000","updated_by":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"reviewer":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"state":"REVIEWER"},{"updated":"2024-09-27 06:14:31.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2024-09-27 07:43:53.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"CC"}],"messages":[{"id":"5ad49261ac5df5196f36830085ebce8fc108abd9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-09-23 15:04:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1dc52d099e60851a250e88be3e09a25e0d336812","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-09-23 15:56:21.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"4ee813cb8d02d40c33cf4aa9bbfb6e48ee990795","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-23 17:01:51.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/42cdd720db284c699f817ec684bf6ecc\n\n- neutron-tempest-plugin-api-yoga https://zuul.opendev.org/t/openstack/build/dd8607137b52413cb970a00f0be64f1c : FAILURE in 41m 12s\n- neutron-tempest-plugin-scenario-linuxbridge-yoga https://zuul.opendev.org/t/openstack/build/5277ac68d8a448f3847ac73e9d197f59 : SUCCESS in 1h 26m 25s\n- neutron-tempest-plugin-scenario-openvswitch-yoga https://zuul.opendev.org/t/openstack/build/1524a23d66bf4ffe857d9416d5cbaf23 : SUCCESS in 1h 18m 57s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-yoga https://zuul.opendev.org/t/openstack/build/a8a6a17215434b79b9c2ab7df10b953e : SUCCESS in 1h 03m 18s\n- neutron-tempest-plugin-scenario-ovn-yoga https://zuul.opendev.org/t/openstack/build/4327083305394c618857f42a5e01e11f : FAILURE in 1h 55m 02s\n- neutron-tempest-plugin-designate-scenario-yoga https://zuul.opendev.org/t/openstack/build/235f27e4b225460f860dd5e8a82c74b9 : SUCCESS in 28m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/15e72c9a01b840f1975e209739763c20 : SUCCESS in 35m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a0c3250368034571b7283917a55e4226 : SUCCESS in 11m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c1c14bbd57b3466e81ce98f017f97761 : SUCCESS in 31m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/db6d238ec0134bc28ed0afa23eabd984 : FAILURE in 4m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e7006713cdc4eb98cc15180fc6af8e6 : SUCCESS in 8m 43s","accounts_in_message":[],"_revision_number":1},{"id":"998570e4f80313be367cdb0a3932d2c1cbe20b02","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-23 17:06:58.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cd5a57672c8d4c9ca941a0462393e729\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8f2dd53f0ed64e4b912b07a83d86d7b3 : SUCCESS in 26m 44s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6bcdf58a65784215af35eec99570caa9 : FAILURE in 8m 22s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"aea88100dfcc926a5051614efe3256f6a8446c0e","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2024-09-24 11:51:45.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e54ce992223d456d9f0125886bfa7d8377546bca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-24 14:05:19.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/754ac3c2b996470cb48c95990499df78\n\n- neutron-tempest-plugin-api-yoga https://zuul.opendev.org/t/openstack/build/18e52813deff45d7a4e0d705e6b22edc : FAILURE in 32m 37s\n- neutron-tempest-plugin-scenario-linuxbridge-yoga https://zuul.opendev.org/t/openstack/build/f94b6836448d48028224525fd49bd7e9 : SUCCESS in 1h 35m 37s\n- neutron-tempest-plugin-scenario-openvswitch-yoga https://zuul.opendev.org/t/openstack/build/4332cd0fc5f24320a01138f8937aaaf3 : SUCCESS in 1h 07m 05s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-yoga https://zuul.opendev.org/t/openstack/build/395341c9288242e39b89e27cf91a8ba6 : SUCCESS in 1h 28m 03s\n- neutron-tempest-plugin-scenario-ovn-yoga https://zuul.opendev.org/t/openstack/build/cce7640032644f8ba8e3504fe1e838f4 : FAILURE in 1h 45m 50s\n- neutron-tempest-plugin-designate-scenario-yoga https://zuul.opendev.org/t/openstack/build/d20237476d1d41d795273998669e1678 : SUCCESS in 40m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7951b6f7a58748978260ce604cbb0c12 : SUCCESS in 34m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3d30401f6b2c41afa7c7a7d93ee807ff : SUCCESS in 14m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ad279fa16ef442bba6c471033d73c230 : SUCCESS in 26m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ae0db7c4e1d443e0a938946f9bc50828 : FAILURE in 3m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f79b02213b0d4dabbd2a5c35995c051e : SUCCESS in 9m 08s","accounts_in_message":[],"_revision_number":1},{"id":"3dc0319c03dcd0a801e355ec0fdd82c2f4d1c338","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-24 14:10:35.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cdf7251209d849e3839e070eba2b9896\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/68bbedb7c7ff4311830095628cb8bcf3 : SUCCESS in 21m 38s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f5af307ab1df44da966e71be4ca623b1 : FAILURE in 6m 50s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6bfabed74278580e849bdd258c095e87fa220bb1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-09-26 09:58:43.000000000","message":"Patch Set 2: Commit message was updated.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n* Workflow+1\n","accounts_in_message":[],"_revision_number":2},{"id":"04d766e80f61b66cab7e0ebc0b5f10683e89369c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-26 11:35:48.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1cb7b2d67a2d4de284f9476263b7490c\n\n- neutron-tempest-plugin-api-yoga https://zuul.opendev.org/t/openstack/build/a78e8c17075742ce91849adfc35fa24f : SUCCESS in 39m 08s\n- neutron-tempest-plugin-scenario-linuxbridge-yoga https://zuul.opendev.org/t/openstack/build/fb1503f5f8314a9f9758b3747f4b7294 : SUCCESS in 1h 14m 39s\n- neutron-tempest-plugin-scenario-openvswitch-yoga https://zuul.opendev.org/t/openstack/build/e82f5ba70c82425e9a7c6abbe9b1d95c : SUCCESS in 1h 11m 25s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-yoga https://zuul.opendev.org/t/openstack/build/1fb24bc15a194fae9380e50dc221b176 : SUCCESS in 1h 35m 14s\n- neutron-tempest-plugin-scenario-ovn-yoga https://zuul.opendev.org/t/openstack/build/e9534f8759c24bb5bfe8ac452d5cb931 : FAILURE in 1h 28m 42s\n- neutron-tempest-plugin-designate-scenario-yoga https://zuul.opendev.org/t/openstack/build/e190598324024bf4bb0e65b821ae9aba : SUCCESS in 42m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e83ec7ad26644591a4399535824fbc1d : SUCCESS in 33m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1686b17458ae421184d9f9f71a41d59c : SUCCESS in 14m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7cba51854a4b4bb3a8c231e7daa91319 : SUCCESS in 28m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/925b06f7767146198ee36ec537a88369 : FAILURE in 2m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2b0f1750df7f4c838d55c05f4e46b6cc : SUCCESS in 8m 56s","accounts_in_message":[],"_revision_number":2},{"id":"7d7f4c0a6aece6d1707c275710fea68e2720f1f0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-26 12:14:02.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/81162fdf77b54392b1801068ac9148b4\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c6af8a06bccc44e98c5bf822dbde5953 : SUCCESS in 27m 05s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f17eb54241294198910411ff2bb9ea66 : FAILURE in 12m 31s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"21b527a0ffd390e98aca1ec37c178abc4e5469a5","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-09-26 13:37:50.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b97444fa1084559e3457328caa4b62396b9c2792","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-26 14:31:14.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a579125809e34c5e93a7b56a1f60dc10\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/112b5c9dc0c0414da3551281dac3548e : SUCCESS in 24m 24s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ae20dd2cb39f476898b9cea2c8fb5cc1 : FAILURE in 9m 49s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a92ce9145c7f06a279c41c453a2139bd8e6297c8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-09-26 15:03:12.000000000","message":"Patch Set 3: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"cbb29a7d957b366a19db237fa3a760cd24c0c12d","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2024-09-26 15:58:14.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"e376550e4a9c16f55c5eb0490029493ddd454e7c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-26 16:13:14.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4723af57dbff4058b97a30bb7117511c\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4ba6847345c6494bb146677e023037be : SUCCESS in 21m 17s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9f91939c2af14910a9af1ce9beae936c : FAILURE in 9m 00s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"a3439ae6d4877f9fd5b77321c1ccdfa5834a7ff3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-26 16:56:23.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/01583ddd27484ed382e1d47a4df1c9b8\n\n- neutron-tempest-plugin-api-yoga https://zuul.opendev.org/t/openstack/build/2af6c4bfd0054a35a83e08f86743d092 : SUCCESS in 1h 00m 09s\n- neutron-tempest-plugin-scenario-linuxbridge-yoga https://zuul.opendev.org/t/openstack/build/b945722748374d5097161ec73c87c6d8 : SUCCESS in 1h 36m 57s\n- neutron-tempest-plugin-scenario-openvswitch-yoga https://zuul.opendev.org/t/openstack/build/436d3bf94863481a91aeb2b896084191 : SUCCESS in 1h 06m 07s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-yoga https://zuul.opendev.org/t/openstack/build/8dd2fb661c234607a9de547ec5d5de4e : SUCCESS in 1h 24m 34s\n- neutron-tempest-plugin-scenario-ovn-yoga https://zuul.opendev.org/t/openstack/build/a5828d211cf14cfb877f471133078c82 : SUCCESS in 1h 27m 35s\n- neutron-tempest-plugin-designate-scenario-yoga https://zuul.opendev.org/t/openstack/build/968a458d4b5f49c18d338e345cd77914 : SUCCESS in 26m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ccfe948fc8fe44b890bc1c6e9a176e6c : SUCCESS in 36m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/de0077af19054927bed3f3394112c022 : SUCCESS in 13m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c3b25352fd54403dbd5969eeb4c36c21 : SUCCESS in 28m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/39fa4d059d4c4a0fad4c7b1f552cc79e : FAILURE in 4m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ccf1852eab504d108769198b0db5d7eb : SUCCESS in 8m 54s","accounts_in_message":[],"_revision_number":3},{"id":"ceebe370901082031aacec494d0ef55d057a133a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-09-26 17:33:30.000000000","message":"Patch Set 4: Published edit on patch set 3.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"11d03c0a5fc60e60469df8c472e0e08100b7257a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-26 19:23:44.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f500bdc7768d4005a29362aa09aea3f3\n\n- neutron-tempest-plugin-api-yoga https://zuul.opendev.org/t/openstack/build/fb17d7fffa4a4a54a296a8281c90f6f8 : SUCCESS in 51m 58s\n- neutron-tempest-plugin-scenario-linuxbridge-yoga https://zuul.opendev.org/t/openstack/build/be626cc5ee94496d87ba82144da7b274 : SUCCESS in 1h 41m 49s\n- neutron-tempest-plugin-scenario-openvswitch-yoga https://zuul.opendev.org/t/openstack/build/d686745a24224c669f91664d6d089d1d : SUCCESS in 1h 14m 39s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-yoga https://zuul.opendev.org/t/openstack/build/31db57dc072748768aac22c87a25e8aa : SUCCESS in 1h 27m 09s\n- neutron-tempest-plugin-scenario-ovn-yoga https://zuul.opendev.org/t/openstack/build/55f52b5308b44601bf9682c308880e53 : SUCCESS in 1h 34m 11s\n- neutron-tempest-plugin-designate-scenario-yoga https://zuul.opendev.org/t/openstack/build/bb218624af1b450b83a6708e0ef76289 : SUCCESS in 35m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0eec7e9fc8ae4d31846d3d721e81ea19 : SUCCESS in 31m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7251bf69b8084457bf2033ab387a7672 : SUCCESS in 13m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/12addb5112104e819b247ca7d5415fe8 : SUCCESS in 28m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/35d3fdd68a1f48c7a3067c712dfb99f7 : FAILURE in 3m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d9bb6e4d941f4684952aca028b165135 : SUCCESS in 9m 29s","accounts_in_message":[],"_revision_number":4},{"id":"3526a91b2f84b1c2fdd013010844689962002a23","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-26 19:25:37.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b0eca5fc6e8f41f68cddf151b18d195f\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/09566f7992ca42c18499a79afa5a9423 : SUCCESS in 25m 46s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0012ba0cbce94f62ab200c74935ebd53 : FAILURE in 10m 00s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"39fa8e04cd0c148e47584c31e755027085bb5b5a","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2024-09-27 06:14:31.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"403671242fbfbf30f4414baa968288d2c5f3ff01","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-09-27 07:26:58.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"85477877e04cf837439dbfe9fd382735dae96c12","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-09-27 07:43:53.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"c295ebcda5e23928cc3fa4f47ed6123169f32274","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-09-27 12:17:25.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"843d20c5602a65b6efe519aa804c3924a8521e16","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-09-27 12:17:28.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"73e1943b7b004c1b5ee3fd3affd982c7cb87b40f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-27 12:55:30.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eb2d52b3c1f44d1e9153ef9689ede974\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/acc0f6e5f8f1465192719dc1967b5d0c : SUCCESS in 20m 42s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b692bb7a7e444e2aaa46457f9f46f8e1 : SUCCESS in 36m 15s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"7f7e721529b6be210aa869e9b867354470d5a47d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-27 14:02:06.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5831f89621d64e67a4d3082c3a154466\n\n- neutron-tempest-plugin-api-yoga https://zuul.opendev.org/t/openstack/build/01b9001086934c5ca0936e6de7e87f0a : SUCCESS in 36m 41s\n- neutron-tempest-plugin-scenario-linuxbridge-yoga https://zuul.opendev.org/t/openstack/build/53e39678646d41c6b447db5dd641be66 : FAILURE in 1h 30m 42s\n- neutron-tempest-plugin-scenario-openvswitch-yoga https://zuul.opendev.org/t/openstack/build/122cf4be95c14e7595e33f124ba75415 : SUCCESS in 1h 14m 25s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-yoga https://zuul.opendev.org/t/openstack/build/44edef7cfa624d1793f2687e8a33ac56 : SUCCESS in 1h 35m 27s\n- neutron-tempest-plugin-scenario-ovn-yoga https://zuul.opendev.org/t/openstack/build/859dbe94eebc4de28d3bd7676794339b : SUCCESS in 1h 36m 00s\n- neutron-tempest-plugin-designate-scenario-yoga https://zuul.opendev.org/t/openstack/build/754eebb2caba433ab71899f115f9175c : SUCCESS in 40m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f35247918e2e4c1abded3da0bdda48cc : SUCCESS in 38m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f7f5d00aa171461786ab7fb9d2479a81 : SUCCESS in 13m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/63147e4b4ffb41c984c6facc78228222 : SUCCESS in 30m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/036809e360c041388406518d15e612a7 : SUCCESS in 28m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/120d68a5ada943bfbeecc732e1e90bcd : SUCCESS in 10m 31s","accounts_in_message":[],"_revision_number":5},{"id":"abb7afbd85c6379cf55da68bbbc65236e7f0f750","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-09-30 07:43:39.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"dad7babdb83290a8774c4bdd8532e64a5e01a7c2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-30 08:17:36.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2198c892dd734ff2a785b6229cefb854\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0225a8b6edc5474c99fe8429ca21c37a : SUCCESS in 32m 14s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/072bb5faf53f47099bd2aedb621d373a : SUCCESS in 24m 29s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"eb994754c209d8ae14df6ad931494df7489ded46","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-09-30 08:28:48.000000000","message":"Uploaded patch set 6: New patch set was added with same tree, parent tree, and commit message as Patch Set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"a44de621f8d4ec9f764e26b52e20af1aa9786cb1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-30 09:04:46.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fff21a2a724245cca3af94ec703ba22c\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8029f163cd064633818c39f35ea107eb : SUCCESS in 33m 51s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/19a720e43a184f49b6e29b8966e3e4de : SUCCESS in 30m 28s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"d2fa47ecd08fe3283dd51d0434c1fb9c0e4a7c1d","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-09-30 09:41:28.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"3bdb42c674062e8f0ccc6fce842729752c94afcd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-30 09:59:26.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c558d1a1664447f49bc8e66a47071337\n\n- neutron-tempest-plugin-api-yoga https://zuul.opendev.org/t/openstack/build/06389097a9cd491781c1eb05ba3b22d3 : SUCCESS in 45m 38s\n- neutron-tempest-plugin-scenario-linuxbridge-yoga https://zuul.opendev.org/t/openstack/build/4f9359701a8b4bcabba663b30577d52c : SUCCESS in 1h 08m 58s\n- neutron-tempest-plugin-scenario-openvswitch-yoga https://zuul.opendev.org/t/openstack/build/ac248c798eda44639dc95e73b12b20e3 : SUCCESS in 47m 18s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-yoga https://zuul.opendev.org/t/openstack/build/bc5faccb325a45d8bd0920fb1a306746 : SUCCESS in 1h 17m 52s\n- neutron-tempest-plugin-scenario-ovn-yoga https://zuul.opendev.org/t/openstack/build/d17e4d58b67b4048a74be887237591de : SUCCESS in 1h 23m 21s\n- neutron-tempest-plugin-designate-scenario-yoga https://zuul.opendev.org/t/openstack/build/e615675869dd434aae514594afa05535 : SUCCESS in 33m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/779fd70547a9405bba17163c6ddc530d : SUCCESS in 37m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/38df817bb0b94a0cabf9a38e206cf874 : SUCCESS in 13m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3fdb963d44384857afb54dacfa892deb : SUCCESS in 28m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/25d8973dc90d4bd7945461d9e1353d63 : SUCCESS in 22m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/588e03e8b3454171a1149de1eea5d1b2 : SUCCESS in 9m 07s","accounts_in_message":[],"_revision_number":6},{"id":"a0c6520ccff3935b82d71abecf9f17d21c6de487","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2024-09-30 13:59:59.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"d9ede14e129c1c848d24bc568855aea4b6b94614","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-30 14:00:44.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"7953aa0d02606cf1c1a5edad5110997647e3abbe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-30 15:36:41.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/619cbf1de15f41c0b6787329f40a4291\n\n- neutron-tempest-plugin-api-yoga https://zuul.opendev.org/t/openstack/build/f0424427e7c94b24a387ccbee454b235 : SUCCESS in 53m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bae3e12b66f1499a9a29eda67cc2115f : SUCCESS in 14m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a2d630cdd68e4e90be35c7aef69dbcb4 : SUCCESS in 30m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6794006ab90d4e8591d11a0b80abcb30 : SUCCESS in 23m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c3980fff7e504d2da0302aae2f59441c : SUCCESS in 6m 48s","accounts_in_message":[],"_revision_number":6},{"id":"bba58932776a4cef3437f61b48ea959aa22341e8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-30 15:36:42.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"1c5836613e7a7689722944dcc70f1013a8f4ad6e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-30 15:41:10.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6b85485e7b5847909a525e8e71b018ac\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e709d723d2cd41aba4ac4a0d3e9e410a : SUCCESS in 1m 09s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"b285d6ff2f178e3b1c7fb7c055d37ac4f8790cfa","revisions":{"08f01670bebd4650d94df98bad7e79513bffc5c8":{"kind":"REWORK","_number":1,"created":"2024-09-23 15:04:52.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/11/930211/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/11/930211/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/11/930211/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/11/930211/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/11/930211/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/11/930211/1"}}},"commit":{"parents":[{"commit":"19ab990dc20aaa7d4fd91fc3aad8aff9407677fb","subject":"Fix regex lines in zuul.d/* files","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/19ab990dc20aaa7d4fd91fc3aad8aff9407677fb"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-09-23 15:04:40.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-09-23 15:04:40.000000000","tz":0},"subject":"[unmaintained-only] Periodic jobs will be weekly only","message":"[unmaintained-only] Periodic jobs will be weekly only\n\nAs agreed in the Neutron CI meeting, the unmaintained branches periodic\njobs will be weekly only [1].\n\n[1]https://meetings.opendev.org/meetings/neutron_ci/2024/neutron_ci.2024-09-23-14.01.log.html#l-97\n\nChange-Id: I0bc95e3705bd3114c36e1ec95701da2e2d13b13f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/08f01670bebd4650d94df98bad7e79513bffc5c8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/08f01670bebd4650d94df98bad7e79513bffc5c8"}]},"branch":"refs/heads/unmaintained/yoga"},"51aa2e032ff6643c119229ae4a78652e3a13079d":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2024-09-26 09:58:43.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/11/930211/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/11/930211/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/11/930211/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/11/930211/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/11/930211/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/11/930211/2"}}},"commit":{"parents":[{"commit":"19ab990dc20aaa7d4fd91fc3aad8aff9407677fb","subject":"Fix regex lines in zuul.d/* files","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/19ab990dc20aaa7d4fd91fc3aad8aff9407677fb"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-09-23 15:04:40.000000000","tz":0},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2024-09-26 09:58:43.000000000","tz":0},"subject":"[unmaintained-only] Periodic jobs will be weekly only","message":"[unmaintained-only] Periodic jobs will be weekly only\n\nAs agreed in the Neutron CI meeting, the unmaintained branches periodic\njobs will be weekly only [1].\n\n[1]https://meetings.opendev.org/meetings/neutron_ci/2024/neutron_ci.2024-09-23-14.01.log.html#l-97\n\nDepends-On: https://review.opendev.org/c/openstack/neutron-tempest-plugin/+/930548\n\nChange-Id: I0bc95e3705bd3114c36e1ec95701da2e2d13b13f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/51aa2e032ff6643c119229ae4a78652e3a13079d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/51aa2e032ff6643c119229ae4a78652e3a13079d"}]},"branch":"refs/heads/unmaintained/yoga","description":"Edit commit message"},"042b002e4d18455771c40ca176a0cbfcbb604c8f":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2024-09-26 15:03:12.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/11/930211/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/11/930211/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/11/930211/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/11/930211/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/11/930211/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/11/930211/3"}}},"commit":{"parents":[{"commit":"19ab990dc20aaa7d4fd91fc3aad8aff9407677fb","subject":"Fix regex lines in zuul.d/* files","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/19ab990dc20aaa7d4fd91fc3aad8aff9407677fb"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-09-23 15:04:40.000000000","tz":0},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2024-09-26 15:03:11.000000000","tz":0},"subject":"[unmaintained-only] Periodic jobs will be weekly only","message":"[unmaintained-only] Periodic jobs will be weekly only\n\nAs agreed in the Neutron CI meeting, the unmaintained branches periodic\njobs will be weekly only [1].\n\n[1]https://meetings.opendev.org/meetings/neutron_ci/2024/neutron_ci.2024-09-23-14.01.log.html#l-97\n\nDepends-On: https://review.opendev.org/c/openstack/neutron-tempest-plugin/+/930594\n\nChange-Id: I0bc95e3705bd3114c36e1ec95701da2e2d13b13f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/042b002e4d18455771c40ca176a0cbfcbb604c8f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/042b002e4d18455771c40ca176a0cbfcbb604c8f"}]},"branch":"refs/heads/unmaintained/yoga","description":"Edit commit message"},"13aac2169f857aa275d2fdd5fc8e23d172ca7c0e":{"kind":"REWORK","_number":4,"created":"2024-09-26 17:33:30.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/11/930211/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/11/930211/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/11/930211/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/11/930211/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/11/930211/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/11/930211/4"}}},"commit":{"parents":[{"commit":"19ab990dc20aaa7d4fd91fc3aad8aff9407677fb","subject":"Fix regex lines in zuul.d/* files","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/19ab990dc20aaa7d4fd91fc3aad8aff9407677fb"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-09-23 15:04:40.000000000","tz":0},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2024-09-26 17:33:30.000000000","tz":0},"subject":"[unmaintained-only] Periodic jobs will be weekly only","message":"[unmaintained-only] Periodic jobs will be weekly only\n\nAs agreed in the Neutron CI meeting, the unmaintained branches periodic\njobs will be weekly only [1].\n\nThis patch is also marking the CI job py39 as non-voting. There is an\nerror during the execution that should be fixed in a follow-up patch.\n\n[1]https://meetings.opendev.org/meetings/neutron_ci/2024/neutron_ci.2024-09-23-14.01.log.html#l-97\n\nDepends-On: https://review.opendev.org/c/openstack/neutron-tempest-plugin/+/930594\nRelated-Bug: #2082234\n\nChange-Id: I0bc95e3705bd3114c36e1ec95701da2e2d13b13f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/13aac2169f857aa275d2fdd5fc8e23d172ca7c0e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/13aac2169f857aa275d2fdd5fc8e23d172ca7c0e"}]},"branch":"refs/heads/unmaintained/yoga"},"ecb737267daa35a5c934d3aa569c75fdab9e16b1":{"kind":"REWORK","_number":5,"created":"2024-09-27 12:17:25.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/11/930211/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/11/930211/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/11/930211/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/11/930211/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/11/930211/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/11/930211/5"}}},"commit":{"parents":[{"commit":"9897b0e8ba2093f200fb0ea04bc5efbb869729f2","subject":"[stable-only] Cap setuptools \u003c71.0.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9897b0e8ba2093f200fb0ea04bc5efbb869729f2"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-09-23 15:04:40.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-09-27 12:16:58.000000000","tz":0},"subject":"[unmaintained-only] Periodic jobs will be weekly only","message":"[unmaintained-only] Periodic jobs will be weekly only\n\nAs agreed in the Neutron CI meeting, the unmaintained branches periodic\njobs will be weekly only [1].\n\n[1]https://meetings.opendev.org/meetings/neutron_ci/2024/neutron_ci.2024-09-23-14.01.log.html#l-97\n\nDepends-On: https://review.opendev.org/c/openstack/neutron-tempest-plugin/+/930594\n\nChange-Id: I0bc95e3705bd3114c36e1ec95701da2e2d13b13f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ecb737267daa35a5c934d3aa569c75fdab9e16b1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ecb737267daa35a5c934d3aa569c75fdab9e16b1"}]},"branch":"refs/heads/unmaintained/yoga"},"b285d6ff2f178e3b1c7fb7c055d37ac4f8790cfa":{"kind":"NO_CHANGE","_number":6,"created":"2024-09-30 08:28:48.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/11/930211/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/11/930211/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/11/930211/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/11/930211/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/11/930211/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/11/930211/6"}}},"commit":{"parents":[{"commit":"582136424455440c5722a8f81af049f7d3f39c44","subject":"[stable-only] Cap setuptools \u003c71.0.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/582136424455440c5722a8f81af049f7d3f39c44"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-09-23 15:04:40.000000000","tz":0},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2024-09-30 08:28:29.000000000","tz":120},"subject":"[unmaintained-only] Periodic jobs will be weekly only","message":"[unmaintained-only] Periodic jobs will be weekly only\n\nAs agreed in the Neutron CI meeting, the unmaintained branches periodic\njobs will be weekly only [1].\n\n[1]https://meetings.opendev.org/meetings/neutron_ci/2024/neutron_ci.2024-09-23-14.01.log.html#l-97\n\nDepends-On: https://review.opendev.org/c/openstack/neutron-tempest-plugin/+/930594\n\nChange-Id: I0bc95e3705bd3114c36e1ec95701da2e2d13b13f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b285d6ff2f178e3b1c7fb7c055d37ac4f8790cfa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b285d6ff2f178e3b1c7fb7c055d37ac4f8790cfa"}]},"branch":"refs/heads/unmaintained/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
