)]}'
{"id":"openstack%2Fneutron~930245","triplet_id":"openstack%2Fneutron~master~I35195eea928f99ba4ae2bcdea51436d250062d47","project":"openstack/neutron","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-12-09 23:05:09.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I35195eea928f99ba4ae2bcdea51436d250062d47","subject":"Convert [ml2] physical_network_mtus to DictOpt","status":"MERGED","created":"2024-09-24 01:49:47.000000000","updated":"2024-12-09 23:07:20.000000000","submitted":"2024-12-09 23:05:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":9,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"930245","meta_rev_id":"bd72ee6cbc31d7fe47cc48b61266fd125bf84dd4","_number":930245,"virtual_id_number":930245,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2024-09-24 02:54:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-12-09 23:05:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2024-10-01 12:55:56.000000000","_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"value":2,"date":"2024-09-25 13:39:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":2,"date":"2024-09-24 21:20:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":1,"date":"2024-10-01 12:55:56.000000000","_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-09-24 02:54:44.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-09-24 04:15:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-09-24 12:50:20.000000000","updated_by":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"reviewer":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"state":"CC"},{"updated":"2024-09-24 21:20:23.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2024-09-25 13:39:13.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2024-10-01 12:55:56.000000000","updated_by":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"reviewer":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"state":"REVIEWER"},{"updated":"2024-12-06 08:20:29.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"CC"}],"messages":[{"id":"1a9e7f3f82c55c034a84f3b00a7af8925c7cd057","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-09-24 01:49:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6bae120a8059d502e0bae7098be8cb7dc4fb36cb","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2024-09-24 02:54:44.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/September-2024/Arista_Tempest_Testrefs/changes/45/930245/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"7c1aa783055073ce3c307cd8c9852ebd8462dffc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-24 04:15:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b1b533c17aee4672ab3ffdf30fc34c64\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/284db8a136054bef80ba93c688d80c19 : SUCCESS in 1h 15m 50s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/c8abb37e9aa84342a66558866bc8cc56 : SUCCESS in 1h 44m 49s\n- neutron-tempest-plugin-ovn-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/96d50cc664e34209bf554bceffb8b1fd : SUCCESS in 2h 18m 07s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/c2d948f2b53242bcac771aa545874958 : SUCCESS in 1h 41m 39s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/424ad6babc3747d1aed2bdd72e20498e : SUCCESS in 37m 33s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/edecaf6a352945ddb4b8450a14e7bf03 : SUCCESS in 1h 04m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b67bc0a65507492ba30157d7225e0e1c : SUCCESS in 11m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/abf9caecb40d49a4826a6599b2223d30 : SUCCESS in 23m 16s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/65ff621b67714860ae60a925001f913a : SUCCESS in 17m 04s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5b231e3bffc34f4b81c38ae8675656df : SUCCESS in 23m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8163a9ff2a0142f1adc6bef2f24f796f : SUCCESS in 11m 17s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/a1eeaf10657e44a5ab5120d400f3bec2 : SUCCESS in 1h 03m 14s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/d1ce0501220b42179e679e92a2c04012 : SUCCESS in 2h 22m 40s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/d4f837b74d3b4d9aaa8627483a6cc9a0 : SUCCESS in 56m 05s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/0d6f56e8dd894238b76b7a90465a889b : SUCCESS in 56m 29s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/9326b04449894dfd9680d8090cd944c3 : SUCCESS in 1h 33m 06s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/e5911352571146668923cf8d7f2793dd : SUCCESS in 1h 09m 25s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/a7c03a89786b42449c3bb308b09719fe : SUCCESS in 1h 04m 26s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/6219e4890f8e490bb9c81866bd75c21d : SUCCESS in 1h 03m 47s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/ca34788f0b264b6e8cfdf232e0767f77 : SUCCESS in 1h 12m 10s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/66cfce5521634b8e9d4939ffcafa5be6 : SUCCESS in 1h 29m 26s","accounts_in_message":[],"_revision_number":1},{"id":"f829fef2516f569289ea4fbbe47af9b5d5f29f80","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-24 04:46:43.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f12436cbca2f4888b138ab7133e4b867\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0dcf070116b1414ea6e37345bb5be06f : SUCCESS in 27m 04s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/cf88951bfd8a4f4998950d97afd319a1 : SUCCESS in 25m 19s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6a58406165f19d56bec91ccb9b74e22340882919","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2024-09-24 12:50:20.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4a0a746a0363b12bddc7bc73a3e40006070b2f8f","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-09-24 21:20:23.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c8cf5485eecb3e4e516e07c14e9dbe73855cef48","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-09-25 13:39:13.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"14ea465b03bf4e229544085dea5e1a537eecd8e6","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2024-10-01 12:55:56.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d72cdba08e807a15178de89365867cd84b834255","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-01 12:56:48.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0f5be67af466a6f64e121ff2369d1cc24cafd884","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-01 15:14:08.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d19ef36c3ce14a5ca291f1f5b9da7924\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/a19460c099394daf940afcf97ea887a0 : SUCCESS in 2h 09m 29s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/01157821272d4658952a7c8a41e09bad : SUCCESS in 2h 04m 59s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/14545b569f1144ac9881c8b54bae0a24 : SUCCESS in 1h 43m 49s\n- neutron-tempest-plugin-ovn-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/77c3acc8ac9e43c6a561daa2f17c1d36 : SUCCESS in 1h 31m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ddc953f8625b4086944c4c49a944e2ef : SUCCESS in 9m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4a8cfd960bde43d4913b9f765e194738 : SUCCESS in 21m 41s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5ac3d110466f4e3ea2643c452333a820 : SUCCESS in 18m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0623f8e99a240cb84529c143f83e77d : SUCCESS in 6m 50s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/a66c99b262a7498d9dc5b9e502ea5894 : FAILURE in 49m 51s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/9ce8b64500af44afaa44125c31b9445c : SUCCESS in 2h 00m 34s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/9f6c4d497e1c406891bd25351ef63543 : SUCCESS in 1h 35m 55s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/3a705ad68abb4a4ba84da11bb29e7d3e : SUCCESS in 1h 02m 04s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/deb24f13dec04c0b866c0bae0fd60a72 : SUCCESS in 48m 49s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/28e0a6f5653c49fc9096323fcca5723f : SUCCESS in 1h 23m 41s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/aff11ec735a84ab9b45f550f29ed51a3 : SUCCESS in 1h 08m 33s","accounts_in_message":[],"_revision_number":1},{"id":"23ef4de5b96e224235ee944b9e38fff5da87b29a","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2024-10-01 15:16:32.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"807df6bc197a0d3a4463740be4ea0b5dc8408a65","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-01 17:16:08.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5bdfadc3e4a24f3799c59c3b30da2c7a\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3b153de172024d2d8da2962a531a2f96 : SUCCESS in 38m 54s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/42daf4e1f9d64fdb937ccc8ff3a2007e : SUCCESS in 33m 39s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2cb39db5a00e9c0928e7b22036f30f4bc6bf9dc3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-01 17:34:56.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d38abc2cc7ac427584549e14d415b5f8\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/e09417d4d42e47ca95f15dd967c36fa0 : SUCCESS in 1h 49m 13s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/66d0609dcd9849a4a47af405a4c4cc7e : SUCCESS in 1h 32m 29s\n- neutron-tempest-plugin-ovn-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/9747fc4744bf419989559ba8454e66db : SUCCESS in 2h 16m 32s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/ee54e95c202f43b99f4df6b3c156e4ed : FAILURE in 17m 06s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/2610e3d1181240b99dfe8e502afed89e : FAILURE in 25m 19s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f47a9c236f844b6b9ba8e74dfc5fede5 : SUCCESS in 49m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6eb18735843140f4b52bf58dbae1966d : SUCCESS in 9m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/246cd8f0dd6d4214a12f2c426d1cca73 : SUCCESS in 21m 46s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/efc4e8a5239d4337afec51ae7550bfec : SUCCESS in 18m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f7931143382647c6ab988ff885cf3326 : SUCCESS in 6m 48s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/70ce90a32af9437cb0aeceb38d3919e2 : SUCCESS in 50m 28s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/82292f004ca64af6b4fefef88b0e2cdc : SUCCESS in 1h 44m 02s\n- neutron-tox-mypy https://zuul.opendev.org/t/openstack/build/9b6906fa9ab64f07a56785072e782c79 : SUCCESS in 4m 59s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/f9c2ce135d11442e8dd8a26f54cc7d0a : SUCCESS in 59m 13s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/c8b0ab206cbc4ad4b3f95acfd109867c : SUCCESS in 1h 18m 52s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/a016b1a9f0b74baea12f7d00de64cdc2 : SUCCESS in 54m 50s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/cea9d2c4ff0a4fd0ba24b32e171b400c : SUCCESS in 1h 00m 28s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/290ef4ddeb134eafacf50112681dbe5e : SUCCESS in 1h 40m 16s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/cc6ee8f0e2cf4e1b9d03fae31ac75d3b : SUCCESS in 1h 37m 21s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/5df8b4c7dfef452eb193d1b2adab5bd4 : SUCCESS in 56m 03s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/bf3e87698ae94d88bae2d504bd75cd74 : SUCCESS in 52m 42s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/9f0b6ff6189d443f9e95671fa2d49a2d : SUCCESS in 1h 09m 47s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/27cb220a544b4933ba2e45740d7da006 : SUCCESS in 1h 09m 31s","accounts_in_message":[],"_revision_number":1},{"id":"55ab58a63df6d3c47dc9e55897e4d83c36a691fc","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2024-10-02 12:45:55.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"dee3cc1802013b3cca7966a40cb1f7242a8d429d","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2024-10-02 12:45:59.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"69166024c8189fc4b36827a4b4abf9694c254529","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-02 13:28:55.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/021e60dab04541f9affa86ca1a3c656b\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4bdcc69bbd014433bb34fa65242ebe6a : SUCCESS in 30m 47s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/fe9e2a6d2b334896a467a7791af56533 : SUCCESS in 32m 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"eb2236ef562a6f07bb5d90d11f8897dee1398aa9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-02 15:32:22.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b12446e953e246c3b278fb1e9425fa2f\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/be318c7d536b47169ce7c9cd66960841 : SUCCESS in 1h 17m 59s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/2b428968272a4769b864c0bbd91320d3 : SUCCESS in 2h 44m 11s\n- neutron-tempest-plugin-ovn-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/ddcf1abba62c4ea696d4fe6d363620c7 : SUCCESS in 1h 42m 41s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/91776e1593ff49db8c9d69a5ab0ca4c0 : SUCCESS in 1h 41m 56s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/203ba78946874bd38332d73a1c314253 : SUCCESS in 35m 19s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e3a798f83bfe4d75a9de0b7a0ab20a82 : SUCCESS in 51m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/46ceddaa0a8c44dfaf5137259aee19fb : SUCCESS in 9m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1b3905d1177148bea439fad2c21ff0ce : SUCCESS in 28m 46s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/914cf710d87447c3adae7ee3b2b7f9f1 : SUCCESS in 26m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e64fda73d7324ac9bc5b75b7da1777c4 : SUCCESS in 6m 53s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/59659627848c4fc7bd538927b412aa69 : SUCCESS in 1h 04m 20s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/5a05a6ed4e5d4420b1397b98525cc69d : SUCCESS in 1h 13m 14s\n- neutron-tox-mypy https://zuul.opendev.org/t/openstack/build/fba015c8d0174e57942bb4473693bc5c : SUCCESS in 5m 11s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/f07ec232991e4101bbf065b9d986cae8 : SUCCESS in 55m 37s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/c7a85e54c43e415f9e3b68328d8865bc : SUCCESS in 1h 55m 02s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/e764a140ebe04a1d9fe1bbf870fc0f80 : SUCCESS in 53m 50s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/3440efb3c4e94b989bb99a884d1a4b7f : SUCCESS in 1h 12m 13s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/defa852b8b46443dac0f9229f133696f : FAILURE in 1h 13m 32s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/520c5258618e45fcb182f130bc26cb7a : SUCCESS in 1h 27m 48s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/4b66fef938ed4cc6bdc6343d2cf15ee0 : SUCCESS in 57m 49s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/214505360fef43cbbfa907c5648f45d2 : SUCCESS in 1h 04m 41s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/1658a77d88fc4eb08a343b1bd0f24fcd : SUCCESS in 1h 02m 39s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/c6855df233bc4be19129c398ad2fe168 : SUCCESS in 1h 19m 40s","accounts_in_message":[],"_revision_number":1},{"id":"df534182a6129b3dcbb44897f26811858360a1f2","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2024-12-06 08:20:29.000000000","message":"Abandoned\n\nThis review is \u003e 4 weeks without comment, and failed Zuul jobs the last time it was checked. We are abandoning this for now. Feel free to reactivate the review by pressing the restore button and leaving a \u0027recheck\u0027 comment to get fresh test results.","accounts_in_message":[],"_revision_number":1},{"id":"efbb360da9909e392fddc47e1741893f6e6a4746","tag":"autogenerated:gerrit:restore","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-12-06 21:57:54.000000000","message":"Restored","accounts_in_message":[],"_revision_number":1},{"id":"b18ad35777fb378a9744c6e4e83c925015b56cf9","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-12-06 21:59:08.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0b5d1eb0d659750e30cd41af60c15e34bc419dc8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-06 23:04:06.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/161c1e610ec9491190918074b2e7cf24\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/87971097379744b2a2eba47dbaeffa5a : SUCCESS in 31m 26s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/49ab741ef69a4dca90ceebca44daf2ca : SUCCESS in 20m 57s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"aa79b1c9f09e1a0b4dbc29be29a3ec64be9a75b2","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2024-12-06 23:04:10.000000000","message":"Patch Set 1:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/December-2024/Arista_Tempest_Testrefs/changes/45/930245/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"ebd9fd2e29d530b8d68630cb5a55b1b77283dfa5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-07 00:09:30.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e49b536add0f4df78d05f7753113d7a2\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/53f0aa811bfa427db178174dc23ef8e6 : SUCCESS in 1h 21m 53s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/48fa2186d6c84154bd1d7023b1941bc6 : SUCCESS in 1h 46m 51s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/475d68e0ae7f434eabd5fa08126b30c0 : SUCCESS in 1h 45m 52s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/2a57dfc4070c44cb9e12fe277eb1c2a7 : SUCCESS in 31m 51s (non-voting)\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d0d4de0a97b94f1bb7486d062a1ca0ca : SUCCESS in 1h 18m 56s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/709d11e9305f4b68963d4f80bf62c341 : SUCCESS in 1h 51m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5e29fe8dba20408d92d11c7f960b7ed2 : SUCCESS in 50m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9333b68e42774aaaa596cdd55e2a656c : SUCCESS in 10m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fdcb159a67b54823919985bf8ff2038f : SUCCESS in 19m 42s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/851700742e654103aa5f2b40197c38e4 : SUCCESS in 30m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/113dec6b90dc459d9163c940154f9225 : SUCCESS in 10m 42s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/10b84ae160824fe9ad141feeab215988 : SUCCESS in 54m 10s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/8c086eea008a4bd69590dd8c25ee7b66 : SUCCESS in 1h 06m 01s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/379e60eaac4e430e9b00b599e73b3411 : FAILURE in 48m 03s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/6ad60817eb6349f2af7732524c04e9b9 : SUCCESS in 1h 47m 51s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/80de8e4b2b5f40af90031d09e8d76565 : SUCCESS in 56m 55s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/ed08968f3de44893a298772fadc5aeb1 : SUCCESS in 1h 13m 22s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/88e72c5a0e374bf78eef6ebf8ea33077 : SUCCESS in 1h 31m 26s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/1658bdda624a4095931a2951fd775263 : SUCCESS in 1h 37m 07s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/59156e67e6c743d494c5c419464898ad : SUCCESS in 58m 09s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/81b869c65b254eec8d262d07eecdeaf5 : SUCCESS in 1h 11m 03s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/5eb30d5acfde4c6e8d271d9d0e55ae57 : SUCCESS in 1h 35m 29s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/02305b9128b94a4ba5b638b01bae861d : SUCCESS in 1h 05m 50s","accounts_in_message":[],"_revision_number":1},{"id":"964f9d265af84dfcf98682dceb4edccc97c46288","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-12-09 17:33:01.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"796d966074cdf85c7ece38606dbe907e49a90e7c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-09 18:00:27.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/739872c4753940ffbd8db038faf4c1a9\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/305feb687ad1439e94a6a26c19c99ea7 : SUCCESS in 23m 26s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/085214405d4e4cb1b99bc0470deead3c : SUCCESS in 22m 40s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a28adfb5eadfab024be241696d572224caa72eca","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2024-12-09 19:23:38.000000000","message":"Patch Set 1:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/December-2024/Arista_Tempest_Testrefs/changes/45/930245/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"93f06ca7c9f0d9a1dff50c077c60aa678f663edb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-09 20:26:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c13088f9de2f425bbee9294894c14a07\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/bb2f054bc1d0469a861aeabf134433a1 : SUCCESS in 1h 45m 41s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/f19aa623cc7e49cc97130fb1f372c504 : SUCCESS in 1h 56m 31s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/62ed86f0fabe408eb9fb5d72fa0eabde : SUCCESS in 2h 13m 28s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/cd1801d8cc5f4d77b1a6f4956cee4e0e : SUCCESS in 43m 36s (non-voting)\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/62101e5abd374ffe8cbe1e7ee1a53042 : SUCCESS in 2h 08m 07s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c3e21b4df4034e248cd5a22b88c8b631 : SUCCESS in 2h 50m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8cb5fe3dadcf44d8bc1a774e96a1c6ae : SUCCESS in 35m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0f4cd3dffe8043138c2bc50cad682030 : SUCCESS in 11m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/83934984c89644e98fbf0afaf6d54387 : SUCCESS in 22m 31s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0b164bac743f41e481dc046afeb6243f : SUCCESS in 21m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/20164f291bdc4aeb8a6ac597b878856b : SUCCESS in 9m 25s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/6888bef890d549ac85bf6e6994b9c77f : SUCCESS in 1h 01m 15s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/35536803f3374bc29b113884db279388 : SUCCESS in 1h 10m 12s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/3066e65d7c854dce8b46c6fb2ad21f5d : SUCCESS in 46m 21s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/14cdcb5b9c0e4ae298096c62e59df95e : SUCCESS in 1h 50m 44s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/2619a267ec5a46a1bb2c2a863eed4c0e : SUCCESS in 1h 02m 51s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/a7983d4e21a74522b48bb1ab2411511f : SUCCESS in 1h 21m 40s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/a85ab53429ca435886f6e10b7dbd0d11 : SUCCESS in 1h 28m 10s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/8a808f9b44334dd0bda10e6aaa427024 : SUCCESS in 1h 25m 22s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/7af2b81c27da4cf2987dd44ef1b99246 : SUCCESS in 1h 08m 01s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/279985d9a3a74f1292b570af49c89944 : SUCCESS in 53m 21s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/c35c201814b34547b61c7ae3c112bda3 : SUCCESS in 57m 14s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/4e864000e26a4e0baa4d140d26189106 : SUCCESS in 1h 07m 36s","accounts_in_message":[],"_revision_number":1},{"id":"65f565c276b947bbbd1529d727f696a4916558a9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-09 20:27:18.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"be14f3448476aeb11167eda1b8aa53f7bb58bae0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-09 23:05:09.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3d3bdf52b8774fd48bbb5f96f867bf99\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/bb8b91e80f3b4accba8febc383376cb1 : SUCCESS in 2h 14m 12s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/89c5343461de4a3f9930b8e139356371 : SUCCESS in 2h 17m 18s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/c5f2fff99c51485fa4928fe42da9f230 : SUCCESS in 2h 36m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9b7fc13bf07e4bcdb70cd2d05695a3bd : SUCCESS in 11m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4408f69a067847c184beb1fa0b3ef6ca : SUCCESS in 23m 48s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e2c1f116f73e43d78fe124bd167fd32d : SUCCESS in 19m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3704c9f8ca9c489eb38e293d462ee7e0 : SUCCESS in 10m 50s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/535c41d35fab4fd38702e519bc8265f8 : SUCCESS in 1h 03m 50s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/8473bbcc06774ea1a92d2aa3a820d626 : SUCCESS in 1h 53m 16s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/6fdb186011154d318cd19f0dd1538439 : SUCCESS in 1h 56m 32s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/d603b5159ed6462cabc721d197708131 : SUCCESS in 1h 09m 58s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/73a1b7f35c784bb1a0d7378358c04b4c : SUCCESS in 1h 08m 44s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/6957531b28c64afbbaa93676f5d3f7b6 : SUCCESS in 1h 05m 16s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/2f9d9ff64cb74ee5bfac22912e99ceda : SUCCESS in 1h 09m 16s","accounts_in_message":[],"_revision_number":1},{"id":"f3a11e25531db6d3ab6980c9df65dc6d53523012","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-09 23:05:09.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"bd72ee6cbc31d7fe47cc48b61266fd125bf84dd4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-09 23:07:20.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2ec61be164d847f7b5ef436e49e24a49\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a1504964db44c4d93e5b543f0b41eb0 : SUCCESS in 1m 51s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ae4b4300f9dacee68ce5a1dfd88c601424d2229d","revisions":{"ae4b4300f9dacee68ce5a1dfd88c601424d2229d":{"kind":"REWORK","_number":1,"created":"2024-09-24 01:49:47.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/45/930245/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/45/930245/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/930245/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/930245/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/930245/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/45/930245/1"}}},"commit":{"parents":[{"commit":"203e28105fd690665189ca68d3c88b62148372bf","subject":"Merge \"refactor: minor ternary `if x then true else false` squash\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/203e28105fd690665189ca68d3c88b62148372bf"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-09-24 01:47:52.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-09-24 01:49:02.000000000","tz":540},"subject":"Convert [ml2] physical_network_mtus to DictOpt","message":"Convert [ml2] physical_network_mtus to DictOpt\n\nThis option accepts a dictionary value. Use the native oslo.config\nimplementation so that its value is parsed at config layer and explicit\nerror is returned to users in case an invalid value is configured.\n\nChange-Id: I35195eea928f99ba4ae2bcdea51436d250062d47\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ae4b4300f9dacee68ce5a1dfd88c601424d2229d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ae4b4300f9dacee68ce5a1dfd88c601424d2229d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
