)]}'
{"id":"openstack%2Fneutron~930582","triplet_id":"openstack%2Fneutron~unmaintained%2Fyoga~I34d5dde2151abf03e125e776871ec84ebfe09fe6","project":"openstack/neutron","branch":"unmaintained/yoga","topic":"bug/2082234","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2024-09-26 17:32:15.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I34d5dde2151abf03e125e776871ec84ebfe09fe6","subject":"Mark CI job py39 as non-voting","status":"ABANDONED","created":"2024-09-26 13:53:05.000000000","updated":"2024-09-26 17:32:15.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"6519aa25896edd223c0ff073f5c0731cb9ec50ad","_number":930582,"virtual_id_number":930582,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-09-26 15:51:40.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-09-26 14:47:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-09-26 15:51:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1d013bb221a63900ffbc4e1c5ca2ef149630db19","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-09-26 13:53:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"28b22844562e2f94df854bdb3dd0bdd70a3cced9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-26 14:47:27.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7c62e98d93294bf99ce900c0edf1c1bc\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9958252a00304ed696eecb616425c723 : SUCCESS in 28m 08s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/826f3503bfdf4ee8b349797505d8d610 : FAILURE in 11m 26s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"61838fc56d673edc69f243b94a858a09c8b9b8d3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-26 15:51:40.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4a09458a76a646fa8541992cf9bf1d0f\n\n- neutron-tempest-plugin-api-yoga https://zuul.opendev.org/t/openstack/build/ccbc9bfbc48b4f4d8d62a449301c84e6 : FAILURE in 49m 10s\n- neutron-tempest-plugin-scenario-linuxbridge-yoga https://zuul.opendev.org/t/openstack/build/902fafdd87f745a3ba7d94b9aa6fc55b : SUCCESS in 1h 38m 09s\n- neutron-tempest-plugin-scenario-openvswitch-yoga https://zuul.opendev.org/t/openstack/build/b0ab978bfed246d89de1c28ba3deae51 : SUCCESS in 1h 10m 16s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-yoga https://zuul.opendev.org/t/openstack/build/01f681c3d3624b34809f8222f9661f51 : SUCCESS in 1h 23m 50s\n- neutron-tempest-plugin-scenario-ovn-yoga https://zuul.opendev.org/t/openstack/build/ae3c9e16379a4e62ba7fb4d0e6b73b7f : FAILURE in 1h 35m 04s\n- neutron-tempest-plugin-designate-scenario-yoga https://zuul.opendev.org/t/openstack/build/224c35ce1d474cc6862f832d13a6e8e1 : SUCCESS in 43m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/498a7e80036e4e37b9f19a8476752387 : SUCCESS in 35m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/377fa3e6f6534c5897238d46570c2d55 : SUCCESS in 12m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d1f0126ffbc04fb492138255f861fa08 : SUCCESS in 27m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ecce09dde18845af8f34192db8136412 : FAILURE in 4m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e03a6487f60b4e9fb638376963aca88c : SUCCESS in 8m 48s","accounts_in_message":[],"_revision_number":1},{"id":"6519aa25896edd223c0ff073f5c0731cb9ec50ad","tag":"autogenerated:gerrit:abandon","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-09-26 17:32:15.000000000","message":"Abandoned\n\nThis patch must be combined with https://review.opendev.org/c/openstack/neutron/+/930211 in order to pass the CI","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"218ffcb32faef8c04076e9b6720a46e4143d1d68","revisions":{"218ffcb32faef8c04076e9b6720a46e4143d1d68":{"kind":"REWORK","_number":1,"created":"2024-09-26 13:53:05.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/82/930582/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/82/930582/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/82/930582/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/82/930582/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/82/930582/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/82/930582/1"}}},"commit":{"parents":[{"commit":"19ab990dc20aaa7d4fd91fc3aad8aff9407677fb","subject":"Fix regex lines in zuul.d/* files","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/19ab990dc20aaa7d4fd91fc3aad8aff9407677fb"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-09-26 13:52:03.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-09-26 13:52:03.000000000","tz":0},"subject":"Mark CI job py39 as non-voting","message":"Mark CI job py39 as non-voting\n\nThere is an error during the execution that should be fixed in a\nfollow-up patch.\n\nRelated-Bug: #2082234\nChange-Id: I34d5dde2151abf03e125e776871ec84ebfe09fe6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/218ffcb32faef8c04076e9b6720a46e4143d1d68"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/218ffcb32faef8c04076e9b6720a46e4143d1d68"}]},"branch":"refs/heads/unmaintained/yoga"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
