)]}'
{"id":"openstack%2Fneutron~930681","triplet_id":"openstack%2Fneutron~unmaintained%2Fyoga~I08e57e3b4c905bd679038c89185e54f6ab79738a","project":"openstack/neutron","branch":"unmaintained/yoga","topic":"fix-setuptools-canonicalize-version","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2024-09-30 09:24:47.000000000","reason":"\u003cGERRIT_ACCOUNT_16688\u003e replied on the change","reason_account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2024-09-30 15:36:38.000000000","reason":"Change was submitted"},"21798":{"account":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"last_update":"2024-09-30 13:59:41.000000000","reason":"\u003cGERRIT_ACCOUNT_21798\u003e replied on the change","reason_account":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2024-09-30 10:06:42.000000000","reason":"\u003cGERRIT_ACCOUNT_8313\u003e replied on the change","reason_account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}}},"hashtags":[],"change_id":"I08e57e3b4c905bd679038c89185e54f6ab79738a","subject":"[stable-only] Cap setuptools \u003c71.0.0","status":"MERGED","created":"2024-09-27 07:42:36.000000000","updated":"2024-09-30 15:39:22.000000000","submitted":"2024-09-30 15:36:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"930681-fix-setuptools-canonicalize-version","meta_rev_id":"0963fb7e5aba15a57a21a5ad0907f7f74f9ef794","_number":930681,"virtual_id_number":930681,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-09-30 15:36:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"recommended":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2024-09-30 09:24:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2024-09-30 13:59:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2024-09-30 10:06:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2024-09-30 13:59:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-09-27 08:27:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-09-27 09:11:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-09-27 10:34:52.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2024-09-27 10:34:59.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2024-09-27 12:15:20.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"}],"messages":[{"id":"c697cd764b0f2fc1859c43f9b043c579bdc18455","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-09-27 07:42:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"943e24fa5a5a1a5e0ccb8ab09c4d9db073c899a9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-27 08:27:16.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d39ea72cce094884b6310b018dd18dec\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/eb5aa51a084a4ac1b9190ca775e5b27b : SUCCESS in 42m 33s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/77798b2349264dabb5e9b6163ebba7d9 : SUCCESS in 41m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"82c21ef48d12965aa8def4a9764bc5fba8f946df","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-27 09:11:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ee96f3e2c31746e5954f89d5581e799a\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b44ca495d7444d999dda409866cd8cc6 : SUCCESS in 40m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9b53b531955c4b78b18d3d1ca8712320 : SUCCESS in 13m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0f66772c16914b14a761e04a1db48555 : SUCCESS in 25m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2fd0c4cb165041838c47a8b53fecebd8 : SUCCESS in 33m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f28984e2ef8d4b568a4b50df41c74c67 : SUCCESS in 10m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ba484cfb974e447599499c1df813ef18 : SUCCESS in 20m 55s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/ed4d1c207fd0482eb9f72be36b81ea3d : SUCCESS in 48m 22s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/fafbd46eef7948d7845c31372ede1b5d : SUCCESS in 1h 18m 58s","accounts_in_message":[],"_revision_number":1},{"id":"241a980a2c75508c876b1c045e7eb37f992f7e24","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-09-27 10:34:52.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"175e22eca942f737abecb2d4045391022fb1f055","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-09-27 12:15:20.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"83459b5bf0f88a3fdb5d315907b9e625bce3f2c3","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-09-27 12:15:57.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"747cd1edaf55ff2b025561c25234296b70be9419","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-09-30 08:28:48.000000000","message":"Uploaded patch set 2: Commit message was updated.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"50c632d4c889ac0ade6eb23fb048eb622fd97dca","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-09-30 08:29:23.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ad2f639b55b2fef538cf05f90a73b7426af671d0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-30 09:07:13.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8ea1592fcae84e288f742c3f75dc3ead\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/10b88be706184224b299a317cccc03b8 : SUCCESS in 34m 03s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d9e3c212b70b4c9eae55a86a38d51e4d : SUCCESS in 36m 24s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"469b589d671db976ebc7725cde57b26032a0a7c6","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-09-30 09:24:47.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"224693566d8a074a3bd9a05906caa1ed91b75b38","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-09-30 10:06:42.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f0deec4fa2b45d93c3945535d04d2c5306a9af85","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-30 10:11:45.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4613d208f88d424ba2b3eba13b18395f\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/22bd26edd7924a498f18c36a61d6c41a : SUCCESS in 37m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3740e4607d974a8cbd8a589e94ba28d8 : SUCCESS in 12m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b4e27ed8b6a14717bc332390651dda31 : SUCCESS in 25m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/450431900d4142e69c126d4f6eb288e2 : SUCCESS in 24m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a13d7b54a30c400fa0779115e9c9cf1f : SUCCESS in 9m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c27f28ec8a5340408012b880bd632460 : SUCCESS in 14m 18s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/0430cee5c64b45419ff0339cf5250385 : SUCCESS in 43m 12s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/54533cc4614146efa2c7f5ad18ff42ac : SUCCESS in 1h 41m 39s","accounts_in_message":[],"_revision_number":2},{"id":"2240aee197f9c4cfeea337052854f0370fbf3b2f","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2024-09-30 13:59:41.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"c2fab77374255fb76c4a643b49debb943298b447","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-30 14:00:07.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"bc2ff290022d37759a673e321a702e48c4d5f4ad","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-30 15:36:38.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/52c48f2b203b462da082d8134a9e9872\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f1bf691b8b6b4b98b4c50fd983bb72f3 : SUCCESS in 15m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8a5c954348164e86849cb8f4a4587e65 : SUCCESS in 30m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/da9bd5292a1f4cafb2a246c5de158795 : SUCCESS in 23m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7436438fc0154b3fbe11acad0e6e4f44 : SUCCESS in 9m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/98e4c0ec96904fce99eb90f8f60bb2f7 : SUCCESS in 24m 53s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/38ce8fb4d13a45d1b81ee6e3989fa477 : SUCCESS in 46m 15s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/ce1e61354b50452b8d14a3de6852b4ca : SUCCESS in 1h 33m 42s","accounts_in_message":[],"_revision_number":2},{"id":"3a3a85c15d36e34818a25f77ca51f960297a63a6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-30 15:36:38.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"0963fb7e5aba15a57a21a5ad0907f7f74f9ef794","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-30 15:39:22.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4c0755ef98d54d82a5896eaffd835e88\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8d3df78549ba49e282ff15d4a2d46c68 : SUCCESS in 1m 07s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fa7f1ecaec2246849987e12ca93c3b8b : SUCCESS in 54s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"582136424455440c5722a8f81af049f7d3f39c44","revisions":{"9897b0e8ba2093f200fb0ea04bc5efbb869729f2":{"kind":"REWORK","_number":1,"created":"2024-09-27 07:42:36.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/81/930681/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/81/930681/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/81/930681/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/81/930681/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/81/930681/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/81/930681/1"}}},"commit":{"parents":[{"commit":"19ab990dc20aaa7d4fd91fc3aad8aff9407677fb","subject":"Fix regex lines in zuul.d/* files","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/19ab990dc20aaa7d4fd91fc3aad8aff9407677fb"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2024-09-27 07:39:53.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2024-09-27 07:39:53.000000000","tz":120},"subject":"[stable-only] Cap setuptools \u003c71.0.0","message":"[stable-only] Cap setuptools \u003c71.0.0\n\npy39 jobs (on ubuntu-focal) started to fail due to recent virtualenv\nrelease (20.26.4) on Yoga (which bundles setuptools), because we have\n\u0027packaging\u003d\u003d21.3\u0027 in this branch that is not compatible with newer\nsetuptools [1].\n\nsetuptools is bundled in virtualenv, so it has to be capped via the\nvirtualenv package. tox also needed to be capped (\u003c4) as gate uses\ntox 3.28.0, but with capping virtualenv we pull in latest tox as well,\nwhich would cause other errors.\n\n[1] https://github.com/pypa/setuptools/issues/4483\n\nChange-Id: I08e57e3b4c905bd679038c89185e54f6ab79738a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9897b0e8ba2093f200fb0ea04bc5efbb869729f2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9897b0e8ba2093f200fb0ea04bc5efbb869729f2"}]},"branch":"refs/heads/unmaintained/yoga"},"582136424455440c5722a8f81af049f7d3f39c44":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2024-09-30 08:28:48.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/81/930681/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/81/930681/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/81/930681/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/81/930681/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/81/930681/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/81/930681/2"}}},"commit":{"parents":[{"commit":"19ab990dc20aaa7d4fd91fc3aad8aff9407677fb","subject":"Fix regex lines in zuul.d/* files","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/19ab990dc20aaa7d4fd91fc3aad8aff9407677fb"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2024-09-27 07:39:53.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2024-09-30 08:27:42.000000000","tz":120},"subject":"[stable-only] Cap setuptools \u003c71.0.0","message":"[stable-only] Cap setuptools \u003c71.0.0\n\npy39 jobs (on ubuntu-focal) started to fail due to recent virtualenv\nrelease (20.26.4) on Yoga (which bundles setuptools), because we have\n\u0027packaging\u003d\u003d21.3\u0027 in this branch that is not compatible with newer\nsetuptools [1].\n\nsetuptools is bundled in virtualenv, so it has to be capped via the\nvirtualenv package. tox also needed to be capped (\u003c4) as gate uses\ntox 3.28.0, but with capping virtualenv we pull in latest tox as well,\nwhich would cause other errors.\n\n[1] https://github.com/pypa/setuptools/issues/4483\n\nCloses-Bug: #2082234\nChange-Id: I08e57e3b4c905bd679038c89185e54f6ab79738a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/582136424455440c5722a8f81af049f7d3f39c44"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/582136424455440c5722a8f81af049f7d3f39c44"}]},"branch":"refs/heads/unmaintained/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
