)]}'
{"id":"openstack%2Fneutron~930889","triplet_id":"openstack%2Fneutron~master~I901a5abb92a5a178d847143266419b1e502154d2","project":"openstack/neutron","branch":"master","topic":"bug/2069581","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2024-10-01 11:57:47.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I901a5abb92a5a178d847143266419b1e502154d2","subject":"Disable the OVN/OVS WGSI experimental jobs","status":"MERGED","created":"2024-09-30 10:32:25.000000000","updated":"2024-10-01 12:00:42.000000000","submitted":"2024-10-01 11:57:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"930889-bug/2069581","meta_rev_id":"21f657af21dd2d812668643345f3ca2f7ca64abc","_number":930889,"virtual_id_number":930889,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-10-01 11:57:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":2,"date":"2024-10-01 09:23:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":2,"date":"2024-09-30 18:27:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":1,"date":"2024-10-01 09:23:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-09-30 10:32:59.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"CC"},{"updated":"2024-09-30 10:32:59.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"CC"},{"updated":"2024-09-30 10:32:59.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"CC"},{"updated":"2024-09-30 10:32:59.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2024-09-30 11:01:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-09-30 12:57:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-09-30 18:27:05.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2024-10-01 09:23:40.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"}],"messages":[{"id":"199d1d6141b8cbb32a6674ec5070ec5b7bb6624a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-09-30 10:32:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"370a258992c42e0cb41023b834c76780a64a9d4c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-30 11:01:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bb1b884c722e4158820b6463eef80191\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a09dbc406d0a4d0a981a5b9ce4d3137d : SUCCESS in 25m 19s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/2309969e895e4ebaa59c62da765115b0 : SUCCESS in 21m 38s\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/1c407bc7bd76438d9ffbd7f47d52b0d7 : SUCCESS in 27m 13s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c769a71788114378464d1b197e73f201591cce5f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-30 12:57:07.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d0119845305b444ea880d0b10136ba32\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/f125734281a4464f97ddd31fddcd125f : SUCCESS in 1h 40m 36s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/786d46eeac4f488dad074aaaf5f226c0 : SUCCESS in 2h 07m 59s\n- neutron-tempest-plugin-ovn-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/65db686efc174441999663b6d2619ba1 : SUCCESS in 1h 52m 51s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/0d3ffa543ec34c4286effa2e0043f384 : FAILURE in 2h 22m 36s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/43a3e734116a45e69c29eaf65f959cf0 : SUCCESS in 30m 42s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7df33baeb5b94d3f8c3793a614db55e7 : SUCCESS in 48m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/23039f34ca7d4bd494ede4cf4790c496 : SUCCESS in 9m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/38d2542dea504743b2b39973430b8334 : SUCCESS in 24m 48s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c04c59e6972c4d89a3f820bb0f24670c : SUCCESS in 21m 28s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e7bf3615cb2048499fe2d025822e8ae3 : SUCCESS in 17m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c4fbd41080864ebc93b9b8a78b95585c : SUCCESS in 6m 26s","accounts_in_message":[],"_revision_number":1},{"id":"008abbdb76c694c622dae956be9695d54b40f8ec","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-09-30 13:37:21.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2b0d077690b21ce6db0ee7adda86a8bb04702296","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-30 15:27:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1e2986ac50b740a6abcfcb5c7bd54fd9\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/e3f336b6ae6c4418a69352423bbf980c : SUCCESS in 1h 24m 11s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/3f5a7bed9d9547e9aabbc5aef4a54b6b : SUCCESS in 1h 32m 11s\n- neutron-tempest-plugin-ovn-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/b77179d6c09f41318ee48b6dc6c246b5 : SUCCESS in 1h 39m 27s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/914d72d6669649d1a74e97072585b50a : SUCCESS in 1h 47m 42s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/f33eb5818ca94b7791148617645d52cd : SUCCESS in 30m 15s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1a6fc2cd45a5443186691500d17d9d84 : SUCCESS in 40m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/96c391d067da48ddaa33bc646ca0a1b6 : SUCCESS in 10m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6627216a0bf34eceb27b86cccdfc6f90 : SUCCESS in 22m 42s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/13a259f52fc040d7a6d689b67b31acc9 : SUCCESS in 21m 26s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8df7b8c40807462c8e235f949cb2465d : SUCCESS in 22m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b8472976ea0a4f0a97605e486ba78ca9 : SUCCESS in 13m 06s","accounts_in_message":[],"_revision_number":1},{"id":"88da4592bba6573cbc04a9b138e31ef0cb77a0dc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-30 15:41:11.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3968a3b90271439b9c3e12b50d63c20e\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0f8a1dd8379f4758a700bf2b0b1fe8e1 : SUCCESS in 26m 20s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/8ec50a8120b640e5a0204f83294a2bf2 : SUCCESS in 21m 49s\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/3f336cd842a14ef5bf3f9583bebd87b8 : SUCCESS in 32m 55s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4015a49288634c90a4be4ed2cbd14810d2d8c890","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-09-30 18:27:05.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d67930d414bfcaecf7db0c92de8592e25a828c16","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-10-01 09:23:40.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e44c5fe9598d237c7ff523da59792093c3b28f0a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-01 09:24:17.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e21dd8568f70bdb320fadd75a77e08aba0de66a0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-01 11:57:47.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f9f0a1b994444ffb9467063f0ea52f32\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/57749ff460c349529a6eaec7e0826a6e : SUCCESS in 1h 22m 28s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/e41c973f63cc4f5c8ea5451cbc9779f6 : SUCCESS in 1h 25m 46s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/08c36c37672044c0b78708267b6d6049 : SUCCESS in 1h 57m 59s\n- neutron-tempest-plugin-ovn-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/4b31956e19b74043a88f314606b80af7 : SUCCESS in 2h 05m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9ca75a89a64f4d45b9ed73005ee12233 : SUCCESS in 9m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/47ed55b1d83840589b2e4f84035f2da1 : SUCCESS in 23m 04s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a23b52e6beec48489390c43c55787b51 : SUCCESS in 21m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f3f458281efa4a42a8680d8a375b70b3 : SUCCESS in 11m 03s","accounts_in_message":[],"_revision_number":1},{"id":"b95911494a3f7285825698770ef6631cb52825d4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-01 11:57:47.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"21f657af21dd2d812668643345f3ca2f7ca64abc","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-01 12:00:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ea5a4c683c614e10a718b7e64ddc5bfe\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/42602d86133d4095b0b49e337f077bec : SUCCESS in 57s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"16fc47127a663c0f54f088de6b81406d7efebd30","revisions":{"16fc47127a663c0f54f088de6b81406d7efebd30":{"kind":"REWORK","_number":1,"created":"2024-09-30 10:32:25.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/89/930889/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/89/930889/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/89/930889/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/89/930889/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/89/930889/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/89/930889/1"}}},"commit":{"parents":[{"commit":"a4671ad9fe01dfc589818278192f81f8568ecb6b","subject":"Merge \"refactor: use the new py39 `dict | dict` syntax to make a union\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a4671ad9fe01dfc589818278192f81f8568ecb6b"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-09-30 10:28:44.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-09-30 10:32:14.000000000","tz":0},"subject":"Disable the OVN/OVS WGSI experimental jobs","message":"Disable the OVN/OVS WGSI experimental jobs\n\nThe following jobs have been disabled:\n* neutron-tempest-plugin-api-ovs-wsgi\n* neutron-tempest-plugin-api-ovn-wsgi\n\nOnce [1] is merged, all ML2/OVS and ML2/OVN jobs will be executed\nwith WSGI module. These jobs will be converted to *-eventlet and\nwill be in the experimental and periodic queues.\n\n[1]https://review.opendev.org/c/openstack/neutron-tempest-plugin/+/930743\n\nRelated-Bug: #2069581\nRelated-Bug: #2069831\nTopic: eventlet-deprecation\nChange-Id: I901a5abb92a5a178d847143266419b1e502154d2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/16fc47127a663c0f54f088de6b81406d7efebd30"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/16fc47127a663c0f54f088de6b81406d7efebd30"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
