)]}'
{"id":"openstack%2Fneutron~931249","triplet_id":"openstack%2Fneutron~master~Iebc895e6c8bd0f1129880244325afa1d6a0181c0","project":"openstack/neutron","branch":"master","topic":"bug/2069831","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2024-10-11 17:27:24.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iebc895e6c8bd0f1129880244325afa1d6a0181c0","subject":"Remove the sleep calls in the ``create_or_update_agent`` method","status":"MERGED","created":"2024-10-03 08:08:04.000000000","updated":"2024-10-11 17:28:35.000000000","submitted":"2024-10-11 17:27:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"931249-bug/2069831","meta_rev_id":"b3df26025230142db9d2ca0f9f73b952f591f3cd","_number":931249,"virtual_id_number":931249,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2024-10-04 07:30:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-10-11 17:27:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2024-10-07 14:27:14.000000000","_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"value":2,"date":"2024-10-11 15:01:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"value":1,"date":"2024-10-11 15:01:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-03 08:40:42.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-10-03 09:30:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-10-03 10:09:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-10-03 13:40:20.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2024-10-07 14:27:14.000000000","updated_by":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"reviewer":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"state":"REVIEWER"},{"updated":"2024-10-11 15:01:05.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"}],"messages":[{"id":"b3c82f81c7a805763dda3eb11ab5086a39f01721","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-10-03 08:08:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"979d6c45a64e426e11810c0cc74881fcfe3613bd","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-10-03 08:08:18.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f983460fd36681177b5ce3a1c7008bd65c0a085b","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2024-10-03 08:40:42.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/October-2024/Arista_Tempest_Testrefs/changes/49/931249/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"b31516020ad8a4f08198a1c29c7e84e12b57518c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-03 09:30:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7cde3f206f22479a91ebf5c8929ad355\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d5201c3934c341e3bb3af67351e9106a : SUCCESS in 21m 11s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/a59bf59400a546da96bac2aec268df8e : SUCCESS in 34m 42s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ea196d380bccddd57ef8c9ad13a977839a92100d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-03 10:09:04.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/cb14e42990e54b57bbd467a41a816097\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/64cb2dc90ebc40ad80b8061c5a2df3b3 : SUCCESS in 1h 19m 09s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/36d9fa35c1ec429abbc4708aef05c886 : SUCCESS in 1h 21m 37s\n- neutron-tempest-plugin-ovn-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/fbb0d48a6e4549d49d5d4434d71f1a2d : SUCCESS in 1h 27m 30s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/4b74b8baf11045c08a24cc4349d3a9c4 : SUCCESS in 1h 29m 45s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/e5f0f4bf98064f11bff94eaa601ac2d7 : SUCCESS in 28m 14s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/450162c9cb40459683e69863491ec03e : SUCCESS in 53m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2f3ba58b2dc344dbb34e0c37d9efce85 : SUCCESS in 10m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f28fe08c93504988974950e1fbe66892 : SUCCESS in 23m 40s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1196fdf495b8445eb0d16e65175de765 : SUCCESS in 20m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/96263e1369874090bb13026420652e90 : SUCCESS in 10m 34s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/d8915ba2ab9f42dd833c9d366ec25949 : SUCCESS in 1h 18m 36s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/57b7c349110e4f169dd9d6aa5abc9f64 : SUCCESS in 1h 11m 12s\n- neutron-tox-mypy https://zuul.opendev.org/t/openstack/build/881124ed3d83444eb5b426757df58cf2 : SUCCESS in 5m 01s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/d597ff8943da48dfa1b0a53dbaceb6b4 : FAILURE in 50m 14s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/eb14d39358454c8aab342df0d1c95690 : SUCCESS in 1h 54m 35s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/9cacff153b8d4c33b2896a8fbd257aa6 : SUCCESS in 39m 36s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/d5d072480fcd407183514a9127ae5ea4 : SUCCESS in 1h 11m 03s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/bf0b3d83dc3848a3914d2a52601e6b16 : SUCCESS in 1h 43m 41s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/ed36a7d349964de19765b3c8cb56de64 : SUCCESS in 1h 29m 31s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/ec165f57409340ee90dabc179239db5f : SUCCESS in 1h 10m 13s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/4f00962e872449038957bb037e93de83 : SUCCESS in 39m 25s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/f20953821b9b4d5ea889e95642c81f1a : SUCCESS in 42m 17s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/0c53c8f19f574ac98b3b3b70561f6a7a : SUCCESS in 1h 11m 10s","accounts_in_message":[],"_revision_number":1},{"id":"35684941617df1f917b86bfea00adefbe3281609","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-03 11:00:32.000000000","message":"Patch Set 1:\n\nBuild failed (experimental pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7873805bbb984ba3b3cf429a9a2a45d8\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/47be5f6689ec4cfe9aa7111cbc152a13 : SUCCESS in 2h 20m 51s\n- neutron-tempest-plugin-openvswitch-distributed-dhcp https://zuul.opendev.org/t/openstack/build/60335b9f8f3a4f5b815c9e22ceade316 : SUCCESS in 1h 18m 12s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-distributed-dhcp https://zuul.opendev.org/t/openstack/build/f5bcb7acd4c54384baa8586073808747 : FAILURE in 1h 25m 06s\n- neutron-tempest-plugin-api-ovs-wsgi https://zuul.opendev.org/t/openstack/build/3459e7aeda6143af9ae2bc577fd03641 : SUCCESS in 1h 20m 45s (non-voting)\n- neutron-tempest-plugin-api-ovn-wsgi https://zuul.opendev.org/t/openstack/build/b20efdcf4e174d27ab8c8826fe5c8eb2 : SUCCESS in 2h 04m 32s (non-voting)\n- neutron-functional-with-uwsgi-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/7622839d7d9342f09df07a3c3903387a : SUCCESS in 1h 02m 16s\n- neutron-fullstack-with-uwsgi-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/ec7741e74ead47dd852d52f2b77f9150 : SUCCESS in 2h 49m 57s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/b5ab78922f7448cd8cc2232734c40504 : SUCCESS in 1h 30m 23s\n- neutron-ovs-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/3f9d90a423b149eb8c9c6da8d92f6cd0 : SUCCESS in 1h 09m 00s\n- neutron-ovn-tempest-with-uwsgi-loki https://zuul.opendev.org/t/openstack/build/2e653afb3e26474181896fca27dbacc7 : SUCCESS in 1h 36m 38s\n- openstack-tox-py312-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/a55c0b7e9a1c46feb64b4c03d0319f87 : SUCCESS in 23m 11s\n- openstack-tox-py312-with-oslo-master https://zuul.opendev.org/t/openstack/build/f66eef0ea64641aaa29c2cf621cbc061 : SUCCESS in 24m 40s\n- openstack-tox-py312-with-ovsdbapp-master https://zuul.opendev.org/t/openstack/build/8bf477b278504af297faa5a9ead9d195 : SUCCESS in 18m 09s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/4ec1992ea2004a6381975bedcde53e0b : SUCCESS in 1h 06m 49s\n- neutron-functional-with-uwsgi-fips https://zuul.opendev.org/t/openstack/build/ceb7503800f74abe962b8ada71ad04cc : FAILURE in 59m 37s\n- neutron-functional-with-pyroute2-master https://zuul.opendev.org/t/openstack/build/82929463c22247a1a20a0530a8a0af2c : SUCCESS in 1h 01m 25s\n- neutron-functional-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/d4c48ae5c9e24f488c64bdb93e95ad88 : SUCCESS in 1h 02m 47s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/2bf967f1f44c4074919f8878e49d1fef : FAILURE in 1h 40m 53s\n- neutron-fullstack-with-uwsgi-fips https://zuul.opendev.org/t/openstack/build/08ec3be6a7f34dacac7eb2a0da880c32 : SUCCESS in 1h 26m 21s\n- neutron-fullstack-with-pyroute2-master https://zuul.opendev.org/t/openstack/build/6268302131484fa4ba65d991444b68a2 : SUCCESS in 2h 00m 59s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/0adb07510da7433e90bbfe8ca69f02db : SUCCESS in 1h 02m 05s\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/ec03550920fc460ab8dd0b618432eaa5 : SUCCESS in 1h 37m 10s\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/6f23ee1bfabd4ead85dca5e2bda71e31 : SUCCESS in 1h 57m 49s\n- neutron-ovs-tempest-with-os-ken-master https://zuul.opendev.org/t/openstack/build/52a60eb3c4644eebba7ad12637fb1314 : SUCCESS in 58m 20s\n- neutron-ovn-tempest-postgres-full https://zuul.opendev.org/t/openstack/build/f1c8e5ad435f4954b974ef53d937ccfe : SUCCESS in 1h 02m 04s\n- neutron-ovn-tempest-mariadb-full https://zuul.opendev.org/t/openstack/build/e6d99ba1edc640e89ca7ffc92ffb83cd : SUCCESS in 1h 06m 11s\n- neutron-ovn-tempest-ipv6-only-ovs-master https://zuul.opendev.org/t/openstack/build/e68a27e0ae6a4f998bb9b73dba8fe53c : SUCCESS in 1h 29m 19s\n- neutron-ovn-tempest-ovs-master-centos-9-stream https://zuul.opendev.org/t/openstack/build/1fa635e34de44047a069f5225e430ffd : SUCCESS in 1h 16m 57s\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/f83b5eace3fd428abbb916bb68605e38 : SUCCESS in 1h 42m 47s\n- neutron-ovn-tempest-full-multinode-ovs-lts https://zuul.opendev.org/t/openstack/build/1a1d6a78caad487aa712e16637999553 : SUCCESS in 1h 56m 22s\n- neutron-ovn-tempest-ovs-master https://zuul.opendev.org/t/openstack/build/ef5293d0ef654f639294b36a9431b1c3 : SUCCESS in 1h 11m 12s\n- neutron-ovn-tempest-ovs-lts https://zuul.opendev.org/t/openstack/build/92b90c651d4746f597d453e3757b1902 : SUCCESS in 1h 11m 22s\n- neutron-ovn-tempest-plugin-ovs-master https://zuul.opendev.org/t/openstack/build/3014ced501c54fb98c1f61b594c7f3f7 : SUCCESS in 2h 01m 02s\n- neutron-ovn-tempest-plugin-ovs-lts https://zuul.opendev.org/t/openstack/build/4aadb2b535c545b3a15e138099c4fd48 : SUCCESS in 1h 32m 22s\n- neutron-ovn-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/71968e921c694164a95c9f09319744f9 : SUCCESS in 1h 05m 44s\n- neutron-ovn-tempest-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/87c51a27e1644185a5ae88fe219045f9 : SUCCESS in 1h 09m 14s\n- neutron-ovs-tempest-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/ccb2b8d3ac4e46faad827ca8480177ab : SUCCESS in 57m 57s\n- neutron-ovs-tempest-fips https://zuul.opendev.org/t/openstack/build/a128576085ce49928d368b0047aa6b92 : SUCCESS in 1h 00m 12s\n- neutron-ovn-tempest-ovs-release-fips https://zuul.opendev.org/t/openstack/build/6ab5e2470bea41478aa21384ec5405aa : SUCCESS in 1h 35m 46s\n- devstack-tobiko-neutron https://zuul.opendev.org/t/openstack/build/ca778e1c019d45a4965ba55364e05858 : SUCCESS in 47m 36s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/d89cd7464eee4aac81b4b951434c5fe6 : SUCCESS in 48m 05s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/4c7bfe210a364d9d9f2b92669e4e2e60 : SUCCESS in 22m 35s\n- neutron-linuxbridge-tempest-plugin-nftables https://zuul.opendev.org/t/openstack/build/e8d6c8a845d14e8886dcf7a1a39d82f6 : SUCCESS in 1h 51m 30s\n- neutron-ovs-tempest-plugin-iptables_hybrid-nftables https://zuul.opendev.org/t/openstack/build/6e97d2a128d145f481fc1379d03ca639 : SUCCESS in 1h 44m 27s\n- devstack-enforce-scope https://zuul.opendev.org/t/openstack/build/ca9fadc3c7c04cf18bc73ad158a294d0 : SUCCESS in 17m 52s\n- openstack-tox-py39-with-oslo-master https://zuul.opendev.org/t/openstack/build/9aa0fac4a6964e569253f9206f42e862 : SUCCESS in 25m 55s\n- openstack-tox-py312-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/1bb42389d7a8409bb77a36413bd5cac0 : SUCCESS in 24m 00s\n- neutron-functional-with-oslo-master https://zuul.opendev.org/t/openstack/build/0d89f3cb64624e90a8631f0423a3e6a3 : SUCCESS in 55m 18s\n- neutron-ovs-tempest-with-oslo-master https://zuul.opendev.org/t/openstack/build/f0b336b6daa640e2af75904679e88d4c : SUCCESS in 1h 28m 57s\n- neutron-ovn-tempest-ovs-release-with-oslo-master https://zuul.opendev.org/t/openstack/build/1a4667633c414eafae9920cb745817d1 : SUCCESS in 1h 38m 41s","accounts_in_message":[],"_revision_number":1},{"id":"a99c023681f60777d330714020464de623797c7e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-10-04 06:55:54.000000000","message":"Patch Set 2: Commit message was updated.\n\nOutdated Votes:\n* Verified+1, Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"f97e86c9a9443ebb21001a7e5e639e6b484b5b10","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2024-10-04 07:30:30.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/October-2024/Arista_Tempest_Testrefs/changes/49/931249/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"1ec0ad7332386f4ea2f4d4167188b418134e77ce","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-04 07:33:19.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/421f6f87b6404ccb9c7e7fa5a3773d00\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5f62e9d28d544dad96141b4dce3fb187 : SUCCESS in 23m 56s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/08eb7a81cddb43e2b1ae3538f77992d6 : SUCCESS in 35m 21s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"1b77600219d829ab5a1e2ef61d6dd5fb9d0228d8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-04 09:02:20.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9192a360c3964130aa396c995469e1d3\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/93d6fd7f39f745e1a7d5984be2169b01 : SUCCESS in 1h 29m 43s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/620d48b8deeb4baeaeaddd1c38142ea7 : SUCCESS in 1h 55m 16s\n- neutron-tempest-plugin-ovn-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/8225c849040546dbab83bd10fd425f25 : SUCCESS in 1h 38m 13s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/0116157d496e4c1281049d275a4d005f : SUCCESS in 1h 46m 40s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d2ae546da42d40cd9a9977a92817d9fd : SUCCESS in 26m 42s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/34fae89b0efc421385c79687c6b83a42 : SUCCESS in 35m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/75d44942f07e4a5691d33f047608f42a : SUCCESS in 7m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/027d1baee95d4d20ae4d99329cd41a6e : SUCCESS in 22m 44s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/873789eedc8640a0bdeca2041be13f72 : SUCCESS in 22m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a38c2e1188a34e22a337735dcfd80d08 : SUCCESS in 10m 15s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/30e7193180e640838120cc3f8b6c77b9 : SUCCESS in 48m 23s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/c5743583256f470aa6583107eeec0e0f : SUCCESS in 54m 35s\n- neutron-tox-mypy https://zuul.opendev.org/t/openstack/build/1e190b1737114c06b11c7798d2b9195d : SUCCESS in 4m 59s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/cc03512caf114ad69ee5e3f644047e4a : SUCCESS in 56m 38s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/b9cea7d96c1c44a0b062641ed13132c8 : SUCCESS in 2h 00m 39s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/fdb042f38b1049b980a6c990e54c0c9e : SUCCESS in 1h 01m 57s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/0cfb49212fc741db98b9b4aa615c8fa0 : SUCCESS in 1h 08m 33s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/4f4ff35521f248ae83fd02933d751bc3 : SUCCESS in 1h 28m 32s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/119010a4cf4046c987d3a547681b2cae : SUCCESS in 1h 31m 34s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/2fce4e80a6f4463fa852383578639e8a : SUCCESS in 58m 49s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/1d3a8d068b58490f8a1f56939daeea14 : SUCCESS in 1h 02m 33s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/bf0b7e7c09e44d19b1464567da401d76 : SUCCESS in 1h 05m 41s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/f4d122d87e6645e4bf98c25c44b5896e : SUCCESS in 1h 08m 29s","accounts_in_message":[],"_revision_number":2},{"id":"1508c6aba0c3a4613b3d0b228b767cd9264dc57a","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2024-10-07 14:27:14.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a016ab4bcf3cf0dc5e931ac7384d3ea406a14983","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-10-11 08:16:10.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9cef6b2db1dfd3fa83b8fdc9fd53cd66ca5ae149","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-10-11 15:01:05.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"b7e7a8c37718077485a6eb564a9e1a3062e01688","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-11 15:01:38.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"1c3dc83438c34e8f015595f54905b07b3f61ea29","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-11 17:27:24.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dfb89b949ffd42c29e1c454688fe3207\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/f5927be935764b10b6dac81d749b53d3 : SUCCESS in 1h 30m 45s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/7f6c1e0475f349c2b814a0c6f01099b3 : SUCCESS in 2h 15m 46s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/0048a810c5954c31a87c54fe306409d0 : SUCCESS in 1h 47m 03s\n- neutron-tempest-plugin-ovn-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/4495245638ab4c418977ad7d887a2ee2 : SUCCESS in 2h 24m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/54facbfbb8f240debf7e09e8ca6080ef : SUCCESS in 10m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ec4dc0f91f3d44b38a325212111d39cc : SUCCESS in 27m 46s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/15f2bcec2d1944558448b6e5df1b7243 : SUCCESS in 19m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c4778743691c450dbb5ed5aa3aea85eb : SUCCESS in 12m 59s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/bd0ac5c1606543bd8455e19678bdd022 : SUCCESS in 48m 17s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/ff68590ea1734ec3a6d6f730ce90782d : SUCCESS in 1h 31m 57s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/7eec27fd862747a6b257c0a0c93f1f09 : SUCCESS in 1h 34m 19s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/4e6bfdbd505c461890e9a5281cb868b5 : SUCCESS in 59m 20s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/8f300791777e4bc0a02dfebd37cf7bd1 : SUCCESS in 57m 38s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/6edf6bc963cc40b882e805841845d1ae : SUCCESS in 1h 08m 53s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/77d0ec7f4d92438a87217cc23278601a : SUCCESS in 1h 16m 28s","accounts_in_message":[],"_revision_number":2},{"id":"7851c0ed8018f85025a7870716dab001f970cbcb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-11 17:27:24.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"b3df26025230142db9d2ca0f9f73b952f591f3cd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-11 17:28:35.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/852ee45dd9434e51b1f9a55dfaccc83d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ddbd238b396412b98d5ccddc7847060 : SUCCESS in 47s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"79ad54f6b7695d8e33d984f7dbf947814d0b2aad","revisions":{"a356ebc0116cd1a6f102a25fc6ccd838fea68bf5":{"kind":"REWORK","_number":1,"created":"2024-10-03 08:08:04.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/49/931249/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/49/931249/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/49/931249/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/49/931249/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/49/931249/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/49/931249/1"}}},"commit":{"parents":[{"commit":"f861f98adcb1298ac0a9dd27c0198b47f73e6013","subject":"Merge \"Flip ovn_emit_need_to_frag to True by default\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f861f98adcb1298ac0a9dd27c0198b47f73e6013"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-10-03 07:32:59.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-10-03 07:32:59.000000000","tz":0},"subject":"WIP - Remove the sleep calls in the ``create_or_update_agent`` method","message":"WIP - Remove the sleep calls in the ``create_or_update_agent`` method\n\nRelated-Bug: #2069831\nTopic: eventlet-deprecation\nChange-Id: Iebc895e6c8bd0f1129880244325afa1d6a0181c0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a356ebc0116cd1a6f102a25fc6ccd838fea68bf5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a356ebc0116cd1a6f102a25fc6ccd838fea68bf5"}]},"branch":"refs/heads/master"},"79ad54f6b7695d8e33d984f7dbf947814d0b2aad":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2024-10-04 06:55:54.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/49/931249/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/49/931249/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/49/931249/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/49/931249/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/49/931249/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/49/931249/2"}}},"commit":{"parents":[{"commit":"f861f98adcb1298ac0a9dd27c0198b47f73e6013","subject":"Merge \"Flip ovn_emit_need_to_frag to True by default\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f861f98adcb1298ac0a9dd27c0198b47f73e6013"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-10-03 07:32:59.000000000","tz":0},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2024-10-04 06:55:54.000000000","tz":0},"subject":"Remove the sleep calls in the ``create_or_update_agent`` method","message":"Remove the sleep calls in the ``create_or_update_agent`` method\n\nA Neutron API call thread should not yield until the result return\nand finalization.\n\nRelated-Bug: #2069831\nTopic: eventlet-deprecation\nChange-Id: Iebc895e6c8bd0f1129880244325afa1d6a0181c0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/79ad54f6b7695d8e33d984f7dbf947814d0b2aad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/79ad54f6b7695d8e33d984f7dbf947814d0b2aad"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
