)]}'
{"id":"openstack%2Fneutron~931864","triplet_id":"openstack%2Fneutron~stable%2F2024.1~Iff86599678ebdd5189d5b56d11f3373c9b138562","project":"openstack/neutron","branch":"stable/2024.1","topic":"bug/2080199-stable/2024.2-stable/2024.1","attention_set":{},"removed_from_attention_set":{"13861":{"account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"last_update":"2024-10-09 16:53:31.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iff86599678ebdd5189d5b56d11f3373c9b138562","subject":"[CI] Functional: Increase Ulimit to 4096","status":"MERGED","created":"2024-10-09 04:02:13.000000000","updated":"2024-10-09 16:55:02.000000000","submitted":"2024-10-09 16:53:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"931864-bug/2080199-stable/2024.2-stable/2024.1","cherry_pick_of_change":931863,"cherry_pick_of_patch_set":1,"meta_rev_id":"8093f70f2c37657fad6a4b8d6e3c6a54b401da86","_number":931864,"virtual_id_number":931864,"owner":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-10-09 16:53:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2024-10-09 06:00:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2024-10-09 14:48:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2024-10-09 14:48:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-09 04:02:13.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"CC"},{"updated":"2024-10-09 05:41:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-10-09 06:00:43.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2024-10-09 14:48:38.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"}],"messages":[{"id":"cb54d92b05624a69a641f5c9cb2c44c99a5fcfaf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2024-10-09 04:02:13.000000000","message":"Patch Set 1: Cherry Picked from branch stable/2024.2.","accounts_in_message":[],"_revision_number":1},{"id":"779817c40586811c0bc967c9310c808ab6c606ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-09 05:41:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6eb8a4a405d9498a878aa0333f6bb13e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4ddd89e4c3b3462d8df4a9209d1cf939 : SUCCESS in 10m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e5a4af08e01d4d899ffdbe8541f1c4fe : SUCCESS in 10m 44s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/415ef437881540dab7aedf05f04d464d : SUCCESS in 48m 14s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/e03686a83f7c4c1bb611321eada78814 : SUCCESS in 1h 33m 57s","accounts_in_message":[],"_revision_number":1},{"id":"aa0a26c6dbf23eecf754737372d752b8f0ae758f","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-10-09 06:00:43.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"55fcaffbef5c0f661c58e1e7619221b3164b8408","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-10-09 14:48:38.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"53d007963d5f6e0474e226d35c15582c27c5e51d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-09 14:49:11.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3314d274f850931cbdb1da35c13a6a919216aaa4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-09 14:49:47.000000000","message":"Patch Set 1:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"49f6ec2dec2cf49318aa7debec164c5af5d5826e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-09 16:53:30.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/61e4950289784a9e9061766a2dd63cd1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7a00f5f07b6f45cdb91361b4a0ea573c : SUCCESS in 9m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ca859f20ccb4648b344a806cd9d4811 : SUCCESS in 10m 30s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/0e0179a65ce8418da5061f77d3627069 : SUCCESS in 58m 43s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/5e99f590e6ed44b2ba295958f878df40 : SUCCESS in 1h 39m 17s","accounts_in_message":[],"_revision_number":1},{"id":"922d3172449b090a4aa7c628d99cd7a0717851e7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-09 16:53:31.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"8093f70f2c37657fad6a4b8d6e3c6a54b401da86","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-09 16:55:02.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/055481effc874ff2841d32846fcd2b43\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cfb033ee7cf24dc8bd4959154b3a14c0 : SUCCESS in 1m 11s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d8a6552b11b2c642bb74507cb83c7fd5d84698af","revisions":{"d8a6552b11b2c642bb74507cb83c7fd5d84698af":{"kind":"REWORK","_number":1,"created":"2024-10-09 04:02:13.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/64/931864/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/64/931864/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/64/931864/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/64/931864/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/64/931864/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/64/931864/1"}}},"commit":{"parents":[{"commit":"5561387e16f528824c0bd35511fa0ca493e47e76","subject":"Merge \"Correct logic error when associating FIP with OVN LB\" into stable/2024.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5561387e16f528824c0bd35511fa0ca493e47e76"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2024-09-10 07:36:32.000000000","tz":120},"committer":{"name":"yatin","email":"ykarel@redhat.com","date":"2024-10-09 04:02:13.000000000","tz":0},"subject":"[CI] Functional: Increase Ulimit to 4096","message":"[CI] Functional: Increase Ulimit to 4096\n\nFunctional tests started to fail with\n\"Too many open files\" randomly, the default ulimit in\nOS is configured to 1024, increasing this to 4096\nto avoid these random failures.\n\nCloses-Bug: #2080199\nChange-Id: Iff86599678ebdd5189d5b56d11f3373c9b138562\n(cherry picked from commit 6970f39a49b83f279b9e0479f7637d03a123a40e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d8a6552b11b2c642bb74507cb83c7fd5d84698af"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d8a6552b11b2c642bb74507cb83c7fd5d84698af"}]},"branch":"refs/heads/stable/2024.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
