)]}'
{"id":"openstack%2Fneutron~932041","triplet_id":"openstack%2Fneutron~master~Ibd032ea77c5bfbc1fa80b3b3ee9ba7d5c36bb1bc","project":"openstack/neutron","branch":"master","topic":"bug/2083682","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2024-10-14 16:26:18.000000000","reason":"Change was submitted"},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2024-10-14 10:50:02.000000000","reason":"\u003cGERRIT_ACCOUNT_11975\u003e replied on the change","reason_account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},"9656":{"account":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"last_update":"2024-10-14 16:26:18.000000000","reason":"Change was submitted"},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2024-10-14 16:26:18.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ibd032ea77c5bfbc1fa80b3b3ee9ba7d5c36bb1bc","subject":"Optimize the SG rule retrieval","status":"MERGED","created":"2024-10-10 09:18:16.000000000","updated":"2024-10-14 16:27:42.000000000","submitted":"2024-10-14 16:26:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"932041-bug/2083682","meta_rev_id":"6faa35b7ccdea2ac619463a08dc7a593e3a2110a","_number":932041,"virtual_id_number":932041,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2024-10-11 14:08:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-10-14 16:26:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2024-10-14 10:50:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2024-10-14 13:43:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":1,"date":"2024-10-14 13:43:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"}],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-10 09:52:20.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-10-10 11:28:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-10-10 11:48:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-10-10 12:32:47.000000000","updated_by":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"reviewer":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"state":"REVIEWER"},{"updated":"2024-10-10 14:47:27.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2024-10-11 07:20:32.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2024-10-14 13:43:49.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"}],"messages":[{"id":"a3ddc27c40012030f5dc0cc3f8d7226cda702be1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-10-10 09:18:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d1f7f49f9b36e824adb46875dad04cc05ff4fe15","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2024-10-10 09:52:20.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/October-2024/Arista_Tempest_Testrefs/changes/41/932041/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"b83900c2c87e0329e9abd9d4017f083a7a054a4c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-10 11:28:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c61691c7aa1841e9a1175e279b3012d5\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ebea15b471974aef810e6eac3f348167 : FAILURE in 31m 09s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/4349084ef6c74bc091b548d91dac0a54 : FAILURE in 38m 03s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"508a7351abc79fbdc3db40d7a34a5fa28e628ab7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-10 11:48:18.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4d611cf7262f47f48841f26a35d142ce\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/aca395ea3bd04156b7f3c12a37c3d5c2 : SUCCESS in 1h 27m 40s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/87d6f0fae74543bdb30c35d6ac488bc0 : SUCCESS in 2h 00m 53s\n- neutron-tempest-plugin-ovn-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/44f716dfa5154a60b92b81ee845f0b32 : SUCCESS in 1h 44m 22s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/2e06ad1fb18744ca967b7e8098c22827 : SUCCESS in 1h 41m 43s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/7ef1794c3f1d488eb05e76b3cf3e4e0e : SUCCESS in 29m 42s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3faa90ca054e4fcd89175dd834e8ba6d : FAILURE in 37m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d26d01dcc6224f1db8cfdde356a2a750 : FAILURE in 5m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dbdd8f1770c54c7a8c2b8536e50faf1b : FAILURE in 24m 02s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/78ddafa741444bdb996616d5ac53458e : RETRY_LIMIT in 1m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c538b06dde3c49f893bc2bd99d8272a7 : RETRY_LIMIT in 1m 23s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/3ce50fe112e644aba4eb97d94b1b21f8 : SUCCESS in 57m 08s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/9deab3d165444adabbf45896f90462c3 : SUCCESS in 1h 15m 07s\n- neutron-tox-mypy https://zuul.opendev.org/t/openstack/build/e8e9104e63b742a7ba3c65edfd714368 : RETRY_LIMIT in 1m 15s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/4b0f90238ecd4ded813b3003fe4fdf4b : FAILURE in 50m 53s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/b2e5d860ef224b1fa8e835739d7f60f7 : FAILURE in 1h 32m 55s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/1c024b724e414a488b83aa98c572190d : SUCCESS in 50m 23s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/d9342188cf7a4ff1a19bc28d6efda386 : SUCCESS in 1h 02m 02s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/10868c9d53be4094a1ebe224996f0720 : SUCCESS in 1h 38m 59s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/f39746ebb9b3416d82396de13a401e1f : SUCCESS in 1h 19m 01s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/ffb7a0abfd354f1db66f89e29a5fcd2f : SUCCESS in 1h 04m 28s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/76d2b7d836684f09abae48e1e3c2299c : SUCCESS in 1h 08m 14s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/5859bea866d94a5fa5fe2aa454711a04 : SUCCESS in 1h 02m 33s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/2dce72f1fb304183b48125815293a56c : SUCCESS in 1h 14m 29s","accounts_in_message":[],"_revision_number":1},{"id":"1dedfb6ce2231a5b3c0b9279444f94b2fe1454a1","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-10-10 14:47:27.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"3a9b7e378f5c808edf3f89c441d991aebeec1dee","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-10-11 05:52:33.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1, Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"e89d3040d3cdb126d4d93405e264ea03b440e99c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-10-11 05:53:38.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"1169ba9a6e8bf0b11adfb5953f623c3ce21b2737","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-10-11 06:00:25.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"97b2113e7c97dcc867801ef4d5eda105d5a9fce0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-10-11 06:00:42.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"2fdfa9c665260f05d263617f62e8bf73656a548d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-10-11 06:10:12.000000000","message":"Uploaded patch set 5: New patch set was added with same tree, parent tree, and commit message as Patch Set 4.","accounts_in_message":[],"_revision_number":5},{"id":"8a494ca718435843c42de60cd615958a721eb322","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2024-10-11 06:25:45.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/October-2024/Arista_Tempest_Testrefs/changes/41/932041/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"926f09a05aa9cca8bd7341a03ff77a16c1f647e9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-11 06:36:22.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0d55ddedebae4d6bb944d8c2486c0cfe\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6ef6faa0a9524ec680a95847d23246fd : SUCCESS in 21m 10s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/c9e4ef066e1046688f28c3f02eef43e5 : SUCCESS in 24m 01s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"827a648da0260b15c805a2375f3160bd4ba74218","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2024-10-11 06:58:26.000000000","message":"Patch Set 3: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/October-2024/Arista_Tempest_Testrefs/changes/41/932041/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"91e009138365e73a75c1597e806613fd8766d55f","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2024-10-11 07:20:32.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"1ae59f315736c120d99463f9f9a6d5740aa4639b","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2024-10-11 07:56:21.000000000","message":"Patch Set 4: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/October-2024/Arista_Tempest_Testrefs/changes/41/932041/4 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":4},{"id":"e3fc5cf251e7114ed77fa76424e267a64e8ebe68","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-11 08:27:16.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d89aa42adf2e45d69e5e4ffc3341d882\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/83018dd93b4d4300982ca39f785331c3 : FAILURE in 1h 50m 26s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/c322cb5369234500a9a1426e625566f5 : FAILURE in 2h 15m 58s\n- neutron-tempest-plugin-ovn-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/022576a370f348518539d12c3b5c098d : FAILURE in 2h 14m 48s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/a6ebc04e8c02401781d4ac16e37c450d : FAILURE in 2h 14m 27s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/2105c445922744738584f5a9b21f340f : SUCCESS in 37m 45s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6bc2ea5b6d99439cacd5d3bca53be6b4 : SUCCESS in 48m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1b7df71daa514fa5a0a3aabff1ac9d61 : SUCCESS in 7m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/88474fcaf3fb462e9cc523facd957e63 : SUCCESS in 24m 13s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e68b470d691e40739c529c3dd8c45bdb : SUCCESS in 22m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d3335c3bf97d441ebf9453a8ee7f440a : SUCCESS in 10m 33s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/867b9eed84e047beadd9eb8b9970b6a9 : SUCCESS in 1h 02m 11s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/bc58d00d0c1a4cdfa3ec8935e8b0f900 : SUCCESS in 1h 13m 19s\n- neutron-tox-mypy https://zuul.opendev.org/t/openstack/build/af5e269eb0ec4442ac9459355897bd77 : SUCCESS in 4m 22s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/9c5605b5f00d43f0bb03e227cfe90347 : SUCCESS in 53m 00s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/ac530467a36d40c9a81c2fb6b6661318 : SUCCESS in 1h 54m 37s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/429e446830be4721a8baba93fd5936b2 : SUCCESS in 1h 05m 59s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/c4299d219d4040e0805f4b0765d63fe2 : SUCCESS in 1h 08m 28s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/cb77ee50b0ce467eac88d3c726148214 : SUCCESS in 1h 24m 55s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/bdcee844065e45ee91d815846b773ca3 : SUCCESS in 1h 23m 11s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/9f83e713b6184a3388bc2176464e492f : SUCCESS in 1h 03m 26s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/8cd63d956ed74cb8bae66d7a7c9c3907 : SUCCESS in 55m 14s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/283860d8926b4ee7a69dcc903f93370a : SUCCESS in 57m 58s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/9e41def655a548b1b6dd9b9802b8b485 : SUCCESS in 1h 11m 03s","accounts_in_message":[],"_revision_number":5},{"id":"2feddee066add53da155f772db57157b48be9d8b","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2024-10-11 08:29:15.000000000","message":"Patch Set 5: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/October-2024/Arista_Tempest_Testrefs/changes/41/932041/5 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":5},{"id":"37ad85859f42d99aa8999514159e6b5d2b6579b7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-10-11 10:00:39.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1, Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"2b231fd6e56a878cf8189edd2c0bfb13c17bd932","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-11 10:44:10.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ceababc802ad43b1a37384ac3bccb9c9\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/93ebe18629564fd3b3c6ffcdba32a34f : SUCCESS in 31m 50s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/dcea440b3daf4da0ab3479bd5cd0cb2e : SUCCESS in 40m 55s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"081a2cdd12d71b02203268a2d7d5fd41c0119878","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2024-10-11 11:45:38.000000000","message":"Patch Set 6: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/October-2024/Arista_Tempest_Testrefs/changes/41/932041/6 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":6},{"id":"4649024475aec1f9954b7386240bd2d97ce42ad0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-11 12:50:23.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b7b8dc6f8dfb4fbc9e2d7a2b98662ec2\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/735c21ed57c54e2ca672acd320c9793d : POST_FAILURE in 1h 32m 33s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/c735ba779076457c886a59423b2abb3a : FAILURE in 2h 03m 19s\n- neutron-tempest-plugin-ovn-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/dc41fa11723142ffbb2d8c1ff63fc4ed : POST_FAILURE in 1h 50m 47s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/576e99a5f08748969d1c2f1c68ad68d9 : FAILURE in 2h 04m 14s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d33609bf0c6f48889ae9781ff9094d05 : SUCCESS in 32m 55s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/01fdb629763b4c5bafd0afbced65c625 : SUCCESS in 28m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0b223de04ffb4fcd92e9ada65bdb29f9 : SUCCESS in 8m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3fad6129c22f4bb78a77495f4322df5a : SUCCESS in 19m 07s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/dc0391f6ca514d3cba2c5816167f128d : POST_FAILURE in 23m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/89d460712d4b4a6ab3a67d45da184eb4 : SUCCESS in 10m 35s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/93d6a0c7672a44fea57edc9df1f2e1cb : SUCCESS in 57m 44s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/e118c75bb5f3497cba168707ae86275e : SUCCESS in 1h 11m 58s\n- neutron-tox-mypy https://zuul.opendev.org/t/openstack/build/523e2c0bf59344a9a664b4dd16ae3dbe : SUCCESS in 6m 43s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/c77a7745d0314bc6b2e277174f7c7dda : SUCCESS in 58m 42s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/683ce088188b4c4bbf3ce24a13e6fea6 : SUCCESS in 2h 30m 50s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/a6d0a483ad484ecc9dea625da57c2937 : POST_FAILURE in 1h 04m 32s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/a53e47a0868542b0809d079a29b14c92 : SUCCESS in 1h 14m 19s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/2bf9f1cf33264bf2b7e7cc3fa97c507b : SUCCESS in 1h 32m 15s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/6bab8b5fd1c24b93b173450e27c3a564 : POST_FAILURE in 1h 32m 31s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/4454a1d21ff34cbeb600b1137a213c87 : SUCCESS in 1h 07m 00s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/a749a0717cb14109acd5a953921cba8a : SUCCESS in 55m 46s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/33b96cf24b3b4c6fb548c1d70fbf6c58 : SUCCESS in 1h 03m 12s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/d5e925d13ddb462b8c71d2387a0584f1 : POST_FAILURE in 1h 20m 21s","accounts_in_message":[],"_revision_number":6},{"id":"0027ba1785826bcd2ade9da3ce14192d338eb2fa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-10-11 13:35:22.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified+1, Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"ca56017287890fdc268631e1ae87b855ff38a50f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-11 14:00:41.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e60a4766c5a04766b1ca5f25456ac338\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/cc404d51808f4725b9b222823f9526c1 : SUCCESS in 21m 24s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/46e59a0d987943dda4c7fbfd4c0391ad : SUCCESS in 23m 32s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"ea36919fa4a2d923339f821e2eb49328e7a310d8","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2024-10-11 14:08:24.000000000","message":"Patch Set 7: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/October-2024/Arista_Tempest_Testrefs/changes/41/932041/7 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":7},{"id":"a982f809ff461819c8daef90486388f1a31f7503","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-11 16:29:30.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f2150b93754742f4915663b55276d17f\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/3d06e6ce347e4f8b9e7ef6d0cd08efae : SUCCESS in 1h 46m 34s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/37d74a1dd7594835b86e93610bd5b067 : SUCCESS in 2h 40m 09s\n- neutron-tempest-plugin-ovn-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/2bd936091b1a4989b79cb179e9e00372 : SUCCESS in 1h 41m 39s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/db2d92d320ee4876983a2a5d8aa889f8 : FAILURE in 2h 45m 58s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/7784f5aba0cf4414befd371486c9d5e3 : SUCCESS in 36m 41s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8101fc93cd6b4a7b8d1fe2d9bef0b546 : SUCCESS in 42m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/17651087342a4d958930dccaa6b55f49 : SUCCESS in 10m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5df767ef4db64148a86ebb3f82b8aa26 : SUCCESS in 24m 15s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d29d7812ba3a4668a63cc00cb93bd676 : SUCCESS in 21m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f03f14da3a67406c80ff90ee871f3059 : SUCCESS in 9m 39s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/e525e635b66640ccb8c7189d2e18b3dd : SUCCESS in 1h 01m 04s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/d2a2d13ca12a43dc9dacf278edb085fd : SUCCESS in 54m 04s\n- neutron-tox-mypy https://zuul.opendev.org/t/openstack/build/056dd2340b0249a3b23dff06bcac15df : SUCCESS in 4m 41s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/7464480bf92e45aab596bf0734966ea2 : SUCCESS in 56m 26s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/4ea20a91fd4c472f85387d667eee6fdc : SUCCESS in 1h 53m 57s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/8206c885fd6a4d2a8bf00763d32da299 : SUCCESS in 49m 27s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/4e107e6310c64393830cd339c778ec61 : SUCCESS in 1h 13m 36s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/88736c2d418e469aa44463180b3123d0 : SUCCESS in 1h 23m 33s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/53a165f2b7bc4ca1ba51703152d9e6fb : SUCCESS in 1h 21m 14s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/3747983bc40b4ebdb0cf43d5f21aa745 : SUCCESS in 1h 22m 37s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/14535f73eddb4bf8a3258473cf18b208 : SUCCESS in 49m 05s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/17e63e918daa4d94b643a8f1172156fe : SUCCESS in 59m 47s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/c0084cc532fa4290b49bf03bb859d5f9 : SUCCESS in 1h 26m 26s","accounts_in_message":[],"_revision_number":7},{"id":"02e4faac3a4a0d43121a7bedf41634ffa0f4e711","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-10-14 07:54:25.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"f8eeefb9fbbfd1d6d6c289d1149d5907b8ec9971","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-14 08:30:50.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9f5b22b73cb547f5b6acfcad01a4b497\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c66b43659a124195be474e764c4054b1 : SUCCESS in 20m 36s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/1fbc155b99ea43f9b167024d5b586528 : SUCCESS in 34m 11s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"f6274779932cad1126172654d698a39029257f87","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-14 10:15:20.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cf09ed7d43474de1a2ebd55a372169e6\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/75ac79bc56944e8f970f51135f15783a : SUCCESS in 1h 46m 52s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/55268763cd324676bc07b6910e60d917 : SUCCESS in 2h 19m 10s\n- neutron-tempest-plugin-ovn-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/b74f77e5e7db4fa9be7281ae72aedb62 : SUCCESS in 2h 02m 20s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/8aafb59c24604261bef475682ade1a04 : SUCCESS in 2h 12m 43s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/88b49108bdf94811ac11500bdd42750a : SUCCESS in 31m 54s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ce326fcc910c4a14b891706a9a151db1 : SUCCESS in 46m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4e60cf95254e43df889965561d0c3477 : SUCCESS in 10m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fa9302172be74f298b5e8f750cec2374 : SUCCESS in 22m 41s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/149398698dba473cb44290d1f79dfd39 : SUCCESS in 22m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca954569062743ffa7101a8bfeafdb7e : SUCCESS in 9m 57s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/6cc3c21c53b14ad0bd95994a763f80da : SUCCESS in 1h 01m 14s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/daa7cb2fa4cd41bd8f8b20c9c8f5c9b8 : SUCCESS in 1h 06m 21s\n- neutron-tox-mypy https://zuul.opendev.org/t/openstack/build/5d10aff0f91e45e3914cf717a1853c1f : SUCCESS in 4m 40s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/0779cc0bfab54cdb94b1f7ce80a9ae6d : SUCCESS in 58m 13s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/8557be03e4004ebcb86258997c48d836 : SUCCESS in 1h 55m 26s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/a6cd579197e945618f850b9f247a28e9 : SUCCESS in 36m 54s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/5dd2b040fa4a43eba573bce95dd98b56 : SUCCESS in 55m 54s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/1bf54ed2a11a4566846afa382a2c00b7 : SUCCESS in 1h 39m 03s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/2549f537128e4ad9ba907baca99a9a26 : SUCCESS in 1h 10m 07s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/60e1b8138f8c4d769558f7af01e72656 : SUCCESS in 57m 40s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/e8f6b94a014b4eb1b8dac061fe7f5e34 : SUCCESS in 53m 17s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/00579a508ae44f3a9d0d421d26809b8f : SUCCESS in 1h 01m 45s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/f35e76c876e4429cb65dca1699e1cb7e : SUCCESS in 1h 13m 37s","accounts_in_message":[],"_revision_number":7},{"id":"6edcff3618f77539931d687385ed0e7e4893ae74","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2024-10-14 10:50:02.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"917ff26bb80506704c54263c7df5c9eb935dea0a","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-10-14 13:43:49.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"cc9f78f5688aed277d7bf09ff111b1653251bb82","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-14 13:44:20.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"d442c1f52d835133cc3a80ef1620444d8e885dc3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-14 16:26:18.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1cdae7ed98f44589983796c0c0ee3d99\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/0c5d29d827e2435bb5f26d3188fea9ef : SUCCESS in 1h 48m 14s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/8338aed829414c9fa763601b97f900d3 : SUCCESS in 2h 19m 33s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/38cf4dce37e84bf5a8930c79ad47b809 : SUCCESS in 2h 40m 05s\n- neutron-tempest-plugin-ovn-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/13bea3775c9c491a8be6f386dfc52cd8 : SUCCESS in 2h 21m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7c33850c0c654204a40a03a08bb2fb5b : SUCCESS in 10m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8ed62d99cf614c96b5e396d5b991e3da : SUCCESS in 24m 06s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/347d48d5ac30409483035c908e3e0a0b : SUCCESS in 22m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/247d480e06734f48af1454f294051de2 : SUCCESS in 9m 28s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/885aa9352a8f4557a4f9b42e6b137c3b : SUCCESS in 45m 12s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/47f9ea9eacb54b30bd4a11b43f0aad3c : SUCCESS in 1h 42m 04s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/d282300a86c0425886c7ff0c3f66b46c : SUCCESS in 1h 10m 52s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/d7d55e31f5dc458593994908bdcf0be8 : SUCCESS in 59m 16s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/6d8f820632c24af7872fbcbc9a7fc23f : SUCCESS in 55m 27s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/a0ce6e8ac7274ce08794a7530a66b6bd : SUCCESS in 1h 08m 01s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/5d91b266bc2c4af4aad37d1d1531919f : SUCCESS in 1h 20m 53s","accounts_in_message":[],"_revision_number":7},{"id":"21916cb96b9845ba647a23ffbc2d5eb38579d064","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-14 16:26:18.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":7},{"id":"6faa35b7ccdea2ac619463a08dc7a593e3a2110a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-14 16:27:42.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6d0d76a8fcd7434eadfb24e6b9a0b348\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bebc1078e73d48738db12572cd110a0d : SUCCESS in 1m 03s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"adbc3e23b7d2251cc7de088e2a757674a41c2f6a","revisions":{"86a26baecfb09e67c2b6e7c5aaf1ff7176acf690":{"kind":"REWORK","_number":1,"created":"2024-10-10 09:18:16.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/41/932041/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/41/932041/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/41/932041/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/41/932041/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/41/932041/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/41/932041/1"}}},"commit":{"parents":[{"commit":"9ec96f5e68f7fd7402889987cf9d60a6ffb65ec1","subject":"Merge \"Use the declarative attribute ``standard_attr_id``\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9ec96f5e68f7fd7402889987cf9d60a6ffb65ec1"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-10-10 08:49:44.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-10-10 09:18:03.000000000","tz":0},"subject":"Optimize the SG rule retrieval","message":"Optimize the SG rule retrieval\n\nThere are some operations where the SG DB object can be used instead of\nthe SG OVO. That saves conversion time, including the conversion of the\nSG rule OVOs, that are child resources of the SG OVO.\n\nThis optimization applies to the following methods:\n* SecurityGroupDbMixin.create_security_group\n* SecurityGroupDbMixin.get_security_group\n* SecurityGroupDbMixin.get_security_groups\n* SecurityGroupDbMixin.update_security_group (partially)\n\nThe Nova query to retrieve the SG list in the \"server list\" command,\nhas been benchmarked. The testing environment had a single SG with\n250 SG rules. Call:\n  \"GET /networking/v2.0/security-groups?id\u003d81f64aa4-2cea-46db-8fea-cd944f106aab\n     \u0026fields\u003did\u0026fields\u003dname HTTP/1.1\"\n\n* Without this patch: around 1.25 seconds\n* With this patch: around 0.025 second (50x improvement).\n\nCloses-bug: #2083682\nChange-Id: Ibd032ea77c5bfbc1fa80b3b3ee9ba7d5c36bb1bc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/86a26baecfb09e67c2b6e7c5aaf1ff7176acf690"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/86a26baecfb09e67c2b6e7c5aaf1ff7176acf690"}]},"branch":"refs/heads/master"},"98bd8a0e1a23b885840b907f492ffb4640b5aebd":{"kind":"REWORK","_number":2,"created":"2024-10-11 05:52:33.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/41/932041/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/41/932041/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/41/932041/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/41/932041/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/41/932041/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/41/932041/2"}}},"commit":{"parents":[{"commit":"92550c9d66006e6f6d64e759ea0e32dc7c9ec3f5","subject":"Merge \"Deprecate ovn_emit_need_to_frag option\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/92550c9d66006e6f6d64e759ea0e32dc7c9ec3f5"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-10-10 08:49:44.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-10-11 05:50:09.000000000","tz":0},"subject":"Optimize the SG rule retrieval","message":"Optimize the SG rule retrieval\n\nThere are some operations where the SG DB object can be used instead of\nthe SG OVO. That saves conversion time, including the conversion of the\nSG rule OVOs, that are child resources of the SG OVO.\n\nThis optimization applies to the following methods:\n* SecurityGroupDbMixin.get_security_groups\n* SecurityGroupDbMixin.update_security_group (partially)\n\nThe Nova query to retrieve the SG list in the \"server list\" command,\nhas been benchmarked. The testing environment had a single SG with\n250 SG rules. Call:\n  \"GET /networking/v2.0/security-groups?id\u003d81f64aa4-2cea-46db-8fea-cd944f106aab\n     \u0026fields\u003did\u0026fields\u003dname HTTP/1.1\"\n\n* Without this patch: around 1.25 seconds\n* With this patch: around 0.025 second (50x improvement).\n\nCloses-bug: #2083682\nChange-Id: Ibd032ea77c5bfbc1fa80b3b3ee9ba7d5c36bb1bc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/98bd8a0e1a23b885840b907f492ffb4640b5aebd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/98bd8a0e1a23b885840b907f492ffb4640b5aebd"}]},"branch":"refs/heads/master"},"e1e7763968f3443d28b500c1da2fb874da6a5ce7":{"kind":"REWORK","_number":3,"created":"2024-10-11 05:53:38.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/41/932041/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/41/932041/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/41/932041/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/41/932041/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/41/932041/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/41/932041/3"}}},"commit":{"parents":[{"commit":"92550c9d66006e6f6d64e759ea0e32dc7c9ec3f5","subject":"Merge \"Deprecate ovn_emit_need_to_frag option\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/92550c9d66006e6f6d64e759ea0e32dc7c9ec3f5"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-10-10 08:49:44.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-10-11 05:53:31.000000000","tz":0},"subject":"Optimize the SG rule retrieval","message":"Optimize the SG rule retrieval\n\nThere are some operations where the SG DB object can be used instead of\nthe SG OVO. That saves conversion time, including the conversion of the\nSG rule OVOs, that are child resources of the SG OVO.\n\nThis optimization applies to the following methods:\n* SecurityGroupDbMixin.get_security_groups\n* SecurityGroupDbMixin.update_security_group (partially)\n\nThe Nova query to retrieve the SG list in the \"server list\" command,\nhas been benchmarked. The testing environment had a single SG with\n250 SG rules. Call:\n  \"GET /networking/v2.0/security-groups?id\u003d81f64aa4-2cea-46db-8fea-cd944f106aab\n     \u0026fields\u003did\u0026fields\u003dname HTTP/1.1\"\n\n* Without this patch: around 1.25 seconds\n* With this patch: around 0.025 second (50x improvement).\n\nCloses-bug: #2083682\nChange-Id: Ibd032ea77c5bfbc1fa80b3b3ee9ba7d5c36bb1bc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e1e7763968f3443d28b500c1da2fb874da6a5ce7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e1e7763968f3443d28b500c1da2fb874da6a5ce7"}]},"branch":"refs/heads/master"},"120373c57b09376833ef540731b7a8f8414cb0c7":{"kind":"REWORK","_number":4,"created":"2024-10-11 06:00:42.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/41/932041/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/41/932041/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/41/932041/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/41/932041/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/41/932041/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/41/932041/4"}}},"commit":{"parents":[{"commit":"92550c9d66006e6f6d64e759ea0e32dc7c9ec3f5","subject":"Merge \"Deprecate ovn_emit_need_to_frag option\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/92550c9d66006e6f6d64e759ea0e32dc7c9ec3f5"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-10-10 08:49:44.000000000","tz":0},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2024-10-11 06:00:42.000000000","tz":0},"subject":"Optimize the SG rule retrieval","message":"Optimize the SG rule retrieval\n\nThere are some operations where the SG DB object can be used instead of\nthe SG OVO. That saves conversion time, including the conversion of the\nSG rule OVOs, that are child resources of the SG OVO.\n\nThis optimization applies to the following methods:\n* SecurityGroupDbMixin.get_security_groups\n* SecurityGroupDbMixin.update_security_group (partially)\n\nThe Nova query to retrieve the SG list in the \"server list\" command,\nhas been benchmarked. The testing environment had a single SG with\n250 SG rules. Call:\n  \"GET /networking/v2.0/security-groups?id\u003d81f64aa4-2cea-46db-8fea-cd944f106aab\n     \u0026fields\u003did\u0026fields\u003dname HTTP/1.1\"\n\n* Without this patch: around 1.25 seconds\n* With this patch: around 0.025 second (50x improvement).\n\nCloses-bug: #2083682\nChange-Id: Ibd032ea77c5bfbc1fa80b3b3ee9ba7d5c36bb1bc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/120373c57b09376833ef540731b7a8f8414cb0c7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/120373c57b09376833ef540731b7a8f8414cb0c7"}]},"branch":"refs/heads/master"},"140c9d02927639c9b4323914809946e6f2b42265":{"kind":"NO_CHANGE","_number":5,"created":"2024-10-11 06:10:12.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/41/932041/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/41/932041/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/41/932041/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/41/932041/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/41/932041/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/41/932041/5"}}},"commit":{"parents":[{"commit":"92550c9d66006e6f6d64e759ea0e32dc7c9ec3f5","subject":"Merge \"Deprecate ovn_emit_need_to_frag option\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/92550c9d66006e6f6d64e759ea0e32dc7c9ec3f5"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-10-10 08:49:44.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-10-11 06:01:08.000000000","tz":0},"subject":"Optimize the SG rule retrieval","message":"Optimize the SG rule retrieval\n\nThere are some operations where the SG DB object can be used instead of\nthe SG OVO. That saves conversion time, including the conversion of the\nSG rule OVOs, that are child resources of the SG OVO.\n\nThis optimization applies to the following methods:\n* SecurityGroupDbMixin.get_security_groups\n* SecurityGroupDbMixin.update_security_group (partially)\n\nThe Nova query to retrieve the SG list in the \"server list\" command,\nhas been benchmarked. The testing environment had a single SG with\n250 SG rules. Call:\n  \"GET /networking/v2.0/security-groups?id\u003d81f64aa4-2cea-46db-8fea-cd944f106aab\n     \u0026fields\u003did\u0026fields\u003dname HTTP/1.1\"\n\n* Without this patch: around 1.25 seconds\n* With this patch: around 0.025 second (50x improvement).\n\nCloses-bug: #2083682\nChange-Id: Ibd032ea77c5bfbc1fa80b3b3ee9ba7d5c36bb1bc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/140c9d02927639c9b4323914809946e6f2b42265"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/140c9d02927639c9b4323914809946e6f2b42265"}]},"branch":"refs/heads/master"},"7a91ccebc52e40a78b70639b6a97c5a7ab76edfb":{"kind":"REWORK","_number":6,"created":"2024-10-11 10:00:39.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/41/932041/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/41/932041/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/41/932041/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/41/932041/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/41/932041/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/41/932041/6"}}},"commit":{"parents":[{"commit":"92550c9d66006e6f6d64e759ea0e32dc7c9ec3f5","subject":"Merge \"Deprecate ovn_emit_need_to_frag option\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/92550c9d66006e6f6d64e759ea0e32dc7c9ec3f5"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-10-10 08:49:44.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-10-11 10:00:28.000000000","tz":0},"subject":"Optimize the SG rule retrieval","message":"Optimize the SG rule retrieval\n\nThere are some operations where the SG DB object can be used instead of\nthe SG OVO. That saves conversion time, including the conversion of the\nSG rule OVOs, that are child resources of the SG OVO.\n\nThis optimization applies to the following methods:\n* SecurityGroupDbMixin.get_security_groups\n* SecurityGroupDbMixin.update_security_group (partially)\n\nThe Nova query to retrieve the SG list in the \"server list\" command,\nhas been benchmarked. The testing environment had a single SG with\n250 SG rules. Call:\n  \"GET /networking/v2.0/security-groups?id\u003d81f64aa4-2cea-46db-8fea-cd944f106aab\n     \u0026fields\u003did\u0026fields\u003dname HTTP/1.1\"\n\n* Without this patch: around 1.25 seconds\n* With this patch: around 0.025 second (50x improvement).\n\nCloses-bug: #2083682\nChange-Id: Ibd032ea77c5bfbc1fa80b3b3ee9ba7d5c36bb1bc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7a91ccebc52e40a78b70639b6a97c5a7ab76edfb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7a91ccebc52e40a78b70639b6a97c5a7ab76edfb"}]},"branch":"refs/heads/master"},"adbc3e23b7d2251cc7de088e2a757674a41c2f6a":{"kind":"REWORK","_number":7,"created":"2024-10-11 13:35:22.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/41/932041/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/41/932041/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/41/932041/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/41/932041/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/41/932041/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/41/932041/7"}}},"commit":{"parents":[{"commit":"92550c9d66006e6f6d64e759ea0e32dc7c9ec3f5","subject":"Merge \"Deprecate ovn_emit_need_to_frag option\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/92550c9d66006e6f6d64e759ea0e32dc7c9ec3f5"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-10-10 08:49:44.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-10-11 13:35:09.000000000","tz":0},"subject":"Optimize the SG rule retrieval","message":"Optimize the SG rule retrieval\n\nThere are some operations where the SG DB object can be used instead of\nthe SG OVO. That saves conversion time, including the conversion of the\nSG rule OVOs, that are child resources of the SG OVO.\n\nThis optimization applies to the following methods:\n* SecurityGroupDbMixin.get_security_groups\n* SecurityGroupDbMixin.update_security_group (partially)\n\nThe Nova query to retrieve the SG list in the \"server list\" command,\nhas been benchmarked. The testing environment had a single SG with\n250 SG rules. Call:\n  \"GET /networking/v2.0/security-groups?id\u003d81f64aa4-2cea-46db-8fea-cd944f106aab\n     \u0026fields\u003did\u0026fields\u003dname HTTP/1.1\"\n\n* Without this patch: around 1.25 seconds\n* With this patch: around 0.025 second (50x improvement).\n\nCloses-bug: #2083682\nChange-Id: Ibd032ea77c5bfbc1fa80b3b3ee9ba7d5c36bb1bc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/adbc3e23b7d2251cc7de088e2a757674a41c2f6a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/adbc3e23b7d2251cc7de088e2a757674a41c2f6a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
