)]}'
{"id":"openstack%2Fneutron~932679","triplet_id":"openstack%2Fneutron~master~I5c94d085304bd1a0f694aee6734dc74b569c389e","project":"openstack/neutron","branch":"master","topic":"bug/2028846","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2024-10-24 16:21:15.000000000","reason":"\u003cGERRIT_ACCOUNT_16688\u003e replied on the change","reason_account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2024-10-23 07:15:14.000000000","reason":"\u003cGERRIT_ACCOUNT_11975\u003e replied on the change","reason_account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2024-10-25 11:14:30.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I5c94d085304bd1a0f694aee6734dc74b569c389e","subject":"[OVN] Remove maintenance method ``add_gw_port_info_to_logical_router_port``","status":"MERGED","created":"2024-10-18 06:13:02.000000000","updated":"2024-10-25 11:14:30.000000000","submitted":"2024-10-25 11:14:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"932679-bug/2028846","meta_rev_id":"c5b3404360977129f409a70f0208e383ec1a018a","_number":932679,"virtual_id_number":932679,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2024-10-18 12:50:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-10-25 11:14:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2024-10-23 07:15:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2024-10-18 17:11:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":1,"date":"2024-10-23 07:15:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-18 06:45:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-10-18 08:05:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-10-18 08:37:41.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-10-18 17:11:26.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2024-10-21 15:32:12.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2024-10-21 15:32:12.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"f8a6ed437105dc375f749598797e9fb88b896d56","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-10-18 06:13:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"68167adb057c288e2600fd6ae6e3006de1fb1259","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-10-18 06:13:39.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a366e7b16dabc632502e0af0d707b4a6e98c8948","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-18 06:45:44.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/01f4b38939be4f8ab0a9e6d883e9c64e\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/772ffb94ef694107937bbeeee9d58cee : SUCCESS in 26m 07s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/691d5b88232643a9a6dee8091288c96e : SUCCESS in 29m 42s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0d21dba2728638d4152171ddfc8111ef5f8e6077","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-18 08:05:09.000000000","message":"Patch Set 2: Verified-1\n\n(6 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5482e713e5234a3c8ac058af0fedaa0c\n\n- neutron-tempest-plugin-ovn-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/1c0a7bbb98784856b9acae7439f96a5e : SUCCESS in 1h 49m 24s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/6718b60e1bf246e3ae41ec7b470c5117 : SUCCESS in 1h 35m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/218dd89f01f54bfc9fb4e2fb593cf14d : SUCCESS in 27m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/693a1bae04a44e0d88e14b960c32fd3d : FAILURE in 4m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b40c673217d44242a65fd460cee09d88 : SUCCESS in 21m 39s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e1a14cfcfad640949b2556ed943f19ad : SUCCESS in 22m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7cc126bdca5a4eecb523f64af12682f9 : SUCCESS in 9m 12s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/cb8c435de81b44299cc8bb427de92c02 : SUCCESS in 54m 17s\n- neutron-tox-mypy https://zuul.opendev.org/t/openstack/build/c44ada9d53424ce0847559f3a87c8875 : SUCCESS in 6m 09s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/93b5e9e6e89a4af58e4f86ae4fb8fd1e : SUCCESS in 1h 00m 13s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/bee264b05cfd4711bbfeaf18342a6533 : SUCCESS in 1h 10m 25s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/0c28cf0268d14cb89202067cbd704380 : SUCCESS in 59m 13s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/6ba91ece471540feb38b235bff347da2 : SUCCESS in 1h 11m 23s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/d1cea23fb34f4118b3a141085f4532e0 : SUCCESS in 1h 18m 20s","accounts_in_message":[],"_revision_number":2},{"id":"836f7772efbdb4609ad0d97cbb887462c596f9c4","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2024-10-18 08:37:41.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/October-2024/Arista_Tempest_Testrefs/changes/79/932679/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"146e19104047f0bd5c77d00bf36f60622e3a17e6","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2024-10-18 09:08:35.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/October-2024/Arista_Tempest_Testrefs/changes/79/932679/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"35adc1b624a044a99c0e4cc038eb75b51358a6bf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-10-18 10:53:20.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nOutdated Votes:\n* Verified+1, Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"50b066d8ea3315bf7a4a7a7af4de4d3ffe4919b7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-18 11:23:57.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2167d35421ba44758e5f992f085a1886\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/291e90c9aea04012a8d8a7e53602d658 : SUCCESS in 25m 49s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/1f08fdb661f54313ac123a4853f4fde7 : SUCCESS in 28m 43s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"55cb2d923b8fca099fc5bb7008e7660da8fabecf","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2024-10-18 12:50:37.000000000","message":"Patch Set 3: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/October-2024/Arista_Tempest_Testrefs/changes/79/932679/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"1c9339b0ceef8b667304714c3e30002ed89b2ffb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-18 12:58:19.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/df9afc2ba04f4726a32cf0071dfb2bc2\n\n- neutron-tempest-plugin-ovn-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/0b8c918a869644a9ba8a58ebb25429f5 : SUCCESS in 1h 53m 00s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/515a505e2e354ba0a1367c2b7f013872 : SUCCESS in 2h 03m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1f2a2ffb79aa4cd799e78662d0fcb9c5 : SUCCESS in 41m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/02fc477d5d9b4276afcf672fc661296d : SUCCESS in 10m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a941badb671447d7957bf0f1efdfbc5d : SUCCESS in 23m 23s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f86b3a30230a4ef081aaae79353c7e04 : SUCCESS in 23m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0c7e74b24cc74abc842d5ae9f0880881 : SUCCESS in 9m 57s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/159cb4fe9d984130ad88fef0cac9536f : SUCCESS in 57m 16s\n- neutron-tox-mypy https://zuul.opendev.org/t/openstack/build/08a5c981d0b646d9b6fb09dbf3aa53e2 : SUCCESS in 4m 55s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/862e198836c74b83bfb7173b1bb762ec : SUCCESS in 54m 42s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/1776759a897e41e6aac3600a58727354 : SUCCESS in 55m 09s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/0a7df9fe0b994301a7b02c7785cc1ead : SUCCESS in 1h 17m 51s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/a34c2682930e4edc85e30f151eee7e66 : SUCCESS in 1h 14m 17s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/71f7671aa9aa4e6f9f196759615c6624 : SUCCESS in 1h 20m 09s","accounts_in_message":[],"_revision_number":3},{"id":"636f4a25184d95c540509525163ec983f60257cd","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-10-18 17:11:26.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"137b7d9698f6cac79bf8096f773146e569d75e14","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2024-10-23 07:15:14.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"81e898ca6b22534896b7e33815a5bb188204818d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-23 07:16:25.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"d48c88b6aef1db7e82e7bf9e4f2cf5feba6b5f60","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-23 10:05:55.000000000","message":"Patch Set 3: Verified-2\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/932678 is needed.","accounts_in_message":[],"_revision_number":3},{"id":"241505f87c14ad6154cf4727098b3673909e890c","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-10-24 06:07:03.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"2fa3f06f0ae26822ebf067848c38bb730fdd846d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-24 06:58:17.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/42d5bc7c4d3f475b95239329f2e3f901\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/72c915a2d5e54a98b609f3e90a86c532 : SUCCESS in 48m 27s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/2d3da6c2b2e64b98a1c9121783201694 : SUCCESS in 40m 17s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"9c47979e485dd1692ef8fe1df4d783b4ec321b74","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-24 08:22:49.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e74020a01467427cbdc38c7742a8a092\n\n- neutron-tempest-plugin-ovn-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/30621608b3434fa69f2b970d9f1f4826 : SUCCESS in 2h 12m 56s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/a153c56ea1f44bf9a397d29866cf7d28 : SUCCESS in 2h 05m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/70e6a20eb4b841bfa044085d2200769a : SUCCESS in 29m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1da6c449a3ba486791337df010fd2ce7 : SUCCESS in 7m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1605b7399553487e9a49be9d722d7fc8 : SUCCESS in 21m 24s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6f7decb2645a44948744b40dcf51c7f7 : SUCCESS in 18m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/105f7179b8634359ac9ece4eb07002b7 : SUCCESS in 13m 27s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/e7f02332ffb4425886f74945c10f0099 : SUCCESS in 1h 13m 34s\n- neutron-tox-mypy https://zuul.opendev.org/t/openstack/build/e0fc0d8cedd2496ba82110a0df7205c0 : SUCCESS in 4m 43s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/1fd9a8cb739146a7aa7d151808a73733 : SUCCESS in 53m 39s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/062aaf5788554ea992c879b3fceb2767 : SUCCESS in 1h 06m 05s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/047d6f6fbcfc4c1e948b3cea2de158b1 : FAILURE in 56m 26s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/f725563db49f4ae2a18de5aa0e5b9524 : SUCCESS in 1h 29m 01s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/27afcceac1aa4ba9b7052002814230d5 : SUCCESS in 1h 14m 28s","accounts_in_message":[],"_revision_number":3},{"id":"807a2ae05dc7583bbd1de2697270db1d932299c7","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-10-24 16:21:15.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"003e9e19572995515a8f98743e49798ae1d8602a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-24 17:21:09.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8da166743b934eabb2f1b1265ef24882\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/601bd5acf2ce4edb9ee76db283bb7071 : SUCCESS in 28m 58s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/62f611f7b7db4f8fa9106dbb7836d9ed : SUCCESS in 33m 47s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"77adbb6ca5b60c7fbaad16c621dcc96c60989284","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-24 19:04:27.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/04a65983b9fa4a8b89d8eebc8ddb2080\n\n- neutron-tempest-plugin-ovn-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/1a97561872a6458da5715a5c767f3920 : SUCCESS in 2h 40m 04s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/a66c765fed214302b3c86e0e09f066c5 : SUCCESS in 2h 07m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/86e8e38eb29846059e6ae8b5dedc4b47 : SUCCESS in 49m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a2fa33ff7fe14452bcd1231c20e519e7 : SUCCESS in 10m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dadb53fe061e46158c438807c803af6a : SUCCESS in 25m 02s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/37cff5f3923d41d8b9e71e1934451cdf : SUCCESS in 18m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2437f4688d44c43ad1865b5e11d45da : SUCCESS in 6m 46s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/1f71ea2f7d414ddda3ca3e3910907131 : SUCCESS in 1h 06m 48s\n- neutron-tox-mypy https://zuul.opendev.org/t/openstack/build/5efb518ccd7e49539b220e053ee8fec7 : SUCCESS in 5m 50s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/152ba85473ff438eaed0483735515f3a : SUCCESS in 58m 16s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/c20c0bcf5f544bfe86d615092eeac5cf : SUCCESS in 1h 01m 26s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/aa3e37e9e06148c38196303be9fc1545 : SUCCESS in 1h 12m 21s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/f512634edc5d4053b41ce6aa9e069689 : SUCCESS in 1h 07m 41s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/be96cee2b54a491a95c91c446c5ffc4b : SUCCESS in 1h 14m 14s","accounts_in_message":[],"_revision_number":3},{"id":"cb9476c098665b9ee4b199d4bf1c0ffbb67f610d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-25 08:20:32.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"8ce411531c9d370e99febadb48c0caa343e17a5a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-25 11:14:30.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/daeec682d1c84698905a50f2eb827aea\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/8de91ec67f7342cebad2d975cd6a4d0f : SUCCESS in 1h 37m 27s\n- neutron-tempest-plugin-ovn-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/918534cf805b4de2b4d6b74e407c9812 : SUCCESS in 1h 56m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/50377e34f2a34fedbd4b1467c9f66395 : SUCCESS in 5m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d3e3f09976804f84a279f1ca90f34c11 : SUCCESS in 18m 59s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/44ea1ff8a1b14dcea6b5ba6db2ed3b43 : SUCCESS in 17m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3bf17beb317b4b329a2872219f2f300c : SUCCESS in 9m 42s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/e16d4a324a7c48899832788c0c667bc1 : SUCCESS in 51m 30s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/ff5bee59ed9d49f7aefc8e02634b8a9b : SUCCESS in 1h 13m 19s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/8cef88336ca344428cd676ab4b3d4804 : SUCCESS in 1h 00m 22s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/b0352f0565294add9818bc9371b48a3a : SUCCESS in 1h 21m 54s","accounts_in_message":[],"_revision_number":3},{"id":"c5b3404360977129f409a70f0208e383ec1a018a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-25 11:14:30.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"6e27923535542cc486566eb98db2636665f7f9c5","revisions":{"985504f40844dca3b923ed7379a0c55eecb6f056":{"kind":"REWORK","_number":1,"created":"2024-10-18 06:13:02.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/79/932679/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/79/932679/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/79/932679/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/79/932679/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/79/932679/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/79/932679/1"}}},"commit":{"parents":[{"commit":"c5f1ac607e6b31dede40348e12c01cc9fe4f5838","subject":"[OVN] Remove maintenance method ``create_router_extra_attributes_registers``","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c5f1ac607e6b31dede40348e12c01cc9fe4f5838"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-10-18 06:10:26.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-10-18 06:12:48.000000000","tz":0},"subject":"[OVN] Remove maintenance method ``add_gw_port_info_to_logical_router_port``","message":"[OVN] Remove maintenance method ``add_gw_port_info_to_logical_router_port``\n\nIt was marked to be removed in E cycle.\n\nRelated-Bug: #2028846\nChange-Id: I5c94d085304bd1a0f694aee6734dc74b569c389e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/985504f40844dca3b923ed7379a0c55eecb6f056"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/985504f40844dca3b923ed7379a0c55eecb6f056"}]},"branch":"refs/heads/master"},"cd0a186642ef6c870a804ca50bc5050e50850060":{"kind":"REWORK","_number":2,"created":"2024-10-18 06:13:39.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/79/932679/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/79/932679/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/79/932679/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/79/932679/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/79/932679/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/79/932679/2"}}},"commit":{"parents":[{"commit":"c5f1ac607e6b31dede40348e12c01cc9fe4f5838","subject":"[OVN] Remove maintenance method ``create_router_extra_attributes_registers``","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c5f1ac607e6b31dede40348e12c01cc9fe4f5838"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-10-18 06:10:26.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-10-18 06:13:32.000000000","tz":0},"subject":"[OVN] Remove maintenance method ``add_gw_port_info_to_logical_router_port``","message":"[OVN] Remove maintenance method ``add_gw_port_info_to_logical_router_port``\n\nIt was marked to be removed in E cycle.\n\nRelated-Bug: #2028846\nChange-Id: I5c94d085304bd1a0f694aee6734dc74b569c389e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/cd0a186642ef6c870a804ca50bc5050e50850060"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/cd0a186642ef6c870a804ca50bc5050e50850060"}]},"branch":"refs/heads/master"},"6e27923535542cc486566eb98db2636665f7f9c5":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2024-10-18 10:53:20.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/79/932679/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/79/932679/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/79/932679/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/79/932679/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/79/932679/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/79/932679/3"}}},"commit":{"parents":[{"commit":"87a0fc5d70b048a933eb3710e7f8c2c35a2746f2","subject":"[OVN] Remove maintenance method ``create_router_extra_attributes_registers``","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/87a0fc5d70b048a933eb3710e7f8c2c35a2746f2"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-10-18 06:10:26.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-10-18 10:38:10.000000000","tz":0},"subject":"[OVN] Remove maintenance method ``add_gw_port_info_to_logical_router_port``","message":"[OVN] Remove maintenance method ``add_gw_port_info_to_logical_router_port``\n\nIt was marked to be removed in E cycle.\n\nRelated-Bug: #2028846\nChange-Id: I5c94d085304bd1a0f694aee6734dc74b569c389e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6e27923535542cc486566eb98db2636665f7f9c5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6e27923535542cc486566eb98db2636665f7f9c5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
