)]}'
{"id":"openstack%2Fneutron~933552","triplet_id":"openstack%2Fneutron~stable%2F2024.2~Ibe8bda2f81bda7a89e3a994db55cd394a18decb8","project":"openstack/neutron","branch":"stable/2024.2","topic":"bug/2083832","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2024-10-29 17:20:35.000000000","reason":"Change was submitted"},"21798":{"account":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"last_update":"2024-10-29 14:40:03.000000000","reason":"\u003cGERRIT_ACCOUNT_21798\u003e replied on the change","reason_account":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}}},"hashtags":[],"change_id":"Ibe8bda2f81bda7a89e3a994db55cd394a18decb8","subject":"[OVN] Update lsp host id when cr port is updated with chassis","status":"MERGED","created":"2024-10-28 09:45:25.000000000","updated":"2024-10-29 17:22:04.000000000","submitted":"2024-10-29 17:20:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"933552-bug/2083832","meta_rev_id":"0a57ce53a2d50599dcd23db0f58388d2dcef7a2d","_number":933552,"virtual_id_number":933552,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-10-29 17:20:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":2,"date":"2024-10-29 14:40:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2024-10-28 15:37:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":1,"date":"2024-10-29 14:40:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":35983,"name":"Aleksandr Chudinov","email":"silentirk@me.com","username":"silentirk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-28 09:45:25.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":35983,"name":"Aleksandr Chudinov","email":"silentirk@me.com","username":"silentirk"},"state":"CC"},{"updated":"2024-10-28 10:08:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-10-28 12:15:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-10-28 15:37:33.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2024-10-29 13:33:14.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"}],"messages":[{"id":"28e95416ccb7b21f06306c6990c063db8c7931be","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-10-28 09:45:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7a4cb9e4145eff2b703c3897245b5b29e59cdce1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-28 10:08:26.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fe6933a91db5495f8a4ea3f3a475de21\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e903179d1d544885b600b8a913813f1e : SUCCESS in 21m 10s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/79c3590c13144800a125faf0554f0834 : SUCCESS in 22m 30s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5f7928dce5902078d8f8700782e6f714d921753a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-28 12:15:56.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f092e0349a654f9eaa26c3c73f5c1e1e\n\n- neutron-tempest-plugin-ovn-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/556f461d03124e01aca55ba67c95c5cb : FAILURE in 1h 46m 35s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/1e727139670848ae972dbb4dee26dd6e : SUCCESS in 2h 29m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9e918c2ecd384dbfbad6d3e11a8ba43a : SUCCESS in 27m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/05953cde451f4a24960465e5c0bae36a : SUCCESS in 9m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9a3f0485ca174cc28f7e73f3568aa143 : SUCCESS in 18m 18s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c0bc34ade2fa461d85ee1faf53491c26 : SUCCESS in 16m 20s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/27a4b19fc9504ed59237ee9f424fc324 : SUCCESS in 20m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/56aab55e1fec42e6a85a0fdcf589ab5e : SUCCESS in 9m 31s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/8df4d44c6d064006b9d0431c7e05af67 : SUCCESS in 45m 53s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/47ef77d03e7d4b10b1d2ab132d82efc1 : SUCCESS in 2h 15m 47s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/eefcc5d40285423b97fdcd4c76fde572 : SUCCESS in 59m 07s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/9b770309c1fc4a7faea92ba08c26ee86 : SUCCESS in 53m 14s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/e3adf34fae55420282579e23e2af9f71 : SUCCESS in 1h 15m 55s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/40873e5a86d74bdbb8c54925e79ec4eb : SUCCESS in 1h 12m 12s","accounts_in_message":[],"_revision_number":1},{"id":"ca4347f42920ea3083e8f317280120089d025925","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-10-28 15:37:33.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a8fca639f695c232d61923cec76ddcc671ee7db5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-28 16:09:00.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/98fc29d57d5844e984d61b567b53accb\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0f6d78bf261a4cb4a0ca34f5489b89a7 : SUCCESS in 29m 18s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/24c5195a65f74a12bd18ea09ac6896ee : SUCCESS in 28m 11s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3c5ea7024f808bf843e29a2413215b9abd8645a2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-28 17:54:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2861bfcef7094497ae31392f5fbdfbda\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/555115bd7ffe4281afd8b745de9c744a : SUCCESS in 2h 13m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0e52a8f898d3422c9985d3fa4de38125 : SUCCESS in 44m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9e4c2255efc945f59c6aa047b02c87a6 : SUCCESS in 9m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/762693b14117400bbd712c325b616659 : SUCCESS in 22m 56s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/dbb85acd468d49379111d3d1105e9c55 : SUCCESS in 18m 04s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c183cc5bc9214c90a2c624fb3873343a : SUCCESS in 17m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f72a90e3a9594e4d8deaf869dccfe9d5 : SUCCESS in 7m 03s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/a7b2230d44e746edbbd14619a055b1f0 : SUCCESS in 58m 57s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/268b704d18114451b9586336bae61506 : SUCCESS in 1h 23m 01s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/e97d73e761fc4c20bb2057c5fb0e7194 : SUCCESS in 45m 48s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/2ceda12857ba44419d922d97c893d22a : SUCCESS in 1h 01m 07s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/7e3a8742e82a479cbd1939769d3ef5b5 : SUCCESS in 54m 18s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/67cf0743aaca46548d89cb695a1e21f2 : SUCCESS in 1h 13m 35s","accounts_in_message":[],"_revision_number":1},{"id":"c9bca27d265c395019a1d092f3dbdc01d131c6a7","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2024-10-29 14:40:03.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6405bff0bb2ba846b22a318b61dc424bd0b1b3f8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-29 14:40:49.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"aa3be4562ce770297187f14aa2b6b70d8c0e7429","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-29 17:20:35.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bcc3ad151bae42aaac805667ed62ff8d\n\n- neutron-tempest-plugin-ovn-2024-2 https://zuul.opendev.org/t/openstack/build/053709db451042c285f1ac0729e25eb0 : SUCCESS in 1h 40m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d93aff88360f4b32bf1ea8561146c468 : SUCCESS in 8m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/705d05dd63b1475ba20893b93973fca1 : SUCCESS in 26m 14s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/dbdc9224d18149239b78dc4ea3f623e7 : SUCCESS in 22m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/486d5d5755704843826a727372b06cd4 : SUCCESS in 8m 08s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/fb178727d1704089b8c3ea8b61aad15c : SUCCESS in 1h 07m 33s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/a1a6b9b2beb64eac8ae7c8805989fc75 : SUCCESS in 1h 27m 50s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/d766e10ae9534c82bda61bc69a76cc43 : SUCCESS in 51m 13s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/05e779f70ca347139c9fca7231558629 : SUCCESS in 47m 25s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/a72b3e13af494b69b6c197b0d00cdde0 : SUCCESS in 1h 04m 10s","accounts_in_message":[],"_revision_number":1},{"id":"6e6401c504674bad9c0526528b188117a066a6d3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-29 17:20:35.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"0a57ce53a2d50599dcd23db0f58388d2dcef7a2d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-29 17:22:04.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bbabfc1e1e4442c488f22a469357ca7b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/768368a302764147bbfc0830c389b3c6 : SUCCESS in 55s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"aff4e4e8f35dce2877ba69eaddbb1e90012e5090","revisions":{"aff4e4e8f35dce2877ba69eaddbb1e90012e5090":{"kind":"REWORK","_number":1,"created":"2024-10-28 09:45:25.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/52/933552/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/52/933552/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/52/933552/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/52/933552/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/52/933552/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/52/933552/1"}}},"commit":{"parents":[{"commit":"5a69d0354f53bf0d5b820575d991c288d97169ba","subject":"\"security_group_rules\" is not a SG selectable field","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5a69d0354f53bf0d5b820575d991c288d97169ba"}]}],"author":{"name":"Aleksandr","email":"silentirk@me.com","date":"2024-10-07 10:06:59.000000000","tz":180},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-10-28 09:45:12.000000000","tz":0},"subject":"[OVN] Update lsp host id when cr port is updated with chassis","message":"[OVN] Update lsp host id when cr port is updated with chassis\n\nWhen a chassisredirect port is updated with chassis, the\nPortBindingChassisEvent event would only update the binding\nhost id in the neutron database, while it is also usefull to keep the\ninformation in the OVN database up to date with the host information.\n\nSimilar to change [1], but for router\u0027s gateway ports.\n\n[1] https://review.opendev.org/c/openstack/neutron/+/896883\n\nOther plugins that connect to the OVN database can then also rely on the\ninformation stored in the OVN DB\u0027s\n\nCloses-Bug: #2083832\n\nChange-Id: Ibe8bda2f81bda7a89e3a994db55cd394a18decb8\n(cherry picked from commit 4b032bdbb2a6843b776c367486d1620ea6ae71a5)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/aff4e4e8f35dce2877ba69eaddbb1e90012e5090"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/aff4e4e8f35dce2877ba69eaddbb1e90012e5090"}]},"branch":"refs/heads/stable/2024.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
