)]}'
{"id":"openstack%2Fneutron~933554","triplet_id":"openstack%2Fneutron~stable%2F2023.2~Ibe8bda2f81bda7a89e3a994db55cd394a18decb8","project":"openstack/neutron","branch":"stable/2023.2","topic":"bug/2083832","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2024-10-29 17:20:50.000000000","reason":"Change was submitted"},"21798":{"account":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"last_update":"2024-10-29 14:40:18.000000000","reason":"\u003cGERRIT_ACCOUNT_21798\u003e replied on the change","reason_account":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}}},"hashtags":[],"change_id":"Ibe8bda2f81bda7a89e3a994db55cd394a18decb8","subject":"[OVN] Update lsp host id when cr port is updated with chassis","status":"MERGED","created":"2024-10-28 09:46:01.000000000","updated":"2024-10-29 17:22:15.000000000","submitted":"2024-10-29 17:20:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"933554-bug/2083832","meta_rev_id":"ffe1dbcf45b9ae522b93853b16951a3d71fff631","_number":933554,"virtual_id_number":933554,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-10-29 17:20:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":2,"date":"2024-10-29 14:40:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2024-10-29 09:08:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":1,"date":"2024-10-29 14:40:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":35983,"name":"Aleksandr Chudinov","email":"silentirk@me.com","username":"silentirk"}],"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-28 09:46:01.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":35983,"name":"Aleksandr Chudinov","email":"silentirk@me.com","username":"silentirk"},"state":"CC"},{"updated":"2024-10-28 10:27:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-10-28 11:39:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-10-29 09:08:08.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2024-10-29 13:32:54.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"}],"messages":[{"id":"03d337b9161f7e0a7dae8d632ba03d12824c549e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-10-28 09:46:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3e852517eb0402cf6c0540e0c8dd8cd68b607b2d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-28 10:27:57.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/80dbb9f4b2e9487a884046fbd6d96653\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5ac3f89eae0f4daa8c545146b63b6f0b : SUCCESS in 40m 18s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f4ab0d33e3264e99a2c1c59f70d2d90a : SUCCESS in 26m 20s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/8ca2789b789546ddab39c31ba939f9d4 : SUCCESS in 25m 32s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/e7881aedb55c4b39ace9a946292ba985 : SUCCESS in 24m 06s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e030e070dc6700ed5c450c4009f96cd14d06d366","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-28 11:39:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b3aac14f563e42a38ea25cd2fa1e83e2\n\n- neutron-tempest-plugin-ovn-2023-2 https://zuul.opendev.org/t/openstack/build/08f8339fa0f24c14acbe6f91ae164848 : SUCCESS in 1h 41m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/61638fbd969742faa8959e38fdaa3122 : SUCCESS in 35m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e125a6eca3e1412f8e3bba3d948ab606 : SUCCESS in 7m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b9cfceeda8234b8d9f251f4b10100624 : SUCCESS in 21m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a1a702a766ff4247b3aefff1d0620ce6 : SUCCESS in 21m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d4e370f40a24414db48ec2080b2993d5 : SUCCESS in 22m 27s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8bce60c7f4294393be1c59b50d348442 : SUCCESS in 19m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c80de7c3c994c88a9d945b9c0430d75 : SUCCESS in 10m 47s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/fba09e54e4e14e18b931462e973fb3a9 : SUCCESS in 37m 14s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/2de1793bcb024b18931e4bbfd0013ac5 : SUCCESS in 1h 45m 46s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/9dc6d901a73a4e149fee615d4ee7e770 : SUCCESS in 54m 29s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/a76dc13856f545da8e24448131292c3e : SUCCESS in 1h 19m 09s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/3c2e142b39c24f3493587647bf737dd8 : SUCCESS in 1h 17m 35s","accounts_in_message":[],"_revision_number":1},{"id":"ddf0ad99e1eff5c0df68f9e893e3ddf831485c5f","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-10-29 09:08:08.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"fb9f7f1c3f5bb24de5a3d0b90b5fd6dce65bba12","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2024-10-29 14:40:18.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0761f3defffaa8189186ee1a02714f22e5a68e31","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-29 14:41:03.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b4a67bc7ae0307373438444fbe4dd251cc4c0962","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-29 17:20:50.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8765ac60366d41508a2bd71774312b61\n\n- neutron-tempest-plugin-ovn-2023-2 https://zuul.opendev.org/t/openstack/build/c296864b12b347949d76ad752f24abb3 : SUCCESS in 1h 50m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5bcd5748a94744c8a92e20fd75d967b8 : SUCCESS in 10m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bf9538257e0b494abfcf4588d894a7bc : SUCCESS in 24m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0248dc5151094f99a6c2eddea32c5e15 : SUCCESS in 24m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/918b79c7f7934156a7143732632ff72f : SUCCESS in 27m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/341fe3e301bc433bb3791f7714567a85 : SUCCESS in 10m 02s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/c4c064cb626c4c12a6447f0da4aa2d05 : SUCCESS in 41m 42s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/9bbcd98a7116410b819bcc0c10a05f54 : SUCCESS in 1h 40m 27s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/3285e123a2e340e8b1aa7e446801b3f4 : SUCCESS in 52m 29s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/63acadcc8e2d42bb825044533ed5be7f : SUCCESS in 51m 48s","accounts_in_message":[],"_revision_number":1},{"id":"a8386bfa11c345eb2051009bd527ab7d7368da9a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-29 17:20:50.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"ffe1dbcf45b9ae522b93853b16951a3d71fff631","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-29 17:22:15.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0f9ebeee682e45feb0ee53b0dbae2a67\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d22160e0ff74c00ad25baa148733eb3 : SUCCESS in 52s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a687142295e1f5184f32fdc040991cd1f12d823d","revisions":{"a687142295e1f5184f32fdc040991cd1f12d823d":{"kind":"REWORK","_number":1,"created":"2024-10-28 09:46:01.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/54/933554/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/54/933554/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/54/933554/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/54/933554/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/54/933554/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/54/933554/1"}}},"commit":{"parents":[{"commit":"e2f0beb99eb97d0ce1202b6218d6b857cbe7e6ce","subject":"Merge \"Revert \"[OVN] Prevent Trunk creation/deletion with parent port bound\"\" into stable/2023.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e2f0beb99eb97d0ce1202b6218d6b857cbe7e6ce"}]}],"author":{"name":"Aleksandr","email":"silentirk@me.com","date":"2024-10-07 10:06:59.000000000","tz":180},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-10-28 09:45:54.000000000","tz":0},"subject":"[OVN] Update lsp host id when cr port is updated with chassis","message":"[OVN] Update lsp host id when cr port is updated with chassis\n\nWhen a chassisredirect port is updated with chassis, the\nPortBindingChassisEvent event would only update the binding\nhost id in the neutron database, while it is also usefull to keep the\ninformation in the OVN database up to date with the host information.\n\nSimilar to change [1], but for router\u0027s gateway ports.\n\n[1] https://review.opendev.org/c/openstack/neutron/+/896883\n\nOther plugins that connect to the OVN database can then also rely on the\ninformation stored in the OVN DB\u0027s\n\nCloses-Bug: #2083832\n\nChange-Id: Ibe8bda2f81bda7a89e3a994db55cd394a18decb8\n(cherry picked from commit 4b032bdbb2a6843b776c367486d1620ea6ae71a5)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a687142295e1f5184f32fdc040991cd1f12d823d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a687142295e1f5184f32fdc040991cd1f12d823d"}]},"branch":"refs/heads/stable/2023.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
