)]}'
{"id":"openstack%2Fneutron~933752","triplet_id":"openstack%2Fneutron~master~I12079de78773f7409503392d4791848aea90cb7b","project":"openstack/neutron","branch":"master","topic":"bug/2085946","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2024-11-05 06:41:09.000000000","reason":"\u003cGERRIT_ACCOUNT_16688\u003e replied on the change","reason_account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2024-11-04 09:06:25.000000000","reason":"\u003cGERRIT_ACCOUNT_11975\u003e replied on the change","reason_account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2024-11-05 10:42:16.000000000","reason":"Change was submitted"},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2024-10-31 14:17:58.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"I12079de78773f7409503392d4791848aea90cb7b","subject":"[OVN] Fix the revision number retrieval method","status":"MERGED","created":"2024-10-30 15:22:36.000000000","updated":"2024-11-05 10:43:43.000000000","submitted":"2024-11-05 10:42:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"933752-bug/2085946","meta_rev_id":"5764eb4f623c74efdfcaa6129eb03d3eb08a3075","_number":933752,"virtual_id_number":933752,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2024-10-31 11:11:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-11-05 10:42:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2024-11-04 09:06:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2024-10-31 14:17:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":1,"date":"2024-11-04 09:06:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-30 15:56:04.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-10-30 17:28:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-10-31 08:38:10.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2024-10-31 08:38:10.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2024-10-31 08:49:40.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"af2376ded9b96953e1f5f4768b461113f3179fe1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-10-30 15:22:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"88c15d28cbaf3219011f3794fc141c8405c38566","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2024-10-30 15:56:04.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/October-2024/Arista_Tempest_Testrefs/changes/52/933752/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"dc2bcf0cfe5783677be21af12c39379378b9ecd6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-30 17:28:16.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2fec30c65cbb42a68ad0801933186335\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/0a30a103da984cf5b860c402e0dec1d5 : SUCCESS in 1h 19m 52s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/10dcd6f2f73947ac89ed812f9a471300 : SUCCESS in 2h 04m 01s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/555246d363c44d33bfe968511a4a16c6 : SUCCESS in 1h 57m 24s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/1d3bee50e45044c0959973b40921f148 : SUCCESS in 37m 43s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9877bd012a15400aaaa80b55ad785906 : SUCCESS in 55m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fa0f77e825e44d45a476f818504249af : SUCCESS in 10m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/33975831dee94803bdfa7ed69cd599ca : SUCCESS in 18m 37s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/895c2800a6c74adfa34e053c968ed5aa : SUCCESS in 19m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c4c8a5e108f94b21b50f6a895834c59d : SUCCESS in 10m 15s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/ac455e5aeb584cd58729a8658683085c : SUCCESS in 1h 03m 11s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/e5b9ea5e2e274e6ca821522e8671c251 : SUCCESS in 45m 56s\n- neutron-tox-mypy https://zuul.opendev.org/t/openstack/build/1923ce0ed4fb48a1956f8b49201d5662 : SUCCESS in 4m 37s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/df66b40cd01f4ec68a8b1772ac5c3efe : FAILURE in 36m 10s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/766dc3c9037f4fb18f3c0ad0492e23f4 : SUCCESS in 1h 40m 28s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/e430b434fc1841eeadbe1177dff6933e : SUCCESS in 55m 49s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/a90518912fd54f76ae0008815031f186 : SUCCESS in 1h 14m 41s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/772ee27ff1554382b929a8ac77566f45 : SUCCESS in 1h 24m 09s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/7a06d3832621400390ad4ba67b1cf13c : SUCCESS in 1h 24m 23s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/7d775ee0f65b4e65b034e94121531895 : SUCCESS in 1h 13m 21s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/284091a643fe4174bd29e3ce61f1afd1 : SUCCESS in 56m 34s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/a1633b5d797246ad9b65ef8c077665ae : SUCCESS in 1h 12m 53s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/e2503fee38854963b05cec378415cf48 : SUCCESS in 1h 14m 39s","accounts_in_message":[],"_revision_number":1},{"id":"266ed8c91e5bd6b2ae101614d1e243d8fb3379cb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-30 17:57:43.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/50bad54d82e94315a0a748c982b40908\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/579daba1b59344909b4f97220cee17a5 : SUCCESS in 32m 16s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/27cc2a7525f54f3782b7211917971b78 : SUCCESS in 37m 17s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"536d2d1a65d18c3cac0325173a658c2f92ccb67a","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-10-31 06:30:13.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7a1c2419df927d4bcf88bec9dd9dc3ec29adef85","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-31 07:14:38.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/58660c112b3b4b7a85b7350c015b8d4e\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1a08f6090fbe43e5ab82645c189391fc : SUCCESS in 32m 27s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/79301efb1208449ead2563f8d4e90480 : SUCCESS in 41m 44s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ae641e63b9942b7c84005eb6c0f630aaf566b475","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-10-31 08:38:39.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5411622d839042794bc4268a17c44cf47360ccef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-31 08:43:15.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4956633345554fb4a89a53a3ed897b8b\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/d290f6bf35484cb399cdf32ca483db07 : SUCCESS in 1h 49m 13s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/c3ed35af625a4c5286f44afd203d2ec5 : SUCCESS in 1h 53m 16s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/5549584f51e14dc59538ee13976a00b5 : SUCCESS in 2h 10m 44s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/87d369d511fb4669bd0e47b6cd3c272a : SUCCESS in 39m 00s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/58b8dc5d694b4b228df1f1a1f6da30b4 : SUCCESS in 29m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/658da611b73d4d1185594319191e7522 : SUCCESS in 11m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0af01a0f63af44418c0bade97bd432b1 : SUCCESS in 19m 59s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d537720a3f064a288a8723fe9e411cc9 : SUCCESS in 22m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a6543790f8943188e0d33dfe13a4b94 : SUCCESS in 7m 24s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/cdbb814ae376497f968649f4871c15b5 : SUCCESS in 44m 53s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/34d0a174fd4e4ad8904f16e6b52ef1a7 : SUCCESS in 51m 41s\n- neutron-tox-mypy https://zuul.opendev.org/t/openstack/build/04287b2bbb9149fca00c99e2c53dc006 : SUCCESS in 4m 48s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/8e65fdb3469241e4b54a3e4a23a77b6b : FAILURE in 48m 52s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/1d6cca27363e4f39a8e5c918d1e9d428 : SUCCESS in 2h 00m 18s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/9aa3fe8c3bef4910a6f7730d2c08018b : SUCCESS in 57m 34s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/3d760dbf1e214eeaabdf2d058f96fc85 : SUCCESS in 46m 05s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/09f1e671190c40d6b967f34d67924826 : SUCCESS in 1h 43m 41s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/822637b9f7164ec7974b088b9fc02ebe : SUCCESS in 1h 24m 11s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/c5439046270449e28d1a40814a81b27e : SUCCESS in 1h 03m 16s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/b5ce99fd417d45ad83991ccc64f656c9 : SUCCESS in 45m 34s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/df7fe84ccb1745da8309b5a9272f8907 : SUCCESS in 49m 25s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/fa83fd5430924ad4b742b17ab6539df3 : SUCCESS in 54m 57s","accounts_in_message":[],"_revision_number":1},{"id":"9b4b76dd9ff90c2b5f4ff6d68e6f41a6eb69797a","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2024-10-31 08:49:40.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"469528a74561ac222f0f1ecd798c681c6f1e592a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-31 09:31:57.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8228c0dce525492c8add4a7456f9b6af\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/83dc70c7e6e442dd878ef1356bed2128 : SUCCESS in 38m 55s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/09109e71eb344c528cb4ad24950ff30c : SUCCESS in 39m 00s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5a889097cf41fa0c7b8237a35a2b64670b09b807","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-10-31 09:51:44.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e0f4de6e11e6e4f6f829ea521583b0c2072344f4","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-10-31 10:36:40.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ad052e7af04a6363a9e1c3f08d86d195d5edf2cc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-10-31 10:38:22.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1, Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"9febadabd51e124d59cf1cb6b9f059673e23f359","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2024-10-31 11:11:32.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/October-2024/Arista_Tempest_Testrefs/changes/52/933752/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"824a3e60e9fd54f96f7bf6842993f0d7f6475df5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-31 12:35:16.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/88404090149540bca43f2ba497dc7186\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/0dbd7c08522342f7994f7a27270a2dbc : SUCCESS in 1h 22m 14s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/3ef096b5fd8641f9b45b96f432d511b6 : SUCCESS in 1h 53m 48s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/66324f23d70e41788149d0e7860c00b6 : SUCCESS in 1h 52m 04s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/c792c3069eb846b8b10f2c8ac9a057b1 : SUCCESS in 32m 55s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/82ead358d97d402b9fa9ebbdca6ee143 : SUCCESS in 43m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a103ddad5692432eb5db416e24540664 : SUCCESS in 12m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/90542455e36b4a06bccc89d935e6cc64 : SUCCESS in 21m 38s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5b4eadedbf9a42f2baca198c1b4a4c68 : SUCCESS in 21m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea16f3ea384d4b16ad7ed4d0003efda6 : SUCCESS in 11m 11s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/5e5ef9dc78204a4b84d205f72ac793c8 : SUCCESS in 58m 48s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/8630eb06ce744d2fa326c3cd97be0bc5 : SUCCESS in 39m 49s\n- neutron-tox-mypy https://zuul.opendev.org/t/openstack/build/ba461e3592564347bfc374400ea11370 : SUCCESS in 4m 44s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/054f4219657449aa90fecdd0adfc0be2 : SUCCESS in 59m 53s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/89926b14817343fa85e535d503c353ff : SUCCESS in 1h 39m 44s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/f62207abb40b40b5af3183f9c71c41a8 : SUCCESS in 55m 15s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/2f9223b825584f59878b680763eeccf1 : SUCCESS in 1h 09m 38s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/2f4919c8207d471b8be376f004301f2f : SUCCESS in 1h 38m 35s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/6457e9d9347744b584cb0a53a7d125f9 : SUCCESS in 1h 32m 36s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/6999f707a2b34f06bda4857a562c6abf : SUCCESS in 1h 13m 17s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/09425f99484e438e9b4824f46bd04426 : SUCCESS in 1h 05m 20s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/62d0b9b200d44230bf91dedabbba72f7 : SUCCESS in 1h 15m 08s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/6cb913a11763416d98b740feebc6acd3 : SUCCESS in 1h 21m 29s","accounts_in_message":[],"_revision_number":2},{"id":"93329a58e46b7a3d22512757aa0dd6932e529eab","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-10-31 14:17:58.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"ff3ad0dc07a08b85573dbf85b62dab1fbe5d681c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-01 02:22:20.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7430c29033964be3826a155cfcff1f61\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/51a1157e0ce64953b385b96d9c7b071d : SUCCESS in 37m 11s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/2557fa272d7a4a388c75aafee4b45b58 : SUCCESS in 44m 36s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"4e3079c80afd10743d0d63c67c6d57e7068bb153","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2024-11-04 09:06:25.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"ed1b9c92d7f3c9b3f57f1759e1e6e7c2a11fee8b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-04 09:07:02.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"36afecc39ef2c073f9357fd1327a7993c58817f3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-04 12:23:00.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f3c07ffb51ca441698f8440db05ea394\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/aedaf336eedc4e388adada1fa3cde797 : SUCCESS in 1h 56m 21s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/96a6120ad4554dc3963ff9cf920bf47e : SUCCESS in 2h 03m 01s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/1fe831a05c94424eb01556267ca39ff1 : SUCCESS in 2h 20m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c3b90ed7027843158259d28799c016df : SUCCESS in 9m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2167321857b9487c9010ac00e07172ad : SUCCESS in 19m 02s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/576834fb2bd6401fad981d043007b98b : SUCCESS in 18m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ec5e885b75948cdacb3dc438a03d2f5 : SUCCESS in 6m 47s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/6bac790d369f4df596ff886251c60e29 : SUCCESS in 41m 34s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/8755bcf6660141839ad29ac42698e20f : FAILURE in 1h 25m 12s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/70f50b7ab403418b8cb4b10175333dbb : SUCCESS in 1h 21m 32s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/79f245955d8b4f29b98d6da7b3556611 : SUCCESS in 54m 19s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/321cfd4dc83c48f3a2db77321c7cc7c0 : SUCCESS in 52m 28s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/3c28c6af69754a378b58b58503c76667 : SUCCESS in 1h 04m 59s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/5603392707214975b9105da800b225bc : SUCCESS in 1h 25m 52s","accounts_in_message":[],"_revision_number":2},{"id":"ee8ac9f5578137a1e89e96a0a76856ef91cbf4f1","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-11-04 15:11:52.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ef1e1fe5dce9c978269a4412d76532a482cd8ba0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-04 17:20:03.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d1da2dc03dd74442ab18ece56cb1273a\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a744a4f1a66645c9b31bb57df324513d : SUCCESS in 38m 20s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/1fbf612d6482432eb1259b5cfba93e2f : SUCCESS in 29m 38s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"6a0a87ffcf52f35715670d7b3cc54ec95f5929eb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-04 17:29:28.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/758aa6afce2e455abf3ea0b430cb9d64\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/c0ff46c2f2334cb4a2820769d053d915 : SUCCESS in 1h 11m 23s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/2b6cea4225e04db28a5fc508af7ad5f9 : SUCCESS in 2h 05m 56s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/0e4b24e0e375446786aae8f75bc2ceb1 : FAILURE in 2h 15m 42s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/65e1ec416c194090993bb93fbbd70e8f : SUCCESS in 31m 52s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b46f162d7b8448d0887b5fb3650e5d98 : SUCCESS in 37m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cb7fee831bc94d9aa65b85b73dfb1f1c : SUCCESS in 10m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/35e20f8559cc468bb1a8f79ff784a532 : SUCCESS in 21m 46s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/49204aa14cd944f6a8033936f1de48c8 : SUCCESS in 20m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/16cdeffb9c324a5691c85ea420eb90d7 : SUCCESS in 10m 34s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/285d38b807774eaf9b9f2d3215762389 : SUCCESS in 45m 03s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/c9c47fe3561e48e0ac7dacf78af7c103 : SUCCESS in 1h 04m 33s\n- neutron-tox-mypy https://zuul.opendev.org/t/openstack/build/5767932cea484e0ca4475041f177b151 : SUCCESS in 4m 35s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/16d21f48e0ed4599ab2b684e34c127a8 : SUCCESS in 53m 19s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/06aad2d2d23a4bb2b13e1460aca402d9 : SUCCESS in 1h 54m 19s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/0bc71763185b49019435b1a2ca30cbe6 : SUCCESS in 1h 01m 16s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/e0586716721940ce988c726b1dabf9cd : SUCCESS in 57m 44s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/47f0e0afa30d419fa8944865df2f86a8 : SUCCESS in 1h 23m 19s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/c9faea656c564470a627a4d947059eec : SUCCESS in 1h 31m 56s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/44ade7439d6e47549cb8f9d316d142da : SUCCESS in 51m 57s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/2844030d95864e39a633740b56349c06 : SUCCESS in 54m 08s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/e89f4da434c94775982f5b47dcfc7711 : SUCCESS in 59m 28s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/086e11066ad74694a32894ff8572e66f : SUCCESS in 1h 12m 47s","accounts_in_message":[],"_revision_number":2},{"id":"dbb07bfff0108f84c5bcf007bbfb1ef701786707","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-11-05 06:41:09.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"959538eee7d82353d6582b471a06916ad4bb8cd0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-05 07:13:30.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c99f6c8071f9438ea8611fa9e64d390c\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d0b7f6f302c04980a8ceb26712b33329 : SUCCESS in 24m 54s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/0515716dcdde4a66989771bb051734cc : SUCCESS in 30m 25s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"fe103db1b86be42e87b8d973716bb28df4133f2d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-05 08:34:04.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d31e3ef1bd0542a2b55b7c524974c18e\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/3e1744014dba4a45985ef36f6b8af9ed : SUCCESS in 1h 27m 12s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/49273d9db43746c39c43662cf1669f8e : SUCCESS in 1h 34m 29s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/f907697fb53446e8a9c7b0b6517e08c0 : SUCCESS in 1h 51m 48s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/54777f86b54f499998b136a2dd2c69b4 : SUCCESS in 30m 24s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ef075f16519b4782a24f6136b0fe2206 : SUCCESS in 31m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/73fb463e0b9b4e96a4bd1ba5360390a3 : SUCCESS in 9m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/780246e076ad4c18bf81aa052bd49b4b : SUCCESS in 18m 26s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/52510b11300d4a5baeb56fdb6d8301b0 : SUCCESS in 18m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2c2787fce7445309ce32fb51d5b96f5 : SUCCESS in 8m 46s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/b93e4b9a9657433d8f1ee4877f604502 : SUCCESS in 36m 12s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/525bc36f615944a182359dad5d9df716 : SUCCESS in 58m 15s\n- neutron-tox-mypy https://zuul.opendev.org/t/openstack/build/9de8e03f9a2b4d0db9172cf818b63bad : SUCCESS in 4m 28s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/a4f7e4b4cb354161b590e59629ea292a : SUCCESS in 1h 00m 47s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/1e9b1a1b49aa43c5ad11670f3213e787 : SUCCESS in 1h 22m 01s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/fbf6b544472d4328bfac714cc408912e : SUCCESS in 53m 19s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/efd1f907a2b040e7b09523a7c52e8b40 : SUCCESS in 1h 10m 42s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/8938da3d850c4a47852f573c2ff5b28f : SUCCESS in 1h 06m 25s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/d0301f1c88904ad1896722087b1d98d0 : SUCCESS in 1h 30m 12s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/207e5b08923c489aa998510219c539bf : SUCCESS in 1h 01m 34s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/cfb378ef78e1476b8cd27d8a5db128ea : SUCCESS in 1h 09m 43s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/40d433e39467478a93c309fc74c0473c : SUCCESS in 1h 05m 53s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/b3765ddd1efa41a5b1397032bdc330b0 : SUCCESS in 52m 19s","accounts_in_message":[],"_revision_number":2},{"id":"9fb5e640f11ec24b9c48c561fd7664f447f7aadd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-05 08:34:37.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"6aa0378ad5836e4b29a2f2f7da91f588bfb386d6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-05 10:42:15.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/514ff70660ee4a23b2366bc6f66d1e67\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/574c7c5300d54a22932c9129fef641fe : SUCCESS in 1h 34m 21s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/20d5cedd8db549078b50f1c62dc6dd9f : SUCCESS in 1h 54m 28s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/e6d1b4bb5b2d441182b3fac50ce0766d : SUCCESS in 1h 30m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8e130ddf4e184ec0aa9bb44ec3601aa1 : SUCCESS in 9m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c8439efdf43f47ee983b29081f153d58 : SUCCESS in 23m 40s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d55d3cec935c40098f2d9aa38ce13527 : SUCCESS in 19m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0af688f44bce4cd587c25a6f5b463359 : SUCCESS in 11m 36s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/87804fa9fd1d4e22a8303b10ed080871 : SUCCESS in 56m 43s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/c1a629648fec4ec7be992eef71a37c2e : SUCCESS in 2h 02m 12s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/ca438039d796451cb731ad5e11843fb3 : SUCCESS in 1h 36m 01s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/ff7c8c353b9d456990f7b89a54b1e492 : SUCCESS in 52m 21s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/853abe1904b74b81a44cbc1cc4612a62 : SUCCESS in 1h 05m 55s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/06b63b6b2d184110aa6b6cf2a8d78591 : SUCCESS in 1h 03m 21s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/c26be2392dee47b98115f1275a9080d9 : SUCCESS in 1h 15m 25s","accounts_in_message":[],"_revision_number":2},{"id":"ed218200e066e2c3bbdf85589c43338ffab8a383","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-05 10:42:16.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"5764eb4f623c74efdfcaa6129eb03d3eb08a3075","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-05 10:43:43.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/392015e107c34dffb1f480e0bc9cad1b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/03ad3c64f41c4b6fb2037d673b4a475b : SUCCESS in 1m 05s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"a298a37fe7ee41d25db02fdde36e134b01ef5d9a","revisions":{"63f60d878fc02f14f468a004ac1592b9765fcdab":{"kind":"REWORK","_number":1,"created":"2024-10-30 15:22:36.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/52/933752/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/52/933752/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/52/933752/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/52/933752/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/52/933752/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/52/933752/1"}}},"commit":{"parents":[{"commit":"2a244beb214b65267db58ed9afc7cf0464832d28","subject":"Merge \"Update allowed address pair documentation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2a244beb214b65267db58ed9afc7cf0464832d28"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-10-30 00:58:16.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-10-30 00:58:16.000000000","tz":0},"subject":"[OVN] Fix the revision number retrieval method","message":"[OVN] Fix the revision number retrieval method\n\nThe \"ovn_revision_numbers\" table has a unique constraint that is a\ncombination of the \"resource_uuid\" and the \"resource_type\". There is\na case where the resource_uuid can be the same for two registers.\nA router interface will create a single Neutron DB register (\"ports\")\nbut it will require two OVN DB registers (\"Logical_Switch_Port\" and\n\"Logical_Router_Ports\"). In this case is needed to define the\n\"resource_type\" when retrieving the revision number.\n\nThe exception \"RevisionNumberNotDefined\" will be thrown if only the\n\"resource_uuid\" is provided in the related case.\n\nCloses-Bug: #2085946\nChange-Id: I12079de78773f7409503392d4791848aea90cb7b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/63f60d878fc02f14f468a004ac1592b9765fcdab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/63f60d878fc02f14f468a004ac1592b9765fcdab"}]},"branch":"refs/heads/master"},"a298a37fe7ee41d25db02fdde36e134b01ef5d9a":{"kind":"REWORK","_number":2,"created":"2024-10-31 10:38:22.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/52/933752/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/52/933752/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/52/933752/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/52/933752/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/52/933752/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/52/933752/2"}}},"commit":{"parents":[{"commit":"2a244beb214b65267db58ed9afc7cf0464832d28","subject":"Merge \"Update allowed address pair documentation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2a244beb214b65267db58ed9afc7cf0464832d28"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-10-30 00:58:16.000000000","tz":0},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2024-10-31 10:38:22.000000000","tz":0},"subject":"[OVN] Fix the revision number retrieval method","message":"[OVN] Fix the revision number retrieval method\n\nThe \"ovn_revision_numbers\" table has a unique constraint that is a\ncombination of the \"resource_uuid\" and the \"resource_type\". There is\na case where the resource_uuid can be the same for two registers.\nA router interface will create a single Neutron DB register (\"ports\")\nbut it will require two OVN DB registers (\"Logical_Switch_Port\" and\n\"Logical_Router_Ports\"). In this case is needed to define the\n\"resource_type\" when retrieving the revision number.\n\nThe exception \"RevisionNumberNotDefined\" will be thrown if only the\n\"resource_uuid\" is provided in the related case.\n\nCloses-Bug: #2085946\nChange-Id: I12079de78773f7409503392d4791848aea90cb7b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a298a37fe7ee41d25db02fdde36e134b01ef5d9a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a298a37fe7ee41d25db02fdde36e134b01ef5d9a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
