)]}'
{"id":"openstack%2Fneutron~933754","triplet_id":"openstack%2Fneutron~master~If0de7235f6d00a3f034b5c9ef2a84d42cc86c517","project":"openstack/neutron","branch":"master","topic":"delete_duplicated_ovn_const","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2024-11-05 16:05:13.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If0de7235f6d00a3f034b5c9ef2a84d42cc86c517","subject":"[OVN] Delete duplicated OVN resource type constants","status":"MERGED","created":"2024-10-30 15:33:24.000000000","updated":"2024-11-05 16:06:33.000000000","submitted":"2024-11-05 16:05:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"933754-delete_duplicated_ovn_const","meta_rev_id":"1d07346ce8c5158daa0eb6f70a7bfdfffff2ec1a","_number":933754,"virtual_id_number":933754,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2024-11-01 22:02:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-11-05 16:05:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2024-11-05 13:34:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2024-11-04 17:02:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":1,"date":"2024-11-05 13:34:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-30 16:28:31.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-10-30 18:16:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-10-31 09:41:03.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2024-11-04 17:02:53.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2024-11-05 13:34:09.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"028f2e0e3ee520b8d402077ef5a698f6593b5fcd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-10-30 15:33:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1d7b4110c683eb075e16fc10333db5c3791656e1","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2024-10-30 16:28:31.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/October-2024/Arista_Tempest_Testrefs/changes/54/933754/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"e694ccd3fa9f7337714f5377e1315ca8993cd90c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-30 18:16:08.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/529ab4c108074736ac5547115d8b39a4\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/2f22d4a0677647f2958c29ce6581c373 : SUCCESS in 1h 49m 45s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/742829a967164aa29d34239097ff58ca : SUCCESS in 1h 30m 40s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/61881cd1838c466d8e7662709bbddf9d : SUCCESS in 2h 03m 55s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/30da547123424d7ba1ef1de3d6387f41 : SUCCESS in 30m 15s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f867f75056f44fa4be00ea8605aba3bd : FAILURE in 25m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b918d01f0b7c4249bdc0fe4deb4667b9 : SUCCESS in 9m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dd869f24e2984be1bdf4732d723fce3d : FAILURE in 22m 47s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/94f8b3ca697840688d38cbc8cbaecf0d : FAILURE in 20m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/838d3f35f51e410ba1f64e5f99ff1e75 : SUCCESS in 11m 49s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/dbe999e720cb4f3a8e4508bb5ffc5e3d : SUCCESS in 1h 24m 58s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/04c9a7ed4b524f1a8eda18b3b5bdd3f3 : SUCCESS in 1h 02m 59s\n- neutron-tox-mypy https://zuul.opendev.org/t/openstack/build/7d8e213cc276483db27aaceb88345625 : SUCCESS in 4m 37s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/f241053b71b349af9bb8e27e6c213358 : SUCCESS in 56m 32s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/1d34e5261c9f4dd58b2583778bfd62f6 : SUCCESS in 2h 40m 06s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/8a5ab1e0ba4a4b388249d9ec11fdd946 : SUCCESS in 47m 27s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/ec7d0cda4f5f4814a2e5a5de57dfa634 : SUCCESS in 1h 06m 35s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/7f127d2f1fda42e3ae2520329d6eb814 : SUCCESS in 1h 36m 03s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/ca20e4419c824011b3a4776fedef53e4 : SUCCESS in 1h 29m 07s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/de3c92683d2841bbb41273a0b8fd6c6b : SUCCESS in 1h 00m 23s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/93206f2dbf944980b17cc7618d35fa01 : SUCCESS in 50m 13s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/b8dd081c1ec7462298a2a91c0c9c0c84 : SUCCESS in 1h 01m 28s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/86f94ba880b2493faf60e20dc74cc8cb : SUCCESS in 1h 01m 54s","accounts_in_message":[],"_revision_number":1},{"id":"1c9c3efeca358ced8c45cc4de192f058ad64bb21","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-30 19:06:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e32c4584a034454a874e9c771bcf7de1\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8561f80f87df4ed8bef612f4ae8920d9 : FAILURE in 27m 43s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/d42c9bf13649453f942ae696edd2d7ca : FAILURE in 27m 13s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"59c06438c2ae2fc24839a12ce6acd8d8d7d52909","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-10-31 10:50:19.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1, Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"8406b7c3c35f5eb339816fc75cf0c19a68bb209a","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2024-10-31 11:44:09.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/October-2024/Arista_Tempest_Testrefs/changes/54/933754/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"cba72e0a850aaeb2a5c940c60b83b64b1d07cd1f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-31 13:05:27.000000000","message":"Patch Set 2: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d779f081f1204ec0b26f7f4f9b95f4fa\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/7a77b7eb53504ff78f3c91ab4b223da9 : SUCCESS in 1h 40m 40s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/e094bb4448a14f6baac5ddd3f2750844 : SUCCESS in 1h 32m 16s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/7d195fb19cc54cefa7673cf5c491ba3b : FAILURE in 1h 48m 31s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/cca2ee6df24a4159b2aac531c1f3728d : SUCCESS in 28m 00s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fa2af07bfe304400ae682623f03a1289 : SUCCESS in 34m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/04fb831dfd4544d7bf6ba265f306b28c : FAILURE in 4m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fb307816fcc04507a6e38f1078b79f7e : SUCCESS in 21m 19s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/cdba35699c064cfea8978a3954c0c6fb : SUCCESS in 20m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd5244c5e24243a8b0d8a87e242c41a3 : SUCCESS in 11m 14s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/d60a29ce41b04abf886f1f4ccb07caf5 : SUCCESS in 58m 10s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/1044d6ea1e31468e8a18ea938d9a4f51 : SUCCESS in 1h 05m 57s\n- neutron-tox-mypy https://zuul.opendev.org/t/openstack/build/3dc13bdd56ed4c7a92e2d9789d49b3b1 : SUCCESS in 5m 00s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/407b2628843449d2a334a1de93aafb4f : SUCCESS in 52m 30s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/828a46c2c97541edacf1ba42173aea88 : SUCCESS in 2h 01m 44s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/ceb03be274c54527924e5695b800da89 : SUCCESS in 1h 02m 51s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/2755bb2474c44a9cb712358de2932c46 : SUCCESS in 1h 18m 07s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/3a9edc60437d49caaf1f6a2453840406 : SUCCESS in 1h 42m 01s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/b9ed47f938ee4c8c83bc0e3afffa5289 : SUCCESS in 1h 25m 38s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/cc77cbe5fc20400ca9b1bcc47962aa8c : SUCCESS in 1h 03m 58s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/e269c1f51bb544f096903bc9cb3a6709 : SUCCESS in 57m 04s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/18b51cbf372943df805757968ce92ce0 : SUCCESS in 1h 04m 40s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/efd7ef824bd74e8aa83fa20a4655d35b : SUCCESS in 1h 12m 00s","accounts_in_message":[],"_revision_number":2},{"id":"475550eaebd35de43919bf4ba896df32ce822557","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-01 03:17:24.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c4d1d46944b94538a2dacff74eeffc01\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1066985a8f9b482baf6e2935ffd4cfa6 : SUCCESS in 50m 55s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/ccc0eeeb3fc54cec917c87a2e433b53c : SUCCESS in 55m 31s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e517ff7c87f19a00e4436902041adb8d6572e1f8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-11-01 21:27:13.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nOutdated Votes:\n* Verified+1, Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"9118d0d3ce0aa930600dff85b54fb339404833ca","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2024-11-01 22:02:45.000000000","message":"Patch Set 3: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/November-2024/Arista_Tempest_Testrefs/changes/54/933754/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"820c0a81d608566b0ef5bbdc370ce5cf7af70519","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-01 22:28:09.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a3b080b812da496d8a11493eb3e6f3c2\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3e8624c30d17485491a11b46059bff68 : SUCCESS in 41m 09s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/4b3ac288353b4bb4a872ff134c037dd1 : SUCCESS in 44m 31s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"296aadc87550b1691045491ed3d7807ec9885552","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-01 23:29:53.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/df267ee4267a4220b4ef5118b40efee3\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/6a07b404860d4a0ebd0d05bd077ddae4 : SUCCESS in 1h 54m 50s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/081c3f9186ce4fe38a4cf1a8741ae145 : SUCCESS in 2h 00m 32s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/e8c5fa40430c40b3a957a41c9b07ed6b : SUCCESS in 1h 47m 19s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/b4eccdc4b2594b4b8d34d8a3d558d59a : SUCCESS in 28m 54s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5a46523379a8462480e2292ae031a6f5 : SUCCESS in 46m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/62173a52bb7341f1a34bfa9cce054546 : SUCCESS in 11m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/30fad48b41374731923f46bc5cb8f05b : SUCCESS in 24m 03s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/cd2f634f96874c0da2aa5d855c210483 : SUCCESS in 16m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a92d459fa57140bdad1f96be24b4fb21 : SUCCESS in 9m 03s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/0c3a3b8428a0422fadf58cb20b20223b : SUCCESS in 1h 03m 42s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/1221de9ebd0149a9b105341edf1eade9 : SUCCESS in 38m 53s\n- neutron-tox-mypy https://zuul.opendev.org/t/openstack/build/68a14dc8934e481eae6caf42df3bf7d9 : SUCCESS in 4m 21s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/d38a415435e8456db2e39850bd75d857 : SUCCESS in 49m 35s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/0dac561e9b354da381276c9eea1bd59b : SUCCESS in 1h 36m 18s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/ebb0788479be48fd87cb3f37d35513df : SUCCESS in 36m 53s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/deefc8df3e5642d0a3a2d9863971576a : SUCCESS in 1h 12m 41s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/a974097dcc9b49ed823efd1fa8b24f12 : SUCCESS in 1h 40m 01s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/a21ea36ee7174aa2a6d0c0786e239863 : SUCCESS in 1h 13m 54s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/76a600d5459c44f098a0e9f3fe343815 : SUCCESS in 52m 43s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/3040bd4b2c6b49d7b02278da8593d9cd : SUCCESS in 44m 21s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/1b05f07b21814918a4ddceebbfc49d56 : SUCCESS in 57m 56s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/3cf979972b484e69b96c827f34e38929 : SUCCESS in 1h 07m 58s","accounts_in_message":[],"_revision_number":3},{"id":"315e54ad98ce95b92a8bb04722946b2e0d2f866f","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-11-04 17:02:53.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"c77e69d9b1b1a93de1660508fb88b2faddbfa8a7","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2024-11-05 13:34:09.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"18c67c7d317a39076dab0016a0a0746495ad32c9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-05 13:34:45.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"bfe73d03a8af83f50c8f9b8db04cc1e76732e867","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-05 16:05:13.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/93a2182c49724fb895e4a16aafc9531b\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/1047bd579ee440b3bb89854b58502c24 : SUCCESS in 1h 30m 45s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/4eefd43179be4d39809c244a0479c47b : SUCCESS in 2h 08m 18s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/2c318fcef1c348b7ab75bf47e9eefa0c : SUCCESS in 2h 07m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f2381219f0ea45c38669136e1f031a60 : SUCCESS in 10m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f645b5bebabc46f09807fd1be02996df : SUCCESS in 22m 32s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d977cb61455e4166a4741d6307749308 : SUCCESS in 24m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6732626b8be64084bb2394b565a3a9bf : SUCCESS in 9m 47s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/218bc6ec55b64598a0b6ad7cb50e5e62 : SUCCESS in 52m 14s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/dbd05fa95a8e41c4b93ae21bd56d45e4 : SUCCESS in 1h 46m 04s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/ba9f6a6b02ed4aa4b43278ea7b77bbc4 : SUCCESS in 1h 29m 12s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/5308b34b9e7e45a0bb44970cbc297498 : SUCCESS in 54m 32s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/93d56a476e14472faed35a8729fbc7b7 : SUCCESS in 57m 33s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/ed82ae2b59634b8992f389bf97cbf696 : SUCCESS in 57m 28s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/63625270487948ccaf8d41756162b8f7 : SUCCESS in 1h 09m 53s","accounts_in_message":[],"_revision_number":3},{"id":"e3becac1549ecff3f9a8f465e4fd85259a0a80e0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-05 16:05:13.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"1d07346ce8c5158daa0eb6f70a7bfdfffff2ec1a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-05 16:06:33.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f580fa26124d44e8b3b225ccc923c0cb\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/776f43b3f857462d947ca33890c0a008 : SUCCESS in 57s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"9bd0b21a370de96dfd6601c002a56168d6bdfbe5","revisions":{"08706908c8db039e888ff2b6c6c83dfd0f96b249":{"kind":"REWORK","_number":1,"created":"2024-10-30 15:33:24.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/54/933754/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/54/933754/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/54/933754/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/54/933754/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/54/933754/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/54/933754/1"}}},"commit":{"parents":[{"commit":"63f60d878fc02f14f468a004ac1592b9765fcdab","subject":"[OVN] Fix the revision number retrieval method","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/63f60d878fc02f14f468a004ac1592b9765fcdab"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-10-30 01:14:51.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-10-30 01:14:51.000000000","tz":0},"subject":"[OVN] Delete duplicated OVN resource type constants","message":"[OVN] Delete duplicated OVN resource type constants\n\nTrivial-Fix\n\nChange-Id: If0de7235f6d00a3f034b5c9ef2a84d42cc86c517\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/08706908c8db039e888ff2b6c6c83dfd0f96b249"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/08706908c8db039e888ff2b6c6c83dfd0f96b249"}]},"branch":"refs/heads/master"},"ae4525ceb271bcbc3e0b370650931450707dd63d":{"kind":"REWORK","_number":2,"created":"2024-10-31 10:50:19.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/54/933754/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/54/933754/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/54/933754/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/54/933754/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/54/933754/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/54/933754/2"}}},"commit":{"parents":[{"commit":"a298a37fe7ee41d25db02fdde36e134b01ef5d9a","subject":"[OVN] Fix the revision number retrieval method","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a298a37fe7ee41d25db02fdde36e134b01ef5d9a"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-10-30 01:14:51.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-10-30 20:32:17.000000000","tz":0},"subject":"[OVN] Delete duplicated OVN resource type constants","message":"[OVN] Delete duplicated OVN resource type constants\n\nTrivial-Fix\n\nChange-Id: If0de7235f6d00a3f034b5c9ef2a84d42cc86c517\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ae4525ceb271bcbc3e0b370650931450707dd63d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ae4525ceb271bcbc3e0b370650931450707dd63d"}]},"branch":"refs/heads/master"},"9bd0b21a370de96dfd6601c002a56168d6bdfbe5":{"kind":"REWORK","_number":3,"created":"2024-11-01 21:27:13.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/54/933754/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/54/933754/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/54/933754/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/54/933754/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/54/933754/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/54/933754/3"}}},"commit":{"parents":[{"commit":"a298a37fe7ee41d25db02fdde36e134b01ef5d9a","subject":"[OVN] Fix the revision number retrieval method","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a298a37fe7ee41d25db02fdde36e134b01ef5d9a"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-10-30 01:14:51.000000000","tz":0},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2024-11-01 21:27:13.000000000","tz":0},"subject":"[OVN] Delete duplicated OVN resource type constants","message":"[OVN] Delete duplicated OVN resource type constants\n\nTrivial-Fix\n\nChange-Id: If0de7235f6d00a3f034b5c9ef2a84d42cc86c517\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9bd0b21a370de96dfd6601c002a56168d6bdfbe5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9bd0b21a370de96dfd6601c002a56168d6bdfbe5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
